)]}'
{"id":"openstack%2Fmanila-tempest-plugin~805330","triplet_id":"openstack%2Fmanila-tempest-plugin~master~If525e97a5d456d6ddebb4bf9bc8ff6190c95a555","project":"openstack/manila-tempest-plugin","branch":"master","hashtags":[],"change_id":"If525e97a5d456d6ddebb4bf9bc8ff6190c95a555","subject":"Fix tests creating share without using share_size","status":"MERGED","created":"2021-08-20 04:40:41.000000000","updated":"2021-08-29 19:32:45.000000000","submitted":"2021-08-29 19:32:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805330","meta_rev_id":"6065d699c2ecc2049066e90a5aa1d9753a58cc0b","_number":805330,"virtual_id_number":805330,"owner":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-29 19:32:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"recommended":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"all":[{"value":1,"date":"2021-08-23 18:55:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":2,"date":"2021-08-29 18:22:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2021-08-23 21:18:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"value":1,"date":"2021-08-29 18:22:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-20 06:18:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-20 07:50:26.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-08-23 18:55:23.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"REVIEWER"},{"updated":"2021-08-23 21:18:47.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-08-26 18:16:31.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-26 18:40:34.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-08-29 18:22:39.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"07b9b7ae41fbc9625d554e2ee43a8fa883ee3a8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2021-08-20 04:40:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56e99473873ef6df7fee65793e136f787b7f6c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 06:18:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02d9adf757764a52a7f279e794d9122a : SUCCESS in 3m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/350a9aab641b4890a6f6eebcde853806 : SUCCESS in 30m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/04bf2a9c3963489485d560ef8960255d : SUCCESS in 31m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6d4fd5463625405cbe0586622f17de64 : FAILURE in 1h 36m 34s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/44bea50dd0864fc18056d42107e4b761 : SUCCESS in 1h 03m 20s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/73b3aae29c4a401085c5731e74339de8 : SUCCESS in 1h 05m 52s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/b4d8e49ac4884801a87cf8d0dd12a162 : SUCCESS in 1h 07m 05s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e29691e28db64670b9084b5f1008d487 : FAILURE in 1h 02m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d7e0314361244b7183c3131c018879f6 : SUCCESS in 1h 10m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/23f6047ee24d4515acaf33d851ecf3cb : FAILURE in 1h 23m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ba4ea6c278304bd4a5173da7c8cf1f99 : SUCCESS in 45m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e5036ba3a9fd46c3808e738c575c0e1f : FAILURE in 1h 20m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/14ef441d4d2d4cc5ad0baf2b83cef2bd : SUCCESS in 20m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecbbb4cec857cdb2397eac04ea829d58a986934f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-20 07:50:26.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/30/805330/1/check/DellEMC_Unity_Manila_dhss_true/39e73d9/DellEMC_Unity_Manila_dhss_true/281 : FAILURE in 3h 09m 32s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/30/805330/1/check/DellEMC_Unity_Manila_dhss_false/6ca069b/DellEMC_Unity_Manila_dhss_false/278 : SUCCESS in 1h 37m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"b05cfa248e14a31bb7dcce1c976238eb1483dec5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2021-08-23 17:56:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"831dce3ad224137ffc35f76b122f9417c3c2e476","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2021-08-23 18:55:23.000000000","message":"Patch Set 2: Code-Review+1\n\nSeems like this is the correct thing to do....","accounts_in_message":[],"_revision_number":2},{"id":"70e78923d579c787a39d00eee863e773d2e4b09b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 19:36:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5558bcf6b8a4028802bc2ec70dc408c : SUCCESS in 3m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/31554a41b67841ee876ec556b822cbe9 : SUCCESS in 33m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/998e91071f8347e2895f33341d6a655f : SUCCESS in 35m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/516ca5183412440f8a175ae0af37f26e : FAILURE in 1h 05m 43s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/297324787d0c4440b85163bce9622952 : SUCCESS in 1h 07m 36s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/2a497a5127e94c949d13a91b532ce8c9 : SUCCESS in 1h 03m 35s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/42ce770f22404043aeaefe927cdeac31 : SUCCESS in 1h 00m 39s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dcbd6f271ad74ef5bfe13a15786ad981 : FAILURE in 1h 03m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/fdf18972a21a44479d019bc2795b0c64 : SUCCESS in 1h 04m 38s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b5c6d8e347034e72af3e7554996c38fe : FAILURE in 1h 31m 15s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9b6c7b9e21064f4bb0c66477ed3a5f20 : SUCCESS in 42m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d633f24b135b4ae794569cd7bdb88e04 : FAILURE in 1h 38m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f61ed2963632441c9688eafbe81040b9 : SUCCESS in 21m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1244676ab3fde87be37fb7cefe051666630ac2a0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-23 21:18:47.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks, LGTM!","accounts_in_message":[],"_revision_number":2},{"id":"a1df4ddd24347c06538651e308aeb78518a67693","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-08-26 16:29:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"343779e6fe451720db0429153179ce9b75a2a882","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 18:16:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2faef16401a94ec0a63d7ec65bdcbc4d : SUCCESS in 4m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/96e8a62f9a2d4c3b86fbcd07654c6280 : SUCCESS in 32m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/708dcfe2a1464cada83697fc4d26c26c : SUCCESS in 36m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/52aec86fba2346bb8e1136fc01548abf : SUCCESS in 1h 25m 33s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/1632a3a3334c4c9fbc14cd30c1571a50 : SUCCESS in 1h 13m 10s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/652a48b903ac4d1281c951ffb624b61d : SUCCESS in 1h 17m 58s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/0cd8c08699df449089b74ac8e80e104c : SUCCESS in 1h 14m 01s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/674dc48bf6d84554827ee454f0d21941 : SUCCESS in 1h 10m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/355cb5f5ba1e4b89b38576e9dcf96dff : SUCCESS in 1h 15m 24s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/05be03e503f64af5a25edd974507208f : FAILURE in 1h 24m 12s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d5cd83ec05294e7e98fe9d7044f5159b : SUCCESS in 43m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7fc71dbdcb2b4999ace10740cbf5de54 : FAILURE in 1h 43m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5cfc2f37c7484e148c8f9d09541426bc : SUCCESS in 23m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72f12ddb6b0ef5e75a27b998c541b678ee77c321","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-26 18:16:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/30/805330/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/a7173e6/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/30/805330/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/5d991fd/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"36592ec7a93561e5d46d98af0bf8a2a0141bf5a9","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-08-29 18:22:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks Felipe, LGTM","accounts_in_message":[],"_revision_number":2},{"id":"2e0fafee8356e45fcb83fd4e737ebcfaff08959d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 18:23:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0fc5cde3ca75e9669aad95eecebd6d59b0a5549f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 19:32:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14c691db470a4d16923cb5dadd59896f : SUCCESS in 4m 43s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5b0e86170f8949bb9cf5e3dfad520699 : SUCCESS in 38m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/63a23a650903443183ccdca4431814a0 : SUCCESS in 37m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/925191a726844152846adc2c46bd2ccb : SUCCESS in 1h 05m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6065d699c2ecc2049066e90a5aa1d9753a58cc0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 19:32:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bef40b3f80c1fed866bebfa6b2a7dafe101d44ad","revisions":{"43615af5753af499a88dc1059bc0208e3297ac01":{"kind":"REWORK","_number":1,"created":"2021-08-20 04:40:41.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/30/805330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/805330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/805330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/805330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/805330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/805330/1"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2021-08-11 01:00:21.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2021-08-18 03:37:15.000000000","tz":-180},"subject":"WIP: Fix share group tests using CONF share_size and thin provision","message":"WIP: Fix share group tests using CONF share_size and thin provision\n\nChange-Id: If525e97a5d456d6ddebb4bf9bc8ff6190c95a555\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/43615af5753af499a88dc1059bc0208e3297ac01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/43615af5753af499a88dc1059bc0208e3297ac01"}]},"branch":"refs/heads/master"},"bef40b3f80c1fed866bebfa6b2a7dafe101d44ad":{"kind":"REWORK","_number":2,"created":"2021-08-23 17:56:07.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/30/805330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/805330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/805330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/805330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/805330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/805330/2"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2021-08-11 01:00:21.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2021-08-23 17:49:51.000000000","tz":-180},"subject":"Fix tests creating share without using share_size","message":"Fix tests creating share without using share_size\n\nSome tests from share group and replication are not creating\nshare following the `share_size` tempest CONF. This patch\nfixed them applying that configuration.\n\nChange-Id: If525e97a5d456d6ddebb4bf9bc8ff6190c95a555\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bef40b3f80c1fed866bebfa6b2a7dafe101d44ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bef40b3f80c1fed866bebfa6b2a7dafe101d44ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
