)]}'
{"id":"openstack%2Fmanila-tempest-plugin~805498","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I8caa456a5822d93a1ebb6d95e873c5ab5f574f32","project":"openstack/manila-tempest-plugin","branch":"master","topic":"DNM","hashtags":[],"change_id":"I8caa456a5822d93a1ebb6d95e873c5ab5f574f32","subject":"DNM","status":"ABANDONED","created":"2021-08-22 05:22:43.000000000","updated":"2021-08-26 04:10:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6e7a2ef653cad9b451a41ec0704135191904329a","_number":805498,"virtual_id_number":805498,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-08-22 08:32:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-22 08:32:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-22 10:29:49.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"eea8b616dd5c6f3d15a1f3858f601eaa16b315c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-22 05:22:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a329c285b713c7400884973abbf4edb25404c89e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 08:32:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec35e0016b334552885d1b553adca684 : SUCCESS in 3m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7c7d2c81366b418a98b9c5acd72daff8 : SUCCESS in 28m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2fae924a50104ab48a47b5b312b2acfe : SUCCESS in 42m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/db3eeeffcca14b1495642d7eafb1f1c5 : FAILURE in 1h 20m 53s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c772821937d84365873948e68d62bd96 : SUCCESS in 1h 12m 03s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/ded39c006d904c219b38b73aeec80f93 : SUCCESS in 1h 01m 19s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/4f19b129775a465f929527a2d07cd5f7 : SUCCESS in 1h 01m 34s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4c677baed4f24bebb9447cf827817844 : FAILURE in 1h 24m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1f558b9f001d41a4b3ba33edba8cdf80 : SUCCESS in 58m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2fa4e92c851046898cf6df1cceb80abb : FAILURE in 1h 41m 23s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9b631c74b1894e89861962510979d037 : SUCCESS in 35m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a4a5fd83a4f74b929dbed10528517a9c : TIMED_OUT in 3h 03m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ecedb4a6ee78447994194eab1a33b536 : SUCCESS in 24m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"609aa25e26e366faa7b23644624041685fb09598","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-08-22 10:29:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true http://publiclogs.emc.com/98/805498/1/check/DellEMC_Unity_Manila_dhss_true/1e424e8/DellEMC_Unity_Manila_dhss_true/291 : FAILURE in 2h 40m 00s\n- DellEMC_Unity_Manila_dhss_false http://publiclogs.emc.com/98/805498/1/check/DellEMC_Unity_Manila_dhss_false/662df4c/DellEMC_Unity_Manila_dhss_false/288 : FAILURE in 1h 47m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6e7a2ef653cad9b451a41ec0704135191904329a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-26 04:10:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3aa20cb9b24972e47c4840df567a56d60c09554","revisions":{"c3aa20cb9b24972e47c4840df567a56d60c09554":{"kind":"REWORK","_number":1,"created":"2021-08-22 05:22:43.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/98/805498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/98/805498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/805498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/805498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/805498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/98/805498/1"}}},"commit":{"parents":[{"commit":"da415b4f1c3107ca05245ed9abd53a3ab67d83e3","subject":"Fix parsing of share group type json","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da415b4f1c3107ca05245ed9abd53a3ab67d83e3"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-22 05:22:24.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-22 05:22:24.000000000","tz":180},"subject":"DNM","message":"DNM\n\nChange-Id: I8caa456a5822d93a1ebb6d95e873c5ab5f574f32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c3aa20cb9b24972e47c4840df567a56d60c09554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c3aa20cb9b24972e47c4840df567a56d60c09554"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
