)]}'
{"id":"openstack%2Fmanila-tempest-plugin~805938","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I53bb9851445038f81032485817389ec31e2b7341","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-09-14 11:50:37.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2022-10-04 13:32:48.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-09-22 15:21:38.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-09-13 12:59:38.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"32411":{"account":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"last_update":"2022-10-04 13:32:48.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-10-04 13:32:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I53bb9851445038f81032485817389ec31e2b7341","subject":"Add S-RBAC tests for manila","status":"MERGED","created":"2021-08-25 04:49:52.000000000","updated":"2022-10-04 13:32:48.000000000","submitted":"2022-10-04 13:32:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":138,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"805938-secure-rbac","meta_rev_id":"5392f232c4b16142917fef2a100b0db6d0108d7e","_number":805938,"virtual_id_number":805938,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-04 13:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-10-04 12:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-10-03 20:00:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-10-04 12:07:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"}],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 09:12:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-25 09:13:03.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-25 10:55:19.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-26 13:16:07.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-10-26 17:58:59.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"CC"},{"updated":"2021-12-08 15:50:58.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2022-01-27 21:24:45.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2022-02-15 13:40:11.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"CC"},{"updated":"2022-05-26 15:44:00.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"CC"},{"updated":"2022-06-17 19:30:42.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-06-23 08:20:37.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-06-23 15:34:42.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"state":"CC"},{"updated":"2022-06-23 15:34:42.000000000","updated_by":{"_account_id":30555,"name":"Fernando Ferraz","display_name":"Fernando Ferraz","email":"fesilva@redhat.com","username":"fernandoperches"},"reviewer":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"state":"REVIEWER"},{"updated":"2022-08-25 13:40:20.000000000","updated_by":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2022-09-13 12:59:38.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-09-14 22:53:04.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"add0b8b50af7b94218fd9d00a536fd43eb909f51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-25 04:49:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e3cc2fcc59187a31c2b33307449bac04a69c910","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-25 05:09:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ffbd75f688ccf6c3e0856139d65f08381cbfdbe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-25 05:53:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"658897ca862ae778f71e7023b3120b1ccee57ba8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-25 06:36:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3d11f4712583ce565a466de3eec9298682a1281b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-08-25 07:11:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9ae9fa0a3eb417b853265a8880d1feef2960896e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 09:12:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e05b69972a4c72bb1c0fa3afd2e3d9 : SUCCESS in 4m 06s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/663557e5ef904245922be1e0cf161c72 : SUCCESS in 33m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/578783d6c1e3427bb92b6c26702a3ba7 : SUCCESS in 36m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8b017ff52fe54dfe82fdff2c4c47c2f4 : SUCCESS in 1h 17m 47s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/11316c7436164ba99086df0903f98fae : SUCCESS in 1h 10m 49s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/36fa02b5f6b34ad885e4537fc2714fdf : SUCCESS in 1h 19m 42s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/bab01174ca444a0bbcfb31c3c2196ce5 : SUCCESS in 1h 05m 24s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cbf11447de4f49708e83c5d06c8daa9d : SUCCESS in 1h 02m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8fc7a10dea2644969219bc874b23aada : FAILURE in 1h 42m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b37b8e1bb12a43beb74cf5f9edf8fa71 : FAILURE in 1h 29m 43s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e0b5b86272824525bbd8b916878d7a19 : SUCCESS in 46m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/597f7f70658b4f3cb991447e3d1959b7 : FAILURE in 1h 57m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3c9d4c4f0eeb444e89186ca6c9e1eea0 : SUCCESS in 25m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ec94fe8db913e9ca85e43480fc8f9e5d6551fcbe","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-08-25 09:13:03.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/38/805938/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/fa76fbd/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/38/805938/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/5c00e48/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"c2eebbae279440357573516fc864cf01377ad354","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-08-25 10:55:19.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila http://publiclogs.emc.com/38/805938/5/check/DellEMC_VNX_Manila/e0b371c/DellEMC_VNX_Manila/136 : SUCCESS in 1h 42m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"e28c7b6b342609df37cbf7a2f38d9f2e911d1c7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-10-26 13:02:10.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"d2b94680163fc3ae7053ca0e6c241295b344f9ef","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-26 13:16:07.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/6/check/DellEMC_Unity_Manila_dhss_true/da2c79c/DellEMC_Unity_Manila_dhss_true/43 : FAILURE in 13m 43s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/6/check/DellEMC_Unity_Manila_dhss_false/96df9c0/DellEMC_Unity_Manila_dhss_false/46 : FAILURE in 12m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"30f2b7ef4fec92e15e740275807b0df7f9d0fb50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 15:02:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75bc7ad721e44f1b991c28a8ca998a31 : SUCCESS in 3m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1e278bfa57514c7d9bd875a6dc2ff744 : SUCCESS in 28m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/00647c75855f469a827321751c7acba8 : SUCCESS in 39m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/37e27d8ec0c843de867e281d7cf65da4 : SUCCESS in 1h 20m 54s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/bb634e8e359647d7a9e5e65c62592da4 : SUCCESS in 1h 22m 28s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/65a1a846f55b4f05a885c3caf1be726e : SUCCESS in 1h 35m 41s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/63fa748ca4cf433a94833b374d93d408 : FAILURE in 1h 07m 50s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/383af23f25b2481c99c33b94ee93ca96 : FAILURE in 1h 12m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4b26050ca17b4e57b5c50c8fd67d511f : FAILURE in 1h 22m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/03dcc7209ae74d7ab411a29818066cdd : FAILURE in 1h 33m 47s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f695449b758245dfb813587dbd67fcb5 : SUCCESS in 52m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/636161c8cfb544419b39bbe1c2198adb : FAILURE in 1h 48m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9a8889711981474db0ed26de01fbd22e : SUCCESS in 26m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2810700e47d3a3243b8a6d2c9e27ceed714dfb3f","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-10-26 17:58:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1408ccf6bdbe26d49b984558386ae87cbbd8002f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-26 18:13:33.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/6/check/DellEMC_Unity_Manila_dhss_true/b19736a/DellEMC_Unity_Manila_dhss_true/47 : FAILURE in 14m 21s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/6/check/DellEMC_Unity_Manila_dhss_false/912e5bc/DellEMC_Unity_Manila_dhss_false/50 : FAILURE in 1m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"683b7aeb4c16d95bcf9b24650b56875da1b20668","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-10-27 12:16:07.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2e5e5e8c6bb92556d7990f4ab9c26a3539f5571d","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-10-27 12:18:12.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"02283c3f4be75ca737eb8200becc8db473c1969a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-27 12:31:33.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/7/check/DellEMC_Unity_Manila_dhss_true/6f03c3b/DellEMC_Unity_Manila_dhss_true/54 : FAILURE in 15m 12s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/7/check/DellEMC_Unity_Manila_dhss_false/bda55e7/DellEMC_Unity_Manila_dhss_false/57 : FAILURE in 14m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"c006ecd8501a6e2cf7b3a067acf911cebc8a22ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 14:41:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5e4156af0774832adaf2ac4020b463b : SUCCESS in 3m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5fa2a6df608849909ac07c8dec44813a : SUCCESS in 37m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/71d0afcd9a3048fba90deefb679c36b7 : SUCCESS in 31m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ee6eb4112db5412c85c8d7dcc720216d : SUCCESS in 1h 02m 06s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/72c4b309d3d144669d8dbe7695443df6 : SUCCESS in 1h 27m 15s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/064211c0896f4682bc3147076d9bacb2 : SUCCESS in 1h 59m 04s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/b764357251234594b6a443b2e2882dd5 : SUCCESS in 1h 21m 10s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1aaf61ba7de34997977279a1c7fcd2db : SUCCESS in 1h 46m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c242887294e44e139d7b808dfb1fd3fd : FAILURE in 2h 04m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/db5630029a264ed0a2c39bfb187efe37 : FAILURE in 2h 20m 14s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cc0fa777a3cc49229f0ee95ebf757ac1 : SUCCESS in 43m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b7a4ead4bcb5453cbde8f4f2affbbb42 : FAILURE in 1h 49m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2b405205ec1f44138fc67fd7c79029c6 : SUCCESS in 25m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"78a1b76142f787e18ff4d0ee898eba69d3aac708","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-10-27 14:41:28.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/38/805938/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/6d03006/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/38/805938/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/308773a/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"3091d104b285035ca47c366f25125db0084c8969","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-10-27 14:54:15.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/38/805938/7/check/DellEMC_VNX_Manila/bf7121a/DellEMC_VNX_Manila/26 : FAILURE in 12m 52s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"0d8f6540bef495a71f67ffc704499f607f438b71","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-27 14:54:56.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/7/check/DellEMC_Unity_Manila_dhss_true/59d01f2/DellEMC_Unity_Manila_dhss_true/55 : FAILURE in 13m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/7/check/DellEMC_Unity_Manila_dhss_false/2a24f9c/DellEMC_Unity_Manila_dhss_false/58 : FAILURE in 12m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"82fa4b5f95db5965530bc0d8eedeee5e586640c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-11-01 12:22:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c94a4d64d760915c832c237e9ea3a05ff349c8fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-11-01 12:59:37.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"fbf999893979e0e8bcab49adb68b00c4ef078080","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-11-01 14:39:03.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e604d3b4b0b225ed97e6d839d09fd9147dba3a3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 15:07:08.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e8483b304964a97a642e6db4d2901e7 : SUCCESS in 3m 51s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7d5701ea98e4403aadff89995a27326a : SUCCESS in 29m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/af067b486ef541e29e6a3129eb11ad4c : SUCCESS in 41m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/43547060d5994bd4a6a50ddc2c33542e : SUCCESS in 1h 20m 33s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/7181ff40f65745aebb88056f6b7549cf : SUCCESS in 1h 43m 02s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/16a94f728609452397744e418333af4e : SUCCESS in 1h 14m 33s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/56970197604445758f602d14497ec729 : SUCCESS in 1h 05m 56s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/250bbcf66f624806ac528e51ef6c8354 : SUCCESS in 1h 29m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/76fc91633b6c44df9edb2b13c61e6757 : FAILURE in 1h 31m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b5e06be38e994033a2b60480aeacee59 : FAILURE in 2h 02m 13s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b0eef108aa594965af41f98f25aa3dcd : SUCCESS in 50m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/74b33b60f33f444dad0d4b1361716e9a : FAILURE in 2h 05m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/86911493f9ba4cb3ab30075db83962e2 : FAILURE in 28m 49s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"234339a7cc093804fb14daa856a713085894fc2a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-01 15:07:31.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/38/805938/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/5240cd9/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/38/805938/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1ebc45f/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"5302375bd6f0f73dd80dcfe79d4a138a88fc98e6","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-11-01 15:33:49.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/38/805938/9/check/DellEMC_VNX_Manila/57a5628/DellEMC_VNX_Manila/31 : FAILURE in 12m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":9},{"id":"2497994f1fa6307828e6919a7086d429315dccec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-11-01 19:24:55.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9dd34898340be7a6f429de881a90becbd67be966","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-11-01 19:27:57.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c6cd4be49e6c4192f8fcc940b313abeecbea865d","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-11-01 19:29:09.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"56c8234ef54aed7e002eb3cfbf88790f4a110c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 22:37:20.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bef7ebfe35a41a88bd47f39c885d94a : SUCCESS in 3m 48s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f862fb0b60ab4c83bd2cbc49d3036352 : SUCCESS in 34m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/14a56433393648568fc39005ff64f300 : SUCCESS in 38m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0f26c8ce9973484f9763ff00ca4e91dd : SUCCESS in 1h 19m 44s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/65ca9327eb3a4e83b7a452b676ff2480 : SUCCESS in 1h 28m 49s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/1e50f3b06b40496584d1ffb8f04e52c2 : SUCCESS in 1h 30m 09s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/8e4673b88cc846eb8d55c2ffb6d9eb0d : SUCCESS in 1h 11m 24s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/279dbcf166b54427a4055d51ab1b437e : SUCCESS in 1h 26m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ffb5973220f8414a98f78d9292fb70de : FAILURE in 1h 28m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e31e19e0d5844ed581e8a2c36e7aa770 : FAILURE in 1h 42m 39s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e89b9560963b490882ef73e5468b4895 : SUCCESS in 57m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/da64a9bc180e4ee68cfd3b86cfd66e6b : TIMED_OUT in 3h 03m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d379f596aa3c48e194a526f5a4319e9d : SUCCESS in 25m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"cb671a0f35b01741eaa0734b1c3f6b728effcb2c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-01 22:37:40.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/38/805938/11/upstream-check/manila-cDOT-manila-tempest-plugin-ss/4cbac58/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/38/805938/11/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/6f4dbd7/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"f360ada43565a4207e46ea8be2c61d5eb55ad8f1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-11-01 22:49:46.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/38/805938/11/check/DellEMC_VNX_Manila/0ebd3bb/DellEMC_VNX_Manila/32 : FAILURE in 12m 13s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":11},{"id":"30de519a14c5efa0fb892137c9a0940f464d27dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-11-16 12:08:46.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"7bb1592b9b92aaf01a8d9bebe44db85b398a789c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 14:14:45.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fde3fee82444df9be8eb1557c0202ff : SUCCESS in 4m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8a6124fc768a4cb19b5530ae3caacc6d : SUCCESS in 32m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fc40cb72efcb416c9962a7630bd9bf8a : SUCCESS in 34m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e525bf128d2546b5a0e1b35417bc7a15 : SUCCESS in 1h 28m 47s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9e5f45b8e3bc42d586c441e95daa8f0b : SUCCESS in 1h 10m 26s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/013c0a20458f49d2a611a80a3fadda46 : SUCCESS in 1h 24m 53s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/111787f2ac9f4d318a5a94d402319572 : SUCCESS in 1h 33m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e6c5cafebe5948feba4f5907405a0286 : SUCCESS in 1h 08m 48s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e7713ca21f7248c0a23e342793f826b9 : FAILURE in 1h 33m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c82bf90daa864cafb5da5cbcda48b86e : FAILURE in 1h 57m 07s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/70962a85fcea4ecbaf5400ee5d9cb0da : SUCCESS in 52m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3eeadf12e8a2485d88fe01fee92a644e : FAILURE in 19m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b46b441bfcae4bdd8c2b8438d2b8047b : SUCCESS in 29m 04s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"294afc025a0d26af10f10e60fd500c4870a2841f","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-16 14:15:04.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/38/805938/12/upstream-check/manila-cDOT-manila-tempest-plugin-ss/4cafe9a/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/38/805938/12/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/665e402/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"feacfd1d067329575eec5e094fc96c6d42786253","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-11-16 15:07:59.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"aba942b195b188f51057d5512725c3d3b63230ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 17:18:14.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f62a0c5f80dd47909f5bc7d737148b3b : SUCCESS in 7m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/44269d0ec1ba4648b304c64b61369f28 : SUCCESS in 42m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e3a13d9520374fc28c6fef5f4422b38e : SUCCESS in 39m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/848390fbd8b941a9a1ccd6e9cba9d9d7 : SUCCESS in 1h 20m 18s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b1a3a34ea0f34a978ba4e6318ac7d6bf : SUCCESS in 1h 22m 30s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/a6eb5d97b0724a04a96cd0fc7b224e6d : SUCCESS in 1h 35m 58s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/73af82fa60114caf8c15fad6aa6ce83a : SUCCESS in 1h 14m 22s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7c93ff63b1eb49c9b7dc9e729e2b7c5e : SUCCESS in 1h 28m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ea0839469a1d4cc8867332439b26d9a2 : FAILURE in 1h 39m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2d5e6843babf4497b1f4bb98b1aebce7 : FAILURE in 2h 01m 45s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/791cd27713d94aa297a8863c0e7f07c2 : SUCCESS in 50m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e238436190504173b5a0f87ae340cf2d : FAILURE in 1h 52m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/938f9d2e112741b0b2e84cc04fd96def : SUCCESS in 27m 44s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"ff1064ac0e740cfa5352b4148c542e07a7cce4e2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-16 17:18:39.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/38/805938/13/upstream-check/manila-cDOT-manila-tempest-plugin-ss/164f0f9/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/38/805938/13/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1517228/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"d78c73838466f1c0d4467f5a139ec42a9c84401e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-11-16 19:40:08.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/38/805938/13/check/DellEMC_VNX_Manila/69cb732/DellEMC_VNX_Manila/40 : SUCCESS in 2h 21m 40s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":13},{"id":"0f8d6ea0ebdabc247a6c0eeefd62c1535000db80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-11-17 15:42:48.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"54608d6e0a7eb1a5beeb1b068c3c00a9bb65e294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 17:46:43.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca0d24268b0c4aff90c106158fdbb28a : SUCCESS in 4m 21s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ad84066ceba94bceb8ce938818cb8379 : SUCCESS in 38m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/526975e63b824a7293c2a6f422120d71 : SUCCESS in 35m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d41f0ac0f6744cfab63623d7d3e2c000 : SUCCESS in 1h 16m 53s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/de5df54e16354758bbc44ff2bddaf18e : SUCCESS in 1h 23m 47s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/131566b1aa6148459604dab0d5fbe1b9 : SUCCESS in 1h 25m 17s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/91737a5936d44162886cea799c2ea23e : SUCCESS in 1h 23m 56s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c77a5707c31e466bad9f0f93e7b1b899 : SUCCESS in 1h 26m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a539d7ff80d94f9b9a6214dbed71e830 : FAILURE in 1h 35m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3013ddcaad68482ca3a5592e2d1d0100 : FAILURE in 1h 57m 53s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/76b107ffed724b818f6af3c1d6c54255 : SUCCESS in 45m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e749ff34674646f084aa7cf704157bb8 : FAILURE in 1h 50m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7929355c153e4aaa9a89cd13fa5e0c17 : SUCCESS in 22m 07s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"4065e703d33e5e8e21dfd9ca8eb67ebcfb2c24f0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-17 17:47:00.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/38/805938/14/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c2c512e/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/38/805938/14/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/6a5fff1/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"9e1a58b0bb0b4c5fabcfb4040cff064ea69fc9cb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-11-17 20:54:15.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/38/805938/14/check/DellEMC_VNX_Manila/493e205/DellEMC_VNX_Manila/43 : SUCCESS in 1h 37m 13s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":14},{"id":"a46c9e53ccec009d215e359417beb4c22baafa3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-11-18 09:36:39.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"ebef3dfd4b7a5f1269a5ec4bee36c2d09e4f4667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 09:37:26.000000000","message":"Patch Set 15: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  Job devstack-enforce-scope in openstack/manila-tempest-plugin is not permitted to shadow job devstack-enforce-scope in openstack/devstack\n\nThe error appears in the following job stanza:\n\n  job:\n      name: devstack-enforce-scope\n      parent: manila-tempest-plugin-lvm\n      description: |\n        This job runs the devstack with scope checks enabled.\n      vars:\n        tempest_test_regex: \u0027manila_tempest_tests.tests.rbac\u0027\n        devstack_localrc:\n          MANILA_ENFORCE_SCOPE: true\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 557, column 3","accounts_in_message":[],"_revision_number":15},{"id":"c86e232ebd09e14720ae0318836c9050e3759625","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-11-18 09:40:03.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"c464d80f47b4145cc3f7522ae57586872b455449","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-11-18 09:44:10.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"0d47d17e00999f9c47e061ef06c0cb13038ef3ea","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-11-18 11:12:28.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"9aa6d04e3a024abe9b3ebb0c587da8806d037858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 11:31:21.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/907bc38037844f5ca06331e0a22fc20f : SUCCESS in 3m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b6a39bb9d34d41d1bb732c2c893d85ff : FAILURE in 16m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1953e8a6f6ab40a39ac3c10519b71af7 : FAILURE in 12m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5a90a17d8f3d402fad2dc3a9c5334026 : FAILURE in 22m 21s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/6e15f43a2f1d486ab49b48607afaf719 : SUCCESS in 1h 02m 35s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/4a5d2a0882f141f09ba51197c3549faa : SUCCESS in 1h 07m 16s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/2b8424755789478b8031b42bdef7bfe1 : SUCCESS in 1h 45m 15s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/7ef73ba984444c78850bceb7b852373c : FAILURE in 33m 24s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ede3251180ad4c319674a6391dd6254b : FAILURE in 34m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/35c29900a0814ff4b28bc775cf73fa67 : FAILURE in 35m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3de64c5881274edeaaca3fad301c9080 : FAILURE in 24m 27s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/448e3898829046869b2f9586357ac82d : FAILURE in 20m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d03751970c3d4d84a505ed096498e1cd : FAILURE in 30m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/84c3e321e6194af8b9a042aceb903a27 : FAILURE in 17m 03s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"e5b826c75770decda5fe8fc739e7e9cad7737823","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-11-18 11:51:17.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"750e82be466df1dbc771d52f5a0e325fab01819c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 13:32:47.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef32a42b017941d1aae72b79071f903f : SUCCESS in 3m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4e717a911bb14676972b97d450cb5a97 : FAILURE in 24m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4ada6c4792824ee0b3c6e9c043b086b9 : FAILURE in 20m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7dfec0b476a74bcfa6a8e1b9324646a3 : FAILURE in 33m 28s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d5074f80c94044ceb91e7cb6af6e6e75 : SUCCESS in 1h 33m 32s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/5674e629c50243f380c51bd9217150cc : SUCCESS in 1h 20m 50s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/73099cd930d740b5adb5c01b77629df0 : SUCCESS in 1h 19m 27s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/1f0b177e39c548b7bd4a37dcc5583d0f : FAILURE in 41m 10s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/68d679a8a61344daafc73fa951daf9c7 : FAILURE in 40m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/143469c4bc72481398c332038761b093 : FAILURE in 47m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c6afaffbc1e0495d99df013316f018fb : FAILURE in 44m 46s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/427fda09e46a4c0988d64e21b5d11af9 : FAILURE in 37m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/13d721779d3a4b7bafa77d23cd8a9160 : FAILURE in 44m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c4b97135d485472abc1bbf3f178e5451 : FAILURE in 19m 29s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"714cc77c1bb8f6e4d54152a2557fb398a9dce37a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-11-23 09:41:46.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"74e7f4736d9d5d8c74ca4fcf64dc76e5ca92d16a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-11-23 10:08:39.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"399fbb756efa771109908278c11c95d487b5b699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-11-23 12:44:30.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"2ec17f1a0bfe0b8bb81b3ed152634fe6b53df282","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 13:30:05.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc4364707c8b4a0f812abab92e23d195 : SUCCESS in 4m 28s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/5ad9ae7fce3e494fb20004deba1970fa : FAILURE in 40m 47s","accounts_in_message":[],"_revision_number":20},{"id":"4d8bb25bd231135dad1b876feb77506b67d8529d","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-11-28 09:39:48.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":20},{"id":"5bf21294ff5e39d76a6301d24274f17a1eafc37a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-08 15:50:58.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"f4520018440e8a59ad744cfb6acc10a2d6418e10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 16:30:51.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c99605eb064f4a56914c086d04557657 : SUCCESS in 4m 09s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/f6aad9d429c54ac0997a0a785f0defa8 : FAILURE in 35m 00s","accounts_in_message":[],"_revision_number":20},{"id":"36f00606db37f060a3f6d2f5e8b7dd827fe7e2ad","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-12 11:40:00.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"be78deeeaeb7e0733670a3ccbcff00d11961aa96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-12 12:05:50.000000000","message":"Patch Set 20:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":20},{"id":"0bd62c293db5635e7ab3961a9c3ccf19460f88c7","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-12 17:15:31.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"eaa923590d41c2992f56ec438221d8bbe2de8385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-12 17:45:06.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0259193270f401ca0792ac0358f8b27 : SUCCESS in 3m 39s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/b36da80083bd4a33ba6954272e62ca10 : FAILURE in 28m 43s","accounts_in_message":[],"_revision_number":20},{"id":"b2f9f0dce38430470c33877da3e6d97ed09aa5fd","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-13 09:31:57.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"6023025deb84bf1ff41442a2d013d76391e3b47e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 10:00:24.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c124325197e48cb8e092d28ed84157c : SUCCESS in 3m 45s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/50369a3fc9e646de8abf7be4c70320b9 : FAILURE in 26m 29s","accounts_in_message":[],"_revision_number":20},{"id":"19e0c7a41b85d882892333aa2dfa8656a9bc62cd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-14 07:44:40.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"a8f8caf32d87916e4c448f27e6a01c7eb76440b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 08:13:23.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0495847ea5734e51b5c894e04cc38477 : SUCCESS in 2m 57s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/6737e998ff96453d9ff1425459b9f555 : FAILURE in 27m 21s","accounts_in_message":[],"_revision_number":21},{"id":"23c18e3b61a304ea88681791aa2446a3db5f2f75","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-14 09:00:09.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"03c611244b4639ed949cdd82603a4812f827f55c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-14 09:17:54.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"1aad12dc859a10d3a8b46f4818affb45965181d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 09:31:41.000000000","message":"Patch Set 23: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":23},{"id":"c830a9b0fc1f28080fbc3b97f5bb268224b3893e","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-14 09:56:32.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"3d9d5328e319b099535f960bdb1fa30b56f07f4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 10:30:03.000000000","message":"Patch Set 23:\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fb5014a9cd542bda3e559e8c17c1534 : FAILURE in 4m 23s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/3e4575cfbdb843beaf274d5194f561e8 : FAILURE in 31m 18s","accounts_in_message":[],"_revision_number":23},{"id":"4bd2847c62114e08a69a9630544b73b7e14b2d8f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-14 19:45:57.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"82ce44dccc0bbcf5d38bba532c1fbf3e32b547e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 20:37:11.000000000","message":"Patch Set 23:\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed968eef98de4d8a804411056c89fa1b : FAILURE in 3m 36s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/7fb54bc80d544ae99b60f57b7ec6d3c2 : FAILURE in 49m 37s","accounts_in_message":[],"_revision_number":23},{"id":"d36434872b5ae1dbda511c7ab0bcd983c7c1eb88","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-21 11:20:08.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"54551937430a75fe08011dec54b6bba54c489437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 12:01:57.000000000","message":"Patch Set 24: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2da2cf2a935247468f1181efe435bb42 : FAILURE in 4m 59s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/21e41393948343ef86ff1de22f54ff83 : FAILURE in 38m 02s","accounts_in_message":[],"_revision_number":24},{"id":"5cefd2b6c4c50a0a0349058cdcb2d33e677ed1d2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-21 12:43:17.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/24/check/DellEMC_Unity_Manila_dhss_true/5a7cc2a/DellEMC_Unity_Manila_dhss_true/176 : FAILURE in 1h 22m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/24/check/DellEMC_Unity_Manila_dhss_false/a5814ed/DellEMC_Unity_Manila_dhss_false/169 : FAILURE in 14m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":24},{"id":"d964a254eb336e533a2d28808011e78f94adcf46","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-21 13:10:33.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"9327dd507fbaadbdd31442eee6ecce4bb1ef1281","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 13:54:38.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a51c3d335b124dffb831e8bf1d3289b4 : SUCCESS in 3m 41s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/aa2d286ec1f84d95bbfc303943f3db2f : FAILURE in 41m 34s","accounts_in_message":[],"_revision_number":25},{"id":"93879a0fbaad29e00b1a3e2937f422dbad9b60e4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-21 14:27:40.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"976c9eaeb8acc35c5dcc1b30d01fa6f8acff8b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 15:22:12.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/255bfbed87ad4ed5a9f773e1c76adb84 : SUCCESS in 3m 50s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/bbedbfdcb9f5493a917372bf419ba946 : FAILURE in 46m 16s","accounts_in_message":[],"_revision_number":26},{"id":"b803748e005a90ed45edddf8d880597191ef671e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-21 15:42:20.000000000","message":"Patch Set 26:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/26/check/DellEMC_Unity_Manila_dhss_true/c7f9f7c/DellEMC_Unity_Manila_dhss_true/178 : FAILURE in 1h 14m 26s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/26/check/DellEMC_Unity_Manila_dhss_false/7b38720/DellEMC_Unity_Manila_dhss_false/171 : FAILURE in 13m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":26},{"id":"42298cca5374bb366dae3cbb447b4f763b9dece4","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-22 16:14:18.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"c2f7bfae2b302ae3bccf83e6abbad1332f8882f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-22 16:32:38.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"0ae93efef6b174ff8d7194b328bb03db56b8591e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 17:12:36.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4101a1b7c71446a3881654a4a9b15905 : SUCCESS in 4m 47s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/76ee2ba44fa7493ba61e0fe56db0692e : FAILURE in 35m 21s","accounts_in_message":[],"_revision_number":27},{"id":"f72d8d4f786d37b5a706b7e6b63929c3d7262163","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-22 19:11:38.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"30a29bad1bb29b73ede2e65c1fb963952ae28f76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 19:52:35.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19f27aeec19c44829c449187ca7a22b6 : SUCCESS in 3m 55s\n- manila-tempest-plugin-rbac https://zuul.opendev.org/t/openstack/build/a4d73e52a6914d9b9ea8bdb6a3dcf0fd : FAILURE in 38m 35s","accounts_in_message":[],"_revision_number":28},{"id":"8c2b97b31f9160cd811a38d709dda896c6982850","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-23 09:28:25.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"ee6958cddb124278a0196a001b6c04fc54eb67b4","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-23 09:32:32.000000000","message":"Patch Set 29:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":29},{"id":"ca13a4c217a9da0e56b53d4b3278779df12f407d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 11:44:14.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96a253afe5a14bc3917a439d00e2c314 : SUCCESS in 3m 38s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a6704bebc5a84d48b29cee1d3a608c8d : SUCCESS in 33m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cf422c82634e4cb6a86f534d700371be : SUCCESS in 40m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a627a53df79548d096cea8b397a1599c : SUCCESS in 1h 12m 51s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5004ee7b375b45588bc885c2eb10379b : SUCCESS in 1h 01m 47s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/fae7ff596d114cc5b00de6bdca7f6005 : SUCCESS in 1h 24m 59s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/76385a0650c54395aae6bc1c549c7bf5 : SUCCESS in 1h 16m 11s\n- manila-tempest-plugin-cephfs-nfs-rbac https://zuul.opendev.org/t/openstack/build/f0c094db564242adb5502659efb97c37 : SUCCESS in 43m 55s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6bd4e8a319c449579ec04e70cc97d146 : FAILURE in 1h 19m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e12c99f45a414822a84ee96c1dd45e0a : SUCCESS in 1h 21m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b66f3d02b81c4b83a9a8f119f314766c : FAILURE in 1h 51m 05s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6eadbf2b376e46ef95da54567dceb44e : SUCCESS in 47m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8b872fef8b394d459ad6f1cde3404177 : FAILURE in 2h 08m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/245be85476e14bb88ff81f3ed602fe71 : SUCCESS in 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"4704646e9db58ab8eb32e342c82da33fed025c47","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-23 11:44:36.000000000","message":"Patch Set 29:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/38/805938/29/upstream-check/manila-cDOT-manila-tempest-plugin-ss/6ab4af3/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/38/805938/29/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/06e9bba/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":29},{"id":"75dab7a74c74071560f4d365291d02237611bcf7","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-12-23 12:57:27.000000000","message":"Patch Set 29:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/38/805938/29/check/DellEMC_VNX_Manila/e3fac12/DellEMC_VNX_Manila/81 : FAILURE in 1h 12m 58s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":29},{"id":"8fc8a725b34cba2d07dd3a064d10b2ecb29d3c2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-11 10:00:38.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"79ba6f02bbb91c1b3477d79c1bc37d05cba2761f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 10:02:16.000000000","message":"Patch Set 30: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  Job manila-tempest-plugin-cephfs-nfs-rbac not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - tempest-plugin-jobs\n      check:\n        jobs:\n          - manila-tempest-plugin-dummy-no-dhss\n          - manila-tempest-plugin-dummy-dhss\n          - manila-tempest-plugin-lvm\n          - manila-tempest-plugin-lvm-xena\n          - manila-tempest-plugin-lvm-wallaby\n          - manila-tempest-plugin-lvm-victoria\n          - manila-tempest-plugin-cephfs-nfs-rbac\n          - manila-tempest-plugin-zfsonlinux:\n              voting: false\n          - manila-tempest-plugin-cephfs-native:\n              voting: false\n          - manila-tempest-plugin-cephfs-nfs:\n              voting: false\n          - manila-tempest-plugin-container:\n              voting: false\n          - manila-tempest-plugin-generic:\n              voting: false\n          - manila-tempest-plugin-glusterfs-nfs:\n              voting: false\n      gate:\n        queue: manila\n        jobs:\n          - manila-tempest-plugin-dummy-no-dhss\n          - manila-tempest-plugin-dummy-dhss\n          - manila-tempest-plugin-lvm\n      experimental:\n        jobs:\n          - manila-tempest-plugin-glusterfs-native:\n              voting: false\n\n  in \"openstack/manila-tempest-plugin/zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":30},{"id":"0766575cbf3fc2a47170dcab83c33bb0d7d0746c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-11 10:40:28.000000000","message":"Patch Set 30:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/30/check/DellEMC_Unity_Manila_dhss_true/d97bdfa/DellEMC_Unity_Manila_dhss_true/216 : FAILURE in 39m 37s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/30/check/DellEMC_Unity_Manila_dhss_false/b515a06/DellEMC_Unity_Manila_dhss_false/206 : FAILURE in 13m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":30},{"id":"854751e2d602256eeca96612a21eb58e471ddf8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-11 11:22:48.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"6fc4e89e439975fd0dcd20b2385b60c32a69bd48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-11 12:27:19.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"b88624499cf62b04d0120cc2a124b6cbdeead1c4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-11 13:14:28.000000000","message":"Patch Set 32:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/32/check/DellEMC_Unity_Manila_dhss_true/78c7bed/DellEMC_Unity_Manila_dhss_true/219 : FAILURE in 46m 50s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/32/check/DellEMC_Unity_Manila_dhss_false/1887065/DellEMC_Unity_Manila_dhss_false/210 : FAILURE in 39m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":32},{"id":"1e858031a6d9fee9671d9a3e9a9c905d4ca42dde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-11 13:25:51.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"84f76566db72f652761019e3febc40cadbe810e1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-11 14:17:11.000000000","message":"Patch Set 33:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/33/check/DellEMC_Unity_Manila_dhss_true/46581b5/DellEMC_Unity_Manila_dhss_true/220 : FAILURE in 51m 05s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/33/check/DellEMC_Unity_Manila_dhss_false/7626b9f/DellEMC_Unity_Manila_dhss_false/211 : FAILURE in 47m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":33},{"id":"d336cb98134cd276b6edc6aea1aa581ee7f61a11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-11 14:39:21.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"f08c1d9f096f20a27bed1e3f18114fa8dbe1573c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-11 15:25:26.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"8ecd5fba5328aa06063dcf5584e9b0728333f6e8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-11 16:15:10.000000000","message":"Patch Set 35:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/35/check/DellEMC_Unity_Manila_dhss_true/1700610/DellEMC_Unity_Manila_dhss_true/222 : FAILURE in 49m 30s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/35/check/DellEMC_Unity_Manila_dhss_false/e2d7888/DellEMC_Unity_Manila_dhss_false/213 : FAILURE in 44m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":35},{"id":"7ed0bd37f89e5f6d9e52b563d8a8951701aea7cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 17:45:11.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ea3a8e289b84ca19e7f06eeb9e5e994 : SUCCESS in 4m 20s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c918c63421294a5291a0a69e769da016 : SUCCESS in 36m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/316a06cd471b4f08a09987e270a63cca : SUCCESS in 41m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8046a725474743e2923787004847a369 : SUCCESS in 1h 20m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5ac610104d454eb1abdb09d63d6d5b13 : SUCCESS in 1h 10m 00s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c991222bc12f4e5da8ee782bd2929923 : SUCCESS in 1h 08m 59s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/4196158673084d69ad7098c935d5f7e3 : SUCCESS in 1h 07m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5be4ff8ee5994d98ad1018ae28f02e82 : SUCCESS in 23m 51s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3bef288ddad04fd095bd73e737257e0f : SUCCESS in 1h 20m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f44005997f71414b9f8f7d5f64b7927b : SUCCESS in 1h 25m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bd5d7be94bd04694842e15f3f7b878b8 : FAILURE in 2h 05m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f78cd41fc6c0494e8328d14dd61089a1 : SUCCESS in 50m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2391d4ee74c1452392fb087281e03b28 : FAILURE in 1h 52m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b7832c62245c4bb4a9acab123891c158 : SUCCESS in 27m 35s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"1370b9810580518668aff1d208740bb94dae1a64","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-11 17:45:32.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/38/805938/35/upstream-check/manila-cDOT-manila-tempest-plugin-ss/96dbc78/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/38/805938/35/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/8c3eb4e/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":35},{"id":"b7a0bc01a66138700b852de7c894e14e65d2eafd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-11 18:35:15.000000000","message":"Patch Set 35:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/35/check/DellEMC_Unity_Manila_dhss_true/6a877bb/DellEMC_Unity_Manila_dhss_true/223 : FAILURE in 49m 50s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/35/check/DellEMC_Unity_Manila_dhss_false/d3abf3b/DellEMC_Unity_Manila_dhss_false/214 : FAILURE in 45m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":35},{"id":"7097a2ccf0bbe1c2bbc5e8fc9b046d0a83c418fa","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-01-18 12:08:50.000000000","message":"Patch Set 35:\n\n(2 comments)","accounts_in_message":[],"_revision_number":35},{"id":"aedede1c7ca768084e7ba04939a47f0b3df3d9c1","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-18 12:21:07.000000000","message":"Patch Set 35:\n\n(2 comments)","accounts_in_message":[],"_revision_number":35},{"id":"cac6abe775d6a28d2d6d90de969397d0ae3efd6f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-01-27 21:24:45.000000000","message":"Patch Set 35:\n\n(3 comments)","accounts_in_message":[],"_revision_number":35},{"id":"63f0febb0764b634c9c85ea382b8636a2978e160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-30 12:58:39.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"3e277d812ec85e741c6abaf9da1fe883dc459789","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-30 13:07:54.000000000","message":"Patch Set 36:\n\n(3 comments)","accounts_in_message":[],"_revision_number":36},{"id":"1ef4cdb3890d0b9b1ea4436a57264ac3d8a0d5e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-30 13:19:03.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"249e16bbbd6cdbce6e9b58426e2c4dc7f471de27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-30 15:22:13.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67cf79ecec6e4a42951d6b9f76447799 : SUCCESS in 3m 04s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c30bec1cc48448dfaaab34c0318df639 : SUCCESS in 36m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bb429a2193754e3aac9a46e2d1763c48 : SUCCESS in 40m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/40729032e47942c4a15372360e7c1514 : SUCCESS in 1h 26m 30s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/86f41c9d0673459cab5414d60e25c8d4 : FAILURE in 17m 58s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b18e23d3c79c4e67aa97ba900758b6c3 : SUCCESS in 1h 14m 42s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/d6ca9446836940da9ceb90e938873c93 : SUCCESS in 1h 21m 34s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/44ddbfeceead4dfc89ed471f27cdd153 : FAILURE in 33m 59s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f7ff152b974545e79c62a5f90e44418b : FAILURE in 1h 14m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f5775adaa73d458a9002a5e246efbbbd : SUCCESS in 1h 57m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1d7bc4e5b6f14648a09dbe68d5b124ac : FAILURE in 1h 55m 44s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fe8978730a184aff85fdb64860e28956 : SUCCESS in 55m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a82dd4e7e7b848bbbc322509c7f8c0b1 : FAILURE in 1h 28m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/11ace137d01543499e1f9de1a153b66b : SUCCESS in 27m 51s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"5924175ba06dc87752cd2afd19968048448cf12b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-30 15:27:22.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"ffd9830d39b31388f9d46f237f6a6507d894584d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-30 17:04:38.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd6f84e87d2547a7bee5565c52ecf8d5 : SUCCESS in 5m 26s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/767e687831e145c194795ed4d68e044f : SUCCESS in 27m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ed9eae266177446bbd5c24a1dd458866 : SUCCESS in 31m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dd685ab898454908aed0df4a7db27668 : SUCCESS in 1h 13m 39s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a532e6519e064988af22f70cedda8099 : FAILURE in 9m 15s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/2b8f8cdabc55485cb59e4612e5f035c5 : SUCCESS in 1h 04m 02s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/e62c93032a7f4d8bac40c21fb8296299 : SUCCESS in 1h 01m 08s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8a09da4d4a53441b941e2badb19b63ad : FAILURE in 24m 02s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/225046fb3e734c9d905c010ae5db407b : SUCCESS in 56m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/eb0d21bc6eb84a51a8200dbdc446128a : SUCCESS in 1h 21m 24s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1fba246c64364f039a63e94e2f595b72 : FAILURE in 1h 22m 03s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c5b0921373ec46928c4251dd89ec329a : SUCCESS in 43m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5a56bf16cbe34e0cb54180b82aeeae71 : FAILURE in 1h 35m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6db30171946f4d3d94c738f584dfada1 : SUCCESS in 19m 54s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"b898ff1cd869a3a082c8db79048b31ed7b4461e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-30 17:33:14.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"3a29866e34e5bb2d957c7627aef99ef68293dadd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-30 19:14:38.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbfdfc2253944fa6bbba4284d7cf983d : SUCCESS in 3m 24s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e605c0b9d89d4b648842f527ce19586b : SUCCESS in 35m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0849ded731544e82b8c5afcb21121aac : SUCCESS in 41m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d6577be8e1534bd39c20c57fb4ea2f51 : SUCCESS in 51m 44s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/984ed9f10f0845ceba59b573e25bbb87 : FAILURE in 7m 02s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/0bfc9ae5735d49e08d67d600f2af34f7 : SUCCESS in 47m 54s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/f6757ffcd9e84af38bb78925d1a8068e : FAILURE in 1h 05m 58s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/35fe253e3a08415ea8aeded9d3d3954d : FAILURE in 22m 13s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a105b1bea52944bbab814d267256420f : FAILURE in 1h 06m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f0b06d057d7346878557d207d2fa9d20 : SUCCESS in 1h 02m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ec09266e91b448e796db2e826d9798ec : FAILURE in 1h 39m 53s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5687184d09fa438ab2d49fa5d833be3b : SUCCESS in 40m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/757a51ff78db4c67a4844c43d6b15424 : FAILURE in 1h 32m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/04e2257f3d21407ca8a437753e97bb24 : SUCCESS in 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"376bff45560ff3402a564344602ff07fef8c8570","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-30 19:49:48.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"df285004e2b856f2bd9169f917193cb79d8baeec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-30 21:07:57.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"a40b3f16fa86549ebbb06c5ba286cf79a4ca1f9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-30 22:45:37.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/633bcabdf6124ede97d806256d1a8441 : SUCCESS in 3m 23s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8005ce91d4d5449085fc233405a30727 : SUCCESS in 33m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8c03ce892b27428dabcf58ccc7e47598 : SUCCESS in 43m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7dc5100cf18d40658a532f1a0f299b72 : SUCCESS in 1h 21m 15s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c28e86103b964a4cbb1ba1b31b91812f : FAILURE in 14m 45s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/35ad6c25728243f3bfd72bc79f97fc98 : SUCCESS in 1h 13m 59s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/cb4923f4a1a64fdaa2a79659e8c8d04c : SUCCESS in 1h 16m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/82892d7e1d1043989ac6b2a506b7313a : SUCCESS in 30m 52s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b57a154d539a4ca6b0c0ea333f90016e : SUCCESS in 1h 20m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/eb166eb6e3a343ee9c26e8f978b9967e : SUCCESS in 1h 30m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e4396b37ab0e4f7b8176623d71c8c615 : FAILURE in 1h 15m 20s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/81f00392515b4cc1b03cc4dbe77c896d : SUCCESS in 43m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a887b9b33bf24ecea0ca649400754c3a : FAILURE in 1h 25m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0a909fa1a36c4ab0be0f5fd7236b0409 : SUCCESS in 25m 58s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"c2000d2e331fe5dd4a95fd5c65cf0d8aade506c2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-31 01:48:29.000000000","message":"Patch Set 41:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/41/check/DellEMC_Unity_Manila_dhss_true/a699ea3/DellEMC_Unity_Manila_dhss_true/278 : SUCCESS in 4h 40m 17s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/41/check/DellEMC_Unity_Manila_dhss_false/900e2c1/DellEMC_Unity_Manila_dhss_false/286 : FAILURE in 1h 19m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":41},{"id":"d15f40100c5e54c1205b40769c3a05655285d142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-31 14:45:21.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"ce2ab5539c52cc4b0d71f37ea51268264cc24ac6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-31 14:48:22.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"7929274ef7b37a94748e3c443de1fcf5eb1d3815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 15:38:22.000000000","message":"Patch Set 43: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8176487fcc6c4995abd4eb6366fa5839 : FAILURE in 3m 51s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e34ea9473ea5403cadf17cc8eb2a2f77 : FAILURE in 19m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/09bd9a2d09454337a073de15edc4bd5e : FAILURE in 17m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b1646d75179b4b69837e3ecf37e6f12a : FAILURE in 33m 35s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f9f4900fd67e455993bd3345c369764d : FAILURE in 36m 53s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/40fa076c07d540bfb484cce75bd9fbde : FAILURE in 37m 09s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/40db25a602884069b8f6e56fc2871768 : FAILURE in 31m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b46379b84bd84e72a5ddf38d8135fa9e : FAILURE in 19m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d2e0f4a2b8134e94895728bdef1a22af : FAILURE in 38m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1d51118179e549afadcd6493c32eb430 : FAILURE in 41m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/47687d951c3f479d995a05e2c7e0a0b3 : FAILURE in 31m 41s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/df8975ea5c6940f7990666d7d2c7197c : FAILURE in 35m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1a1588084b284925b3338ff0796b22fb : FAILURE in 37m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/616798ea43da4b81b76f4fa4454d21e8 : FAILURE in 23m 27s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"c700cec16eeed39d26ea9a32c08a3bc64ef086dd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-31 15:54:15.000000000","message":"Patch Set 43:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/43/check/DellEMC_Unity_Manila_dhss_true/b8db336/DellEMC_Unity_Manila_dhss_true/280 : FAILURE in 1h 05m 34s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/43/check/DellEMC_Unity_Manila_dhss_false/8b01494/DellEMC_Unity_Manila_dhss_false/288 : FAILURE in 34m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":43},{"id":"8cfdded58e7af0bb25265be91c2db6d8100374f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-31 15:54:52.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"872a71c463adafc244969a7815d5238602a388bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 18:33:18.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1623183024004ff189277342b105dc91 : SUCCESS in 6m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/06debd3679854a108ddc04b2a61ace3e : SUCCESS in 25m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7ffec8f65f8846b29ba80903be59802f : SUCCESS in 35m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/eb503edee2bc4cf893b98a0a3df618ae : SUCCESS in 1h 03m 50s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/abcbcd5abd534fe69756eec678a09263 : SUCCESS in 50m 23s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/53696908279c4c0fbeba5e20fd977caa : SUCCESS in 59m 59s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/0924707d105b44899795556e0140558b : SUCCESS in 56m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d890e6c68bcc4ffc942716a67ad1d6d5 : FAILURE in 29m 22s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/82bef19581194eaa9cb775087db60e0a : FAILURE in 1h 09m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d6f8843f47b44e2d83e3f74c924131e9 : SUCCESS in 1h 15m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/44ac42178e8d43be8b2116841e2b1e1d : FAILURE in 1h 36m 12s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e6e2c77ae46145229ebcbc57d15798e8 : SUCCESS in 46m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/62f21b33195542a28be339a0b5f1d273 : FAILURE in 1h 35m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/de4bcf8641cb4f82b0d260dc13544191 : SUCCESS in 23m 04s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"e311e9ca3714a0e99604ae897e76e70a6e1cdd08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-31 19:16:14.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"e4c7299ff9bf790a5d61809678ab28c68ffdac0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 21:15:11.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b51a23b368946549d7d67d1582107fa : SUCCESS in 6m 37s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/176cb9f18c4f4a669e8ac4a4afea0d99 : SUCCESS in 29m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/392d18edad1843f8a287285f6084039f : SUCCESS in 50m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9a78d215084a447198a437777c28b2d8 : SUCCESS in 1h 15m 43s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/2e2f1b4ad22749cb9d42582b1028db2f : SUCCESS in 1h 20m 19s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/4d10182408a24c828df9fa65c7b5f29f : SUCCESS in 1h 18m 46s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/9fd0503920f24afb8fb5b938df12c398 : SUCCESS in 57m 21s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/760c5dadbafe4eb3883637b2d5c86e76 : SUCCESS in 21m 20s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/37c23f826fc34aa99b62308dbb7ca7f6 : SUCCESS in 1h 06m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/241f841be1d74adaae45b3f78f9daf89 : SUCCESS in 1h 12m 23s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9a6fc962a87344448c7296db39a664a5 : FAILURE in 1h 47m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/672069787bb74779bd7de25b65ae6636 : SUCCESS in 40m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b993ddc0a7cf4a16931838878ad7a2ec : FAILURE in 1h 26m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ce2c4595432a4f7bb3058d3eba07a040 : FAILURE in 20m 58s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"9e58772b04d7f17380cc9d8e273c499003b20638","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-31 21:59:24.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/45/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b48cfa2/ : FAILURE in 40m 17s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/45/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/5322cf9/ : FAILURE in 41m 45s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":45},{"id":"e7563f6474e41496584a4b939843f98bd7af2708","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-31 22:41:42.000000000","message":"Patch Set 45:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/45/check/DellEMC_Unity_Manila_dhss_true/2d11aed/DellEMC_Unity_Manila_dhss_true/282 : SUCCESS in 3h 25m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/45/check/DellEMC_Unity_Manila_dhss_false/50acb67/DellEMC_Unity_Manila_dhss_false/290 : SUCCESS in 1h 11m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":45},{"id":"157b5c946e579b5cea1c58dcbc8ffe1670c1313b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-10 14:44:47.000000000","message":"Patch Set 45:\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"4aace1a5060617a6f99b9ed1933abcc0c361c69d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-11 15:58:35.000000000","message":"Topic rbac removed","accounts_in_message":[],"_revision_number":45},{"id":"6bbfd420a6f8b623c8a78acc36668c42c0533b2c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-11 15:58:41.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":45},{"id":"a8a82892e137d2ee24277df763b8807c0c8b0811","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-13 09:58:31.000000000","message":"Uploaded patch set 46.","accounts_in_message":[],"_revision_number":46},{"id":"81016010536e355866bbc00be552161ca7f19825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 11:35:42.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caf87bb8445b4342bc38da885c1aea7d : SUCCESS in 2m 41s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/69d0f8f0a1474d519b4dc30c6e30024b : SUCCESS in 26m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b54b7303f7754917948aed179e309e13 : SUCCESS in 34m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/285639a5716445f88a7e8702661b0a03 : SUCCESS in 1h 02m 07s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/2b883dd739f64502b53bd86cd9bd9345 : SUCCESS in 57m 21s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e720a4693a5c42f5b2a33cf35e4292ed : SUCCESS in 50m 17s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/b488085e841a4a1c8491deed58d2527b : SUCCESS in 55m 26s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/efe945209b4e4b1d8efe1886c8457fc0 : FAILURE in 20m 24s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b0ebd936f55e4b01a1840f2e7ba8eb0e : SUCCESS in 1h 06m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/694df6b4ef1541b39969061979df5611 : SUCCESS in 1h 06m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9ba53f992fae4225b3b01c77cfa4192d : FAILURE in 1h 26m 41s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cb044deefff5477cbdfe5bb2eade0eb3 : SUCCESS in 36m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b9c1766b84104b28a5421956beafdb36 : FAILURE in 1h 29m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0450f7d6e977484aac1afaa7c186a888 : SUCCESS in 16m 29s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"b0c51860d6ca1423d5116aeafc95c8660c7314d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-13 12:00:30.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"b7fb5e1ffa3b8af1507ed307ba3bde4fa3c176a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-13 13:27:25.000000000","message":"Uploaded patch set 48.","accounts_in_message":[],"_revision_number":48},{"id":"ed23c3179175c2458d91758ab8022012b38b6404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 15:14:24.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c71c68c9ffe549a8a84495078072bf95 : SUCCESS in 3m 11s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/efb2580add34400faedc0596a6cc54bd : SUCCESS in 23m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/08bcbf54c26146b7a6f707d53d54fb8a : SUCCESS in 28m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1c4b6101a1834319946941d9e3622e06 : SUCCESS in 1h 07m 35s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/7530993aeef342a8b5ac4c8396c33422 : SUCCESS in 54m 48s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/fcc545f13b0648a8aa68360d2493799f : SUCCESS in 56m 16s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/672c347cbddf4fe4ab7507b753531d57 : SUCCESS in 1h 20m 14s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/86d984988ca149aa9f04cd905b0dd6d0 : SUCCESS in 21m 36s\n- manila-tempest-plugin-dummy-dhss-rbac https://zuul.opendev.org/t/openstack/build/74e20b39fde843c5a96634c833607bf0 : FAILURE in 14m 51s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0de1b090ef0b498f80cdbfd20590c7c7 : SUCCESS in 1h 11m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3832574e505540e4b9430783df161606 : FAILURE in 1h 10m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/91d5a47be6f34af08321e17f2e5c3280 : FAILURE in 1h 34m 18s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5e813a31ece24cad8798d5585d1456fb : SUCCESS in 42m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/000d420ed128448fbe2d6d8b1e396ab6 : FAILURE in 1h 45m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2714f679c25547b6b5c5e675f6b66993 : SUCCESS in 20m 07s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"94322370b44cab41b28a9e3bcb389b4a76174cd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-13 15:40:40.000000000","message":"Uploaded patch set 49.","accounts_in_message":[],"_revision_number":49},{"id":"0a287ea91ab4d0b43979c6f0328f2c7d41a7f663","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-13 17:49:23.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/024be29ac95046cabb13e60617ec297d : SUCCESS in 3m 41s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c92facb9f0d54897aec53a9c49726d45 : SUCCESS in 20m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0682171d5d3644028bd041ecd09705f6 : SUCCESS in 36m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bc6ccc1aa21b48c38395a8bbd6f04f1f : SUCCESS in 57m 02s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f85bf47d19074366bad069b528563de9 : SUCCESS in 51m 33s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/17fb2419ed7c4ab1bc2308533883f155 : SUCCESS in 51m 47s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/1a93630b8d6141eb86ef48b43a6dffca : SUCCESS in 2h 07m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4dc54da9091b44478523ab414f8016a2 : SUCCESS in 23m 03s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6036f0c6b81e4cbe8c1934c4dd243de7 : SUCCESS in 57m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bdca54bcb61b43d8953efff7dbeb4289 : SUCCESS in 1h 01m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/164e7c58ee1944b5bf956affc5c4e707 : FAILURE in 1h 40m 49s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7da67689a0414dc38acc0994aaf9efc1 : SUCCESS in 44m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/32cac3a4edb645c9b9db2d3c6bccf417 : FAILURE in 1h 48m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ba646e2ef708440ba493a86b32c60959 : SUCCESS in 16m 02s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"4a598910284c7abf3a794f543b090293de0795ea","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-13 20:48:40.000000000","message":"Patch Set 49:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/49/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b79dbea/ : SUCCESS in 2h 58m 58s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/49/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/aca4b8a/ : SUCCESS in 1h 49m 47s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":49},{"id":"abd6b00c662f747a3c55d7eea323431996442639","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-15 13:39:03.000000000","message":"Uploaded patch set 50: Commit message was updated.","accounts_in_message":[],"_revision_number":50},{"id":"03cf14c152425391530a19c62a1b5e27a3507a78","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-15 13:50:27.000000000","message":"Patch Set 50:\n\n(1 comment)","accounts_in_message":[],"_revision_number":50},{"id":"5c976fcad7988f592acd5dab980ee310ace41952","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 16:11:30.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe8caf94b6b749cc8f91af757036e5b2 : SUCCESS in 4m 54s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ebce234d027442afb71b9d9cf25880e1 : SUCCESS in 39m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/90408efcb74340e09e7dc6ba890806a3 : SUCCESS in 39m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d05dcac79ece41cd98c4f2d11e24a69b : SUCCESS in 1h 20m 48s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/cabc63002fa7404e87f52d1f4eafc8fc : SUCCESS in 1h 03m 45s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/fd324e0d6feb4835bf19d8f105c51436 : SUCCESS in 1h 16m 56s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/37518e9f35174638ab0e5bd0e19f6d34 : SUCCESS in 1h 17m 43s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/178f29b7d7a64f9c85fdbc6580f45f38 : SUCCESS in 35m 31s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5d9746444ab14e0d9746f64791064bee : SUCCESS in 1h 22m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/058430088f724d21a21f52146e4bc7af : SUCCESS in 1h 25m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d240421f01d34bc4b1a814cd0afad6fd : FAILURE in 1h 40m 11s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/155eb767db454c119d826b6580ef9804 : SUCCESS in 54m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/76e709e8fb3e40ceb7ffb8c0312be151 : FAILURE in 2h 22m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/625f066b731c400980228de21b34c2a4 : SUCCESS in 28m 50s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"bc139677fede9e12c8c0319b19ebabe77c99283b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-02-15 17:58:02.000000000","message":"Patch Set 50:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/38/805938/50/check/DellEMC_VNX_Manila/da6b490/DellEMC_VNX_Manila/139 : FAILURE in 56m 37s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":50},{"id":"c0a2c2c0c44b7c4227f53d998f1d267d9d73e876","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-15 19:13:46.000000000","message":"Patch Set 50:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/50/upstream-check/manila-cDOT-manila-tempest-plugin-ss/fe532c6/ : SUCCESS in 3h 01m 58s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/50/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/9eaf964/ : SUCCESS in 1h 53m 24s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":50},{"id":"10277f38fe55560633e3e1fbb73a7b5670fdba60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-22 13:24:47.000000000","message":"Uploaded patch set 51.","accounts_in_message":[],"_revision_number":51},{"id":"0d24b93239e40df37bdeaac5d5d80fab2df01198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 16:07:42.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c31fc4f55a144783af2ca8c44f5bcceb : SUCCESS in 3m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/34586abaa49d421b96d73f3efaac3a78 : SUCCESS in 36m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4481ea0187f84dd084c77e8139230e3a : SUCCESS in 37m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c1ab795d5e714222a74e7ce839baefd7 : FAILURE in 1h 21m 08s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/b09f0a96b6ef4be1812a1a95b5cb0a98 : SUCCESS in 1h 10m 26s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/edac174d7dd542ac872c4a9fbdee8929 : SUCCESS in 1h 34m 13s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/812cd3d97e3840dba8371f1da2b9d68a : SUCCESS in 1h 08m 18s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0b9999f1085b4ff38e511be40ffaa893 : SUCCESS in 34m 07s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d484e91601ec45169bf1027c5803a33d : SUCCESS in 1h 16m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8c1770d1943e4633b950335ca0781151 : SUCCESS in 1h 40m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f48d508dca4b4acaa801fec53d118a9f : FAILURE in 2h 02m 18s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6d7a913bc4be457f95a6a5857e0725e3 : SUCCESS in 49m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/db2ac2f85988487aa01a88aa61a48acc : FAILURE in 2h 35m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0ea86472083044b386e7b5c630fd66f5 : SUCCESS in 31m 57s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"444208fbe2fee6a28e3c38ffa92d6052a6e68d0c","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-23 05:40:02.000000000","message":"Patch Set 51:\n\n(1 comment)","accounts_in_message":[],"_revision_number":51},{"id":"58fb4c2404c0e63a7a2124676de432bd08aced64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 07:13:11.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/841f10ea595a4e92a8ef1276c043f449 : SUCCESS in 4m 11s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6ae2f8991a6641e7a818741421d57416 : SUCCESS in 25m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a6490df834ff45dcb25ba3552abe3860 : SUCCESS in 27m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/335c6b41b8374601a2e2b41290ec0f4c : SUCCESS in 1h 13m 31s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/064ab44ef96d4731b38628d1406b17b9 : SUCCESS in 49m 55s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/7df64d6b5f324b01bc00eb778fe16032 : SUCCESS in 1h 00m 18s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/6638a779f9314c0c8ba4e2e3fe538884 : SUCCESS in 49m 01s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7320040e7b094793a7353260c5b53b04 : SUCCESS in 25m 21s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1656e18dc39f4729bd7d908fc0bc5ad4 : SUCCESS in 1h 15m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/69411c245da247d8815b0a07aa89d393 : SUCCESS in 1h 17m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/42fd5460fa394737a12936f110532b31 : FAILURE in 1h 31m 13s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4b80839a55c249e18501b821abe1d954 : SUCCESS in 49m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9a8eb274c1fb44a2beca668a82fa1793 : FAILURE in 1h 21m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9f90b02bac71449b90ec8aedbd80a536 : SUCCESS in 19m 21s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"cb373a24a6e5702d49384623189668a12a4563f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-05-24 13:58:10.000000000","message":"Patch Set 52: Patch Set 51 was rebased","accounts_in_message":[],"_revision_number":52},{"id":"27abca676dd79738bacde51c87a4b1dddaee6337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 16:22:26.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1c82c1022a94d31930f4e2579e75e8c : SUCCESS in 3m 04s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/653707e8633d4563a36422b8f07a6926 : SUCCESS in 34m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4cc61311c01d4d4d9e2783618b9fdda7 : SUCCESS in 47m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fa1a38fa900a4c25ba072f4b384f84aa : SUCCESS in 1h 04m 26s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/7e7d00e0a3ec42798444ad7a8447825a : SUCCESS in 1h 05m 39s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/406ddfd91f674acc9caa3d64db247a48 : SUCCESS in 1h 16m 37s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/5a606100cd334b8b995bb5eeb6562d10 : SUCCESS in 1h 11m 44s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/97d503ea7a104cc797fa12db61be6faf : SUCCESS in 34m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1a440e94c1cb4901b02aa8980e57612c : SUCCESS in 1h 08m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/112927f816674245be4abea05d671c04 : SUCCESS in 1h 21m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4dda7a90e0184f1d8f533deaa805d155 : FAILURE in 1h 47m 48s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/89b71d458f734530992fb54a34e2643c : SUCCESS in 1h 04m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8f262ccc8eda493da1fe8149ffc60838 : SUCCESS in 2h 21m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4c5405632634430fb888b6e2c75627d6 : SUCCESS in 22m 00s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"8ae8176e2b966c520ac2c909d5d85f19ec59ff44","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-24 16:24:00.000000000","message":"Patch Set 52:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/38/805938/52/check/DellEMC_VNX_Manila/bb0bdd1/DellEMC_VNX_Manila/274 : FAILURE in 1m 15s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":52},{"id":"a76e5219335ddf218dc615da7fae5081cb4a8409","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-24 16:34:37.000000000","message":"Patch Set 52:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/52/check/DellEMC_Unity_Manila_dhss_true/ffb7ab8/DellEMC_Unity_Manila_dhss_true/491 : FAILURE in 2h 36m 12s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/52/check/DellEMC_Unity_Manila_dhss_false/6262f8a/DellEMC_Unity_Manila_dhss_false/512 : SUCCESS in 1h 00m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":52},{"id":"d68d9d3e92a398e93122890f14ad4e27ddc97f64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-06-14 04:25:50.000000000","message":"Uploaded patch set 53.","accounts_in_message":[],"_revision_number":53},{"id":"71c182b8671fb618b90728dc6c95a344bf6ba948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 07:25:37.000000000","message":"Patch Set 53: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cdf98b3899e4941aaf3dc795a516729 : SUCCESS in 3m 11s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f8a8e1a866574de0acce1a2168f15520 : SUCCESS in 30m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5a66b0cf8f2f4800b2a27146c25f4b03 : SUCCESS in 33m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/209e70f8a40a4ecba8a82fb6fad45246 : SUCCESS in 59m 46s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/72f2723b0c87477487414e91df21e80a : FAILURE in 42m 19s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/7036c397580a433da913f7190d76bf94 : FAILURE in 32m 22s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/abe032f334834f1eb60bb096a987bbe9 : FAILURE in 43m 06s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d1741fc692304f7faac539698872ca4e : SUCCESS in 31m 40s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bf49efc1c13844fc9e9dde09db70e700 : FAILURE in 1h 10m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ecc60dd85a6d4cf9bbe32de9023227c2 : SUCCESS in 1h 38m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a913b5a69f7d4e4d9d768d7b081ecee2 : FAILURE in 1h 30m 14s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/08e37181d35143bcbd4dd878023c3e8f : FAILURE in 52m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ad58f8ec29854fffa161ca35b2a2da7f : SUCCESS in 1h 02m 52s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/70b0d22f03a5477dac34b209a2a5549a : FAILURE in 2h 37m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ac8fd172ef5e4cbcbfee72bf73103216 : SUCCESS in 29m 35s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"f8f284407ff72f31aceac35aa606f0dd5897fc7e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-15 00:33:50.000000000","message":"Patch Set 53:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/53/check/DellEMC_Unity_Manila_dhss_true/b701c73/DellEMC_Unity_Manila_dhss_true/550 : FAILURE in 6h 06m 31s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/53/check/DellEMC_Unity_Manila_dhss_false/22a00af/DellEMC_Unity_Manila_dhss_false/573 : SUCCESS in 3h 29m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":53},{"id":"c75e9643b54146526e7ffdb5f172d2cfe9e1ecf1","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-06-17 19:30:42.000000000","message":"Patch Set 53: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":53},{"id":"f5acaaec97aefce2cec06e5dcc686e08415c7387","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-06-19 05:23:34.000000000","message":"Uploaded patch set 54.","accounts_in_message":[],"_revision_number":54},{"id":"22b51f7d49fccf7f218194f8ec3a70a411b66402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-19 07:01:13.000000000","message":"Patch Set 54: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/106d1fd501eb42d79629fcf5dca981d3 : SUCCESS in 2m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c8921fb5dbe9421e954651d46988b707 : SUCCESS in 20m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2218b918b445459290d1e8d33f24d0bd : SUCCESS in 45m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bef3279c4460432e9b96638179c6790c : SUCCESS in 58m 50s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/4dfb768b59ed4df99bae3bc005556163 : FAILURE in 30m 49s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/71194e27f7764abbb94d39a46106d6d8 : FAILURE in 33m 40s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/5fa815698786461eb5dff03c52fe36f4 : FAILURE in 37m 59s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ff082ceff51944bda63dfdddf36196ca : SUCCESS in 13m 32s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/debf3dfc59504dfebb86479f7892a869 : FAILURE in 54m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bf92d05c6d06457eb990fc54ef8d1912 : FAILURE in 59m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/28e5b3ffcbaf4c869ed6d1900b3423cf : FAILURE in 1h 36m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/00eca9c13ab74bffac9ccea47aac5890 : SUCCESS in 36m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7840fbd8d72348c0a202a76ac54f03ac : SUCCESS in 1h 03m 59s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/975fe04778004e4fa3294d8bbebe2341 : FAILURE in 1h 34m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a692ed8d88e749b9b30afdcd3d020602 : SUCCESS in 18m 21s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"0bf23649bcb54aff9e5ef6346e951dddf2de0710","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-06-19 08:13:44.000000000","message":"Uploaded patch set 55.","accounts_in_message":[],"_revision_number":55},{"id":"2d9003cf725cda23c31c53234022091509298a1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-06-19 08:32:53.000000000","message":"Uploaded patch set 56.","accounts_in_message":[],"_revision_number":56},{"id":"c8317403f5132b3cc8cf78d3f7fcb3383f503fd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-06-19 09:27:01.000000000","message":"Uploaded patch set 57.","accounts_in_message":[],"_revision_number":57},{"id":"87cb11e5a2cfd01b8431efb97b501c853dc7a1c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-06-19 09:47:01.000000000","message":"Uploaded patch set 58.","accounts_in_message":[],"_revision_number":58},{"id":"72d2e3106d26673ee09380684f354e180f1c8ea3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-06-19 10:39:01.000000000","message":"Uploaded patch set 59.","accounts_in_message":[],"_revision_number":59},{"id":"fa7cf1dd328340efd58d87f2bd26ad17956da3a9","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-06-19 11:42:33.000000000","message":"Patch Set 59:\n\n(9 comments)","accounts_in_message":[],"_revision_number":59},{"id":"5f8933d81e1e78cd4361a73ae8320e9ae227f2e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-19 12:25:13.000000000","message":"Patch Set 59: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73bc2b881bb9416c995501a2bfa84616 : SUCCESS in 3m 01s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5704355c09f945dabc9042b5e901a19e : SUCCESS in 23m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8942bbb12c8a4244a9859aad30fa1bdc : SUCCESS in 27m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9b2f12793aee4c59a9efafba797e019e : SUCCESS in 55m 40s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/eca82f6e80a04f189cc35dc60be8aabc : FAILURE in 30m 41s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/0732a508c3ce4bd7aa26d577a82f4368 : FAILURE in 33m 57s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/2daa5c59c5d449b3bd36af0e8e91ac86 : FAILURE in 30m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/714e8e99045f4ee68d1d22823f36b324 : SUCCESS in 20m 09s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0268b8d451bb4205a7c6763f2e4b76dd : FAILURE in 55m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0a314ecb4b6b4b90a8e676eb0a1f92d9 : SUCCESS in 58m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f72a235f4d9e4d899b836f1ca5ec464e : FAILURE in 1h 15m 08s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9232ecda8d5c491699649058c9af9a3b : SUCCESS in 39m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/31831ea72ecc45848230d16e2eb593de : SUCCESS in 58m 32s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f9d741647b6c47edb5c68f65b27d3131 : FAILURE in 1h 44m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1af2a6c7ed094943aa8afad958beb2b3 : FAILURE in 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"24a678ffcf268a158dc387bc4552e9f4c33618c8","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-06-22 15:17:13.000000000","message":"Patch Set 59:\n\n(1 comment)","accounts_in_message":[],"_revision_number":59},{"id":"ba63f3a90197ee74e22948f7bb875abc61208799","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 17:26:10.000000000","message":"Patch Set 59:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c980f49b6b694bf8a159f6b267c444fb : SUCCESS in 3m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b0b5967e7b764ed6bc838c86674ff4cf : SUCCESS in 40m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8d810b41b82642eea5962f67cc22d7c8 : SUCCESS in 43m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8874f35e582d468886531eee2dc1be0f : SUCCESS in 1h 14m 05s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/d314785bff13421d8ad3b118e05dab73 : SUCCESS in 1h 20m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f06a73f0481b424d8a468d7230f62521 : FAILURE in 1h 03m 03s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/10b56b68d8164a3a836b03e8e9e0432d : SUCCESS in 1h 20m 07s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a0b4bb4e3b5641d489c889176c053414 : SUCCESS in 30m 17s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4d7d29e0f0054e1ca76eb30d9177ab51 : SUCCESS in 1h 18m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9a638a062c884d6d8fb481f9f5fc414e : SUCCESS in 1h 16m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1717cc049fc54e1b9c24906925005f40 : FAILURE in 1h 39m 50s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/19294fbdde7643989f9d41b9a0d817db : SUCCESS in 47m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f2f93cb4b78f4bf795b46152af6dfac9 : SUCCESS in 1h 24m 33s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/938d418a2505468d9f967824bc2632ed : SUCCESS in 2h 03m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/baf6c640523b4475b0a592f2caf995f4 : SUCCESS in 34m 11s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"4f7ab65dd2f2e73f94c2ad1d5efa85298625f903","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-06-23 04:16:29.000000000","message":"Patch Set 59:\n\n(1 comment)","accounts_in_message":[],"_revision_number":59},{"id":"12ab6b56caed117834563aad2d280407f926f610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 06:12:01.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c2317704db4433d9523ce6296d56be1 : SUCCESS in 4m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a3d9369a1fa84118a5f1e426f607450a : SUCCESS in 26m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5405c7b84b7340b1b58cd635da3cc3c3 : SUCCESS in 31m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8a00c7370dc34273908f1cef575ac5ea : SUCCESS in 1h 02m 58s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/1621ab8f509849b8a6d1a7f9f182a0ce : SUCCESS in 1h 08m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3fd0423413824e41a429329aa68dbc8e : SUCCESS in 55m 32s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/057ae5a0fd984d38a6fd81ebad94b963 : SUCCESS in 1h 05m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/38f94d15a77b4b57af8ba04e81106432 : SUCCESS in 24m 29s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0f981e9103304f5baefad6dfff6fb9a4 : FAILURE in 1h 10m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7e225c6c7b064ea5a38574e7d8fbf606 : SUCCESS in 1h 07m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8876db8c074949ab98b23c507480fc03 : FAILURE in 1h 31m 03s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4d0e9f470ced4f349d5186c5a908c328 : SUCCESS in 41m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1a76a120c9a247d4bddd1027d078e9d5 : SUCCESS in 58m 33s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/48b55ae1abc64b73aecfe8effaf74bdc : SUCCESS in 1h 45m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/df3bcdcc283d48c7b8e8b33d032ce806 : SUCCESS in 20m 15s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"afcccd29e8d4087277e58ec78397c39c61a094bd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-23 08:20:37.000000000","message":"Patch Set 59:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/59/upstream-check/manila-tempest-plugin-ontap-dhss/e7374f3/ : FAILURE in 2h 06m 47s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/59/upstream-check/manila-tempest-plugin-ontap-no-dhss/1b7edc2/ : RETRY_LIMIT in 8m 24s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":59},{"id":"bf5eaf1156886dbc2cf9928f946ace8f40b6b477","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-23 09:28:12.000000000","message":"Patch Set 59:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/59/check/DellEMC_Unity_Manila_dhss_true/c44edc2/DellEMC_Unity_Manila_dhss_true/568 : FAILURE in 3h 15m 56s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/59/check/DellEMC_Unity_Manila_dhss_false/f1287f0/DellEMC_Unity_Manila_dhss_false/595 : SUCCESS in 1h 26m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":59},{"id":"6c1b14c2cb441747395b42a158ca5cdb82a5213b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-27 15:17:42.000000000","message":"Topic s-rbac removed","accounts_in_message":[],"_revision_number":59},{"id":"4d65ee26dbbbe397920d740b7558f370de92e513","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-27 15:17:46.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":59},{"id":"f381ea4c0b556d04f40b6f43d5ca83c626717248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-07-03 07:45:14.000000000","message":"Uploaded patch set 60.","accounts_in_message":[],"_revision_number":60},{"id":"d69cf3661413795c419ef1e0acbc40c5b4ccd81f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 09:54:14.000000000","message":"Patch Set 60: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e875879c248e411b8df3c0b02ed2575f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47da3b6158974e33b6de2278fe958922 : SUCCESS in 2m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b30881f6faff45eeb5830afe764e529d : SUCCESS in 1h 09m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/37d9cb739c014784baa7c2d784f84b39 : SUCCESS in 1h 51m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fccbfc98a2aa495f993a4eb149f568d3 : SUCCESS in 1h 06m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/299a6040a608451b942474ca1c24f68c : SUCCESS in 1h 10m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c571b4fab6df49ec9f1b6eebb90b92bd : FAILURE in 2h 05m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4ee2b50e4d814a858856228080c77ba8 : SUCCESS in 1h 09m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/644a32cc985740c181faa075cc2511da : SUCCESS in 31m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9f9113dd031348238e92b3b6d2667c1b : SUCCESS in 47m 21s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e81517093d274d5a8f3d5ac57f35c3be : SUCCESS in 23m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/76f78b22316e4203aa810f71ae16cabe : SUCCESS in 47m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4af2c12bfcf04144bc1b9c2c900e519c : SUCCESS in 20m 57s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"4a1adb67733ff048e32ff6cac93cd42dadf912c7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-03 11:46:42.000000000","message":"Patch Set 60:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/60/check/DellEMC_Unity_Manila_dhss_true/38b703b/DellEMC_Unity_Manila_dhss_true/598 : FAILURE in 4h 01m 13s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/60/check/DellEMC_Unity_Manila_dhss_false/0af8a0a/DellEMC_Unity_Manila_dhss_false/626 : SUCCESS in 1h 13m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":60},{"id":"805ff1962667f4dcc1141d14191f14b1f8d1d99d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-03 11:48:18.000000000","message":"Patch Set 60:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/60/upstream-check/manila-tempest-plugin-ontap-dhss/7efbae1/ : FAILURE in 1h 46m 35s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/60/upstream-check/manila-tempest-plugin-ontap-no-dhss/8ce18de/ : SUCCESS in 1h 52m 55s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":60},{"id":"58fe5b1e1ef9ab5b1ddad57968646feee252a62b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-07-07 12:15:27.000000000","message":"Uploaded patch set 61.","accounts_in_message":[],"_revision_number":61},{"id":"113027adccf16fe55caf56eb8249cd97085748f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 14:12:37.000000000","message":"Patch Set 61: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a56ff458bcd4c1289f64d9e229d7cc6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/464a49dd122f453b87c730e0d97ac547 : SUCCESS in 2m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/54000167563c46cba663a85330efb402 : SUCCESS in 1h 02m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5c916a2ff78e418ebac0344009c1719e : SUCCESS in 1h 48m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/80d0f7fddd4547c6b4d0ebe2c8f2b822 : SUCCESS in 1h 04m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/fe434c2d45bc4e1a87d621e9664796b9 : SUCCESS in 1h 06m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/600318222b844287bfaf4a98ccbbc615 : FAILURE in 1h 53m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f21aeef1d7e04f169346dcdbe86011b1 : FAILURE in 1h 00m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/502418cd8450405e8f725792960cdfb8 : SUCCESS in 26m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0d315811e4de40b9b27bfaba6e7ec08e : SUCCESS in 29m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c836685c4c6b40429f11fbe27ad7c4c6 : SUCCESS in 23m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e575d0bc01ea44cf8c65395a163f8418 : SUCCESS in 56m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d8b12ad1ac0d46c881c32fd02d717ceb : SUCCESS in 20m 46s (non-voting)","accounts_in_message":[],"_revision_number":61},{"id":"ef3da34db65bd2a6552c285057ab6fe660ea2002","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-07-07 14:22:44.000000000","message":"Patch Set 61: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":61},{"id":"511b3ee809280fd92b202fa403c58b126a71dca7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-07-07 15:10:44.000000000","message":"Uploaded patch set 62.","accounts_in_message":[],"_revision_number":62},{"id":"c5630c46a5dbd3453cdde9a8a671a1b1d475190f","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-07-07 15:11:46.000000000","message":"Patch Set 61:\n\n(1 comment)","accounts_in_message":[],"_revision_number":61},{"id":"29990356f688bfc9ecda7245a53fc9ee925c3075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 17:51:51.000000000","message":"Patch Set 62: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69f869817ce146b5a2e4bd61aa5ae8a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d391df328ae84462a000025ed3bf8c83 : SUCCESS in 2m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/83aee32372214793b169885e48590c57 : SUCCESS in 1h 17m 22s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ab25ea7696aa4661af7bf50a2dc1f250 : SUCCESS in 2h 34m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/51e44860bf644788a2775de11d9171b7 : SUCCESS in 56m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5408cd0527be4866ae6df3825c85d99f : SUCCESS in 1h 23m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a0f62a613e0447b5b8c826ed903ed0ca : FAILURE in 1h 15m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f75347c779b2470698cd185c5e96ad8e : SUCCESS in 1h 24m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/39fc740b1b4a4c8ca681bf4f0dccbcb5 : SUCCESS in 34m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/91b3f69da2a54660b34ca547a072ea58 : SUCCESS in 45m 31s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8f0e213b38b947b187b9b8b6ab79edc6 : SUCCESS in 20m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/982d21a0c908480db59a2458c94d55df : SUCCESS in 53m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0307d7eba2a74c24841fc0191793655a : SUCCESS in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"d87fc20b783566fa65178adcb6e84ab82add43da","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-07 18:02:24.000000000","message":"Patch Set 62:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/62/upstream-check/manila-tempest-plugin-ontap-dhss/7b547eb/ : RETRY_LIMIT in 8m 39s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/62/upstream-check/manila-tempest-plugin-ontap-no-dhss/4d0101a/ : RETRY_LIMIT in 8m 29s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":62},{"id":"7e3e5d1710f31517f1b9effb7f3ca06dea9aaca4","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-09-13 12:59:38.000000000","message":"Patch Set 62: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":62},{"id":"14b3a133a89b54d6d9e2d5f1978f8534b5e330fb","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-09-13 13:10:59.000000000","message":"Patch Set 62: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":62},{"id":"409b6a61e8672fb4da63366afe31e369705fb104","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-09-13 14:03:56.000000000","message":"Uploaded patch set 63.","accounts_in_message":[],"_revision_number":63},{"id":"d7da247877fe72faa8ae918123fc48b03c3d96cf","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-09-13 14:05:35.000000000","message":"Patch Set 62:\n\n(2 comments)","accounts_in_message":[],"_revision_number":62},{"id":"e8110c66664833b59064b397e145d4bbe11fd4fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 16:01:37.000000000","message":"Patch Set 63: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/364c964da6904a64a3648ced81642c16\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5afe219ae0ad4a8db99c14c15a8536c6 : SUCCESS in 5m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b8493d453e7f43d2bde6e861a175b12c : FAILURE in 1h 15m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5b362ed031db47969ba297ec85188ab9 : SUCCESS in 1h 55m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4a5548bba2a7461191cfe6ccd9fe5ea8 : SUCCESS in 1h 19m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3883da09789a46bea1863ee5e785cc23 : SUCCESS in 1h 31m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/63935db5649d4f0d9645f1910edca0f9 : FAILURE in 1h 14m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8803db10b6d5479882cabd5d27259670 : SUCCESS in 1h 07m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/714304530ebd49e2a14bdd00e57a2e17 : FAILURE in 30m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4a243b904f694fd08a7a7a6b23fc3ab3 : SUCCESS in 38m 51s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2d2048f91d454f6ab9b46090b529cfc5 : SUCCESS in 1h 07m 17s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/18f28857e0f140ebbe6f7b2a5b559bfb : SUCCESS in 1h 01m 39s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/8d8fb8519d744037abf749cbb2b73cb0 : SUCCESS in 1h 03m 38s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/15774897ce16449287847b97d68ce37a : SUCCESS in 24m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2384af4b547f4bda8d3c057b0ef076d9 : SUCCESS in 44m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c79236a18ed643a6b8fa968a2d252594 : SUCCESS in 24m 22s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"aa718ad625d699e9211b0d45001e15a8fe9bf58e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-13 16:29:16.000000000","message":"Patch Set 63:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/63/check/DellEMC_Unity_Manila_dhss_true/c4dcb21/DellEMC_Unity_Manila_dhss_true/843 : FAILURE in 2h 25m 06s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/63/check/DellEMC_Unity_Manila_dhss_false/fcbf966/DellEMC_Unity_Manila_dhss_false/891 : FAILURE in 10m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":63},{"id":"98b89d38bbdb397cbb81ebd8e1853c7d05fa4bdc","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-09-14 04:45:41.000000000","message":"Patch Set 63:\n\n(1 comment)","accounts_in_message":[],"_revision_number":63},{"id":"ad6d45555cab52ff9fefb55b1e4529083537c7e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 06:29:44.000000000","message":"Patch Set 63:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aac6bb8ec07c418d826069fd28b93ece\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b25cf23feff94e808ee30ea96ab90bbd : SUCCESS in 3m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6947f04e73384473b5610c52fa431702 : SUCCESS in 57m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7b92b78f72f54c18a5e2d47e5bdbd108 : FAILURE in 1h 37m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/75a554ee78984c5a8bc495565e64e45d : SUCCESS in 1h 00m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2c41f0111755416388420039cb23ff2c : SUCCESS in 1h 04m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/859c5bd98d91417e8782f07b3e2311f3 : FAILURE in 1h 41m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3161545937364a95be35d9b8c264f52c : SUCCESS in 57m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e0d01b7c7d994f0b8a02ab598f5b5f92 : SUCCESS in 27m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/789bdaf35fdc4f429bd83bd3bffddab1 : SUCCESS in 31m 03s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/6a83998fab4c4f1cbcc0f60b62e65929 : FAILURE in 1h 03m 54s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/22a4a88b936d463c9f78c1e14a24ec07 : FAILURE in 1h 13m 56s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/181afdb279764a7195ee72e49973d4d8 : SUCCESS in 58m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d6c61b9d2f144ae99662389d0e087fa9 : SUCCESS in 36m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5772030c9c274ec78a5c35797301c1f7 : SUCCESS in 43m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/41f2e24c85404ec89cf04916835797be : SUCCESS in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"ecc29ccb73687da3736d54b8539e331cf16abad1","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-09-14 06:40:32.000000000","message":"Patch Set 63:\n\n(1 comment)","accounts_in_message":[],"_revision_number":63},{"id":"e262c8c26d0f5a2b9d7881c69d94025ac315633c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 08:39:31.000000000","message":"Patch Set 63: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9df1db11126140939183c863474fb11f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/887fbe99d8584825b1e0b945f0e07927 : SUCCESS in 3m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ed121ad8141d46b2963486802f3ad062 : SUCCESS in 1h 31m 34s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a5d6ef3613454fea8ef49635ff164904 : FAILURE in 1h 54m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/96b6da87071d414aaaf161cccc6cf82b : FAILURE in 59m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/db56e0c9972a4885b4a53472c0d5ca1d : SUCCESS in 1h 23m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e2a772f34df4490e9693c8a80061fccc : FAILURE in 1h 52m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/28e43e8fd2dd425892b6ac1cfa994f4c : SUCCESS in 1h 24m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/43632f00acd743b1aca87dfebcbb9170 : SUCCESS in 36m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f0eb9069f62c4ccfa71064e53a352f28 : SUCCESS in 43m 39s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/b4d4b9d7b43b4ad59d302962c3f84576 : SUCCESS in 1h 29m 42s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c852e2bc34f54c31bbbf9f95f73048bd : SUCCESS in 1h 21m 10s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/65395a4f36214b3ca5002d9f5a63ce1f : SUCCESS in 1h 12m 59s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/15a89f6df42440558f7ba30e4ca014c6 : SUCCESS in 33m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cdfdbc501078492cbebc12f55e29f88f : SUCCESS in 57m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/43e76b57a4f347418d82a9651ffa7d1a : SUCCESS in 29m 46s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"f7d4598579bd2282f6ce8db8c6a536c594a717d3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-14 09:47:55.000000000","message":"Patch Set 63:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/63/check/DellEMC_Unity_Manila_dhss_true/e9e687b/DellEMC_Unity_Manila_dhss_true/849 : FAILURE in 1h 08m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/63/check/DellEMC_Unity_Manila_dhss_false/588a523/DellEMC_Unity_Manila_dhss_false/898 : FAILURE in 3m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":63},{"id":"43057baa2a1cbbeed3eed5125baa0816fb261f95","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-14 09:56:27.000000000","message":"Patch Set 63:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/63/upstream-check/manila-tempest-plugin-ontap-dhss/a63192c/ : FAILURE in 32m 53s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/63/upstream-check/manila-tempest-plugin-ontap-no-dhss/e674de5/ : FAILURE in 1h 15m 58s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":63},{"id":"8b3ae94f4eeaad1225fb7e237f43f46673adc430","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-09-14 11:50:37.000000000","message":"Patch Set 63: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":63},{"id":"eda75dee1c317e8f12fba324740192531f7b6d59","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-09-14 15:42:50.000000000","message":"Patch Set 63: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":63},{"id":"3bf72609e02169f1bde35a92a0c8ff99bd668e7f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-14 22:53:04.000000000","message":"Patch Set 63: Code-Review-1\n\n(27 comments)","accounts_in_message":[],"_revision_number":63},{"id":"d52bf789ce8b7529a821ebf9ffafa711daa5b2cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-09-15 09:55:52.000000000","message":"Uploaded patch set 64.","accounts_in_message":[],"_revision_number":64},{"id":"87026c35ab90d65d1b2df867989ea8335f0eec5a","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-09-15 09:56:13.000000000","message":"Patch Set 63:\n\n(11 comments)","accounts_in_message":[],"_revision_number":63},{"id":"98c4c6f2b12e47cdf4b219fd2648bb7c3f74bb3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 12:21:39.000000000","message":"Patch Set 64: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d903a92e1ae46c0ae167e6d2910176a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7158f51f9804f4aae2e9101d79af1ea : SUCCESS in 3m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bd7cbda5cf02414abd11b8aa0d7c2349 : SUCCESS in 1h 16m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/866d5134dcc94716b7b8a60436f9bf04 : SUCCESS in 2h 18m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b5b391e0ba724f8a8d11ed80a47d70ca : SUCCESS in 1h 31m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/de3a6b2e415649dfa966a6f87d4553df : SUCCESS in 1h 26m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/66aba47062bb4d6ca159836e5945eba9 : FAILURE in 1h 59m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/59142addc4a84253ad01bad7dfc0516d : SUCCESS in 1h 12m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8caa4a2cea6a43db825461f412c44198 : SUCCESS in 37m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/461ac5f97f5f4d83bfd0bf895104d36d : SUCCESS in 39m 11s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/24d02a738ec3420d8987d249fa09287e : SUCCESS in 1h 23m 16s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3febd5121dd64060804dbc49d8e05344 : SUCCESS in 1h 13m 59s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a98a8abd18b0440e8c3f6181f73000d9 : SUCCESS in 1h 18m 12s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/12028489e2534807a6ea203f934e574a : SUCCESS in 22m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/569fd27ab23743b5a68da8acedae8363 : SUCCESS in 43m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/50d845644876436a9fd531a6ec9aefe1 : SUCCESS in 21m 37s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"b1be772a3c6f944d7b36b0b9a52e9d971e88d1f6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-15 14:12:46.000000000","message":"Patch Set 64:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/64/check/DellEMC_Unity_Manila_dhss_true/e6dafdd/DellEMC_Unity_Manila_dhss_true/865 : FAILURE in 1h 33m 27s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/64/check/DellEMC_Unity_Manila_dhss_false/439625b/DellEMC_Unity_Manila_dhss_false/916 : FAILURE in 41m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":64},{"id":"df37b50ab528b165377126260e9f8d522088baea","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-15 15:04:47.000000000","message":"Patch Set 64:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/64/upstream-check/manila-tempest-plugin-ontap-dhss/fbd2006/ : FAILURE in 2h 41m 43s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/64/upstream-check/manila-tempest-plugin-ontap-no-dhss/6afed54/ : FAILURE in 2h 15m 06s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":64},{"id":"f2a9cd89d0e4f0a4bab6638c9a4e208ec8945f55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-09-21 09:56:00.000000000","message":"Uploaded patch set 65.","accounts_in_message":[],"_revision_number":65},{"id":"58b40bc9dea95f2342dbcc7cd6f6a8e259b64584","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-09-21 09:57:14.000000000","message":"Patch Set 65:\n\n(18 comments)","accounts_in_message":[],"_revision_number":65},{"id":"3f90a87a7b3fdf8661fdf1a6d676b9bd54f9cdd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 12:20:54.000000000","message":"Patch Set 65: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24eb7889377742f4b9bbc304ec58650d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21406c893a96478d9457a6f9b8346bb3 : SUCCESS in 3m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3c6c022a7fe14c20928ecb50c46809a1 : SUCCESS in 1h 32m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d11393ff2e9446b898efae82a41800c4 : SUCCESS in 2h 09m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/042f86864fee4c3aa635aa1840ee8363 : FAILURE in 2h 01m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f43057b4a88f4eb892202f035e4bea7c : SUCCESS in 1h 26m 00s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a33090e14bd449f8a65c4f1c2e6784dd : FAILURE in 1h 29m 58s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4f1d70f9d7a342aa8891d5a495f45d73 : FAILURE in 1h 10m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6393373b578f4accbdcae2a3bdcef068 : SUCCESS in 40m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/af0f3d80c864426eb1f6d5d687ced1b8 : SUCCESS in 40m 27s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/5a350d7ad74845e18b1aa55aa677dcc6 : FAILURE in 1h 54m 53s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/cc43815949f2496eb8f8ee559029a038 : FAILURE in 1h 30m 10s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b9a5c13c01f743e7bc0b5117b40d2866 : SUCCESS in 1h 16m 31s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f96c4a8b787544bbab7482a4a00c6cdd : FAILURE in 22m 51s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2030234305144677bbfbe24f54d31874 : SUCCESS in 52m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/600c4bad79a244b4a7186f678222d434 : SUCCESS in 39m 43s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"870247ca641fcece2e255070af3efa971e3aedc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-21 15:08:18.000000000","message":"Patch Set 66: Commit message was updated.","accounts_in_message":[],"_revision_number":66},{"id":"6738ce36561aea39b1617e7fa3cba5ebe17a5028","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-21 15:08:42.000000000","message":"Patch Set 66:\n\n(1 comment)","accounts_in_message":[],"_revision_number":66},{"id":"ce650235cc664013b9c20073fd2bae4b031addcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 18:19:49.000000000","message":"Patch Set 66: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3d244650eb4445f8d09482a39d3b3cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f8c6274c34d41a4a7fcb93814141166 : SUCCESS in 6m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c937511fafe54adb95229455fec9ae81 : SUCCESS in 1h 21m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/aa85c1c4d34446ea93bb17e962576161 : FAILURE in 3h 01m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7bd9551713aa43de8a9966219d06279b : SUCCESS in 1h 24m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/900b12a5f09a465a84a2c728bac63c36 : SUCCESS in 1h 46m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/788638f6508a4b6eb27e1ef313f27f6c : FAILURE in 2h 06m 58s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b9d1529104964037a5da38b942234acd : SUCCESS in 1h 12m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f26d396ffcdf41059197bacbe593fedb : SUCCESS in 37m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/78d6ba43dcd848cb8d781b3469c1cfae : SUCCESS in 41m 56s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/6c510c22f93d4c68acc62f9bc6749b75 : SUCCESS in 1h 12m 41s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/732915b1b30a48ffa7335e83d0d5b33c : SUCCESS in 1h 06m 43s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/5f19345ef14f44f2bffd5c1372182e41 : SUCCESS in 1h 03m 57s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/cc18f7eca7d04f3c8193e9a60d9f5803 : FAILURE in 34m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b23779ea9bf34756b82393c4bd6677a8 : SUCCESS in 45m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ed41f19e22b64e6dbedc1eddb9480917 : SUCCESS in 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"32b9632dca3ca7181699f8322a603b15f41d2d4d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-21 19:32:24.000000000","message":"Patch Set 66:\n\n(2 comments)","accounts_in_message":[],"_revision_number":66},{"id":"9c2ebd701f7883090156b6e836e1fc3035be30a6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-21 19:48:54.000000000","message":"Patch Set 66:\n\n(1 comment)","accounts_in_message":[],"_revision_number":66},{"id":"3002c6f73757c3daf4c71972a883e1100c2f6b15","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-22 04:08:05.000000000","message":"Patch Set 66:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/66/check/DellEMC_Unity_Manila_dhss_true/c9d0cd2/DellEMC_Unity_Manila_dhss_true/912 : FAILURE in 1h 43m 28s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/66/check/DellEMC_Unity_Manila_dhss_false/76693b4/DellEMC_Unity_Manila_dhss_false/971 : SUCCESS in 1h 12m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":66},{"id":"661cd0c4c711e4296602884d40e0ba62512d32b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-09-22 07:25:50.000000000","message":"Uploaded patch set 67.","accounts_in_message":[],"_revision_number":67},{"id":"fb914df83861025cc21d7f4c9e0e44a748fcfeda","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-22 08:46:52.000000000","message":"Patch Set 67:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/67/check/DellEMC_Unity_Manila_dhss_true/f465358/DellEMC_Unity_Manila_dhss_true/914 : FAILURE in 1h 20m 46s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/67/check/DellEMC_Unity_Manila_dhss_false/34295cf/DellEMC_Unity_Manila_dhss_false/973 : SUCCESS in 1h 18m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":67},{"id":"84121be1ec53cdf7cff80c93e9bef2cd37092964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 09:14:03.000000000","message":"Patch Set 67: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b36ab215ae584df8a4c687e198e102d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e334cbc3bdc4aa193380d702fa004cb : SUCCESS in 3m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9e7e39cab0684832a399030ba58a3c37 : SUCCESS in 1h 06m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f526fd0bd5434a6a8c05d0f49a9f87dc : FAILURE in 1h 47m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/40d1ad1ff0c743289292e10c5e25b040 : SUCCESS in 1h 12m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/37ae8ea4c9884e79b6d6ac411437306e : SUCCESS in 1h 07m 36s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5fde7f4c3d7343338793fa7ea1e00b9b : FAILURE in 1h 34m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c9cfd0ee2b3b4418ac667faad75be150 : FAILURE in 1h 03m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3e2b36e05df4465395248c445fe6c617 : SUCCESS in 28m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/71e9b1c726bb4d618f01d1bc29872542 : SUCCESS in 35m 38s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/47a5fa8b04e94f80aa9d76fc77b6624e : SUCCESS in 1h 10m 22s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/8091d97b75b74241aad6bb7c9fe1d99d : SUCCESS in 1h 00m 43s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d366f3bf5a6648358971edf7f1c320f5 : SUCCESS in 1h 00m 28s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2e764c381dbf4b68aeb5f18530b93d09 : FAILURE in 13m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/07a95ce2d62b4ebdb187930e66810692 : SUCCESS in 42m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/978b30560c844921a350ad3623b62dda : FAILURE in 24m 27s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"be00baf7a39745e31851b40da357e9cce49fc849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-09-22 09:43:08.000000000","message":"Uploaded patch set 68: Commit message was updated.","accounts_in_message":[],"_revision_number":68},{"id":"b4130529e494d65fcb52f33c74745f663744118b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-22 11:56:20.000000000","message":"Patch Set 68:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/68/check/DellEMC_Unity_Manila_dhss_true/72e5280/DellEMC_Unity_Manila_dhss_true/916 : FAILURE in 1h 16m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/68/check/DellEMC_Unity_Manila_dhss_false/56ff38b/DellEMC_Unity_Manila_dhss_false/975 : SUCCESS in 1h 11m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":68},{"id":"8dfc793250deade2cee722c6118c1524c10391d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 11:58:05.000000000","message":"Patch Set 68: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7a7acea11a14c11a91bcf08e83d741d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aacdfdc68ed4155bc60e9c1122766a2 : SUCCESS in 3m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/18725f391ee6472f8dca069a93722639 : SUCCESS in 2h 01m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/06a5921485824e549acf8d0849d3cfb6 : SUCCESS in 2h 10m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/52bd93fa5bb14b1cac2836ae88dce9e4 : SUCCESS in 1h 21m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9063192d2454467a9e81c4e7421e1c52 : SUCCESS in 1h 25m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/892976129be443359f2ec8082184d3e8 : FAILURE in 1h 28m 00s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d2c6319542b04d6ba057892a568c325d : FAILURE in 1h 17m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/587d48ccc7d84e29880d85a385d9b310 : SUCCESS in 42m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/add92afee8684adb85fef0d164ad9916 : SUCCESS in 49m 21s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/035ba11bcec54a7dbd3c73ea17ab4d89 : SUCCESS in 1h 04m 25s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/026dd6a23ede46fba599d331bb57297b : SUCCESS in 1h 14m 46s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/81ea44906ef34ceca13e10a9480c4b0c : SUCCESS in 1h 25m 33s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d63bc2dd4e3f487f983c870233fb9b3c : SUCCESS in 22m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c4689873b89246f982667722461afe65 : SUCCESS in 44m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/15aaf1b7bab04859b584310ce8c278ae : SUCCESS in 31m 14s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"27dd400c825526186288ca0a56a1082a0722f773","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-22 13:14:33.000000000","message":"Patch Set 68:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/68/check/DellEMC_Unity_Manila_dhss_true/e19b2cd/DellEMC_Unity_Manila_dhss_true/917 : FAILURE in 1h 16m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/68/check/DellEMC_Unity_Manila_dhss_false/0285fb8/DellEMC_Unity_Manila_dhss_false/976 : SUCCESS in 1h 14m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":68},{"id":"1672d7336dae12d5cc9b0d28cc60375af0cea46c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-09-22 13:59:44.000000000","message":"Patch Set 68: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":68},{"id":"77eb3cacf77f8a4a08d5bbccdfb3d52d8f84efac","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-22 15:21:38.000000000","message":"Patch Set 68: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":68},{"id":"b08984cc0545a4342565e7c910146bbce6aa2594","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-09-23 12:45:19.000000000","message":"Patch Set 68:\n\n(1 comment)","accounts_in_message":[],"_revision_number":68},{"id":"d12638135146e41a26fc332a42fb456280f132aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 14:56:21.000000000","message":"Patch Set 68: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/46b6bdc35ba140e096dc9ce943fcf0af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bba2a08f4d074c3791a5b9674cd3be84 : SUCCESS in 7m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/13ede5e135784e2bbc2b05ddc7f1eefc : SUCCESS in 1h 15m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bb7bfca426544aa1b660959f4fc0ce8f : SUCCESS in 2h 10m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e4b048c759e345d39889ac6ecd3d3bf7 : SUCCESS in 1h 33m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3c99d954095b4a86aabfa72e0f51f56e : SUCCESS in 1h 29m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d2f2b17f2bfa49a5af7235382e660a58 : FAILURE in 1h 56m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e6839b9d771c4bb0b606d182bc21eca5 : SUCCESS in 1h 20m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7aa45f21338540f9a154e9be2c783cfe : SUCCESS in 37m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8438d81b28274633b2780e8c231cd412 : SUCCESS in 41m 10s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/d47e08ac2ed04de4b20e5f5bb54aa152 : FAILURE in 1h 33m 08s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/6283ad3a3f8b4680a50ad726330e2de5 : SUCCESS in 1h 33m 06s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/00106a5e20814f81a10eac2692835c5f : SUCCESS in 1h 15m 41s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/44eb00e8b728464fb719cf70e9bff1e2 : SUCCESS in 34m 16s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/df381932b4f344f0999baec553ed8d40 : SUCCESS in 57m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/73c5197e7bf74a2eaed0e205a5c356c2 : SUCCESS in 30m 28s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"6eee22673c94e70dc533efdff6d153ba52d952d0","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-09-23 14:59:29.000000000","message":"Patch Set 68:\n\n(1 comment)","accounts_in_message":[],"_revision_number":68},{"id":"b60d49ecd0eaf8a0e8b1a6cb7a306a27664b4991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 17:30:55.000000000","message":"Patch Set 68: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00abbab94a514686bf0ffc3f2ba4e092\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a931304ef7a648f782ff7c08da07745e : SUCCESS in 2m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c0638a5303ed41ff8452937be8f65fbf : SUCCESS in 1h 18m 00s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ef3476b83ec24612a839e1cd1b177957 : SUCCESS in 2h 07m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a2b50f4c21ed424eb64c61a84966910b : SUCCESS in 1h 17m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/25440c263325475faaf086bef8c63ed4 : SUCCESS in 1h 34m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/31ef88b5f58343bdb89ad3253a24609b : FAILURE in 2h 25m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0131b81ff29f450ab320b7dd858ce080 : FAILURE in 1h 14m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ed7e5b9c1f0c492ea913960ca82bbc30 : SUCCESS in 30m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2badaecfee844863bb45a9c0b704d260 : SUCCESS in 42m 42s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/a0c3a7a56254491984dc493bb5bcacfb : SUCCESS in 1h 13m 33s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3e4aeef4c3ef48e597dccd8b8070b08d : SUCCESS in 1h 05m 52s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/f1b9d61aee1b4a6f8f1338d12ed9e61b : SUCCESS in 1h 08m 56s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/46cd5a163a0b4da6bda2693c6a27f728 : SUCCESS in 31m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1f2f77c6bdb343c0a869a452adbb0431 : SUCCESS in 51m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e2bf76978ec74c97b5e652bb0a3e9168 : SUCCESS in 25m 45s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"188dba84f16c5dc4297d0201c5090b31455b4647","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-23 20:31:27.000000000","message":"Patch Set 68:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/68/upstream-check/manila-tempest-plugin-ontap-dhss/20504db/ : FAILURE in 2h 09m 56s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/68/upstream-check/manila-tempest-plugin-ontap-no-dhss/1071bdb/ : SUCCESS in 1h 16m 20s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":68},{"id":"933eb370b666e7206c3705dfb28795aefe7cf749","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-09-27 16:17:28.000000000","message":"Patch Set 68: Code-Review+2","accounts_in_message":[],"_revision_number":68},{"id":"8c8da96ae006b625ab8e56ea36b441ce3764b8b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 15:27:59.000000000","message":"Patch Set 68: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":68},{"id":"187b5906001dc1e9a0a1546a5fe0013147df10bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 16:34:59.000000000","message":"Patch Set 68: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd72a52750864b8ba8538412069a7cf0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fec9ea8691546c5ab12f1821ecd473f : SUCCESS in 4m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/01c5693110694efda7aa0de8158c6644 : FAILURE in 1h 06m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/44353fdef29d4b49a65b36a2ac85b6be : SUCCESS in 30m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7a0927cd646e4cb4b68d92b55e053f8f : SUCCESS in 37m 07s","accounts_in_message":[],"_revision_number":68},{"id":"259f65d785c5c9416ef0d37b159466c767a33aa0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-29 16:59:02.000000000","message":"Patch Set 68:\n\n(1 comment)","accounts_in_message":[],"_revision_number":68},{"id":"4b374cfa5622c23e2f387d9941deb0b5b8f8a710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 20:09:05.000000000","message":"Patch Set 68: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ef1ad2d217c43d89bcfdf7c167d41fa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56c7ec7a74df4c519e4fecbf00ef4e41 : SUCCESS in 3m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4cb33e90903842d39598f620ef450e0c : SUCCESS in 1h 25m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/35083c44e2fe447ba3a51b422cd70c8f : SUCCESS in 2h 02m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/201d8163f13b4800843277aa525bbd3a : SUCCESS in 1h 13m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/54b56f8aaecf459896a0a3e965eebbd6 : SUCCESS in 1h 11m 40s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/715eda2f51a34907bc9de1cb9baad360 : TIMED_OUT in 3h 04m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4137f2bd1a4b4d3280175c927ed5db26 : FAILURE in 1h 45m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3cb49e58a89d4532b2260f9eb147d567 : SUCCESS in 29m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fbd3303af1a84af0bb7b6aa2c4912f11 : SUCCESS in 33m 48s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/9b96ef9f07ca455794622446444a2388 : SUCCESS in 1h 27m 56s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/18e25ebce2a64d83afcc61dda812160a : SUCCESS in 1h 10m 36s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/18d70158fc5e4b1cb7628a22d22061b5 : SUCCESS in 1h 41m 21s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/914d3dd8fb3b4d239033a288e56ff9c5 : FAILURE in 28m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0416156112024014a4bf7ee78522614b : SUCCESS in 53m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d9ba00e98a164631af9c32f5e1665455 : SUCCESS in 20m 34s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"7ca38fe93c2563495d14c1549abc3ef3b71af33a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-29 22:02:24.000000000","message":"Patch Set 68:\n\n(1 comment)","accounts_in_message":[],"_revision_number":68},{"id":"f5d209245cc44570c5ea225f38f09bb377ff2834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 00:24:06.000000000","message":"Patch Set 68:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2b5ff974ce84ac3ac7ede78937c5ea1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23813e8146664ad580b42b46e307d1bf : SUCCESS in 2m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/76d4a5ae686f4880a9f3cc085246949a : SUCCESS in 1h 22m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fbeb0a1de9d546c185cc2bab9c0a4dd9 : FAILURE in 2h 16m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/124ba934b7174dfba784c2d6eb5f02be : SUCCESS in 1h 12m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/df2f970567654c569a714c55ccf2bbfa : SUCCESS in 1h 35m 37s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/782d3185278748e4ba25d57b206c5b3d : FAILURE in 2h 06m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a0f6127a992b40d79db1c98bca89cbd6 : SUCCESS in 1h 20m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c7abb4dad1784ecc8938436b87ca14f0 : SUCCESS in 37m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9d91569fb95d43788fed0561d70c5a3c : SUCCESS in 39m 21s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/1ba800d6154a46008f14e149b796e83b : SUCCESS in 1h 16m 09s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/729044e98a3f41fc8e1c0803ab504d6d : SUCCESS in 1h 10m 06s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a7e02473ed5c452fa90064ff3ca11b67 : SUCCESS in 1h 14m 29s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/79e03b4bb607420ca969bcfb376ea262 : FAILURE in 33m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/56f0d79fb5d548e99b8fcc267af24012 : SUCCESS in 56m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1b3327f18125449888884243f212773c : SUCCESS in 29m 51s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"3a8b53c5185fea891e75925f000e83ecc4b8da07","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-30 00:51:37.000000000","message":"Patch Set 68: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":68},{"id":"af3e6112965e92149fd2353c8db13be0d3a1930f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-09-30 12:29:53.000000000","message":"Patch Set 68: -Code-Review","accounts_in_message":[],"_revision_number":68},{"id":"21e334f597f713790d9130eb8c917038cfb70b90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-02 07:13:57.000000000","message":"Uploaded patch set 69.","accounts_in_message":[],"_revision_number":69},{"id":"8b067acbbe0c01c61d678cdab4051f8e5379caaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-02 09:15:23.000000000","message":"Patch Set 69: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/acd92cc3371f46d1a69ec443120c821e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24108364aa1e415487068f98698dc001 : SUCCESS in 3m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/886f720fca6a47deb29d0aa1fed5baf6 : SUCCESS in 1h 06m 28s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fb6af654b4a04269be43f131dec2c47d : SUCCESS in 1h 59m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cb24c0bfccdc4e04bcf86a5940af7fe9 : SUCCESS in 1h 13m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/54be493a5e3f40bcab537cf97c8e0453 : FAILURE in 1h 33m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bb75a59d59b04cd2a1a5fac8b01c1f7b : FAILURE in 1h 35m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f631babf05b44df68758bd731bbd8e24 : SUCCESS in 1h 08m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/52de1b4655cc4ff6af308f2d236220b6 : SUCCESS in 30m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a1b7697b999f41ce822fd253f819ad04 : SUCCESS in 36m 22s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/afc3b44ffd1c42738d460644fdbf7453 : SUCCESS in 1h 02m 00s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/54afd22309a94099ada81f02af437079 : SUCCESS in 1h 01m 11s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e450523a01714f8da9c0d68702a6b07d : SUCCESS in 1h 08m 32s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ae6301e235ed419494aab839f79fb0a6 : FAILURE in 21m 09s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/23cf826e94064f12b3eb2779183140fc : SUCCESS in 55m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/10c65172c062486d93a665b11e9789fe : SUCCESS in 18m 35s (non-voting)","accounts_in_message":[],"_revision_number":69},{"id":"760dbcb38d5158c18d980ac17ba2c2b3efcf6a71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-02 09:30:54.000000000","message":"Uploaded patch set 70.","accounts_in_message":[],"_revision_number":70},{"id":"b0042ed552be90c833910516e7b19fca8c474224","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-02 11:05:10.000000000","message":"Patch Set 70: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02753cb9bfd745a494f23ac61413cbcc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b9f19f5d174b239b1f02136b3385e9 : SUCCESS in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c71a2a975beb4fd2bea871e2e0b993f6 : SUCCESS in 58m 22s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/664daeee8a944ea6b2746955009f1e95 : SUCCESS in 1h 30m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1fed253d0e29498c8df1a824d25edf32 : SUCCESS in 1h 05m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9eac7d73b98a41d5a39b3982499beecc : SUCCESS in 1h 33m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4cabbc131d314e238c57efe7224ae4f9 : FAILURE in 1h 32m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ecb85626db36407d87a4d24277c1fd32 : FAILURE in 1h 04m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a1fa4993a6d547cfb8bb6600158f57f0 : SUCCESS in 30m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/34acad328eac4faab4cd435ba1467fc5 : SUCCESS in 33m 39s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/ef1bd4183ea64df9b02b648fbff8cbf3 : SUCCESS in 1h 19m 31s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/28ba4d99c48344388f881d63b4bcd307 : FAILURE in 1h 03m 45s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/35748f6a3b934c2998138edfdf125775 : SUCCESS in 56m 12s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0c786de40da047b3b55e10b510f902d0 : SUCCESS in 26m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5528abc2b19049709f30e73f37b104ff : SUCCESS in 50m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8263cc54756f447fa14bbf8c88c4a222 : SUCCESS in 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"f01b9014650f634aaed28e503aee9b51b4708dba","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-02 11:13:56.000000000","message":"Patch Set 70:\n\n(1 comment)","accounts_in_message":[],"_revision_number":70},{"id":"86441e9d7fa4414d9bf9edbd6bab94ada4e253f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-02 13:06:01.000000000","message":"Patch Set 70:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/977c203713b5408fb2438e6906348474\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17b8cb271fc34a71b2349677c67aad3f : SUCCESS in 5m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d82e18a206414748b9e96e59f7750069 : FAILURE in 1h 15m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/23d22c21d5114b6bbc67c9ff04213653 : SUCCESS in 1h 49m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1e7758a9f3334b43880ba18d1d41ee99 : SUCCESS in 1h 16m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/aae2e160eb104559844082c505043850 : SUCCESS in 1h 19m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/408f2a1623084144b9a8c4e98752c28c : FAILURE in 1h 39m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/30e83bec538f4ad19d0911af7757bd6d : FAILURE in 1h 09m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c7f22a37fd5742f4a2aa523b326e88ba : SUCCESS in 25m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b635e8ce32be4e22bd22f87a45cf7ad3 : SUCCESS in 43m 26s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/7b693b989c8245d6bff875b8e6b38602 : SUCCESS in 1h 17m 29s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e4c3dacfe508483aa593ddd38d2d65c5 : SUCCESS in 1h 10m 47s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/6348da8cfbc44a32ad54766ee79c6dc6 : SUCCESS in 1h 11m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5692dcb4d7a3425fb7f9632301de390a : SUCCESS in 28m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9db2e22ba32648e78e2eb703597dc322 : SUCCESS in 55m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e8c1b2811c4c435fbf8b66c9d563e8fb : FAILURE in 31m 43s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"eb4c11a9533ecb833ebeb0535ce1e6991b0e4f01","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-02 13:06:07.000000000","message":"Patch Set 70:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/70/check/DellEMC_Unity_Manila_dhss_true/3a05151/DellEMC_Unity_Manila_dhss_true/954 : FAILURE in 3h 34m 52s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/70/check/DellEMC_Unity_Manila_dhss_false/07f14ac/DellEMC_Unity_Manila_dhss_false/1015 : FAILURE in 1h 22m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":70},{"id":"73bc1efe6dd759a3e948a889420283660ea7fbd4","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-03 05:53:42.000000000","message":"Patch Set 70:\n\n(1 comment)","accounts_in_message":[],"_revision_number":70},{"id":"17dc0fb2ce73a78328e4eff0872b63461d132dd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 08:16:07.000000000","message":"Patch Set 70: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3c424e903ac471aaf58ed0b89da62c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2e9e9c384914e64a31527072fe36523 : SUCCESS in 3m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d46b3168b36b4dcaa474d0cfc7da0383 : SUCCESS in 1h 32m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b5fe73389ce545a1a6c8b7d849b23513 : FAILURE in 2h 17m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/46be62810b894665a068f7d151e6428b : SUCCESS in 1h 33m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5843360c953847e7aabf68c9ffbf5e84 : FAILURE in 1h 38m 00s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c784dfa3b5b5476fb4ca51bde4708fd5 : FAILURE in 1h 16m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d8d037c8f02643059c79aea63fdd7ba1 : SUCCESS in 1h 39m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1fac812ca97a4426b3198a5172a4a826 : SUCCESS in 36m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fdd70bbbf3ab4f5392fd6d2f5a5fbc52 : SUCCESS in 40m 18s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2c00ff13b6d24f1b9f6afd6559407002 : SUCCESS in 1h 38m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/1fb7f02abc1a4cd996549ec9499323fc : SUCCESS in 1h 13m 54s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e15bb9c7648848d19fe958273740c95a : SUCCESS in 1h 14m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a50e8120c05044dfa967ef3ba78ecab3 : SUCCESS in 31m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/384450f68db845929beb5c3ac71727cc : SUCCESS in 53m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/594c7f9f7a784aeeac13c75ee7b3caef : SUCCESS in 28m 56s (non-voting)","accounts_in_message":[],"_revision_number":70},{"id":"f12b46ad7c0559466e18df6a9495fb5859c99abb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-03 10:02:13.000000000","message":"Patch Set 70:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/70/upstream-check/manila-tempest-plugin-ontap-dhss/d9f565f/ : SUCCESS in 1h 45m 10s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/805938/70/upstream-check/manila-tempest-plugin-ontap-no-dhss/d96daa8/ : SUCCESS in 1h 18m 40s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":70},{"id":"678227ccee1e089bf875b39e4b32f9ad2a17236e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-03 14:01:56.000000000","message":"Patch Set 70:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/805938/70/check/DellEMC_Unity_Manila_dhss_true/ca6c10e/DellEMC_Unity_Manila_dhss_true/956 : FAILURE in 2h 28m 07s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/805938/70/check/DellEMC_Unity_Manila_dhss_false/4391284/DellEMC_Unity_Manila_dhss_false/1017 : SUCCESS in 1h 16m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":70},{"id":"9d3b58afed8a8c79127f9d44cd9779e8209a6af7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-10-03 20:00:46.000000000","message":"Patch Set 70: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":70},{"id":"b946df9db7e5e3a385883bfe990355a3a7f59ade","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-10-04 12:07:58.000000000","message":"Patch Set 70: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":70},{"id":"f7dd2c4a31f874a693f5cf6b0a81d2af0b71c45f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 12:08:55.000000000","message":"Patch Set 70: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":70},{"id":"4b02e219a75f53b21f3bcc08e5f8c30f3cecfe9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 13:32:48.000000000","message":"Patch Set 70: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9dc08e176c148a5bbab6e0dcbe9a50c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a75c5b65a1449c984cb4f9f2756f61c : SUCCESS in 5m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bedf6107e98f4325a81453272b82a331 : SUCCESS in 1h 19m 22s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b653bef3f8d34d90bd156ba35b4b946d : SUCCESS in 30m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c06e7e3a759b43138b1780ce050ac224 : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":70},{"id":"5392f232c4b16142917fef2a100b0db6d0108d7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-04 13:32:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":70}],"current_revision_number":70,"current_revision":"7b2566a61ef2dc2d4c84460b54ded70d11ed4a79","revisions":{"f77b7eed531f9f103f80ba756ec1426aebadddd7":{"kind":"REWORK","_number":1,"created":"2021-08-25 04:49:52.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/1"}}},"commit":{"parents":[{"commit":"8ccbbebe6d424265ff717197b9a323917518e8c5","subject":"[CI] Temporarily disable IPv6 scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8ccbbebe6d424265ff717197b9a323917518e8c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-25 04:24:45.000000000","tz":180},"subject":"rbac","message":"rbac\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f77b7eed531f9f103f80ba756ec1426aebadddd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f77b7eed531f9f103f80ba756ec1426aebadddd7"}]},"branch":"refs/heads/master"},"3c8cf9ebea41c7d5ca1d5a6598e962c4d64f049f":{"kind":"REWORK","_number":2,"created":"2021-08-25 05:09:03.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/2"}}},"commit":{"parents":[{"commit":"8ccbbebe6d424265ff717197b9a323917518e8c5","subject":"[CI] Temporarily disable IPv6 scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8ccbbebe6d424265ff717197b9a323917518e8c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-25 05:06:50.000000000","tz":180},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3c8cf9ebea41c7d5ca1d5a6598e962c4d64f049f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3c8cf9ebea41c7d5ca1d5a6598e962c4d64f049f"}]},"branch":"refs/heads/master"},"bb63bdba047a1ab8de109b484f00db95d8c8a2be":{"kind":"REWORK","_number":3,"created":"2021-08-25 05:53:09.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/3"}}},"commit":{"parents":[{"commit":"8ccbbebe6d424265ff717197b9a323917518e8c5","subject":"[CI] Temporarily disable IPv6 scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8ccbbebe6d424265ff717197b9a323917518e8c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-25 05:52:46.000000000","tz":180},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bb63bdba047a1ab8de109b484f00db95d8c8a2be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bb63bdba047a1ab8de109b484f00db95d8c8a2be"}]},"branch":"refs/heads/master"},"7f8e9d7dfd959479617772844c1e0b567b22bb45":{"kind":"REWORK","_number":4,"created":"2021-08-25 06:36:36.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/4"}}},"commit":{"parents":[{"commit":"8ccbbebe6d424265ff717197b9a323917518e8c5","subject":"[CI] Temporarily disable IPv6 scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8ccbbebe6d424265ff717197b9a323917518e8c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-25 06:36:11.000000000","tz":180},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f8e9d7dfd959479617772844c1e0b567b22bb45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f8e9d7dfd959479617772844c1e0b567b22bb45"}]},"branch":"refs/heads/master"},"10e862aff3239998483fc11e8aa608e17c378517":{"kind":"REWORK","_number":5,"created":"2021-08-25 07:11:13.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/5"}}},"commit":{"parents":[{"commit":"8ccbbebe6d424265ff717197b9a323917518e8c5","subject":"[CI] Temporarily disable IPv6 scenario tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8ccbbebe6d424265ff717197b9a323917518e8c5"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-08-25 07:10:51.000000000","tz":180},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/10e862aff3239998483fc11e8aa608e17c378517"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/10e862aff3239998483fc11e8aa608e17c378517"}]},"branch":"refs/heads/master"},"1ff5209c266d3b94678f40b9b0646873242400d5":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-10-26 13:02:10.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/38/805938/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/6"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2021-10-26 13:02:10.000000000","tz":0},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ff5209c266d3b94678f40b9b0646873242400d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ff5209c266d3b94678f40b9b0646873242400d5"}]},"branch":"refs/heads/master","description":"Rebase"},"c7f2f0097f4019c036f9116167c677da3b354581":{"kind":"REWORK","_number":7,"created":"2021-10-27 12:16:07.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/7"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-10-27 12:15:48.000000000","tz":180},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c7f2f0097f4019c036f9116167c677da3b354581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c7f2f0097f4019c036f9116167c677da3b354581"}]},"branch":"refs/heads/master"},"52b52fe2017704bb0d206835854618bf6afa18e6":{"kind":"REWORK","_number":8,"created":"2021-11-01 12:22:48.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/38/805938/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/8"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2021-11-01 12:22:28.000000000","tz":-240},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/52b52fe2017704bb0d206835854618bf6afa18e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/52b52fe2017704bb0d206835854618bf6afa18e6"}]},"branch":"refs/heads/master"},"083f36a8f0e64002a693f3f1e4fa94685cbc331a":{"kind":"REWORK","_number":9,"created":"2021-11-01 12:59:37.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/38/805938/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/9"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2021-11-01 12:59:28.000000000","tz":-240},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/083f36a8f0e64002a693f3f1e4fa94685cbc331a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/083f36a8f0e64002a693f3f1e4fa94685cbc331a"}]},"branch":"refs/heads/master"},"062fc21639ad614dbfb2ccc84da5e5e5d795e352":{"kind":"REWORK","_number":10,"created":"2021-11-01 19:24:55.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/38/805938/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/10"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2021-11-01 19:22:25.000000000","tz":-240},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/062fc21639ad614dbfb2ccc84da5e5e5d795e352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/062fc21639ad614dbfb2ccc84da5e5e5d795e352"}]},"branch":"refs/heads/master"},"4d2cf38fcc9b7f346cac28113fef03d613cc3931":{"kind":"REWORK","_number":11,"created":"2021-11-01 19:27:57.000000000","uploader":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"ref":"refs/changes/38/805938/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/11"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"Vida Haririan","email":"vhariria@redhat.com","date":"2021-11-01 19:27:42.000000000","tz":-240},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d2cf38fcc9b7f346cac28113fef03d613cc3931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4d2cf38fcc9b7f346cac28113fef03d613cc3931"}]},"branch":"refs/heads/master"},"7f707d18ed322a592905455a8c7bb39e70b7d1bc":{"kind":"REWORK","_number":12,"created":"2021-11-16 12:08:46.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/12"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-11-16 12:08:24.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f707d18ed322a592905455a8c7bb39e70b7d1bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7f707d18ed322a592905455a8c7bb39e70b7d1bc"}]},"branch":"refs/heads/master"},"5ccecbb95f363a1b38a23af2554a883541c591e0":{"kind":"REWORK","_number":13,"created":"2021-11-16 15:07:59.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/13"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-11-16 15:07:40.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ccecbb95f363a1b38a23af2554a883541c591e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5ccecbb95f363a1b38a23af2554a883541c591e0"}]},"branch":"refs/heads/master"},"bbb013b4abe2517aba644d5497d3859236618bcf":{"kind":"REWORK","_number":14,"created":"2021-11-17 15:42:48.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/14"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-11-17 15:42:02.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bbb013b4abe2517aba644d5497d3859236618bcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bbb013b4abe2517aba644d5497d3859236618bcf"}]},"branch":"refs/heads/master"},"abd9a1ee11ba7f22ce0661c07288d9672b0757e1":{"kind":"REWORK","_number":15,"created":"2021-11-18 09:36:39.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/15"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-11-18 09:36:09.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/abd9a1ee11ba7f22ce0661c07288d9672b0757e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/abd9a1ee11ba7f22ce0661c07288d9672b0757e1"}]},"branch":"refs/heads/master"},"f65f688fee9017a92735e98f47c2284166a7b773":{"kind":"REWORK","_number":16,"created":"2021-11-18 09:40:03.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/16"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-11-18 09:39:45.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f65f688fee9017a92735e98f47c2284166a7b773"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f65f688fee9017a92735e98f47c2284166a7b773"}]},"branch":"refs/heads/master"},"10727230fd6f116f54cc914085d94296224e3e1e":{"kind":"REWORK","_number":17,"created":"2021-11-18 09:44:10.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/17"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-11-18 09:43:47.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/10727230fd6f116f54cc914085d94296224e3e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/10727230fd6f116f54cc914085d94296224e3e1e"}]},"branch":"refs/heads/master"},"f25b55b5b28a03af06853b8876e31cf015c39941":{"kind":"REWORK","_number":18,"created":"2021-11-23 09:41:46.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/18"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-11-23 09:39:56.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/818689\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f25b55b5b28a03af06853b8876e31cf015c39941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f25b55b5b28a03af06853b8876e31cf015c39941"}]},"branch":"refs/heads/master"},"20e131965d34f11d1e3a2a541da497dc82449c0a":{"kind":"REWORK","_number":19,"created":"2021-11-23 10:08:39.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/19"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-11-23 10:08:16.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/818689\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/20e131965d34f11d1e3a2a541da497dc82449c0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/20e131965d34f11d1e3a2a541da497dc82449c0a"}]},"branch":"refs/heads/master"},"ea0d3124c1832141e460e34e4553725aecbf53fd":{"kind":"REWORK","_number":20,"created":"2021-11-23 12:44:30.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/20"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-11-23 12:44:09.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/818689\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ea0d3124c1832141e460e34e4553725aecbf53fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ea0d3124c1832141e460e34e4553725aecbf53fd"}]},"branch":"refs/heads/master"},"6c9d88b325d5b1909503b8e3b8e72959e0308946":{"kind":"REWORK","_number":21,"created":"2021-12-14 07:44:40.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/21"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-12-14 07:44:15.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/818689\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6c9d88b325d5b1909503b8e3b8e72959e0308946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6c9d88b325d5b1909503b8e3b8e72959e0308946"}]},"branch":"refs/heads/master"},"b53e3464d0275a252f1835c1d959d777078ac7f6":{"kind":"REWORK","_number":22,"created":"2021-12-14 09:00:09.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/22"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-12-14 08:59:10.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/818689\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b53e3464d0275a252f1835c1d959d777078ac7f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b53e3464d0275a252f1835c1d959d777078ac7f6"}]},"branch":"refs/heads/master"},"ec3897c9c964b69b8e79488605496915234d5ec9":{"kind":"REWORK","_number":23,"created":"2021-12-14 09:17:54.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/23"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-12-14 09:17:35.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/818689\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ec3897c9c964b69b8e79488605496915234d5ec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ec3897c9c964b69b8e79488605496915234d5ec9"}]},"branch":"refs/heads/master"},"58d894fa6a5e09e6f7d1abe60cfb3aceb013405f":{"kind":"REWORK","_number":24,"created":"2021-12-21 11:20:08.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/24"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-12-21 11:18:39.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/818689\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58d894fa6a5e09e6f7d1abe60cfb3aceb013405f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/58d894fa6a5e09e6f7d1abe60cfb3aceb013405f"}]},"branch":"refs/heads/master"},"5cf27a535c0b14d22e7e9789e1f81994dc1a0906":{"kind":"REWORK","_number":25,"created":"2021-12-21 13:10:33.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/25"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-12-21 13:10:13.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/818689\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5cf27a535c0b14d22e7e9789e1f81994dc1a0906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5cf27a535c0b14d22e7e9789e1f81994dc1a0906"}]},"branch":"refs/heads/master"},"357f4adcbe06917f0fb8ed9a29c5cb82b913ff41":{"kind":"REWORK","_number":26,"created":"2021-12-21 14:27:40.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/26"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-12-21 14:27:20.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/818689\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/357f4adcbe06917f0fb8ed9a29c5cb82b913ff41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/357f4adcbe06917f0fb8ed9a29c5cb82b913ff41"}]},"branch":"refs/heads/master"},"cde989f815a91de6bdffe32fcd578d5583eb1b50":{"kind":"REWORK","_number":27,"created":"2021-12-22 16:32:38.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/27"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-12-22 16:32:19.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/818689\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cde989f815a91de6bdffe32fcd578d5583eb1b50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cde989f815a91de6bdffe32fcd578d5583eb1b50"}]},"branch":"refs/heads/master"},"93e7131084257e25608e603a0ca79e0ca841f5a6":{"kind":"REWORK","_number":28,"created":"2021-12-22 19:11:38.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/28"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-12-22 19:11:17.000000000","tz":120},"subject":"[WIP] Add RBAC tests for manila","message":"[WIP] Add RBAC tests for manila\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/818689\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/93e7131084257e25608e603a0ca79e0ca841f5a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/93e7131084257e25608e603a0ca79e0ca841f5a6"}]},"branch":"refs/heads/master"},"52c1bca86e25197bbe01099e462aafa16029bd9d":{"kind":"REWORK","_number":29,"created":"2021-12-23 09:28:25.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/29"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-12-23 09:22:30.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/818689\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/52c1bca86e25197bbe01099e462aafa16029bd9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/52c1bca86e25197bbe01099e462aafa16029bd9d"}]},"branch":"refs/heads/master"},"d9103ef3c758ad99fbd62187a6234234d2451b2e":{"kind":"REWORK","_number":30,"created":"2022-01-11 10:00:38.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/30"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-11 09:59:59.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d9103ef3c758ad99fbd62187a6234234d2451b2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d9103ef3c758ad99fbd62187a6234234d2451b2e"}]},"branch":"refs/heads/master"},"e8448bffb591f5f85ffc61b231405b7b8dc35897":{"kind":"REWORK","_number":31,"created":"2022-01-11 11:22:48.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/31"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-11 11:22:28.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e8448bffb591f5f85ffc61b231405b7b8dc35897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e8448bffb591f5f85ffc61b231405b7b8dc35897"}]},"branch":"refs/heads/master"},"aed372bd7e6017a643e2c58bd7b4dbae432d3f1b":{"kind":"REWORK","_number":32,"created":"2022-01-11 12:27:19.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/32"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-11 12:26:59.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/aed372bd7e6017a643e2c58bd7b4dbae432d3f1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/aed372bd7e6017a643e2c58bd7b4dbae432d3f1b"}]},"branch":"refs/heads/master"},"15b402842e70173769256669fe380f997b2dd353":{"kind":"REWORK","_number":33,"created":"2022-01-11 13:25:51.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/33"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-11 13:25:12.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/15b402842e70173769256669fe380f997b2dd353"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/15b402842e70173769256669fe380f997b2dd353"}]},"branch":"refs/heads/master"},"65f1224eb4bc2d54b31f3aa080b864b670bdfcdb":{"kind":"REWORK","_number":34,"created":"2022-01-11 14:39:21.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/34"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-11 14:38:54.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/65f1224eb4bc2d54b31f3aa080b864b670bdfcdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/65f1224eb4bc2d54b31f3aa080b864b670bdfcdb"}]},"branch":"refs/heads/master"},"164c46d5888e0770b53b614cfc144e6710140641":{"kind":"REWORK","_number":35,"created":"2022-01-11 15:25:26.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/35"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-11 15:25:04.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/164c46d5888e0770b53b614cfc144e6710140641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/164c46d5888e0770b53b614cfc144e6710140641"}]},"branch":"refs/heads/master"},"b77bded811dbc4e407f82ea6ca4d4037dc4ed450":{"kind":"REWORK","_number":36,"created":"2022-01-30 12:58:39.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/36"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-30 12:58:10.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b77bded811dbc4e407f82ea6ca4d4037dc4ed450"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b77bded811dbc4e407f82ea6ca4d4037dc4ed450"}]},"branch":"refs/heads/master"},"dbc7bfd0d8d2e77b43c1b0ac44bf7aa26b32e83b":{"kind":"REWORK","_number":37,"created":"2022-01-30 13:19:03.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/37"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-30 13:18:44.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dbc7bfd0d8d2e77b43c1b0ac44bf7aa26b32e83b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dbc7bfd0d8d2e77b43c1b0ac44bf7aa26b32e83b"}]},"branch":"refs/heads/master"},"7e52a97d20723a5812d5d040fa9f489fe5dc26ef":{"kind":"REWORK","_number":38,"created":"2022-01-30 15:27:22.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/38"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-30 15:22:04.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7e52a97d20723a5812d5d040fa9f489fe5dc26ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7e52a97d20723a5812d5d040fa9f489fe5dc26ef"}]},"branch":"refs/heads/master"},"fa45121f51f613b9fa16c98ee83364744c922c03":{"kind":"REWORK","_number":39,"created":"2022-01-30 17:33:14.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/39"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-30 17:32:56.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fa45121f51f613b9fa16c98ee83364744c922c03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fa45121f51f613b9fa16c98ee83364744c922c03"}]},"branch":"refs/heads/master"},"6b454d90d8c911b7325da752f0f90c7fb4291328":{"kind":"REWORK","_number":40,"created":"2022-01-30 19:49:48.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/40"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-30 19:49:27.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6b454d90d8c911b7325da752f0f90c7fb4291328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6b454d90d8c911b7325da752f0f90c7fb4291328"}]},"branch":"refs/heads/master"},"3cc13cc5b7754eefcf21326c462b8000d04ecbb8":{"kind":"REWORK","_number":41,"created":"2022-01-30 21:07:57.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/41"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-30 21:07:32.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3cc13cc5b7754eefcf21326c462b8000d04ecbb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3cc13cc5b7754eefcf21326c462b8000d04ecbb8"}]},"branch":"refs/heads/master"},"147825fb4d46ccc94c4110a87936afb2928201e5":{"kind":"REWORK","_number":42,"created":"2022-01-31 14:45:21.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/42"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-31 14:45:02.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/147825fb4d46ccc94c4110a87936afb2928201e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/147825fb4d46ccc94c4110a87936afb2928201e5"}]},"branch":"refs/heads/master"},"5cc5796e84c591a3419bc8a00af65f195f5a6a5b":{"kind":"REWORK","_number":43,"created":"2022-01-31 14:48:22.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/43"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-31 14:47:58.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5cc5796e84c591a3419bc8a00af65f195f5a6a5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5cc5796e84c591a3419bc8a00af65f195f5a6a5b"}]},"branch":"refs/heads/master"},"430b9ce0dfeefa6fc5b73a39a0f24898a8e54bf4":{"kind":"REWORK","_number":44,"created":"2022-01-31 15:54:52.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/44"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-31 15:54:32.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/430b9ce0dfeefa6fc5b73a39a0f24898a8e54bf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/430b9ce0dfeefa6fc5b73a39a0f24898a8e54bf4"}]},"branch":"refs/heads/master"},"d8b18fb31784bc6db2fdf9afcbeb55fad29e022e":{"kind":"REWORK","_number":45,"created":"2022-01-31 19:16:14.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/45"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-01-31 19:15:51.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d8b18fb31784bc6db2fdf9afcbeb55fad29e022e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d8b18fb31784bc6db2fdf9afcbeb55fad29e022e"}]},"branch":"refs/heads/master"},"44b6265a006c47545716ce3241da63a8685dcf23":{"kind":"REWORK","_number":46,"created":"2022-02-13 09:58:31.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/46"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-02-13 09:58:09.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/44b6265a006c47545716ce3241da63a8685dcf23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/44b6265a006c47545716ce3241da63a8685dcf23"}]},"branch":"refs/heads/master"},"14391108b957cd4394e1974734d0a2170f2ac489":{"kind":"REWORK","_number":47,"created":"2022-02-13 12:00:30.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/47"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-02-13 12:00:10.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/14391108b957cd4394e1974734d0a2170f2ac489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/14391108b957cd4394e1974734d0a2170f2ac489"}]},"branch":"refs/heads/master"},"2015fe8ad93f399fda86cf7a86f9a070a846da6b":{"kind":"REWORK","_number":48,"created":"2022-02-13 13:27:25.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/48"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-02-13 13:27:05.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2015fe8ad93f399fda86cf7a86f9a070a846da6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2015fe8ad93f399fda86cf7a86f9a070a846da6b"}]},"branch":"refs/heads/master"},"330121fa11cafbb9c8e51322c483b2c3c8ef5c04":{"kind":"REWORK","_number":49,"created":"2022-02-13 15:40:40.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/49"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-02-13 15:40:15.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/330121fa11cafbb9c8e51322c483b2c3c8ef5c04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/330121fa11cafbb9c8e51322c483b2c3c8ef5c04"}]},"branch":"refs/heads/master"},"b5c2f489a4079384319e315bef5cd762f78ca1e4":{"kind":"NO_CODE_CHANGE","_number":50,"created":"2022-02-15 13:39:03.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/50"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-02-15 13:33:13.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5c2f489a4079384319e315bef5cd762f78ca1e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5c2f489a4079384319e315bef5cd762f78ca1e4"}]},"branch":"refs/heads/master"},"167a8d72c26d69732a9af012ad12d32de1d3b5ca":{"kind":"REWORK","_number":51,"created":"2022-02-22 13:24:47.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/51"}}},"commit":{"parents":[{"commit":"e82105ea449ac07190aadbe8d584a11e6d9e9e70","subject":"Merge \"Add stable/xena job and drop stable/ussuri job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e82105ea449ac07190aadbe8d584a11e6d9e9e70"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-02-22 13:24:33.000000000","tz":120},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/167a8d72c26d69732a9af012ad12d32de1d3b5ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/167a8d72c26d69732a9af012ad12d32de1d3b5ca"}]},"branch":"refs/heads/master"},"9f2654fa9405809124acda44fece51beb66315c7":{"kind":"TRIVIAL_REBASE","_number":52,"created":"2022-05-24 13:58:10.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/52"}}},"commit":{"parents":[{"commit":"a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab","subject":"Define queues at project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","date":"2022-05-24 13:58:10.000000000","tz":0},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9f2654fa9405809124acda44fece51beb66315c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9f2654fa9405809124acda44fece51beb66315c7"}]},"branch":"refs/heads/master","description":"Rebase"},"4c5a15e43a4b4a9234774ac44d31b70e030032bf":{"kind":"REWORK","_number":53,"created":"2022-06-14 04:25:50.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/53"}}},"commit":{"parents":[{"commit":"cc8a651faeac27731a981eb54765da72a162d93a","subject":"Split generic job by tests and reduce concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc8a651faeac27731a981eb54765da72a162d93a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-06-14 04:25:37.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4c5a15e43a4b4a9234774ac44d31b70e030032bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4c5a15e43a4b4a9234774ac44d31b70e030032bf"}]},"branch":"refs/heads/master"},"88624c6295ffe9cb9feffff51b85c37a93b3e749":{"kind":"REWORK","_number":54,"created":"2022-06-19 05:23:34.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/54"}}},"commit":{"parents":[{"commit":"cc8a651faeac27731a981eb54765da72a162d93a","subject":"Split generic job by tests and reduce concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc8a651faeac27731a981eb54765da72a162d93a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-06-19 05:23:23.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/88624c6295ffe9cb9feffff51b85c37a93b3e749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/88624c6295ffe9cb9feffff51b85c37a93b3e749"}]},"branch":"refs/heads/master"},"735c474140fcbe6d076b27fba5def89c4c31991b":{"kind":"REWORK","_number":55,"created":"2022-06-19 08:13:44.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/55"}}},"commit":{"parents":[{"commit":"cc8a651faeac27731a981eb54765da72a162d93a","subject":"Split generic job by tests and reduce concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc8a651faeac27731a981eb54765da72a162d93a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-06-19 08:13:32.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/735c474140fcbe6d076b27fba5def89c4c31991b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/735c474140fcbe6d076b27fba5def89c4c31991b"}]},"branch":"refs/heads/master"},"3f73d99561ef4e25ee61739df612c1978a6e8634":{"kind":"REWORK","_number":56,"created":"2022-06-19 08:32:53.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/56"}}},"commit":{"parents":[{"commit":"cc8a651faeac27731a981eb54765da72a162d93a","subject":"Split generic job by tests and reduce concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc8a651faeac27731a981eb54765da72a162d93a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-06-19 08:32:40.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f73d99561ef4e25ee61739df612c1978a6e8634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3f73d99561ef4e25ee61739df612c1978a6e8634"}]},"branch":"refs/heads/master"},"b386bdeafbb17d18d96086bffdc1fadd38ac6fc2":{"kind":"REWORK","_number":57,"created":"2022-06-19 09:27:01.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/57"}}},"commit":{"parents":[{"commit":"cc8a651faeac27731a981eb54765da72a162d93a","subject":"Split generic job by tests and reduce concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc8a651faeac27731a981eb54765da72a162d93a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-06-19 09:25:52.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b386bdeafbb17d18d96086bffdc1fadd38ac6fc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b386bdeafbb17d18d96086bffdc1fadd38ac6fc2"}]},"branch":"refs/heads/master"},"e3d61043aecd88a9d35eb2f04948549dc755dada":{"kind":"REWORK","_number":58,"created":"2022-06-19 09:47:01.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/58"}}},"commit":{"parents":[{"commit":"cc8a651faeac27731a981eb54765da72a162d93a","subject":"Split generic job by tests and reduce concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc8a651faeac27731a981eb54765da72a162d93a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-06-19 09:46:42.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e3d61043aecd88a9d35eb2f04948549dc755dada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e3d61043aecd88a9d35eb2f04948549dc755dada"}]},"branch":"refs/heads/master"},"6f0ed4b3be78bb54d90c302b2a26e6c2cb3a036a":{"kind":"REWORK","_number":59,"created":"2022-06-19 10:39:01.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/59"}}},"commit":{"parents":[{"commit":"cc8a651faeac27731a981eb54765da72a162d93a","subject":"Split generic job by tests and reduce concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc8a651faeac27731a981eb54765da72a162d93a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-06-19 10:38:51.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f0ed4b3be78bb54d90c302b2a26e6c2cb3a036a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f0ed4b3be78bb54d90c302b2a26e6c2cb3a036a"}]},"branch":"refs/heads/master"},"019ecea5b950f845b254e4ce7e69687ea83e65b9":{"kind":"REWORK","_number":60,"created":"2022-07-03 07:45:14.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/60"}}},"commit":{"parents":[{"commit":"d12285b25b40031ec90743c45b095d1bbcba4ca4","subject":"Follow up for CI jobs refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d12285b25b40031ec90743c45b095d1bbcba4ca4"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-07-03 07:45:02.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/019ecea5b950f845b254e4ce7e69687ea83e65b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/019ecea5b950f845b254e4ce7e69687ea83e65b9"}]},"branch":"refs/heads/master"},"de50b183f0ee11651164f148aedf317833b0985d":{"kind":"REWORK","_number":61,"created":"2022-07-07 12:15:27.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/61"}}},"commit":{"parents":[{"commit":"b42c4f84deb69ab7bb33baf97f4f7fe8094eb570","subject":"Merge \"Add FIPS testing job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b42c4f84deb69ab7bb33baf97f4f7fe8094eb570"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-07-07 12:15:15.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/de50b183f0ee11651164f148aedf317833b0985d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/de50b183f0ee11651164f148aedf317833b0985d"}]},"branch":"refs/heads/master"},"28bf7d7505341b6d39dd3c81cc2c152e4f386f44":{"kind":"REWORK","_number":62,"created":"2022-07-07 15:10:44.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/62"}}},"commit":{"parents":[{"commit":"b42c4f84deb69ab7bb33baf97f4f7fe8094eb570","subject":"Merge \"Add FIPS testing job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b42c4f84deb69ab7bb33baf97f4f7fe8094eb570"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-07-07 15:10:33.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/28bf7d7505341b6d39dd3c81cc2c152e4f386f44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/28bf7d7505341b6d39dd3c81cc2c152e4f386f44"}]},"branch":"refs/heads/master"},"6b05221577dc9540c3ab3cc160dd144913db59fe":{"kind":"REWORK","_number":63,"created":"2022-09-13 14:03:56.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/63"}}},"commit":{"parents":[{"commit":"b42c4f84deb69ab7bb33baf97f4f7fe8094eb570","subject":"Merge \"Add FIPS testing job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b42c4f84deb69ab7bb33baf97f4f7fe8094eb570"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-09-13 14:03:45.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6b05221577dc9540c3ab3cc160dd144913db59fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6b05221577dc9540c3ab3cc160dd144913db59fe"}]},"branch":"refs/heads/master"},"9692088d617f3bfb2f40232ad166824d881effd8":{"kind":"REWORK","_number":64,"created":"2022-09-15 09:55:52.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/64"}}},"commit":{"parents":[{"commit":"b42c4f84deb69ab7bb33baf97f4f7fe8094eb570","subject":"Merge \"Add FIPS testing job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b42c4f84deb69ab7bb33baf97f4f7fe8094eb570"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-09-15 09:55:41.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9692088d617f3bfb2f40232ad166824d881effd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9692088d617f3bfb2f40232ad166824d881effd8"}]},"branch":"refs/heads/master"},"a83e4dbcbccc8996bed2670c6db9b06a43f63ae2":{"kind":"REWORK","_number":65,"created":"2022-09-21 09:56:00.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/65"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-09-21 09:55:49.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a83e4dbcbccc8996bed2670c6db9b06a43f63ae2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a83e4dbcbccc8996bed2670c6db9b06a43f63ae2"}]},"branch":"refs/heads/master"},"1b2e2aee83f9e9119475259af4f6d4777ac41832":{"kind":"NO_CODE_CHANGE","_number":66,"created":"2022-09-21 15:08:18.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/38/805938/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/66"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-09-21 15:08:18.000000000","tz":0},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/858761\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1b2e2aee83f9e9119475259af4f6d4777ac41832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1b2e2aee83f9e9119475259af4f6d4777ac41832"}]},"branch":"refs/heads/master","description":"Edit commit message"},"659da362a40b3e8b5b811ef91e68edad7110ad03":{"kind":"REWORK","_number":67,"created":"2022-09-22 07:25:50.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/67"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-09-22 07:25:32.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/659da362a40b3e8b5b811ef91e68edad7110ad03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/659da362a40b3e8b5b811ef91e68edad7110ad03"}]},"branch":"refs/heads/master"},"2d497becce94bcf813b780cb271a843b47122d0b":{"kind":"NO_CODE_CHANGE","_number":68,"created":"2022-09-22 09:43:08.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/68"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-09-22 09:42:46.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/858761\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2d497becce94bcf813b780cb271a843b47122d0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2d497becce94bcf813b780cb271a843b47122d0b"}]},"branch":"refs/heads/master"},"4410556b902b03bb8a54ebcdae27c6db97b43ac3":{"kind":"REWORK","_number":69,"created":"2022-10-02 07:13:57.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/69"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-02 07:13:43.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/858761\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4410556b902b03bb8a54ebcdae27c6db97b43ac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4410556b902b03bb8a54ebcdae27c6db97b43ac3"}]},"branch":"refs/heads/master"},"7b2566a61ef2dc2d4c84460b54ded70d11ed4a79":{"kind":"REWORK","_number":70,"created":"2022-10-02 09:30:54.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/38/805938/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/805938/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/805938/70"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2021-04-14 07:31:31.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-02 09:30:44.000000000","tz":180},"subject":"Add S-RBAC tests for manila","message":"Add S-RBAC tests for manila\n\nThe tests validate policy share actions for admin,member\nand reader users in a project scope.\nThis patch adds tests for the following resources/features:\n- share\n- share snapshot\n- share type\n- share network\n- scheduler stats\n\nBelow settings were used for ease of code development\nand testing:\n1. DHHS \u003d False\n2. Dummy driver as it supports all features\n3. Manila-tempest-plugin as a base repo to test the changes\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/858761\nChange-Id: I53bb9851445038f81032485817389ec31e2b7341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7b2566a61ef2dc2d4c84460b54ded70d11ed4a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7b2566a61ef2dc2d4c84460b54ded70d11ed4a79"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[]}
