)]}'
{"id":"openstack%2Fmanila-tempest-plugin~808048","triplet_id":"openstack%2Fmanila-tempest-plugin~master~If9c123363d71ac6bc89e743efe4dc2a00299cf5a","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/affinity-antiaffinity-filter","hashtags":[],"change_id":"If9c123363d71ac6bc89e743efe4dc2a00299cf5a","subject":"Add tests for scheduler hints of share create.","status":"MERGED","created":"2021-09-09 10:48:37.000000000","updated":"2021-09-17 18:26:18.000000000","submitted":"2021-09-17 18:26:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"808048-bp/affinity-antiaffinity-filter","meta_rev_id":"cfec04d628849e12318c94f1aaa6d1c80876b759","_number":808048,"virtual_id_number":808048,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-17 18:26:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"recommended":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"all":[{"value":1,"date":"2021-09-16 20:18:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":2,"date":"2021-09-17 17:19:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2021-09-16 19:32:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":1,"date":"2021-09-17 14:50:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":1,"date":"2021-09-17 17:19:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"}],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-09 10:49:54.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},"state":"REVIEWER"},{"updated":"2021-09-09 10:49:54.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2021-09-09 10:49:54.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2021-09-09 10:49:54.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2021-09-09 10:49:54.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2021-09-09 10:49:54.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2021-09-09 12:35:49.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-09-09 12:36:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-10 10:20:16.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-10 11:00:46.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-16 20:06:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"},"state":"REVIEWER"},{"updated":"2021-09-16 20:06:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2021-09-16 20:06:49.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2021-09-16 20:18:52.000000000","updated_by":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"}],"messages":[{"id":"962186124324742ea4af28e961f5c414e2ba409f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-09-09 10:48:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91a0df19ae549745a3148b4bf96e2ed372988f92","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-09 12:35:49.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/48/808048/1/check/DellEMC_Unity_Manila_dhss_true/6c2cfdf/DellEMC_Unity_Manila_dhss_true/331 : SUCCESS in 1h 46m 54s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/808048/1/check/DellEMC_Unity_Manila_dhss_false/d2eb200/DellEMC_Unity_Manila_dhss_false/327 : SUCCESS in 1h 25m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"0c9b823c14c3fe2bce3aba0fccac1e2f121a91ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 12:36:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82e0913787504096bbb2a38c6d8242a6 : SUCCESS in 4m 25s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/662e0c0597c54a32af1ef47118bdb323 : SUCCESS in 33m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/88f66074a8c24a0d98b550abe94ee7e9 : SUCCESS in 34m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1a08a9b9331d4eca85d4f06b9402bcd1 : SUCCESS in 1h 00m 24s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9b2b2e5091d14cbd867733118f31e2ef : SUCCESS in 1h 13m 26s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/6da7cac7f4734c09b00a009066f722f8 : SUCCESS in 57m 30s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/596385bd6b124ee0b70c545633406481 : FAILURE in 1h 02m 46s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/961b609d36b14fa899d8d7a9a92a696d : FAILURE in 1h 20m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9790d84f153a47a7b69f51590979878f : SUCCESS in 1h 19m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d77a0caf76c94ba0be0cfd7b72e4017b : FAILURE in 1h 34m 05s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/33cdd633bae942a4abf90dda98526075 : SUCCESS in 51m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a46d5802325a4acda25b7604f989d83b : FAILURE in 1h 19m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1c7419c602d74ab4bd75217d65ffe28d : SUCCESS in 23m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9408947b57abaaa3e7505927d5d672763fedd18","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-09-09 13:49:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"97d860227c5ca5cbd140692fb602fbcc17019b35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 15:41:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0ee727e7298421fa91850a828eaee62 : SUCCESS in 4m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3a3ad177683e4258b8bd059fdd087fcc : SUCCESS in 32m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d0c28b40f7dc4cd4bff30f88c57c4534 : SUCCESS in 32m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/be6ed3cf4535478d8941c9fb017bbb90 : SUCCESS in 1h 04m 16s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/010ffb1cc8aa45eca798527d82a6a6fc : SUCCESS in 56m 41s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/7e6b7b33652b45878603ac95b4ee42ab : SUCCESS in 1h 21m 44s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/eb361a4a6a5e425da320d4f7fac76482 : FAILURE in 1h 06m 59s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/476e97e4bdef413b8afa10daa76a3058 : SUCCESS in 1h 08m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1f87582bc83e4697a810b1da66388cb5 : SUCCESS in 1h 06m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cdc995413e8443efade400d2f2f50c4e : FAILURE in 1h 22m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/56f26de36d6d471ab1777c371e418466 : SUCCESS in 50m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3ab2698fd9cb440384674b14e87488b3 : FAILURE in 1h 47m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/724ef0c9a70e422fabda764482ca8117 : SUCCESS in 22m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c46cd45f0c706a40faaee8d6a78efa0d44c08866","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-09-10 08:37:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0387ce7647fe0e82b1582e91a111ee8fe8739ec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 10:19:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15e24089d021406db10476912c7749ea : SUCCESS in 4m 53s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0c091a87818b4a6b94a60dd15efc5561 : SUCCESS in 30m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0926065139744051b8303b2b7366b01b : SUCCESS in 43m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fbfd287e936342a6a85109c72214cbe8 : SUCCESS in 1h 02m 51s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a8175af40c9344bb80b3cc0a8d0cc47c : SUCCESS in 1h 34m 08s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/828001d5b2b54961865ca9f08c421979 : SUCCESS in 1h 08m 41s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/54a9011e86e5451b9e5752793663ec97 : SUCCESS in 1h 13m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/14c695c056804235a250cbc3b2dae8a5 : SUCCESS in 1h 16m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7d9c7ac5e920491cac5ace6abd0dee72 : SUCCESS in 1h 09m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4d00cac7fc0b4e29b0f20c081eddb64f : FAILURE in 1h 23m 31s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c4cec9f6783142fc8673af78fad5c24a : SUCCESS in 45m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a5ddccc9908c4e5d8002e855a8373692 : FAILURE in 1h 36m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ab670a186555424a9d1cdf0e5c9368da : FAILURE in 32m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"36aad9b5f8dedd2ef27e66add2ac4cdca4f028f6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-09-10 10:20:16.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/808048/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b2cc3b8/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/808048/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f6d79e4/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"3a2ab82f23fcd0533d279646d54e3ea32e26fad3","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-09-10 11:00:46.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/48/808048/1/check/DellEMC_VNX_Manila/76c32e7/DellEMC_VNX_Manila/172 : FAILURE in 40m 34s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"a0ff5abfbf9c88885231329dab3a876fe5769226","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-10 11:46:14.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/48/808048/1/check/DellEMC_Unity_Manila_dhss_true/ce08eee/DellEMC_Unity_Manila_dhss_true/334 : FAILURE in 40m 56s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/808048/1/check/DellEMC_Unity_Manila_dhss_false/de1bfd0/DellEMC_Unity_Manila_dhss_false/330 : FAILURE in 1h 26m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2d5bb43b59f7e9799a6bcd43375d791688f18f68","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-09-13 03:26:23.000000000","message":"Patch Set 1: Code-Review-1\n\nhi, Kiran Pawar.\nthanks for your change.\ni think we also need add negative test.\nfor example\nmanila_tempest_tests/tests/api/test_scheduler_hints_negative.py\nin this file to test\nscheduler_hint \u003d {\"same_host\": \"%s\" % self.share_a[\"id\"],\n                  \"different_host\": \"%s\" % self.share_a[\"id\"]}\nwe will got scheduler erro(NotFound).","accounts_in_message":[],"_revision_number":1},{"id":"434b1c536c61f9c7a5666f7f34e3a78e3c3731ae","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-13 23:50:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)\n\nThanks for following up on this change, Kiran. Please see some comments inline","accounts_in_message":[],"_revision_number":1},{"id":"49c3a4d285e0cc295319025d89d8fc4d04ebe7ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-09-14 11:30:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"923d161b6540d5fb9d52c1ed2153644dd04f4bc1","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-09-14 11:31:14.000000000","message":"Patch Set 1:\n\n(6 comments)\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e hi, Kiran Pawar.\n\u003e thanks for your change.\n\u003e i think we also need add negative test.\n\u003e for example\n\u003e manila_tempest_tests/tests/api/test_scheduler_hints_negative.py\n\u003e in this file to test\n\u003e scheduler_hint \u003d {\"same_host\": \"%s\" % self.share_a[\"id\"],\n\u003e                   \"different_host\": \"%s\" % self.share_a[\"id\"]}\n\u003e we will got scheduler erro(NotFound).\n\nAdded,\n\nFixed all comments from Goutham","accounts_in_message":[],"_revision_number":1},{"id":"8f42765b5e01e67852881a8a9671b633f29afa3b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-14 13:16:23.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/48/808048/2/check/DellEMC_Unity_Manila_dhss_true/64dde5f/DellEMC_Unity_Manila_dhss_true/340 : FAILURE in 1h 24m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/808048/2/check/DellEMC_Unity_Manila_dhss_false/97c3fd6/DellEMC_Unity_Manila_dhss_false/336 : SUCCESS in 1h 45m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"8265ad0f2a687f1a82aa7f76e3d64b0ba93fb19a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 14:03:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/962fa109c113443d99cb2ad71bfc35d6 : SUCCESS in 4m 35s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b5ee860e2b344ff794d61102289d2296 : SUCCESS in 31m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/16d8ff2a05f24af0b1c191ab4d0e9282 : SUCCESS in 36m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fba7f966dfd345679d130196ebcba960 : SUCCESS in 58m 01s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/19c67f00f02c45bbb8fc4116146ae5fa : SUCCESS in 1h 06m 50s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/92de919d5e274fa8a5ff581e304c3688 : SUCCESS in 1h 21m 46s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/031066076f16475bbb373fadc7df9278 : FAILURE in 1h 08m 10s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d84084e76eac46f79eda5f137a1afdea : SUCCESS in 1h 10m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5d0c535504aa4676999ff338ec8810c4 : SUCCESS in 1h 15m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0eebb03b844f4299b24db20d23bb8d53 : FAILURE in 1h 49m 15s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d344a5d2443b486f966c86474f945248 : SUCCESS in 45m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3efe784342794f53ae21f8f4f1f010e1 : FAILURE in 2h 27m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e00b47c70fbd4cdfaa35c8344429dffc : SUCCESS in 25m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3fbe8571b0bc8da8fea4c0161d6e7763ff2d7402","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-09-14 15:12:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4a77ad32d2f22779c4131268e72a7fdceb1801be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 16:48:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29556b0cf99e4992a96b0e1101804c87 : SUCCESS in 4m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ed4833c906024c21b0454d22742f3e86 : SUCCESS in 33m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/57870e0c4128497a96173237b05b0e11 : SUCCESS in 36m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/59bd48870d8942b7959150746de37610 : SUCCESS in 1h 05m 18s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c6624b9f63c54fb6988f7ec17644532a : SUCCESS in 1h 14m 54s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/5f3419d4ce984a6480cf3c4c8722fd7d : SUCCESS in 1h 09m 50s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/eedfd55bb97546e6be4b812a6b005817 : SUCCESS in 1h 06m 36s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/95bfa2d525cc421ab558684ae17cfc73 : FAILURE in 1h 17m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/090f718ee97e4c5e8c35247eac744fa5 : SUCCESS in 1h 22m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/feae016011d4430c8b3033b769808435 : FAILURE in 1h 30m 24s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e672c00193c64082ab72812457103093 : SUCCESS in 36m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9f7106adb3914d8bae23bbe17802dace : FAILURE in 1h 33m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fe650e69ac054bf4a627c5222df8e983 : SUCCESS in 21m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18a6f265812c0cb2239c7a7fc22977005cf8314e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-09-14 16:48:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/808048/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/73121ca/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/808048/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/7569b89/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"f179f1ca3fba83d7c4a8db691e5cc34f39d204bb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-14 18:14:32.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/48/808048/2/check/DellEMC_Unity_Manila_dhss_true/45a3468/DellEMC_Unity_Manila_dhss_true/341 : FAILURE in 1h 25m 56s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/808048/2/check/DellEMC_Unity_Manila_dhss_false/1b34afe/DellEMC_Unity_Manila_dhss_false/337 : SUCCESS in 1h 24m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"7121148c4bbf6390b2e0500d903b64188f8e3d4c","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-09-14 18:16:12.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/48/808048/2/check/DellEMC_VNX_Manila/c5eb9fa/DellEMC_VNX_Manila/175 : FAILURE in 1h 27m 36s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"e595bb766297e8e655a591b3627adf5aee818070","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-14 19:40:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nHi Kiran, \n\nLGTM for the most part - thanks for the changes. One more thing to address however, please see the comment inline regarding API micro-versioning. \n\nThis tempest plugin is \"branchless\" - meaning that users can run the tests in this plugin against older stable branches of manila. Your new tests shouldn\u0027t be run against stable branches because they lack support for API version 2.65.","accounts_in_message":[],"_revision_number":2},{"id":"4da455af5a1bf5c49e472e7c6dd49560b91a9ce3","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2021-09-15 01:13:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"db416de1f293860bd13c928d60718046dfcc632d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-09-15 07:06:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"417cdce2daaaa3df98236c87d814776fc39499f1","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2021-09-15 07:07:11.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nfixed comments from haixin and Goutham","accounts_in_message":[],"_revision_number":3},{"id":"3245a03fbd9f56a103debdb959eb7535360bfaaf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-15 09:28:23.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/48/808048/3/check/DellEMC_Unity_Manila_dhss_true/02ff1b3/DellEMC_Unity_Manila_dhss_true/348 : FAILURE in 1h 25m 02s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/48/808048/3/check/DellEMC_Unity_Manila_dhss_false/48460c9/DellEMC_Unity_Manila_dhss_false/344 : SUCCESS in 1h 55m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a1b733ae40d47abf7cc1fa5fe2f16c12526c187e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 10:21:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b04846d041146b68dc1409f6e80ce1c : SUCCESS in 5m 39s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/69306e21f0104c7882c1b8eb6eedd070 : SUCCESS in 40m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8d3a66416da14089a30c4a0803216898 : SUCCESS in 41m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4893b72a2a2541f4bcee18544ee4d9a7 : SUCCESS in 1h 13m 41s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/5441102b7dae4d2b811b2d2e024d960d : SUCCESS in 1h 14m 49s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/1e006cec23ca425d9dc8dcc0eb560168 : SUCCESS in 1h 06m 23s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/fe58ed4504d54913b5d5d919ef8478ab : SUCCESS in 1h 06m 08s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/101fbdae3b0b44e1bfa40aba5985604b : SUCCESS in 1h 09m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0714295f38a2464eacee3a0fcd187eab : FAILURE in 1h 16m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/abc7b27699344ab189cc7be1b9e42a53 : FAILURE in 1h 44m 26s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d21f0cf6ca8d41c38b3bbcbada94d054 : POST_FAILURE in 52m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/77e943da5cd1483e9c4b50915a16f9d5 : TIMED_OUT in 3h 06m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c993184b68584aef8c68e3db4963d90d : SUCCESS in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fab27c0f793da3e2ebcacf562974dd92e7e7df94","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-09-15 10:21:59.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/48/808048/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f18ebb4/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/48/808048/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/ae0fc26/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"c9aeef8c4fe34dd5135efce9f302b13a703487d7","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-09-15 12:02:54.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/48/808048/3/check/DellEMC_VNX_Manila/7f54119/DellEMC_VNX_Manila/181 : SUCCESS in 1h 40m 59s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"3452fc853c145f5448aff0681bb6e9ed1f5f7468","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-09-16 19:32:33.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks Kiran, LGTM","accounts_in_message":[],"_revision_number":3},{"id":"b117331a289141f1917165ef53be1e0fffc7413a","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2021-09-16 20:18:52.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM, thanks Kiran!","accounts_in_message":[],"_revision_number":3},{"id":"e4e920c134dae738c1a1fdc44b2d5c3c08b318f4","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2021-09-17 14:50:47.000000000","message":"Patch Set 3: Code-Review+1\n\nThanks Kiran, looks good.","accounts_in_message":[],"_revision_number":3},{"id":"13e162b9a5b04776c8beac1abc8f8b563f1b44d3","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-09-17 17:19:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThanks Kiran, LGTM!","accounts_in_message":[],"_revision_number":3},{"id":"c15b431c3eaa909c4f7d3996dfbf3904aba81146","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 17:19:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"51212f231319856d740e16f6ca782dab7105d80e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 18:26:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb15ba131b34358a98076f1e5d7a901 : SUCCESS in 3m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f595a7d4b1b1467daaf110de7b2916e5 : SUCCESS in 36m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bb8be0095e0f423f8554f1702890fc67 : SUCCESS in 42m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7a3baf880a1e43e28ace9a2450d76de2 : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":3},{"id":"cfec04d628849e12318c94f1aaa6d1c80876b759","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 18:26:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"98b3e8782d20d193974173c5763236bb0478a289","revisions":{"9eebe1dcfa1bf6fe1d622aa968335af05ed4c8b7":{"kind":"REWORK","_number":1,"created":"2021-09-09 10:48:37.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/48/808048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/808048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/808048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/808048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/808048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/808048/1"}}},"commit":{"parents":[{"commit":"e4a1e88ef36b5867d2e7f5a1383bff1064f93623","subject":"Merge \"Add run tests as thin provisioning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4a1e88ef36b5867d2e7f5a1383bff1064f93623"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-09-09 10:43:50.000000000","tz":120},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-09-09 10:48:32.000000000","tz":120},"subject":"Add tests for scheduler hints of share create.","message":"Add tests for scheduler hints of share create.\n\nTests for both scheduler hints e.g. same_host and different_host are\nadded and guarded by multiple backends check. Default version updated to\n2.65.\n\nDepends-on: Ic42d8a0c1d22e77ae64e0ca014607b28fd336467\nPartially implements: bp/affinity-antiaffinity-filter\n\nChange-Id: If9c123363d71ac6bc89e743efe4dc2a00299cf5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9eebe1dcfa1bf6fe1d622aa968335af05ed4c8b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9eebe1dcfa1bf6fe1d622aa968335af05ed4c8b7"}]},"branch":"refs/heads/master"},"2c2781260669d17405353234bbdd7fc0b93175f6":{"kind":"REWORK","_number":2,"created":"2021-09-14 11:30:21.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/48/808048/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/808048/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/808048/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/808048/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/808048/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/808048/2"}}},"commit":{"parents":[{"commit":"e4a1e88ef36b5867d2e7f5a1383bff1064f93623","subject":"Merge \"Add run tests as thin provisioning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4a1e88ef36b5867d2e7f5a1383bff1064f93623"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-09-09 10:43:50.000000000","tz":120},"committer":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-09-14 11:19:05.000000000","tz":120},"subject":"Add tests for scheduler hints of share create.","message":"Add tests for scheduler hints of share create.\n\nTests for both scheduler hints e.g. same_host and different_host are\nadded and guarded by multiple backends check. Default version updated to\n2.65.\n\nDepends-on: Ic42d8a0c1d22e77ae64e0ca014607b28fd336467\nPartially implements: bp/affinity-antiaffinity-filter\n\nChange-Id: If9c123363d71ac6bc89e743efe4dc2a00299cf5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2c2781260669d17405353234bbdd7fc0b93175f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2c2781260669d17405353234bbdd7fc0b93175f6"}]},"branch":"refs/heads/master"},"98b3e8782d20d193974173c5763236bb0478a289":{"kind":"REWORK","_number":3,"created":"2021-09-15 07:06:34.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/48/808048/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/48/808048/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/808048/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/808048/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/808048/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/48/808048/3"}}},"commit":{"parents":[{"commit":"e4a1e88ef36b5867d2e7f5a1383bff1064f93623","subject":"Merge \"Add run tests as thin provisioning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4a1e88ef36b5867d2e7f5a1383bff1064f93623"}]}],"author":{"name":"kpdev","email":"kinpaa@gmail.com","date":"2021-09-09 10:43:50.000000000","tz":120},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2021-09-14 20:33:45.000000000","tz":0},"subject":"Add tests for scheduler hints of share create.","message":"Add tests for scheduler hints of share create.\n\nTests for both scheduler hints e.g. same_host and different_host are\nadded and guarded by multiple backends check. Default version updated to\n2.65.\n\nDepends-on: Ic42d8a0c1d22e77ae64e0ca014607b28fd336467\nPartially implements: bp/affinity-antiaffinity-filter\n\nChange-Id: If9c123363d71ac6bc89e743efe4dc2a00299cf5a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/98b3e8782d20d193974173c5763236bb0478a289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/98b3e8782d20d193974173c5763236bb0478a289"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
