)]}'
{"id":"openstack%2Fmanila-tempest-plugin~809330","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ib463a38cb731f1a61c5eee185343304ee991956e","project":"openstack/manila-tempest-plugin","branch":"master","topic":"manila_01","attention_set":{},"removed_from_attention_set":{"31280":{"account":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"last_update":"2022-01-06 00:09:24.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-01-06 00:09:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib463a38cb731f1a61c5eee185343304ee991956e","subject":"Pre-existing metadata items, set new share metadata","status":"MERGED","created":"2021-09-16 07:48:44.000000000","updated":"2022-01-14 11:52:29.000000000","submitted":"2022-01-06 00:09:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"809330-manila_01","meta_rev_id":"51854e25692da5fefa26d0fd26dfc50b91fde223","_number":809330,"virtual_id_number":809330,"owner":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-06 00:09:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-01-05 22:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-02 08:45:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-01-05 22:58:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-16 10:03:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-16 12:53:31.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-09-17 02:45:11.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-17 06:16:10.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-21 03:07:50.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2021-09-21 03:07:50.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2021-09-21 03:07:50.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-01-05 22:58:54.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-01-14 08:59:03.000000000","updated_by":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"reviewer":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"state":"CC"}],"messages":[{"id":"a055413f8092e69bfd845763b8fe2307c1c2d03a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-09-16 07:48:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b005d55bea702ea4f87650294051760e39358ddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 10:03:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e961b59103540ad96853ed124ede419 : SUCCESS in 4m 25s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f85a4fca6b914492bf4584a83c45d4df : SUCCESS in 28m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9ad43554346241f0a8df88ccbc6970c6 : SUCCESS in 36m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e93a0a59220646da9a43618e5849d685 : FAILURE in 59m 28s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/dd5b6c69284f49c8b968568ee8f21e70 : SUCCESS in 1h 28m 13s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/37002514413541459094d97d1f538adc : SUCCESS in 1h 06m 17s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/de0c70d95efb4dab8939ae7fce95c5d4 : SUCCESS in 1h 12m 59s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/570648ebb12f48b8a4c425cbeb628197 : SUCCESS in 1h 05m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0477491a47b54d4cae98d75653c9d263 : FAILURE in 1h 07m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/58d1fa19f1a5414bb6c7895c2d8b453b : FAILURE in 1h 29m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c1e62420ac5c4db998e8517c4d68a9bf : SUCCESS in 46m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/58bc4ad6eedb416dae51ef549c52c93b : FAILURE in 2h 07m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0f09a6c5790d46ddaf6f6ac46ed26fc5 : SUCCESS in 25m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4918aa0f85a52125570356aeabf8d720db17647a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-09-16 11:35:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"297728e41440d724604a54b6d2a0cbeddbe9eb7f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-09-16 12:53:31.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/30/809330/2/check/DellEMC_Unity_Manila_dhss_true/38cbd00/DellEMC_Unity_Manila_dhss_true/351 : FAILURE in 3m 36s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/30/809330/2/check/DellEMC_Unity_Manila_dhss_false/6710ca4/DellEMC_Unity_Manila_dhss_false/347 : SUCCESS in 1h 17m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b3ddc9c8266e8f3de2f120b6bc94060218f69edf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 13:22:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c95204ece8d4a3e9593f5453768cf44 : SUCCESS in 4m 12s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e89c3d69b02a40398db80d6a1abd1c5f : SUCCESS in 30m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d8bf980a91824f84adc2ad498b85a1a4 : SUCCESS in 33m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/82e7d2b171d149e3b03800dff21cb290 : SUCCESS in 1h 06m 10s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d2946b350ada430aad6784eb570008d2 : SUCCESS in 1h 01m 07s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/adfcd5f5dfc44c1ebcac8d78b73e662f : FAILURE in 1h 19m 47s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/f655335bb17f4b05beee59d66d6823bf : SUCCESS in 1h 07m 17s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3ccb6076fdd9439e85976a6f4f22d5e1 : FAILURE in 1h 09m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c7af848c83a143cfabe00c425487eb46 : SUCCESS in 1h 26m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/002fe79b395b4db5b3fd685f144c1f67 : FAILURE in 1h 38m 28s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dba538d19cbf4e5995496853a51569d4 : SUCCESS in 46m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d4d98fe300ae49f29d49b415b918ac26 : FAILURE in 1h 43m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0c0d786d36db4116844a31b284913698 : SUCCESS in 28m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d88a2fdd3e657e837bf85f5d1ac4539cf956e61","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-09-17 00:48:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2287c544f237bc8d3ede05b2ca37bf0a929e0ba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 02:44:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1237035b5c11415daf6396f8706425dc : SUCCESS in 3m 54s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e93b9ed18f4c4c8187f5e607a806f0ac : SUCCESS in 41m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/26e6b75f33aa485bb8d428374ae55048 : SUCCESS in 40m 08s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2f16efb42d48431a811dc860607f84a3 : SUCCESS in 1h 22m 58s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/cdc440d0c88c4f0bbcc034786eab565a : SUCCESS in 1h 11m 10s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/48996ae44d414b3caeb289ed1b696b92 : SUCCESS in 1h 06m 24s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/edfd18439671421a9da1a9807caea9ae : SUCCESS in 1h 18m 19s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d99d354880b34681bbda7cb3e7ba583d : SUCCESS in 1h 15m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/472419ae66d04d39b675209ca41043e9 : SUCCESS in 1h 12m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4d4e83963431402cae1bbf2e0d0cc426 : FAILURE in 1h 31m 20s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6fdfea2e2dc74904a3b4b8a88fefc8a9 : SUCCESS in 44m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a8eab87fe82c4cff84747cc9dd8db06f : FAILURE in 1h 51m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6342cfd7e0ff4115ae8c23735f34541c : SUCCESS in 25m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c0f23168d17f35eade0c8a0e079dcd3aa0d726f3","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-09-17 02:45:11.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/30/809330/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/90ec407/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/30/809330/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/acc209d/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"8466dadb46b869d42550fd6e6c064968553e7fbd","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-09-17 03:20:22.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Build failed.\n\u003e \n\u003e - manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/30/809330/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/90ec407/ : NOT_REGISTERED\n\u003e - manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/30/809330/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/acc209d/ : NOT_REGISTERED\n\u003e \n\u003e Comment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI\n\nrun-NetApp CI","accounts_in_message":[],"_revision_number":2},{"id":"57b52a467019b7bb1d466b18ee36342326d0ce59","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-09-17 06:16:10.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/30/809330/2/check/DellEMC_VNX_Manila/1119cf1/DellEMC_VNX_Manila/188 : SUCCESS in 1h 14m 30s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"7ef29d840152b08e5be7a64834339dfde77f43c6","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2021-12-30 14:02:21.000000000","message":"Patch Set 2: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6a55f0fd60c8aa10af3cc93534e055ce19941d80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-12-31 06:13:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0f4565831c97f4ee8a5d91a2c409be1d0f98ea2b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-31 07:22:37.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/30/809330/3/check/DellEMC_Unity_Manila_dhss_true/dfedea2/DellEMC_Unity_Manila_dhss_true/183 : FAILURE in 1h 08m 35s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/30/809330/3/check/DellEMC_Unity_Manila_dhss_false/9f09042/DellEMC_Unity_Manila_dhss_false/176 : FAILURE in 13m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"3858bfb35ae0e04268416952c8bfabd1166538f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-31 08:08:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47ba906972c04bc4a008928b1a7351dd : SUCCESS in 3m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d2b2e504bd904de2a1b83145fad9bc52 : SUCCESS in 32m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4aa0e16dcaaa42539dde98120c15e097 : SUCCESS in 31m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a75d0c2c47ef4be1b9583f5c9744eed8 : SUCCESS in 59m 00s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/bedcbcbadd3d4a7685ec4bf6a19c4aa7 : SUCCESS in 1h 02m 00s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/1c605163c9b24b4ea1b50690d2839e68 : SUCCESS in 47m 51s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/73a1a69c6f4841d08d6044031866199f : SUCCESS in 55m 48s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6971a057b2614dbdbbd1a6e640310688 : FAILURE in 1h 05m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6bb760f1dfb5419188f927e1b279d8c5 : FAILURE in 1h 10m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b68e44c1992843b38a2385f14a6d9468 : FAILURE in 1h 19m 16s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5f855336ed7e4b6ebede1ebefcdff3ca : SUCCESS in 39m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/94722a21e14145b5befb935588f10d80 : FAILURE in 1h 52m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a75fb8b9dfb341158031d953395a7606 : SUCCESS in 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"583d695863acf3668c8947997699f922dd38468e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-31 08:09:09.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/30/809330/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/93872cd/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/30/809330/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/0004d94/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"2019881f4627ce6eb75f5b6ce68504b049d117e6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-12-31 08:25:19.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/30/809330/3/check/DellEMC_Unity_Manila_dhss_true/d5823f6/DellEMC_Unity_Manila_dhss_true/184 : FAILURE in 16m 13s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/30/809330/3/check/DellEMC_Unity_Manila_dhss_false/56f79a1/DellEMC_Unity_Manila_dhss_false/177 : FAILURE in 10m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"fb8d0f0784240bf19d2443370840625773a22dd7","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-12-31 08:58:53.000000000","message":"Patch Set 3:\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0589f078ed9a7a6464cb54989206f418ee5466bb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-12-31 09:35:09.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/30/809330/3/check/DellEMC_VNX_Manila/c10adaf/DellEMC_VNX_Manila/90 : FAILURE in 1h 26m 04s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"e1ed9ecd4a56fea2aff92a5e78848d7db034bd97","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-02 08:45:51.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"23ff0ec7ffeedf8e12217ae504e30a5874ffaa5c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-01-05 22:58:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e7ff4f4adadfb0e82d6cff15551e9cd7b9f0b355","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-05 22:59:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"05f33b1b0a2b658484bb3e1633e987c4c7045e14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 00:09:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d4bfbfd08c4269b6462457d837a364 : SUCCESS in 4m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a2815160095c4ef08c197d97366766d3 : SUCCESS in 30m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0f076fc0736940f8b1e6c8c0e1bdce9b : SUCCESS in 33m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/691944759bd148a4aa4883028a502433 : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":3},{"id":"81e20f52442abebd862b9eae31ab02e284e1237d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 00:09:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d684100e8572bb720273d007c72cd7a421b7bd36","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2022-01-14 08:59:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"51854e25692da5fefa26d0fd26dfc50b91fde223","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-01-14 11:52:29.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/30/809330/3/check/DellEMC_VNX_Manila/e1891f4/DellEMC_VNX_Manila/107 : FAILURE in 2h 53m 08s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"032a8fbeed4684b03de98fcc3c40c8376dc57188","revisions":{"f9804b1e52c1031e963a82616e69078997388a0f":{"kind":"REWORK","_number":1,"created":"2021-09-16 07:48:44.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/30/809330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/809330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/809330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/809330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/809330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/809330/1"}}},"commit":{"parents":[{"commit":"e4a1e88ef36b5867d2e7f5a1383bff1064f93623","subject":"Merge \"Add run tests as thin provisioning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4a1e88ef36b5867d2e7f5a1383bff1064f93623"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-09-16 07:44:14.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-09-16 07:48:39.000000000","tz":480},"subject":"Pre-existing metadata items, set new share metadata","message":"Pre-existing metadata items, set new share metadata\n\nChange-Id: Ib463a38cb731f1a61c5eee185343304ee991956e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f9804b1e52c1031e963a82616e69078997388a0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f9804b1e52c1031e963a82616e69078997388a0f"}]},"branch":"refs/heads/master"},"f823aa18abd4efad2ad2155c856ddb151c3ed116":{"kind":"REWORK","_number":2,"created":"2021-09-16 11:35:50.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/30/809330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/809330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/809330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/809330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/809330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/809330/2"}}},"commit":{"parents":[{"commit":"e4a1e88ef36b5867d2e7f5a1383bff1064f93623","subject":"Merge \"Add run tests as thin provisioning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4a1e88ef36b5867d2e7f5a1383bff1064f93623"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-09-16 07:44:14.000000000","tz":480},"committer":{"name":"lkk","email":"likangkang01@inspur.com","date":"2021-09-16 11:35:42.000000000","tz":0},"subject":"Pre-existing metadata items, set new share metadata","message":"Pre-existing metadata items, set new share metadata\n\nChange-Id: Ib463a38cb731f1a61c5eee185343304ee991956e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f823aa18abd4efad2ad2155c856ddb151c3ed116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f823aa18abd4efad2ad2155c856ddb151c3ed116"}]},"branch":"refs/heads/master"},"032a8fbeed4684b03de98fcc3c40c8376dc57188":{"kind":"REWORK","_number":3,"created":"2021-12-31 06:13:49.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/30/809330/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/809330/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/809330/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/809330/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/809330/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/809330/3"}}},"commit":{"parents":[{"commit":"e4a1e88ef36b5867d2e7f5a1383bff1064f93623","subject":"Merge \"Add run tests as thin provisioning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e4a1e88ef36b5867d2e7f5a1383bff1064f93623"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-09-16 07:44:14.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-12-31 03:19:12.000000000","tz":480},"subject":"Pre-existing metadata items, set new share metadata","message":"Pre-existing metadata items, set new share metadata\n\nAdding new metadata items will not delete pre-existing metadata items, if items already exist, they will be updated with new values.\n\nChange-Id: Ib463a38cb731f1a61c5eee185343304ee991956e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/032a8fbeed4684b03de98fcc3c40c8376dc57188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/032a8fbeed4684b03de98fcc3c40c8376dc57188"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
