)]}'
{"id":"openstack%2Fmanila-tempest-plugin~813463","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Id7860ba14ffb8b64ece5f9c2e13f2f4c96b9b8e8","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2021-11-09 21:38:25.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Id7860ba14ffb8b64ece5f9c2e13f2f4c96b9b8e8","subject":"[DNM] Temporarily disable IPv6 setup on LVM and CEPH NFS","status":"NEW","created":"2021-10-11 16:59:05.000000000","updated":"2021-11-10 00:53:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"11a732a438bb2230b4f83a02f5e3eb0bfbc12ccd","_number":813463,"virtual_id_number":813463,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-10 00:53:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-11 18:45:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-11 18:52:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2021-10-11 22:24:30.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-12 00:13:15.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"b351c9e036ed5d7e5f7b8cdf080b8ad27aa7595f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-10-11 16:59:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89c86261b9e47198e22caaa590fb69bab74e183c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 18:45:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ee342a963b4249903c97c004e91cb4 : SUCCESS in 5m 01s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2695513633ea47aea02c976b95bfc110 : SUCCESS in 29m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fdd780b129c44b29bb704f5e385b3ec6 : SUCCESS in 33m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7220419d4eb64671b9d1a720c41ef08e : FAILURE in 52m 05s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/26ed7f47641d497e9d35c8e051e1bd92 : FAILURE in 43m 58s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/b3a44d800cbb4746ab5d6e33f925b0dd : FAILURE in 42m 24s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/1095f4dd3f5e406b99e1815f2149bb1e : FAILURE in 40m 23s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8a0ca6e8de344953951e9e4d1fc4f68d : SUCCESS in 1h 00m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/68e709cdceb245e5b9cbb7954bed403f : SUCCESS in 1h 10m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fcac9de7cbea44c4973d033666d99ef1 : FAILURE in 50m 07s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bbb03f1d9a0143d89028331bde622ed5 : SUCCESS in 45m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9a9b02274a834928b2a66b1485350e66 : FAILURE in 1h 42m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ae30c523969248ebbce96af58477a2b7 : SUCCESS in 23m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"238d6cae32bbf9c4b942f82fa34dbf9c178376e2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-11 18:52:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/813463/1/check/DellEMC_Unity_Manila_dhss_true/a76a0c6/DellEMC_Unity_Manila_dhss_true/6 : FAILURE in 1h 51m 30s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/813463/1/check/DellEMC_Unity_Manila_dhss_false/1376582/DellEMC_Unity_Manila_dhss_false/6 : SUCCESS in 1h 34m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"b40356f0d890d69ecc5844356c6bd1869395784d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-10-11 18:57:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"826e25e26c3c05615d4f49dd90559ca129ba45c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 20:31:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a985385d6e4a168d8bd8853cd177bf : SUCCESS in 3m 47s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6572dec4738041819092fc497bc6b67e : SUCCESS in 35m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ec3fa036bf6149f6898eb88a202abebd : SUCCESS in 34m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/49da1c6ea3084b76beb0e6d9796ffcf1 : FAILURE in 37m 09s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d02b2ebf0ae74ade953ffeb963db75d4 : FAILURE in 43m 30s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/c17ec1aeda2345bba09325ae94e8f16c : FAILURE in 43m 39s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/56805b1319b94e95b976fbe10465c1f3 : FAILURE in 41m 37s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dfc84000cd0246c399249e7472dfd31c : SUCCESS in 1h 22m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ff6d92de2a1442dea768fae09ed0cbcf : SUCCESS in 1h 11m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4777f61c40a84d24a00e4c76bd360574 : FAILURE in 54m 03s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/30c59e109058495190239886a38acb44 : SUCCESS in 53m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c9259c53ca354622aebac3f1d22f1652 : FAILURE in 1h 31m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9194b060b6754b698898da87fb3c0b58 : SUCCESS in 31m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5fa503fd6dbf7aacb9118511d26ab62f965fb62c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-10-11 20:49:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fe5da2a5c92f59f45bf85f7ef1f9bcefe7fc912d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 22:24:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4efb5f55e16e4a088fba7633e746e8de : SUCCESS in 4m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a4ad0fc264f1435fbc373ef3d47ad8eb : SUCCESS in 34m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0fe229ce392941a7a70b1f2d319a27b9 : SUCCESS in 44m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/eb093f39c8ae4051ab220ed0dfa4b8e3 : SUCCESS in 1h 04m 24s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/4ea17a3af5f3499e875493ea5552dd7b : SUCCESS in 1h 06m 26s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/93ab133615214305b91601e2db78b7ac : SUCCESS in 1h 15m 06s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/f607acbd500e4b64a4534394b53a4330 : SUCCESS in 1h 04m 01s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/818ed743f7334da99b21b8d8aba8bbf4 : SUCCESS in 1h 21m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e0b0b38eb9144e19836e039db265b15d : SUCCESS in 1h 21m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/eb364b0c58014918a2eb1b107309b56d : FAILURE in 54m 01s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/13ad6ce6807e42df85d771765811b59d : SUCCESS in 34m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/149cf78fc7384fe8895bad7dadc3c8a5 : FAILURE in 1h 32m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e921cdbd4bf545e2a176a70519d6bb39 : SUCCESS in 21m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0e4a573dbc0fc53c68f72bcd50dd0fbd6aa7d4e4","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-10-11 22:24:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/63/813463/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/be74687/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/63/813463/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/55daea6/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"9b1472cd1270d55a0f6740a2304577711c809fcf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-11 22:46:07.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/813463/3/check/DellEMC_Unity_Manila_dhss_true/1e24bbb/DellEMC_Unity_Manila_dhss_true/9 : SUCCESS in 1h 56m 12s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/813463/3/check/DellEMC_Unity_Manila_dhss_false/d26d80c/DellEMC_Unity_Manila_dhss_false/10 : SUCCESS in 1h 20m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"f43890e7f7f42f613af4744ac78b0fb6e94092ca","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-10-12 00:13:15.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/63/813463/3/check/DellEMC_VNX_Manila/204f6fe/DellEMC_VNX_Manila/3 : FAILURE in 1h 48m 48s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"29489201ff901d6889bbe357e3662d62ef7165fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-10-26 17:12:44.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"10635c535478c8656801d0479cf238ac9e9e5977","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-26 17:28:11.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/813463/4/check/DellEMC_Unity_Manila_dhss_true/4a5fe07/DellEMC_Unity_Manila_dhss_true/45 : FAILURE in 15m 13s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/813463/4/check/DellEMC_Unity_Manila_dhss_false/59129b3/DellEMC_Unity_Manila_dhss_false/48 : FAILURE in 15m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"8e1764d790b2fcf107d04efec6cba10c10615b67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 20:32:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d093519ad2ff4d4eb493c26bee0096a5 : SUCCESS in 3m 33s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ab0e656a7ab646ae9aa6489e23fe27b3 : SUCCESS in 32m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9220163fbbed4092bc0f5cb9a755f3fc : SUCCESS in 41m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/08a9cc970f5f495c92a02038a74c4055 : FAILURE in 23m 28s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d8f29244696d442896554696ab6e6bf7 : FAILURE in 23m 18s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/08bc20f3d23b4b33bc811579981b0758 : FAILURE in 26m 04s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/ba4690fc253c464c89a173ef6398cb5b : FAILURE in 15m 38s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ab3d9abdcd424e3484023226c352247c : SUCCESS in 1h 15m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/47f4d02a2d6d4c2e9aad17e3ccc8aa98 : SUCCESS in 1h 24m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6d7015a4ef4e4b2ba45762f5dad08bd8 : FAILURE in 23m 12s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/87450ddeb51a456fa0ad1494d00a5aae : SUCCESS in 48m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bfa04292586e4f059a24c25f41b83a37 : TIMED_OUT in 3h 07m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b3d3f6a9f25e441aa3a8bf3f7d9e4081 : SUCCESS in 26m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5cd69e708f34539bde62d39a59a7ec841e4fb60c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-10-26 22:08:14.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"358d7946482fa272aa6133f0451c94acc459eb90","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-26 22:22:48.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/813463/5/check/DellEMC_Unity_Manila_dhss_true/d32f8d2/DellEMC_Unity_Manila_dhss_true/50 : FAILURE in 14m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/813463/5/check/DellEMC_Unity_Manila_dhss_false/4421a3d/DellEMC_Unity_Manila_dhss_false/53 : FAILURE in 12m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"332105d6ad2032387a354695a0fdec5847fee7d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-27 00:04:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00f7de73c1b34eff9b80731bed86b859 : SUCCESS in 5m 17s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d4cb362fb05c4f37913a02a737dfcd5a : SUCCESS in 29m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/741aa53b1712440faaf2700e0b2d4e01 : SUCCESS in 33m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5804d8f1e0fc402e8ad0d75082f269c5 : SUCCESS in 1h 21m 34s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/ad07374fc6e842e69ba415db4721c3b7 : SUCCESS in 1h 05m 46s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/66e3f8b59f2b41d8866b461a10e7c8ba : SUCCESS in 1h 06m 58s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/72b72b6109ab482d8e20509ab49022d3 : FAILURE in 30m 01s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f4d007e988334edaa5aec7212ce9018b : FAILURE in 1h 17m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c6b172f97e6c4cdb8e1e3eb94e6dbf29 : FAILURE in 1h 23m 27s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/59ad4e4d764d4854b54c86a277f0a7b1 : FAILURE in 1h 40m 01s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d692ca4f9b2b402e8717468ad9bb530b : SUCCESS in 43m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/301817a2dcdd489c8b33944471aac6c3 : FAILURE in 1h 53m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ca33eb74c4c84c149eae7f2a120a455b : FAILURE in 22m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"91a2bf1a6108c565387d0a4b29a6388b09d188dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-10-28 21:27:15.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"bc4ec2e7d02d4415e6ab54c18a870fcbdd3b04b0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-28 21:40:54.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/813463/6/check/DellEMC_Unity_Manila_dhss_true/95e7e7e/DellEMC_Unity_Manila_dhss_true/58 : FAILURE in 13m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/813463/6/check/DellEMC_Unity_Manila_dhss_false/ae91f7b/DellEMC_Unity_Manila_dhss_false/61 : FAILURE in 12m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"1f3c5f8c0722b46385c3ebd0ade3bb838445b690","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-10-28 23:02:43.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"a3885848aeb49a1aba317d7a31023ece10f75063","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-28 23:16:20.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/813463/7/check/DellEMC_Unity_Manila_dhss_true/ac33643/DellEMC_Unity_Manila_dhss_true/59 : FAILURE in 13m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/813463/7/check/DellEMC_Unity_Manila_dhss_false/54f470e/DellEMC_Unity_Manila_dhss_false/62 : FAILURE in 12m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"517db4bb11e018fd2807306b2472ec82f2290577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 00:43:54.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4619b6326084ea5ae8f37b33b3ef991 : SUCCESS in 5m 06s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a8ad00efd61648aab59abe7c7788fc36 : SUCCESS in 30m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dbd3ed644be64d4ba8e4bf6b55944a98 : SUCCESS in 35m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7d5af46b4bdc448a83c1be55e81f7404 : SUCCESS in 1h 27m 44s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d0605406e9ed42c68858970613711c1b : SUCCESS in 1h 00m 26s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/acfd8a32298346fc98f0a8b5dd118e84 : SUCCESS in 1h 04m 56s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/72a45d26040449dda59b8774985a91b0 : SUCCESS in 1h 21m 52s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/951081ee3d0d4f5681ee1fcbaa40a173 : FAILURE in 1h 05m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/11759e75457f42d6a3f0dfe154a6b764 : SUCCESS in 1h 28m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bd22f0c7ec4f4fc78d6993fec477181f : FAILURE in 1h 38m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8be5d28421414a89a0a1dbcb6e59dd09 : SUCCESS in 43m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3dc6a350789f493881fd36e7f2b428d6 : FAILURE in 1h 21m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a045161b26944c3980fe88a6cecfd761 : SUCCESS in 23m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fd5c2c82bfd83f58bce916cdfc2178a883192cb1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-10-29 00:44:09.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/63/813463/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/56af69a/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/63/813463/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/ba7df47/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"a7140b43980ad202969da55cd8b59ab7b80d10c1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-10-29 00:56:12.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/63/813463/7/check/DellEMC_VNX_Manila/77867b4/DellEMC_VNX_Manila/27 : FAILURE in 12m 04s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":7},{"id":"16d04ae8010ade1633efcebc79fd11a6b9d0a531","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-10-29 00:58:27.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/813463/7/check/DellEMC_Unity_Manila_dhss_true/ded0ab4/DellEMC_Unity_Manila_dhss_true/60 : FAILURE in 14m 20s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/813463/7/check/DellEMC_Unity_Manila_dhss_false/80055c9/DellEMC_Unity_Manila_dhss_false/63 : FAILURE in 12m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"5f14eab239674c033d4587c5ed125e16979bf3a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-11-09 20:29:14.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"049d750a82316812605d51ec5d6d628208d98f97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 20:30:21.000000000","message":"Patch Set 8: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  Nodeset ubuntu-bionic-2-node already defined\n\nThe error appears in the following nodeset stanza:\n\n  nodeset:\n      name: ubuntu-bionic-2-node\n      nodes:\n        - name: primary\n          label: ubuntu-bionic\n        - name: secondary\n          label: ubuntu-bionic\n      groups:\n        - name: subnodes\n          nodes:\n            - secondary\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/nodesets.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":8},{"id":"811cff11568c0d45b621e0f6bb97a5e73a9db788","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-11-09 20:38:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1c9c1301a21da4acfc6f87dc2504cba8f567e85c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-11-09 20:52:37.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"d47b48b4cc852baa889a92d2e87a1dd5e9e9a2fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 20:53:45.000000000","message":"Patch Set 9: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  Nodeset ubuntu-bionic-2-node already defined\n\nThe error appears in the following nodeset stanza:\n\n  nodeset:\n      name: ubuntu-bionic-2-node\n      nodes:\n        - name: primary\n          label: ubuntu-bionic\n        - name: secondary\n          label: ubuntu-bionic\n      groups:\n        - name: subnodes\n          nodes:\n            - secondary\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/nodesets.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":9},{"id":"a0ffd1bb82eced7feaf2e9f76d81327777ca38a9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-11-09 21:37:24.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"98db9fe96ec369898407600b8f10cc4656495999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 21:38:25.000000000","message":"Patch Set 9:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/openstack-zuul-jobs on branch master.  The error was:\n\n  Nodeset ubuntu-bionic-2-node already defined\n\nThe error appears in the following nodeset stanza:\n\n  nodeset:\n      name: ubuntu-bionic-2-node\n      nodes:\n        - name: primary\n          label: ubuntu-bionic\n        - name: secondary\n          label: ubuntu-bionic\n      groups:\n        - name: subnodes\n          nodes:\n            - secondary\n\n  in \"openstack/openstack-zuul-jobs/zuul.d/nodesets.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":9},{"id":"4dbe64bbc73599635a3a1900e152a073c513d6ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-11-09 21:44:22.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"014558bd813be4ccf560d2591196deceae37744a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 00:53:01.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2121dbbb4bb438d918177a20e95bc4e : SUCCESS in 3m 34s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9f7565c734ee4895af36fd3d57d91e42 : SUCCESS in 28m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b49fae25fd6a4503a271f0dbe691f19b : SUCCESS in 42m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/19f09bb42e83401a9fc30b51de8875a5 : SUCCESS in 1h 10m 42s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/7e6a5fa5de8e44a5917c64d0a9c1487f : SUCCESS in 1h 04m 58s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/cbcc909a7b7c4d56bdf1e78a6244d468 : SUCCESS in 1h 14m 11s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/42a12f0a2c594610a750d1322c55ef04 : SUCCESS in 1h 00m 02s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/01cf5094a68c4a01bd5ba27687a4b492 : SUCCESS in 1h 29m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3bcf72bd381a47b884ed814fd809d972 : FAILURE in 1h 36m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5e87dd0ba3024ddc8f0c9bea6064c979 : FAILURE in 1h 42m 34s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/03bd31ca96d644459f31d55046d1db44 : SUCCESS in 41m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c4e17966b8314176ab0b5d876748be41 : FAILURE in 3h 02m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/44a9a4df80b047158ddaaf07fc2f99c3 : SUCCESS in 27m 15s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"11a732a438bb2230b4f83a02f5e3eb0bfbc12ccd","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-10 00:53:17.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/63/813463/10/upstream-check/manila-cDOT-manila-tempest-plugin-ss/a4d596e/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/63/813463/10/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/15d146f/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"9714e948160046592de6589a03319c6adbba13fa","revisions":{"3fd264991382a89d12a38508e45c2c44d4591840":{"kind":"REWORK","_number":1,"created":"2021-10-11 16:59:05.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/63/813463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/813463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/1"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-11 16:58:00.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-11 16:58:00.000000000","tz":-180},"subject":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS","message":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS\n\nChange-Id: Id7860ba14ffb8b64ece5f9c2e13f2f4c96b9b8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3fd264991382a89d12a38508e45c2c44d4591840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3fd264991382a89d12a38508e45c2c44d4591840"}]},"branch":"refs/heads/master"},"a74f99dcb1ea5df181afad1beae6293246dcb967":{"kind":"REWORK","_number":2,"created":"2021-10-11 18:57:29.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/63/813463/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/813463/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/2"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-11 16:58:00.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-11 18:56:33.000000000","tz":-180},"subject":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS","message":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS\n\nChange-Id: Id7860ba14ffb8b64ece5f9c2e13f2f4c96b9b8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a74f99dcb1ea5df181afad1beae6293246dcb967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a74f99dcb1ea5df181afad1beae6293246dcb967"}]},"branch":"refs/heads/master"},"ea3bb8b8dbeb20750cc3993bbf26170f485a7860":{"kind":"REWORK","_number":3,"created":"2021-10-11 20:49:40.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/63/813463/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/813463/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/3"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-11 16:58:00.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-11 20:49:28.000000000","tz":-180},"subject":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS","message":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS\n\nChange-Id: Id7860ba14ffb8b64ece5f9c2e13f2f4c96b9b8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ea3bb8b8dbeb20750cc3993bbf26170f485a7860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ea3bb8b8dbeb20750cc3993bbf26170f485a7860"}]},"branch":"refs/heads/master"},"53b40788af9f8e38069e0a425cc50a6a1a20247f":{"kind":"REWORK","_number":4,"created":"2021-10-26 17:12:44.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/63/813463/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/813463/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/4"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-11 16:58:00.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2021-10-26 17:12:20.000000000","tz":0},"subject":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS","message":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS\n\nChange-Id: Id7860ba14ffb8b64ece5f9c2e13f2f4c96b9b8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/53b40788af9f8e38069e0a425cc50a6a1a20247f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/53b40788af9f8e38069e0a425cc50a6a1a20247f"}]},"branch":"refs/heads/master"},"dac42f46e889457804bc2df6a72c09e40453a704":{"kind":"REWORK","_number":5,"created":"2021-10-26 22:08:14.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/63/813463/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/813463/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/5"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-11 16:58:00.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2021-10-26 22:08:14.000000000","tz":0},"subject":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS","message":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS\n\nChange-Id: Id7860ba14ffb8b64ece5f9c2e13f2f4c96b9b8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dac42f46e889457804bc2df6a72c09e40453a704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dac42f46e889457804bc2df6a72c09e40453a704"}]},"branch":"refs/heads/master"},"cd878cbd13247535c7d7d4dc7b7d577a1a80f773":{"kind":"REWORK","_number":6,"created":"2021-10-28 21:27:15.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/63/813463/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/813463/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/6"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-11 16:58:00.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2021-10-28 21:27:07.000000000","tz":0},"subject":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS","message":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS\n\nChange-Id: Id7860ba14ffb8b64ece5f9c2e13f2f4c96b9b8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd878cbd13247535c7d7d4dc7b7d577a1a80f773"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cd878cbd13247535c7d7d4dc7b7d577a1a80f773"}]},"branch":"refs/heads/master"},"8b222d30d4e55dc820d2dc556cb28a55ce485b7f":{"kind":"REWORK","_number":7,"created":"2021-10-28 23:02:43.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/63/813463/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/813463/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/7"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-11 16:58:00.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2021-10-28 23:02:38.000000000","tz":0},"subject":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS","message":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS\n\nChange-Id: Id7860ba14ffb8b64ece5f9c2e13f2f4c96b9b8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8b222d30d4e55dc820d2dc556cb28a55ce485b7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8b222d30d4e55dc820d2dc556cb28a55ce485b7f"}]},"branch":"refs/heads/master"},"f53ab96d7444d5fc6a254dd5ac19aa9535870d9a":{"kind":"REWORK","_number":8,"created":"2021-11-09 20:29:14.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/63/813463/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/813463/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/8"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-11 16:58:00.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2021-11-09 20:29:14.000000000","tz":0},"subject":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS","message":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS\n\nChange-Id: Id7860ba14ffb8b64ece5f9c2e13f2f4c96b9b8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f53ab96d7444d5fc6a254dd5ac19aa9535870d9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f53ab96d7444d5fc6a254dd5ac19aa9535870d9a"}]},"branch":"refs/heads/master"},"b4ceadc109341870fabc1b94023e1b8950672d02":{"kind":"REWORK","_number":9,"created":"2021-11-09 20:52:37.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/63/813463/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/813463/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/9"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-11 16:58:00.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2021-11-09 20:52:37.000000000","tz":0},"subject":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS","message":"DNM - Temporarily disable IPv6 setup on LVM and CEPH NFS\n\nChange-Id: Id7860ba14ffb8b64ece5f9c2e13f2f4c96b9b8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b4ceadc109341870fabc1b94023e1b8950672d02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b4ceadc109341870fabc1b94023e1b8950672d02"}]},"branch":"refs/heads/master"},"9714e948160046592de6589a03319c6adbba13fa":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2021-11-09 21:44:22.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/63/813463/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/813463/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/813463/10"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-11 16:58:00.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-11-09 21:44:02.000000000","tz":-180},"subject":"[DNM] Temporarily disable IPv6 setup on LVM and CEPH NFS","message":"[DNM] Temporarily disable IPv6 setup on LVM and CEPH NFS\n\nChange-Id: Id7860ba14ffb8b64ece5f9c2e13f2f4c96b9b8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9714e948160046592de6589a03319c6adbba13fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9714e948160046592de6589a03319c6adbba13fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
