)]}'
{"id":"openstack%2Fmanila-tempest-plugin~815965","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I19ed529a1b55c75e8f48a52edc9095be97a485a5","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-03-02 20:06:06.000000000","reason":"Change was submitted"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-03-02 18:59:23.000000000","reason":"Douglas Viroel replied on the change"},"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2022-02-25 17:26:51.000000000","reason":"Ashley Rodriguez replied on the change"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-02-23 19:07:39.000000000","reason":"Goutham Pacha Ravi replied on the change"}},"hashtags":[],"change_id":"I19ed529a1b55c75e8f48a52edc9095be97a485a5","subject":"Add support to chose ssh key type","status":"MERGED","created":"2021-10-29 14:10:05.000000000","updated":"2022-03-16 19:25:45.000000000","submitted":"2022-03-02 20:06:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815965","meta_rev_id":"c741a5f40923de950a39bc62853ec4415c189abb","_number":815965,"virtual_id_number":815965,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 20:06:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2022-03-02 18:59:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2022-02-25 17:26:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":2,"date":"2022-02-23 19:07:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2022-03-02 18:59:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}],"CC":[{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-29 15:44:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-11 04:10:39.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-22 13:09:55.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-02-22 13:09:55.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2022-02-22 13:09:55.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"3513c7ec4f62e733d3f30c70741679fd96abce6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-10-29 14:10:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ef6c4051bfdb86577e895fb5f824e3a3e409cd3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 15:44:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ff3d26cb7744b0d943629142dbf5805 : SUCCESS in 4m 41s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c04dd00ba4c142f485d06029ccf9e7e6 : SUCCESS in 30m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fb4592131ef3412eb988932d8247fc82 : SUCCESS in 33m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7c4d150e7902437794f94967b711fa4e : FAILURE in 55m 30s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/445b853b86324e62b0f9b9894b008549 : FAILURE in 16m 32s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/97476879f7304a0783bf6639f2cef172 : FAILURE in 16m 23s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/522016ec99cf4a158be7a6d5abafdc42 : FAILURE in 16m 51s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b50b56e7eb914787875a75e846d15dcf : FAILURE in 59m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/33edae8e32c74306b1211ac1390b2040 : FAILURE in 1h 02m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1f7c55b1305b40948412c67fb8423b20 : FAILURE in 1h 31m 04s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e01dd7043ccb40d2b85eeb6cee4accd6 : SUCCESS in 42m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3a4ca9e9ff9744dcb955bc1357cbf79c : FAILURE in 1h 30m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a6299f69c39b457abf06abbe01b885ab : SUCCESS in 22m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fdd92c34ec3978f0551ba2f5105222c336ceb983","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-11-17 13:01:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7917b2909da8987ed633110e7540a435aea6313d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 14:52:54.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"b7dac279bfdb2dac9f428a9d7a20a02c3781a5dc","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-04 18:17:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a024a8115bc04e4947eb5a585d115f8e8710a7c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-04 20:34:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2efdef26a7714bbbb4ac91cc1f3d7687 : SUCCESS in 3m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2aa409af0b9f4be390fea76d94e7d748 : SUCCESS in 29m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e57f83077bba435b96116bb7fce283ec : SUCCESS in 36m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7acfdecc1bfb458dbe2aa7700e86da2c : SUCCESS in 1h 14m 40s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/81d77182fff746fdb7f37e93bd3a84ad : SUCCESS in 1h 05m 26s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a4754c0e76954a939645f29e6b10e4cb : SUCCESS in 1h 21m 56s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/5e31ed43b76942af84f7d0d42bd7837c : SUCCESS in 1h 09m 12s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d6d5d2896de7456093f2e0e072b2c973 : FAILURE in 1h 07m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9f0855be7b034d4d93fb2a60fdfe64bd : FAILURE in 1h 26m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/10f07ba4e65d46959502cafb3dacefca : FAILURE in 1h 39m 07s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/18c6e1d327794403b7c45ad52a8164cb : SUCCESS in 52m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e38d1b7f31f74affb89975fba17d74cf : FAILURE in 2h 08m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9277151a711c4f9aba4dcc164ce3529d : SUCCESS in 27m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fbd6f75b19034b0fecc76067f57a135589acf7e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-08 14:59:52.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1d03ab577d8dad418ad0dc14d1424fd5cefc59d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 17:16:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c01b325a174f41f99cab0ff31fa21f10 : SUCCESS in 5m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5e5f137e79ef4a6385b35b421b10436a : SUCCESS in 28m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5122ca2265434404a7d6570fbcec53c4 : SUCCESS in 44m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e231b3ac0f424f8590168e656148fff2 : SUCCESS in 1h 22m 26s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3f7a3ddd4dcf470aad05e4e0eafac535 : SUCCESS in 1h 18m 16s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/03e74c2d16404bbfbed7288c06ea8be5 : SUCCESS in 1h 13m 32s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/f27d435af5ab49979af98b5551eb69b9 : FAILURE in 1h 12m 25s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f096842354db4b3e81b29ca40a1a680b : SUCCESS in 1h 16m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2e384ecb58124beba68bfc83061fc7b2 : FAILURE in 1h 40m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2a32dc26d6974b6aaa394db7e10549b4 : FAILURE in 1h 44m 35s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b04ae51335b047a5a4d092282c9ab85a : SUCCESS in 46m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ee75175c05b841fd96abc08e8df769b8 : FAILURE in 2h 08m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0be9cfbedb0a439fb0b9022c5b7f4f41 : FAILURE in 21m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a9e17bfa1bf149125f2783a0c1368d73490d0f20","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-10 23:07:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eb2270f5f8841046960d35dd8f13c2dab8ebd371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 01:14:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72792ac1b6cb4de4add3eff1f32c74cf : SUCCESS in 3m 43s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4f992c220ed448e1a02949520bd13c9d : SUCCESS in 27m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/94b6351c80354ea7915e62a153079395 : SUCCESS in 33m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/da775ce4b754403cafa54a7e9f88e741 : SUCCESS in 1h 01m 29s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c14f9edb130f401fbca9168e2d5d36f4 : SUCCESS in 1h 01m 35s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/4706e9469db242a9b050bccb630df563 : SUCCESS in 57m 09s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/54d4fd2ee8144b81afa2f3339f21eda3 : SUCCESS in 51m 40s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0945d76cab054102901eb8103e71a79e : SUCCESS in 1h 08m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6c732e6587a643aba4339f83ac9aaa8d : SUCCESS in 1h 33m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5e63a21cd5354a7999dd84d4e7e0ce6d : FAILURE in 1h 40m 40s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c0a8351aac044c06840338b597282fc1 : SUCCESS in 54m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f9916311673a401f8a687c597e354048 : FAILURE in 2h 04m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/839f67845a0248739f187cb1507dd118 : SUCCESS in 32m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b7f0af46b968f93e0cb14987f25f9c899effc70e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-11 04:10:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/815965/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f8e1339/ : SUCCESS in 2h 55m 36s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/815965/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/3f6c5dd/ : SUCCESS in 2h 06m 43s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"51f7564b12a2353047215d3faf2f9b6123ffb348","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-22 13:09:38.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"3abea3f186db4d2035e06c1f8a0c8328556158ae","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-02-22 13:22:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"54060d8567d1ae7c28a007b51d58dc4aa0016cbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 15:14:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85ff389fd15c430bac73016f73dcee7c : SUCCESS in 5m 04s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ebef7428f75f44d3b478646e0ac4150c : SUCCESS in 34m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d5abf54789074f47b2024db79c51c2df : SUCCESS in 31m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/12f83a03d6b74ca1937624ebb147559e : SUCCESS in 1h 13m 55s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/96ceb60d6cdc4f29b58b8a35d2a6c9ed : SUCCESS in 1h 03m 05s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d9ad76fdbad44e1aa5bb9fc8cc7f23a6 : SUCCESS in 1h 13m 03s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/b7d6bc77371a4ad1be58889aef4872ab : SUCCESS in 1h 23m 01s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/548d1a0555874521ab76a049d7da97fc : SUCCESS in 1h 26m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b98b4d341c3545629fe7f9900b979f8a : SUCCESS in 1h 24m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2e0967e800ee403e90169b7634fccfe3 : FAILURE in 1h 57m 26s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1481a814b0d64c99aa88870c7aea4403 : SUCCESS in 53m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6d1f8e02260c44b7a79a8361b8fca218 : FAILURE in 2h 00m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5cf4fb21afe84130bb1fc1518b334ea1 : SUCCESS in 23m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"da22902455cd15f4a413311ed91934f6417846f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-22 20:24:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bc1b85292dfa87386a6bb7de5f509ae249c8be97","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-22 20:24:23.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a7c0aac5dc6defbdddb9bf285cee0032490eb6e2","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-02-22 21:21:29.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e5096126981bd65a01b3e93e7ad52e6581a9529a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-22 21:26:32.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"02cb65e197e501d9cdd052c45cec0d1631e8942f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 22:19:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b693f402beaf48ddbceae26bfcd2e287 : SUCCESS in 3m 59s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5f547975769d4b488f20f91bc2cf5359 : SUCCESS in 27m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/be74ee21af04460e857affc772565e92 : SUCCESS in 32m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bcee34696b0348ecbbe5ff73bcccd130 : FAILURE in 25m 29s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a2c9104de21d43039b84e463839d884f : SUCCESS in 59m 56s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/2b5de1a37e4d4934998e7143c8923756 : SUCCESS in 59m 57s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/57994f408e2245ee8f79d52db4f61569 : SUCCESS in 1h 22m 56s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/282c2f9ea11648a68e1ee1ae61e98e91 : FAILURE in 1h 40m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5a7c3f1c2c8e44c9b741b7c8079908fc : SUCCESS in 1h 40m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7627dec77ca3460eb5322a34253f0fa7 : FAILURE in 1h 48m 41s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6e8a0728f62040fcb22e98bc7aecfcef : SUCCESS in 1h 03m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/354ee11f5e844f8fb3bf6613984fd2ac : FAILURE in 1h 53m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/aa22a6f3cf194c0bac124db2b8984188 : SUCCESS in 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d07866ecf964105898758409f81b574f4f92fe0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-23 15:04:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2196dffd15cb5837fbb3250d05efde42473a5fbf","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-23 15:05:01.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2558b5b9d32d9cf709f812a0a81b4e0367022d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 17:00:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3f4d67b173441f18a7e14ea5669abac : SUCCESS in 3m 34s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/28c2b19f0c6545ddaa3f435b4177d9bb : SUCCESS in 33m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/96f2f8ccd4ce4c3480ac3dd08b79c598 : SUCCESS in 44m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/96b6fee000124fc08dfcf2c22cb446e4 : SUCCESS in 1h 31m 49s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f00c7b3e47a142cf841318022674cd95 : SUCCESS in 1h 11m 02s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/0700146281a94f0bbd2e3fe164976c37 : SUCCESS in 1h 15m 01s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/0304905a8782439985220eb90fa09b65 : SUCCESS in 1h 18m 29s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2cb147e48ab748cf9d2dcc7253a2df01 : SUCCESS in 1h 22m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ae9495432b6b46fdaef5e7a38946e44e : FAILURE in 1h 48m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1450814a4a684b05a45e05b77680323f : FAILURE in 1h 49m 39s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3d41694b1a014c1fa7b7d4474b606bfd : SUCCESS in 55m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/75f9e8c7d6ed46b78a3886a646b582ac : FAILURE in 1h 33m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c62770e93c8642d8ba1f77a92008bd93 : SUCCESS in 31m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0009f5effd6841d9dc9e51f5041c7718cce1eb5e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-23 19:07:39.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"54d26b9865bd0d7b658cbab26f92d626b8975f56","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-02-25 17:26:51.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b2199334617648a578f72cfaa87f650f84c56db9","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-03-02 18:59:23.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e8cc9afb06dfa4d8213569d3ae22f4ca85b7ab23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 19:00:17.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"978da888087ea72f61059106e533cc750a115d28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 20:06:06.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaa9a9c8d66d488e98207f4ca65b0927 : SUCCESS in 5m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/380de34ff4164505ac6a536fbcd589c6 : SUCCESS in 28m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/07fa15e9817d4cbcb89a36b7d609114b : SUCCESS in 30m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0fdd2e9da2014ba0a9d8ae12d3e8c2ff : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":6},{"id":"6321d96f6ebbb14296dd0df9edf10e53088b6bcf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 20:06:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"c741a5f40923de950a39bc62853ec4415c189abb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-16 19:25:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a894bae49a7d911d448e30a3ef8c9fd9d748e0f7","revisions":{"7a27615dd5094594a4d2042dd995e7dde54f5d95":{"kind":"REWORK","_number":1,"created":"2021-10-29 14:10:05.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/65/815965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/815965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/1"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-29 14:07:59.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-29 14:07:59.000000000","tz":-180},"subject":"WIP/DNM Add support to chose ssh key type","message":"WIP/DNM Add support to chose ssh key type\n\nDepends-On: I891fd8131623a927fef637d4acac574a8fb18c10\nChange-Id: I19ed529a1b55c75e8f48a52edc9095be97a485a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7a27615dd5094594a4d2042dd995e7dde54f5d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7a27615dd5094594a4d2042dd995e7dde54f5d95"}]},"branch":"refs/heads/master"},"48cd24fd188d07ac673c5023eb9088cccfa04783":{"kind":"REWORK","_number":2,"created":"2021-11-17 13:01:42.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/65/815965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/815965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/2"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-29 14:07:59.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2021-11-17 13:01:18.000000000","tz":0},"subject":"WIP/DNM Add support to chose ssh key type","message":"WIP/DNM Add support to chose ssh key type\n\nDepends-On: I891fd8131623a927fef637d4acac574a8fb18c10\nChange-Id: I19ed529a1b55c75e8f48a52edc9095be97a485a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/48cd24fd188d07ac673c5023eb9088cccfa04783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/48cd24fd188d07ac673c5023eb9088cccfa04783"}]},"branch":"refs/heads/master"},"0ecf39cf0c1f8cadfad350a59a40e91f46f6a9c5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-02-08 14:59:52.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/65/815965/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/815965/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/3"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-29 14:07:59.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2022-02-08 14:59:52.000000000","tz":0},"subject":"WIP/DNM Add support to chose ssh key type","message":"WIP/DNM Add support to chose ssh key type\n\nChange-Id: I19ed529a1b55c75e8f48a52edc9095be97a485a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ecf39cf0c1f8cadfad350a59a40e91f46f6a9c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ecf39cf0c1f8cadfad350a59a40e91f46f6a9c5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"19f9d82e2f56f7b929e4030905abd2599c7e010d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-02-22 13:09:38.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/65/815965/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/815965/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/4"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-29 14:07:59.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2022-02-22 13:09:38.000000000","tz":0},"subject":"Add support to chose ssh key type","message":"Add support to chose ssh key type\n\nTempest will always use rsa keys. This change adds\nthe possibility to the admins to set their preferred\nssh key type via config file.\n\nChange-Id: I19ed529a1b55c75e8f48a52edc9095be97a485a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/19f9d82e2f56f7b929e4030905abd2599c7e010d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/19f9d82e2f56f7b929e4030905abd2599c7e010d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"854f6dd395c9cdd970893233aea6a6c6fdb15b8c":{"kind":"REWORK","_number":5,"created":"2022-02-22 20:24:04.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/65/815965/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/815965/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/5"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-29 14:07:59.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-02-22 20:23:52.000000000","tz":-180},"subject":"Add support to chose ssh key type","message":"Add support to chose ssh key type\n\nTempest will always use rsa keys. This change adds\nthe possibility to the admins to set their preferred\nssh key type via config file.\n\nChange-Id: I19ed529a1b55c75e8f48a52edc9095be97a485a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/854f6dd395c9cdd970893233aea6a6c6fdb15b8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/854f6dd395c9cdd970893233aea6a6c6fdb15b8c"}]},"branch":"refs/heads/master"},"a894bae49a7d911d448e30a3ef8c9fd9d748e0f7":{"kind":"REWORK","_number":6,"created":"2022-02-23 15:04:39.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/65/815965/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/815965/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/815965/6"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-10-29 14:07:59.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-02-23 15:04:26.000000000","tz":-180},"subject":"Add support to chose ssh key type","message":"Add support to chose ssh key type\n\nTempest will always use rsa keys. This change adds\nthe possibility to the admins to set their preferred\nssh key type via config file.\n\nChange-Id: I19ed529a1b55c75e8f48a52edc9095be97a485a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a894bae49a7d911d448e30a3ef8c9fd9d748e0f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a894bae49a7d911d448e30a3ef8c9fd9d748e0f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
