)]}'
{"id":"openstack%2Fmanila-tempest-plugin~818134","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I14768075a2ea60781e03e979b8f015ac0d7abd64","project":"openstack/manila-tempest-plugin","branch":"master","hashtags":[],"change_id":"I14768075a2ea60781e03e979b8f015ac0d7abd64","subject":"[WIP] Increase SSH exec timeout in manila jobs","status":"NEW","created":"2021-11-16 19:12:10.000000000","updated":"2021-11-16 23:55:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"955455b6de9a79828f4b035edeb344f372da0c0f","_number":818134,"virtual_id_number":818134,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-11-16 21:39:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-16 21:39:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-16 21:39:42.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-16 23:55:30.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"3f909561f2859bebb90d7a78dca399c0d9641f67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-11-16 19:12:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"589766e7a9a3562bca51ba870f273a9fd921c61b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 21:39:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d23c96504024abfb3b31a742f23abf4 : SUCCESS in 5m 12s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e752bfdc2c394b3f98d46cfd1ead47cc : SUCCESS in 28m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/26ed95631cbe447ba4786c962cb8b774 : SUCCESS in 37m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9df4ab9d4f0642f8a7a6a77974246098 : SUCCESS in 1h 07m 16s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/44525b4b952943ad99e0cc472d05ae07 : SUCCESS in 1h 06m 08s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/033eef7ae6b64021a73ea14665faa552 : SUCCESS in 1h 07m 28s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/a1975b273ddf4bf49c7c88325670252b : SUCCESS in 1h 13m 42s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8f237e8eb0d94ff7861f379e32128799 : FAILURE in 1h 14m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b9086b9cd0a945ffb2d5ef30d3403c77 : FAILURE in 1h 22m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c1a82d0537894b63a85b67c985001ce1 : FAILURE in 1h 55m 05s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3b94d40a7d1649b785341a54d8b70650 : SUCCESS in 48m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b5d3b8c87ddf4629afb4de9526b919fd : FAILURE in 2h 16m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5ce87277ddc347588547b624cfb4f54f : SUCCESS in 26m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d11cd8545bea13595015dcc7fb2064e11747653","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-11-16 21:39:42.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/34/818134/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2f49af9/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/34/818134/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2ab4bed/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"955455b6de9a79828f4b035edeb344f372da0c0f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2021-11-16 23:55:30.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/34/818134/1/check/DellEMC_VNX_Manila/46ad6e5/DellEMC_VNX_Manila/41 : SUCCESS in 2h 15m 53s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22f33cd846ed34ec95a27f1cd1d08024b9623f5b","revisions":{"22f33cd846ed34ec95a27f1cd1d08024b9623f5b":{"kind":"REWORK","_number":1,"created":"2021-11-16 19:12:10.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/34/818134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/818134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/818134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/818134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/818134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/818134/1"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-11-16 19:10:19.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2021-11-16 19:10:19.000000000","tz":-180},"subject":"[WIP] Increase SSH exec timeout in manila jobs","message":"[WIP] Increase SSH exec timeout in manila jobs\n\nSome jobs may fail depending on the nodes they run uppon. In order\nto mitigate this issue, the timeout could be increased and we\u0027d\nhave a lower probability of those failures caused by timeouts\nwhile running commands.\n\nChange-Id: I14768075a2ea60781e03e979b8f015ac0d7abd64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/22f33cd846ed34ec95a27f1cd1d08024b9623f5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/22f33cd846ed34ec95a27f1cd1d08024b9623f5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
