)]}'
{"id":"openstack%2Fmanila-tempest-plugin~818496","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I6133f51f1ec2d5a12141efc119e48ee283fd2dd2","project":"openstack/manila-tempest-plugin","branch":"master","topic":"manila_03","attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-01-07 00:54:37.000000000","reason":"lkk replied on the change"}},"removed_from_attention_set":{"31280":{"account":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"last_update":"2022-01-07 00:54:37.000000000","reason":"lkk replied on the change"}},"hashtags":[],"change_id":"I6133f51f1ec2d5a12141efc119e48ee283fd2dd2","subject":"Maximum number of share replicas allowed from V2.53","status":"NEW","created":"2021-11-19 01:11:24.000000000","updated":"2022-01-07 00:54:37.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8350f5d75479e45329f48d931ba605065f1407c4","_number":818496,"virtual_id_number":818496,"owner":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-06 13:36:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 02:43:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 03:29:33.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-01-05 23:51:21.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2022-01-06 13:36:33.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"0118f35ef0d59684be0f13d2262578e94ab2b859","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-11-19 01:11:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d36b107504d409dcb51a45619fa7607a3f654353","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 02:43:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd998624c294947ae83ed52a91c6f2d : SUCCESS in 3m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/93887053dc694c59ad3070bc88622917 : FAILURE in 21m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/241385a1886f4c2daceeef38bee5b692 : FAILURE in 21m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3691dbdcc1734ea3983cc1669992efbf : FAILURE in 41m 47s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/38d8b56e125e405d897a1c303c99d6dc : SUCCESS in 1h 25m 17s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/004040935f784ba780a40a4edfe2fe73 : SUCCESS in 1h 19m 19s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/ed59a9b1403947439fa6ed340a168aa3 : SUCCESS in 1h 18m 25s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c2233aa5b84246708617e5c99cd6ee02 : FAILURE in 35m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8b8fb6813edf4c53a999f30f1b00f0b9 : FAILURE in 37m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/558ffe26087548c28f772633985f26fa : FAILURE in 42m 47s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/69e9703a6ab84c8c9f4fe6530cea6cfa : FAILURE in 31m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/443355d99f544ff9a561f6b5c277ce87 : FAILURE in 38m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/dcc6d103802949deb20b33da24eb763b : FAILURE in 21m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7eb909153d1cf3395081e1da41a94bc65937009","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-11-19 03:18:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58ec94ad6f2878f1c850211d3fbd76d98558b285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 04:45:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7e20727105f42d4a5be2e27829a18d3 : SUCCESS in 3m 15s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b87c6085b1e2451b9623cb72fce607b6 : FAILURE in 22m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/039ba3193bf24c1396938d29137bf4ac : FAILURE in 15m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a9d53b41a51c453484b7eeb8d8d5e1d8 : FAILURE in 38m 31s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/aff7965d311b4ed2a862cf62db794d73 : SUCCESS in 1h 00m 00s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/2879ec5c351d4362b7accd585f2e0d01 : SUCCESS in 1h 19m 49s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/1239b5fc7a454dfb86276088c08103c4 : SUCCESS in 1h 22m 39s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2e403b1d85d94632a934e82e04dbc599 : FAILURE in 32m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4b5a1fcc20bb4f6eb76adc7f70146a11 : FAILURE in 37m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/020367e6a9b14188810ae1b8fe39e1a1 : FAILURE in 35m 11s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/25f1c5867d2e408e9ceade86441f3a3b : FAILURE in 33m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5bab277ed9c14406880baf5e0851da3c : FAILURE in 34m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0dd9d4234661487bb9674d87d5340b15 : FAILURE in 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c94023c4c1d4a58e6c5d875edb13081195f88487","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-11-22 00:49:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c301966fcc9664d5be85a21feaf73d396b4ea78e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 02:35:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d580e2f173e4604aee33f708e1ec302 : SUCCESS in 3m 08s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/545638e4bef3410d9ffd5e4f7e051311 : FAILURE in 23m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e837fdcab05244f4a74f50229d28e447 : FAILURE in 13m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d4c261e212bf4734b269150eb90d71f4 : FAILURE in 33m 54s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3d6645c9d6e447848a3467cc77497bed : SUCCESS in 1h 20m 57s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/e3fea0f8250f43ea9bf7473da5b9fb35 : SUCCESS in 1h 01m 53s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/9e773fdb7ba8462dbd936691881a88b2 : SUCCESS in 1h 45m 17s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/39a76a66c13b44bca9ae9b9389b0709d : FAILURE in 46m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7c62aa6b55354119b9d9b2cfe9cfc395 : FAILURE in 31m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/276e24b3becc436c965d037ab7defa1e : FAILURE in 45m 12s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/060c70632d86457281682a6a363d5ea6 : FAILURE in 25m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2e63a8dff1a4475cbe9c70bec0fe40d2 : FAILURE in 25m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f3c846100f1e4443911cea5f3d691063 : FAILURE in 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"930b442d3693c2703af713b456c96381079c7bb0","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-11-23 01:20:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"897f897574f6caeb1b3b99a896734d05a5e131cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 03:47:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/155a6c53f3be41009d716d8f1b222e89 : SUCCESS in 3m 21s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f46c07a19a1c40dc921a96689f2e6df8 : FAILURE in 23m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/66e7a0854ef74ba98a1ae2dff77b2b81 : FAILURE in 39m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1000672106ce4184853b7af2eece69e9 : SUCCESS in 1h 19m 12s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/73687e89624f4d13b9af5b7e682047c4 : SUCCESS in 52m 53s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/5d7b8ad981b94b9884b58968ea991651 : SUCCESS in 1h 00m 32s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/36900d132c644c13871e1aa496d080f4 : SUCCESS in 53m 30s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4b99cfb0ff424ac7953e2a29244f7379 : FAILURE in 1h 42m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cb998c64393c40a38b5add3be3d9b58f : FAILURE in 1h 33m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ecafa7ddc4ba4a51a1355781969efa73 : FAILURE in 2h 03m 56s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c8df4fb11c78488397f809d9b2c577b4 : FAILURE in 33m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e0884ff8345c4a17aacb389fa0e2a477 : FAILURE in 2h 22m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/63767d8e51b64458bf93ff0426979d9b : FAILURE in 17m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59163cbe12778fb43c1ea932766de75f67557e74","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-11-24 02:28:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"839a03f88b7c570f4bd361c69198afdd68371b9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 04:36:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb2d046114054415ba30ef562bc983ab : SUCCESS in 5m 20s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9a172e9583834e638650cef8a223290f : FAILURE in 31m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/564c12f10ae848e68ee3e36d637efe8d : FAILURE in 37m 24s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0194c5061a2a4d77ac17b43f6be8ffa9 : SUCCESS in 1h 21m 45s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/ffe7ef63207b475c8ca213dc582a281a : SUCCESS in 1h 08m 07s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/c882afa2ed31402bb244bf0947aa45fd : SUCCESS in 1h 07m 55s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/6b7ff0d6657c48bcbd238725f172ecc5 : SUCCESS in 1h 02m 55s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/75ff406a1c2f48efbffb6b22aa7cec09 : SUCCESS in 1h 25m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1d26d85e46d9404692821f422e71e90c : FAILURE in 1h 22m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/50f7a197eb4949f1b973ba12b50f92e5 : FAILURE in 1h 56m 55s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/48c24703e05f4826a40f730646b6a0c7 : FAILURE in 50m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aeb1e21d8301463881ff078464417c8b : FAILURE in 1h 46m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/00b94af601cb496ab24d532291aade45 : FAILURE in 22m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6fa72d5215a1a2896c6063e88f61e48178a897a","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-11-24 10:15:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b272b65b2541551596a2ec532d4df8eca3f1aad9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 12:34:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6cb65694b54a3da1d1f15612b7ba34 : SUCCESS in 3m 36s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a578b82b7726405cb164e3d59e288cf8 : FAILURE in 26m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9a6761a75b104c35882d06331ff27d09 : FAILURE in 30m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b17f53600e8f46ba9508c4a5dc3bfddf : SUCCESS in 1h 37m 15s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/52978684a4c74dc2b923ce38ec74e3bc : SUCCESS in 1h 09m 27s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/b561cca4a94d4cd39ac937d9bf74897f : SUCCESS in 1h 14m 09s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/05682aef1609421b8efed76504ac044e : SUCCESS in 1h 04m 34s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/22d59943cbef441a9eb2bc7181fb7212 : SUCCESS in 1h 20m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3f6a0170725c46c088dc9fa3d09c8ca0 : FAILURE in 1h 20m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f2114c4777f9403cafcf365d8a45b025 : FAILURE in 2h 12m 09s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/13437d7c19f74ce7a03c2f6628697d7c : FAILURE in 49m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/689e933826a54413bead33873fa077d8 : FAILURE in 2h 03m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fce2aa44dd174194b7ba4ed9b491461d : FAILURE in 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"36e5fa1ff56debc2b405877288f5b77d302f16c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-11-25 02:35:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5965aa5b024d9d947fba193a638abdbd8725c916","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2021-11-25 03:29:33.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/96/818496/2/check/DellEMC_Unity_Manila_dhss_true/0cebc68/DellEMC_Unity_Manila_dhss_true/100 : FAILURE in 15m 26s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/96/818496/2/check/DellEMC_Unity_Manila_dhss_false/7ea1759/DellEMC_Unity_Manila_dhss_false/101 : FAILURE in 32m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"6829f0a93c29a2a0142c5dac9770fa7d25fefb12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 05:05:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46e911cda44b4ded9b8452de814ef8d4 : SUCCESS in 5m 23s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d249c995b52d46748db3750501b904b7 : FAILURE in 40m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6382d929caf449ceb60e06d7a420de61 : FAILURE in 30m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2bcb6297be074040a6b8ed40fa224a05 : SUCCESS in 1h 17m 38s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/84070c6ead0249c2adfab7fd155b926d : SUCCESS in 1h 41m 15s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/4f122c5232794e4ebbb50fccae31527f : SUCCESS in 1h 15m 07s\n- manila-tempest-plugin-lvm-ussuri https://zuul.opendev.org/t/openstack/build/4cbd3dce77d44f83abaa5a492c78f763 : SUCCESS in 1h 00m 06s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/26bce30a86704e309d5acbe1f726cd7d : SUCCESS in 1h 05m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d82bb2cb0174494f9a26c64cd410a713 : FAILURE in 1h 28m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c559b66b7f944ac89cebed9bfdea7ec5 : FAILURE in 1h 38m 34s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e1800243ee2d4f5e92da836ab5f11861 : FAILURE in 44m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3ed152aa29f742c4ab111812cb10364f : FAILURE in 2h 12m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/df4648997aa24a5d957118fce0afd9ca : FAILURE in 25m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7fbbec4f452388d398964dc116ea9cd2ff53fc02","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-11-25 10:18:46.000000000","message":"Abandoned\n\nabandon","accounts_in_message":[],"_revision_number":2},{"id":"fd931f3178e1bbfe8d52b7d725bab0010614aced","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-01-05 23:51:21.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fecc92373528cca724948a1e8eee642b64381236","tag":"autogenerated:gerrit:restore","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-06 11:25:33.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"59f0e67a2b48e765b1302d30e277b2ae430a4271","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-06 11:26:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa8008425acce9a7c3bd034386a50c54e81784e5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-06 12:11:09.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/96/818496/3/check/DellEMC_Unity_Manila_dhss_true/b305cae/DellEMC_Unity_Manila_dhss_true/196 : FAILURE in 44m 24s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/96/818496/3/check/DellEMC_Unity_Manila_dhss_false/2d4eac8/DellEMC_Unity_Manila_dhss_false/185 : FAILURE in 9m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"01441ade6a15c3b6b1fbf641ced627331525a598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 13:36:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/331ec50e6a444652aba99a3bd992fca0 : SUCCESS in 3m 32s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a5825a5cf7684fa0a0a36861f9b0a66f : SUCCESS in 35m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e2c3ece738cf4b768e3518a7395730df : SUCCESS in 39m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c31d2aa9df0841c8aa52afeecd81ce4f : SUCCESS in 1h 13m 03s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/dde04c6161814ee2ab1a9037f9729d62 : SUCCESS in 1h 08m 26s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c4399d21a42a4c77b8e00b91acafb848 : SUCCESS in 1h 14m 43s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/9526eb433e074552842feab9dee83a6b : SUCCESS in 1h 01m 29s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ae97d44d9a994e01b32e95bb909cb2fb : FAILURE in 1h 11m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/dad0aef7b8c04b2b8b1edbdf528bb2dc : SUCCESS in 1h 18m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7d5e59ff195a4911bd3235764fecb373 : FAILURE in 1h 44m 56s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/02b4f627c4fa4bc8817b0ca2acd2df01 : SUCCESS in 46m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1bb28c6025cb469ea23c9c15753da72e : FAILURE in 2h 05m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f750d5c2db7c43a19e7c8994ddb2c334 : SUCCESS in 23m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ef08a15dcb987a2b16bc7ce3ec02418a4c5b591d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-06 13:36:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/96/818496/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/4d83896/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/96/818496/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/ebbcbfb/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"2550306ef694912298083950d16a8c83f66e824a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-06 14:17:55.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/96/818496/3/check/DellEMC_Unity_Manila_dhss_true/6b5a76b/DellEMC_Unity_Manila_dhss_true/197 : FAILURE in 41m 27s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/96/818496/3/check/DellEMC_Unity_Manila_dhss_false/2ed05cd/DellEMC_Unity_Manila_dhss_false/186 : FAILURE in 9m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"8350f5d75479e45329f48d931ba605065f1407c4","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-07 00:54:37.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"64390b1cd0ffda8fa2567106cc424a8617733957","revisions":{"411d00e17a626f0b349ad7ee41ec320b0b5f438a":{"kind":"REWORK","_number":1,"created":"2021-11-19 01:11:24.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/96/818496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/96/818496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/818496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/818496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/818496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/818496/1"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-19 01:10:56.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-19 01:10:56.000000000","tz":480},"subject":"Maximum number of share replicas allowed from V2.53","message":"Maximum number of share replicas allowed from V2.53\n\nChange-Id: I6133f51f1ec2d5a12141efc119e48ee283fd2dd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/411d00e17a626f0b349ad7ee41ec320b0b5f438a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/411d00e17a626f0b349ad7ee41ec320b0b5f438a"}]},"branch":"refs/heads/master"},"f51e6d3a46f4e756a9dba483e1ef4f05648428c8":{"kind":"REWORK","_number":2,"created":"2021-11-25 02:35:27.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/96/818496/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/96/818496/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/818496/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/818496/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/818496/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/818496/2"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-19 01:10:56.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-25 02:31:51.000000000","tz":480},"subject":"Maximum number of share replicas allowed from V2.53","message":"Maximum number of share replicas allowed from V2.53\n\nMaxTotalShareReplicas and MaxTotalReplicaGigabytes allowd from v2.53\n\nChange-Id: I6133f51f1ec2d5a12141efc119e48ee283fd2dd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f51e6d3a46f4e756a9dba483e1ef4f05648428c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f51e6d3a46f4e756a9dba483e1ef4f05648428c8"}]},"branch":"refs/heads/master"},"64390b1cd0ffda8fa2567106cc424a8617733957":{"kind":"REWORK","_number":3,"created":"2022-01-06 11:26:27.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/96/818496/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/96/818496/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/818496/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/818496/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/818496/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/96/818496/3"}}},"commit":{"parents":[{"commit":"9c4a97b150e15b814acd4feb4da858a0eeff881e","subject":"Fix check to server migration complete test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9c4a97b150e15b814acd4feb4da858a0eeff881e"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-19 01:10:56.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-06 11:17:28.000000000","tz":480},"subject":"Maximum number of share replicas allowed from V2.53","message":"Maximum number of share replicas allowed from V2.53\n\nMaxTotalShareReplicas and MaxTotalReplicaGigabytes allowd from v2.53\n\nChange-Id: I6133f51f1ec2d5a12141efc119e48ee283fd2dd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/64390b1cd0ffda8fa2567106cc424a8617733957"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/64390b1cd0ffda8fa2567106cc424a8617733957"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
