)]}'
{"id":"openstack%2Fmanila-tempest-plugin~821280","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I1e019ec913f8933d50afae6739352686725f4780","project":"openstack/manila-tempest-plugin","branch":"master","topic":"stable-jobs","attention_set":{},"removed_from_attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2021-12-13 18:40:47.000000000","reason":"Douglas Viroel replied on the change"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2021-12-13 19:54:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e019ec913f8933d50afae6739352686725f4780","subject":"Add stable/xena job and drop stable/ussuri job","status":"MERGED","created":"2021-12-09 19:46:54.000000000","updated":"2021-12-13 19:54:42.000000000","submitted":"2021-12-13 19:54:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"821280-stable-jobs","meta_rev_id":"c26beceb99174054d5e0d8c520e170d2a64baf72","_number":821280,"virtual_id_number":821280,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-13 19:54:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2021-12-13 17:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2021-12-13 17:29:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2021-12-13 18:40:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"CC":[{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-09 19:47:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-10 23:53:44.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2021-12-10 23:54:50.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2021-12-11 01:30:54.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-12-13 17:15:31.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"c4d98b864345e8f9d998501eb29d3d1b7325ce3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-09 19:46:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b76afbadcb630ea06d574d50775fa96f2be2538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 19:47:43.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  Job manila-tempest-plugin-lvm-ussuri not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - tempest-plugin-jobs\n      check:\n        jobs:\n          - manila-tempest-plugin-dummy-no-dhss\n          - manila-tempest-plugin-dummy-dhss\n          - manila-tempest-plugin-lvm\n          - manila-tempest-plugin-lvm-wallaby\n          - manila-tempest-plugin-lvm-victoria\n          - manila-tempest-plugin-lvm-ussuri\n          - manila-tempest-plugin-zfsonlinux:\n              voting: false\n          - manila-tempest-plugin-cephfs-native:\n              voting: false\n          - manila-tempest-plugin-cephfs-nfs:\n              voting: false\n          - manila-tempest-plugin-container:\n              voting: false\n          - manila-tempest-plugin-generic:\n              voting: false\n          - manila-tempest-plugin-glusterfs-nfs:\n              voting: false\n      gate:\n        queue: manila\n        jobs:\n          - manila-tempest-plugin-dummy-no-dhss\n          - manila-tempest-plugin-dummy-dhss\n          - manila-tempest-plugin-lvm\n      experimental:\n        jobs:\n          - manila-tempest-plugin-glusterfs-native:\n              voting: false\n\n  in \"openstack/manila-tempest-plugin/zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"f2c0d8980865082ddef750bb8d5919bde40639f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-09 23:10:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"962881bf8aee18389254924809f7187254d66d76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 02:31:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/177cdc8e12cf48418d02ac53f24e8722 : SUCCESS in 3m 08s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/74f2a191125f4dcdb2dc229a083c7a43 : SUCCESS in 27m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dbec89f2a2bc42b7bd5e82d84bce7f4d : SUCCESS in 30m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fac5f683b55f4805a41c357a8ab6b166 : SUCCESS in 1h 04m 39s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/dbee7ddd6e6444fcb5287d3f039e260a : FAILURE in 1h 03m 47s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/5538c5f180fe4313b7594a9c5bf7a095 : SUCCESS in 1h 03m 40s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/60d2c43ee18c4a799836e4ae7c167b0e : SUCCESS in 1h 47m 43s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aa70426e71b648b8a782f21f64dae6e6 : SUCCESS in 1h 07m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c79fab40e076449eb8f6ae39a6582187 : FAILURE in 1h 20m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cbd88e51a34d44639f1576e595907283 : FAILURE in 1h 46m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/00dbff04d95c42beb0ba66c1c0802a35 : SUCCESS in 34m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/954482b9a92246d7a633c93b3fd31750 : FAILURE in 2h 18m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/33a3c54772ec4f54983bbdd797187fd7 : FAILURE in 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9d3573582f476ce18fd23b4acbf1ac4100387bfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-10 20:26:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"719937587a19316809c5689130b98ee44f87f975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-10 22:12:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1446489a753145e882b49160e78fc400 : SUCCESS in 3m 11s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d8ce9f61367d4b0b9f686d42273543fe : SUCCESS in 24m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/91af28e766ea442790b73df532954a58 : SUCCESS in 35m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dc7736fa3c864e019a372efed8ce1c89 : SUCCESS in 1h 37m 34s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/7c55be5d5a5c4be59fc01c4f48582423 : SUCCESS in 1h 03m 34s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/44e5390e8c8c40e5acbf112542ba4e99 : FAILURE in 1h 09m 52s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/e0480ad6ddda44128c339f8e5a20d2a2 : SUCCESS in 1h 11m 37s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f4deed7095504bbaa8c1cd71559b390a : SUCCESS in 1h 19m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6744dd2d51c3437599c0df39bc82e1e3 : FAILURE in 20m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4b95ffd333614da59f828401059e63b1 : FAILURE in 20m 48s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/159234a548b342b4adcc10477fbeb9f3 : SUCCESS in 46m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9f459973fbad4e1ebf36efc3b2875854 : FAILURE in 1h 02m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1e9a0ceb56334ecf8e98e6bbfa541812 : SUCCESS in 25m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b2b31a25f2b2bb7f6707008e2c4bc38ba0f1cc4e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-12-10 23:53:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f3d09bb6edf69ebe16296e5b9deff25faddefc46","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-12-10 23:54:50.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d92a4fbb430280cb0820d272e05f8464c34e882c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-11 01:30:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10cc788157c74557824e3691760bbf7e : SUCCESS in 3m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2b8e36e226364251a71205eb16ce9a99 : SUCCESS in 25m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c7443d76bac349efb599bcc394d55d4f : SUCCESS in 28m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5fe69ea11f5445b59ad5c1830cf4e1c1 : SUCCESS in 1h 05m 54s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/7726b9702e88433ba0997843f7621f52 : SUCCESS in 56m 12s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/8739e8254e34405193cd313bcc5af321 : SUCCESS in 1h 02m 45s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/8fc222b463c547a3893e2b34fcd397eb : SUCCESS in 1h 01m 40s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d609482624054972a71b958779bbc315 : SUCCESS in 1h 12m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6976a607e64645038cc980374259fc39 : FAILURE in 12m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a6647d4f139344dc8d7474f506d7396e : FAILURE in 13m 14s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e0037dd41a2646d09b85c0176433d685 : SUCCESS in 39m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/92eced0199314a7f9cb5109bad3512f6 : FAILURE in 1h 35m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/44ed73dbaf524fc49b52d57b51b53152 : SUCCESS in 18m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fc86677033239191b1e8c37c317a75124dd4d4d6","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2021-12-11 01:30:54.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/80/821280/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/2b2f4dd/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/80/821280/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/caa187a/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"4ecd8ec4b61d13ca48ba8f18d5164b188360ec90","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2021-12-13 17:15:31.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1291f27be7fbc19f636fbc58c562667d2553dd27","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-12-13 17:29:12.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8bf07d7f047202a5564c425b943d2c1ce16b11e3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2021-12-13 18:34:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1c556b52ed2b67af0e7aea916b9675ae420dfaf0","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-12-13 18:40:47.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2f8c53ee09be3e8cd43cbb9038eff569e2f5e168","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 18:41:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"42097748f1c0345e8edf2a38095250b045b8f73d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 19:54:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb2f7382c7df457fbe6c7e745d1076a1 : SUCCESS in 3m 37s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/768067043a1e40a08a8ed7d619548c3d : SUCCESS in 23m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2e06d8658d89419abf1bc8d5f6b41c54 : SUCCESS in 30m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/314d6e7c83044de1a34dcab926d86fc5 : SUCCESS in 1h 12m 23s","accounts_in_message":[],"_revision_number":3},{"id":"c26beceb99174054d5e0d8c520e170d2a64baf72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-13 19:54:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b4c94c8c4ee0ba8f514bb833fb283a0107f206ef","revisions":{"fc1fa460d3e8c1e3433fcd38dd563d95b733f376":{"kind":"REWORK","_number":1,"created":"2021-12-09 19:46:54.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/80/821280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/821280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/821280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/821280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/821280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/821280/1"}}},"commit":{"parents":[{"commit":"2d0c8694a4a61acbf523438b5406c781958e0f1f","subject":"Add enforce_scope config option for Manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2d0c8694a4a61acbf523438b5406c781958e0f1f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-12-09 19:46:40.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-12-09 19:46:40.000000000","tz":-480},"subject":"Add a stable/xena job and drop the stable/ussuri job","message":"Add a stable/xena job and drop the stable/ussuri job\n\nstable/ussuri has now been transitioned to EM (Extended Maintenance) [1].\nSo we can stop testing manila-tempest-plugin changes against that branch,\nand introduce a stable/xena job.\n\nDropping the stable/ussuri job does not mean that manila-tempest-plugin\ndoes not support running against that branch. Extended maintenance\nbranches (stable/ussuri, train and stein at the moment) still get\nper-commit testing when changes are posted to the manila repo against\nthose branches.\n\nChange-Id: I1e019ec913f8933d50afae6739352686725f4780\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc1fa460d3e8c1e3433fcd38dd563d95b733f376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc1fa460d3e8c1e3433fcd38dd563d95b733f376"}]},"branch":"refs/heads/master"},"7413f72dce198251a2d0d100aec8ebf23a80add1":{"kind":"REWORK","_number":2,"created":"2021-12-09 23:10:27.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/80/821280/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/821280/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/821280/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/821280/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/821280/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/821280/2"}}},"commit":{"parents":[{"commit":"2d0c8694a4a61acbf523438b5406c781958e0f1f","subject":"Add enforce_scope config option for Manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2d0c8694a4a61acbf523438b5406c781958e0f1f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-12-09 19:46:40.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-12-09 23:10:10.000000000","tz":-480},"subject":"Add a stable/xena job and drop the stable/ussuri job","message":"Add a stable/xena job and drop the stable/ussuri job\n\nstable/ussuri has now been transitioned to EM (Extended Maintenance) [1].\nSo we can stop testing manila-tempest-plugin changes against that branch,\nand introduce a stable/xena job.\n\nDropping the stable/ussuri job does not mean that manila-tempest-plugin\ndoes not support running against that branch. Extended maintenance\nbranches (stable/ussuri, train and stein at the moment) still get\nper-commit testing when changes are posted to the manila repo against\nthose branches.\n\nChange-Id: I1e019ec913f8933d50afae6739352686725f4780\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7413f72dce198251a2d0d100aec8ebf23a80add1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7413f72dce198251a2d0d100aec8ebf23a80add1"}]},"branch":"refs/heads/master"},"b4c94c8c4ee0ba8f514bb833fb283a0107f206ef":{"kind":"REWORK","_number":3,"created":"2021-12-10 20:26:11.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/80/821280/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/80/821280/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/821280/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/821280/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/821280/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/80/821280/3"}}},"commit":{"parents":[{"commit":"2d0c8694a4a61acbf523438b5406c781958e0f1f","subject":"Add enforce_scope config option for Manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2d0c8694a4a61acbf523438b5406c781958e0f1f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-12-09 19:46:40.000000000","tz":-480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2021-12-10 20:26:01.000000000","tz":-480},"subject":"Add stable/xena job and drop stable/ussuri job","message":"Add stable/xena job and drop stable/ussuri job\n\nstable/ussuri has now been transitioned to EM (Extended Maintenance) [1].\nSo we can stop testing manila-tempest-plugin changes against that branch,\nand introduce a stable/xena job.\n\nDropping the stable/ussuri job does not mean that manila-tempest-plugin\ndoes not support running against that branch. Extended maintenance\nbranches (stable/ussuri, train and stein at the moment) still get\nper-commit testing when changes are posted to the manila repo against\nthose branches.\n\nChange-Id: I1e019ec913f8933d50afae6739352686725f4780\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b4c94c8c4ee0ba8f514bb833fb283a0107f206ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b4c94c8c4ee0ba8f514bb833fb283a0107f206ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
