)]}'
{"id":"openstack%2Fmanila-tempest-plugin~823771","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I2fbbd4aa8c3848a95731c1936080c8815c05612e","project":"openstack/manila-tempest-plugin","branch":"master","topic":"master_06","attention_set":{"31280":{"account":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"last_update":"2022-01-07 08:03:23.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-01-19 09:26:53.000000000","reason":"Liron Kuchlani replied on the change"}},"hashtags":[],"change_id":"I2fbbd4aa8c3848a95731c1936080c8815c05612e","subject":"Admin reset and force-delete snapshot in any state","status":"NEW","created":"2022-01-07 05:53:51.000000000","updated":"2022-01-19 09:26:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"570a7fccb529f7532327c3723c5273d38e353866","_number":823771,"virtual_id_number":823771,"owner":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-07 10:44:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"disliked":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"all":[{"value":2,"date":"2022-01-18 22:19:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-01-19 09:26:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-07 08:03:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-07 10:45:09.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-07 11:45:01.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-01-18 22:19:04.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-01-18 22:19:14.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"}],"messages":[{"id":"ac37eb63adeb4420304161c3df8be771d5a03427","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-07 05:53:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4eed80eeb23e3ddaa7e45952ed3831c2ca4493c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 08:03:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b43a52a3ef4e41cab1c2e6893bfac611 : SUCCESS in 3m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c350ad59770245d5960dd07311a6e8a9 : FAILURE in 26m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ea2aaabafe164dd5990fe8e8373057b5 : FAILURE in 39m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/748bc25df037474ea25e77ca9b9de2b0 : SUCCESS in 1h 11m 24s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/27d6933b7cc342e0a689373c646f2506 : SUCCESS in 1h 02m 14s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9e9cd16c2445405cafd991403a9ae5a2 : SUCCESS in 1h 19m 28s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/d7a7e33df082421b98d8dd87052f7908 : SUCCESS in 1h 15m 04s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6db6481e334e4878bf930fd038158e75 : SUCCESS in 1h 15m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e6e938c4fec14d19b83cf86951015d39 : FAILURE in 1h 23m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/49f96c91452a42ea963a5b1657041427 : FAILURE in 1h 40m 43s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e8a24d6e85174b948a3cd094ddc5c09d : SUCCESS in 1h 00m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4a472f743ab94778abe424db0627772c : FAILURE in 2h 08m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bc0db099547c4d34ab18ad4ccf3a2fec : SUCCESS in 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8de7c49d56d3515606c27735c8ae8325bacb442f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-07 08:17:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48a3fa1868fa0080935db6a45766d78afc11d077","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 10:44:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d9bf1c37e0f4ec38b662b755811a30e : SUCCESS in 3m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f9b2e23e57a547bba0cd464157ff1c9c : SUCCESS in 27m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2f4cc0fb26a048e38e3cb4e081aaa7b3 : SUCCESS in 27m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6af81fe9fe3c47958903de435b09bb51 : SUCCESS in 1h 08m 09s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/8fec898e90b142148b6e68332c8dd8ac : SUCCESS in 1h 00m 56s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/cb332cd6ec3f4bb1b5532fba8bf30acf : SUCCESS in 1h 10m 25s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/91131d3971f0470e9fb0d2d3bb576d35 : SUCCESS in 1h 28m 20s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9bf0de0b50b84190aba5115c635227ec : SUCCESS in 1h 15m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/38b811d6bad84536861c971595e5641b : FAILURE in 1h 20m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4192db399ddb4dd2859aeda9c6faa29c : FAILURE in 1h 55m 16s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1ec70c761bc84a888fd4470884f970f4 : SUCCESS in 49m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5a2bfff6433c41838ade9a952100f26d : FAILURE in 2h 25m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b6c30764d0464451a4c2854f26cb6021 : SUCCESS in 20m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5a0e7205e828a9a926e27c796f6e4799f8f1abfb","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-07 10:45:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/71/823771/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c8edcb3/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/71/823771/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/279df8b/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"93757515da9d647face7da86a54cee29fd5e4082","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-07 11:45:01.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/71/823771/2/check/DellEMC_Unity_Manila_dhss_true/2bbc4ad/DellEMC_Unity_Manila_dhss_true/203 : FAILURE in 38m 47s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/71/823771/2/check/DellEMC_Unity_Manila_dhss_false/24ae3cd/DellEMC_Unity_Manila_dhss_false/192 : FAILURE in 9m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"923f50f33a29fdcf443aec6b6522b98ba6f522de","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-01-18 22:19:04.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"02daa70ba98a0218c66e7e358da78e9d1d6fb363","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-19 09:26:53.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"570a7fccb529f7532327c3723c5273d38e353866","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-01-19 09:26:58.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f88c2ab0490eb7103a5be9cd9e0fa5c40e4f8283","revisions":{"5e75dc9c73bdfd89fd4a910af1c9a84523de3395":{"kind":"REWORK","_number":1,"created":"2022-01-07 05:53:51.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/71/823771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/71/823771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/823771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/823771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/823771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/823771/1"}}},"commit":{"parents":[{"commit":"e480c24e177b21bd24511a388095ed51d0e2b2cd","subject":"Merge \"Pre-existing metadata items, set new share metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e480c24e177b21bd24511a388095ed51d0e2b2cd"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 05:53:37.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 05:53:37.000000000","tz":480},"subject":"Admin reset and force-delete snapshot in any state","message":"Admin reset and force-delete snapshot in any state\n\nChange-Id: I2fbbd4aa8c3848a95731c1936080c8815c05612e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5e75dc9c73bdfd89fd4a910af1c9a84523de3395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5e75dc9c73bdfd89fd4a910af1c9a84523de3395"}]},"branch":"refs/heads/master"},"f88c2ab0490eb7103a5be9cd9e0fa5c40e4f8283":{"kind":"REWORK","_number":2,"created":"2022-01-07 08:17:50.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/71/823771/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/71/823771/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/823771/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/823771/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/823771/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/71/823771/2"}}},"commit":{"parents":[{"commit":"e480c24e177b21bd24511a388095ed51d0e2b2cd","subject":"Merge \"Pre-existing metadata items, set new share metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e480c24e177b21bd24511a388095ed51d0e2b2cd"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 05:53:37.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 08:14:06.000000000","tz":480},"subject":"Admin reset and force-delete snapshot in any state","message":"Admin reset and force-delete snapshot in any state\n\nReset snapshot status and force-delete snapshot. Valid states: error_deleting, deleting, available, migrating, error, creating, server_migrating, migrating_to.\n\nChange-Id: I2fbbd4aa8c3848a95731c1936080c8815c05612e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f88c2ab0490eb7103a5be9cd9e0fa5c40e4f8283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f88c2ab0490eb7103a5be9cd9e0fa5c40e4f8283"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
