)]}'
{"id":"openstack%2Fmanila-tempest-plugin~823774","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I14481f5a63990fb6f44dd5dcb40d24b74862a98f","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{"31280":{"account":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"last_update":"2022-01-12 20:43:56.000000000","reason":"Goutham Pacha Ravi replied on the change"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-02-06 08:00:31.000000000","reason":"Liron Kuchlani replied on the change"}},"removed_from_attention_set":{"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-02-06 08:00:31.000000000","reason":"Liron Kuchlani replied on the change"}},"hashtags":[],"change_id":"I14481f5a63990fb6f44dd5dcb40d24b74862a98f","subject":"Force create snapshot for invalid share","status":"NEW","created":"2022-01-07 07:36:51.000000000","updated":"2022-02-06 08:00:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":11,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"faa672ec2b76e4723ca61b27c57b7314011d352a","_number":823774,"virtual_id_number":823774,"owner":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-13 04:50:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"recommended":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":2,"date":"2022-01-13 07:13:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-02-04 18:02:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-07 09:43:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-07 11:06:14.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-01-11 11:01:45.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-11 20:36:39.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2022-01-12 20:43:56.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-02-04 18:02:36.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-02-04 22:53:40.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"}],"messages":[{"id":"2355b34d8ce2f049d7794be44a0b0d1891e78730","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-07 07:36:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d01ecfe1fbdd41b6e32df57da0656648920c5946","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 09:43:38.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b75ff46877045c8a36f3d5ea4da3868 : FAILURE in 3m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5d99708d0e6848d6b61785410cc1f5b3 : FAILURE in 33m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/049842774c684e9f8261e5b628786572 : FAILURE in 42m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8d101bf3f0bc4bf8b3289ce7d8d64133 : FAILURE in 1h 24m 30s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/9497be0e70cd42eb89ecd0eb41b630f0 : FAILURE in 1h 20m 37s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9262368395924dc9a5d56653af09108c : FAILURE in 1h 46m 08s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/e5be8ddb419549a4a344db2f92438962 : FAILURE in 1h 14m 42s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ef7b4dbe6f5c40ada1cdd2a182bab7f5 : FAILURE in 1h 23m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/faef485bb9a845918673d1228b7fc62e : FAILURE in 1h 22m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8ecbcb61f2104023ad6dcf3674714949 : FAILURE in 2h 01m 37s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9229cde1d98442c3bafd78cf3f45493c : SUCCESS in 52m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/264fc527cb95486f8b1f3e9ff843936e : FAILURE in 1h 50m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7fc6f9e15729462dbff16eb06b2dcdf5 : SUCCESS in 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"347a92bccf3abdf5438ba6c0ee5c4b5ec5a9264b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-07 10:27:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8bf879c54c090d490444a728f9df15730da62d5c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-07 11:06:14.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/74/823774/2/check/DellEMC_Unity_Manila_dhss_true/d599c41/DellEMC_Unity_Manila_dhss_true/202 : FAILURE in 38m 28s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/74/823774/2/check/DellEMC_Unity_Manila_dhss_false/0c8f566/DellEMC_Unity_Manila_dhss_false/191 : FAILURE in 9m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"e8bf29fa18db70bda080befd37ca9bc74371f77e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-07 12:40:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f15b5ce5faae44119edd5730d945544b : SUCCESS in 3m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/71b7b6d0d2784a1eb23ecb19e8fd0d61 : FAILURE in 34m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d74d67edff1b4475bd0a0baf60da2cda : FAILURE in 41m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/56f972cf89e94efbad11257cacd88c44 : FAILURE in 1h 16m 00s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/18c9bd7537874018b1cc5f46754e0afd : FAILURE in 1h 05m 19s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b86b449f1a474db5899e8e92d437a676 : FAILURE in 1h 08m 50s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/5c0f165b702f46f7a11142fce83d262d : FAILURE in 1h 17m 35s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/77e43bb66bbe4aa5a02727d31ee4167c : FAILURE in 1h 11m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/de6df22e5992479297cc570dc1ca26a3 : FAILURE in 1h 17m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/96612f64b7974ddc92e5ad4b30168a2c : FAILURE in 1h 55m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2c8a518f8c404c5b90dd5000b776274a : SUCCESS in 55m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c8164574136d4ca8b7819d50805e1b3b : FAILURE in 2h 09m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/25ea62a9277244da8498d43660949569 : SUCCESS in 27m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b456f88f328ac05c25013c13c9439432caef897e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-08 03:05:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"39c5e23897a373c54acbe96e121b3d03d1eca163","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-08 03:28:05.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/74/823774/3/check/DellEMC_Unity_Manila_dhss_true/bf866aa/DellEMC_Unity_Manila_dhss_true/207 : FAILURE in 22m 13s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/74/823774/3/check/DellEMC_Unity_Manila_dhss_false/3cdfa83/DellEMC_Unity_Manila_dhss_false/196 : FAILURE in 15m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"717c91b5c368a2fca74ae5db2e338055e2cef998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 05:01:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f2180b5a3b40c7b307d611bac62a08 : SUCCESS in 4m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5bc9a921872b4cab87273cab054f7b41 : FAILURE in 32m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f36ff1f227ae4688aa903e3478b7d5d2 : FAILURE in 38m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/55c69821633648d7aec8d02a0479394e : FAILURE in 1h 40m 26s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a38c009688144dbe8cd6e8e0c1c53563 : FAILURE in 1h 03m 38s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/fabe3562e28845689e71470620f7c878 : FAILURE in 1h 15m 04s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/9ff90b4f9f00449da49a40e5fc8ca100 : FAILURE in 1h 13m 56s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b9f24b07f4564e83910812c23de0e287 : FAILURE in 1h 22m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ef101d7408fe4c69b29eaccf7904c469 : FAILURE in 1h 09m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/72d2c54766b8475d9e9e8ed1151a8033 : FAILURE in 1h 29m 34s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fef4e6e2dd604007acf60e63bd389a19 : SUCCESS in 57m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b85d1139d7b24dcfb627a815d7b9cf86 : FAILURE in 1h 46m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/59496488bc0b4cf0a97db7d9e4f84a40 : SUCCESS in 28m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0c17f1efab345948afa8335bec5e429e55543eb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-08 06:22:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1b5b9ec0d04135c9531a61056558bf41f6d4c06b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-08 07:14:23.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/74/823774/4/check/DellEMC_Unity_Manila_dhss_true/8b0909f/DellEMC_Unity_Manila_dhss_true/208 : FAILURE in 51m 52s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/74/823774/4/check/DellEMC_Unity_Manila_dhss_false/927e401/DellEMC_Unity_Manila_dhss_false/197 : FAILURE in 28m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"6278c489fda0666042193321debd6a474c8418df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-08 08:22:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ec1034a1a294b9aa787dbfadc8d7221 : SUCCESS in 3m 12s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d98e8cd5f2414e53b6f53eb5d5e73d36 : FAILURE in 36m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/59f694b7a9134305a849736d5be70e87 : FAILURE in 47m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4337498268a54954b18a7b0994dbda24 : FAILURE in 1h 34m 47s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/eb67f6bdacf04263bae3b5cef9970cd4 : FAILURE in 1h 05m 33s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/40a9f819323b41368bcea9ee18d809b1 : FAILURE in 1h 06m 53s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/a9bbe334df0b4291846018d5d19848aa : FAILURE in 1h 25m 42s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f0dff8acd9234481966ec13aa6ea7522 : FAILURE in 1h 37m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/906ae57bf487451cb8fc81c50d19f326 : FAILURE in 1h 31m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1df1f5631f4a47beb09a3d8f4d630ec0 : FAILURE in 1h 42m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b80dde8a4c954fb9904c2763ae308c6e : SUCCESS in 55m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a349c32f36fa47cea819dd33c77932b8 : FAILURE in 1h 55m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c0c5ce1d8614463c9c29d27d1752d853 : SUCCESS in 27m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9de30637012516f4560984e2541f18d450d1f4f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-10 02:07:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d8019dc3009f61229acea168c834273cf16d88f4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-10 02:47:35.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/74/823774/5/check/DellEMC_Unity_Manila_dhss_true/afcca1b/DellEMC_Unity_Manila_dhss_true/209 : FAILURE in 39m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/74/823774/5/check/DellEMC_Unity_Manila_dhss_false/231edc5/DellEMC_Unity_Manila_dhss_false/198 : FAILURE in 10m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"cc031fdcc213669e46a3971ac70257209921af60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 04:42:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00e50f23f5264094b98bdcbcb00225ea : SUCCESS in 4m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/35c3000cb6724ce9b420ea334efb37f7 : FAILURE in 36m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0ae7f3caaa4e452985d312a21fa49a81 : FAILURE in 38m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6d5cfd87926e4952ac6ae0dfedd14563 : SUCCESS in 1h 20m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a3db66779c7647df920ece42a80181d4 : SUCCESS in 1h 08m 50s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/05daba6e6b1046de9ebff2e6735b50f1 : SUCCESS in 1h 14m 26s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/22fdde0a81fd44cd8c1beb67434b23e7 : SUCCESS in 1h 40m 47s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/72ced1e7e09f42029263d1385ae07157 : FAILURE in 1h 26m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d779dcceef174081954b07121026f0ec : FAILURE in 1h 33m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/405c279474c14057bfec70ced001010c : FAILURE in 2h 06m 30s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6ca3b65fdcd14f2c91d9ce2b50c6cfc6 : SUCCESS in 56m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a3c85714e0e24c0483a234420ea42ab4 : FAILURE in 1h 46m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/674b578c7466419bb803fb691c80386c : SUCCESS in 30m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3616dbb3b65bf066b2cbf40ee18b69edf839e0e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-11 08:43:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"24a0e147d1e2976675ba241051d6072ab2bf1577","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-11 09:22:53.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/74/823774/6/check/DellEMC_Unity_Manila_dhss_true/5a94347/DellEMC_Unity_Manila_dhss_true/215 : FAILURE in 38m 44s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/74/823774/6/check/DellEMC_Unity_Manila_dhss_false/9775456/DellEMC_Unity_Manila_dhss_false/205 : FAILURE in 13m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"e2ed69180c8730136ce87355937c7a7642cf66e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 11:01:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2763b6e60b1499b92a71f72067a6434 : SUCCESS in 3m 22s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0505e92b1c13453ea3cb92a8d552121d : SUCCESS in 34m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/745f24f0dc5f432c8446a18521231d82 : SUCCESS in 42m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f74826959a194cf8aa2c3bcd98e71ab7 : SUCCESS in 1h 19m 09s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/facf425be4624f31a22df140e3b204c4 : SUCCESS in 1h 10m 44s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/0df78e3479e447b7a9f970c3c5c90860 : SUCCESS in 1h 11m 22s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/ba6e081e30524956869ccbd37e500e0a : SUCCESS in 1h 08m 14s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cc62f3d3eb7b490b8e5c9ff66ccbdacf : SUCCESS in 1h 20m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7d87d1f7120d43128db37af91b389189 : FAILURE in 1h 59m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b4c57f6ea21045559fd7b15b89bcb2e9 : FAILURE in 1h 39m 38s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ecd5bd940d4b4642955c7b9bb57d4138 : SUCCESS in 45m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e68804c19da942feb7ded27f9f72edae : FAILURE in 2h 09m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ccabb146003b4b819ae6308ec60c5e57 : SUCCESS in 22m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e7304b4c30994ea124cb5758722f71392eced033","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-11 11:01:45.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/74/823774/6/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f0ccaf6/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/74/823774/6/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/176d6f0/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"f58d583b8ed8fc50f5d763fc51c9432c7dc46f52","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-11 12:09:54.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/74/823774/6/check/DellEMC_Unity_Manila_dhss_true/d3519f5/DellEMC_Unity_Manila_dhss_true/217 : FAILURE in 43m 10s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/74/823774/6/check/DellEMC_Unity_Manila_dhss_false/942a2fc/DellEMC_Unity_Manila_dhss_false/208 : FAILURE in 38m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"1346b86f48777b0b9197193d85f9581a3d784aa0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-01-11 20:36:39.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f4b5d5447378b25e889c9b0fbcf87b92a7137fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-12 00:53:35.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3cb00cad6c8e3b99e128cebebbb3d9a4e15c1822","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-12 01:37:51.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/74/823774/7/check/DellEMC_Unity_Manila_dhss_true/0adeaa5/DellEMC_Unity_Manila_dhss_true/227 : FAILURE in 44m 02s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/74/823774/7/check/DellEMC_Unity_Manila_dhss_false/5780589/DellEMC_Unity_Manila_dhss_false/218 : FAILURE in 36m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"26a79ac38bb6062ad72f9a8c4da25df873ac9e84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 03:14:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84bbb4c3bd0f4a09b3ae3f10bced8610 : SUCCESS in 3m 12s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/31d2dd4578464e7cbfa05eb0a3f3485e : SUCCESS in 34m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d2c8f2b1fd534bcdb85e8b9d177c46aa : SUCCESS in 35m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/13489988947646b8ac522adfb607c2de : SUCCESS in 1h 16m 46s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/82bb73b9d3674d099d59fc36183d4932 : SUCCESS in 1h 09m 59s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e6e3a367bda04f229e716f6f0e89a70d : SUCCESS in 1h 02m 17s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/365f5db046014e18a6f235d5f3317b60 : SUCCESS in 1h 08m 33s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/69a7b9dc90e94c43bfd2052acf11ffe6 : SUCCESS in 1h 23m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/06f6abddeda14e39a111c90db02cc570 : SUCCESS in 1h 17m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/589effcaf2e0433b8ac9ecc3fe907639 : FAILURE in 1h 32m 15s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/457cffed812049b887f078836ee9a4c4 : SUCCESS in 51m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5152a660ae014e58ae5c863d7c1c5c34 : FAILURE in 2h 16m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/eaba89a5ec6f4713bb3c24d1959934e1 : SUCCESS in 21m 42s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4739daadd17a4dfbcbcf9a2b97839cc5545bd8f2","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-12 03:15:13.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/74/823774/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e0bac35/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/74/823774/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/fa052b9/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"a362f657c6e5b401de3455d7e301e7726d84bebd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-12 04:01:20.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/74/823774/7/check/DellEMC_Unity_Manila_dhss_true/c1165c2/DellEMC_Unity_Manila_dhss_true/228 : FAILURE in 46m 16s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/74/823774/7/check/DellEMC_Unity_Manila_dhss_false/4e2790b/DellEMC_Unity_Manila_dhss_false/219 : FAILURE in 44m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"6b0b7c8bfc58e8132917fbc62d0c8513a73a13c9","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-12 05:44:56.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4da56b22f80213dcf1224a185d8372da8a6d6863","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-01-12 20:43:56.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f0bf359fbf014c2c375bb87de196fd95b12cc152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-13 00:31:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"db284a38f5ef7fa3ce7bbbc0dad51b53e39c69d4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-13 01:17:27.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/74/823774/8/check/DellEMC_Unity_Manila_dhss_true/d65a41b/DellEMC_Unity_Manila_dhss_true/229 : FAILURE in 46m 12s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/74/823774/8/check/DellEMC_Unity_Manila_dhss_false/3ac7976/DellEMC_Unity_Manila_dhss_false/220 : FAILURE in 41m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"c8ff9bcf462782621de0208dbf6eb20500149fba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 02:44:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9efbdeca7b74b098f19ea1a7241917a : SUCCESS in 3m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/65084237c265476da532213137f4cb0e : FAILURE in 34m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/84ae356bcaa540af8920cfd31c34070e : FAILURE in 35m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e95c21333fd841e28a2b1fb77230b4af : SUCCESS in 1h 13m 42s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/9273f61034594603a9c8a31892974c6e : FAILURE in 1h 04m 52s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/0a1d07a5f4994ff892cfc05bb51c4f6c : SUCCESS in 1h 07m 48s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/fee0e8f9cd6d45caaea8ff79323f6310 : SUCCESS in 1h 02m 00s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8777b06d2aa34ca9be1f30ddcf96d9b8 : FAILURE in 1h 30m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/75eeab55cfe547b49a5d10f71eb111b8 : FAILURE in 1h 12m 38s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4fdb9be6f04a4e3186c86ae4656eb703 : FAILURE in 1h 44m 07s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/79da3e9622f04070a39e7e545acd7643 : SUCCESS in 47m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/53b568c167a04f9ba5ad1612997e9257 : FAILURE in 2h 11m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f2709aebcbc844488c63fc1da03d6200 : SUCCESS in 19m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6e4bb37f01141e5a6e91410c3ce8d235c17d12e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2022-01-13 02:49:52.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1864d20db88bdc7c0efe3a816804cb70f1847897","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-13 03:40:55.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/74/823774/9/check/DellEMC_Unity_Manila_dhss_true/7bf5025/DellEMC_Unity_Manila_dhss_true/230 : FAILURE in 50m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/74/823774/9/check/DellEMC_Unity_Manila_dhss_false/016a8ca/DellEMC_Unity_Manila_dhss_false/221 : FAILURE in 47m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"303bb1f05094fea188553dbda6bc4c87b7fede90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 04:50:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/291f6d220464499a8298a3541d44f67a : SUCCESS in 5m 22s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2e9ac2ab05d9414cb623c3abf4ace960 : SUCCESS in 35m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9735175f2a2d4bb9849c69a54a417767 : SUCCESS in 40m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/44ce60c53d014fffaf054f60d1683a01 : SUCCESS in 1h 18m 31s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d1ed7431dd2a44a29c88b9ae014881d4 : SUCCESS in 1h 10m 04s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/6928fd41b7ca435bb3ea0046d34516c3 : SUCCESS in 1h 26m 12s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/1eead158034f45728961d9960c413995 : SUCCESS in 1h 15m 34s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b521ecf5820541f2843cfea80cfe9a69 : SUCCESS in 1h 11m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f9d0053ad6744db98f82a6f3c4e7d31a : FAILURE in 1h 19m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1669dbf90e8543ecb5f54d1d9b8a83ab : FAILURE in 1h 51m 09s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b1daf6c341ed44c69533eaddfb33c615 : SUCCESS in 58m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/95dc5c84a0bf4c17a78415ddbef6c600 : FAILURE in 1h 32m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0614eded89b04ffe8e839ca9c6a70945 : SUCCESS in 28m 14s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c69e217e34f677c09537fb1b30944b168faf3139","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-13 04:50:38.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/74/823774/9/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ff84e7b/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/74/823774/9/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/d03bd57/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"4143f7616f32f23501106a47dd2eb3247e6428cf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-13 05:35:04.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/74/823774/9/check/DellEMC_Unity_Manila_dhss_true/af1aa64/DellEMC_Unity_Manila_dhss_true/231 : FAILURE in 44m 31s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/74/823774/9/check/DellEMC_Unity_Manila_dhss_false/861bcda/DellEMC_Unity_Manila_dhss_false/222 : FAILURE in 38m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"54e5f3808f0727f8fe3d5e9c4beae59158efe584","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-01-13 07:13:17.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"553fdbbae16d0089d4264a30bca53be33fc95a5c","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-02-04 18:02:36.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2277c92b3c570520c17f21983876d1809587b437","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-04 22:53:42.000000000","message":"Topic master_07 removed","accounts_in_message":[],"_revision_number":9},{"id":"faa672ec2b76e4723ca61b27c57b7314011d352a","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-06 08:00:31.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"69fcf8fea13a2b5609348fe5e530f7b01d28dd0b","revisions":{"40a3c5e140f139cc9821300a02dd972863459cec":{"kind":"REWORK","_number":1,"created":"2022-01-07 07:36:51.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/74/823774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/823774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/1"}}},"commit":{"parents":[{"commit":"e480c24e177b21bd24511a388095ed51d0e2b2cd","subject":"Merge \"Pre-existing metadata items, set new share metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e480c24e177b21bd24511a388095ed51d0e2b2cd"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 07:36:37.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 07:36:37.000000000","tz":480},"subject":"Force create snapshot for invalid share","message":"Force create snapshot for invalid share\n\nChange-Id: I14481f5a63990fb6f44dd5dcb40d24b74862a98f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40a3c5e140f139cc9821300a02dd972863459cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40a3c5e140f139cc9821300a02dd972863459cec"}]},"branch":"refs/heads/master"},"6a9460f2a55d3963842afc6d0729c6a1a5e9ebf5":{"kind":"REWORK","_number":2,"created":"2022-01-07 10:27:32.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/74/823774/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/823774/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/2"}}},"commit":{"parents":[{"commit":"e480c24e177b21bd24511a388095ed51d0e2b2cd","subject":"Merge \"Pre-existing metadata items, set new share metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e480c24e177b21bd24511a388095ed51d0e2b2cd"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 07:36:37.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 10:24:14.000000000","tz":480},"subject":"Force create snapshot for invalid share","message":"Force create snapshot for invalid share\n\nSet force\u003dtrue to force snapshot creation when the share is busy performing other operations.\n\nChange-Id: I14481f5a63990fb6f44dd5dcb40d24b74862a98f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6a9460f2a55d3963842afc6d0729c6a1a5e9ebf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6a9460f2a55d3963842afc6d0729c6a1a5e9ebf5"}]},"branch":"refs/heads/master"},"2c13c5406a226ac7fda91e7d7ce22811c44407aa":{"kind":"REWORK","_number":3,"created":"2022-01-08 03:05:38.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/74/823774/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/823774/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/3"}}},"commit":{"parents":[{"commit":"e480c24e177b21bd24511a388095ed51d0e2b2cd","subject":"Merge \"Pre-existing metadata items, set new share metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e480c24e177b21bd24511a388095ed51d0e2b2cd"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 07:36:37.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-08 03:05:09.000000000","tz":480},"subject":"Force create snapshot for invalid share","message":"Force create snapshot for invalid share\n\nSet force\u003dtrue to force snapshot creation when the share is busy performing other operations.\n\nChange-Id: I14481f5a63990fb6f44dd5dcb40d24b74862a98f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2c13c5406a226ac7fda91e7d7ce22811c44407aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2c13c5406a226ac7fda91e7d7ce22811c44407aa"}]},"branch":"refs/heads/master"},"35c6c4c5138fd725f84b9cfdf92fd9d1f8dedecb":{"kind":"REWORK","_number":4,"created":"2022-01-08 06:22:17.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/74/823774/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/823774/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/4"}}},"commit":{"parents":[{"commit":"e480c24e177b21bd24511a388095ed51d0e2b2cd","subject":"Merge \"Pre-existing metadata items, set new share metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e480c24e177b21bd24511a388095ed51d0e2b2cd"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 07:36:37.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-08 06:21:48.000000000","tz":480},"subject":"Force create snapshot for invalid share","message":"Force create snapshot for invalid share\n\nSet force\u003dtrue to force snapshot creation when the share is busy performing other operations.\n\nChange-Id: I14481f5a63990fb6f44dd5dcb40d24b74862a98f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/35c6c4c5138fd725f84b9cfdf92fd9d1f8dedecb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/35c6c4c5138fd725f84b9cfdf92fd9d1f8dedecb"}]},"branch":"refs/heads/master"},"d20bbd0c85c5c6c9ad51dbab4ee555b45dda18de":{"kind":"REWORK","_number":5,"created":"2022-01-10 02:07:58.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/74/823774/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/823774/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/5"}}},"commit":{"parents":[{"commit":"e480c24e177b21bd24511a388095ed51d0e2b2cd","subject":"Merge \"Pre-existing metadata items, set new share metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e480c24e177b21bd24511a388095ed51d0e2b2cd"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 07:36:37.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-10 02:05:33.000000000","tz":480},"subject":"Force create snapshot for invalid share","message":"Force create snapshot for invalid share\n\nSource share status must be available when create snapshot\n\nChange-Id: I14481f5a63990fb6f44dd5dcb40d24b74862a98f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d20bbd0c85c5c6c9ad51dbab4ee555b45dda18de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d20bbd0c85c5c6c9ad51dbab4ee555b45dda18de"}]},"branch":"refs/heads/master"},"528336bf19a000f9cd12a80b57d2ad25923d8f4d":{"kind":"REWORK","_number":6,"created":"2022-01-11 08:43:54.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/74/823774/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/823774/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/6"}}},"commit":{"parents":[{"commit":"e480c24e177b21bd24511a388095ed51d0e2b2cd","subject":"Merge \"Pre-existing metadata items, set new share metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e480c24e177b21bd24511a388095ed51d0e2b2cd"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 07:36:37.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-11 08:43:03.000000000","tz":480},"subject":"Force create snapshot for invalid share","message":"Force create snapshot for invalid share\n\nSource share status must be available when create snapshot\n\nChange-Id: I14481f5a63990fb6f44dd5dcb40d24b74862a98f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/528336bf19a000f9cd12a80b57d2ad25923d8f4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/528336bf19a000f9cd12a80b57d2ad25923d8f4d"}]},"branch":"refs/heads/master"},"a39a91cf0589bdacd7e21cc8c83a515c8cea0a8d":{"kind":"REWORK","_number":7,"created":"2022-01-12 00:53:35.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/74/823774/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/823774/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/7"}}},"commit":{"parents":[{"commit":"e480c24e177b21bd24511a388095ed51d0e2b2cd","subject":"Merge \"Pre-existing metadata items, set new share metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e480c24e177b21bd24511a388095ed51d0e2b2cd"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 07:36:37.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-12 00:53:07.000000000","tz":480},"subject":"Force create snapshot for invalid share","message":"Force create snapshot for invalid share\n\nSource share status must be available when create snapshot\n\nChange-Id: I14481f5a63990fb6f44dd5dcb40d24b74862a98f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a39a91cf0589bdacd7e21cc8c83a515c8cea0a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a39a91cf0589bdacd7e21cc8c83a515c8cea0a8d"}]},"branch":"refs/heads/master"},"40452b8c1b7f7b6563e4f9cbe46672e261348483":{"kind":"REWORK","_number":8,"created":"2022-01-13 00:31:02.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/74/823774/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/823774/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/8"}}},"commit":{"parents":[{"commit":"e480c24e177b21bd24511a388095ed51d0e2b2cd","subject":"Merge \"Pre-existing metadata items, set new share metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e480c24e177b21bd24511a388095ed51d0e2b2cd"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 07:36:37.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-13 00:30:37.000000000","tz":480},"subject":"Force create snapshot for invalid share","message":"Force create snapshot for invalid share\n\nSource share status must be available when create snapshot\n\nChange-Id: I14481f5a63990fb6f44dd5dcb40d24b74862a98f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40452b8c1b7f7b6563e4f9cbe46672e261348483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40452b8c1b7f7b6563e4f9cbe46672e261348483"}]},"branch":"refs/heads/master"},"69fcf8fea13a2b5609348fe5e530f7b01d28dd0b":{"kind":"REWORK","_number":9,"created":"2022-01-13 02:49:52.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/74/823774/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/74/823774/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/74/823774/9"}}},"commit":{"parents":[{"commit":"e480c24e177b21bd24511a388095ed51d0e2b2cd","subject":"Merge \"Pre-existing metadata items, set new share metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e480c24e177b21bd24511a388095ed51d0e2b2cd"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-07 07:36:37.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2022-01-13 02:49:24.000000000","tz":480},"subject":"Force create snapshot for invalid share","message":"Force create snapshot for invalid share\n\nSource share status must be available when create snapshot\n\nChange-Id: I14481f5a63990fb6f44dd5dcb40d24b74862a98f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69fcf8fea13a2b5609348fe5e530f7b01d28dd0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/69fcf8fea13a2b5609348fe5e530f7b01d28dd0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
