)]}'
{"id":"openstack%2Fmanila-tempest-plugin~825244","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I63c0bb579fd6636f2bd80280ec6d7319538e982c","project":"openstack/manila-tempest-plugin","branch":"master","topic":"train-tempest-cap","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-01-19 19:23:23.000000000","reason":"Ghanshyam replied on the change"}},"hashtags":[],"change_id":"I63c0bb579fd6636f2bd80280ec6d7319538e982c","subject":"DNM: test tempest 26.1.0 tag","status":"ABANDONED","created":"2022-01-19 03:51:53.000000000","updated":"2022-01-27 16:36:33.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1b8efa984f8ce78c1b9b65b1483a6ba4d397ac8d","_number":825244,"virtual_id_number":825244,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-01-19 06:16:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-19 06:16:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-19 06:16:52.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-19 07:13:45.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-01-19 09:09:55.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"edfdddd9cf0945aa6760ac1fc008875936a9af0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-19 03:51:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5af222a4ff9ffb033b10f14e8593fcc63d899ae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 06:16:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef01d8499fc34217a79db3044627605f : SUCCESS in 4m 43s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5ea0089dc0444b0bbae884b914847a9c : SUCCESS in 36m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1ff55624f10a4c95a09ec64a7df946c7 : SUCCESS in 33m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/337faf04403647a48bf1165eac95f15c : SUCCESS in 1h 13m 50s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/03dd04946d2c479e99e4814419d52e0b : SUCCESS in 1h 04m 05s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b0cd1ff98dfb4b769a24bb6d1b149de2 : SUCCESS in 1h 10m 42s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/cfffd503fba3457aa5be830b6e3b77fe : SUCCESS in 1h 09m 03s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/64dfc57b34ab4569b0e8e3842cf471ec : FAILURE in 1h 17m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cfe9976b7fb14befb2241ce6feca8b46 : FAILURE in 1h 16m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/16616df94fef4f8089fe289f7a289322 : FAILURE in 1h 43m 14s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/51e6ff9f8feb4a5d9a86aea41ee6743f : SUCCESS in 51m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9772ebc3bf40487bb4ea97d55ed0cad3 : FAILURE in 2h 16m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/291bf6938819441da290d76d585395f6 : SUCCESS in 28m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d0168cee60870f5585d91fa696395d65df6bde8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-19 06:16:52.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/44/825244/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/68d8abd/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/44/825244/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/a41fc14/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"502d94c38eed388ce29ec32315aedf299c47372d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-01-19 07:13:45.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/44/825244/1/check/DellEMC_Unity_Manila_dhss_true/d62deeb/DellEMC_Unity_Manila_dhss_true/250 : SUCCESS in 3h 21m 38s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/44/825244/1/check/DellEMC_Unity_Manila_dhss_false/6be03e8/DellEMC_Unity_Manila_dhss_false/245 : FAILURE in 40m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"3952eca4f6bd4ea9e3007a77f67f9437bc656773","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-01-19 09:09:55.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/44/825244/1/check/DellEMC_VNX_Manila/1bc1a7e/DellEMC_VNX_Manila/118 : FAILURE in 2h 53m 06s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"520c9d6704ff264cc4fd58a6a91711771d5eafe1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-19 19:23:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55bb9b54f6526e96ef513e57b128758d9fb4c28c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 23:44:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c17cf3e70b24402192d56065c3d6ab53 : SUCCESS in 4m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f729c663924543e4837846e706ea3aa9 : SUCCESS in 36m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a4bf3fbb1bd64ba786962c30087425b0 : SUCCESS in 38m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/154550a383b34d7fa706daf2f44c881e : SUCCESS in 1h 08m 35s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/672bd6cf0c8a4e05ad1fa29ae3489a61 : SUCCESS in 1h 18m 14s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/69893ca5a8c141c9a8efcdc0d5df5206 : SUCCESS in 1h 13m 40s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/b84b685507ba4b6eb1804409b1356889 : SUCCESS in 1h 16m 23s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/045bc2a4de0e4ff6ad6df4a9881c7b61 : SUCCESS in 1h 17m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c0e3961527ad49a989af315c29809208 : FAILURE in 1h 38m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7686224cedd14ead9171825042dc8bca : FAILURE in 1h 36m 51s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c03a78c108334a62b74e9f9e008f4158 : SUCCESS in 48m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5c728d827183457184b8e9f74defa732 : FAILURE in 2h 10m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a1a8301882304b8098ef89ddba7b3f07 : SUCCESS in 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4d2911811519134867ef153e618574d4a603625e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-19 23:44:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://logs.openstack.netapp.com/logs/44/825244/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/aa88b96/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://logs.openstack.netapp.com/logs/44/825244/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/91359cf/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"bdae6d52ab5746a18134eb71ff7ec24dfa9c7b5b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-01-20 02:37:38.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/44/825244/1/check/DellEMC_VNX_Manila/4ba13cf/DellEMC_VNX_Manila/120 : FAILURE in 2h 53m 06s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"1b8efa984f8ce78c1b9b65b1483a6ba4d397ac8d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-27 16:36:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"17127e2cc9b3388fa3f29ed0b8db5a5ecdc5aec4","revisions":{"17127e2cc9b3388fa3f29ed0b8db5a5ecdc5aec4":{"kind":"REWORK","_number":1,"created":"2022-01-19 03:51:53.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/44/825244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/44/825244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/825244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/825244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/825244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/44/825244/1"}}},"commit":{"parents":[{"commit":"717e0ad258eb3aaa562244753f7131668d0eaa90","subject":"Merge \"Added test test_create_snapshot_over_quota_limit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/717e0ad258eb3aaa562244753f7131668d0eaa90"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-01-19 03:50:43.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-01-19 03:50:43.000000000","tz":-360},"subject":"DNM: test tempest 26.1.0 tag","message":"DNM: test tempest 26.1.0 tag\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/822339\nChange-Id: I63c0bb579fd6636f2bd80280ec6d7319538e982c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/17127e2cc9b3388fa3f29ed0b8db5a5ecdc5aec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/17127e2cc9b3388fa3f29ed0b8db5a5ecdc5aec4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
