)]}'
{"id":"openstack%2Fmanila-tempest-plugin~826747","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I3e7d3a49cec6af347368757f30d4478841dabeaa","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/metadata-for-share-resources","attention_set":{},"removed_from_attention_set":{"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2022-02-24 16:57:47.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2022-02-24 15:40:26.000000000","reason":"haixin replied on the change"}},"hashtags":[],"change_id":"I3e7d3a49cec6af347368757f30d4478841dabeaa","subject":"Remove share metadata test","status":"MERGED","created":"2022-01-27 17:59:59.000000000","updated":"2022-02-24 16:57:47.000000000","submitted":"2022-02-24 16:57:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"826747-bp/metadata-for-share-resources","meta_rev_id":"694f2c13e279b9b7b1193e0360201020a0fda045","_number":826747,"virtual_id_number":826747,"owner":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-24 16:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-02-24 15:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-02-24 02:06:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2022-02-24 15:40:26.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-02-24 15:40:26.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"CC":[{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-27 18:01:26.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-01-27 20:16:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-27 20:17:47.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-08 07:25:13.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"CC"},{"updated":"2022-02-24 15:31:17.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-02-24 15:40:26.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"}],"messages":[{"id":"76a8c9046648f9611312ef19c1dd3727f6019e54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-01-27 17:59:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"712e90b395f884dbbb9513fbbecf81e63769ebde","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-01-27 18:01:26.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6cee1394a5caecc85f239083074fb0b2e0cc65f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 20:16:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/551c70112bd143269e41809d5d0e0405 : SUCCESS in 4m 34s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/80d26ee89c9e426cbb8181a97deb581e : SUCCESS in 37m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0226985264834c89bb38364a7bbb1224 : SUCCESS in 38m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/068673ffa4384fdf95d4c5cd7600570b : SUCCESS in 1h 25m 57s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/7fc8207f297446bc8a244c1d29013849 : SUCCESS in 1h 20m 52s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/5b2dbfbb5f944b838af8a312b1f6b50c : SUCCESS in 1h 44m 48s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/2931d139c64b4e7685b4d8a2ea2233a7 : SUCCESS in 1h 10m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/611abe82025b49c2aecc5f0506477467 : SUCCESS in 1h 21m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8420aaee846947749424c5c4bc5081c6 : SUCCESS in 1h 25m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/875554a71cb44d9eb5cb2591fe60750f : FAILURE in 1h 53m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2f7d03d453af4009ab40ded3acab169d : SUCCESS in 56m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fe922d5a2f5643589e7cfad0019c9a4e : FAILURE in 2h 09m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cca78d8f54d244028eab646909cbf757 : SUCCESS in 28m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a2f1b99549235747cc976e7f8119bab8db373f8","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-27 20:17:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/826747/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/78a11b6/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/826747/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/b74132a/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"4f86c34b758646a98169294fa3dafc1091f3ad1a","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-01-27 21:18:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/826747/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/46e9691/ : NOT_REGISTERED\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/47/826747/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/bef8ffe/ : NOT_REGISTERED\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"49d95a94ed5351dc2c254114f3d122cf589550a1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-01-28 18:36:39.000000000","message":"Topic remove-meta-test removed","accounts_in_message":[],"_revision_number":1},{"id":"6834fea8a396bdfea7c4a3dba621a39db1bc11fe","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-01-28 18:36:53.000000000","message":"Topic set to bp/metadata-for-share-resource","accounts_in_message":[],"_revision_number":1},{"id":"27907ee01ae9db2f8519b72742562e1366519ee6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-01-28 18:37:13.000000000","message":"Topic bp/metadata-for-share-resource removed","accounts_in_message":[],"_revision_number":1},{"id":"2328718b9ed4dc3e0ca66f25d3373807cb6c27f1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-01-28 18:37:22.000000000","message":"Topic set to bp/metadata-for-share-resources","accounts_in_message":[],"_revision_number":1},{"id":"bbc5ea9a34e092ec94957a8a4576555301b8e1db","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-02-08 07:25:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"94a06cc24036df9de12a17e84796d7b3d6bc1aed","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-15 22:32:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bce65efcc2d0b5dc22584f2416f7c4ed7ca9d5fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-02-23 23:31:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"45c22e5e4760c2958403623438ca509687d9548c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 01:46:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a575bd2a7a4446e78845763c57c663ea : SUCCESS in 3m 12s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/54986782f89a4249965f5c71280a7c2f : SUCCESS in 37m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c2292e5a07de4986974d157c77ab22fe : SUCCESS in 36m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/17c0f3ca5b5b45db862cd48b3adcede5 : SUCCESS in 1h 21m 26s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5d94ed1e2eed4cd1a559d7c1b2f53408 : SUCCESS in 52m 24s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/0d78b1ba1157469aa505fee5220c39b9 : SUCCESS in 56m 26s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/5c295a1fe7d14c61ba4a8b9555907a74 : SUCCESS in 1h 00m 49s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/836a802593444e7db18e6cbe183bdc49 : SUCCESS in 1h 25m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/459b1ce680454c56b1b139b9f9e96c19 : SUCCESS in 1h 32m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f0f7503a840d43e79651233253574c17 : FAILURE in 2h 10m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/58b9ae313e684dd09999c4dff47eae9f : FAILURE in 22m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f5c16b7bd18d49b4bdd4b27024c63184 : FAILURE in 1h 55m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c3c9b4d6a05042a3a099981af0c591cb : SUCCESS in 18m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac28ec6bd30117d38919b4576b52c5a3efa00a10","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-24 02:06:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"070dab8d0da9949dd654d43d6495d64d0b379f3a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-24 15:31:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5254157996b04ac2597c0be37f57548989950ee1","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-02-24 15:40:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"894334e038d247e4ee6eb17d18f0cde7145e9999","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 15:44:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f507779bf7f1a8792b51778543179bc5af52033b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 16:57:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d05e58eb55dc497988a27ce407a6f664 : SUCCESS in 3m 26s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f55eb671f14a426680ca17460d1f307d : SUCCESS in 27m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a940953fe4244a2c95f40e81f6f6bae5 : SUCCESS in 33m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8c0e8f303fa34362a15f95d3cfa715f3 : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":2},{"id":"694f2c13e279b9b7b1193e0360201020a0fda045","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-24 16:57:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0cfaa06369acb8d5b42f6fd499653de617313833","revisions":{"13c78ffd0810ebfa6eebf55344af7fc96dfd7d5a":{"kind":"REWORK","_number":1,"created":"2022-01-27 17:59:59.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/47/826747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/47/826747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/826747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/826747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/826747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/826747/1"}}},"commit":{"parents":[{"commit":"73e0b6fdc450b7539bf6a97349f5774177ffb24a","subject":"Added code to show share network neutron_net_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/73e0b6fdc450b7539bf6a97349f5774177ffb24a"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-01-27 17:30:17.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-01-27 17:58:25.000000000","tz":0},"subject":"remove share metadata test","message":"remove share metadata test\n\nRemoves test_try_delete_metadata_with_unsupport_format_key\nfor metadata api changes\n\nChange-Id: I3e7d3a49cec6af347368757f30d4478841dabeaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/13c78ffd0810ebfa6eebf55344af7fc96dfd7d5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/13c78ffd0810ebfa6eebf55344af7fc96dfd7d5a"}]},"branch":"refs/heads/master"},"0cfaa06369acb8d5b42f6fd499653de617313833":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2022-02-23 23:31:06.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/47/826747/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/47/826747/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/826747/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/826747/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/826747/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/826747/2"}}},"commit":{"parents":[{"commit":"88225a342c5ee08368f76459cf12415fcf701e52","subject":"Merge \"Follow up fixes to recycle bin tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/88225a342c5ee08368f76459cf12415fcf701e52"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-01-27 17:30:17.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-02-23 23:17:26.000000000","tz":0},"subject":"Remove share metadata test","message":"Remove share metadata test\n\nRemoves test_try_delete_metadata_with_unsupport_format_key\nfor metadata api refactoring introduced in [1].\n\nThis test relates to this bug [2], but fails in the code refactoring\nmentioned above. Now that the router handles metadata endpoints, the\nstripping of \u0027.html\u0027 or \u0027.xml\u0027 in the key is unecessary.\n\nFor more dicussion see [3].\n\n[1] https://review.opendev.org/c/openstack/manila/+/824648\n[2] https://bugs.launchpad.net/manila/+bug/1774353\n[3]\nhttps://meetings.opendev.org/meetings/manila/2022/manila.2022-02-03-15.00.log.html#l-77\n\nChange-Id: I3e7d3a49cec6af347368757f30d4478841dabeaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0cfaa06369acb8d5b42f6fd499653de617313833"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0cfaa06369acb8d5b42f6fd499653de617313833"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
