)]}'
{"id":"openstack%2Fmanila-tempest-plugin~829758","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ic5527243c0b08dbbfbc849a7c2e79ff927c23098","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-02-27 12:36:09.000000000","reason":"Change was submitted"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-02-27 12:36:09.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-02-23 15:38:49.000000000","reason":"Goutham Pacha Ravi replied on the change"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-02-27 12:36:09.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-02-27 11:04:39.000000000","reason":"Liron Kuchlani replied on the change"}},"hashtags":[],"change_id":"Ic5527243c0b08dbbfbc849a7c2e79ff927c23098","subject":"Add unstable tag to test with intermittent failures","status":"MERGED","created":"2022-02-17 19:21:18.000000000","updated":"2022-02-27 12:36:09.000000000","submitted":"2022-02-27 12:36:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"829758","meta_rev_id":"e2724cd994b2885ae3d23555fcfac454d3605412","_number":829758,"virtual_id_number":829758,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-27 12:36:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-02-23 15:38:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-27 11:04:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-27 11:04:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-17 21:33:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-17 22:29:17.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-17 22:32:54.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2022-02-17 22:32:54.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-02-17 22:32:54.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-02-17 22:32:54.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-02-18 00:04:51.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-02-18 00:47:56.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"28d1300f31c55df0e0d1c534200a33e2d8ad632b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-17 19:21:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09f49e042939f8dadfd632591f19b86cc62e93ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-17 21:33:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cba801a21bfa455799eead201eea1fba : SUCCESS in 4m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/22864fdf11744951ac2b222076a7a7ae : SUCCESS in 30m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/94dd40108dcf4597be1cb7d3fc135802 : SUCCESS in 30m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b35ad4d7b50a470284634f512ceeb0a6 : SUCCESS in 1h 11m 16s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3ccd59c77e4646eb9d6f91cd118047f4 : SUCCESS in 1h 00m 43s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/19ee6318f722450ea85d0cdb8ae7cfd6 : SUCCESS in 58m 13s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/1a49c6e472324f1b9b3fecc7e87372ce : SUCCESS in 1h 04m 58s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e8930deb0fb044b0bb4ba39bdca53ca7 : FAILURE in 1h 11m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ebf5c6df976f47b8903cd5afdc489478 : SUCCESS in 1h 08m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6d1a9938dad24ab08dd3a23e991e3537 : FAILURE in 1h 43m 01s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/99b49c56018c4e87894106db9698b655 : SUCCESS in 43m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ea293893595f4763a28701608aca405c : SUCCESS in 2h 08m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7487a14fb5e6437f82d49b83fcb181f7 : SUCCESS in 18m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af31b87fb68c33732981c3e0e388107b5e904e8b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-02-17 22:29:17.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/58/829758/1/check/DellEMC_VNX_Manila/eb30774/DellEMC_VNX_Manila/152 : FAILURE in 55m 39s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"d2d8f8c0c6f0dd7232a33714778cdd2724946e2d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-02-18 00:04:51.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/58/829758/1/check/DellEMC_Unity_Manila_dhss_true/723284e/DellEMC_Unity_Manila_dhss_true/286 : SUCCESS in 4h 43m 19s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/58/829758/1/check/DellEMC_Unity_Manila_dhss_false/034cceb/DellEMC_Unity_Manila_dhss_false/294 : FAILURE in 16m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2644f5f757a525173cf48620f8c8379a5d6e4a2c","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-02-18 00:47:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/829758/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/64343d0/ : POST_FAILURE in 3h 14m 09s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/58/829758/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/7153cac/ : POST_FAILURE in 1h 53m 48s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"8a1b210b89337efaa0448e9c9c1f062d4b3d1743","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-21 08:46:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ca365345099825de9d82b99ef1a066be8aa1d84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-22 20:27:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6499bf62861f271cd60d720220a63a77f2923fd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 22:24:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84d89fa69d3a408fa91a3015e6db5d34 : SUCCESS in 3m 23s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d34fce03a25e4c6aaca59334968d7f47 : SUCCESS in 26m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/158c5285d7b24ae3aa305054ff6cbb40 : SUCCESS in 30m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/37ddef4c815841c1a3e505873e8619e9 : SUCCESS in 1h 09m 40s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/1ba9ccce571d4aef85e1d810c48c6d5b : SUCCESS in 1h 02m 22s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b6c9ef47c2eb442db9e0e6a704217724 : SUCCESS in 1h 25m 14s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/80f653ac0f3946b0bd0bcb78298cc15a : SUCCESS in 59m 52s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c3d39085bd4a4d9786df545d5c493a31 : SUCCESS in 1h 13m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c96baa56bc1c4464a244014286c862ef : SUCCESS in 1h 24m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1cc1f55e5093494282d7eaba96431bb8 : FAILURE in 1h 42m 46s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ca9961fe4926483ca02082f89f1ff0ef : SUCCESS in 47m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/63e41f0ff468475285d3ce76c0617b5a : SUCCESS in 1h 53m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4570840a3adc4f419147ba1d8457c675 : SUCCESS in 19m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6c9e301b8642578b218af1cc69ec43cf3b9e26d9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-22 22:26:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"005ec23f2506db8649f3e6ed54130c4a317d2100","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-22 22:31:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"40a92162b56538c5d612fac4fa52d5025fdf85ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-22 22:34:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9c034e8e33259e78a38d11003718f5761c30dd1a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-02-22 22:34:48.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2318e9331f98b441d64ec14b9572b53b143226c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 01:05:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8448056da71846ae8a10135365f36602 : SUCCESS in 3m 29s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0303e7cf404c444b8c8da2bfdb83f014 : SUCCESS in 39m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/df3c85bcbd7b4ee1ad83e61ce356b4e5 : SUCCESS in 33m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1f20c380e236409ebbe44438bf7b9ba4 : SUCCESS in 1h 21m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/2df00de3fdce42628b0dbb2f8f11a894 : SUCCESS in 1h 14m 20s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/668af63a9a294df2a9cd6d8ef14ecb0d : SUCCESS in 1h 03m 04s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/df6a21bcffa348a1ac6fa5c0529c0326 : SUCCESS in 1h 02m 28s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ce338ac56bdb44ce97236221a898d159 : FAILURE in 1h 11m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cc52eddaf7274cdaa8af3dfa4d4e2f67 : SUCCESS in 1h 26m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8757334d1090420b99db6fa207c9f584 : FAILURE in 1h 55m 51s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fb4ac143c9bc462bb6cb2ecb5d2614b8 : SUCCESS in 43m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d7254c8f09504da8a3e2cbdd232e68ff : SUCCESS in 2h 28m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ce7f8302c95a479db7691685f6f4cfad : SUCCESS in 19m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2cb7119786b20e8c6acb2d2e34c2305b88a83e9c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-23 15:38:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"39c6daa99d1e94a5735b2b457a895799c0f8a274","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-02-23 15:38:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"837c51929aa49a0aa85cdefb9245d44e0d820cfc","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-02-27 11:04:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d86a7ea1e58e8b8588ad955693925ed2ec2afa2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-27 11:05:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0c1a3a475938270039486acdec9d70d78a5bebeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-27 12:36:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc5c678969dc4ac791e410490cc08a0e : SUCCESS in 6m 01s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e6626ce22fe244b2b9f2de0944e7eec0 : SUCCESS in 27m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/37bc0ee978b34ec9a9261655122ba8f5 : SUCCESS in 38m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c1ca947772294f7096c7c115ec9a8580 : SUCCESS in 56m 55s","accounts_in_message":[],"_revision_number":3},{"id":"e2724cd994b2885ae3d23555fcfac454d3605412","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-27 12:36:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4ca16841cb575aff89ac94cb9a5f256ae475a82d","revisions":{"600c01462fe9a8bc1a86d382ffbd32b6a5564bd7":{"kind":"REWORK","_number":1,"created":"2022-02-17 19:21:18.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/58/829758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/829758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/829758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/829758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/829758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/829758/1"}}},"commit":{"parents":[{"commit":"73e0b6fdc450b7539bf6a97349f5774177ffb24a","subject":"Added code to show share network neutron_net_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/73e0b6fdc450b7539bf6a97349f5774177ffb24a"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-02-17 19:18:36.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-02-17 19:21:16.000000000","tz":-180},"subject":"Add unstable tag to test with intermittent failures","message":"Add unstable tag to test with intermittent failures\n\nThe test \u0027test_create_extend_and_write\u0027 started failing\nintermittently when run under CIFS protocol. This change adds\na unstable test tag to the test to the result will be ignored\nwhen it fails again.\n\nCloses-Bug: #1903922\nChange-Id: Ic5527243c0b08dbbfbc849a7c2e79ff927c23098\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/600c01462fe9a8bc1a86d382ffbd32b6a5564bd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/600c01462fe9a8bc1a86d382ffbd32b6a5564bd7"}]},"branch":"refs/heads/master"},"76bcd471f0d24eec826242a24a179f19e10f7d10":{"kind":"REWORK","_number":2,"created":"2022-02-22 20:27:43.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/58/829758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/829758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/829758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/829758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/829758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/829758/2"}}},"commit":{"parents":[{"commit":"73e0b6fdc450b7539bf6a97349f5774177ffb24a","subject":"Added code to show share network neutron_net_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/73e0b6fdc450b7539bf6a97349f5774177ffb24a"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-02-17 19:18:36.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-02-22 20:27:29.000000000","tz":-180},"subject":"Add unstable tag to test with intermittent failures","message":"Add unstable tag to test with intermittent failures\n\nThe test \u0027test_create_extend_and_write\u0027 started failing\nintermittently when run under CIFS protocol. This change adds\na unstable test tag to the test to the result will be ignored\nwhen it fails again.\n\nCloses-Bug: #1903922\nChange-Id: Ic5527243c0b08dbbfbc849a7c2e79ff927c23098\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/76bcd471f0d24eec826242a24a179f19e10f7d10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/76bcd471f0d24eec826242a24a179f19e10f7d10"}]},"branch":"refs/heads/master"},"4ca16841cb575aff89ac94cb9a5f256ae475a82d":{"kind":"REWORK","_number":3,"created":"2022-02-22 22:34:20.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/58/829758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/58/829758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/829758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/829758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/829758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/58/829758/3"}}},"commit":{"parents":[{"commit":"73e0b6fdc450b7539bf6a97349f5774177ffb24a","subject":"Added code to show share network neutron_net_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/73e0b6fdc450b7539bf6a97349f5774177ffb24a"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-02-17 19:18:36.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-02-22 22:34:06.000000000","tz":-180},"subject":"Add unstable tag to test with intermittent failures","message":"Add unstable tag to test with intermittent failures\n\nThe test \u0027test_create_extend_and_write\u0027 started failing\nintermittently when run under CIFS protocol. This change adds\na unstable test tag to the test to the result will be ignored\nwhen it fails again.\n\nCloses-Bug: #1903922\nChange-Id: Ic5527243c0b08dbbfbc849a7c2e79ff927c23098\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4ca16841cb575aff89ac94cb9a5f256ae475a82d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4ca16841cb575aff89ac94cb9a5f256ae475a82d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
