)]}'
{"id":"openstack%2Fmanila-tempest-plugin~831165","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ib78df90a5e21b88d9088b0b6f349261236662923","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-03-28 19:24:56.000000000","reason":"Goutham Pacha Ravi replied on the change"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-04-21 16:51:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib78df90a5e21b88d9088b0b6f349261236662923","subject":"Remove _create_port and create_keypair methods","status":"MERGED","created":"2022-02-28 09:42:53.000000000","updated":"2022-04-22 03:23:12.000000000","submitted":"2022-04-21 16:51:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831165-tempest-scenario-manager-cleanup","meta_rev_id":"a74b303fb40f685d1c4afd5778d5f5a305727c24","_number":831165,"virtual_id_number":831165,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-21 16:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-04-13 19:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-03-29 18:41:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-15 21:01:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-04-13 19:02:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-28 10:30:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-01 14:03:37.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-07 08:18:09.000000000","updated_by":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-03-15 21:01:27.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-03-29 18:41:22.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-03-29 21:37:30.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-29 21:38:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"cf0fd7f2288828463f0f14cfb8f22945cb37c9f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-02-28 09:42:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42a2b29925192d90980f5a509eb54c0a9e60560a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 10:30:11.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bffdc7ea71fe491b8986350892ec8ba6 : FAILURE in 4m 34s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7456716b402342a3b371d04328ac4433 : FAILURE in 17m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cabf3b31209c4bd48bc92ce1e3867191 : FAILURE in 14m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0f0e9e169f484745a801916cb4fb27b1 : FAILURE in 36m 16s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/33dae732770545ae8baefe6ceee762bf : FAILURE in 33m 32s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b0945282923c4bba876cc4c465532344 : FAILURE in 37m 47s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/9eecefc4667441c8a21c1ae963f25bf8 : FAILURE in 36m 11s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/14aae9bf03ba4d33aa9d79470ffcbc93 : FAILURE in 30m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/52141a16deaf4334897d5e6cd936259d : FAILURE in 37m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6750d8e8ccd2461b90f7eb4756e53250 : FAILURE in 41m 34s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7505443c475f4137a2b6577ac6506790 : FAILURE in 33m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6309f0b5f6034e44a6aee7e0fe6e71ba : FAILURE in 38m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6c7b927a74104359a5158e0de1c250ea : FAILURE in 24m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecfaae794d05317e4f3938e3d7b2fc83d74f0732","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-01 08:57:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6ba1cc31d252a0079dcf90e8d5b9bc6ff4babc79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 11:10:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb79762aa33a49fdb7b043e274d63d9c : SUCCESS in 4m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d9e8b8b575c3492e96e4caaba1af6688 : SUCCESS in 41m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/07c750971dac43419c8e2226b9fbf59a : SUCCESS in 47m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bcea6df3a2a44f0b8a9139a94e5e6ab5 : SUCCESS in 1h 20m 21s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d176edf4eaee4c7b885a4e5a4d75c594 : SUCCESS in 1h 12m 17s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/68857966a1544913876a9d2fc2721c20 : SUCCESS in 52m 46s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/67ad9b23c0c14f6288cc7415eabe2008 : SUCCESS in 1h 19m 23s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a96672efed2141b1ac569295ac1a5675 : FAILURE in 1h 28m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/db97c623b2cf428dac380816c56b29fe : SUCCESS in 2h 06m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/203fbe2b7fbd44e7a78bd785f63f7a62 : FAILURE in 1h 53m 46s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/74c6df05887a49c486c19fd4be904468 : SUCCESS in 1h 00m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0e584ce2c90c432f83e113295fa79295 : FAILURE in 1h 51m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/53d08f5ea1b542758e89956168631daf : SUCCESS in 28m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7cc86b2e040ffa1a5f926445e42325fdf46f3a1","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-01 14:03:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/831165/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/09a39a0/ : SUCCESS in 2h 52m 34s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/831165/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/1da6776/ : SUCCESS in 2h 01m 40s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"374466f5f31fdd3c7727b87ef829e81e837aba48","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-02 10:45:26.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"ba744306ad2dbe7aaf9e92bc3f05aca88b87b8ef","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-07 17:29:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fba7793b055affb7e0ad9833348c3d7d8905809a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-08 10:05:50.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e5aa6dd2be4e45cb7422c65aeab2223a03e238d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-08 10:47:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4d8faa06e75c338f5dec015223cf8176873dab0a","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-08 10:55:28.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"444ed48d5c9bbef126732153ff086d9f54e0c70c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 13:12:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b14633cde2e405bb007d38c584f96c3 : FAILURE in 4m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8d730a32e1984027b1e7a9dfa685c432 : SUCCESS in 38m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/14e539999513431a8125804bde408aa1 : SUCCESS in 39m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/80dba8511ee74a33bb7c0d8c58963e70 : SUCCESS in 1h 23m 53s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/48ac1c49f3604ab1848b779a4d3ccd51 : SUCCESS in 1h 13m 38s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e1e34bb04d5c486daac1eb29a260ebca : SUCCESS in 1h 49m 54s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/ffdc1d01d88c44ecaeee7e68b07471f7 : SUCCESS in 1h 21m 07s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/451bb7a90d394d17b389bac47e6cdda4 : SUCCESS in 1h 27m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c61d447ae2654f78a031d497dad90408 : SUCCESS in 1h 34m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/89d205a53b1e4f8a8de703f513a30cd3 : FAILURE in 2h 16m 56s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8d82f09f1a3445ef958196453c8ccbdc : SUCCESS in 1h 19m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c20deb6541464c1c812ff17961c994dc : FAILURE in 2h 18m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1c40a433d10e4d37ba9f8c690a7163b7 : SUCCESS in 31m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5e9b99429cf200cb7bd4361c6b75b65ace93b12a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-08 14:18:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f8d450424c0c96f34e59f94012f0b0789ad1cb98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 16:29:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/667d157af4a14474abbfa2412235a49d : SUCCESS in 5m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/700349fec8444415ab4d1a21c5daa416 : SUCCESS in 38m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9b0ef3e326c541c3a88056bd1c2ecc2d : SUCCESS in 32m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5828510b761147aeb8b5c38ac751a5ee : SUCCESS in 1h 23m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/983416354af04ba1910fd4dc898ce4a0 : SUCCESS in 1h 21m 54s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/20ea130406b5433ab13222682e880b46 : SUCCESS in 1h 25m 19s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/868585c2960a43adb02e14b0d1a24abd : SUCCESS in 1h 15m 04s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4ef506bb8260456597c8d1eea78dfc65 : SUCCESS in 1h 34m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0ceb871e58424a19b80123200fe36856 : FAILURE in 1h 35m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/59e8a6f547dd4f15916cd171b89a4cba : FAILURE in 2h 06m 18s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7abdd46f549344fbbfedcab9f239400f : SUCCESS in 57m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/68b34a34219149908cafaeeb38f19729 : SUCCESS in 1h 56m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d6ac718021c74147bfb2afc80f056d9c : SUCCESS in 28m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"216bd6f606a520c70ae009d848d81009331a5d19","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-08 18:53:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"556b288f7f9c76ad63f158803b387150698a1355","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-09 07:01:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5af25d6114bb434e0b88d77eca1d9f77028edcb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-09 07:31:00.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent, and commit message as Patch Set 6.","accounts_in_message":[],"_revision_number":7},{"id":"060535e262d6ed83239312e4f1dd09b6e05e88ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-09 07:44:33.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent, and commit message as Patch Set 7.","accounts_in_message":[],"_revision_number":8},{"id":"e2bff68c8f0bd468c2a69aa5a6c169b20f8f4bd7","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-09 08:13:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4e87919dfaa42e5dd08d43df0e6b298f574826d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 09:58:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5e741ca63d4ab18d58cc978804131f : SUCCESS in 4m 37s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/757da95bfdd8452387e930e50c2cb466 : SUCCESS in 31m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/604fdcb936ce408582646cc8aa97494f : SUCCESS in 31m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f5e4904014984fd78b8936cb0f81deda : SUCCESS in 1h 15m 31s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/ae3f5de5e85b4ad1af24f8830b43f908 : SUCCESS in 1h 02m 46s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e423d111de6b418aa4d35802dfa25761 : SUCCESS in 1h 08m 02s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/0f6876886585408f8921704f3e681b36 : SUCCESS in 1h 07m 38s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/50d3e930727e408195396a740ebb1a7a : SUCCESS in 1h 11m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0ebed74138644240bc47af6797c5c8c7 : SUCCESS in 1h 21m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/57d17b616160454284c73620b9342c5b : FAILURE in 2h 12m 04s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/13e4d7b41f2d472a8cee5fca9d4620f8 : SUCCESS in 1h 00m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f2c15e54f9384b44b84348228fef2e98 : FAILURE in 2h 00m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3f6621bfc543456180fd770c21e70290 : SUCCESS in 24m 49s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"392faaed7fdc2836bf3c0246b4bf3247f22a2c8d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-03-15 21:01:27.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0f879325a87f04ce3f733fe55f977522d5bae9f1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-28 19:24:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"458f051af5e3eafe018e5bdb22cc2c8e58171298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 00:53:30.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a49653a3445e4a0ba12eebcfb75dd350 : SUCCESS in 5m 39s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/28146b755d274188b02b481cfd5970de : SUCCESS in 42m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/622dbb6bed304b5da951c431db7ccb55 : SUCCESS in 39m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8add2362c3f048b4814d941fffbbe9ac : SUCCESS in 1h 23m 48s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/4517b655ec7e45229424c90d4a2cd9ae : SUCCESS in 1h 00m 47s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/edf4c932e478466783ff716cd8183072 : SUCCESS in 1h 07m 40s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/c8972b4e8b49432ea3e9b07a7370c444 : FAILURE in 1h 12m 59s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/84bdfe78d58b4071844526afdcb7a50f : FAILURE in 1h 19m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a0b05bcfc59f4fc5b6c91e6caae3f458 : SUCCESS in 1h 19m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b10b35f08c6043f2b075f1f1ae531bf7 : FAILURE in 2h 09m 11s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a64077555b0e4a34a2bfc449805f906c : SUCCESS in 52m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7673abae40ea4da7a28585fd880f7e83 : FAILURE in 2h 28m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2459962306004cc1aa3c7620cb554bfc : SUCCESS in 25m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4de3a196afc396049a9d23998229398015a0df78","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-03-29 18:41:22.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"45d161151b0fd878800d71ef779b76935e83be0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 21:34:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a2eb51b62f3491084fbd136ca239f2e : SUCCESS in 4m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9a04973195ce462a8d5a1fd6f0cdb4be : SUCCESS in 32m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d81570810a574b61b877a88aa7d6063b : SUCCESS in 40m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b48d36dae2bd4836872e7a372967867e : SUCCESS in 1h 31m 49s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/868e18cfd1324d579971dabd71141e4f : SUCCESS in 1h 09m 16s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/7a432fb8d9824351a9f45380e2d279eb : SUCCESS in 1h 04m 22s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/125841b6433949c1a9f5fbe361c59d35 : SUCCESS in 1h 19m 29s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6a34e81070fd4202910fcf4f9722bfa4 : SUCCESS in 1h 27m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f290dfc83a934f8ea4216712c9a050de : FAILURE in 1h 36m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1db0a25fd0424db2b9bcfcf6de347c21 : FAILURE in 2h 00m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/28972e51fd7446c1b1b2e22ee747724f : SUCCESS in 1h 00m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a1f02f77b5fd4b18a54d6c1b398ce20d : SUCCESS in 2h 04m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c07287c00d5345d28ae148eb1e4dc7a8 : SUCCESS in 24m 26s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"261a1928f39c7d17913113a45ab32f2d80987ef1","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-29 21:37:30.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/65/831165/8/check/DellEMC_VNX_Manila/38b5f4d/DellEMC_VNX_Manila/206 : FAILURE in 2m 28s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"180de26e6ec1d62df49be3294d96371a9f679d36","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-29 21:37:53.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/831165/8/upstream-check/manila-cDOT-manila-tempest-plugin-ss/166af26/ : FAILURE in 2m 44s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/65/831165/8/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/e9c1f12/ : FAILURE in 2m 47s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"f2dda2173c361a22b7a84a9e237fb9cd29811b3e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-29 21:38:30.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/65/831165/8/check/DellEMC_Unity_Manila_dhss_true/d24bb59/DellEMC_Unity_Manila_dhss_true/328 : FAILURE in 2m 57s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/65/831165/8/check/DellEMC_Unity_Manila_dhss_false/77105a8/DellEMC_Unity_Manila_dhss_false/334 : FAILURE in 3m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"d2df72f4c3d15e37a0a683bffde70f30ffdea557","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-04-13 19:02:03.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"05e5e0b5ebc4298e0ddeb4dc8f9dd1549e174220","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 19:02:46.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"7d5960ab9a5a927516e3f6a32d039bafe7c8f93a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 19:24:17.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/700dc40439634caeb70cbd619d25b9bf : SUCCESS in 3m 38s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b3bc541d00b34026b85e9492b9682fe5 : FAILURE in 17m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/eba4db352ce24f22bad848a9970c8323 : FAILURE in 10m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a4ff2775f8ec4f389ebd5858012b74a4 : FAILURE in 12m 08s","accounts_in_message":[],"_revision_number":8},{"id":"b496bf494b30085c04a223cc87b34fbe236119d4","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-20 06:21:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d6833daeae67987152e41c90f16bea84ac06b89d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 08:11:19.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/299ad793e62d448a8e066bd1883a1466 : SUCCESS in 3m 26s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6f66b6c13dc642f1a4480e183289d544 : SUCCESS in 38m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/facdf7d7281a4801b256a7e98ebc5153 : SUCCESS in 37m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/786869bdb72246a0bac6ddb05ba5061d : FAILURE in 33m 19s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e472f3f9e9954924a02e2906fcca5806 : SUCCESS in 1h 10m 10s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d7979a00fe104abc8699dfdddf0103a9 : SUCCESS in 1h 08m 33s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/b2ae2cb7c1644d178863321c8e217901 : SUCCESS in 1h 08m 55s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a9d22bb84bc7452ca97e5514475c5494 : SUCCESS in 1h 16m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b1afd9082bdf4fc58acd499264e7e44f : FAILURE in 1h 42m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/85d97f357d4f4fe7a634ec5f4cc48032 : FAILURE in 35m 40s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2908943f6113485cbe82c5ee6c2ef101 : SUCCESS in 1h 04m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d2ef8c26a5a443dfad94c4d14a4cbe81 : FAILURE in 1h 32m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d60120b2e4e243ce9ebbb0c6efe8c2fc : SUCCESS in 26m 33s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"67ac8c1041c382fdabfcbe5a3324d6c992ec5c7f","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-21 10:16:49.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5b710c1236b1542f136ffb32e5bc06159284dacd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 11:05:35.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12743e54ceb34cf8aef8ccad7f3f33fc : SUCCESS in 3m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5d0ad982eef44401abfc5eb96940a8dc : SUCCESS in 35m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3733cc93ac9f4451a37cb7c032e70a53 : SUCCESS in 41m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8d7aa266c4864663aa0a92329bbd897a : FAILURE in 35m 01s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f032fccb68dc43ceb1a6d82f59b2c773 : FAILURE in 34m 06s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/64588c40e3df43f685f271f321f9532c : FAILURE in 37m 41s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/437618d6edd042ee8a719e65506231fe : FAILURE in 35m 17s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a655190512614c6a8594a2b14a1cb0b8 : FAILURE in 37m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/76b659c8fde24ddaa5362cf12c778156 : FAILURE in 35m 23s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/063c49bbd23343e8922871e84f480db9 : FAILURE in 36m 46s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/febb76dceb744f52815d62d2f6c01eb0 : FAILURE in 40m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c3168a5b22d74a04a59b9ca3a2ae22eb : FAILURE in 38m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f16d6803cc46472ea6bf8d5b7740d57c : SUCCESS in 27m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b0108c976df22ab2c2c241e9c1bb80b7b9230ad9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-04-21 13:07:54.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f78f85c83a37d5372cce8f78cd30cff7715351ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 15:38:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aaab52f4498451e936c6aec60bb4c1e : SUCCESS in 3m 21s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4ce0fdfe337e45c5bcfd3fb64f9b0abb : SUCCESS in 28m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8d4079927a2f41488c679f45ccc639bc : SUCCESS in 35m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/549c937b9b2547599e8dc2d1b0b4296a : SUCCESS in 1h 21m 14s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/dc2e470e6d1e4bebb0135a55b544fb5b : SUCCESS in 1h 05m 07s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/42de3fdabb834c279bec9c54627d7fcd : SUCCESS in 1h 16m 17s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/6c4aad3ef65b4d638261ee7cf7fbc92c : SUCCESS in 1h 06m 21s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dea8e713d2f740989fa0bbdbe0e9c5ce : SUCCESS in 57m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/945d94bbfd7b405ab4d8b6516f213fe7 : SUCCESS in 1h 32m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4f39164b5feb4f62bb1474086f25c3a2 : FAILURE in 1h 56m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c7858e1e7f57487f9346cb7ee47bde73 : SUCCESS in 1h 11m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5c23bb04b09c48978213cc13925fc3c6 : SUCCESS in 2h 27m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/aa8a8cbf660c46429c58e408df8a8d22 : SUCCESS in 30m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"092e5f57df99d9fb944cb208cb261b655ba959e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 15:39:48.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"12baf828cedca9899092c423db57e247bd4a8145","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-04-21 15:40:22.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/65/831165/8/check/DellEMC_VNX_Manila/44c0b6b/DellEMC_VNX_Manila/215 : FAILURE in 1m 11s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":8},{"id":"0fc7c8cb70d853cd5143e88b9d78020d0315237a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 16:51:02.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/739d23996fa3450abc06c187d46e729a : SUCCESS in 3m 43s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7e30ded2fc8c4d97843416441ede0b81 : SUCCESS in 40m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d31dd0718c204d61af3a33daad8a4eed : SUCCESS in 34m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4905b47e660642b0ae9e0a1e632a77b8 : SUCCESS in 1h 10m 29s","accounts_in_message":[],"_revision_number":8},{"id":"dd234383a5974311e0d0ab9712fc05419162301b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 16:51:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"a74b303fb40f685d1c4afd5778d5f5a305727c24","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-22 03:23:12.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/65/831165/8/check/DellEMC_Unity_Manila_dhss_true/c7740c2/DellEMC_Unity_Manila_dhss_true/405 : FAILURE in 4h 34m 27s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/65/831165/8/check/DellEMC_Unity_Manila_dhss_false/b87739f/DellEMC_Unity_Manila_dhss_false/424 : SUCCESS in 1h 24m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"290ef29ae05833f26e0c3eb5225135cb3e3d0b25","revisions":{"8aafdb604dc85f8b8fc4b54b884938ea759843d1":{"kind":"REWORK","_number":1,"created":"2022-02-28 09:42:53.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/65/831165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/831165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/1"}}},"commit":{"parents":[{"commit":"a825bea637683e78f3eac91b59f361ca5e6b9a04","subject":"Merge \"Add unstable tag to test with intermittent failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a825bea637683e78f3eac91b59f361ca5e6b9a04"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:41:04.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:42:51.000000000","tz":60},"subject":"Manila scenario manager cleanup","message":"Manila scenario manager cleanup\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_port\n\t* create_keypair\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ib78df90a5e21b88d9088b0b6f349261236662923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8aafdb604dc85f8b8fc4b54b884938ea759843d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8aafdb604dc85f8b8fc4b54b884938ea759843d1"}]},"branch":"refs/heads/master"},"f7b616fd76ba458f42706f1b739b0328a2ad6956":{"kind":"REWORK","_number":2,"created":"2022-03-01 08:57:45.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/65/831165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/831165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/2"}}},"commit":{"parents":[{"commit":"a825bea637683e78f3eac91b59f361ca5e6b9a04","subject":"Merge \"Add unstable tag to test with intermittent failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a825bea637683e78f3eac91b59f361ca5e6b9a04"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:41:04.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-01 08:57:13.000000000","tz":60},"subject":"Manila scenario manager cleanup","message":"Manila scenario manager cleanup\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_port\n\t* create_keypair\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ib78df90a5e21b88d9088b0b6f349261236662923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f7b616fd76ba458f42706f1b739b0328a2ad6956"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f7b616fd76ba458f42706f1b739b0328a2ad6956"}]},"branch":"refs/heads/master"},"3d3399b9aead16db795c030f2880db6934af4339":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-03-08 10:05:50.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/65/831165/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/831165/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/3"}}},"commit":{"parents":[{"commit":"a825bea637683e78f3eac91b59f361ca5e6b9a04","subject":"Merge \"Add unstable tag to test with intermittent failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a825bea637683e78f3eac91b59f361ca5e6b9a04"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:41:04.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-08 10:05:50.000000000","tz":0},"subject":"Remove _create_port and create_keypair methods","message":"Remove _create_port and create_keypair methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_port\n\t* create_keypair\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ib78df90a5e21b88d9088b0b6f349261236662923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d3399b9aead16db795c030f2880db6934af4339"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d3399b9aead16db795c030f2880db6934af4339"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2e0eff6b3ee83aefb0e1ce3d4aaec0d80b73e2c6":{"kind":"REWORK","_number":4,"created":"2022-03-08 10:47:49.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/65/831165/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/831165/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/4"}}},"commit":{"parents":[{"commit":"a825bea637683e78f3eac91b59f361ca5e6b9a04","subject":"Merge \"Add unstable tag to test with intermittent failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a825bea637683e78f3eac91b59f361ca5e6b9a04"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:41:04.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-08 10:47:27.000000000","tz":60},"subject":"Remove _create_port and create_keypair methods","message":"Remove _create_port and create_keypair methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_port\n\t* create_keypair\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ib78df90a5e21b88d9088b0b6f349261236662923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e0eff6b3ee83aefb0e1ce3d4aaec0d80b73e2c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e0eff6b3ee83aefb0e1ce3d4aaec0d80b73e2c6"}]},"branch":"refs/heads/master"},"78173fe2be8e4f8adfa17c039d4a96c8a99384f2":{"kind":"REWORK","_number":5,"created":"2022-03-08 14:18:15.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/65/831165/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/831165/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/5"}}},"commit":{"parents":[{"commit":"a825bea637683e78f3eac91b59f361ca5e6b9a04","subject":"Merge \"Add unstable tag to test with intermittent failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a825bea637683e78f3eac91b59f361ca5e6b9a04"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:41:04.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-08 14:17:57.000000000","tz":60},"subject":"Remove _create_port and create_keypair methods","message":"Remove _create_port and create_keypair methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_port\n\t* create_keypair\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ib78df90a5e21b88d9088b0b6f349261236662923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78173fe2be8e4f8adfa17c039d4a96c8a99384f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78173fe2be8e4f8adfa17c039d4a96c8a99384f2"}]},"branch":"refs/heads/master"},"99defc70fff73a94aa123b3c63b86fdc70106c0c":{"kind":"REWORK","_number":6,"created":"2022-03-09 07:01:54.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/65/831165/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/831165/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/6"}}},"commit":{"parents":[{"commit":"a825bea637683e78f3eac91b59f361ca5e6b9a04","subject":"Merge \"Add unstable tag to test with intermittent failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a825bea637683e78f3eac91b59f361ca5e6b9a04"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:41:04.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-09 07:01:06.000000000","tz":60},"subject":"Remove _create_port and create_keypair methods","message":"Remove _create_port and create_keypair methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_port\n\t* create_keypair\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ib78df90a5e21b88d9088b0b6f349261236662923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/99defc70fff73a94aa123b3c63b86fdc70106c0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/99defc70fff73a94aa123b3c63b86fdc70106c0c"}]},"branch":"refs/heads/master"},"4b30892d6b20dc786c08a59f71120b9534fcd454":{"kind":"NO_CHANGE","_number":7,"created":"2022-03-09 07:31:00.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/65/831165/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/831165/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/7"}}},"commit":{"parents":[{"commit":"a825bea637683e78f3eac91b59f361ca5e6b9a04","subject":"Merge \"Add unstable tag to test with intermittent failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a825bea637683e78f3eac91b59f361ca5e6b9a04"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:41:04.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-09 07:30:43.000000000","tz":60},"subject":"Remove _create_port and create_keypair methods","message":"Remove _create_port and create_keypair methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_port\n\t* create_keypair\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ib78df90a5e21b88d9088b0b6f349261236662923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b30892d6b20dc786c08a59f71120b9534fcd454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b30892d6b20dc786c08a59f71120b9534fcd454"}]},"branch":"refs/heads/master"},"290ef29ae05833f26e0c3eb5225135cb3e3d0b25":{"kind":"NO_CHANGE","_number":8,"created":"2022-03-09 07:44:33.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/65/831165/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/65/831165/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/65/831165/8"}}},"commit":{"parents":[{"commit":"a825bea637683e78f3eac91b59f361ca5e6b9a04","subject":"Merge \"Add unstable tag to test with intermittent failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a825bea637683e78f3eac91b59f361ca5e6b9a04"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:41:04.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-09 07:43:40.000000000","tz":60},"subject":"Remove _create_port and create_keypair methods","message":"Remove _create_port and create_keypair methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_port\n\t* create_keypair\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ib78df90a5e21b88d9088b0b6f349261236662923\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/290ef29ae05833f26e0c3eb5225135cb3e3d0b25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/290ef29ae05833f26e0c3eb5225135cb3e3d0b25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
