)]}'
{"id":"openstack%2Fmanila-tempest-plugin~831168","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I0c2fa6d195e1e31ed077bdc4c4e21fa6758dfddb","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-03-29 16:29:00.000000000","reason":"Goutham Pacha Ravi replied on the change"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-04-21 17:18:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c2fa6d195e1e31ed077bdc4c4e21fa6758dfddb","subject":"Remove create_server and rebuild_server methods","status":"MERGED","created":"2022-02-28 09:59:45.000000000","updated":"2022-04-21 17:18:33.000000000","submitted":"2022-04-21 17:18:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831168-tempest-scenario-manager-cleanup","meta_rev_id":"57c9381fb71466301ab27606d13283e8b9fb72eb","_number":831168,"virtual_id_number":831168,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-21 17:18:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-04-20 18:27:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-03-29 16:29:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-03-21 12:36:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":1,"date":"2022-03-29 18:43:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2022-03-15 21:01:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-04-20 18:27:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-28 10:49:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-01 14:44:49.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-07 08:18:30.000000000","updated_by":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-03-15 21:01:45.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-03-21 12:36:40.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-03-22 00:21:47.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-29 18:43:33.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-04-20 18:27:52.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"d9d369f8a5b942103f0820103f55bffea51ff459","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-02-28 09:59:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7d34031f8bb446d047ae67a25d8c4f37133a10d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 10:49:52.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/291a80395c4447ce9158619bb1506e43 : FAILURE in 4m 52s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2e0b215995ae4ff1b5e4610372f79a31 : FAILURE in 24m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/be856aecf8af4bd49d7d6c7746497c44 : FAILURE in 24m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b1680dfbeeb54cf39f64e33f6dd1e65e : FAILURE in 39m 30s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/df98ad7497b3498da12b68de4c7911cd : FAILURE in 36m 25s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/463823f65ce64422be484a4ee52b7d18 : FAILURE in 40m 17s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/996bff0c777047d1b450fd2659590f25 : FAILURE in 37m 23s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/46d928b38aaa41ffb119f00a9f633237 : FAILURE in 32m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3c4edd08bc0f4685b9c9490dbba4ab9d : FAILURE in 36m 37s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a38febcd02244d089d5a867435935d82 : FAILURE in 26m 35s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/98bf84187f604ab5876febfc7dd8dda4 : FAILURE in 33m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bb00507e5f0f4bed8b73d8206b8ae796 : FAILURE in 42m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/335219bd15664c5ab5a6f7d857c4b471 : FAILURE in 20m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aecee6e977006942da94f24cd5d1e603b981e306","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-01 09:01:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"88afc8ddfd2907dd0ef81d49ba39c659947048c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 11:43:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93521117e8c04e10b655478c5566e762 : SUCCESS in 4m 58s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b1828eb774b04e48acea139c04eb6323 : SUCCESS in 37m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6aa27fc1f1f84201a51b39744bec2cc0 : SUCCESS in 37m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ba3516dc7864427abbb8e3a9867cae16 : SUCCESS in 1h 26m 22s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a7d0a28944a64e4c9196fb2e7a457b99 : SUCCESS in 1h 03m 06s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e51f6d5793354dc5a2b1b3e9f1657d87 : SUCCESS in 1h 36m 15s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/327c235e328946db96a38ba2c62548bf : SUCCESS in 1h 53m 58s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3f80d2633faa48d7a45a438ba545f8bb : SUCCESS in 1h 22m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5e60ee2d1df543ba83ba6eb4c23ea35d : SUCCESS in 1h 28m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e0f6498ef7064909943be4ae15f135e0 : FAILURE in 1h 46m 17s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b4e0b1e217224030b7d8c8ce7ded2e38 : SUCCESS in 1h 02m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/29fd4a2329374c5b9a5442d2c9723e1d : FAILURE in 2h 35m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/312273f077c848ee914a7d35af08ce18 : SUCCESS in 31m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df4f1a53be34c9684010f72a4a7de356711e9c1d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-01 14:44:49.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/831168/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ce83bc7/ : SUCCESS in 3h 00m 51s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/831168/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f073c4f/ : FAILURE in 6m 40s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"705291cf7ab2883e19a47e8e22344e21194f011b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-02 10:29:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9b5d50a0d64728aeff33e9500fcede4ca1c6995b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 12:27:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a10e137a0484ddf8e1f81f4f1bd9c21 : SUCCESS in 5m 39s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4a1ef7ff317144a790896051744beadb : SUCCESS in 34m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/52bda4336481444f99e792f9494dd4bd : SUCCESS in 35m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4b80b8dbfb734abbbcce95956b5d247b : SUCCESS in 1h 10m 00s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d280a6e0e6494c5a9947e6f8085c6a4a : SUCCESS in 1h 01m 07s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/cc3796eb5a4946c49d181790136f76f9 : SUCCESS in 1h 30m 32s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/4bd0b4cf26e14201a64ca25ba7b80cfd : FAILURE in 1h 17m 51s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fab6bf69f2854fa1912fc1c65ce33854 : SUCCESS in 1h 15m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/600f01266cbe43eaa47a521e137fc57d : SUCCESS in 1h 51m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0f838cee5b28457b88490dec0e672c3e : FAILURE in 1h 36m 07s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bd4adf13bb66425abab1dcf2d4407979 : SUCCESS in 45m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6b68a98b1fdd42ac9ba87aa8de419095 : FAILURE in 1h 46m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ecb60c2553384b119be10bb1aed563fd : SUCCESS in 34m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a6cfb93fff0c1551367400db58cc7048c96028a6","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-03 11:30:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"da6f7384f74243ed4c96c6cecfeaa1ab796fd4a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-03 13:53:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5bd91cef2d6425faa57266f5819e789 : SUCCESS in 5m 13s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/731f69e505c946e2aa37fa15c25a7bbb : SUCCESS in 40m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c954a1b204294edd9919af9d897e5c95 : SUCCESS in 40m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b3ffa6b6be67417dbb77523ce38c223c : SUCCESS in 1h 19m 46s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/01c871f832a54fa093714eefff2f3a5c : SUCCESS in 1h 30m 14s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/f6da470b0224476589d3ae457b0ae535 : SUCCESS in 1h 29m 47s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/c2ab27ce51834cc398ff3a2b29951147 : SUCCESS in 1h 28m 38s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2d364d7814d44435b1baf823f0c14d8e : FAILURE in 1h 10m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/95996dd4a5754fddb684013250c3c80a : SUCCESS in 1h 50m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1b4964557aa94e55801353f71430af9a : FAILURE in 2h 16m 22s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b3b8459188a74133a084ea0d502780d5 : SUCCESS in 54m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/025ccdcfa1cb479f9637da4401be87ea : FAILURE in 1h 40m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/dd8de12d70464a338f9e268ada19a648 : SUCCESS in 32m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3c842872d69297f099709f6837f309af7c09b263","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-03 13:54:31.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"45854e67dc071d6b0af947294edddb7c79036ea7","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-03 16:51:30.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/831168/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e3504dc/ : SUCCESS in 2h 58m 09s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/831168/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/26ccb44/ : SUCCESS in 1h 54m 11s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"ebea0cf84d114f85387d21b362d70dc582bb656a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-07 13:22:21.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c991f277c5b02b0735a3b7a47b8e345467cf827a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-07 15:27:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/039dda07c9914fa6ae4b35b91b485933 : SUCCESS in 5m 09s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c98596f23e1f491c9decc73eff06e79b : SUCCESS in 28m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3f1a43db3a604fc2ab73b7b82761e387 : SUCCESS in 36m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/78a2272cca714ac392fc86ddb5ae4321 : SUCCESS in 1h 21m 50s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d4847ff6035f4c4ab01f036567fc029d : SUCCESS in 1h 04m 52s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e0df738b85784500877772d27555630d : SUCCESS in 1h 14m 44s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/f3be6018c052414eae51d5d8d5772b7e : SUCCESS in 1h 13m 30s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a7df9b5dec274c0dacc621b704d6d383 : SUCCESS in 1h 09m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ffe485594e32416c9390283719b7e018 : SUCCESS in 1h 22m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/51d329045f28400ca07b71c2adfb10a8 : FAILURE in 1h 48m 10s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7948b1499d03400b85b36015ec2fb673 : SUCCESS in 46m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/da37d6ac0b3841cc8d7885f12d796da6 : FAILURE in 2h 00m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ec3a5cf2bc694c8182dd8885e86b0370 : SUCCESS in 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"634317eab835ac44ddaef4a5c5818d235468967d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-08 10:05:14.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9c85d2720083b18183fcff530929c92631ea140f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-08 10:41:33.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"7d063fab87717fa0b156b347f92bc58a539bcab3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-08 10:51:41.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"f53cfa52b475785e99972987af02fa0684932d21","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-08 10:55:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c3f25647868c4ce01722be0ebb028e9803a50c69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 13:00:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a14836cdf0a341fb91e393d6472a85bb : FAILURE in 5m 32s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4de333f07c1c492bbca862536a80105c : SUCCESS in 31m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3d562a9a9b4e4463956617c1a4fd7046 : SUCCESS in 32m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6c9bed9f2fd64afda7330fe6bd338f35 : SUCCESS in 1h 20m 06s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/46eb880295db4cb3931b4d52a575c669 : SUCCESS in 1h 01m 53s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/caf8eae3df3640c5b40ca15310985194 : SUCCESS in 1h 10m 24s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/30194f56ff9e4a2490ef23b3142e2381 : SUCCESS in 59m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/92e1ab79bfe04119bd880bf4ce1c67f9 : SUCCESS in 1h 16m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d9f4e310c08b41a4960cbf5c43888113 : FAILURE in 1h 21m 37s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/57e7475e5bfe45fa955ad5cabc4ad817 : FAILURE in 2h 06m 08s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1cf476c45d524989b191f4f9bb7a6125 : SUCCESS in 53m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e3eb998cb9634b758b913636d7d4e3c8 : FAILURE in 2h 00m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9442f6dce74748e2b1fcd362bcfc2087 : SUCCESS in 23m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d6168d219fd69de3b3716ba7abc523c28daf04e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-08 14:20:53.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"3fbc1e20037ba91d1d0041fb78922e7de11967aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 17:33:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7730ca6b65cd4e75b609e441b7fa01e7 : SUCCESS in 5m 40s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8ccff502ce654384983b266d411ac8b1 : SUCCESS in 32m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/303a082ee28b47fe92f610f21beb2c32 : SUCCESS in 39m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8613677deaab42448ef869ebf6ef84cc : SUCCESS in 1h 30m 02s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3cd3ea5faaad4cc3939db4d91021bb7d : SUCCESS in 1h 05m 13s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9599b5d63a2c450189924a2789ad62bd : SUCCESS in 1h 21m 04s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/0589e8bd7d6940a0a5163f42926bc53f : SUCCESS in 1h 37m 49s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e0f66a689443434490f6604ecc2a7a4c : FAILURE in 1h 35m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/581edd273ec946d192bac625f3f59989 : SUCCESS in 1h 18m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/808fceeb797542faba4340898f995dd0 : FAILURE in 1h 47m 23s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e2b2ef529c474f2f868ae979d036468e : SUCCESS in 57m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/dcc57440065d4f848cc63c16cca1830f : TIMED_OUT in 3h 05m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8423245cfb8043e88f6801aafc3eb8f4 : SUCCESS in 28m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4febdbe2d4122b55374aa4100a5301c794d7e1ec","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-08 18:49:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"33bd338130cf45b953d28f72d1467e8844df8dbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-09 07:12:21.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5fe5d6a92d8ba04b2be360d8031b891bea06709a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-09 07:27:09.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent, and commit message as Patch Set 8.","accounts_in_message":[],"_revision_number":9},{"id":"49627dc92b062f51e453058cd00b2693c2223bcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-09 07:47:41.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3959d7c6fadda06232202538569bd5130045aa92","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-09 07:53:02.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8229e4498111df61c5e7fbcace29bea2cdf5c085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 10:26:03.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e5beabf1c594f0983502101edceb816 : SUCCESS in 4m 33s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d9755322eb244c86bc8ff5ddefdf787f : SUCCESS in 32m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/12961ab954854c73aaee04c224f1da4b : SUCCESS in 33m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/22b20db24f324f3b824b8455e02a7bc0 : SUCCESS in 1h 21m 53s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/57a2c7b7f38a47a4b0cbb4c35c9bd983 : FAILURE in 1h 10m 15s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3aa5a955af184e81a8d501c25494d851 : FAILURE in 1h 26m 21s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/45a13e7454cf44b9b79e004a123d7d05 : SUCCESS in 1h 34m 00s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b5f184a7e2b84af4af8ed252cdc39fb6 : SUCCESS in 1h 16m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/58f00089cb504c998291f8fc21b4a357 : FAILURE in 1h 31m 00s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a13d50d61f3d4d9787023fb4bf432931 : FAILURE in 2h 04m 04s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/56b36638659c48bdb9f71e2b3bfdde9d : SUCCESS in 1h 19m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8f97cf303f5a412ca2cdca4deb534297 : FAILURE in 2h 36m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f6f01bd661c048b4a93f51a26bdf8526 : SUCCESS in 22m 47s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"38adb30418275536bc11434e7cb6b68b93b9846f","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-09 12:31:31.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bf9e881e49ebf798e6c0971504925b773da14279","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-09 12:32:09.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7581b0b0f9c3a9230602afba1faf87e3fec68a36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 14:29:40.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eca81423869940ce8dc1fb3c201f405b : SUCCESS in 5m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dee567254c7542c0871f6ddd1d756bb4 : SUCCESS in 29m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5129a3678fb84881a8bd40527d455abf : SUCCESS in 37m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a073fec4e13c46fcad8e150cde5555e2 : SUCCESS in 1h 18m 59s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/fbec71157bfe4bab9b7eb6c892247f13 : SUCCESS in 1h 17m 57s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/2c48f710bed043ffa567fba6938ed8c7 : SUCCESS in 1h 03m 18s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/a23d6d3f38334a77857bf140eaee9164 : SUCCESS in 1h 41m 35s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2531349c67d345349d9e979dee2b2085 : SUCCESS in 1h 12m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b5396c0860a249f39fce5443f9a516e1 : SUCCESS in 1h 39m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e1076effa62849ac83c3cf29f0a21695 : FAILURE in 1h 53m 14s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4ee7ea68388543c889c39c521eae0701 : SUCCESS in 56m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/110722a74c7345268058451d2f68bdd5 : FAILURE in 1h 30m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8fcb9efc3f2b4dd6b463dfa808fb38cd : SUCCESS in 28m 38s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7eec05af656639e3123d70326e506c210b2d4e7e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-03-15 21:01:45.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bab709bcd80faac6ee29c39a9b177a223439dd9a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-03-15 21:02:10.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bab2b7c6aeb9275dfc954731c329d01b1db2f33e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-16 18:57:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"80d3ab367de93bec8fb48622da69126ed0789733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 21:44:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf64e9f9ccd84c4d8b714dba1f056132 : SUCCESS in 5m 24s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/85d7145d85e041e992c781654613db86 : SUCCESS in 40m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3524e88ed4c34970a90a980581b47540 : SUCCESS in 42m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/eed8982512bb4e009a0ad9312797b4fa : SUCCESS in 1h 17m 41s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/949d70e12d604c0d9e972f3eac5e0370 : SUCCESS in 1h 07m 21s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a51fa7f8180e469c845d8694e903ba2b : SUCCESS in 1h 51m 55s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/b9b84c363b06462787ae006f782cdaa0 : SUCCESS in 1h 07m 13s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f9e58463b96249e38d83040948513b3e : SUCCESS in 1h 54m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4f5df7b93cdd444684958bb5c100a036 : SUCCESS in 1h 29m 17s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f20cc2e1ebef4ebe993a99c5bc26d01f : FAILURE in 1h 42m 35s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/39db6c10363d4357a3055c644da22214 : SUCCESS in 1h 03m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c96572134830457e944b8feb112fd3a8 : FAILURE in 2h 41m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/110c4af0872b49aea963de89c1c47e5a : SUCCESS in 27m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5179942f51b742768738765a01b7b629304b16b4","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-03-21 12:36:40.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7230382be5f5eac64e294bfff1fa988fa0d5c34b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-21 17:56:21.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"62883f528eb5d44ee9adde89d3978cfea2edb0fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 19:36:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a793cfc6e34f4d1d9e8ecdd8149ed2b2 : SUCCESS in 4m 59s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dd6de916d2414cc3ab270996861fc2df : SUCCESS in 34m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7967dd178557493995e854f590325fc7 : SUCCESS in 41m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/356fa3ec0a8443ad9207d2cc872a8bb6 : SUCCESS in 1h 21m 43s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/1f9412d2c53e422e8b5d87f51e30b47b : SUCCESS in 1h 08m 36s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a9918fe5969445bf9f634f00dcbbd6cd : SUCCESS in 1h 12m 40s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/9b4a629607da4311a571eaf9f7a0af6e : SUCCESS in 1h 30m 53s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/305c481796394301a81d27e54c437e19 : SUCCESS in 1h 21m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c646f8c29f9f477fa51351e978f004e3 : SUCCESS in 1h 21m 13s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/959010d616b6450f9a3668ddfeb36ed3 : FAILURE in 1h 35m 01s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0b7d2f9a9021488685e071ba0a421fd9 : SUCCESS in 1h 02m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c3d048ce60c9477bbe22c0134d52e74e : FAILURE in 1h 34m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3e4e4580ad26464bb80681b4ec93b538 : SUCCESS in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4cb895c91659804aa7b4c029f1ba9a7db032b743","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-22 00:21:47.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/68/831168/10/check/DellEMC_VNX_Manila/040b760/DellEMC_VNX_Manila/170 : FAILURE in 2h 53m 51s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":10},{"id":"6f9c1008cd36da7a1fa976c3c34dfb210d8f5952","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-23 18:49:03.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6cbe0fc6b73b3af27afc2c835681a37a483dbd8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 21:26:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ad8cab21d0e42a887d98f6c4bd72735 : SUCCESS in 5m 01s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4f4d63d11a714cc8b65c09cde27cca20 : SUCCESS in 31m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d873042067f641dd92ddc651bc06a65c : SUCCESS in 37m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1d51d43439aa41a393ff20a7447613c4 : SUCCESS in 1h 10m 27s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f736752484e8427084871a5f34fd3db2 : SUCCESS in 1h 00m 54s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3e6b7c8b04814e8387b23197ebdb6b61 : SUCCESS in 1h 06m 33s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/1045400f738347ba8ddbc4af2c15c532 : SUCCESS in 1h 04m 17s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ef179eda4a3a4788ab6ea4e219d48d2b : SUCCESS in 1h 12m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/42d11135e00a4d4788c31ee93d1c3019 : FAILURE in 2h 13m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/40b6c5d2ea564be996c1b776ea8dd383 : FAILURE in 1h 48m 54s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ae36a68346504f79a9b08ce34595b635 : SUCCESS in 52m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1f061d850f3348d09204c21f64da64d8 : SUCCESS in 2h 30m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6794341d21904208b2059438d0b6b724 : SUCCESS in 22m 46s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a15509209938d8ab812268d4e9e5a0daf772db27","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-23 21:29:22.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/831168/10/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c558d79/ : FAILURE in 2m 57s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/68/831168/10/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/0070ee0/ : FAILURE in 2m 54s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"8ee8ed39c80bc1dedc9e24bf3a4f709e3c21f535","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-24 03:52:11.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/68/831168/10/check/DellEMC_VNX_Manila/a2237b0/DellEMC_VNX_Manila/188 : FAILURE in 2h 53m 53s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":10},{"id":"f7f3bbdfee894fbcbe6025dd5c2a1f9e146ea834","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-24 12:42:48.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"075d73cf8cca09a9748b41a0759df234ec9fd075","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-03-28 20:38:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b62d498035506cc00b043f9d464619193c0f15f2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-29 16:29:00.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a7a44a11c79d8e8ddc255a394bbb12149de338fe","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-03-29 18:43:33.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3e78da417bcfbf9832d2b0a7bb4c443346d9a269","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-04-20 18:27:52.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1189414d1d1dfecb4132a6d8aaed85b8f266342a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 15:39:48.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"7ee06c2a9ea89d8e15733a332d0e6e1046f8d36e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 17:18:33.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3af8f1c69a35472f8cf7bb18999bd01e : SUCCESS in 4m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d202ee679a9746478f97b4505e14e81b : SUCCESS in 29m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/be7e9f60e00e4cc08fdac090ca51f511 : SUCCESS in 39m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/df29189119aa49619fd3a96dc6083bee : SUCCESS in 1h 34m 14s","accounts_in_message":[],"_revision_number":10},{"id":"57c9381fb71466301ab27606d13283e8b9fb72eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 17:18:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a02c455af93838e4247be7348630bc83b138cf6c","revisions":{"92c67c1415b12b295b94ed9979c9a33b4de2169e":{"kind":"REWORK","_number":1,"created":"2022-02-28 09:59:45.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/68/831168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/831168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/1"}}},"commit":{"parents":[{"commit":"8aafdb604dc85f8b8fc4b54b884938ea759843d1","subject":"Manila scenario manager cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8aafdb604dc85f8b8fc4b54b884938ea759843d1"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:58:37.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:58:37.000000000","tz":60},"subject":"Manila scenario manager cleanup","message":"Manila scenario manager cleanup\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* create_server\n\t* rebuild_server\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0c2fa6d195e1e31ed077bdc4c4e21fa6758dfddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/92c67c1415b12b295b94ed9979c9a33b4de2169e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/92c67c1415b12b295b94ed9979c9a33b4de2169e"}]},"branch":"refs/heads/master"},"6bb4c89bfe15501dfe861ff0a702b9aa0f94f565":{"kind":"REWORK","_number":2,"created":"2022-03-01 09:01:48.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/68/831168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/831168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/2"}}},"commit":{"parents":[{"commit":"8aafdb604dc85f8b8fc4b54b884938ea759843d1","subject":"Manila scenario manager cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8aafdb604dc85f8b8fc4b54b884938ea759843d1"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:58:37.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-01 09:01:37.000000000","tz":60},"subject":"Manila scenario manager cleanup","message":"Manila scenario manager cleanup\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* create_server\n\t* rebuild_server\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0c2fa6d195e1e31ed077bdc4c4e21fa6758dfddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6bb4c89bfe15501dfe861ff0a702b9aa0f94f565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6bb4c89bfe15501dfe861ff0a702b9aa0f94f565"}]},"branch":"refs/heads/master"},"cee978a7c4c5a6c1678d62c365a8e4c0b766f2a0":{"kind":"REWORK","_number":3,"created":"2022-03-02 10:29:19.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/68/831168/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/831168/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/3"}}},"commit":{"parents":[{"commit":"f7b616fd76ba458f42706f1b739b0328a2ad6956","subject":"Manila scenario manager cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f7b616fd76ba458f42706f1b739b0328a2ad6956"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:58:37.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-02 10:28:42.000000000","tz":60},"subject":"Manila scenario manager cleanup","message":"Manila scenario manager cleanup\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* create_server\n\t* rebuild_server\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0c2fa6d195e1e31ed077bdc4c4e21fa6758dfddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cee978a7c4c5a6c1678d62c365a8e4c0b766f2a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cee978a7c4c5a6c1678d62c365a8e4c0b766f2a0"}]},"branch":"refs/heads/master"},"adc2706accc95a2d53145c5efa408b35ec459dd5":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-03-08 10:05:14.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/68/831168/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/831168/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/4"}}},"commit":{"parents":[{"commit":"f7b616fd76ba458f42706f1b739b0328a2ad6956","subject":"Manila scenario manager cleanup","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f7b616fd76ba458f42706f1b739b0328a2ad6956"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:58:37.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-08 10:05:14.000000000","tz":0},"subject":"Remove create_server and rebuild_server methods","message":"Remove create_server and rebuild_server methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* create_server\n\t* rebuild_server\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0c2fa6d195e1e31ed077bdc4c4e21fa6758dfddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/adc2706accc95a2d53145c5efa408b35ec459dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/adc2706accc95a2d53145c5efa408b35ec459dd5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d91f5753419053ac2b50c5edac24fd69c3da3496":{"kind":"NO_CHANGE","_number":5,"created":"2022-03-08 10:41:33.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/68/831168/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/831168/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/5"}}},"commit":{"parents":[{"commit":"3d3399b9aead16db795c030f2880db6934af4339","subject":"Remove _create_port and create_keypair methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d3399b9aead16db795c030f2880db6934af4339"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:58:37.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-08 10:41:33.000000000","tz":0},"subject":"Remove create_server and rebuild_server methods","message":"Remove create_server and rebuild_server methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* create_server\n\t* rebuild_server\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0c2fa6d195e1e31ed077bdc4c4e21fa6758dfddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d91f5753419053ac2b50c5edac24fd69c3da3496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d91f5753419053ac2b50c5edac24fd69c3da3496"}]},"branch":"refs/heads/master","description":"Rebase"},"2d628bc03f3b6cb4c31e2b596d65d68a09f5460c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-03-08 10:51:41.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/68/831168/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/831168/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/6"}}},"commit":{"parents":[{"commit":"2e0eff6b3ee83aefb0e1ce3d4aaec0d80b73e2c6","subject":"Remove _create_port and create_keypair methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2e0eff6b3ee83aefb0e1ce3d4aaec0d80b73e2c6"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:58:37.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-08 10:51:14.000000000","tz":60},"subject":"Remove create_server and rebuild_server methods","message":"Remove create_server and rebuild_server methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* create_server\n\t* rebuild_server\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0c2fa6d195e1e31ed077bdc4c4e21fa6758dfddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2d628bc03f3b6cb4c31e2b596d65d68a09f5460c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2d628bc03f3b6cb4c31e2b596d65d68a09f5460c"}]},"branch":"refs/heads/master"},"09d9b2f10741bb9516f9ce325111293ae254d45c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-03-08 14:20:53.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/68/831168/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/831168/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/7"}}},"commit":{"parents":[{"commit":"78173fe2be8e4f8adfa17c039d4a96c8a99384f2","subject":"Remove _create_port and create_keypair methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/78173fe2be8e4f8adfa17c039d4a96c8a99384f2"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:58:37.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-08 14:20:36.000000000","tz":60},"subject":"Remove create_server and rebuild_server methods","message":"Remove create_server and rebuild_server methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* create_server\n\t* rebuild_server\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0c2fa6d195e1e31ed077bdc4c4e21fa6758dfddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/09d9b2f10741bb9516f9ce325111293ae254d45c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/09d9b2f10741bb9516f9ce325111293ae254d45c"}]},"branch":"refs/heads/master"},"9da4e1836e894375e0bdade409213bf90f272de2":{"kind":"REWORK","_number":8,"created":"2022-03-09 07:12:21.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/68/831168/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/831168/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/8"}}},"commit":{"parents":[{"commit":"3d3399b9aead16db795c030f2880db6934af4339","subject":"Remove _create_port and create_keypair methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d3399b9aead16db795c030f2880db6934af4339"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:58:37.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-09 07:05:43.000000000","tz":60},"subject":"Remove create_server and rebuild_server methods","message":"Remove create_server and rebuild_server methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* create_server\n\t* rebuild_server\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0c2fa6d195e1e31ed077bdc4c4e21fa6758dfddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9da4e1836e894375e0bdade409213bf90f272de2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9da4e1836e894375e0bdade409213bf90f272de2"}]},"branch":"refs/heads/master"},"1277992087ed1f2cf7188bfb2560855603ad4783":{"kind":"NO_CHANGE","_number":9,"created":"2022-03-09 07:27:09.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/68/831168/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/831168/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/9"}}},"commit":{"parents":[{"commit":"3d3399b9aead16db795c030f2880db6934af4339","subject":"Remove _create_port and create_keypair methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d3399b9aead16db795c030f2880db6934af4339"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-02-28 09:58:37.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-09 07:26:11.000000000","tz":60},"subject":"Remove create_server and rebuild_server methods","message":"Remove create_server and rebuild_server methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* create_server\n\t* rebuild_server\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0c2fa6d195e1e31ed077bdc4c4e21fa6758dfddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1277992087ed1f2cf7188bfb2560855603ad4783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1277992087ed1f2cf7188bfb2560855603ad4783"}]},"branch":"refs/heads/master"},"a02c455af93838e4247be7348630bc83b138cf6c":{"kind":"REWORK","_number":10,"created":"2022-03-09 07:47:41.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/68/831168/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/68/831168/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/68/831168/10"}}},"commit":{"parents":[{"commit":"290ef29ae05833f26e0c3eb5225135cb3e3d0b25","subject":"Remove _create_port and create_keypair methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/290ef29ae05833f26e0c3eb5225135cb3e3d0b25"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-09 07:46:54.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-09 07:46:54.000000000","tz":60},"subject":"Remove create_server and rebuild_server methods","message":"Remove create_server and rebuild_server methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* create_server\n\t* rebuild_server\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0c2fa6d195e1e31ed077bdc4c4e21fa6758dfddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a02c455af93838e4247be7348630bc83b138cf6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a02c455af93838e4247be7348630bc83b138cf6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
