)]}'
{"id":"openstack%2Fmanila-tempest-plugin~834220","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I6e4db6b89f89d2374580827e2334635dc5ead5b9","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/cephadm_deploy","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-04-13 14:53:00.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-04-13 07:05:55.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"25402":{"account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"last_update":"2023-04-06 09:31:30.000000000","reason":"\u003cGERRIT_ACCOUNT_25402\u003e replied on the change","reason_account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2023-04-13 16:09:49.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-04-13 08:15:46.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I6e4db6b89f89d2374580827e2334635dc5ead5b9","subject":"[CI] Add cephadm based native cephfs test job","status":"MERGED","created":"2022-03-17 19:25:27.000000000","updated":"2023-04-13 16:09:49.000000000","submitted":"2023-04-13 16:09:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":49,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834220-bp/cephadm_deploy","meta_rev_id":"fe2f1b1a4c2f8fbbe62d34b8faf76dde92035ddb","_number":834220,"virtual_id_number":834220,"owner":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-13 16:09:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2023-04-13 07:05:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":1,"date":"2023-04-06 17:12:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":2,"date":"2023-04-13 08:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2023-04-13 14:53:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-17 19:26:45.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2022-03-17 20:05:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-22 19:39:00.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-04-07 05:13:12.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2022-04-07 13:53:31.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-08-16 19:40:27.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-04-05 23:17:35.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2023-04-05 23:17:35.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"14b37f10e21f8e8ba10157b1fbda86b41ebd71a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-03-17 19:25:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dff739728dcbc8e976358783220a45e0cfe435ec","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-03-17 19:26:24.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0ffc3674ce33209b1e838e4a5938308eaccae10f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 20:05:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4b3a48ff66b4ba885978b1723fb0675 : SUCCESS in 4m 58s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/01a6aabc775d4b01bee0cd98d0060fd0 : FAILURE in 34m 46s","accounts_in_message":[],"_revision_number":1},{"id":"24b2fbd592752717d31e6aaae460fe7bc4c284b0","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-03-17 20:12:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"362c03092f84f56d1e2b06e31e6a0cca1310d676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-17 20:15:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e3dd6a97733fe3db6bbff2af20b5e7a5055e5236","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-17 20:16:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"38d7e35f9209388b0f5dbfbd456424935a42dd5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 21:00:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd83f4245c2439dbaa09ce23a9d1c96 : SUCCESS in 5m 03s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6ed82abccb304dc29950eff01e04028e : FAILURE in 39m 19s","accounts_in_message":[],"_revision_number":2},{"id":"28121e266d51dcd7380856c1464827683368b1f5","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-18 11:04:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b805c0b46c83c2d3e5f90de3c11756b40add6522","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-18 14:48:22.000000000","message":"Topic bp/use-cephadm-nfs-ganesha removed","accounts_in_message":[],"_revision_number":2},{"id":"7e56342a1303b12e5734d2cf656894ef5037c559","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-18 14:48:24.000000000","message":"Topic set to bp/cephadm_deploy","accounts_in_message":[],"_revision_number":2},{"id":"81cf2f954101732560450b2937a68863e11d793e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-21 18:03:12.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"22e35c1ef0c025d1a1a12d73dafffcf548972213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 19:08:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3280cf85cc7f49eb88431acb737c2055 : SUCCESS in 4m 49s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/eec3da8b75c3446199e05b463ef77159 : FAILURE in 1h 00m 11s","accounts_in_message":[],"_revision_number":3},{"id":"d2d817cc7cb42c98b98da3b6a81e5fa06a338e77","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-21 20:53:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"365851bbf9be205bf006a78d9369f6c8ebe1b26e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 21:52:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7873b89531e40b9a2c7bcbc06e31324 : SUCCESS in 5m 19s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/eaa351b47b184ed6a262e67294b78276 : FAILURE in 53m 09s","accounts_in_message":[],"_revision_number":3},{"id":"fdf450de9d4af59b94bc2cd2bddc9db532501786","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-22 09:51:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fba44d7f952252e0c6f5864588f210ad8e77b2b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-22 11:16:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"aeb6e712205b9d841dc2bf1a9842c16dea61352d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 11:20:26.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  The key \"suppress_errors_in_cleanup\" appears more than once; duplicate\n  keys are not permitted.\n\nThe error appears in the following stanza:\n\n  suppress_errors_in_cleanup: True\n                backend_names: cephfsnative1\n                capability_storage_protocol: cephfs\n                default_share_type_name: cephfstype\n                enable_cephx_rules_for_protocols: cephfs\n                enable_protocols: cephfs\n                image_password: manila\n                multitenancy_enabled: False\n                run_share_group_tests: False\n                suppress_errors_in_cleanup: True\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 580, column 15","accounts_in_message":[],"_revision_number":5},{"id":"68158326f57fdc8cdf35fa1197c8bd82f7d389d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-22 11:22:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e9188a69ca6c608ed36207c57b441ac70a99f107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 11:46:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9f5d4c2657c4d4489e888c400c3c09b : SUCCESS in 6m 08s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bea69efa67374e90825fbf3de4a1b366 : FAILURE in 19m 31s","accounts_in_message":[],"_revision_number":6},{"id":"3e9dd578df299790fc6d1ea54601a07151f72f96","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-22 11:56:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a0abb14fb0a232c0f841d4c1235cc7ba886b0798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:17:32.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42df3ee948fe4a328ffc30d970b3277d : SUCCESS in 4m 57s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/cff71220b1d54ccba85daa9343de5995 : FAILURE in 15m 42s","accounts_in_message":[],"_revision_number":6},{"id":"5e7668267580e0a1d5ef6af4f5ce087197285f3b","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-22 17:03:04.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"41826a92b1f435fe5a5597785258d5b031a7f5c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 18:38:50.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da6daa8b6ed144c3be790b1f1b509cf0 : SUCCESS in 5m 20s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b6bd9ccb18204899825dcd6899f86908 : FAILURE in 1h 33m 20s","accounts_in_message":[],"_revision_number":6},{"id":"fff621e11c79b1a67428ef0705cd2b57b4104448","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-22 19:39:00.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/6/check/DellEMC_Unity_Manila_dhss_true/19201b7/DellEMC_Unity_Manila_dhss_true/310 : FAILURE in 3h 45m 03s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/6/check/DellEMC_Unity_Manila_dhss_false/efb3110/DellEMC_Unity_Manila_dhss_false/317 : SUCCESS in 1h 06m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"68d1a69d5832f456f41d37e6762e1ed37a3d7e27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-24 11:39:38.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a992bc6f14933712e95266a6ecf66b1ea573bd21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 13:15:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5d21cebd0934b6bb7262a4a6baaf282 : SUCCESS in 5m 08s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/97985f5f8a8749b8905a1799baca1e14 : FAILURE in 1h 33m 05s","accounts_in_message":[],"_revision_number":7},{"id":"0bcbf5edcba9835657cda3ceba0ef251f2c8d5d3","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-24 13:52:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d2c267afa1331ca8ed60dc933a9550fa083f0f5d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-24 15:58:36.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/7/check/DellEMC_Unity_Manila_dhss_true/0b229ee/DellEMC_Unity_Manila_dhss_true/318 : FAILURE in 4h 18m 43s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/7/check/DellEMC_Unity_Manila_dhss_false/ff7ce67/DellEMC_Unity_Manila_dhss_false/325 : SUCCESS in 1h 21m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"9434190f9377164b19f5b701c2afd46c9a7f01a0","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-28 07:42:49.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fb8c922d1464d8246dba9a4c8d3f6c63e0320ce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 10:28:40.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04870ba4fa1740a184045776a156d99a : SUCCESS in 6m 09s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/55879e50a6b242af9d27f7ea1defd1a4 : FAILURE in 1h 41m 30s","accounts_in_message":[],"_revision_number":7},{"id":"ac269c8035553d25691f4619f46e59672d839448","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-28 10:39:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7c6dce898e85fe10fd77e30524967872d7d518de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-29 14:02:25.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"a9f616a940e5bf4dcc5eb99441c65a8c41eca379","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-29 14:06:09.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/8/check/DellEMC_Unity_Manila_dhss_true/e7db2a7/DellEMC_Unity_Manila_dhss_true/325 : FAILURE in 2m 53s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/8/check/DellEMC_Unity_Manila_dhss_false/ee3ad52/DellEMC_Unity_Manila_dhss_false/331 : FAILURE in 3m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"502e867e70cc7014ed71c57bf322e125e5fce7c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 15:43:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ce7309cad7b49cd959e27692e5e7353 : SUCCESS in 5m 25s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9dc6640a69654ed69bc23d5d57feb878 : FAILURE in 1h 39m 28s","accounts_in_message":[],"_revision_number":8},{"id":"f35d29b7d66f2f8f174e3b7619c35a0d4fc0be65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-29 18:42:28.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"643721c26b52be23ee364a24d990c9c22e6799d3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-29 18:46:11.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/9/check/DellEMC_Unity_Manila_dhss_true/b4fb545/DellEMC_Unity_Manila_dhss_true/327 : FAILURE in 3m 06s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/9/check/DellEMC_Unity_Manila_dhss_false/fb25a04/DellEMC_Unity_Manila_dhss_false/333 : FAILURE in 3m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"62488e1b224a9e975746f639e31eaf69a76ac40f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 21:18:01.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fad9999bf9f645ca8c2f0092f60f00e1 : SUCCESS in 5m 35s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/687e28e3ddd74458bc38c58b4a5ae28a : FAILURE in 1h 47m 54s","accounts_in_message":[],"_revision_number":9},{"id":"1ed39483498e16c42c22c970918d0f6aacd03ed7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-30 06:10:45.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"a775baa25271140e1878daab806f6fb04126a22e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 07:57:57.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57eb608631d04817b6adad6c5409a34c : SUCCESS in 4m 40s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/58f4054346b6441bbb04b56fc1400349 : FAILURE in 1h 42m 20s","accounts_in_message":[],"_revision_number":10},{"id":"35dc5f2f5c332c780446f711adcdb25084c31f79","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-30 09:36:29.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/10/check/DellEMC_Unity_Manila_dhss_true/0c68002/DellEMC_Unity_Manila_dhss_true/333 : ABORTED in 2h 02m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/10/check/DellEMC_Unity_Manila_dhss_false/b9f0aa6/DellEMC_Unity_Manila_dhss_false/338 : FAILURE in 33m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"439526a3ab5524f69eca153eacb14189461fbf96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-30 16:42:16.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"5348a69ef460d0dd10adc80ded1cf18a727c9634","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-30 18:04:04.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/11/check/DellEMC_Unity_Manila_dhss_true/e0477be/DellEMC_Unity_Manila_dhss_true/336 : FAILURE in 1h 13m 53s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/11/check/DellEMC_Unity_Manila_dhss_false/2202d70/DellEMC_Unity_Manila_dhss_false/342 : SUCCESS in 1h 21m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"1fb0ce05e5cc9cfea6b9ef009663ee777d9bc3ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 18:39:47.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e02e6a7701d44a6b8f8dda0a80702b50 : SUCCESS in 3m 18s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/92897bb8174242f0acfd741e767f5b65 : FAILURE in 1h 50m 13s","accounts_in_message":[],"_revision_number":11},{"id":"0bd304eca18c0fecac770114ca81b583614c4535","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-03-31 13:40:38.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6069279f4342003d6d6ead7841f53830b04b3db7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-31 15:05:21.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/12/check/DellEMC_Unity_Manila_dhss_true/453edff/DellEMC_Unity_Manila_dhss_true/337 : FAILURE in 1h 24m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/12/check/DellEMC_Unity_Manila_dhss_false/b754f5e/DellEMC_Unity_Manila_dhss_false/343 : FAILURE in 1h 00m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"4cdaafb54b64f1253304ca6558ba1ef9638d107b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 16:22:50.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea4633ac88ae4bfb8cf9505b68997204 : SUCCESS in 3m 32s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/72014a36bc2e43b2a5b01058a420dc97 : FAILURE in 2h 38m 57s","accounts_in_message":[],"_revision_number":12},{"id":"60f0a7761003d1ea60a5a3cb31b9bc9b2bf63a97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-04-01 11:38:31.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"fc055ec540489ac70b4c1fe4880ef7191cb165e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 12:52:33.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b301b7c0954883b7bf4bed8d6bd293 : SUCCESS in 3m 24s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9bafaa4e199746369e045856ac32c7de : FAILURE in 1h 11m 32s","accounts_in_message":[],"_revision_number":13},{"id":"6b00ded9b48f0181c66dd2fd0b7b2427b8d6aebc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-04-01 13:46:09.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"4e0ef6063cc68be30a119fb0b479a12fd4d688c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 16:55:51.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4963cf49eb74e44800d9969aef93327 : SUCCESS in 4m 00s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/abd68cb5137d4c67b3231cf1ba756e9c : TIMED_OUT in 3h 03m 55s","accounts_in_message":[],"_revision_number":14},{"id":"cb282d01036fe8cd0f2c20869134880d6a842724","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-04-01 19:27:05.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"3978bdc2b83f288b4acbe2058c47f3579fda251c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 22:36:02.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77febda332214b95b3cfecd9e2fa0192 : SUCCESS in 3m 27s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b60ae36ae896494eadfe4379e6efaa11 : TIMED_OUT in 3h 03m 33s","accounts_in_message":[],"_revision_number":15},{"id":"ea5c94c10acd432ea3eb2768e05376a1e58666d2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-02 19:50:55.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/15/check/DellEMC_Unity_Manila_dhss_true/779e70b/DellEMC_Unity_Manila_dhss_true/357 : SUCCESS in 1h 01m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/15/check/DellEMC_Unity_Manila_dhss_false/c2e35b9/DellEMC_Unity_Manila_dhss_false/376 : FAILURE in 28m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"48a7183016ac3e53e7c7af70f4d05d79513b2908","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-04-03 08:21:28.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"7973972dd9c74cbbb794a0bc140bb6ec71926676","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-04-03 08:22:59.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"9545503365d77b3d9c8c2ede76cf995921643876","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 11:40:23.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a04704a16ba74330a5528a6c2029a3b7 : SUCCESS in 2m 59s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/fd89e3cf7a10420aad745f71f4e5e1f8 : TIMED_OUT in 3h 03m 33s","accounts_in_message":[],"_revision_number":15},{"id":"d85e55f6289e94f836c62a903c26798edcb221c9","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-04-03 12:54:11.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"95ece467075d2813340415e7e8fc7823c3fbb600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-03 16:03:20.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/959ebd8f2a0f45ca811da18615c8376e : SUCCESS in 3m 01s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/df70fece149e4891b2c298e7b41a6b53 : TIMED_OUT in 3h 03m 32s","accounts_in_message":[],"_revision_number":15},{"id":"5035e0fa0da330c1c72d8e0e6b327bb7d8f4c797","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-04-04 06:58:04.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"932104ff4f7707500654b23172c5830f4563aaad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 07:48:22.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":15},{"id":"51246df212c6d556d8732bfdc900236524f2f648","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-04-04 07:55:15.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a0050e6259b3d546e78380caf6629646b82f8f0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 10:19:33.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75c3a7c94fe049499a5aada703147d25 : SUCCESS in 3m 04s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2f749002baf74d638394d80f2b22c72e : FAILURE in 2h 22m 43s","accounts_in_message":[],"_revision_number":15},{"id":"23985ee01cc3aa482be3e9cd683a2dd022522674","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-04-04 17:14:46.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"52d380bc84bd213c2d8d8e84ebf03f6b4035b76d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 19:58:02.000000000","message":"Patch Set 16: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df932a7b8f354393bb1573472dcb824a : FAILURE in 3m 43s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/83e343cc99f04620a13f50d85e6fc375 : FAILURE in 2h 42m 06s","accounts_in_message":[],"_revision_number":16},{"id":"c64c4412ab91410c787220ba3e45c99b03cd3518","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-04-04 19:58:56.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"f8f679c851dde652a486358a8b613ee9d274fd65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 23:08:17.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/193ae0bd62ef453abb2e6389c0384824 : SUCCESS in 3m 38s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6530b00c0b7b45a4949b9bdcab70e3c2 : TIMED_OUT in 3h 03m 36s","accounts_in_message":[],"_revision_number":17},{"id":"e8cb13d236e60856f478713312c6a214cd601909","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-04-05 09:04:46.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"3206858025ae0d94a9bc7915fa79084319e0a727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 11:12:14.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cec1ec639da847768f4cb77bb754fb07 : SUCCESS in 3m 16s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e1e2e730240c433485cfa5e539cd9a50 : FAILURE in 2h 05m 04s","accounts_in_message":[],"_revision_number":18},{"id":"e2b6735c08124903f822735f17b2987819c93802","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-04-07 05:13:12.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"7d08caf1466a6d016790dc3d970fd2ae6ebbfe24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 07:10:36.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97b099a8bae9456d8d3e1797dcd6ac3f : SUCCESS in 3m 20s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/068a5517da5e47039eae54d92ab72111 : FAILURE in 1h 52m 14s","accounts_in_message":[],"_revision_number":18},{"id":"23f089f0ced80ddf8625f20adcb300afce07d341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-04-07 13:52:48.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"ed7ab757d84a950ec4d50ac18d69a96323ceb199","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-04-07 13:53:31.000000000","message":"Patch Set 19: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"cba5adb87485ebf99e65b5dd8113ba1d73b71e57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 14:30:26.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf3b48ddeca94609b48fd65c6a262634 : SUCCESS in 3m 41s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a69fd16a6d8444e48892592e9b3048a1 : FAILURE in 35m 20s","accounts_in_message":[],"_revision_number":19},{"id":"c14bd25ee7ea38de92b820e7cf81f203426e9da9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-04-07 17:58:27.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"409e34266f832634011a083594d7d011b665d337","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-04-07 17:58:38.000000000","message":"Patch Set 20: Workflow-1","accounts_in_message":[],"_revision_number":20},{"id":"f1b43555b5441257cd91e0d026f07e49f5645421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 20:22:02.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b041c50d0c4c429c8106b7f4900036 : SUCCESS in 2m 54s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/359fd17b4ebd4192878c84b3850d63be : FAILURE in 2h 17m 50s","accounts_in_message":[],"_revision_number":20},{"id":"42bbb4ba99bb7021f9d512c4f3697ed10372d462","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-07 23:39:44.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/20/check/DellEMC_Unity_Manila_dhss_true/4eee2a3/DellEMC_Unity_Manila_dhss_true/363 : FAILURE in 6m 30s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/20/check/DellEMC_Unity_Manila_dhss_false/1eb0a4c/DellEMC_Unity_Manila_dhss_false/382 : FAILURE in 1h 24m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":20},{"id":"e041d8c702040b5c90812e1df7eba4ed93cf4397","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-04-14 13:11:03.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"f59b2bfd198b115d5c60a0dbd33548b026b91fe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 15:36:35.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a59bb2b816f4aaaba34adcbdd8784fe : SUCCESS in 3m 28s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ceac780c7ba84c5298f0980f0537e18d : FAILURE in 2h 20m 00s","accounts_in_message":[],"_revision_number":21},{"id":"559445ab38f7e697af5a92a0fa9a732713bbe0ec","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-14 17:21:13.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/21/check/DellEMC_Unity_Manila_dhss_true/6473870/DellEMC_Unity_Manila_dhss_true/385 : FAILURE in 4h 09m 44s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/21/check/DellEMC_Unity_Manila_dhss_false/2400fdf/DellEMC_Unity_Manila_dhss_false/404 : SUCCESS in 1h 23m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":21},{"id":"3abf4d9fe285c89b48f76824ffc29ee847fe10fa","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-04-18 12:50:31.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"2950cbf4a8411973c8382279e662efcc3da98261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 15:06:54.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba8fe70ed3b14204ab6a91e887bad2fd : SUCCESS in 3m 12s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a47500fc3002477f82993633d2a96edb : FAILURE in 2h 14m 36s","accounts_in_message":[],"_revision_number":21},{"id":"8dfa3105f1560a07bd4dc84c689fe33890a6c8b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-02 06:58:22.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"16bbe62bfd67a7efc7df784132eadc9dff5fa162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 08:58:36.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f7be3c7f8f4466ea90bcac3b9674012 : SUCCESS in 2m 59s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2a38baf179544ba3a2f2d22520dbc4f6 : FAILURE in 1h 58m 46s","accounts_in_message":[],"_revision_number":22},{"id":"589824e9fc3a40bc07da6b44045bc4096b1dc383","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-02 09:37:19.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/22/check/DellEMC_Unity_Manila_dhss_true/ecdde5b/DellEMC_Unity_Manila_dhss_true/524 : FAILURE in 2h 38m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/22/check/DellEMC_Unity_Manila_dhss_false/820ac79/DellEMC_Unity_Manila_dhss_false/547 : SUCCESS in 1h 04m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":22},{"id":"03d7ffb9375cb14f7acac324f18779df3921e64c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-13 11:54:13.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"b212932e65233b3747c81976618867cdb36df21f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-13 12:10:32.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"fbe9ed0469ce968c4e100809ec69a5df3a9f3c37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 14:39:32.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c18edb643d84affa6666ca41d9e1ed6 : SUCCESS in 2m 57s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/39a7902aca154cba979b422867099f8e : FAILURE in 2h 23m 59s","accounts_in_message":[],"_revision_number":24},{"id":"d8d7f217cb7f4e998490bff83f2a408cb0c9fb98","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-13 18:00:55.000000000","message":"Patch Set 24:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/24/check/DellEMC_Unity_Manila_dhss_true/2639eb7/DellEMC_Unity_Manila_dhss_true/545 : FAILURE in 5h 50m 08s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/24/check/DellEMC_Unity_Manila_dhss_false/3360d9c/DellEMC_Unity_Manila_dhss_false/568 : FAILURE in 52m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":24},{"id":"2238fdd810b6e3a7adb95a55d0070eb28bd2b6e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-07-12 11:11:41.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"47e57fb1266f9eee56e96be9eb4cd12524e7427b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-12 13:44:27.000000000","message":"Patch Set 25:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/25/check/DellEMC_Unity_Manila_dhss_true/8ef0aec/DellEMC_Unity_Manila_dhss_true/617 : FAILURE in 2h 32m 31s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/25/check/DellEMC_Unity_Manila_dhss_false/4f43719/DellEMC_Unity_Manila_dhss_false/645 : SUCCESS in 1h 04m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":25},{"id":"cfdedebb58ecdc0c5e82f862a9c94122022f1d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 14:13:32.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/10618ad1dcbf4461a2d799add63cd343\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1020f8b727e84fa880ca77c6d9658f18 : POST_FAILURE in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/df5024fe8c4b4f569e5f7552194083b5 : POST_FAILURE in 1h 29m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3137e2208638460b8c6ad364d412a8fe : FAILURE in 2h 55m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c35f60c375594802b9c84cae1b60795a : FAILURE in 1h 51m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/510c12a76d6e42a38856e18176edd411 : SUCCESS in 1h 42m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/74c783ec47f94eb9b776daf09dfa77bc : FAILURE in 1h 58m 22s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/62c3e2281f3044df91801f6e075461f9 : SUCCESS in 1h 11m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d992ac8ef20c43f7a7e3c9a38422a985 : FAILURE in 1h 38m 34s","accounts_in_message":[],"_revision_number":25},{"id":"11710b54388b50b21a86abb8129906b95b9abc3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-07-12 15:42:57.000000000","message":"Patch Set 26: Published edit on patch set 25.","accounts_in_message":[],"_revision_number":26},{"id":"51e5270357877719f817ce2ccce6ec76d5b3ef65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-07-12 16:37:48.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"5f4d011a7db31ae0175cdd195fa75d21f853f8fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 19:44:24.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fc6fc90886f4251a5d20d07c2da1e75\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c6a191186de4d3ca7e2ec1c139f4930 : SUCCESS in 3m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9643194500d345b5b2b4210439104ac2 : SUCCESS in 1h 10m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4c5afbd61cce4c179f54eb3f29296361 : SUCCESS in 1h 50m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/48abe762a844441192a5a526a30962be : SUCCESS in 1h 21m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d5cf522e2f7745d68777e34c796b966f : FAILURE in 1h 25m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/47f0c3d2ee274b239fe5ac5cb6b7bbd4 : FAILURE in 1h 41m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0a1840ee5a21432c9de1fecfcd3495ce : TIMED_OUT in 3h 04m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3cb358a9693c469b9259e79df610e38c : FAILURE in 26m 45s","accounts_in_message":[],"_revision_number":27},{"id":"5f095a5147a1407ccfabdb76a0463aeee4b8df62","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-12 19:59:10.000000000","message":"Patch Set 27:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/27/check/DellEMC_Unity_Manila_dhss_true/82349d7/DellEMC_Unity_Manila_dhss_true/619 : FAILURE in 2h 28m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/27/check/DellEMC_Unity_Manila_dhss_false/e1423ab/DellEMC_Unity_Manila_dhss_false/648 : SUCCESS in 1h 00m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":27},{"id":"19670069d62425f175b88c411602f8cd17286a16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-07-29 14:04:24.000000000","message":"Patch Set 28: Commit message was updated.","accounts_in_message":[],"_revision_number":28},{"id":"e8467f9e7b400a65e28e74ec9b5aba3c310c1896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 16:20:31.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d5b6518daba4127b6dbc03ccfe126ed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38cec99b02b649f391b41ebfeb585dac : SUCCESS in 2m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e994465b29074d08b2d42823923e18cd : SUCCESS in 1h 11m 13s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8423ae909f97484e9ac049cc26d939cc : SUCCESS in 1h 40m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1d6056920ac24aaaa1842b048d0f3d6f : SUCCESS in 1h 20m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/fbd7aee1e6b84727971673e028348035 : SUCCESS in 1h 20m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a661771cc8c54ecdb0dafbea624d5e7a : FAILURE in 2h 11m 07s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a1ca7e52b8aa4ac893848cb10fa9980c : SUCCESS in 1h 12m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7f6ff53af829467e99d215a25c5819d7 : FAILURE in 2h 09m 45s","accounts_in_message":[],"_revision_number":28},{"id":"dd7dbc1706e1499f9363330ec2a5680725b8c2a8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-29 17:20:47.000000000","message":"Patch Set 28:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/28/check/DellEMC_Unity_Manila_dhss_true/d144de6/DellEMC_Unity_Manila_dhss_true/665 : FAILURE in 3h 16m 07s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/28/check/DellEMC_Unity_Manila_dhss_false/49845d5/DellEMC_Unity_Manila_dhss_false/698 : FAILURE in 15m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":28},{"id":"6406d5e3e486949942f43391b72ccb737dbf93b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-01 18:47:50.000000000","message":"Uploaded patch set 29: New patch set was added with same tree, parent, and commit message as Patch Set 28.","accounts_in_message":[],"_revision_number":29},{"id":"8a303fd429bbbca37f524f59b15f134df4ab62a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 20:55:28.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba44cf4a733d4796892c1e0c8ccfb45c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f83784966e34c618fafde5b93acab7a : SUCCESS in 7m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4da0bd395dcd41a78959b41221d6e00a : SUCCESS in 1h 14m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ad3396ba64a4490aa1b3d2823cc41943 : SUCCESS in 2h 05m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f9ff45a0c6b44fea89e4740e3fcee6d1 : SUCCESS in 1h 21m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3fcb6c2b1cde49ee82c667d05e841ec3 : SUCCESS in 1h 44m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/197bb1548f7f4ba2b3d97d13f811066c : FAILURE in 1h 45m 33s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5835b09b65cc4da186ee3ef3542beaa0 : FAILURE in 1h 19m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7da7891ce5764f89981fccceb3f782ea : FAILURE in 1h 36m 10s","accounts_in_message":[],"_revision_number":29},{"id":"bf6526f4c514da344d2f3fbc61ebf8c61795f47a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-01 22:09:29.000000000","message":"Patch Set 29:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/29/check/DellEMC_Unity_Manila_dhss_true/693f829/DellEMC_Unity_Manila_dhss_true/668 : FAILURE in 3h 21m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/29/check/DellEMC_Unity_Manila_dhss_false/ccef270/DellEMC_Unity_Manila_dhss_false/703 : FAILURE in 19m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":29},{"id":"0ffee45d5ccc03c16a5a0423f463c6ef7c5d9d1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-02 07:51:38.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"86773e9e74c125592a36fe4a73d6cacac41b117c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 07:52:01.000000000","message":"Patch Set 30: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  Job manila-tempest-plugin-cephfs-cephadm-base not defined\n\nThe error appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-cephfs-native-cephadm\n      description: Scenario tests CephFS Native (DHSS\u003dFalse) deployed with cephadm\n      parent: manila-tempest-plugin-cephfs-cephadm-base\n      vars:\n        tempest_test_regex: manila_tempest_tests.tests.api\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 682, column 3","accounts_in_message":[],"_revision_number":30},{"id":"ecf61f2f99f3e108122f9ee442049ce8ea68cc3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-02 09:10:19.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"275b106d1ab5a595fcb1ef53d14b04dad64104bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 11:36:52.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/697b52ece4e74276b3be578e7af023b9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80c05e155f8e482b889e443ce34f9fd3 : SUCCESS in 3m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/313ef1f1108e4074973b7abb3a2fc90f : FAILURE in 1h 12m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1b3f18d475d14b448c43d6a07289780e : SUCCESS in 2h 09m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a7b91e626d0f461880442bad25896c33 : SUCCESS in 1h 25m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/20306001ef834df2b6928c59c815f744 : SUCCESS in 1h 18m 36s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e04760dd01e243988a7af1439448cc85 : FAILURE in 2h 24m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/365541a3e2854fc58884469e6750446c : FAILURE in 1h 07m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0c42eab8b280443ba7c9d635bcbd53ec : FAILURE in 55m 19s\n- manila-tempest-plugin-cephfs-native-cephadm-scenario https://zuul.opendev.org/t/openstack/build/4c03475283af42f19163205af473e0ef : FAILURE in 2h 15m 53s","accounts_in_message":[],"_revision_number":31},{"id":"7b41de6c74170df705c4c66911091b31de225c20","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-02 12:00:16.000000000","message":"Patch Set 31:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/31/check/DellEMC_Unity_Manila_dhss_true/d7ded9c/DellEMC_Unity_Manila_dhss_true/672 : FAILURE in 2h 49m 43s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/31/check/DellEMC_Unity_Manila_dhss_false/1edcd4f/DellEMC_Unity_Manila_dhss_false/708 : FAILURE in 13m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":31},{"id":"91087a7949e407aab86160bd73f57ab2683570df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-02 16:00:55.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"f262bff7cd1c6d580b41147f36ff62e0c64ab595","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-02 17:43:40.000000000","message":"Patch Set 32:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/32/check/DellEMC_Unity_Manila_dhss_true/9e797c8/DellEMC_Unity_Manila_dhss_true/673 : FAILURE in 1h 42m 30s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/32/check/DellEMC_Unity_Manila_dhss_false/3ae0e65/DellEMC_Unity_Manila_dhss_false/709 : FAILURE in 18m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":32},{"id":"df8165e12a388594fcd76e54a07cf9c58f609ff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 18:13:47.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/07ed4688690a4d03bd42dde58f3e8863\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58822b60b5304c21a52eab23432ac66a : SUCCESS in 4m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b0c3a6dbb7004f2ea0af225ae4ee6ece : SUCCESS in 1h 06m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3827fc643e3d478d9a988886f24ab241 : SUCCESS in 1h 51m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0a0e72200fbb4c46a72f520f885265dd : SUCCESS in 1h 07m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e6fc7ca6d73440cb91d6085f91e06a11 : FAILURE in 1h 17m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c07b8b37a2284efeb41c6c95da9b7b03 : FAILURE in 1h 52m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8aadc87c56bd40d7aa18774c5de0031f : SUCCESS in 1h 08m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8a7806251b8546c19a6943e00a3bdc02 : SUCCESS in 46m 12s\n- manila-tempest-plugin-cephfs-native-cephadm-scenario https://zuul.opendev.org/t/openstack/build/eb5cf155530d48279a0137887c0fd710 : FAILURE in 2h 11m 00s","accounts_in_message":[],"_revision_number":32},{"id":"b692a44e1f50acc5e36abccc68dbfb89e59134cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-10 13:31:06.000000000","message":"Patch Set 33: Published edit on patch set 32.","accounts_in_message":[],"_revision_number":33},{"id":"5e9ca1835e01f22b06200139676303336a9904ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 15:44:37.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7c1a139bdb54d48a2431eda0a30d7cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f5bbc25766e4ecf9c54fc0e705cba72 : SUCCESS in 2m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/775429a6d8604f19a48ace15c7828883 : SUCCESS in 1h 14m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/659eee1eb8474bfb97b565226fe747c9 : SUCCESS in 2h 08m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6f61d924316d40babc51a54fc88dfabc : SUCCESS in 1h 22m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/80a9d0ae8e3a44f29a1629b324c01e97 : SUCCESS in 1h 06m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/74e665fba4c044d4bf8117676e62e354 : FAILURE in 1h 20m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/71086897c6ef48f79685235f775e976e : FAILURE in 1h 10m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bb9ce3cf938549aa9491ffac1d632af5 : SUCCESS in 41m 47s\n- manila-tempest-plugin-cephfs-native-cephadm-scenario https://zuul.opendev.org/t/openstack/build/f639305899684fab8ba7676709111c4c : FAILURE in 1h 29m 15s","accounts_in_message":[],"_revision_number":33},{"id":"ebd29bc1493bb14fc5b37dccbe9633fae472578c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-10 16:30:52.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"6d0eb2040e9bcac5a664113e1766801e5da1b91a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 18:45:46.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed1418eca4894def846fe054cf9041b6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af446d1b5f7141909fc37c4b8aa131ba : SUCCESS in 6m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ea436ea3803743949c250ed095462be3 : SUCCESS in 1h 12m 22s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5a42a560584240ab93b494ceda7943b3 : SUCCESS in 2h 06m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/dcb43365eefc4e6b86773d94170d0f70 : SUCCESS in 1h 19m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/dd59e8fc974e41e5969e59cb1533784b : SUCCESS in 1h 31m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/31903d473a274fc0a08b4d0fbf552a8d : FAILURE in 1h 43m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/82f6e6b45f824988b90ff6adf6d6df41 : SUCCESS in 1h 12m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8a4d8d4632b3406a974de949e8370071 : SUCCESS in 57m 59s\n- manila-tempest-plugin-cephfs-native-cephadm-scenario https://zuul.opendev.org/t/openstack/build/0323400339e548709e057f71f3b5eb7e : FAILURE in 2h 01m 50s","accounts_in_message":[],"_revision_number":34},{"id":"f6c5cd2349614ff4ae0d14c6872581025c4d165f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-11 03:08:56.000000000","message":"Patch Set 34:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/34/check/DellEMC_Unity_Manila_dhss_true/a0f81f4/DellEMC_Unity_Manila_dhss_true/695 : FAILURE in 4h 37m 08s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/34/check/DellEMC_Unity_Manila_dhss_false/f589323/DellEMC_Unity_Manila_dhss_false/732 : FAILURE in 17m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":34},{"id":"47317e9137293204e0b7d49d98fc8319c1a15ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-11 10:17:12.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"a0fd98559fde1d6a230df6e960bda475901470d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-11 10:18:02.000000000","message":"Uploaded patch set 36.","accounts_in_message":[],"_revision_number":36},{"id":"0185f57aab627674868de3377dbb3fd87542147f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 12:41:03.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f471f509d7fe421a9d7fa417bde925cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa226a69a764cfaa4474c5d27dbfb4c : SUCCESS in 2m 26s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/db303083e765469a9f521eafeaeb9aea : SUCCESS in 41m 53s\n- manila-tempest-plugin-cephfs-native-cephadm-scenario https://zuul.opendev.org/t/openstack/build/380f9fae0acd476f88dcf5b7a367a027 : FAILURE in 2h 20m 21s\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/efb0a5572f364c08ac72c435d48bf6cf : FAILURE in 1h 56m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/80f4b632a4f3476bb27dcd904a45e3fa : SUCCESS in 29m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7621ecdf1a8e430ebe4ee87dac34957b : SUCCESS in 29m 22s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c7131d45e05f448aa39300e9e937dc5a : SUCCESS in 1h 14m 43s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/9093c3d44d0f4ab7a7e790e9111ba79b : SUCCESS in 1h 19m 31s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b43933f2fbd740ecb14eb03a1d99c1dc : SUCCESS in 59m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3bb2313611624992a624a1f2949e2fbc : SUCCESS in 43m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2ce960753f8642609fec8a9c926753d5 : SUCCESS in 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"886c264bbef19faeb75f91c5d5ad4dffc40f62df","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-11 14:36:45.000000000","message":"Patch Set 36:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/36/check/DellEMC_Unity_Manila_dhss_true/143f6da/DellEMC_Unity_Manila_dhss_true/697 : FAILURE in 4h 18m 19s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/36/check/DellEMC_Unity_Manila_dhss_false/ea1073e/DellEMC_Unity_Manila_dhss_false/734 : FAILURE in 13m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":36},{"id":"0af5fdfd4f81eeea0fa0a110d72087aab8b6f36a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-11 16:18:06.000000000","message":"Patch Set 37: Commit message was updated.","accounts_in_message":[],"_revision_number":37},{"id":"8b3b7c5f7037ea3edcda12d6bd9f942567d4f9ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-11 16:28:41.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"eea8c8fcea69c9bb1828c2f8901bedbe00300e1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 18:15:41.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb5794a0123840d382c5bc8134dd3fc2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fada1bde9c743baa97c451c13cb49c8 : SUCCESS in 3m 59s\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ae843444fe8541aebceba038e17ff48b : SUCCESS in 46m 38s\n- manila-tempest-plugin-cephfs-native-cephadm-scenario https://zuul.opendev.org/t/openstack/build/032bcf8826544af5a4b96f3e5a22ef92 : SUCCESS in 1h 24m 54s\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/bcd37d7f3cdb4d9a8222451bcd1fc110 : FAILURE in 1h 45m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ca3ac3acf3654d3d9471649c48ebca49 : SUCCESS in 37m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6f9663a8ccc74cefa5e909d7a2a402aa : SUCCESS in 37m 50s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/04c2d515ce4f4db3b86df6aad6a806be : SUCCESS in 1h 22m 35s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/246cc81e97f54fe39c587a9be903aa01 : SUCCESS in 1h 13m 30s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/701f75a96d034f868e68d7d334dc7e72 : SUCCESS in 1h 12m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/05763160f8a143659c049c2b1bbcf88b : SUCCESS in 52m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b4353684e1bc4c0db07aa1159aa8b3c0 : SUCCESS in 26m 17s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"014631b6a4f5e97f4c8f65e9d6ef8bb95424dd33","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-11 20:48:07.000000000","message":"Patch Set 38:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/38/check/DellEMC_Unity_Manila_dhss_true/1e79853/DellEMC_Unity_Manila_dhss_true/699 : FAILURE in 4h 19m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/38/check/DellEMC_Unity_Manila_dhss_false/82a86fe/DellEMC_Unity_Manila_dhss_false/736 : FAILURE in 13m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":38},{"id":"80806bf821d102a4cdf70c7537e426ad27250bd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-16 14:23:51.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"613679a2078fdcac7ba75d231fbc95c0d19bb39d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-16 14:25:57.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"f4ad7ca853541598d6ecd25f2046a9ca37ee09d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 16:27:22.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaf7ceeb7e8a4112b569a315e629ab37\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8a02bdc3b4e4f4a885e2ccf09e20732 : SUCCESS in 6m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/957e0864d003468494d27a3ec1be1a62 : SUCCESS in 1h 11m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/25914a00662d4fb181eb582c36856253 : FAILURE in 1h 59m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1bf2a4809b8f42c38c8fc3facc207cb7 : SUCCESS in 1h 17m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5d14136b04cb477484ff115ae64ebbfd : SUCCESS in 1h 10m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8692217f89974c169dfa7f78b6dd0c91 : FAILURE in 1h 35m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f72f1fa1d944457097ad96fa5a34301c : SUCCESS in 1h 16m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a96a0b235bab474988df7d7e27030f63 : FAILURE in 1h 24m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c78ea524a85f406fb59fe20eb8207239 : SUCCESS in 1h 03m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/185e2f58199f46bbacfc9ae2f50ff3ae : SUCCESS in 28m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/92932636b8794639bba29d9fb6c10cde : SUCCESS in 34m 17s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/53550e31d73c4c5d8c3b5fd22bfb30d5 : SUCCESS in 1h 26m 16s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a5dc142234a94f7ab6bd3c609ace389f : SUCCESS in 1h 03m 58s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/ea50188444ff44de9fbce10f246717d2 : SUCCESS in 1h 08m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/700e16bcea8942aea8120ddcc0d785e1 : SUCCESS in 50m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3e01bcc8f910474182f7bd8d57a8d1f6 : SUCCESS in 23m 23s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"7703b8ba6c8c8d1b0a88d7c9dab29786889c89eb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-16 16:45:25.000000000","message":"Patch Set 40:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/40/check/DellEMC_Unity_Manila_dhss_true/0e432ed/DellEMC_Unity_Manila_dhss_true/708 : FAILURE in 1h 13m 37s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/40/check/DellEMC_Unity_Manila_dhss_false/bdc7a47/DellEMC_Unity_Manila_dhss_false/745 : FAILURE in 14m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":40},{"id":"16dadaf11b713354d47896e4549abe988ce281bb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-16 19:40:27.000000000","message":"Patch Set 40:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/40/upstream-check/manila-tempest-plugin-ontap-dhss/eced290/ : TIMED_OUT in 3h 11m 28s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/40/upstream-check/manila-tempest-plugin-ontap-no-dhss/359dde4/ : SUCCESS in 1h 15m 11s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":40},{"id":"62c7d6ec931d1a46c0177298c520110416b2a4d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-19 11:40:04.000000000","message":"Uploaded patch set 41.","accounts_in_message":[],"_revision_number":41},{"id":"dc74b7de25e0962e7b099dbfaa49b29d340a10bc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-19 11:40:35.000000000","message":"Patch Set 41:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/41/check/DellEMC_Unity_Manila_dhss_true/274a337/DellEMC_Unity_Manila_dhss_true/730 : FAILURE in 1s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/41/check/DellEMC_Unity_Manila_dhss_false/2148b20/DellEMC_Unity_Manila_dhss_false/767 : FAILURE in 13s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":41},{"id":"441b1522f83d44574637f598d410a975d2324cef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 14:31:50.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b455cef666654ba18b310db0ada0f009\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71fade5723f6488fb21b3b14ecf8486d : SUCCESS in 4m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/36b998f00f504435bed7dad3d08aa53f : SUCCESS in 1h 04m 47s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/20ed2585ce69470f87d9812cd4de9732 : SUCCESS in 2h 08m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/39507e74b624496ea6a1788c54df3a91 : SUCCESS in 1h 26m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cfa2779b9db947069e88b782a5697670 : FAILURE in 1h 35m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/79d7bcaf93534711ba926c16a232cce7 : FAILURE in 1h 49m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b76c581e383c48509be9a477885581ba : FAILURE in 1h 00m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/276d5b60ea9f429593c9d7033bd26a40 : FAILURE in 2h 50m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f14f167c21cd411fa0a914f85359d99f : SUCCESS in 1h 02m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f6cffcfa769045bb8fe397c71d8d4563 : SUCCESS in 28m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/682a41a3f3ed4aa09ad43aa966f29f0d : SUCCESS in 29m 43s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/67c9632bd15a4c2b8c6bf0794379cd2b : SUCCESS in 1h 06m 34s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a6fef5585119420992f39d00888255f5 : SUCCESS in 57m 01s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/ed322ad0a19448958cdfc735c7def015 : SUCCESS in 1h 11m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5f6c73b9cfbc45f29c80790e6a3fae6c : SUCCESS in 45m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9d82d0c992bd4891ad530806c3a5c498 : SUCCESS in 32m 00s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"3cf80ae32179e0336c10914a308b19bc6008112b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-19 14:32:22.000000000","message":"Patch Set 41:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/41/check/DellEMC_Unity_Manila_dhss_true/8e3daab/DellEMC_Unity_Manila_dhss_true/731 : FAILURE in 1s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/41/check/DellEMC_Unity_Manila_dhss_false/c298d96/DellEMC_Unity_Manila_dhss_false/768 : FAILURE in 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":41},{"id":"4dfe9cf1b8912a32814ce02197f6bd1a63b1c20c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-19 16:15:26.000000000","message":"Patch Set 41:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/41/upstream-check/manila-tempest-plugin-ontap-dhss/1dff053/ : FAILURE in 1h 42m 01s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/41/upstream-check/manila-tempest-plugin-ontap-no-dhss/50fa24c/ : RETRY_LIMIT in 8m 36s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":41},{"id":"2e28b4f90c2991e82e2967f3ece89654969a478a","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-19 17:32:56.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"3de20ec3376c405e0df926c85d9aca2a935bbf77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 19:47:14.000000000","message":"Patch Set 41:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/327b193595ee4dabb9252db645e89182\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47412f7c59df437a92e50a42c660936f : SUCCESS in 3m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f541490440e04f1290ee0e6831f6ed67 : SUCCESS in 1h 11m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a3741e67d80f4b79855ac6de8f1698a3 : FAILURE in 2h 12m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/94537700b0114878bd512e33ba749299 : SUCCESS in 1h 04m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/d00630cbf3b840519022e791f221fa5d : SUCCESS in 1h 12m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5956b41eb1ab4894810918d84b56959a : FAILURE in 1h 26m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/623c8b5b5cb345df9b4f6aa2e7479903 : FAILURE in 1h 07m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3f7f18df23a244d1a03cceab22a1112d : FAILURE in 1h 28m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d0e09e7153ac4ef089e1d8997ca91c67 : SUCCESS in 1h 12m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c5e9efd597ed46a1b1765d61fa9d4849 : SUCCESS in 31m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7aa795c2ca2a4fda9f795318df47c0ac : SUCCESS in 32m 18s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/9d02c79ef44641a6b7f5f64d4d2d85c3 : SUCCESS in 1h 12m 58s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d000c26559434526b14b4f0202fca99e : SUCCESS in 48m 31s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d474afee174e47a7ad94ba71915aeda3 : SUCCESS in 1h 03m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0c0826b08c074989bd8c7e73f901f1e3 : SUCCESS in 56m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/93d6dc71c2284d30a5a7d8469ea7f8f0 : SUCCESS in 25m 20s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"7ed37678a3ccdabf1d58a4fd3a5464d9f1ae30a9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-19 19:47:40.000000000","message":"Patch Set 41:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/41/check/DellEMC_Unity_Manila_dhss_true/5135966/DellEMC_Unity_Manila_dhss_true/736 : FAILURE in 1s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/41/check/DellEMC_Unity_Manila_dhss_false/beb48b2/DellEMC_Unity_Manila_dhss_false/773 : FAILURE in 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":41},{"id":"7bfaffdac2f592a376763d9a99bd1f3fb90df3ee","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-19 22:57:40.000000000","message":"Patch Set 41:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/41/upstream-check/manila-tempest-plugin-ontap-dhss/ce652d1/ : TIMED_OUT in 3h 09m 18s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/41/upstream-check/manila-tempest-plugin-ontap-no-dhss/04176b8/ : RETRY_LIMIT in 8m 38s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":41},{"id":"4dba0677f429a5b46c367451caf7b4537db1204d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-24 16:43:28.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"1f24cbc25cb075a0ec42d690e1a3cf443275af3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 18:36:50.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a00d9041916143ecad63c5ee1fb64ad1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae5110e07e02461eb74384b313bf00b2 : SUCCESS in 4m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/968fc69ec47d47e0a943f52b9792ad01 : SUCCESS in 1h 39m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ab1fc1fdfcd64b2d942ab5ecd1317c29 : SUCCESS in 1h 44m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/da1458992b784901aa48596b52b5c0a8 : SUCCESS in 56m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/14428eb658684826ae34a299a18010a2 : SUCCESS in 1h 29m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1bf0d6741d4942e9b6c97a855ce0fb40 : FAILURE in 1h 42m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/51021db6b00f45c89b9168eae1742771 : SUCCESS in 1h 15m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/97877e99659c4de79b904fb7bc752336 : FAILURE in 1h 29m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/846b5f8efaa74d7f8d432c3ff88f524c : SUCCESS in 1h 21m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8d333a2371af4bb7b18e3c7f8d2bad9d : SUCCESS in 41m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fc12ec9c3c7b4aadb851e7fbb2cd0298 : SUCCESS in 43m 21s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/d3dd93dad3e8488ab44b30b3451ae37d : SUCCESS in 1h 22m 01s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/47e258343cc64a2ea79b6eb16e1cb384 : SUCCESS in 1h 03m 55s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b478180f485e407691068efd00271da3 : SUCCESS in 55m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e7ce9eb626344966b713d8898c04d26f : SUCCESS in 46m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4f29b27a22004cfab8dc2fef9a82b18e : SUCCESS in 29m 05s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"2b84714bfc366b392aecae02ecb792b698612a2a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-24 20:30:21.000000000","message":"Patch Set 42:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/42/check/DellEMC_Unity_Manila_dhss_true/53d891b/DellEMC_Unity_Manila_dhss_true/747 : FAILURE in 3h 46m 35s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/42/check/DellEMC_Unity_Manila_dhss_false/d4c70f4/DellEMC_Unity_Manila_dhss_false/785 : FAILURE in 13m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":42},{"id":"c2b5af4b160f4f52f8de539cdc1123b82ff5ded5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-24 21:47:41.000000000","message":"Patch Set 42:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/42/upstream-check/manila-tempest-plugin-ontap-dhss/c8f9fe5/ : TIMED_OUT in 3h 09m 36s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/42/upstream-check/manila-tempest-plugin-ontap-no-dhss/6797d8d/ : SUCCESS in 1h 51m 40s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":42},{"id":"82b954dde889b2051353d6084a19652b85a559ac","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-08-31 02:33:42.000000000","message":"Patch Set 42: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"0367da78555ed1a2cbc9129332bd33f8057103b9","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-08-31 06:04:29.000000000","message":"Patch Set 42: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":42},{"id":"694d104a0f0f718ebf32fb77813d35752564106f","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-31 10:37:15.000000000","message":"Patch Set 42:\n\n(2 comments)","accounts_in_message":[],"_revision_number":42},{"id":"ab865bac935d3998dd2ee181acec42bba16a649e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-31 10:55:56.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"8caa189ec6244515805902fa9494698d58269301","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 12:50:45.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33d591db46b3463593931c7bd3ef5285\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b476ef5ba6c94d139dcb593e3eb4fda2 : SUCCESS in 3m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2a56c91c9ae540f88710f5c485e71dc1 : SUCCESS in 1h 14m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b75f2baa942348adb04d3cb591b46609 : SUCCESS in 1h 52m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6edaf9e21f9c4e91a5a8f43301b7864a : SUCCESS in 1h 06m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/af0e906b21f14a069ef2111e760d5f7b : SUCCESS in 1h 14m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/550f872e7382436d87df6dc0c86ed1ee : FAILURE in 1h 50m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f17f9efb062d41daa592e0635eae01fb : FAILURE in 19m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4922916dd8cc4063ac3747b056382957 : FAILURE in 1h 17m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cb86f36b6c994a9fb66e130049f21b50 : SUCCESS in 1h 22m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f958ae0e6c1b4689a9434e78acc5c311 : SUCCESS in 34m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0b0cb2821c20403b8e1905389e96e8e6 : SUCCESS in 39m 42s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/4bfdf780f85d456dabd733545cd68b2b : SUCCESS in 1h 13m 26s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/6c281256a3eb48e49149517326629554 : SUCCESS in 1h 07m 59s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/fc7b76b0e6a548b98d22d2e0f263218c : SUCCESS in 1h 10m 13s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d1d1d7ca9c534bcf9269bf71043b1e5c : SUCCESS in 57m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/34012ce345a3497f9c5990400a836d33 : SUCCESS in 27m 13s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"da72f10f31066b21b237b6a5cae364de127e5f94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-08-31 13:56:42.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"1980eeb253d87e8e792df3e6b77d719c1f550468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 16:27:18.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02dbc936e5904395b02c796aa58f7e30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7bdf95c43d64f018687c9d653646078 : SUCCESS in 4m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6c314fa998944415aa8f51015f55a9fe : SUCCESS in 1h 23m 28s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/6164c064d58a4eae8f3bad581c43b9dc : SUCCESS in 2h 25m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/09ccc8fa6664460d963e3e8a3621ab42 : SUCCESS in 1h 48m 42s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4af408319c76487ea6dd5f7e275079b4 : SUCCESS in 1h 23m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a89e36ed5a354cb6b545f4b29bfa1da8 : FAILURE in 1h 55m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e1d21b58121745a785356f96a55cc4e0 : FAILURE in 1h 20m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/65596baab64041778993167cc4e64f09 : FAILURE in 1h 41m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3550c2c370f14e2e967a02aeaeddcdb1 : FAILURE in 1h 28m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1722ac9a810e42d1b8a0d038ea3376df : SUCCESS in 37m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dd3a087402e24423b05b49a2cd1a81ee : SUCCESS in 39m 08s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/0721b78fd4b84f7ba28163a09251be88 : SUCCESS in 1h 28m 57s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/fd8e0a828755436ab4aa412d057e20a2 : SUCCESS in 1h 27m 00s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c1fa39d73412467b895383db9f21300f : SUCCESS in 1h 51m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0551e0bd3aab4006864036cd073c9bc6 : SUCCESS in 54m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/38ca8e3562464873b8c4a92c2cfeb563 : SUCCESS in 34m 22s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"334105482bce31b94ab6e24dddc338d2b8d5fb57","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-31 18:29:18.000000000","message":"Patch Set 44:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/44/upstream-check/manila-tempest-plugin-ontap-dhss/1f85151/ : FAILURE in 2h 00m 59s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/44/upstream-check/manila-tempest-plugin-ontap-no-dhss/8da1b16/ : SUCCESS in 1h 09m 47s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":44},{"id":"b5e2bf21dd20de483a745720ceaeac8a45857c92","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-01 07:08:29.000000000","message":"Patch Set 44:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/44/check/DellEMC_Unity_Manila_dhss_true/c8708e8/DellEMC_Unity_Manila_dhss_true/771 : FAILURE in 2h 34m 20s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/44/check/DellEMC_Unity_Manila_dhss_false/367376a/DellEMC_Unity_Manila_dhss_false/811 : SUCCESS in 1h 15m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":44},{"id":"492c9572a5c39c9a282b9de070ab219422e159a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-09-01 14:40:32.000000000","message":"Patch Set 45: Published edit on patch set 44.","accounts_in_message":[],"_revision_number":45},{"id":"792270044b51f6efe983dc2a16ca1ff212e346de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 17:14:50.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/270b5b91a7054b6a9fa6e2ee371c9842\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff1b985dab2b4b348a3f11ef18148091 : SUCCESS in 3m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7a603eab921142599e95457bcb1fc8aa : SUCCESS in 1h 29m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/59ff15b02d68438182066e70f637b5ea : FAILURE in 2h 19m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c459ff21ba204d3aa5d46faba65dd102 : SUCCESS in 1h 20m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3af8bc5c737b42f4963c6b26dc298f32 : SUCCESS in 1h 39m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/dda749338c9a4192aa23116b67c21ad3 : FAILURE in 1h 05m 48s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/52ea63c2272b47e9853845698fbccf73 : FAILURE in 24m 27s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/018b538820224ede9db3ecc7ade2d9d7 : FAILURE in 26m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2f747f85087c4a1bad0eb0755c391dc5 : FAILURE in 1h 19m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aeac2f92d08b4a908bae14a67ece9ad9 : SUCCESS in 31m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0c5133829a23404ebca80aea8d3fb90a : SUCCESS in 47m 34s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2ae7499339bc4874a0fa493116bf5e61 : SUCCESS in 1h 19m 31s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/441b82041869470885e7acea76b3c6b6 : SUCCESS in 1h 16m 46s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e2d97f6bd207461a92fd52f63a0ac7a1 : SUCCESS in 1h 06m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/731de3a61a8e419399d3dcb65a089cff : SUCCESS in 52m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c5aa65731122453494314d32668d75ad : SUCCESS in 34m 13s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"d473d9fd3dfdba20a1990c9161fcbf1f07a3430e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-01 19:26:31.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/45/upstream-check/manila-tempest-plugin-ontap-dhss/2605447/ : SUCCESS in 1h 30m 06s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/45/upstream-check/manila-tempest-plugin-ontap-no-dhss/3b14bfc/ : FAILURE in 2h 04m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":45},{"id":"5d9a975d00bb53b27d00aee7c91e22a3e3d21422","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-01 19:37:24.000000000","message":"Patch Set 45:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/45/check/DellEMC_Unity_Manila_dhss_true/bbe1d53/DellEMC_Unity_Manila_dhss_true/777 : FAILURE in 2h 43m 24s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/45/check/DellEMC_Unity_Manila_dhss_false/f43bbbd/DellEMC_Unity_Manila_dhss_false/819 : SUCCESS in 1h 07m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":45},{"id":"e9ffbd8a95c1882755d322882eab46ca624ded04","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-09-01 21:57:33.000000000","message":"Patch Set 45: Code-Review+1","accounts_in_message":[],"_revision_number":45},{"id":"a6adcf68206f02a3119f7bc95bb109835917f3c6","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-09-02 10:12:05.000000000","message":"Patch Set 45:\n\n(2 comments)","accounts_in_message":[],"_revision_number":45},{"id":"8cdbc3b138042cf4bddb5ff25da13e5c8d42750a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 12:40:09.000000000","message":"Patch Set 45:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54f49e293b694a35968718f34d8df525\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/414a07e660dc4b0295a01d804eaefdd4 : SUCCESS in 3m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bec2945b4cf34ee48c208c85ee76164b : SUCCESS in 1h 07m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9475885755d64a7b814009de36eff9fc : SUCCESS in 2h 20m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/51e5e8f1d1344a4ea184bdd32b4bf8cb : SUCCESS in 1h 10m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1fafe2c1bacd45168e64d3a123187613 : FAILURE in 1h 18m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f9c2990b70ec434fa59fc1f60236bc2f : FAILURE in 55m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/447deb21c3b240219930d0df268f1f07 : SUCCESS in 1h 11m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/54ed70cbc6ad4c3aa453af6fb5f12bbd : FAILURE in 59m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e211f23cd9814117adfb9a6bb30289a2 : SUCCESS in 1h 18m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2521f0a048864524bceb00ee354cdf04 : SUCCESS in 39m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1d55ab99177e4cec87a39c506cf3eba0 : SUCCESS in 36m 52s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/0b54364b41734a90a56826e6cd1c6ceb : SUCCESS in 1h 24m 51s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/094598dc5d474cdcab32467d937045f2 : SUCCESS in 1h 11m 25s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/700c1e1841e64d45a0fd3d4a5ac9eb4f : SUCCESS in 1h 40m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9ef010746d79481ba0d6c7a5ac44dbd5 : SUCCESS in 56m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/58fe261ddc8744d2ac433937b946c03d : SUCCESS in 26m 27s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"f73a08768fa8b9a399644f5eabe5ef032d18f0fc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-02 14:34:00.000000000","message":"Patch Set 45:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/45/upstream-check/manila-tempest-plugin-ontap-dhss/1cf7055/ : SUCCESS in 1h 53m 01s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/45/upstream-check/manila-tempest-plugin-ontap-no-dhss/a39f949/ : RETRY_LIMIT in 11m 33s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":45},{"id":"0e425b4e939b5b2b86b7046ce331235fef5e633a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-02 16:54:40.000000000","message":"Patch Set 45:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/45/check/DellEMC_Unity_Manila_dhss_true/e841870/DellEMC_Unity_Manila_dhss_true/781 : FAILURE in 3h 34m 38s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/45/check/DellEMC_Unity_Manila_dhss_false/0c39c9b/DellEMC_Unity_Manila_dhss_false/823 : FAILURE in 1h 03m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":45},{"id":"3509a47061f0fb53b9109e7ec751e1cbb49ec951","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-09-05 11:03:40.000000000","message":"Patch Set 46: Published edit on patch set 45.","accounts_in_message":[],"_revision_number":46},{"id":"6d4cc5b68fe52c1b83612a807fa81bbca0a60c43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 12:49:52.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca7e3956ff124e3985fb9b7a832c3001\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b5ea16dbc44ef8885359c3e90d64c5 : SUCCESS in 3m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0b04d8e957a0491e90055cd5dd608764 : SUCCESS in 1h 28m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/da947eb651fb4580b7f64a684c6ffb42 : SUCCESS in 1h 44m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bd58a13517ec48689d4a31b94a3eafdd : SUCCESS in 1h 05m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/67db12863d2b45c48be05abbfa6ac329 : SUCCESS in 1h 07m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ed08e583c92043be995b5e4a8e079e82 : FAILURE in 43m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/29e8a577bd094914bb5954c1ed602651 : SUCCESS in 1h 13m 37s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e37942d36c4443ce8dc144344499f825 : FAILURE in 44m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fd66b151b7104d7ab74374f68979c1fe : SUCCESS in 1h 11m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/533c30492bf047048089f5dba1e74614 : SUCCESS in 43m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c96a98c9dfa4467195784564ece2115c : SUCCESS in 40m 15s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/fe08217b563f43ba9df8d698b4863fcf : SUCCESS in 1h 10m 25s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a6d3faa535124f5b88e42d42e73a8ede : SUCCESS in 1h 03m 14s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/8e41cfbef8a44b79b073efb7fefb2bd0 : SUCCESS in 1h 19m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e74cf322c2fe4770b9028f9ee01c113c : SUCCESS in 1h 00m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/14cc3a860222436abb8625876be20540 : SUCCESS in 24m 23s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"81b43703918920b8c59eec851982a3349ff60fe4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-05 12:52:13.000000000","message":"Patch Set 46:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/46/check/DellEMC_Unity_Manila_dhss_true/c882150/DellEMC_Unity_Manila_dhss_true/786 : FAILURE in 1h 09m 09s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/46/check/DellEMC_Unity_Manila_dhss_false/c850e07/DellEMC_Unity_Manila_dhss_false/828 : FAILURE in 42m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":46},{"id":"ad920e3431222c4e41f70a71186a011a567863c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-09-05 14:51:51.000000000","message":"Patch Set 47: Published edit on patch set 46.","accounts_in_message":[],"_revision_number":47},{"id":"e7db38ec988b6555502f4f5326fe3fc3a164ff9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 17:21:25.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b792beeb26464f50bf4590c04d7d454d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d045f2e1a8f348f08727c7bca633edee : SUCCESS in 3m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8c7315bb47784faa9790014dd35fe57e : SUCCESS in 1h 50m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/017fed793d3341cd9817ca978bbfe3f1 : FAILURE in 2h 22m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/53172d7699c340f9b9171da1a8761962 : SUCCESS in 1h 18m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7f25acd316d5448ab85ea957cfd83af9 : FAILURE in 1h 38m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/feae991a42fc4daa84a0651b02a751d3 : FAILURE in 1h 02m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c4240a2a985a4ce7b5b49fbfeeefebd5 : SUCCESS in 1h 41m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8f5b3f9bfa184687a14496277fd615fe : FAILURE in 53m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/67abb5fa217e4289a2eef8884503aa82 : SUCCESS in 1h 41m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d1d23bc625a848de81800e27d24ddcdb : SUCCESS in 39m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3e20e32c2e2e4b8eb31ef033b485d2db : SUCCESS in 45m 23s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/dd458a4692814c0d8205514121eec1fb : SUCCESS in 1h 17m 42s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/ab284a396ae44451a9ba539d9a2d6800 : FAILURE in 1h 28m 46s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/02cd02f88f7d4c859e1909b3e01d9b69 : SUCCESS in 1h 09m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b1b380e90aa549b4ba8b56ce3ee2d7f9 : SUCCESS in 57m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9d578dff2ce74b3a9efe2c458ebf3bcf : SUCCESS in 32m 25s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"8cd5699133fe83c1798c397aee30830467aceecc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-05 19:07:05.000000000","message":"Patch Set 47:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/47/check/DellEMC_Unity_Manila_dhss_true/951a66a/DellEMC_Unity_Manila_dhss_true/789 : FAILURE in 1h 11m 00s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/47/check/DellEMC_Unity_Manila_dhss_false/2d59fd4/DellEMC_Unity_Manila_dhss_false/831 : FAILURE in 42m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":47},{"id":"9d19b865c72d4cf79acfcc91c29d4bf79c944c4b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-06 22:26:30.000000000","message":"Patch Set 47: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"c2928aa7bf3748f0aa6d086641f2d868d11b15ea","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-08 14:51:28.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"7d61df494dfe24e6861e3ee99211a9028d669130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 17:26:06.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36179f70f8e745ea9809dc5a99e381f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e73f0aa9f5d45c8b4979ffe73625ccc : SUCCESS in 5m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9eb927cd7ca54de1b46b203b8a83e67f : SUCCESS in 1h 22m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/52805fe467a64d69810df11b35c278cf : SUCCESS in 2h 03m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8acf1f540cae4054a3a1e28ff3bfb4c9 : SUCCESS in 1h 27m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9a6f4377e52e4c9b84ae58f14ee63604 : SUCCESS in 1h 37m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/59416790bfee402c84a36b891228c044 : FAILURE in 1h 46m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7cecc1043511499785192001409e5666 : SUCCESS in 1h 31m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2eb38c3a764d4d23ae1b95df39e4957a : FAILURE in 2h 31m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fe513be0db8440a7aaa354c0f6144a64 : SUCCESS in 1h 14m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c57ce6332ddb4efcb837bb2ebc2c00e5 : SUCCESS in 40m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/653a773b66a246908a7401461704b284 : SUCCESS in 43m 29s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/f6b9e71b1b5045bb84918ae30427681b : SUCCESS in 1h 24m 47s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f24afb66c65747e2a9c528ee6238268e : SUCCESS in 1h 19m 34s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/102587664e574786bb92e6c68992d489 : SUCCESS in 1h 34m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8291dc9e2640468c96164f52e25f1ad4 : SUCCESS in 1h 07m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e9e81d857b784b94831eb140f538bf08 : SUCCESS in 33m 27s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"edcff32fe77338755282f26eca2e9fcd0507adbc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-08 19:29:29.000000000","message":"Patch Set 47:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/47/upstream-check/manila-tempest-plugin-ontap-dhss/4a0badf/ : SUCCESS in 2h 02m 44s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/47/upstream-check/manila-tempest-plugin-ontap-no-dhss/a5ffc2b/ : SUCCESS in 1h 33m 09s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":47},{"id":"d1af5bd96ecf43f55bf52cc26dff40ee79add2ae","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-08 21:13:19.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"8ae65f767e08fa0dc1e4bd2c9d838994299de190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 00:24:45.000000000","message":"Patch Set 47:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b3b2e222740460ab9c54c2eba6eec3b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92ea3e665fd9455e8d3d23ba0c66474a : SUCCESS in 3m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/81a32e0976034a319eec8e2d77b8267e : SUCCESS in 1h 20m 28s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/58303291bdcb4feea7a3a2e106c8f349 : SUCCESS in 2h 11m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e6d35ffc41024c999e86d608f65b5c93 : SUCCESS in 1h 15m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bf61a26ccd0e46209b7acd3d7086e270 : FAILURE in 2h 00m 24s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6b99b472558844bb9a2c603d75a7bbb9 : TIMED_OUT in 3h 05m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/59cc8e7cc9d84678881daaeabc996279 : SUCCESS in 1h 18m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/faa0601ac10a4a80a89def402defc6ae : FAILURE in 1h 46m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c07fa59e0d074d0b8abc5ccd13246c94 : SUCCESS in 1h 26m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6ada8ffbc6b34d42b0640cc76a60983c : SUCCESS in 44m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/56d8a01d774a4821a4ebd9177ecd0069 : SUCCESS in 38m 09s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/4c5404596da4490f85bc0b6be8bd8ee8 : SUCCESS in 1h 05m 24s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/bff1852c5e3b4abe94f2b45a98e4c428 : SUCCESS in 1h 23m 53s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d3f006722fd443259b95423faf31e831 : SUCCESS in 1h 14m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/102222dd05a7406d98561f1ba3533ce7 : SUCCESS in 43m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/71fbbfe1d44b45d09cfd0eb7d6439d04 : SUCCESS in 23m 53s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"2bea6a23a520db5eb2c164aac5570460d7fe3315","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-09 03:34:53.000000000","message":"Patch Set 47:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/47/upstream-check/manila-tempest-plugin-ontap-dhss/019de3d/ : TIMED_OUT in 3h 09m 15s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/47/upstream-check/manila-tempest-plugin-ontap-no-dhss/254bcce/ : FAILURE in 2h 03m 31s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":47},{"id":"b01775ab109b06e1c2143d47cdf5331ec1b728a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-09-09 14:10:04.000000000","message":"Patch Set 48: Commit message was updated.","accounts_in_message":[],"_revision_number":48},{"id":"2275d060d14bb35746609378bd2d193abd7585fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 16:57:18.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc7e248e052d40008bd0d0860ef6974a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1111907dd57f4739ad6474a13b86ce5b : SUCCESS in 3m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8f8f2d545e834c73baced97d5b52004e : FAILURE in 1h 34m 22s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/77c02a1c28524e82bd01ff2b8bf66354 : SUCCESS in 2h 21m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7de75e20fe32482da9017fc85f90d4fb : SUCCESS in 1h 13m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/196f0bec6a184bfa81c4056dcdeb19c2 : SUCCESS in 1h 08m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/28a9ed20aa1b42ac8a7d5f24c578b23d : FAILURE in 1h 58m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e6da80cbd8f24807a5f061f88614b3b0 : SUCCESS in 1h 31m 27s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c1e86099b3704ef1b0d77b3a092ca76e : FAILURE in 2h 41m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3efd0644c5744d469f9e2ac5d0fa769c : SUCCESS in 1h 17m 23s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/74fd4c0e7d5e4a0a9e538b498fd5e5bf : SUCCESS in 26m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7d50ee2998104e239a145c0d0a225137 : SUCCESS in 47m 29s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/8e6db3e45b20411ab2734e9136e6b4f7 : SUCCESS in 1h 29m 02s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/b394d26ade22496b9699936ce71bc6db : SUCCESS in 1h 04m 31s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/230178a1b1854349a39156304f48cd33 : SUCCESS in 1h 16m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b54dca62194b443db8a331db42c4c30f : SUCCESS in 52m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cdf1b7f3ab2b41378f299fe21e3d2b20 : FAILURE in 21m 22s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"b8f70accbc1d94f7472048e1af0cb17ada2a611d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-10 07:25:33.000000000","message":"Patch Set 48:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/48/check/DellEMC_Unity_Manila_dhss_true/71a976c/DellEMC_Unity_Manila_dhss_true/829 : FAILURE in 3h 02m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/48/check/DellEMC_Unity_Manila_dhss_false/62c1b88/DellEMC_Unity_Manila_dhss_false/875 : SUCCESS in 59m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":48},{"id":"7808fa88e2d2d7759a4b4614edfe4b91caca9f04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-09-30 09:38:33.000000000","message":"Patch Set 49: Commit message was updated.","accounts_in_message":[],"_revision_number":49},{"id":"23263cf7caef41f686d94ab641f593c88781f13f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 09:43:37.000000000","message":"Patch Set 49: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/859950 is needed.","accounts_in_message":[],"_revision_number":49},{"id":"eb56b3b6447c9f881c2e3ec2938dc233811dfb39","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-30 12:32:25.000000000","message":"Patch Set 49:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/49/check/DellEMC_Unity_Manila_dhss_true/88bde1f/DellEMC_Unity_Manila_dhss_true/947 : FAILURE in 2h 48m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/49/check/DellEMC_Unity_Manila_dhss_false/9894444/DellEMC_Unity_Manila_dhss_false/1007 : SUCCESS in 57m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":49},{"id":"3d63942614356b5d62dc77229c870a5a94c6371d","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-09-30 13:31:52.000000000","message":"Patch Set 49:\n\n(1 comment)","accounts_in_message":[],"_revision_number":49},{"id":"d0ae86071e038e9b5ead56f754f2774da36f5bc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 16:44:02.000000000","message":"Patch Set 49:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b668234b84a34535a6672a69beb01dbf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3daf36a0132c4064a08f24496353cb9f : SUCCESS in 3m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ced039512f1c463589eca3907d4193e2 : SUCCESS in 1h 30m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d99a12e108bf4068bb3618a2b2e31fe5 : SUCCESS in 2h 28m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c969fc4f740040129b8d509806168e81 : SUCCESS in 1h 33m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/60baadca9cc44fef9b4e149dbc6ebe6e : SUCCESS in 1h 39m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/97d4a0ac486d4c55b0e43ad223d992d6 : TIMED_OUT in 3h 04m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/75e4f954cc9b4891bb9a5f59bc4fcfcb : FAILURE in 1h 33m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ee4e73b96e3d44859cba6378b01a3efe : FAILURE in 1h 29m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/35f7f83adc414c3891f59b0c7b60e518 : FAILURE in 1h 32m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cd689bb830324ef7a3ea14e04611f6ea : SUCCESS in 44m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/42ed9eb5801742e8a636c8940e71a012 : SUCCESS in 42m 16s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/d696736a8b624d77925cb9d121778290 : SUCCESS in 1h 44m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/4502d40aa7b64293a9c4373e2dc6f096 : SUCCESS in 1h 07m 55s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/2e60d9caa8b7437db462420a5e794351 : FAILURE in 1h 28m 50s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d40c81e996e0439193db38d44bd388b2 : SUCCESS in 49m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/63ae46bc789b45f7a140c9c3438de4c8 : SUCCESS in 34m 45s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"99931a54e53cc255f88681b8b32ebc9185060b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-30 17:06:22.000000000","message":"Patch Set 50: Published edit on patch set 49.","accounts_in_message":[],"_revision_number":50},{"id":"a94493797716165210a93f436f520a0a965c6cb3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-30 17:06:57.000000000","message":"Patch Set 50:\n\n(1 comment)","accounts_in_message":[],"_revision_number":50},{"id":"7b1d10d1434075ae85f9ef08394d014ec8fc48f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 20:03:14.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98b534aedb834ca48c4799b2dfc0abcf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bbb02d986724612a26b0f50ce77ae4e : SUCCESS in 2m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ceb3928048724972b5c073c6088969b1 : SUCCESS in 1h 22m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/82a54153fdda47f796277ec51c19d5fe : FAILURE in 2h 50m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/20101cce454e48b6b2b4def3ad6a942f : SUCCESS in 1h 33m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/31c5c81c33cf41f8ad4244b56623dbed : FAILURE in 2h 23m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0e33d190271d4ab8a97d5e214a69e4ba : FAILURE in 1h 43m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c094a73eec7f470b9b22db51218cb421 : FAILURE in 2h 08m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/04957e17ab3147e0b7bd79c91654f397 : FAILURE in 1h 47m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fa5b713f01774bb3ac0660b8a61dff41 : SUCCESS in 1h 23m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4f656900846c45459298f51bc4d0890b : SUCCESS in 33m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7eb3beac4c2640c5873f3924beb7fab0 : SUCCESS in 46m 30s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/ac27bc73efe741ef835b87b5ba6dbe58 : SUCCESS in 1h 37m 06s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d49e3fcc0d894219a5f615b4592cd7ea : SUCCESS in 1h 15m 09s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/dcb5a01afddc4704ac476b5d35f18532 : SUCCESS in 1h 12m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/010b392a59a9479eb018272e389825b4 : SUCCESS in 44m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bda9c71fa90e4685914c94cdc8323623 : SUCCESS in 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"fc3b849e99bc3628f2c1a2569f07ad3e8c6e6982","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-30 22:10:23.000000000","message":"Patch Set 50:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/50/upstream-check/manila-tempest-plugin-ontap-dhss/d1a2a91/ : FAILURE in 2h 05m 54s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/50/upstream-check/manila-tempest-plugin-ontap-no-dhss/f68a861/ : SUCCESS in 1h 43m 21s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":50},{"id":"a804313630cd07ecd26bd2b181539c97fd04719c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-01 04:31:14.000000000","message":"Patch Set 50:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/50/check/DellEMC_Unity_Manila_dhss_true/c8444a1/DellEMC_Unity_Manila_dhss_true/952 : FAILURE in 2h 58m 20s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/50/check/DellEMC_Unity_Manila_dhss_false/0277028/DellEMC_Unity_Manila_dhss_false/1013 : FAILURE in 1h 36m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":50},{"id":"d829521525c45bf0519d060fd066d4303a1c35cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-10-26 08:55:19.000000000","message":"Patch Set 51: Patch Set 50 was rebased","accounts_in_message":[],"_revision_number":51},{"id":"9c100d4c74dd236e1727e44beb23dd1b58d7def1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 11:38:46.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b59cbc7b73ab4a1d9ef73f68a24016f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b7c05d368b74c2b8769dbd9b27bb4b1 : SUCCESS in 3m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/89591c677c6746ad856e0ed3acc20cb5 : SUCCESS in 1h 24m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c9ec147dda054edeb69b0b5da224c0cc : FAILURE in 2h 24m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7b9db81d9f9f4d6891998d4928685015 : SUCCESS in 1h 21m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/099806e71b2146cfa2679202ba982701 : SUCCESS in 1h 29m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6652b6ef148c45348d97041feddbe67b : FAILURE in 1h 36m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/eaa9331cd33548789f4970456946ef8c : SUCCESS in 1h 14m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/949fde42d7a14c77988c880d5b24805b : FAILURE in 2h 38m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d27c5500972f4c6e80c3fd6817f03334 : SUCCESS in 1h 08m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4fbba17b75b749e4b3878da98b88bf72 : SUCCESS in 37m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/db5491b67e5a44bdacc397c21cfbf8b7 : SUCCESS in 40m 23s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/1f4f6e8335b54d98bf2380ba8e12cf97 : SUCCESS in 1h 10m 41s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/43c09e21a23d4ca4a24b4db03af302ff : SUCCESS in 46m 15s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/81bd3800188146f6900ea4b96a5f4bb1 : SUCCESS in 1h 12m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/04a61a12aa6e43a2bcd81afbdee865f0 : SUCCESS in 28m 30s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e133055de9a94762af7aba67c4e6d796 : SUCCESS in 38m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1ddeb5ec207e40e78d0729517f4b7f03 : SUCCESS in 25m 39s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"e2ab60c8620c4586de6354af6430c55c7f6b6b07","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-26 12:04:18.000000000","message":"Patch Set 51:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/51/check/DellEMC_Unity_Manila_dhss_true/6335160/DellEMC_Unity_Manila_dhss_true/998 : FAILURE in 3h 08m 44s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/51/check/DellEMC_Unity_Manila_dhss_false/3342ed5/DellEMC_Unity_Manila_dhss_false/1062 : SUCCESS in 1h 12m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":51},{"id":"99c2adfc1f24ef67bd722069b099d0d012bf55b5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-26 14:49:29.000000000","message":"Patch Set 51:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/51/upstream-check/manila-tempest-plugin-ontap-dhss/858c059/ : TIMED_OUT in 3h 09m 27s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/834220/51/upstream-check/manila-tempest-plugin-ontap-no-dhss/e8c64e1/ : SUCCESS in 1h 55m 42s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":51},{"id":"3184be5293f0cdf762d6c0b6cc6845925f05d533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-03-01 19:23:00.000000000","message":"Patch Set 52: Patch Set 51 was rebased","accounts_in_message":[],"_revision_number":52},{"id":"2969e4b02e4e9e1192c7eabf610e9446bf452556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-03-01 19:25:38.000000000","message":"Patch Set 53: Published edit on patch set 52.","accounts_in_message":[],"_revision_number":53},{"id":"72dbe33ef0340bfa1a131729f96c8ae17016f832","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 21:31:22.000000000","message":"Patch Set 53: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbe319d6d200468b95c119379a97d8d1\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c43803a10a8d4ccf9aad8dbbd2ef541e : FAILURE in 25m 51s\n- manila-tempest-plugin-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3a83aa6ab94d4d629a74bf6cc86261aa : FAILURE in 1h 59m 55s","accounts_in_message":[],"_revision_number":53},{"id":"eb27d99809e5d5dddc42878f173f6f15c7e92bbe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-01 23:29:05.000000000","message":"Patch Set 53:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/834220/53/check/DellEMC_Unity_Manila_dhss_true/56f9a7f/DellEMC_Unity_Manila_dhss_true/1331 : FAILURE in 51m 30s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/834220/53/check/DellEMC_Unity_Manila_dhss_false/8c6e924/DellEMC_Unity_Manila_dhss_false/1431 : SUCCESS in 1h 06m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":53},{"id":"e2e448d088d36c462d03a1a9896cd4f3403d0643","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-04 23:59:04.000000000","message":"Uploaded patch set 54.","accounts_in_message":[],"_revision_number":54},{"id":"cbeaeefa7eba019b36bc04e65b194c3ff5179279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 01:14:16.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66b8da2b09d74985bc6d7a3cf1b5b29d\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/46c9185263cb4c6da7ded694e4c6414d : SUCCESS in 1h 09m 46s","accounts_in_message":[],"_revision_number":54},{"id":"499deedd5263336c4286992841fc80b640fb31b1","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2023-04-05 07:55:06.000000000","message":"Patch Set 54:\n\n(2 comments)","accounts_in_message":[],"_revision_number":54},{"id":"e9b9bad53514a517f9179ce6710aac3a37291612","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2023-04-05 16:28:08.000000000","message":"Patch Set 54: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":54},{"id":"30fd52c4ada0f18cc142c9f8ea9d17f7792bbc39","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2023-04-05 16:28:14.000000000","message":"Patch Set 54:\n\n(1 comment)","accounts_in_message":[],"_revision_number":54},{"id":"4e10d8b56df6d1729a18c1dd9f39607bb22c2774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 17:49:28.000000000","message":"Patch Set 54: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f699d6b6e0b4ccc81b9727e6079dd10\n\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/378b8b4601614528a6d598496195857d : FAILURE in 1h 15m 33s","accounts_in_message":[],"_revision_number":54},{"id":"31ff1902f6f9aa85ebfc8b45b609db044c1d1202","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-05 19:28:10.000000000","message":"Patch Set 54:\n\n(2 comments)","accounts_in_message":[],"_revision_number":54},{"id":"f4824b4b8e64acc648959046155d0e1e60acbafc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-05 19:36:03.000000000","message":"Uploaded patch set 55.","accounts_in_message":[],"_revision_number":55},{"id":"6fa70c1910e52dd03e250bac7a17b4caceda793d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-05 19:36:51.000000000","message":"Patch Set 55:\n\n(4 comments)","accounts_in_message":[],"_revision_number":55},{"id":"b231643386cb68b952882d537d9fe22ed7227acb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 22:00:23.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba1372c95de94d7fb004760d80fab18b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c67bdc451b95407bac65564433972c4a : SUCCESS in 2m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c6f47d3f2e284959850410921662df83 : SUCCESS in 1h 20m 28s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c5694042d4fc4783b5486fb8ecf2a9d0 : SUCCESS in 2h 18m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/013ff36212124016ad349c6e2c3b45e0 : SUCCESS in 57m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e15c41f1ff634081865d06dffa5b52a8 : SUCCESS in 1h 09m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/85380b011d2c48acb22bc2d81a43533d : FAILURE in 1h 08m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3950b78093fd42fba0b8db8f6e56b332 : SUCCESS in 1h 17m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/186c9258362c4d0b82cb2decfbd88698 : SUCCESS in 42m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fe2899068aea4fbc830db1e479c77517 : SUCCESS in 35m 18s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/f308991d1c244a84ada316f5fb52df97 : SUCCESS in 1h 20m 20s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d024241e16d648b3a236c772edb3ae5e : SUCCESS in 49m 23s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/cb354de9808845f5af915bde7f28da47 : SUCCESS in 21m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c3b5fb4ac16644c89c9ab4b6f0bef764 : SUCCESS in 57m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/da2198f3846b44bc9ed8404412501541 : SUCCESS in 29m 13s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"b8b8a81b9ab041760c7ddd72b8f4f319412a4ff1","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2023-04-06 09:31:30.000000000","message":"Patch Set 55: Code-Review+1","accounts_in_message":[],"_revision_number":55},{"id":"b0a7a5efa380b4ff8752e4ca5ae861aaf6ac3f63","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-04-06 12:28:49.000000000","message":"Patch Set 55: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":55},{"id":"2b01cf91c64d1aea7bac8554f36963a70f0d265d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-06 15:13:00.000000000","message":"Patch Set 55:\n\n(1 comment)","accounts_in_message":[],"_revision_number":55},{"id":"fc11d7129b1573e6829ad39dc8afcba47ad4a066","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-06 15:13:48.000000000","message":"Patch Set 56: Commit message was updated.","accounts_in_message":[],"_revision_number":56},{"id":"d4a77b3019a49503332b8f2447c87d60f489acaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:56:57.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebbb3086dc9d45beb0a052b168cfc755\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d31f6691e794e4399db1573798b89c2 : SUCCESS in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c72d95b747ed4303a7f66ec8a0f692f4 : FAILURE in 1h 29m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d68b17f100c54068b46b62ea88c99a0d : SUCCESS in 1h 40m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ccb257352f7c4033936a67d1feb01b60 : SUCCESS in 1h 19m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/796f51ecbf04481898efd7a529c89ed8 : SUCCESS in 54m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/86dbce84986d47acaea16d8b875cea86 : FAILURE in 29m 18s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fd63a1d3c2fa4895944909e9739a7a62 : SUCCESS in 1h 01m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b89a3fea440b429a87cabd3f25bb37b0 : SUCCESS in 35m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9ce8ecd6b91c4c309b14135b85d56f8e : SUCCESS in 32m 49s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/9e50afa123bf4683a13c612b9be34f2e : SUCCESS in 59m 40s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/55f4034d8cc345559a191ab576c30a82 : SUCCESS in 1h 30m 31s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/bdd2f21de1c948119a5baa19dc90d6cc : SUCCESS in 26m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/741b955c73c1498fb5909d80e5c9f104 : SUCCESS in 36m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/05f5ce53adaa42b4aef9954570983d79 : SUCCESS in 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"245ce962cfc833f1d4d82f6e3c08b5419bab82f1","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2023-04-06 17:12:50.000000000","message":"Patch Set 56: Code-Review+1","accounts_in_message":[],"_revision_number":56},{"id":"bea67513e44a1d2b2129d54774e1e6eb31c40006","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-10 17:44:55.000000000","message":"Patch Set 56:\n\n(2 comments)","accounts_in_message":[],"_revision_number":56},{"id":"4f8b6ed0192886ec58459bb16f7435334f615e80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 19:44:02.000000000","message":"Patch Set 56:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99631c46120944758b013ecf51a3a23d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29df2ef187004ddc91dc0fc3cf9e6bc1 : SUCCESS in 2m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/78ba77013b6e44c59bd082baf0f8c450 : SUCCESS in 1h 09m 34s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/228ba3a85ab3437aad68d32429237247 : FAILURE in 1h 55m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f8e8d9946b2d49f891ac6180c978d8eb : SUCCESS in 1h 29m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9c0da30e76a74045b305065a7a58e8df : SUCCESS in 1h 06m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b22c4b48b05a46f1baa96ca27f075d55 : FAILURE in 1h 28m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bc8c81db78ca4506a3c057c321a54dff : SUCCESS in 1h 13m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/941bdc1bad454510951ab0084a52ebe9 : SUCCESS in 30m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/33a41ac9d35c40c294f427de61eae5fe : POST_FAILURE in 59m 00s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/22816db0af3f45d6a7c093f2f5cc22b9 : SUCCESS in 1h 53m 22s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/95683636c2414520ac1b41fa95a63c78 : SUCCESS in 56m 12s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e7f58eadad4c4fd28035e5e73844d258 : SUCCESS in 25m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/873942e1db634a4d96eabce7545f2780 : FAILURE in 46m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/53953fe9bc6943f79158adbcec3cad38 : SUCCESS in 17m 43s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"3a835808866c3aee8ccfe0a0cef0148f9707862d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-11 22:29:07.000000000","message":"Patch Set 56:\n\n(1 comment)","accounts_in_message":[],"_revision_number":56},{"id":"9728f2a911ca8c0ea680488e6ba0a2cb91b7b8e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 01:16:57.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c03dde5fabeb4ccaac6a71476c9c3234\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aabdaf849bc488999cb5da46ca3156d : SUCCESS in 2m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b3484748514a435689f8bc54f971baaa : SUCCESS in 1h 28m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b8f5f6966a1846fe819913c7cb7368cb : FAILURE in 2h 41m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e3e76c55eac54b3f8fb7a17827c1a99f : SUCCESS in 1h 10m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3215594d069f44f491c74e534b2c05b7 : SUCCESS in 55m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2d057d7a66db481b81257208d9d5a3b9 : FAILURE in 1h 23m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8e62ee23580d40d1969f578606be5b5f : SUCCESS in 1h 30m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/badd4bdfcf734383ba944aad6d0fa276 : SUCCESS in 26m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/948782cfa38641979b9f99bd258c518f : SUCCESS in 44m 03s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3394cf8c6f784bbab24298eb0b37f557 : SUCCESS in 1h 06m 26s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/2ea1b7a49c404aa6a9a232ddef0c555c : SUCCESS in 1h 02m 27s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f980cb51359343f1a10cb6d2f417c98a : SUCCESS in 25m 39s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/561d87931a784fc9b6c417cd0acfb92e : FAILURE in 33m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8ac3ac1a76574e64a94f79c7046bcdf2 : SUCCESS in 16m 04s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"6a119996b8cb9b213f3829258a8462dcb811e7b1","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2023-04-13 07:05:55.000000000","message":"Patch Set 56: Code-Review+2","accounts_in_message":[],"_revision_number":56},{"id":"9b8cb005ff3155dbbcc64da58c49a68574ca540f","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-04-13 08:15:46.000000000","message":"Patch Set 56: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":56},{"id":"b9cad980511849793d8e00061768a739300fe62f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-13 14:53:00.000000000","message":"Patch Set 56: Workflow+1","accounts_in_message":[],"_revision_number":56},{"id":"b1b919a79c894f878b535bbc6560f184371564a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 14:53:38.000000000","message":"Patch Set 56: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":56},{"id":"bb099c01914b3d42581550b37bb8703ebe84ee93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 16:09:47.000000000","message":"Patch Set 56: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16ee23d52b624c848340c5cebaec9e6b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c42407673aab43e3a58844b93a633f66 : SUCCESS in 2m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f99c68ed7b0f45aaa4a84105b8234557 : SUCCESS in 1h 11m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a32ce7c0460b47a7a81bd9fe99026555 : SUCCESS in 38m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b64db47373174c4797b9fb820352ec9e : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":56},{"id":"fe2f1b1a4c2f8fbbe62d34b8faf76dde92035ddb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 16:09:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":56}],"current_revision_number":56,"current_revision":"60626d51e9a46b0c16cbb7124038053e4ce9219b","revisions":{"1bf8c0c741214956e76c63a5670000da2f3401a4":{"kind":"REWORK","_number":1,"created":"2022-03-17 19:25:27.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/1"}}},"commit":{"parents":[{"commit":"669e92c68e963089213d3deb8734ef61c6d6e9d9","subject":"Merge \"Add support to chose ssh key type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/669e92c68e963089213d3deb8734ef61c6d6e9d9"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:25:48.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1bf8c0c741214956e76c63a5670000da2f3401a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1bf8c0c741214956e76c63a5670000da2f3401a4"}]},"branch":"refs/heads/master"},"20fb65f51f71057c5f2f48a07f9188fc7851d170":{"kind":"REWORK","_number":2,"created":"2022-03-17 20:15:53.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/2"}}},"commit":{"parents":[{"commit":"669e92c68e963089213d3deb8734ef61c6d6e9d9","subject":"Merge \"Add support to chose ssh key type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/669e92c68e963089213d3deb8734ef61c6d6e9d9"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-03-17 20:15:51.000000000","tz":60},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/20fb65f51f71057c5f2f48a07f9188fc7851d170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/20fb65f51f71057c5f2f48a07f9188fc7851d170"}]},"branch":"refs/heads/master"},"6f99208b12100c35941088a3085132dc73aba349":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-03-21 18:03:12.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/3"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-03-21 18:03:12.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f99208b12100c35941088a3085132dc73aba349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f99208b12100c35941088a3085132dc73aba349"}]},"branch":"refs/heads/master","description":"Rebase"},"1a876ce2d1bc1fd79d4d3d2045b1df73b2910f63":{"kind":"REWORK","_number":4,"created":"2022-03-22 09:51:19.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/4"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-03-22 09:51:17.000000000","tz":60},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1a876ce2d1bc1fd79d4d3d2045b1df73b2910f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1a876ce2d1bc1fd79d4d3d2045b1df73b2910f63"}]},"branch":"refs/heads/master"},"c7378b8f4a8097b4c9935a6a01c2a00cf28901c7":{"kind":"REWORK","_number":5,"created":"2022-03-22 11:16:00.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/5"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-03-22 11:15:37.000000000","tz":60},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c7378b8f4a8097b4c9935a6a01c2a00cf28901c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c7378b8f4a8097b4c9935a6a01c2a00cf28901c7"}]},"branch":"refs/heads/master"},"9bbfd7001bdac6593134cc23f56fd1c7719a3079":{"kind":"REWORK","_number":6,"created":"2022-03-22 11:22:29.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/6"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-03-22 11:22:25.000000000","tz":60},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9bbfd7001bdac6593134cc23f56fd1c7719a3079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9bbfd7001bdac6593134cc23f56fd1c7719a3079"}]},"branch":"refs/heads/master"},"dc9cc3673438249020132daffcbe313a1c091069":{"kind":"REWORK","_number":7,"created":"2022-03-24 11:39:38.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/7"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-03-24 11:39:30.000000000","tz":60},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc9cc3673438249020132daffcbe313a1c091069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc9cc3673438249020132daffcbe313a1c091069"}]},"branch":"refs/heads/master"},"18e86523e9e0cf5fd05f4eafe077db35f3ca057b":{"kind":"REWORK","_number":8,"created":"2022-03-29 14:02:25.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/8"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-03-29 14:02:21.000000000","tz":120},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/18e86523e9e0cf5fd05f4eafe077db35f3ca057b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/18e86523e9e0cf5fd05f4eafe077db35f3ca057b"}]},"branch":"refs/heads/master"},"b4bdd259f0777627814f1ccbc2aaddc4c994ea51":{"kind":"REWORK","_number":9,"created":"2022-03-29 18:42:28.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/9"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-03-29 18:42:14.000000000","tz":120},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b4bdd259f0777627814f1ccbc2aaddc4c994ea51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b4bdd259f0777627814f1ccbc2aaddc4c994ea51"}]},"branch":"refs/heads/master"},"2dbedd3f63f39b371153e3039becf7be90c09f44":{"kind":"REWORK","_number":10,"created":"2022-03-30 06:10:45.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/10"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-03-30 06:10:32.000000000","tz":120},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2dbedd3f63f39b371153e3039becf7be90c09f44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2dbedd3f63f39b371153e3039becf7be90c09f44"}]},"branch":"refs/heads/master"},"3bc164bd0b7e2275ea1e6446aa2a6d7ce0505807":{"kind":"REWORK","_number":11,"created":"2022-03-30 16:42:16.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/11"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-03-30 16:42:10.000000000","tz":120},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3bc164bd0b7e2275ea1e6446aa2a6d7ce0505807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3bc164bd0b7e2275ea1e6446aa2a6d7ce0505807"}]},"branch":"refs/heads/master"},"85f9b0a9f02c51961c3d251ed450bf739540462a":{"kind":"REWORK","_number":12,"created":"2022-03-31 13:40:38.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/12"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-03-31 13:40:30.000000000","tz":120},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/85f9b0a9f02c51961c3d251ed450bf739540462a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/85f9b0a9f02c51961c3d251ed450bf739540462a"}]},"branch":"refs/heads/master"},"df053d8f5a14fa09774eff720f5aa01f48cb944c":{"kind":"REWORK","_number":13,"created":"2022-04-01 11:38:31.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/13"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-04-01 11:38:22.000000000","tz":120},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/df053d8f5a14fa09774eff720f5aa01f48cb944c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/df053d8f5a14fa09774eff720f5aa01f48cb944c"}]},"branch":"refs/heads/master"},"562249c2e379df9b667fa66729ee9441c6242640":{"kind":"REWORK","_number":14,"created":"2022-04-01 13:46:09.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/14"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-04-01 13:46:08.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/562249c2e379df9b667fa66729ee9441c6242640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/562249c2e379df9b667fa66729ee9441c6242640"}]},"branch":"refs/heads/master"},"572b52ebbb331fde5cef9f02bfc80e40bbe50421":{"kind":"REWORK","_number":15,"created":"2022-04-01 19:27:05.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/15"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-04-01 19:27:04.000000000","tz":120},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/572b52ebbb331fde5cef9f02bfc80e40bbe50421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/572b52ebbb331fde5cef9f02bfc80e40bbe50421"}]},"branch":"refs/heads/master"},"e73a1e51a606b80240141a408e2292b6c79b343f":{"kind":"REWORK","_number":16,"created":"2022-04-04 17:14:46.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/16"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-04-04 17:14:34.000000000","tz":120},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e73a1e51a606b80240141a408e2292b6c79b343f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e73a1e51a606b80240141a408e2292b6c79b343f"}]},"branch":"refs/heads/master"},"467bf76dfc3b5feff12feb2908311d5e862a97ce":{"kind":"REWORK","_number":17,"created":"2022-04-04 19:58:56.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/17"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-04-04 19:48:27.000000000","tz":120},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/467bf76dfc3b5feff12feb2908311d5e862a97ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/467bf76dfc3b5feff12feb2908311d5e862a97ce"}]},"branch":"refs/heads/master"},"dab6d45c6b0a8a0f5487808635d49dd7c680399b":{"kind":"REWORK","_number":18,"created":"2022-04-05 09:04:46.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/18"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-04-05 09:04:41.000000000","tz":120},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dab6d45c6b0a8a0f5487808635d49dd7c680399b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dab6d45c6b0a8a0f5487808635d49dd7c680399b"}]},"branch":"refs/heads/master"},"e9dc1b807e45d1dae43fc2778597579a19b96437":{"kind":"REWORK","_number":19,"created":"2022-04-07 13:52:48.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/20/834220/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/19"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-07 13:52:47.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e9dc1b807e45d1dae43fc2778597579a19b96437"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e9dc1b807e45d1dae43fc2778597579a19b96437"}]},"branch":"refs/heads/master"},"8612e65f158bb39e2c536b6ef92cf3ceb0791f97":{"kind":"REWORK","_number":20,"created":"2022-04-07 17:58:27.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/20/834220/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/20"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-07 17:58:27.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8612e65f158bb39e2c536b6ef92cf3ceb0791f97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8612e65f158bb39e2c536b6ef92cf3ceb0791f97"}]},"branch":"refs/heads/master"},"4ad72f708db740ea26890bfe9bd6412d067d9986":{"kind":"REWORK","_number":21,"created":"2022-04-14 13:11:03.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/21"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-04-14 13:11:01.000000000","tz":120},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4ad72f708db740ea26890bfe9bd6412d067d9986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4ad72f708db740ea26890bfe9bd6412d067d9986"}]},"branch":"refs/heads/master"},"6d42cfa090c537b9950b533c417631867e4e8d07":{"kind":"REWORK","_number":22,"created":"2022-06-02 06:58:22.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/22"}}},"commit":{"parents":[{"commit":"a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab","subject":"Define queues at project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-06-02 06:58:19.000000000","tz":120},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6d42cfa090c537b9950b533c417631867e4e8d07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6d42cfa090c537b9950b533c417631867e4e8d07"}]},"branch":"refs/heads/master"},"99578724f49154e4a3800eb2c86612469596beee":{"kind":"REWORK","_number":23,"created":"2022-06-13 11:54:13.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/20/834220/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/23"}}},"commit":{"parents":[{"commit":"cc8a651faeac27731a981eb54765da72a162d93a","subject":"Split generic job by tests and reduce concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc8a651faeac27731a981eb54765da72a162d93a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-06-13 11:54:11.000000000","tz":120},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/99578724f49154e4a3800eb2c86612469596beee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/99578724f49154e4a3800eb2c86612469596beee"}]},"branch":"refs/heads/master"},"81d8c18354804f087b2e7276ec8e4af10cd79218":{"kind":"REWORK","_number":24,"created":"2022-06-13 12:10:32.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/20/834220/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/24"}}},"commit":{"parents":[{"commit":"cc8a651faeac27731a981eb54765da72a162d93a","subject":"Split generic job by tests and reduce concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc8a651faeac27731a981eb54765da72a162d93a"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-06-13 12:10:32.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/826484\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/81d8c18354804f087b2e7276ec8e4af10cd79218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/81d8c18354804f087b2e7276ec8e4af10cd79218"}]},"branch":"refs/heads/master"},"9970841a38362b1a14f96deaa21cd7d74bc4996d":{"kind":"REWORK","_number":25,"created":"2022-07-12 11:11:41.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/25"}}},"commit":{"parents":[{"commit":"684ab8a44a1209366841fffd1bad973bb2365797","subject":"Merge \"Remove nova-network security groups methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/684ab8a44a1209366841fffd1bad973bb2365797"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-07-12 11:12:47.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9970841a38362b1a14f96deaa21cd7d74bc4996d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9970841a38362b1a14f96deaa21cd7d74bc4996d"}]},"branch":"refs/heads/master"},"70491e97ba832ab995f957d3fe08f11f092d0048":{"kind":"REWORK","_number":26,"created":"2022-07-12 15:42:57.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/26"}}},"commit":{"parents":[{"commit":"684ab8a44a1209366841fffd1bad973bb2365797","subject":"Merge \"Remove nova-network security groups methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/684ab8a44a1209366841fffd1bad973bb2365797"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-07-12 15:42:56.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70491e97ba832ab995f957d3fe08f11f092d0048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70491e97ba832ab995f957d3fe08f11f092d0048"}]},"branch":"refs/heads/master"},"e481a180f0c780b172c603253e43eff0ff4b7ad5":{"kind":"REWORK","_number":27,"created":"2022-07-12 16:37:48.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/27"}}},"commit":{"parents":[{"commit":"684ab8a44a1209366841fffd1bad973bb2365797","subject":"Merge \"Remove nova-network security groups methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/684ab8a44a1209366841fffd1bad973bb2365797"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-07-12 16:37:48.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e481a180f0c780b172c603253e43eff0ff4b7ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e481a180f0c780b172c603253e43eff0ff4b7ad5"}]},"branch":"refs/heads/master"},"b49d4b8192a015cadb0af667801f1acbb5517937":{"kind":"NO_CODE_CHANGE","_number":28,"created":"2022-07-29 14:04:24.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/28"}}},"commit":{"parents":[{"commit":"684ab8a44a1209366841fffd1bad973bb2365797","subject":"Merge \"Remove nova-network security groups methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/684ab8a44a1209366841fffd1bad973bb2365797"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-07-29 14:04:24.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/851521\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b49d4b8192a015cadb0af667801f1acbb5517937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b49d4b8192a015cadb0af667801f1acbb5517937"}]},"branch":"refs/heads/master","description":"Edit commit message"},"108784afc71c01f354326bd7b6600e1d9958a595":{"kind":"NO_CHANGE","_number":29,"created":"2022-08-01 18:47:50.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/29"}}},"commit":{"parents":[{"commit":"684ab8a44a1209366841fffd1bad973bb2365797","subject":"Merge \"Remove nova-network security groups methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/684ab8a44a1209366841fffd1bad973bb2365797"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-01 18:48:01.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/851521\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/108784afc71c01f354326bd7b6600e1d9958a595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/108784afc71c01f354326bd7b6600e1d9958a595"}]},"branch":"refs/heads/master"},"c0572f9331170ddb15385d41f3adbfb323d22e53":{"kind":"REWORK","_number":30,"created":"2022-08-02 07:51:38.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/30"}}},"commit":{"parents":[{"commit":"386c04bc6d95c94c8943fd9a087d7c9c63ed679f","subject":"Merge \"Update tempest version required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/386c04bc6d95c94c8943fd9a087d7c9c63ed679f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-02 07:51:46.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/851521\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0572f9331170ddb15385d41f3adbfb323d22e53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0572f9331170ddb15385d41f3adbfb323d22e53"}]},"branch":"refs/heads/master"},"5d0954e2826b5d9bddaaca32a2d880b172888e3d":{"kind":"REWORK","_number":31,"created":"2022-08-02 09:10:19.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/31"}}},"commit":{"parents":[{"commit":"386c04bc6d95c94c8943fd9a087d7c9c63ed679f","subject":"Merge \"Update tempest version required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/386c04bc6d95c94c8943fd9a087d7c9c63ed679f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-02 09:10:27.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/851521\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5d0954e2826b5d9bddaaca32a2d880b172888e3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5d0954e2826b5d9bddaaca32a2d880b172888e3d"}]},"branch":"refs/heads/master"},"842d408f25754242a9877956bba3b9689f3f54df":{"kind":"REWORK","_number":32,"created":"2022-08-02 16:00:55.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/32"}}},"commit":{"parents":[{"commit":"386c04bc6d95c94c8943fd9a087d7c9c63ed679f","subject":"Merge \"Update tempest version required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/386c04bc6d95c94c8943fd9a087d7c9c63ed679f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-02 16:01:04.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/851521\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/842d408f25754242a9877956bba3b9689f3f54df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/842d408f25754242a9877956bba3b9689f3f54df"}]},"branch":"refs/heads/master"},"e0b35e871c342c9591b63085e08975a0e1a8416b":{"kind":"REWORK","_number":33,"created":"2022-08-10 13:31:06.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/33"}}},"commit":{"parents":[{"commit":"386c04bc6d95c94c8943fd9a087d7c9c63ed679f","subject":"Merge \"Update tempest version required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/386c04bc6d95c94c8943fd9a087d7c9c63ed679f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-10 13:31:05.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/851521\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e0b35e871c342c9591b63085e08975a0e1a8416b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e0b35e871c342c9591b63085e08975a0e1a8416b"}]},"branch":"refs/heads/master"},"a5dfa16f59c3bc8571f69ea236dd794803233cb3":{"kind":"REWORK","_number":34,"created":"2022-08-10 16:30:52.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/34"}}},"commit":{"parents":[{"commit":"386c04bc6d95c94c8943fd9a087d7c9c63ed679f","subject":"Merge \"Update tempest version required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/386c04bc6d95c94c8943fd9a087d7c9c63ed679f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-10 16:30:52.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/851521\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5dfa16f59c3bc8571f69ea236dd794803233cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a5dfa16f59c3bc8571f69ea236dd794803233cb3"}]},"branch":"refs/heads/master"},"8af9a4788f81675287e36644f8055471bbf6b2c5":{"kind":"REWORK","_number":35,"created":"2022-08-11 10:17:12.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/35"}}},"commit":{"parents":[{"commit":"386c04bc6d95c94c8943fd9a087d7c9c63ed679f","subject":"Merge \"Update tempest version required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/386c04bc6d95c94c8943fd9a087d7c9c63ed679f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-11 10:17:08.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/851521\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8af9a4788f81675287e36644f8055471bbf6b2c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8af9a4788f81675287e36644f8055471bbf6b2c5"}]},"branch":"refs/heads/master"},"64022e9340d46bad595f05fca9eed3c63248b0f1":{"kind":"REWORK","_number":36,"created":"2022-08-11 10:18:02.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/36"}}},"commit":{"parents":[{"commit":"386c04bc6d95c94c8943fd9a087d7c9c63ed679f","subject":"Merge \"Update tempest version required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/386c04bc6d95c94c8943fd9a087d7c9c63ed679f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-11 10:18:00.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/851521\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/64022e9340d46bad595f05fca9eed3c63248b0f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/64022e9340d46bad595f05fca9eed3c63248b0f1"}]},"branch":"refs/heads/master"},"0a0b70c96d14e3ea63387fcd9d827a0a3db9abcd":{"kind":"NO_CODE_CHANGE","_number":37,"created":"2022-08-11 16:18:06.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/37"}}},"commit":{"parents":[{"commit":"386c04bc6d95c94c8943fd9a087d7c9c63ed679f","subject":"Merge \"Update tempest version required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/386c04bc6d95c94c8943fd9a087d7c9c63ed679f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-11 16:18:06.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/848987\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a0b70c96d14e3ea63387fcd9d827a0a3db9abcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a0b70c96d14e3ea63387fcd9d827a0a3db9abcd"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a1ca0b285f2e26f215b53aabf0821281917b6133":{"kind":"REWORK","_number":38,"created":"2022-08-11 16:28:41.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/38"}}},"commit":{"parents":[{"commit":"386c04bc6d95c94c8943fd9a087d7c9c63ed679f","subject":"Merge \"Update tempest version required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/386c04bc6d95c94c8943fd9a087d7c9c63ed679f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-11 16:28:40.000000000","tz":0},"subject":"[WIP][CI] Add CephFS Native with cephadm job","message":"[WIP][CI] Add CephFS Native with cephadm job\n\nAdds a job for CephFS Native with a cephadm\ndeployed Ceph cluster\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/848987\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a1ca0b285f2e26f215b53aabf0821281917b6133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a1ca0b285f2e26f215b53aabf0821281917b6133"}]},"branch":"refs/heads/master"},"dbe98a51b1dafa8a0c56b7cb53d00b5225fda279":{"kind":"REWORK","_number":39,"created":"2022-08-16 14:23:51.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/39"}}},"commit":{"parents":[{"commit":"386c04bc6d95c94c8943fd9a087d7c9c63ed679f","subject":"Merge \"Update tempest version required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/386c04bc6d95c94c8943fd9a087d7c9c63ed679f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-16 14:18:20.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/848987\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dbe98a51b1dafa8a0c56b7cb53d00b5225fda279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dbe98a51b1dafa8a0c56b7cb53d00b5225fda279"}]},"branch":"refs/heads/master"},"9e6aa6b9d5065685940d104884fc538b58caa2f9":{"kind":"REWORK","_number":40,"created":"2022-08-16 14:25:57.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/40"}}},"commit":{"parents":[{"commit":"386c04bc6d95c94c8943fd9a087d7c9c63ed679f","subject":"Merge \"Update tempest version required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/386c04bc6d95c94c8943fd9a087d7c9c63ed679f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-16 14:26:09.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/848987\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9e6aa6b9d5065685940d104884fc538b58caa2f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9e6aa6b9d5065685940d104884fc538b58caa2f9"}]},"branch":"refs/heads/master"},"5b1e34f24c0acd099cfdefa6a2375e956b01bbe8":{"kind":"REWORK","_number":41,"created":"2022-08-19 11:40:04.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/41"}}},"commit":{"parents":[{"commit":"386c04bc6d95c94c8943fd9a087d7c9c63ed679f","subject":"Merge \"Update tempest version required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/386c04bc6d95c94c8943fd9a087d7c9c63ed679f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-19 11:40:20.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/848987\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5b1e34f24c0acd099cfdefa6a2375e956b01bbe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5b1e34f24c0acd099cfdefa6a2375e956b01bbe8"}]},"branch":"refs/heads/master"},"d69bb0599d9e381767d9cade0a90a1b720668024":{"kind":"REWORK","_number":42,"created":"2022-08-24 16:43:28.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/42"}}},"commit":{"parents":[{"commit":"386c04bc6d95c94c8943fd9a087d7c9c63ed679f","subject":"Merge \"Update tempest version required\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/386c04bc6d95c94c8943fd9a087d7c9c63ed679f"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-24 16:43:47.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/848987\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/854436\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d69bb0599d9e381767d9cade0a90a1b720668024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d69bb0599d9e381767d9cade0a90a1b720668024"}]},"branch":"refs/heads/master"},"fc5e67bf5e2b4ecc0935239bc7eb5a59f4e7aa5f":{"kind":"REWORK","_number":43,"created":"2022-08-31 10:55:56.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/43"}}},"commit":{"parents":[{"commit":"f67893ccccb59151b69719bf44c3a09aae541b11","subject":"Use OVS with IPv6 data path testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f67893ccccb59151b69719bf44c3a09aae541b11"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-31 10:56:47.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/848987\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/854436\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc5e67bf5e2b4ecc0935239bc7eb5a59f4e7aa5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc5e67bf5e2b4ecc0935239bc7eb5a59f4e7aa5f"}]},"branch":"refs/heads/master"},"29b2462805572b6219479039555272bdbd0516fa":{"kind":"REWORK","_number":44,"created":"2022-08-31 13:56:42.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/44"}}},"commit":{"parents":[{"commit":"f67893ccccb59151b69719bf44c3a09aae541b11","subject":"Use OVS with IPv6 data path testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f67893ccccb59151b69719bf44c3a09aae541b11"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-08-31 13:57:34.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/848987\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/854436\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/29b2462805572b6219479039555272bdbd0516fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/29b2462805572b6219479039555272bdbd0516fa"}]},"branch":"refs/heads/master"},"6debab7d2db40b5671da46630931fabffa54870e":{"kind":"REWORK","_number":45,"created":"2022-09-01 14:40:32.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/45"}}},"commit":{"parents":[{"commit":"f67893ccccb59151b69719bf44c3a09aae541b11","subject":"Use OVS with IPv6 data path testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f67893ccccb59151b69719bf44c3a09aae541b11"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-09-01 14:40:32.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/848987\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/854436\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6debab7d2db40b5671da46630931fabffa54870e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6debab7d2db40b5671da46630931fabffa54870e"}]},"branch":"refs/heads/master"},"e9d18e5a152d7c605e1fe6338de112d20d3b7c7c":{"kind":"REWORK","_number":46,"created":"2022-09-05 11:03:40.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/46"}}},"commit":{"parents":[{"commit":"f67893ccccb59151b69719bf44c3a09aae541b11","subject":"Use OVS with IPv6 data path testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f67893ccccb59151b69719bf44c3a09aae541b11"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-09-05 11:03:40.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/848987\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/854436\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e9d18e5a152d7c605e1fe6338de112d20d3b7c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e9d18e5a152d7c605e1fe6338de112d20d3b7c7c"}]},"branch":"refs/heads/master"},"7fae550fc12c3b3738b69fc659d83b920720808a":{"kind":"REWORK","_number":47,"created":"2022-09-05 14:51:51.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/47"}}},"commit":{"parents":[{"commit":"f67893ccccb59151b69719bf44c3a09aae541b11","subject":"Use OVS with IPv6 data path testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f67893ccccb59151b69719bf44c3a09aae541b11"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-09-05 14:51:51.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/848987\nDepends-On: https://review.opendev.org/c/openstack/devstack-plugin-ceph/+/854436\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7fae550fc12c3b3738b69fc659d83b920720808a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7fae550fc12c3b3738b69fc659d83b920720808a"}]},"branch":"refs/heads/master"},"959b584a4bb0fd713db5abb4170ec457a18f40b9":{"kind":"NO_CODE_CHANGE","_number":48,"created":"2022-09-09 14:10:04.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/48"}}},"commit":{"parents":[{"commit":"f67893ccccb59151b69719bf44c3a09aae541b11","subject":"Use OVS with IPv6 data path testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f67893ccccb59151b69719bf44c3a09aae541b11"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-09-09 14:10:04.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/959b584a4bb0fd713db5abb4170ec457a18f40b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/959b584a4bb0fd713db5abb4170ec457a18f40b9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9fcb5fe3ebbbcb40b32395a1f790567e582206e2":{"kind":"NO_CODE_CHANGE","_number":49,"created":"2022-09-30 09:38:33.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/49"}}},"commit":{"parents":[{"commit":"f67893ccccb59151b69719bf44c3a09aae541b11","subject":"Use OVS with IPv6 data path testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f67893ccccb59151b69719bf44c3a09aae541b11"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-09-30 09:38:33.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/859950\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9fcb5fe3ebbbcb40b32395a1f790567e582206e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9fcb5fe3ebbbcb40b32395a1f790567e582206e2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"891bba6820c2ebe01d8aa728580ba3c733d01862":{"kind":"REWORK","_number":50,"created":"2022-09-30 17:06:22.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/20/834220/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/50"}}},"commit":{"parents":[{"commit":"f67893ccccb59151b69719bf44c3a09aae541b11","subject":"Use OVS with IPv6 data path testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f67893ccccb59151b69719bf44c3a09aae541b11"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-09-30 17:06:22.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/859950\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/891bba6820c2ebe01d8aa728580ba3c733d01862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/891bba6820c2ebe01d8aa728580ba3c733d01862"}]},"branch":"refs/heads/master"},"67ee09af1cc85d663cd477155c2c43d68e72c2bd":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2022-10-26 08:55:19.000000000","uploader":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"ref":"refs/changes/20/834220/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/51"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-10-26 08:55:19.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/859950\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/67ee09af1cc85d663cd477155c2c43d68e72c2bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/67ee09af1cc85d663cd477155c2c43d68e72c2bd"}]},"branch":"refs/heads/master","description":"Rebase"},"0f355bae6309128e6a363012f431676d09d09830":{"kind":"TRIVIAL_REBASE","_number":52,"created":"2023-03-01 19:23:00.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/20/834220/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/52"}}},"commit":{"parents":[{"commit":"0f4e845830812271fc232e0f9079129788e12948","subject":"Merge \"[RBAC] Generate unique arguments for access rule creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f4e845830812271fc232e0f9079129788e12948"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-03-01 19:23:00.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/859950\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f355bae6309128e6a363012f431676d09d09830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f355bae6309128e6a363012f431676d09d09830"}]},"branch":"refs/heads/master","description":"Rebase"},"bc778c6b68ffd3c0c7bd2c4778153de4afbf79bb":{"kind":"REWORK","_number":53,"created":"2023-03-01 19:25:38.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/20/834220/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/53"}}},"commit":{"parents":[{"commit":"0f4e845830812271fc232e0f9079129788e12948","subject":"Merge \"[RBAC] Generate unique arguments for access rule creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f4e845830812271fc232e0f9079129788e12948"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-03-01 19:24:38.000000000","tz":0},"subject":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs","message":"[CI] Add CephFS Native and CephFS NFS with cephadm jobs\n\nAdds jobs for CephFS Native and CephFS NFS with a cephadm\ndeployed Ceph cluster.\n\nDepends-On: https://review.opendev.org/c/openstack/manila/+/859950\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bc778c6b68ffd3c0c7bd2c4778153de4afbf79bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bc778c6b68ffd3c0c7bd2c4778153de4afbf79bb"}]},"branch":"refs/heads/master"},"4850f2cf74526211903c2a58a7242860eef3673a":{"kind":"REWORK","_number":54,"created":"2023-04-04 23:59:04.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/20/834220/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/54"}}},"commit":{"parents":[{"commit":"0f4e845830812271fc232e0f9079129788e12948","subject":"Merge \"[RBAC] Generate unique arguments for access rule creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f4e845830812271fc232e0f9079129788e12948"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-04-04 23:58:46.000000000","tz":-420},"subject":"[CI] Add CephFS Native cephadm job","message":"[CI] Add CephFS Native cephadm job\n\nAdd jobs for CephFS Native  with a cephadm deployed Ceph cluster.\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4850f2cf74526211903c2a58a7242860eef3673a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4850f2cf74526211903c2a58a7242860eef3673a"}]},"branch":"refs/heads/master"},"8acf259c2cb42287f3d3c1c86cbbc9cd0ac2c38d":{"kind":"REWORK","_number":55,"created":"2023-04-05 19:36:03.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/20/834220/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/55"}}},"commit":{"parents":[{"commit":"0f4e845830812271fc232e0f9079129788e12948","subject":"Merge \"[RBAC] Generate unique arguments for access rule creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f4e845830812271fc232e0f9079129788e12948"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-04-05 19:35:48.000000000","tz":-420},"subject":"[CI] Add cephadm based native cephfs test job","message":"[CI] Add cephadm based native cephfs test job\n\nAdd a new job for testing manila\u0027s Native CephFS driver\nwith a cephadm deployed Ceph cluster. This job inherits\nfrom the existing \"base\" job and overrides the\nceph deployment strategy to use cephadm.\n\nWith this commit, we will use this new cephadm based job\ninstead of the package based deployment to test this\nrepository.\n\nOnce all CI consumers adopt the new cephadm job, we\u0027ll\nrewrite the \"base\" job definition to only care about\ncephadm based deployment.\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8acf259c2cb42287f3d3c1c86cbbc9cd0ac2c38d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8acf259c2cb42287f3d3c1c86cbbc9cd0ac2c38d"}]},"branch":"refs/heads/master"},"60626d51e9a46b0c16cbb7124038053e4ce9219b":{"kind":"NO_CODE_CHANGE","_number":56,"created":"2023-04-06 15:13:48.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/20/834220/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/20/834220/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/20/834220/56"}}},"commit":{"parents":[{"commit":"0f4e845830812271fc232e0f9079129788e12948","subject":"Merge \"[RBAC] Generate unique arguments for access rule creation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f4e845830812271fc232e0f9079129788e12948"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-03-17 19:24:28.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-04-06 15:13:48.000000000","tz":0},"subject":"[CI] Add cephadm based native cephfs test job","message":"[CI] Add cephadm based native cephfs test job\n\nAdd a new job for testing manila\u0027s Native CephFS driver\nwith a cephadm deployed Ceph cluster. This job inherits\nfrom the existing \"base\" job and overrides the\nceph deployment strategy to use cephadm.\n\nWith this commit, we will use this new cephadm based job\ninstead of the package based deployment to test this\nrepository. As with the existing job, this new job is\nnon-voting as well. If the job\u0027s stable for a period of\ntime, we can decide to promote it to being \"voting\".\n\nOnce all CI consumers adopt the new cephadm job, we\u0027ll\nrewrite the \"base\" job definition to only care about\ncephadm based deployment.\n\nChange-Id: I6e4db6b89f89d2374580827e2334635dc5ead5b9\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/60626d51e9a46b0c16cbb7124038053e4ce9219b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/60626d51e9a46b0c16cbb7124038053e4ce9219b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
