)]}'
{"id":"openstack%2Fmanila-tempest-plugin~834426","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I519fd4de0b9f3ff20c355894c3e797d6c19e1835","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-04-13 19:01:40.000000000","reason":"Goutham Pacha Ravi replied on the change"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-05-04 12:05:32.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-05-04 12:05:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I519fd4de0b9f3ff20c355894c3e797d6c19e1835","subject":"Remove ping_ip_address \u0026 _log_console_output","status":"MERGED","created":"2022-03-21 08:14:05.000000000","updated":"2022-05-04 12:05:32.000000000","submitted":"2022-05-04 12:05:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"834426-tempest-scenario-manager-cleanup","meta_rev_id":"9bd2c3ee7b42880fb419a2508b5767d57973f0ca","_number":834426,"virtual_id_number":834426,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-04 12:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-04-13 19:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-04-13 08:00:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2022-05-04 10:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-05-04 10:45:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-21 10:50:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-21 11:53:39.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-03-21 12:53:47.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-22 19:52:07.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 13:04:24.000000000","updated_by":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-04-13 08:00:47.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-05-04 10:26:11.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"ac0020cdf15b8654d5214130ed73f22e026599fb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-21 08:14:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c951d355f97f554f9dcdab447ec3b6138a5380ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 10:50:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b88877696def472ab3372aec0f2a10a3 : SUCCESS in 5m 49s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ca67aa66259448e9892ca58def1202b8 : SUCCESS in 35m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ec742719c4204baea67015ebcf15608b : SUCCESS in 40m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1e1120e1bb084fffb597f0cbd0b635f1 : SUCCESS in 1h 24m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/380ccbcc6b40401daf01447a3054c84a : SUCCESS in 1h 09m 50s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d4db9406aab3469691dd82298c15b497 : SUCCESS in 59m 50s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/ae8ba3fe29cd4a4eafba615beaa21e8e : SUCCESS in 1h 09m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/917c15e147ef4ef7aabfa8f45f1df358 : SUCCESS in 1h 12m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/159cc00ae20e467a873b5731bfcb3bb3 : SUCCESS in 1h 22m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c5206145647b4cbeb7f8803fc4395f6f : FAILURE in 1h 35m 51s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/20e62933792042c9a7ac33d44b8a9a05 : SUCCESS in 59m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9351f48d720f426896cc6f0494c9f0d6 : FAILURE in 2h 30m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0c9c254df27d41ad97ce487954056ca3 : SUCCESS in 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26ba8f8bbac6e18ffb3731eb54e01bb35ab7b15c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-21 11:53:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/26/834426/1/check/DellEMC_Unity_Manila_dhss_true/d952e2c/DellEMC_Unity_Manila_dhss_true/300 : FAILURE in 3h 39m 15s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/26/834426/1/check/DellEMC_Unity_Manila_dhss_false/bb95a1f/DellEMC_Unity_Manila_dhss_false/306 : SUCCESS in 1h 17m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"9396af68223216070d515d5df22aa6c20c6f6e91","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-21 12:53:47.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/26/834426/1/check/DellEMC_VNX_Manila/2d980d4/DellEMC_VNX_Manila/168 : FAILURE in 2h 03m 28s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"95d99cebb5e448a3dcaa43997b0f206d69ceb4ea","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-22 15:28:34.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"39f9399b480b42f07587d185b05cb3b9a84a6fc3","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-22 15:50:13.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"b0042374d45fe9a5bf9d89d68fbca3a7b6776795","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-22 15:52:55.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6da0e885e23f0c935cf7a13093bfe420fda9c087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 19:05:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ce0fe847b874aeeb1c33df2f8cffe8b : SUCCESS in 5m 23s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/04b543fba8124bc0b8b3734e145e6e33 : SUCCESS in 33m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2a3eef3161404685acab740dc13b33ae : SUCCESS in 38m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8ba185cb204149cba461a403c23f7fdb : SUCCESS in 1h 10m 01s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/b95c998e3d45488a8e724c9c7eaea015 : SUCCESS in 56m 43s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/eef37562a7a9483ebcf71ed5116da684 : SUCCESS in 1h 06m 19s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/6004f990eec34af696fcbc3ae6d79b86 : SUCCESS in 1h 05m 29s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/34f5b833192d433c8820974bb3c6e09a : SUCCESS in 1h 16m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3cb8fcf281554099bc4efce286da6376 : SUCCESS in 1h 29m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7f70d30c97d849a6923fc8c99b99b389 : FAILURE in 1h 37m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/171556949f84404caba7eee065b8caab : SUCCESS in 53m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/48e88e91bf5847608788a3ac69c2dceb : TIMED_OUT in 3h 05m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7ff70bfbef434a0e9bf4575e2ecb1d33 : SUCCESS in 25m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bc8bfe55ee3264b148c22b31448ec0120044f006","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-22 19:52:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/834426/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/955ef21/ : FAILURE in 3m 14s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/26/834426/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f1a8a9c/ : FAILURE in 7m 34s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"1e6e8a9a45176e64dee5292b24bc6deda2f65d54","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-22 23:06:40.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/26/834426/2/check/DellEMC_Unity_Manila_dhss_true/17dc94d/DellEMC_Unity_Manila_dhss_true/311 : FAILURE in 3h 27m 38s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/26/834426/2/check/DellEMC_Unity_Manila_dhss_false/5ce9eed/DellEMC_Unity_Manila_dhss_false/319 : SUCCESS in 1h 16m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b46c06ff50fa7d31e7dde8d33365c111240cec8f","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-23 07:46:20.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/26/834426/2/check/DellEMC_VNX_Manila/864ab27/DellEMC_VNX_Manila/181 : FAILURE in 2h 28m 24s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"ea5e3acdad1347be09e4d3727a97975e2db7f0e5","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-24 10:22:22.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"ce172dd1270e1cbdc359467c8eae61d5845ac074","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-13 08:00:47.000000000","message":"Patch Set 2: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0f40de19a704cc09441854cad765862293a7578b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-04-13 19:01:40.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"38f4844a8d9fc5998360c15820fa71424dd5e537","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-05-04 10:26:11.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b6bbe18ffb54e2049b3d7e79e851b951d3b03a3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-04 10:45:03.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99a422821560ce3906e36ec659cb712c1438bf09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 10:46:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a6e901312151382d45cca4f8f9acdc7d4997f515","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 12:05:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44b19eea68f446dbb37e04c87be50c78 : SUCCESS in 3m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c169a25dfbe14bad8c043ae0de3e8afc : SUCCESS in 26m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4435ff6be30c46d6a55160a43f9214dd : SUCCESS in 34m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8c29828e2f7646b78747f127fe976e47 : SUCCESS in 1h 18m 16s","accounts_in_message":[],"_revision_number":2},{"id":"9bd2c3ee7b42880fb419a2508b5767d57973f0ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 12:05:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"164898cd76d84a776198952e18255d010f256285","revisions":{"a88b0efc76c16b85e299bd7957d14bc4e417ca5f":{"kind":"REWORK","_number":1,"created":"2022-03-21 08:14:05.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/26/834426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/26/834426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/834426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/834426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/834426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/834426/1"}}},"commit":{"parents":[{"commit":"a02c455af93838e4247be7348630bc83b138cf6c","subject":"Remove create_server and rebuild_server methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a02c455af93838e4247be7348630bc83b138cf6c"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 08:12:38.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 08:12:38.000000000","tz":60},"subject":"Remove ping_ip_address \u0026 _log_console_output","message":"Remove ping_ip_address \u0026 _log_console_output\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* ping_ip_address\n\t* _log_console_output\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I519fd4de0b9f3ff20c355894c3e797d6c19e1835\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a88b0efc76c16b85e299bd7957d14bc4e417ca5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a88b0efc76c16b85e299bd7957d14bc4e417ca5f"}]},"branch":"refs/heads/master"},"164898cd76d84a776198952e18255d010f256285":{"kind":"NO_CHANGE","_number":2,"created":"2022-03-22 15:52:55.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/26/834426/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/26/834426/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/834426/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/834426/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/834426/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/26/834426/2"}}},"commit":{"parents":[{"commit":"a02c455af93838e4247be7348630bc83b138cf6c","subject":"Remove create_server and rebuild_server methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a02c455af93838e4247be7348630bc83b138cf6c"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 08:12:38.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-22 15:52:36.000000000","tz":60},"subject":"Remove ping_ip_address \u0026 _log_console_output","message":"Remove ping_ip_address \u0026 _log_console_output\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* ping_ip_address\n\t* _log_console_output\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I519fd4de0b9f3ff20c355894c3e797d6c19e1835\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/164898cd76d84a776198952e18255d010f256285"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/164898cd76d84a776198952e18255d010f256285"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
