)]}'
{"id":"openstack%2Fmanila-tempest-plugin~834427","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I083d713c638de50b1549b4c10b9eb38f58b85fab","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-05-07 07:42:13.000000000","reason":"Goutham Pacha Ravi replied on the change"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-05-07 08:56:34.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-04-13 08:11:12.000000000","reason":"Martin Kopec replied on the change"}},"hashtags":[],"change_id":"I083d713c638de50b1549b4c10b9eb38f58b85fab","subject":"Remove _disassociate_floating_ip \u0026 _create_network","status":"MERGED","created":"2022-03-21 08:20:48.000000000","updated":"2022-05-07 08:56:34.000000000","submitted":"2022-05-07 08:56:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834427-tempest-scenario-manager-cleanup","meta_rev_id":"07e9b6330b294957d63bb3c3780657802edb1757","_number":834427,"virtual_id_number":834427,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-07 08:56:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-05-07 07:42:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-04-13 08:11:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2022-05-04 23:15:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-05-07 07:42:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-21 09:43:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-21 15:25:21.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-03-22 11:09:16.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-22 18:07:41.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 13:04:36.000000000","updated_by":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-03-29 18:49:28.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-05-04 23:15:52.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"62da4b33c87e064b98c78ba3a1f422b66bdf9847","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-21 08:20:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c0920d097b2cea73bcecdeded201165a789b7ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 09:43:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/168f82ca3a404a409af771dd38545bbc : SUCCESS in 5m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/980f7ddb1c224fe19ec8c435f21f1ef9 : SUCCESS in 31m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ea8ddf90a7194fb9b6b037bccb328f0e : SUCCESS in 34m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/30d7d8a482e745ef8661f8b4dfc2edef : FAILURE in 46m 50s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/6696dd7cebf340d19f265f4996597ed4 : FAILURE in 38m 44s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b952ac75ccdb4eb3b04378d8395a31b1 : FAILURE in 1h 01m 56s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/636f912b275743ed83a649b7662f274a : FAILURE in 1h 02m 47s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/90ce2ee499834b91b0d0e89245b6de4d : FAILURE in 46m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/852240b467e94f79a62d25d26e7a99d9 : FAILURE in 49m 23s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a6103769b9a8477683f0e30d2286f0d3 : FAILURE in 1h 20m 05s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0a616d3aede843d3931a3fe0a55ff720 : SUCCESS in 1h 00m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0b09ecba960c4bea823652b17bccbd05 : FAILURE in 1h 13m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/833de68f5f2e40f49bcd38c02c6292e1 : SUCCESS in 25m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9390e03dc036fe6f0f11b8d9d40b6283a6f10136","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-21 15:25:21.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/27/834427/1/check/DellEMC_Unity_Manila_dhss_true/3a836b1/DellEMC_Unity_Manila_dhss_true/301 : FAILURE in 3h 31m 43s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/27/834427/1/check/DellEMC_Unity_Manila_dhss_false/98a24a3/DellEMC_Unity_Manila_dhss_false/307 : FAILURE in 1h 38m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"bf6b433fa221a5df3a2d0dcb064255a00dea77a0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-22 06:59:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"114699f16441f66c563e2938944149ea4d24dccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 08:54:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5de171c34a7b4f77a27ddc9f9c9fe09f : SUCCESS in 4m 37s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7aa68f31f1154b258f88debaaed47de9 : SUCCESS in 39m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/61897a168fae413382d6c980800acf33 : SUCCESS in 41m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/743fdf965c8b4e91b3953ccad97071ce : SUCCESS in 1h 16m 03s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3a5694be02f044bebd84310fa094d6bb : SUCCESS in 1h 08m 24s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b07f8bd1447e45808f785a61e9141b6c : SUCCESS in 1h 18m 41s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/b302d62f88b04adb92b5322b2f80b3aa : SUCCESS in 1h 13m 07s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ca6b671319054da791378caa1d6ff334 : SUCCESS in 1h 18m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/81adce2d102849d595d06d23c1ce78db : SUCCESS in 1h 39m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6eceaa8fc0d943c89f882a4adf4306ab : FAILURE in 1h 48m 55s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ea6146e5104743b1bc8b8c27961bd69b : SUCCESS in 52m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/64809092b0cd46d98a37df41f6cdd201 : SUCCESS in 1h 44m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/dd0e3577e27a47be902c9e83f083c3c5 : SUCCESS in 28m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b626a58ada143836bfdf63ed83e8f94d811a5d4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-22 10:38:55.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/27/834427/2/check/DellEMC_Unity_Manila_dhss_true/f850a7c/DellEMC_Unity_Manila_dhss_true/306 : FAILURE in 3h 38m 35s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/27/834427/2/check/DellEMC_Unity_Manila_dhss_false/75cf763/DellEMC_Unity_Manila_dhss_false/313 : SUCCESS in 1h 20m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"eddf884c547b3481b3a308ed191823558e5d1486","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-22 11:09:16.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/27/834427/2/check/DellEMC_VNX_Manila/d8d9e02/DellEMC_VNX_Manila/172 : FAILURE in 2h 14m 55s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"ccf7c926948b0af83e2fcd501eaa128d721ba603","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-22 15:53:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b5849f04a8764cfea07ed2a4fb288475db5ae54b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 18:00:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/836909e99e8c4d268a324d52aac5a75d : SUCCESS in 5m 27s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b6303433db214654b2a680af3fe66639 : SUCCESS in 36m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8d260a80b9564d6fba2d729a9a520440 : SUCCESS in 37m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/798dc8fc610149968c07d6a6aabe15d8 : SUCCESS in 1h 17m 20s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/cc52e78c4a7e45d0ae58fde775e490cc : SUCCESS in 1h 23m 43s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/58879fc4ff9b44bcb3fd91e0a1060592 : SUCCESS in 1h 07m 21s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/00ad2ed914c64bbcb5b306cb3f260eb8 : SUCCESS in 1h 07m 44s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1e6e34c67b66404abbf8db53700a8da2 : FAILURE in 1h 37m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/dbc38942144e457e98e96ff0c47d729b : SUCCESS in 1h 21m 00s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/683bc053e10340a2910c844d97858956 : FAILURE in 2h 02m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d11f2067d30141e7a962155bdcc6c3c5 : SUCCESS in 1h 04m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8f0b82c79d884bfc9803416342260489 : FAILURE in 1h 19m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bfc573cd88e24e10aab1855b028a9ef2 : SUCCESS in 23m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fba60197a2e3745007f058317a6e0af5bb442053","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-22 18:07:41.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/27/834427/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b5007ea/ : FAILURE in 3m 13s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/27/834427/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/77c5617/ : FAILURE in 6m 28s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"7d802da82e7217d9b23a3fb60e980965aaae0f37","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-22 22:28:34.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/27/834427/3/check/DellEMC_VNX_Manila/bd5cc54/DellEMC_VNX_Manila/177 : FAILURE in 2h 15m 02s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"453d8f3a8661d8df0e7c618448d74a1ac4f56202","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-23 03:31:42.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/27/834427/3/check/DellEMC_Unity_Manila_dhss_true/c76a1ae/DellEMC_Unity_Manila_dhss_true/312 : FAILURE in 4h 24m 57s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/27/834427/3/check/DellEMC_Unity_Manila_dhss_false/271ec2a/DellEMC_Unity_Manila_dhss_false/320 : SUCCESS in 1h 39m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"4f95dcf341e510bbb5762eb39e2206de130101d8","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-24 10:22:31.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"23ae7b261936a9fdac82d4a74ea3d2f820a66f66","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-03-29 18:49:28.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3ba24149eaf2ecfe53b1a7f29c1e7d7b1a23a818","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:20:12.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a8dcdb3429907fd6457b93d6ab7cce2220c82532","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:22:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"48091ae979bb72033303ad9128c0e26e1c9b9cef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 11:05:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/101216daae864755bbc851e42efc9be6 : SUCCESS in 3m 21s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d7d9a2615d6648adbb00340042b73af5 : SUCCESS in 30m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/07d2dd97e8574fe0b017688bf2ed613f : SUCCESS in 27m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/350a1c499992415fb7fd5b2fe78840fa : SUCCESS in 1h 11m 33s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/db58a1958fd94e70900cfde02b810a32 : SUCCESS in 56m 01s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/66f080416a0c4b4ea0e859b0a4c0964a : SUCCESS in 1h 01m 04s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/f16139b0663c451b94bdbdcba486d506 : SUCCESS in 1h 06m 23s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/054c6a4e60c442ef9a3a41275bbb2f22 : SUCCESS in 1h 37m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ad1d61b9b8c5409c96b0d28de64e4d1f : SUCCESS in 1h 29m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/239f4904550148239bf49cbcd24ea11d : FAILURE in 1h 21m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/44a8e8b4af664dc8acbb6c27ab06bd01 : SUCCESS in 1h 03m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/358cce54ba844af3aa78e950f15046a9 : FAILURE in 1h 38m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/79e47fff915c4c6e9bd86d8870f6cbd7 : SUCCESS in 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"14e1e2f480abd8a7e13a80aba3a3058741d339ab","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-01 11:09:08.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/27/834427/4/upstream-check/manila-cDOT-manila-tempest-plugin-ss/fbb7d6e/ : FAILURE in 2m 47s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/27/834427/4/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/f8f0d67/ : FAILURE in 2m 48s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"a0f2c00cb5826ee010d6bdfdf96bc0c02ae8a956","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-01 16:38:45.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/27/834427/4/check/DellEMC_Unity_Manila_dhss_true/b712a04/DellEMC_Unity_Manila_dhss_true/341 : FAILURE in 4h 42m 59s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/27/834427/4/check/DellEMC_Unity_Manila_dhss_false/81c738b/DellEMC_Unity_Manila_dhss_false/359 : FAILURE in 27m 30s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"acdfe344ae9a468bcef79df937f7718ab753b479","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-13 08:11:12.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5095559d47a77b68ba4b65d440031102327c67a2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-04 23:15:52.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"021a2637e8d9db73f967e235124ad127b935fca8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-07 07:42:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7640ce963ef82026e8b344a1c9d012575d0f32d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 07:43:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"be63b16bc2e1d677ec1557bd39b2a29b50205c8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 08:56:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2654548e66c84772b7a3cae88e148330 : SUCCESS in 3m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f5f728e5198e497799da6d82f04522c7 : SUCCESS in 32m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7f295ec374fd41a9bca229d9728caebe : SUCCESS in 29m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6b4f190f783647edb47f34bd30a0be0c : SUCCESS in 1h 12m 59s","accounts_in_message":[],"_revision_number":4},{"id":"07e9b6330b294957d63bb3c3780657802edb1757","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 08:56:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1118f3ebe4773a1119685fada4b368d0d554f986","revisions":{"1ca1f64275fafc9f3bcb8ef5f7a7bcf281141b9e":{"kind":"REWORK","_number":1,"created":"2022-03-21 08:20:48.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/27/834427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/27/834427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/1"}}},"commit":{"parents":[{"commit":"a88b0efc76c16b85e299bd7957d14bc4e417ca5f","subject":"Remove ping_ip_address \u0026 _log_console_output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a88b0efc76c16b85e299bd7957d14bc4e417ca5f"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 08:18:53.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 08:18:53.000000000","tz":60},"subject":"Remove _disassociate_floating_ip \u0026 _create_network","message":"Remove _disassociate_floating_ip \u0026 _create_network\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _disassociate_floating_ip\n\t* _create_network\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I083d713c638de50b1549b4c10b9eb38f58b85fab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ca1f64275fafc9f3bcb8ef5f7a7bcf281141b9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ca1f64275fafc9f3bcb8ef5f7a7bcf281141b9e"}]},"branch":"refs/heads/master"},"4a36559c435f07865677a00ec3da3b1abec18624":{"kind":"REWORK","_number":2,"created":"2022-03-22 06:59:56.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/27/834427/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/27/834427/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/2"}}},"commit":{"parents":[{"commit":"a88b0efc76c16b85e299bd7957d14bc4e417ca5f","subject":"Remove ping_ip_address \u0026 _log_console_output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a88b0efc76c16b85e299bd7957d14bc4e417ca5f"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 08:18:53.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-22 06:58:55.000000000","tz":60},"subject":"Remove _disassociate_floating_ip \u0026 _create_network","message":"Remove _disassociate_floating_ip \u0026 _create_network\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _disassociate_floating_ip\n\t* _create_network\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I083d713c638de50b1549b4c10b9eb38f58b85fab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a36559c435f07865677a00ec3da3b1abec18624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a36559c435f07865677a00ec3da3b1abec18624"}]},"branch":"refs/heads/master"},"f01c60eac31465af0b9425e7ede4ba540b8192f8":{"kind":"NO_CHANGE","_number":3,"created":"2022-03-22 15:53:26.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/27/834427/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/27/834427/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/3"}}},"commit":{"parents":[{"commit":"164898cd76d84a776198952e18255d010f256285","subject":"Remove ping_ip_address \u0026 _log_console_output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/164898cd76d84a776198952e18255d010f256285"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 08:18:53.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-22 15:53:26.000000000","tz":0},"subject":"Remove _disassociate_floating_ip \u0026 _create_network","message":"Remove _disassociate_floating_ip \u0026 _create_network\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _disassociate_floating_ip\n\t* _create_network\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I083d713c638de50b1549b4c10b9eb38f58b85fab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f01c60eac31465af0b9425e7ede4ba540b8192f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f01c60eac31465af0b9425e7ede4ba540b8192f8"}]},"branch":"refs/heads/master","description":"Rebase"},"1118f3ebe4773a1119685fada4b368d0d554f986":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-04-01 09:20:12.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/27/834427/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/27/834427/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/834427/4"}}},"commit":{"parents":[{"commit":"164898cd76d84a776198952e18255d010f256285","subject":"Remove ping_ip_address \u0026 _log_console_output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/164898cd76d84a776198952e18255d010f256285"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 08:18:53.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:20:12.000000000","tz":0},"subject":"Remove _disassociate_floating_ip \u0026 _create_network","message":"Remove _disassociate_floating_ip \u0026 _create_network\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _disassociate_floating_ip (not used)\n\t* _create_network\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I083d713c638de50b1549b4c10b9eb38f58b85fab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1118f3ebe4773a1119685fada4b368d0d554f986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1118f3ebe4773a1119685fada4b368d0d554f986"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
