)]}'
{"id":"openstack%2Fmanila-tempest-plugin~834431","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I0911cda7fcf461b9786f9534689083c7ac3a2035","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-05-07 07:43:10.000000000","reason":"Goutham Pacha Ravi replied on the change"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-05-10 14:53:48.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-05-10 14:53:48.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-05-10 14:53:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0911cda7fcf461b9786f9534689083c7ac3a2035","subject":"Remove get_remote_client \u0026 check_vm_connectivity","status":"MERGED","created":"2022-03-21 09:21:55.000000000","updated":"2022-05-10 14:53:48.000000000","submitted":"2022-05-10 14:53:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"834431-tempest-scenario-manager-cleanup","meta_rev_id":"0596d25802cff2a40eb9b247d2d7d4031fade920","_number":834431,"virtual_id_number":834431,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-10 14:53:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-05-07 07:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-04-13 08:14:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-04 23:16:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-05-10 13:38:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-21 11:20:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-21 19:06:35.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-03-22 13:22:03.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-22 18:20:29.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 13:04:51.000000000","updated_by":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-03-29 19:43:26.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2022-04-13 08:14:50.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-05-04 23:16:22.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-05-09 16:21:17.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"7dddf79827497bea3056daaec9f292e5507f4b67","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-21 09:21:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a8ce3e8889161d6c85f47cc8421dfc2052e35db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 11:20:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e92b2cdb60b7420c8588e1a2a923fc17 : SUCCESS in 5m 13s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/55c37dd9701844358b2bc5649f024d0e : SUCCESS in 37m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5509f902abd8428c8098f70ee45c207d : SUCCESS in 43m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b1521b9861684fa787e76ca537b4b8c8 : FAILURE in 50m 10s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/34cbbee585f2489982372b36ebfca847 : FAILURE in 47m 24s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/271c5b9af63a403f9e3754827e8f54f3 : FAILURE in 1h 07m 41s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/2246294c9219454da0f1b3b01808b848 : FAILURE in 50m 10s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6c3d314fb2e347289d46688fc72143e3 : FAILURE in 41m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c6751519710f41bdad62d5fb57775041 : FAILURE in 52m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0e9937afc932470b85b5d1d45275f9a9 : FAILURE in 1h 21m 59s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9b53ee29612943aeaf6f50c29a68269e : SUCCESS in 1h 01m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/84a76a685da649febce6bf5ce8793573 : FAILURE in 1h 56m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/592755f44a7c458fb33f150e35c74a19 : SUCCESS in 34m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cf71cedb75668ea6d41ac33bf9f9818c6f3369a2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-21 19:06:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/31/834431/1/check/DellEMC_Unity_Manila_dhss_true/8491e8a/DellEMC_Unity_Manila_dhss_true/302 : FAILURE in 3h 41m 09s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/834431/1/check/DellEMC_Unity_Manila_dhss_false/e386eeb/DellEMC_Unity_Manila_dhss_false/308 : FAILURE in 1h 53m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"15f8d247b4af5a1dbf5ef97286b9ef9c1406ea36","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-22 07:00:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"03e6460f719aebc70874084ef55891c75a776642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 09:08:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32de93abe5b541a4a240ed44b8e9d5f2 : SUCCESS in 4m 35s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ae85ed7ce5684caea500cd06f0e7a658 : SUCCESS in 32m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1081b82b9b094bd6ab086cec554c8d63 : SUCCESS in 33m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e6699a24c35145e69868a1e42a6de971 : SUCCESS in 1h 13m 38s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/016975c0a2974634b7d936adbb9197d1 : SUCCESS in 1h 05m 27s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/ff7a71bad3634800ba17f890fe065b6d : SUCCESS in 59m 26s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/55610a45a8d44d2ebd688992a41e521e : SUCCESS in 1h 01m 57s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/07a6fd6761d343dfa11115f4201a4937 : SUCCESS in 1h 15m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7c8e92d676ed41238ec15fb8b0560dfe : SUCCESS in 1h 12m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/986f82691d5e45379eb71452345e0a9a : FAILURE in 1h 54m 58s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/939417cec3b04d29a0322b06a26ec2a7 : SUCCESS in 52m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/59ed1d28f8cc49e4aa6a9c8bd90854b4 : FAILURE in 2h 06m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/83bd53ee523d4455a19e776356d9447c : SUCCESS in 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9b48e903250354d51c0dd08424b3bff0fcf47bb7","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-22 10:32:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f8d07cbb3288ce5fe1023399ba8bcf5eebcd5fc8","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-22 13:22:03.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/31/834431/2/check/DellEMC_VNX_Manila/5de88bd/DellEMC_VNX_Manila/173 : FAILURE in 2h 12m 41s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"58d7b24def19294b2f10c21d5ed6d4616f7b7504","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-22 14:26:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/31/834431/2/check/DellEMC_Unity_Manila_dhss_true/3c8f163/DellEMC_Unity_Manila_dhss_true/307 : FAILURE in 3h 47m 47s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/834431/2/check/DellEMC_Unity_Manila_dhss_false/45f4605/DellEMC_Unity_Manila_dhss_false/314 : SUCCESS in 1h 29m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"27ee3a1e965e73abcdc5e0f3b84126df1fa56d99","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-22 15:53:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"81680f3ea7535c1359fa8c1e7fe9a23cdee752fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 18:16:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f3c08e031a462ea9dc49701e2aeacd : SUCCESS in 5m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a504a90983d2481eb35c3fe50628e8f9 : SUCCESS in 36m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b4335c269b6c4d98a2862089a8fefb7c : SUCCESS in 49m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9b31920f4fc2446997d6b1ca5f198ab2 : SUCCESS in 1h 15m 44s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/ecf3d43275fc4578aaec74fa91a520b4 : SUCCESS in 1h 24m 34s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/6487bdccfddd4571ac5e286a1dbdf06a : SUCCESS in 1h 22m 50s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/dca4f60dbe1e49818eb832f7f05caab4 : SUCCESS in 1h 13m 53s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9f931293008346f0b820202e4c94ca60 : SUCCESS in 1h 10m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/83fd63b1529348f4862394971ea6c054 : SUCCESS in 1h 23m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/65a10888c0544a3cb00b36bc364efd15 : FAILURE in 1h 36m 26s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ae72b18c27514c63b70b34a3bd7bb014 : SUCCESS in 53m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4d5ab8a3402546de85112f19516916cd : FAILURE in 2h 12m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/97d4bbde61ea45e59056c201a368a7b9 : SUCCESS in 28m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8d3942e75b4a3aee499239f93c9a4469894a6dab","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-22 18:20:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/834431/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/8ecbdc3/ : FAILURE in 3m 12s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/834431/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/ff00942/ : FAILURE in 3m 05s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"0fafb591ccc1e6ba7517820ba8516ea79c489c8b","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-23 00:55:38.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/31/834431/3/check/DellEMC_VNX_Manila/c131a1e/DellEMC_VNX_Manila/178 : FAILURE in 2h 26m 58s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"f63d956baaba528677990e63a538684804f2df24","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-23 04:31:14.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/31/834431/3/check/DellEMC_Unity_Manila_dhss_true/d090097/DellEMC_Unity_Manila_dhss_true/313 : ABORTED in 59m 31s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/834431/3/check/DellEMC_Unity_Manila_dhss_false/ecf0130/DellEMC_Unity_Manila_dhss_false/321 : FAILURE in 1h 22m 55s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d67dec44c12c5a3762b2b285010c007bbd16c501","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-24 10:22:40.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"0bbd15fd17d7724630932ea597010f88cff25d49","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-03-29 19:43:26.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"429dd91aad67b4e38ce2e5a5e465e4e28ae8ccc0","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 08:00:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aafe5a3b1beb7c9598940bdfc3b4140d53417b4c","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 08:18:05.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6af0613b09dcc8bb8990c2fc2e3730acda8a7ff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:20:32.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"562afc8f8e183610a656e07bb468469ce813e268","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:23:42.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"5300bcd07a773ce13b08a5286557901c6a2e0354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 11:29:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c92944e896524c1da9576742024cb529 : SUCCESS in 3m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/533970fb27574f34ba8ff7dc4fc7db6b : SUCCESS in 26m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fb93573eecdf4b90bb3532fc455dc948 : SUCCESS in 31m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4cfaed17ef2c48d19dbbf8e001e9003a : SUCCESS in 1h 14m 11s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/129c5cd61ac543b196262a8a410c4621 : SUCCESS in 46m 10s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/f73ec7b0146649e0bc5353ce11eac9ad : SUCCESS in 54m 14s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/bc97b3af83a840f4bc38a4d0cb115159 : SUCCESS in 56m 31s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/beea807ec9104e3f8d88f189a6e73e1c : SUCCESS in 1h 12m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/aadd2b9e2cbd408bb34da87a5e9613bd : SUCCESS in 1h 35m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c3d1c7d2bbbc4ee38775f1150b053e0b : FAILURE in 1h 40m 20s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2e3fed78f7c149fc85bb90867cea945b : SUCCESS in 47m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bb731f0614954314976cc0b672e823b9 : FAILURE in 1h 59m 54s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/65ca07ea208f484094cb1fdfcc541fe2 : SUCCESS in 24m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"abc8e3934ea4c3bf3a375904f62c71f1fcc63bbe","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-01 11:35:33.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/834431/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b2c114a/ : FAILURE in 2m 40s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/834431/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/5acd12f/ : FAILURE in 2m 41s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"e0d7502f3ef1a9117b2d585d987c7af5c5d7195c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-01 21:14:08.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/31/834431/5/check/DellEMC_Unity_Manila_dhss_true/422839c/DellEMC_Unity_Manila_dhss_true/342 : FAILURE in 4h 35m 23s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/834431/5/check/DellEMC_Unity_Manila_dhss_false/be5aec7/DellEMC_Unity_Manila_dhss_false/360 : FAILURE in 28m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"95c989537d21a2cae3fb18def9a004332e8ddaf9","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-13 08:14:50.000000000","message":"Patch Set 5: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b396e0012655bea72ce0063db9f9299fe75eb277","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-04 23:16:22.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"91d1740f8e3540a329f556f22c56831822821eb9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-07 07:43:10.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"93bd453469feeb6fd63ca09c31d183c661d44ad7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 09:30:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af838615454d45519162749d3c474f84 : SUCCESS in 3m 11s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a29b220e3e444632b1b0fa899c3312c6 : SUCCESS in 25m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/50729bd78ed64950bed56a85f5872218 : FAILURE in 35m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0e1f6d9c08824767afb1f7b2ba6f0769 : SUCCESS in 1h 13m 31s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/cc6ab25d6fca43879fd12e2b32bad016 : SUCCESS in 1h 07m 29s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/2d72af498371462198658e55d9939576 : SUCCESS in 1h 04m 41s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/75404a2f13334518b4b1e25351650438 : SUCCESS in 57m 32s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a73973c3828c48efb9edd16e6c8fca60 : SUCCESS in 1h 12m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/78ec5d3fc1824641afc2de12a1e8edd9 : SUCCESS in 1h 15m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/74697915e5e84823a072f6c155c6ad03 : FAILURE in 1h 29m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d1d92266a36b45c8a4b0be2eb05cfaa6 : SUCCESS in 56m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b4828e9665d9401fa87409dbe50db6d2 : SUCCESS in 1h 45m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3070fb66b3334516ab22f21b5807319f : SUCCESS in 22m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"adc94008c3d2e76c8bedf8de53d1803ffc451688","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-09 16:21:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1a1e92581705581c5eb57799678d591ddb3dbfd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 18:59:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b0c2bc598344b98a647af56d612b77f : SUCCESS in 3m 23s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/38e5976630ac4dd597334d0cb1df6d05 : SUCCESS in 33m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ad4e9dc86424435fb93b4f953083f58f : SUCCESS in 30m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1ed0860542114d1b87676e38c91a5d09 : SUCCESS in 1h 07m 56s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/efc84d8c3dc8407ebeea85383e49c9c8 : SUCCESS in 1h 20m 10s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/94151aa0eaaf4837baf67f7d88064bb6 : SUCCESS in 1h 03m 40s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/ec449acc389e4d06b77feaeeecc0d83d : SUCCESS in 49m 16s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/74971c5a22d040dbb3dcd6996f6dc3fa : SUCCESS in 1h 15m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e35a43e5d091454c9d11f8e113500443 : SUCCESS in 1h 18m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4922bfc8e209472889063f1a3ac0f609 : FAILURE in 1h 31m 03s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f9c8c6913b3245ba98f55ef0effcb8db : SUCCESS in 58m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/13e1f16db606414d97ba7e043997ca2a : FAILURE in 2h 35m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a73cc92fa6b44a7a8e11cec2bdb3cc56 : SUCCESS in 20m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"98292ac84d3257312b33c923f25d4716c30b128d","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-09 19:01:01.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/31/834431/5/check/DellEMC_VNX_Manila/7645345/DellEMC_VNX_Manila/258 : FAILURE in 1m 16s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"341ad7faa1e486a23ea23e743399ebeb2ba0ed6d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-10 00:13:30.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/31/834431/5/check/DellEMC_Unity_Manila_dhss_true/6beddb4/DellEMC_Unity_Manila_dhss_true/468 : FAILURE in 57m 12s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/834431/5/check/DellEMC_Unity_Manila_dhss_false/cca9463/DellEMC_Unity_Manila_dhss_false/489 : SUCCESS in 1h 29m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"481bd7e3866dbb5897b9ecab39fd13ade31fe751","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-10 13:38:57.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f887f22327c9bfaae89804b9fd7b854ff9cd2062","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 13:40:49.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ccc9c41ea20b99b1d67161495f8387689b4dead6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 14:53:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6c58cd15ea24aa48fd9b37f6fc8bace : SUCCESS in 3m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/43fee2175c3e4ecda50fe4bfecbe7b54 : SUCCESS in 30m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fa78891372b04035a8dd3c32b5b8d4aa : SUCCESS in 35m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7982295d25de477ab26ff3842a560019 : SUCCESS in 1h 12m 11s","accounts_in_message":[],"_revision_number":5},{"id":"0596d25802cff2a40eb9b247d2d7d4031fade920","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 14:53:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fd4e2f325cf3eedfc929bca9c0398e2640e91650","revisions":{"f1be9c3671c061fe1c0ef49a05d4b6cde2a81cac":{"kind":"REWORK","_number":1,"created":"2022-03-21 09:21:55.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/31/834431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/834431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/1"}}},"commit":{"parents":[{"commit":"1ca1f64275fafc9f3bcb8ef5f7a7bcf281141b9e","subject":"Remove _disassociate_floating_ip \u0026 _create_network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1ca1f64275fafc9f3bcb8ef5f7a7bcf281141b9e"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 09:16:30.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 09:17:34.000000000","tz":60},"subject":"Remove get_remote_client \u0026 check_vm_connectivity","message":"Remove get_remote_client \u0026 check_vm_connectivity\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* get_remote_client\n\t* check_vm_connectivity\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0911cda7fcf461b9786f9534689083c7ac3a2035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1be9c3671c061fe1c0ef49a05d4b6cde2a81cac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1be9c3671c061fe1c0ef49a05d4b6cde2a81cac"}]},"branch":"refs/heads/master"},"8cda00b6a1e29dcf34594b9e87d608ec3b57cf0d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-22 07:00:15.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/31/834431/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/834431/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/2"}}},"commit":{"parents":[{"commit":"4a36559c435f07865677a00ec3da3b1abec18624","subject":"Remove _disassociate_floating_ip \u0026 _create_network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a36559c435f07865677a00ec3da3b1abec18624"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 09:16:30.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-22 07:00:15.000000000","tz":0},"subject":"Remove get_remote_client \u0026 check_vm_connectivity","message":"Remove get_remote_client \u0026 check_vm_connectivity\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* get_remote_client\n\t* check_vm_connectivity\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0911cda7fcf461b9786f9534689083c7ac3a2035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8cda00b6a1e29dcf34594b9e87d608ec3b57cf0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8cda00b6a1e29dcf34594b9e87d608ec3b57cf0d"}]},"branch":"refs/heads/master","description":"Rebase"},"b67d7ce410987d49087b49ca5be819a32c317fb1":{"kind":"NO_CHANGE","_number":3,"created":"2022-03-22 15:53:34.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/31/834431/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/834431/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/3"}}},"commit":{"parents":[{"commit":"f01c60eac31465af0b9425e7ede4ba540b8192f8","subject":"Remove _disassociate_floating_ip \u0026 _create_network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f01c60eac31465af0b9425e7ede4ba540b8192f8"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 09:16:30.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-22 15:53:34.000000000","tz":0},"subject":"Remove get_remote_client \u0026 check_vm_connectivity","message":"Remove get_remote_client \u0026 check_vm_connectivity\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* get_remote_client\n\t* check_vm_connectivity\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0911cda7fcf461b9786f9534689083c7ac3a2035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b67d7ce410987d49087b49ca5be819a32c317fb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b67d7ce410987d49087b49ca5be819a32c317fb1"}]},"branch":"refs/heads/master","description":"Rebase"},"158283b27adfac98e9b2c8272b3365c176474d07":{"kind":"NO_CHANGE","_number":4,"created":"2022-04-01 09:20:32.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/31/834431/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/834431/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/4"}}},"commit":{"parents":[{"commit":"1118f3ebe4773a1119685fada4b368d0d554f986","subject":"Remove _disassociate_floating_ip \u0026 _create_network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1118f3ebe4773a1119685fada4b368d0d554f986"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 09:16:30.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:20:32.000000000","tz":0},"subject":"Remove get_remote_client \u0026 check_vm_connectivity","message":"Remove get_remote_client \u0026 check_vm_connectivity\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* get_remote_client\n\t* check_vm_connectivity\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0911cda7fcf461b9786f9534689083c7ac3a2035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/158283b27adfac98e9b2c8272b3365c176474d07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/158283b27adfac98e9b2c8272b3365c176474d07"}]},"branch":"refs/heads/master","description":"Rebase"},"fd4e2f325cf3eedfc929bca9c0398e2640e91650":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-04-01 09:23:42.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/31/834431/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/834431/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/834431/5"}}},"commit":{"parents":[{"commit":"1118f3ebe4773a1119685fada4b368d0d554f986","subject":"Remove _disassociate_floating_ip \u0026 _create_network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1118f3ebe4773a1119685fada4b368d0d554f986"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 09:16:30.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:23:42.000000000","tz":0},"subject":"Remove get_remote_client \u0026 check_vm_connectivity","message":"Remove get_remote_client \u0026 check_vm_connectivity\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n        * get_remote_client\n        * check_vm_connectivity\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I0911cda7fcf461b9786f9534689083c7ac3a2035\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fd4e2f325cf3eedfc929bca9c0398e2640e91650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fd4e2f325cf3eedfc929bca9c0398e2640e91650"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
