)]}'
{"id":"openstack%2Fmanila-tempest-plugin~834434","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I2f6b61d868f9fc68cf000627a5a905083af6ace7","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-05-12 17:57:38.000000000","reason":"Goutham Pacha Ravi replied on the change"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-05-16 19:39:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2f6b61d868f9fc68cf000627a5a905083af6ace7","subject":"Remove create_floating_ip \u0026 get_server_ip methods","status":"MERGED","created":"2022-03-21 09:39:11.000000000","updated":"2022-05-16 19:39:02.000000000","submitted":"2022-05-16 19:39:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"834434-tempest-scenario-manager-cleanup","meta_rev_id":"7a9a769c2ed3330eb37eb6a5470794e1dbd82cc1","_number":834434,"virtual_id_number":834434,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-16 19:39:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-05-12 17:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-05-04 23:16:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-05-16 18:15:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-21 11:42:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-22 02:43:33.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-03-22 13:01:08.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-23 03:05:25.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 13:05:08.000000000","updated_by":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-05-04 23:16:43.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"117096153af3244543deb28ee96be8b64c88b17b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-21 09:39:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bd7d58009fcc1619c58dfdbd6744718ebb9633e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 11:42:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72c238560fbd48c39be9ebeaf1ed7d2d : SUCCESS in 5m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b7e1358f50df436bb2e9d601524636ad : SUCCESS in 27m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0f4fa67efb034c81922a52a4ea65eab7 : SUCCESS in 40m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/eb02c143fb0c4d899215841cf81f7bc5 : FAILURE in 47m 07s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d72168ce122d4366a02a405b2655b1b3 : FAILURE in 48m 57s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d7f656bd0b6743ccb57186e95a76f4e5 : FAILURE in 1h 00m 20s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/015602b0561043468c0be42ee6fb961b : FAILURE in 1h 11m 14s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2487af0ab4ab4392b64ee5101025017a : FAILURE in 53m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f1750ed7b527455391f298df97986ce2 : FAILURE in 51m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/56743be9824443688ae965dd70ab303c : FAILURE in 1h 25m 36s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/33edbf779de64a54a46e131afa222b6e : SUCCESS in 1h 32m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/dfc019a956094bc8891ff77033be2881 : FAILURE in 1h 57m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2c12bb0c6a234d809d5f3c6d2194b449 : SUCCESS in 29m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0b0fd24bdb7e908fc3a77d3031649ae5300e71d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-22 02:43:33.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/834434/1/check/DellEMC_Unity_Manila_dhss_true/61b7c18/DellEMC_Unity_Manila_dhss_true/304 : ABORTED in 3h 19m 21s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/834434/1/check/DellEMC_Unity_Manila_dhss_false/a33abe4/DellEMC_Unity_Manila_dhss_false/310 : FAILURE in 1h 35m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"b189fcf9ede9114954f62e6eef24d0175d01c9b9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-22 07:00:29.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"73e0d23f6ca7787e161188a6639260a2d8059bed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 09:32:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ca8dea634a143a79c8ae329f7674a81 : SUCCESS in 5m 32s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9cd3beb018034e5a9e16fa0a1c7ade04 : SUCCESS in 30m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e37ac6c1d64e4969ab5ddf20a2fcdcc4 : SUCCESS in 1h 01m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a2ef97758ae5493486aa387019e23274 : SUCCESS in 1h 02m 19s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/ca0288bc57904dddb089f0cd6e975753 : SUCCESS in 59m 15s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/2e4c79c309014bd68f8bc74cc3622f8e : SUCCESS in 1h 12m 40s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/89c552bffd6240cea5813b8a1a0219eb : SUCCESS in 1h 13m 26s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0fa6e3089f864c8ea4dcbe0eb1e01bdb : FAILURE in 1h 45m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c8392264d87e43a88cdbb2a8143a8aec : SUCCESS in 1h 24m 03s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/72e9246e8ab6485c94b7d674f6043bcb : FAILURE in 1h 47m 12s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8f8a168eea9e48e8bd185ea96888fc7a : SUCCESS in 46m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/dedfda92eb88413b9e5d3e32852653f9 : FAILURE in 2h 26m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0d0f993e181e4382a256e08ac387abd7 : SUCCESS in 22m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"471a84e1dd193edf9d97b8f7f12211af45b9b6ef","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-22 10:40:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c051861c34dea0a083c0fb569aee7ca08aab722a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:57:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54982926b48c45e1b568342643609999 : SUCCESS in 5m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d439dcb5fb1e4b4388fdc213b5680fc8 : SUCCESS in 28m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4cbc842b26354d7f88f7dbea59945760 : SUCCESS in 34m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ab56c03022354343846da011b973b551 : SUCCESS in 1h 21m 50s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/43805ecebcda461e82ff8f5364f7cbc6 : SUCCESS in 1h 05m 10s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/5bffdbf5e15347c99932392852b3d730 : SUCCESS in 1h 09m 32s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/d0085c924c4e42768f2229ef9cba0146 : SUCCESS in 1h 15m 57s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/676fb4e9e1ff4cba909673346e21db96 : SUCCESS in 1h 22m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c67909753ba447bcbf1ad2eee7ef7970 : SUCCESS in 1h 46m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9f2104ac5fed434f8a3edbe5847b3acc : FAILURE in 2h 06m 42s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cc89f960bb4c4949ada3cf666ae61d09 : SUCCESS in 58m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/47bfe871fc884fcb834e02853348bafa : FAILURE in 1h 36m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a8b4efd460364f84bfd5c9e39cfddb74 : SUCCESS in 29m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"beb2077ddafe72d986125990898e2e3ebe4f469d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-22 13:01:08.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/834434/2/upstream-check/manila-cDOT-manila-tempest-plugin-ss/29c1f5b/ : FAILURE in 2m 58s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/834434/2/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/781a847/ : FAILURE in 2m 49s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e498b2a2e1e3b9fb6bb200c7e6ebec6101039930","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-22 15:53:44.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"476be2f3ca9cacbef913c50c80c43d6553b4679b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 18:28:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fde93c270524176a42e0a627b58fcb6 : SUCCESS in 5m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cc2a2960b71542f98d803beec8978b8d : SUCCESS in 33m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/563070741a494e7395b702f9490ba5df : SUCCESS in 35m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/43289ce755fc4b5cbac5bd3b7861922d : SUCCESS in 1h 21m 03s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c5eef04c2172442c95dbc4a1bf043563 : SUCCESS in 1h 03m 30s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/291ee060b0ac434c9dd611fbc58af585 : SUCCESS in 1h 36m 00s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/79278ec326684562b4c20da4a126f6c4 : SUCCESS in 1h 12m 56s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3be8161d1b6947b4bbdf1e89b9721441 : SUCCESS in 1h 22m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9aeb4a78007044ad80684d322e8134d5 : SUCCESS in 1h 35m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/57b247a5dad742848e43a08ebb87b1dc : FAILURE in 1h 49m 06s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e54b5d2b3d93403095153483094769eb : SUCCESS in 1h 05m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/af3646309825482c9af3cf2394b6bee3 : FAILURE in 2h 24m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/995affa07eed44cb89e963d092efda64 : SUCCESS in 28m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5123e4214cf7e7e248c5c52b1203686f76377cdf","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-22 18:31:43.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/834434/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/035be2b/ : FAILURE in 3m 14s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/834434/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/2b11dc1/ : FAILURE in 3m 15s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"7db8da87d83898d2fdc97612c4451507f52190bc","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-23 03:05:25.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/34/834434/3/check/DellEMC_VNX_Manila/f05ba6c/DellEMC_VNX_Manila/179 : FAILURE in 2h 09m 41s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"1e46cd8bc577cf2ce7a78e39d803676e23b0fb34","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-24 10:22:50.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"90488395ff3d473659802b148577c1148f113a83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:20:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"929fc152485d0a09a80fdff10ff0e77941094197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:24:00.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"92e31149f68cf636a402a64aa3778a39bcae4318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 11:28:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466bc39beb7b4bfab0d93c1542e81857 : SUCCESS in 4m 21s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7d845b26f7eb479c9810058564a5bfc5 : SUCCESS in 32m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f478a6f354e64cbdab7c12476a07bcac : SUCCESS in 33m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bfee2b415a214fa287c75507ffd5c464 : SUCCESS in 1h 09m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/2445ec4227444a0ebfd7c376a99dbc84 : SUCCESS in 1h 06m 08s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/4998e56ac76e49279cdabfd8a87fee60 : SUCCESS in 1h 06m 34s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/4fb13170dbfe4066a24a6ba387a3423b : SUCCESS in 1h 12m 43s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f42ad6f4cfb04a46a31adaf9956f36ce : FAILURE in 1h 12m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/94307e10aa7b40bcab3b3074ac6e950e : SUCCESS in 1h 28m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/df76cc2f415c46c2928d6a8c7a0c1bf5 : FAILURE in 1h 59m 44s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cef4bd5519774842a97fbaed7f011f78 : SUCCESS in 55m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/efc9f3536d474c4f9d977ec5d17f9b47 : SUCCESS in 1h 50m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e6b6ee03fab14c1aba262568db1dddcb : SUCCESS in 22m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c2d5ee7e50001d0be8c0874123b4bbdcde909506","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-01 11:31:41.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/834434/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e6c3ff5/ : FAILURE in 2m 49s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/34/834434/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/9b278e1/ : FAILURE in 2m 41s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"827a5a457ae4c164348da0eb11f148d613ef780d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-02 06:04:56.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/834434/5/check/DellEMC_Unity_Manila_dhss_true/c83537a/DellEMC_Unity_Manila_dhss_true/344 : FAILURE in 4h 33m 34s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/834434/5/check/DellEMC_Unity_Manila_dhss_false/d2ec5c1/DellEMC_Unity_Manila_dhss_false/362 : FAILURE in 28m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"c0d1b013f909d0c7d179945a50a6d6e77f3f9903","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-04 23:16:43.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f30db7586481635c563f53f8184b53939d915a06","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-12 17:57:38.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a57ca8799c9983150867889a7fa376896a4ade06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 19:49:57.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa5d456a2ae04121b9bc5afff3896e52 : SUCCESS in 3m 17s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f02bbbac026f4c64ac866e722802fc55 : SUCCESS in 35m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fc64f7c481b64bd198c305bab923e10e : SUCCESS in 40m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/902271876b2a4df2bbee3c107adec203 : SUCCESS in 1h 12m 41s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/9f193e637edd4b3aaf5da9f2b92d765d : SUCCESS in 1h 12m 15s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/07656dcd828d4477b0bad40f31c75b33 : SUCCESS in 1h 04m 52s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/6bb9476f4a6c40ebac75f3aac3ba789d : FAILURE in 1h 20m 30s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2fd6a8d3208441a48b5fd0a06077eca4 : SUCCESS in 1h 28m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/287beead25d1439f92cf877f10b3236c : SUCCESS in 1h 20m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/855fa60265bf44af809b4562ea1e0c87 : FAILURE in 1h 47m 02s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fa8514d273c14702b04016510523259c : SUCCESS in 56m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f0036ca73af54684b0582f8d1324aa31 : FAILURE in 1h 22m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3eb99d0ebf8d40cda0006eb26fc18a64 : SUCCESS in 23m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"95b4a1ca6b8085a03bb1f533272ec707381fb722","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-13 12:46:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fc7f8b20aae1b44b6ba847f7e83fe48cedc7ee7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 15:09:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63d2b1858427416891ac3c82f535f8bc : SUCCESS in 3m 06s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a73ee18c95bc45a08c1fa55cedd2d851 : SUCCESS in 39m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/111c46e27e26441e8da5983386965bbf : SUCCESS in 43m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fed25f8c261f4e77b127fcf8b6294e61 : SUCCESS in 1h 25m 08s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/71388e78d2b44ef783562f326aa8b11f : SUCCESS in 1h 06m 55s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e0c90a655f4a418886b9892187baaf34 : SUCCESS in 1h 16m 35s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/0c6ee282159640f685ed5dcd2a282d49 : SUCCESS in 1h 13m 14s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7e7dd42187e04db28cb29a92b8bc41a1 : FAILURE in 1h 30m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/74bb0800033e4744954102785bda5d9b : SUCCESS in 1h 20m 38s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/793ba5eeaa2542309ebac2a1b6d66115 : FAILURE in 1h 50m 05s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a52a984f817748a38f85f6025ba2b25b : SUCCESS in 1h 04m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d92c12d1a4694481bbfede1fb18bb227 : FAILURE in 2h 18m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cd0fef53da824fdbab12c71cd9df447a : SUCCESS in 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0d3c1d0b17f9e51e5bf7a4cd2a2f221c30bac0ed","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-13 15:11:21.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/34/834434/5/check/DellEMC_VNX_Manila/6ccac27/DellEMC_VNX_Manila/266 : FAILURE in 1m 19s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":5},{"id":"e52d8fc8eec9851641b8d9148eb0b7b12166c1eb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-13 19:16:43.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/34/834434/5/check/DellEMC_Unity_Manila_dhss_true/cc7969a/DellEMC_Unity_Manila_dhss_true/480 : FAILURE in 3h 22m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/34/834434/5/check/DellEMC_Unity_Manila_dhss_false/a7770b1/DellEMC_Unity_Manila_dhss_false/501 : SUCCESS in 1h 06m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"e5f95e635fb33cc694f87acbc6264509cd880611","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-16 18:15:43.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"4bf1398b9be2b69b3fd640d27d5d87164f1e6d1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 18:16:27.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ad6e293e790e7ff8f6f2790db5e23b444e8f84ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 19:39:01.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9126ad1d19784f9b85709a4c68041e6f : SUCCESS in 3m 08s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/890f89f2c6794756ac24e1060c9e2a02 : SUCCESS in 33m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/92485c9fdfac4323afda80fddb80ec93 : SUCCESS in 34m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/455f2c55548c489f836bb5131c1b52f5 : SUCCESS in 1h 21m 53s","accounts_in_message":[],"_revision_number":5},{"id":"7a9a769c2ed3330eb37eb6a5470794e1dbd82cc1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 19:39:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ee3009ecb0d8e898f140e9cb4303ddc1eae85e0d","revisions":{"ed259e504a16f8c5d9e944945432dd88db460890":{"kind":"REWORK","_number":1,"created":"2022-03-21 09:39:11.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/34/834434/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/834434/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/1"}}},"commit":{"parents":[{"commit":"6f6d4ed12812fa2d8a2b77597f430eec486ca0ab","subject":"Remove create_timestamp \u0026 get_timestamp methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6f6d4ed12812fa2d8a2b77597f430eec486ca0ab"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 09:38:49.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 09:38:49.000000000","tz":60},"subject":"Remove create_floating_ip \u0026 get_server_ip methods","message":"Remove create_floating_ip \u0026 get_server_ip methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nBarbican tempest plugin still uses nova-network old interface for floating ip related methods, this patch removes them and use tempests\u0027s ones which already use neutron\u0027s interface.[2]\n\nReplaced methods:\n\t* create_floating_ip\n\t* get_server_ip\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n[2] https://opendev.org/openstack/tempest/commit/6428139de18291c916852ae48130fb8a7b4db318\n\nChange-Id: I2f6b61d868f9fc68cf000627a5a905083af6ace7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ed259e504a16f8c5d9e944945432dd88db460890"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ed259e504a16f8c5d9e944945432dd88db460890"}]},"branch":"refs/heads/master"},"147e14523e421462d044d2997e1359e9247bafaf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-22 07:00:29.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/34/834434/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/834434/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/2"}}},"commit":{"parents":[{"commit":"dcf6235408e3223f4f053581131a8ee5cce5903d","subject":"Remove create_timestamp \u0026 get_timestamp methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dcf6235408e3223f4f053581131a8ee5cce5903d"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 09:38:49.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-22 07:00:29.000000000","tz":0},"subject":"Remove create_floating_ip \u0026 get_server_ip methods","message":"Remove create_floating_ip \u0026 get_server_ip methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nBarbican tempest plugin still uses nova-network old interface for floating ip related methods, this patch removes them and use tempests\u0027s ones which already use neutron\u0027s interface.[2]\n\nReplaced methods:\n\t* create_floating_ip\n\t* get_server_ip\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n[2] https://opendev.org/openstack/tempest/commit/6428139de18291c916852ae48130fb8a7b4db318\n\nChange-Id: I2f6b61d868f9fc68cf000627a5a905083af6ace7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/147e14523e421462d044d2997e1359e9247bafaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/147e14523e421462d044d2997e1359e9247bafaf"}]},"branch":"refs/heads/master","description":"Rebase"},"cabc256f1023e99323a4e08aba65f0ea31a6e8c3":{"kind":"NO_CHANGE","_number":3,"created":"2022-03-22 15:53:44.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/34/834434/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/834434/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/3"}}},"commit":{"parents":[{"commit":"161e7b56a6b7314910fb3329f9d88b0a1bd8f73e","subject":"Remove create_timestamp \u0026 get_timestamp methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/161e7b56a6b7314910fb3329f9d88b0a1bd8f73e"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 09:38:49.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-22 15:53:44.000000000","tz":0},"subject":"Remove create_floating_ip \u0026 get_server_ip methods","message":"Remove create_floating_ip \u0026 get_server_ip methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nBarbican tempest plugin still uses nova-network old interface for floating ip related methods, this patch removes them and use tempests\u0027s ones which already use neutron\u0027s interface.[2]\n\nReplaced methods:\n\t* create_floating_ip\n\t* get_server_ip\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n[2] https://opendev.org/openstack/tempest/commit/6428139de18291c916852ae48130fb8a7b4db318\n\nChange-Id: I2f6b61d868f9fc68cf000627a5a905083af6ace7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cabc256f1023e99323a4e08aba65f0ea31a6e8c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cabc256f1023e99323a4e08aba65f0ea31a6e8c3"}]},"branch":"refs/heads/master","description":"Rebase"},"bf2904fd349a701ef170f7170e872f1ce8aeb505":{"kind":"NO_CHANGE","_number":4,"created":"2022-04-01 09:20:42.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/34/834434/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/834434/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/4"}}},"commit":{"parents":[{"commit":"0a9550689d8230b40f38495f03dba08c66bcae98","subject":"Remove create_timestamp \u0026 get_timestamp methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a9550689d8230b40f38495f03dba08c66bcae98"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 09:38:49.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:20:42.000000000","tz":0},"subject":"Remove create_floating_ip \u0026 get_server_ip methods","message":"Remove create_floating_ip \u0026 get_server_ip methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nBarbican tempest plugin still uses nova-network old interface for floating ip related methods, this patch removes them and use tempests\u0027s ones which already use neutron\u0027s interface.[2]\n\nReplaced methods:\n\t* create_floating_ip\n\t* get_server_ip\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n[2] https://opendev.org/openstack/tempest/commit/6428139de18291c916852ae48130fb8a7b4db318\n\nChange-Id: I2f6b61d868f9fc68cf000627a5a905083af6ace7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bf2904fd349a701ef170f7170e872f1ce8aeb505"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bf2904fd349a701ef170f7170e872f1ce8aeb505"}]},"branch":"refs/heads/master","description":"Rebase"},"ee3009ecb0d8e898f140e9cb4303ddc1eae85e0d":{"kind":"NO_CHANGE","_number":5,"created":"2022-04-01 09:24:00.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/34/834434/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/34/834434/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/34/834434/5"}}},"commit":{"parents":[{"commit":"a505b60961aad2e2a6ced5153bb53747bac05ac6","subject":"Remove create_timestamp \u0026 get_timestamp methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a505b60961aad2e2a6ced5153bb53747bac05ac6"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-21 09:38:49.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:24:00.000000000","tz":0},"subject":"Remove create_floating_ip \u0026 get_server_ip methods","message":"Remove create_floating_ip \u0026 get_server_ip methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nBarbican tempest plugin still uses nova-network old interface for floating ip related methods, this patch removes them and use tempests\u0027s ones which already use neutron\u0027s interface.[2]\n\nReplaced methods:\n\t* create_floating_ip\n\t* get_server_ip\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n[2] https://opendev.org/openstack/tempest/commit/6428139de18291c916852ae48130fb8a7b4db318\n\nChange-Id: I2f6b61d868f9fc68cf000627a5a905083af6ace7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee3009ecb0d8e898f140e9cb4303ddc1eae85e0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee3009ecb0d8e898f140e9cb4303ddc1eae85e0d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
