)]}'
{"id":"openstack%2Fmanila-tempest-plugin~834817","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie17c1d9349e98ce016801bf47c1cb9dba18b51f6","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-06-08 12:22:49.000000000","reason":"Goutham Pacha Ravi replied on the change"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-07-11 19:52:34.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-07-11 19:52:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie17c1d9349e98ce016801bf47c1cb9dba18b51f6","subject":"Remove check_floating_ip_status \u0026 _associate_floating_ip methods","status":"MERGED","created":"2022-03-23 07:37:25.000000000","updated":"2022-07-11 20:57:11.000000000","submitted":"2022-07-11 19:52:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"834817-tempest-scenario-manager-cleanup","meta_rev_id":"ce979fe1eabd3276733770296ec46e2f06c03733","_number":834817,"virtual_id_number":834817,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-11 19:52:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-07-08 20:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-07-11 18:41:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-23 09:41:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 09:45:06.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-23 12:03:10.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 13:05:37.000000000","updated_by":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-04-02 08:06:10.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-04-13 08:18:52.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-05-04 19:20:41.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-05-04 23:17:07.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-07-08 16:46:45.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"f1565a0732faaf6ecfb9fa728dbeb3d04549bb4b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-23 07:37:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d7c8128166a76f80c592c9572f047676263cb21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 09:41:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2fbb788becb40a3a14731e9660e2df2 : SUCCESS in 5m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e01d2dee22e44115a728b6db655db6ad : SUCCESS in 36m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0bf2af6a463f4d3589791c61d37a704e : SUCCESS in 45m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2dc6005967744ba5b488207e36325165 : SUCCESS in 1h 17m 38s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/b791d307dbb64cdcb20696eca4a7fc23 : SUCCESS in 1h 03m 30s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/605447b5fa1048faa42a177f288eedd5 : SUCCESS in 1h 09m 05s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/b8f028720bc5447c8d8bb6b695e824fa : SUCCESS in 1h 10m 40s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9f5742ff09d34547a6722f00f930c098 : FAILURE in 1h 16m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5de5f3c14be94273903dd9f6c2fc9e03 : SUCCESS in 1h 19m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fb52eab153664896bf3ecf31e88a766a : FAILURE in 1h 58m 50s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f2b082f6ce62442b8e77bae0af3e753c : SUCCESS in 56m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9a6d9d8c57b44385a805361394ce32ca : FAILURE in 1h 15m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/00b589243d264288b2a0e859ffd14aca : SUCCESS in 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19e498720a21621505817b4ae52939b33c9d0f12","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-23 09:45:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/834817/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/b22d1c3/ : FAILURE in 2m 49s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/834817/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/efc5f49/ : FAILURE in 3m 02s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"a725e13643e7ec20daa1aad478c03677e9fce07e","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-23 12:03:10.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/17/834817/1/check/DellEMC_VNX_Manila/6c5db97/DellEMC_VNX_Manila/182 : FAILURE in 2h 21m 15s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"49732eb5ee40a8f542326b3fab59d5c6d450ff1f","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-24 10:22:58.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"efde9e131c4e148675e273ed7d3b7dca7d6d0577","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:20:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"69956b7f25f9d63a93fad13d0c9e7a943cdc4b58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:24:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"97b1a77ad2a910a9ddce163fe488baf071a70d30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 11:33:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ecc66a84f16490cb9727f445fe3641b : SUCCESS in 3m 53s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/90e4924bd73f4ee6a9220790a4f78ffd : SUCCESS in 29m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/36c7face27a549c88a259ab63058a68e : SUCCESS in 43m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fb74ebbcc6834f29a4678baae8aa48ec : SUCCESS in 1h 07m 51s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/4c0f805ccb3c499fbe6c9d1e583c67f1 : SUCCESS in 1h 22m 57s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/462311c710df49ae9025faf88623584a : SUCCESS in 1h 21m 41s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/ecb5124c59644046a1a04cefbc05d105 : SUCCESS in 1h 36m 09s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9403d96beece4171a86dd7b696862c0b : SUCCESS in 1h 16m 48s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/24fe8e865fd9485089d0ae4d2f1f93bb : SUCCESS in 1h 45m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d8422244bf4f499f8d241b0ce2339bff : FAILURE in 1h 53m 08s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a100ed4d3e6840e887971919e0fc0390 : SUCCESS in 1h 04m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8c2853a3b2594cecb912938be1d4e225 : SUCCESS in 2h 04m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e31d829b7109410781bd1db361e90803 : SUCCESS in 29m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bd86d2f4956a1169d8da6732f8591a7765ee8660","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-01 11:48:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/834817/3/upstream-check/manila-cDOT-manila-tempest-plugin-ss/52273f2/ : FAILURE in 2m 43s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/834817/3/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/3609e1f/ : FAILURE in 2m 41s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"e1b4a5132b20b704e6767cdc7ec0f4338c23816d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-02 08:06:10.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/17/834817/3/check/DellEMC_Unity_Manila_dhss_true/6170d9b/DellEMC_Unity_Manila_dhss_true/346 : ABORTED in 8m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/17/834817/3/check/DellEMC_Unity_Manila_dhss_false/daaf404/DellEMC_Unity_Manila_dhss_false/364 : FAILURE in 28m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"f3f3d9e4aa7bb5e19fea0802cd4d34586a5b9b80","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-04-13 08:18:52.000000000","message":"Patch Set 3: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"051d0c165745399c5c270bd00dc76f99cada61f4","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-05-04 19:20:41.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"876f413870bf6ee8e2923cb383db5298643003b8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-04 23:17:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bab3dedeec6de471cc33772de5bfcde3ee84fbfb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-08 12:22:49.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"919521e7e9af792bcb365726d0ef233a3e26c853","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 15:38:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4635ac4c5d304059bb70b3e5dcf77e2a : SUCCESS in 3m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0498b1244dde48579229f716693ae532 : SUCCESS in 40m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c48fdb5986aa4982b790207f8e2fae58 : SUCCESS in 32m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b22e44e9618c4af68affc96f73f6e884 : SUCCESS in 1h 05m 19s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d3b0fc09df714db69775e9fca4d95204 : SUCCESS in 1h 14m 28s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/ced08d5aa26c4c4fa9982ec8f9e2f504 : SUCCESS in 59m 54s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/78bb0c0847e24ee2865c0da0bd0409e8 : FAILURE in 1h 05m 10s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/057d2c787a6f4c739f1e86d8445fb537 : FAILURE in 1h 26m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a889403641b141d28b2e970e831aebb5 : SUCCESS in 1h 28m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/66a51a0ad98a45269a54ed1842af57fb : FAILURE in 33m 42s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7215c6943f6043b49dcf54a57aaf40d9 : SUCCESS in 1h 08m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a8b94d72df2047449d4e7fefdadac94e : TIMED_OUT in 3h 08m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a6189298a1dd42259e3f607ece2717e2 : FAILURE in 17m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"60ae88c4960580283315ae1313704a37c6159e9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-06-17 08:11:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"94a4c06cda0deab65ea718f93fae0dab90e315e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 10:34:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a2e5af6cfdc6460c83528354a2bd1e21 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b4e3aeee32046b68f4d4ca2f4799417 : SUCCESS in 4m 12s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ec1fd59c140f4d389dbd7d470f5b5e33 : SUCCESS in 29m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c389146b33534426a88b9cb80d9b0320 : SUCCESS in 30m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/909988204a6f421c8a107fd1ec9be3a7 : SUCCESS in 1h 08m 46s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/1b4123dfd3e64896b59a9872592397c7 : FAILURE in 44m 19s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/47a0d261e82c4e5ea5a895f69f169f3f : FAILURE in 39m 40s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/4231d71b6c7b4e0c890987188c0ce437 : FAILURE in 41m 02s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8c341dd2fee34a8ba92d9b0b4c8b7a46 : FAILURE in 1h 05m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b3339f752fdd4d9287476c9c91b34c87 : FAILURE in 1h 40m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4a87894fc4f74555b1a39fb90f7f8e63 : FAILURE in 1h 59m 03s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c33aa7c54a694c2d8ea75564a80e5db6 : SUCCESS in 47m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/92f43a6ae413448b812da371a5ebb8d0 : SUCCESS in 1h 15m 03s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9d7a40479ddd45898c3383049d99726f : FAILURE in 2h 06m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d93e21d01def4dce9fb7642eeaa38014 : SUCCESS in 23m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2ee10dd151628069d311d15ddeb88d789a5d4bad","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-06-20 15:20:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5dbdf59dda58c1b41912d4ac8dda9d8705000190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 18:05:14.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/77a0fe7fcf344cd59f418bbd57155564 : SUCCESS in 2m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16918bfd0fff4f6f9c0a350a79f14d06 : SUCCESS in 3m 08s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/955d7c3d05af4b57ab7b29ee158f8d1a : SUCCESS in 39m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/029291bdb66b46898631f85ca62c8ecc : SUCCESS in 44m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d63c2ce979744921ba7a01e21d02438c : SUCCESS in 1h 07m 13s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/5b2a39566faf4233bbf15d5c3efb161c : FAILURE in 44m 15s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/02dc9d87e5e246498b46328b7f65a306 : FAILURE in 35m 14s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c6af8ce144ed4282941a43a3e2cc54c9 : FAILURE in 38m 00s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b462ff2f83034c35aa583b62b5f641b0 : SUCCESS in 1h 08m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/afe5c2e250b44cce8ebfb4a618e86c77 : SUCCESS in 1h 11m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/72ddc5b4a0dd4e4588229f3a7ead52e3 : FAILURE in 1h 32m 44s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9f3aab6e1b604235ae417631540dcb25 : SUCCESS in 54m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7309e7bbdfb54427ac6b113ea6c68e46 : FAILURE in 2h 22m 29s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/103139efc1694e268084c1d15a931f83 : FAILURE in 2h 35m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7a0929f192c84afca1aebc415399f240 : SUCCESS in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2aa0b739eab601a9c4f52b3f319f05342745bb9d","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-06-21 08:09:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ec2de902e9fc5c8e68954ef77780b1263a7d301e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-08 14:37:52.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"51845446ed67e1a593b90c5a8c2814655928f910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 16:37:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/194b01ec7062471f85bb002beb388d2d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a3482819330847c799c24e582792d75c : SUCCESS in 2m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecd9b834e7f743f4ab19d996151e2ba4 : SUCCESS in 3m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a85fb7f5751d4dda83025ce35ede67d4 : SUCCESS in 1h 04m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f690fe32efcf4c26b1a2344fb5afc549 : SUCCESS in 1h 55m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/170c81baeb80459d87b149defa7f6721 : SUCCESS in 1h 07m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/521b5bfa9ca745fdb1473f147607c8ee : SUCCESS in 1h 28m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/40e0525cf9984d2c8171675fd07708a7 : FAILURE in 1h 57m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0f3451f64f8741f99facca3eb13cff25 : SUCCESS in 1h 09m 03s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e72008e5ff1b4b5589563bbc361aeaa6 : SUCCESS in 30m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c203b44d4bf248e3a38b8586dc771f40 : SUCCESS in 29m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/015e93e85b7d4122811b7401060696b4 : SUCCESS in 46m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8dcb2fc726224de9a156adfed5527ae8 : SUCCESS in 22m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"98ad240aaa7d0ca59ee4a7a5ec1f6780e4c7eca4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-08 16:46:45.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/834817/4/upstream-check/manila-tempest-plugin-ontap-dhss/a3e364c/ : RETRY_LIMIT in 8m 26s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/834817/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/488b5cd/ : RETRY_LIMIT in 7m 59s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"33ac22f8c0e65262fc423ceb5ee36ad28d921539","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-08 20:32:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"85c7dd20401571768251f0c059751b95cd08885d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-08 20:32:31.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f2763db19da5da57781a29e391c63e2e29011b0f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-08 23:03:40.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/17/834817/4/check/DellEMC_Unity_Manila_dhss_true/23bb040/DellEMC_Unity_Manila_dhss_true/606 : FAILURE in 2h 20m 46s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/17/834817/4/check/DellEMC_Unity_Manila_dhss_false/14a1231/DellEMC_Unity_Manila_dhss_false/633 : SUCCESS in 1h 03m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"bd31bc98cbd2188bf9695801f414b48cea487b55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 23:42:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd24d5d973ff418f89733d51884cd872\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cedbdc1294074e75b5f4e40f4980f8bb : SUCCESS in 1m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53ffd3bcc76246b099b5feb2c2b4d40f : SUCCESS in 2m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0afacc3c8f7c4f27bfa6f46b8bfcd461 : SUCCESS in 1h 30m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c29e5ff99e224675b05ce208ba36acc0 : TIMED_OUT in 3h 02m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4c13870cc1fd4dab8111b0989013ec3f : FAILURE in 1h 47m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/fd9233712b9f425fbf0e7a5662b241f8 : SUCCESS in 1h 00m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0409d36aa28c4543868d729b557d0dd4 : FAILURE in 1h 25m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6a8a887e260d42359d5d7ce00599844f : SUCCESS in 1h 16m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ec6eae22cb814f68bdbb50a1877c4521 : SUCCESS in 22m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cee6d883247c4d18b0882341f25acc42 : SUCCESS in 35m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fdcb6bfc7c114e8cb97211f6db273413 : SUCCESS in 36m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c31c5888a07347ffabf67fabf21bd4ea : SUCCESS in 17m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"653b3ef32bb5f4ce1c8ea6966cf600a1ef57a0ac","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-08 23:51:05.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/834817/4/upstream-check/manila-tempest-plugin-ontap-dhss/c896d85/ : RETRY_LIMIT in 7m 31s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/834817/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/269c350/ : RETRY_LIMIT in 7m 50s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"c1864d975e07577b490efc3edc199c48728fb0f6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-09 14:45:40.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/17/834817/4/check/DellEMC_Unity_Manila_dhss_true/a035cbe/DellEMC_Unity_Manila_dhss_true/612 : FAILURE in 2h 40m 16s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/17/834817/4/check/DellEMC_Unity_Manila_dhss_false/4c1fdb1/DellEMC_Unity_Manila_dhss_false/640 : SUCCESS in 1h 04m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"65265415ffa6652c3078b5bd8a3860295498716c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-11 16:11:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"32ca14a99f95ddc5ca8a55902a02679c32f517e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 18:01:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e5734d83ca44b359719eb1bde2a146b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/22edacacd16d47b190072dba2cccc23e : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29ae3b9951b7424796d30bcb1851b0c4 : SUCCESS in 3m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/57c2a29db01149c7a7e146bc4324bb1b : SUCCESS in 1h 19m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8a2fab3b392947da93a8bc30e6886f7a : SUCCESS in 1h 48m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/13724b8bfd5e428bb1c668392edf2ba8 : SUCCESS in 1h 09m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5fdf7244befd44baa9e88c5ffc76b74b : SUCCESS in 1h 23m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/21a132bc540d494e97062c487808df52 : FAILURE in 1h 45m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fb3b153cda6340259124a9d6cb2df51a : FAILURE in 1h 09m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/33b18ddc2f5d4ebebe1786ff9b4c2426 : SUCCESS in 30m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5795d851a14c46588c73c9deba10d823 : SUCCESS in 40m 35s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/8527aa35a74e4f1b935595433e501ccb : SUCCESS in 1h 08m 49s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/52a8d6119ed747b8adcb0600a547912a : SUCCESS in 1h 00m 43s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/7a57a3105f76461497036e75d69a764f : SUCCESS in 1h 03m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a7890d9d652b4290b6b99e6eb4baae90 : SUCCESS in 42m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/252a11ce7b5940178da2271862de909a : SUCCESS in 22m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4b5830100490de625041fc3634713529f8876ae9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-11 18:41:51.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"32667eb4d6bf25c35f41a7fe196efaec6e583515","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 18:42:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7e720ed8a97b329a5635a101f261e9298d8b5870","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-11 19:34:51.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/834817/4/upstream-check/manila-tempest-plugin-ontap-dhss/e0c83a5/ : FAILURE in 43m 25s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/17/834817/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/cf21297/ : RETRY_LIMIT in 54m 33s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"c8e9306a6f970a901be0c771e78232e2aeebcd30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 19:52:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd54fc4974c54e79a2e6a050792015e3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/566260f02c0e4e5287856c2f9eddd23b : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/266f6734b7de4cd5921489e8da6e809c : SUCCESS in 3m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/19f52c9c33e54cee927360c56821a9ac : SUCCESS in 1h 09m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/53631838bae1409395e7c6366f9008bc : SUCCESS in 34m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2599576eba304e6996dc6ac03fe20427 : SUCCESS in 42m 47s","accounts_in_message":[],"_revision_number":4},{"id":"a44ec7b7d9a10c08a4020c1ef3ea081669d7422d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 19:52:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ce979fe1eabd3276733770296ec46e2f06c03733","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-11 20:57:11.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/17/834817/4/check/DellEMC_Unity_Manila_dhss_true/46cac20/DellEMC_Unity_Manila_dhss_true/613 : FAILURE in 2h 55m 34s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/17/834817/4/check/DellEMC_Unity_Manila_dhss_false/b5ddbb7/DellEMC_Unity_Manila_dhss_false/641 : SUCCESS in 1h 03m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c1f7b9024480ceb2be091f68869d9c003af9577c","revisions":{"46d628d5a9d827a492bfeeeacafb6818081f51d9":{"kind":"REWORK","_number":1,"created":"2022-03-23 07:37:25.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/17/834817/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/834817/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/1"}}},"commit":{"parents":[{"commit":"479af8e71b3ffd7799a89d01bb0c06c304b737a2","subject":"Remove check_public_network_connectivity \u0026 _get_server_port_id_and_ip4 methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/479af8e71b3ffd7799a89d01bb0c06c304b737a2"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-23 07:35:47.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-23 07:35:47.000000000","tz":60},"subject":"Remove check_floating_ip_status \u0026 _associate_floating_ip methods","message":"Remove check_floating_ip_status \u0026 _associate_floating_ip methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _associate_floating_ip\n\t* check_floating_ip_status\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ie17c1d9349e98ce016801bf47c1cb9dba18b51f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/46d628d5a9d827a492bfeeeacafb6818081f51d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/46d628d5a9d827a492bfeeeacafb6818081f51d9"}]},"branch":"refs/heads/master"},"4735497222a897761ed9d133b22abfcc8423b0ff":{"kind":"NO_CHANGE","_number":2,"created":"2022-04-01 09:20:51.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/17/834817/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/834817/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/2"}}},"commit":{"parents":[{"commit":"48457ff77e6405a1e106369cbce57324af49ae08","subject":"Remove check_public_network_connectivity \u0026 _get_server_port_id_and_ip4 methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/48457ff77e6405a1e106369cbce57324af49ae08"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-23 07:35:47.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:20:51.000000000","tz":0},"subject":"Remove check_floating_ip_status \u0026 _associate_floating_ip methods","message":"Remove check_floating_ip_status \u0026 _associate_floating_ip methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _associate_floating_ip\n\t* check_floating_ip_status\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ie17c1d9349e98ce016801bf47c1cb9dba18b51f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4735497222a897761ed9d133b22abfcc8423b0ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4735497222a897761ed9d133b22abfcc8423b0ff"}]},"branch":"refs/heads/master","description":"Rebase"},"7b29a5140642b16ee25da5a1c015dca4d7a04545":{"kind":"NO_CHANGE","_number":3,"created":"2022-04-01 09:24:09.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/17/834817/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/834817/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/3"}}},"commit":{"parents":[{"commit":"f880ce3b694010069f9a1601518eee252e95f59b","subject":"Remove check_public_network_connectivity \u0026 _get_server_port_id_and_ip4 methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f880ce3b694010069f9a1601518eee252e95f59b"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-23 07:35:47.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:24:09.000000000","tz":0},"subject":"Remove check_floating_ip_status \u0026 _associate_floating_ip methods","message":"Remove check_floating_ip_status \u0026 _associate_floating_ip methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _associate_floating_ip\n\t* check_floating_ip_status\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ie17c1d9349e98ce016801bf47c1cb9dba18b51f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7b29a5140642b16ee25da5a1c015dca4d7a04545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7b29a5140642b16ee25da5a1c015dca4d7a04545"}]},"branch":"refs/heads/master","description":"Rebase"},"c1f7b9024480ceb2be091f68869d9c003af9577c":{"kind":"REWORK","_number":4,"created":"2022-06-17 08:11:08.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/17/834817/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/17/834817/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/17/834817/4"}}},"commit":{"parents":[{"commit":"f880ce3b694010069f9a1601518eee252e95f59b","subject":"Remove check_public_network_connectivity \u0026 _get_server_port_id_and_ip4 methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f880ce3b694010069f9a1601518eee252e95f59b"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-23 07:35:47.000000000","tz":60},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-06-17 08:11:00.000000000","tz":0},"subject":"Remove check_floating_ip_status \u0026 _associate_floating_ip methods","message":"Remove check_floating_ip_status \u0026 _associate_floating_ip methods\n\nAs tempest.scenario.manager was announced stable interface\nin Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _associate_floating_ip\n\t* check_floating_ip_status\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\nThe commit also bumps the minimal version of tempest in the\nrequirements to tempest\u003e\u003d27.0.0.\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ie17c1d9349e98ce016801bf47c1cb9dba18b51f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c1f7b9024480ceb2be091f68869d9c003af9577c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c1f7b9024480ceb2be091f68869d9c003af9577c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
