)]}'
{"id":"openstack%2Fmanila-tempest-plugin~834851","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Idd323660b21ea109d5f24efc0edab216723c8dca","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-07-13 13:51:11.000000000","reason":"Change was submitted"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-07-13 13:51:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd323660b21ea109d5f24efc0edab216723c8dca","subject":"Remove check_remote_connectivity related  methods","status":"MERGED","created":"2022-03-23 11:30:10.000000000","updated":"2022-07-13 21:39:55.000000000","submitted":"2022-07-13 13:51:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834851-tempest-scenario-manager-cleanup","meta_rev_id":"66d8da42ac48d04ba19dbb7b47d4292004c1b01b","_number":834851,"virtual_id_number":834851,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-13 13:51:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-07-13 09:06:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-07-12 18:45:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-07-13 09:06:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-23 13:46:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 14:00:18.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 13:06:50.000000000","updated_by":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-04-02 12:50:25.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-05-04 23:18:13.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-07-12 21:39:42.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"66f9cd96da566a603f4f2d3b0606655e238d09b6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-23 11:30:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9edf8406b824108cc3c0bb35dea0418d98c5fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 13:46:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a88edaddc304ab7a3c59b34d7a613f5 : SUCCESS in 5m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5523e2f79b82420bba0aaa7df42351c6 : SUCCESS in 34m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5e020136845e48d39342be38c326ea42 : SUCCESS in 35m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e4107c7c054a4b9392c6505872a49c7e : SUCCESS in 1h 29m 43s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/9cbecf1233c5473bbde1dac58dbeb2d4 : SUCCESS in 1h 25m 58s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/648c2844c7804564ad052e5c4f18288e : SUCCESS in 1h 25m 01s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/71470f79e64a40e6a592ae9c30e7b37c : SUCCESS in 1h 46m 29s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6c11ba75520b455fa9dc7e2fdae9ef9d : SUCCESS in 1h 20m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/92a187f9e7684a3b8e66aa6cf9c17164 : SUCCESS in 1h 20m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/416e9c61566c476bafc534a190028c51 : FAILURE in 2h 08m 31s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bdab7a93d5eb45a1995dcccb81a355a8 : SUCCESS in 1h 04m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/87a40659498347e7a36ab14f580f719b : FAILURE in 2h 10m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/adb3ae54a06844a49913a570db503778 : SUCCESS in 31m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffd607659f8163355d006ce5349a5e7ce30b9bc0","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-23 14:00:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/834851/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/e6d8347/ : FAILURE in 12m 57s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/834851/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/719da68/ : FAILURE in 2m 45s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"341c2347ce7304c4ef232999a6de9e642900d2ba","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-24 10:23:15.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"5b6982b75c3f1ad18cc61de7f1cb776eba268274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:16:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8d75add36382e2efa10e4f0cf3860d50d86dfae8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:18:15.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"40a31276126c0655626dfc1ddc86d2065743f4dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:21:08.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"dcb440ac31a80dd46402a81b252fb57e22864537","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:24:53.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"dc61ab38bb157aec2f78a34c006ee2b6ac086055","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 11:53:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db412b73ef84e37bd17486ad069f34b : SUCCESS in 3m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/687d17032e6a44b597418dff4df92e0c : SUCCESS in 35m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/811e97f072444a86bd3d7e9be4a2f8aa : SUCCESS in 42m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/79f6187152bd4ae4adf814a3eff57631 : SUCCESS in 1h 31m 20s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/4aff5154dc69480e9d331de13f8978a8 : SUCCESS in 1h 21m 48s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/26bc157fe87949079f35cb33cabce9c1 : SUCCESS in 1h 24m 45s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/0c5f06ed90b44c4bafd829ed8e39697c : SUCCESS in 52m 43s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e018cad9968948989a13b0bb2927ab42 : SUCCESS in 1h 04m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6d426ba9b62348bd840f98d617940640 : SUCCESS in 1h 31m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ff25b8d9159240d3a374af1f25c71ffe : FAILURE in 1h 41m 09s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e1346e0e9cc7475c8811705bc7bd2e0f : SUCCESS in 1h 09m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b6424eb43f9d4fe8be97c18599ef9682 : SUCCESS in 2h 23m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2a7b7e1552c04f81b92261a46bc61bcb : SUCCESS in 23m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"03eca5e4cef17cc546099870f0f85704207d4e51","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-01 12:00:01.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/834851/5/upstream-check/manila-cDOT-manila-tempest-plugin-ss/15d2134/ : FAILURE in 2m 40s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/834851/5/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/409afd0/ : FAILURE in 2m 34s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"51234dbe93d4a74eaf43529c3dfef406617ba366","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-02 12:50:25.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/834851/5/check/DellEMC_Unity_Manila_dhss_true/aff0fbd/DellEMC_Unity_Manila_dhss_true/350 : SUCCESS in 58m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/834851/5/check/DellEMC_Unity_Manila_dhss_false/3f3fa80/DellEMC_Unity_Manila_dhss_false/368 : FAILURE in 27m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"714cdc083666ffeedad89714731b3587a8537bcb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-04 23:18:13.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f8598e13e7827fcabc092677341bd5df66be6321","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-08 12:23:57.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"59a268b07fcfd2dfb46e92f58be35cf1bcf47617","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 14:54:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c08f8fc62ba4fdcb8e4091a006a731f : SUCCESS in 4m 18s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/473aeab3a829497b848ba555516a0499 : SUCCESS in 41m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/071db45f8f8e44dc8931b6ba4a3c375a : SUCCESS in 41m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9bcc3e960a0f4f5d99bb2498fe3adb54 : SUCCESS in 1h 13m 42s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/0ae8b38517644f05aad9e4e213d83baa : FAILURE in 1h 05m 54s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3f143440b59c4469bf529f35d9288fd6 : SUCCESS in 1h 30m 23s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/fe6bb8670d4b4ec1a4bdb8b5100a582d : FAILURE in 43m 48s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/554ff8cb688c43cdbe968bef8b3a6660 : FAILURE in 1h 36m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/87c51d895f504f2cb20397f4627c8794 : SUCCESS in 1h 36m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9749a7f9fab141ecadd5c3b36a2af8e3 : FAILURE in 11m 39s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/376667c66f46456fb61a5c6c777cad26 : SUCCESS in 1h 07m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/94a44628271149e8bfcbef06d7cec651 : SUCCESS in 2h 22m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9bd6d5340d8648b5886bac2db076f486 : FAILURE in 31m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2545ed8def631fa8a253bce30ce1e52a1e26cbbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-12 18:45:03.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"265e84cb91b78b197d95439f9648d2085e43441c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 20:53:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c50f36f69d4d4152a59e24d8d91442c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d594b17a5c4467cad0c437566cbc38a : SUCCESS in 7m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c4f0315e06244faeb4854dbb28eee8b1 : SUCCESS in 1h 17m 47s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b235d04d67af49bf837e7fb1c264705e : SUCCESS in 2h 05m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/618e74f40982497e858591d14338b9d9 : FAILURE in 1h 49m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/54331b91940a474185796140d7e9d5dd : SUCCESS in 1h 20m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e2eab349206a4b9d8c53e32c349cc21a : FAILURE in 1h 17m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/37d620a1c28848bbb899c46460aa9c8c : SUCCESS in 1h 13m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/40faf0333a3a45ae96feab79c26f552e : SUCCESS in 39m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0e26779c7e3a480394d609dfb156a9a8 : SUCCESS in 43m 34s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/e70a64bd9fef4e15ad622552f1ba1384 : SUCCESS in 1h 19m 28s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/58cee4ffd6ae46a28adb8b6d189ffab2 : SUCCESS in 1h 05m 03s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/7ed4f4224ae8441088e5bda6cb64c920 : SUCCESS in 1h 11m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f04dd90e919d459f91a32d1b208975b4 : SUCCESS in 49m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3f826607b5b1456080a0f10b8ad25569 : SUCCESS in 32m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3331044e07f3b4050942a586a17b5156e182b9a1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-12 21:39:42.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/834851/6/upstream-check/manila-tempest-plugin-ontap-dhss/7a3ae8b/ : FAILURE in 45m 10s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/834851/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/ca7c64d/ : RETRY_LIMIT in 8m 14s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"3a36307ddd9558ac297098548ff7c7c4d36e7834","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-13 02:21:58.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/834851/6/check/DellEMC_Unity_Manila_dhss_true/0ce7e51/DellEMC_Unity_Manila_dhss_true/622 : FAILURE in 1h 18m 40s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/834851/6/check/DellEMC_Unity_Manila_dhss_false/73e7811/DellEMC_Unity_Manila_dhss_false/651 : SUCCESS in 1h 02m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"2dbeb827794dfcb1d21dcbb2cffe8856d2328cfd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-13 09:06:08.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"86249189837d13db80d8b41776e191483e46e0af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 09:06:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"cd0f858ac76d01a8fb14ba6e87d7800d352fb978","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 10:21:22.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d010ef24796b4b7ea4facf74c38aca29\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e97d0ac55f4278921f7f9f14a7fc2c : SUCCESS in 5m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0305a270f03542e79933ce79a8eb106c : FAILURE in 1h 13m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3ad6c526291944858a8cc12779dbff16 : SUCCESS in 31m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a3f4c5c4b23b4d0e81dae9f03151664a : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":6},{"id":"09ca17ee75915f56230a6429a22fb5644aa71c66","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-13 10:34:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"03c5d3e8d418038637c6f8ef1fb8900b0ec57b1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 12:39:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f6e9b88a91a4b4abd40de4edcc95489\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/270ac88126e4479d804039c08cc30f7d : SUCCESS in 4m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c2e03dc065f84460a91645a773b71f84 : SUCCESS in 1h 11m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/714978e0e5224350a6e50da4953d42c6 : FAILURE in 1h 51m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ff5829224d2d4311bd4130b7559c9685 : SUCCESS in 1h 19m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/475185bf655148c896a9d33cbed39568 : FAILURE in 1h 30m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ee7bebd566f447fd9eb00bf4d96a704b : FAILURE in 2h 04m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/43dc71139e484705a268803ec0a6ca1f : SUCCESS in 1h 07m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/61ef25cbedef43afb28499415a886a51 : SUCCESS in 30m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/445b51ade368432e920fec7bbf0a914c : SUCCESS in 35m 11s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/8a770234120049a3890542c649e731c6 : SUCCESS in 1h 04m 38s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/838d3bb3abea4703a1e6a2ae0b289583 : SUCCESS in 1h 15m 49s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/53e519c9f07b4e72ba6b1614319e5ebd : SUCCESS in 56m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0bf835874f0641b6ad4c104a3d57bcde : SUCCESS in 43m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/467c228ee887433588f965029d26d066 : SUCCESS in 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d06b50a26dcb344a141fa5212739238390ec6ffb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 12:40:28.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"853c64982be61dfe1afbf1bc701e817678d62816","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-13 13:25:55.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/834851/6/upstream-check/manila-tempest-plugin-ontap-dhss/912ffac/ : FAILURE in 45m 27s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/834851/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/129313f/ : RETRY_LIMIT in 8m 15s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"86a4940faf4bd453bc643e7cf35ca0c985090f8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 13:51:10.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e337d868c1cb472dad7e08fd950b4f29\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/651ad5ef5d1042f7883bec7ff53e7925 : SUCCESS in 3m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/40f6cc32d70d45d98eb1874d71e6d5bc : SUCCESS in 1h 03m 26s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/12e9c5ebdf1f4a618440fb5097764ec8 : SUCCESS in 31m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f2903234e8a248b3817ed72e0d46f321 : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":6},{"id":"4c43c7491319f3d4447438c567eaf4c1d178af78","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 13:51:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"66d8da42ac48d04ba19dbb7b47d4292004c1b01b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-13 21:39:55.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/834851/6/check/DellEMC_Unity_Manila_dhss_true/b4e9be0/DellEMC_Unity_Manila_dhss_true/631 : FAILURE in 1h 11m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/834851/6/check/DellEMC_Unity_Manila_dhss_false/be798d9/DellEMC_Unity_Manila_dhss_false/663 : SUCCESS in 59m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"0ec7f43c51520a0627b030fbd099aad042330c20","revisions":{"56fe950d0ee15f02a6da6bdb61beaf4e26106c18":{"kind":"REWORK","_number":1,"created":"2022-03-23 11:30:10.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/51/834851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/834851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/1"}}},"commit":{"parents":[{"commit":"c0b4700a2372b3c0d103b54549eecbd97b1f8642","subject":"Remove _get_router \u0026 _create_router methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0b4700a2372b3c0d103b54549eecbd97b1f8642"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-23 11:28:12.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-23 11:28:12.000000000","tz":60},"subject":"Remove check_remote_connectivity relatd  methods","message":"Remove check_remote_connectivity relatd  methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _check_remote_connectivity\n\t* check_remote_connectivity (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Idd323660b21ea109d5f24efc0edab216723c8dca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/56fe950d0ee15f02a6da6bdb61beaf4e26106c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/56fe950d0ee15f02a6da6bdb61beaf4e26106c18"}]},"branch":"refs/heads/master"},"fde7501f17485d5ae78963ab8f50f95a1fb812df":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-01 09:16:30.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/51/834851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/834851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/2"}}},"commit":{"parents":[{"commit":"c0b4700a2372b3c0d103b54549eecbd97b1f8642","subject":"Remove _get_router \u0026 _create_router methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0b4700a2372b3c0d103b54549eecbd97b1f8642"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-23 11:28:12.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:16:08.000000000","tz":120},"subject":"Remove check_remote_connectivity related  methods","message":"Remove check_remote_connectivity related  methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _check_remote_connectivity\n\t* check_remote_connectivity (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Idd323660b21ea109d5f24efc0edab216723c8dca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fde7501f17485d5ae78963ab8f50f95a1fb812df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fde7501f17485d5ae78963ab8f50f95a1fb812df"}]},"branch":"refs/heads/master"},"0afd7edc9d9490bc1fe70fe6ae22124ca0c8b812":{"kind":"NO_CHANGE","_number":3,"created":"2022-04-01 09:18:15.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/51/834851/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/834851/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/3"}}},"commit":{"parents":[{"commit":"91944c0cdff0e17b2ca9a2bd4e6c991475da03f7","subject":"Remove _get_router \u0026 _create_router methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/91944c0cdff0e17b2ca9a2bd4e6c991475da03f7"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-23 11:28:12.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:18:15.000000000","tz":0},"subject":"Remove check_remote_connectivity related  methods","message":"Remove check_remote_connectivity related  methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _check_remote_connectivity\n\t* check_remote_connectivity (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Idd323660b21ea109d5f24efc0edab216723c8dca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0afd7edc9d9490bc1fe70fe6ae22124ca0c8b812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0afd7edc9d9490bc1fe70fe6ae22124ca0c8b812"}]},"branch":"refs/heads/master","description":"Rebase"},"8f782ac7d62b871aeac551b932548c51a867c30d":{"kind":"NO_CHANGE","_number":4,"created":"2022-04-01 09:21:08.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/51/834851/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/834851/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/4"}}},"commit":{"parents":[{"commit":"efc0254934b9b314409ceb371a283475c360b77c","subject":"Remove _get_router \u0026 _create_router methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/efc0254934b9b314409ceb371a283475c360b77c"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-23 11:28:12.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:21:08.000000000","tz":0},"subject":"Remove check_remote_connectivity related  methods","message":"Remove check_remote_connectivity related  methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _check_remote_connectivity\n\t* check_remote_connectivity (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Idd323660b21ea109d5f24efc0edab216723c8dca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8f782ac7d62b871aeac551b932548c51a867c30d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8f782ac7d62b871aeac551b932548c51a867c30d"}]},"branch":"refs/heads/master","description":"Rebase"},"04c1b0e25a6833e6a3217429296b61325469f0c9":{"kind":"NO_CHANGE","_number":5,"created":"2022-04-01 09:24:53.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/51/834851/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/834851/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/5"}}},"commit":{"parents":[{"commit":"bc623a06905eeb7c814a9ff3eb105142afeff2a0","subject":"Remove _get_router \u0026 _create_router methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/bc623a06905eeb7c814a9ff3eb105142afeff2a0"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-23 11:28:12.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:24:53.000000000","tz":0},"subject":"Remove check_remote_connectivity related  methods","message":"Remove check_remote_connectivity related  methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _check_remote_connectivity\n\t* check_remote_connectivity (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Idd323660b21ea109d5f24efc0edab216723c8dca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/04c1b0e25a6833e6a3217429296b61325469f0c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/04c1b0e25a6833e6a3217429296b61325469f0c9"}]},"branch":"refs/heads/master","description":"Rebase"},"0ec7f43c51520a0627b030fbd099aad042330c20":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-07-12 18:45:03.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/834851/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/834851/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/834851/6"}}},"commit":{"parents":[{"commit":"75791ccf28d094fd43b6573b1003c7c7792ad214","subject":"Remove _get_router \u0026 _create_router methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/75791ccf28d094fd43b6573b1003c7c7792ad214"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-23 11:28:12.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-12 18:45:03.000000000","tz":0},"subject":"Remove check_remote_connectivity related  methods","message":"Remove check_remote_connectivity related  methods\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _check_remote_connectivity\n\t* check_remote_connectivity (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Idd323660b21ea109d5f24efc0edab216723c8dca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ec7f43c51520a0627b030fbd099aad042330c20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0ec7f43c51520a0627b030fbd099aad042330c20"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
