)]}'
{"id":"openstack%2Fmanila-tempest-plugin~835029","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I9b3245be474c3e23622de8e80e391247f2cf866f","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-07-13 21:31:33.000000000","reason":"Change was submitted"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-07-13 21:31:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b3245be474c3e23622de8e80e391247f2cf866f","subject":"Remove _create_security_group \u0026 _create_empty_security_group","status":"MERGED","created":"2022-03-24 10:13:20.000000000","updated":"2022-07-13 21:31:33.000000000","submitted":"2022-07-13 21:31:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835029-tempest-scenario-manager-cleanup","meta_rev_id":"a0ba38c959264e6bc2d405171f69ec10a63d3b8a","_number":835029,"virtual_id_number":835029,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-13 21:31:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-07-13 11:46:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-12 18:45:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-07-13 11:46:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-24 12:17:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-24 12:21:47.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 13:07:44.000000000","updated_by":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-03-24 14:32:58.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 16:15:10.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-05-04 23:18:54.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-07-12 23:31:16.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-13 18:48:48.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"}],"messages":[{"id":"77ad323bdea54bddd1efced696d4bc00a1713a33","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-24 10:13:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0db063f33fee5434f23bb75fc149ab980b4966ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 12:17:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97efdc8dcb524ee28edb82e81e6c338d : SUCCESS in 5m 11s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/014f1b55f554492e9033d9fdeb911425 : SUCCESS in 40m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3c02e7c7c97a41d7919040bf3a2d0693 : SUCCESS in 40m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fce1c8e8b94342afa4500012e9ecbd0b : SUCCESS in 1h 09m 46s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/74ca6a4d077c4519b1d6d30d6fe3974a : SUCCESS in 1h 27m 20s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/0d8cd27939c14aac93c88e8161c47f8e : SUCCESS in 1h 27m 54s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/2e4eec849888474db38048c9b49ec861 : SUCCESS in 1h 33m 20s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/da0015cc28b9442e8d0777c34a22893e : SUCCESS in 1h 31m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0f9c55ed06af40539163a19b39a46d1a : SUCCESS in 1h 29m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2c902f7a054142e2a57eccc7ea7073f3 : FAILURE in 1h 33m 44s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ed56928389b14c398720b6e15098e8a7 : SUCCESS in 54m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/97bafcd0d9c74bbd93fe2a969c65521c : FAILURE in 1h 58m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2a210f2692d34dda9009bddb28a97eb8 : SUCCESS in 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d8188536ac0ff0f007be5323e5add38c6e7c3204","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-24 12:21:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/835029/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/c60dc4d/ : FAILURE in 2m 49s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/835029/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/22901fd/ : FAILURE in 2m 52s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e1a1bc5c18f9704499af2583fe021fb5741f8aea","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-24 12:27:58.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"b15fe3eedf7a2d83ba6030364370098de50dd917","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-24 14:32:58.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/29/835029/1/check/DellEMC_VNX_Manila/375d514/DellEMC_VNX_Manila/195 : FAILURE in 2m 38s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"9f532fd5b6b1c89723700d85869f1b35867d105a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-24 16:15:10.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/29/835029/1/check/DellEMC_Unity_Manila_dhss_true/2b2c8f8/DellEMC_Unity_Manila_dhss_true/322 : FAILURE in 2m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/29/835029/1/check/DellEMC_Unity_Manila_dhss_false/76093f8/DellEMC_Unity_Manila_dhss_false/329 : FAILURE in 3m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"702e89ce2b81e4aee5b6662a45aba373f0ecf946","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-28 19:19:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"120a2ab45c87ec883eed0f14f29eb203ea54d358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 00:09:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbc2f5e6e60449d896c2a82a39f61ab7 : SUCCESS in 5m 22s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/304838aec7c640bb80d4d48ff884f409 : SUCCESS in 41m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5941442026fe41d6a55c4ff99f875b9d : SUCCESS in 39m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8f618b49f5c74c4e905bd8e2973a1de2 : SUCCESS in 1h 19m 04s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c815acb22c874360a1656be7a260dfef : SUCCESS in 1h 11m 48s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3e2868e4641b4885b918a90ee7985928 : SUCCESS in 1h 12m 43s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/e00267761b464b5693caf6c1e2cd2de4 : SUCCESS in 1h 15m 41s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c6b9fd27137c473faed74ed3fa1c7c3a : SUCCESS in 1h 08m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f20c1c2c30344c5ba1ac1e75cec4ffdf : SUCCESS in 1h 34m 38s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/969455c78e7d46aba60d4f63796416b3 : FAILURE in 1h 44m 18s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7370752b522c484d999e4fd19d968bd1 : SUCCESS in 1h 17m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/41e488381a6245499338a975b785e5e8 : SUCCESS in 2h 03m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/53daf5c8cd5d4391b9b97ccf39579c0c : SUCCESS in 26m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5730f04f572c5a6c7f40c97f5ca1436e16010e59","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-29 00:12:06.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/29/835029/1/check/DellEMC_VNX_Manila/c2304bf/DellEMC_VNX_Manila/204 : FAILURE in 2m 22s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"b32aaff33385895066729e29440923fd59959001","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-29 00:12:59.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/835029/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/8a9e725/ : FAILURE in 2m 48s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/835029/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/0931325/ : FAILURE in 2m 43s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"f72c376729cd6cf26106bed556a86c0e20a1b17e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-29 00:13:15.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/29/835029/1/check/DellEMC_Unity_Manila_dhss_true/3beef8c/DellEMC_Unity_Manila_dhss_true/323 : FAILURE in 2m 58s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/29/835029/1/check/DellEMC_Unity_Manila_dhss_false/81da24f/DellEMC_Unity_Manila_dhss_false/330 : FAILURE in 3m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"c5186e9cb6cd8652158368b66d04f5393c2ce882","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-29 16:17:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ad3a795772e39cbdf60a939662f7106b765ef35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:14:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"319c401084d4faeed4041cc97daa20fb087d449e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:16:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c32fde5948e93fb2d99b1dd62ad10988583e2bb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:18:41.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"b92e24820d89bcb01a9316e69d13e5452f1bdd67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:21:28.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"4dd737d545f7b343aa2c84b5bf3a4c56f8aaa63e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:25:06.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"7b2b3a1759a56fe10c9c42e2d4825a5cb5a01333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 12:31:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b884c96416c043d889aac3d690d48f03 : SUCCESS in 3m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/46f7f4c113ab4ab699bf35e38c12179f : SUCCESS in 34m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/779b31728ca144cb9568d640060bca47 : SUCCESS in 38m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8d9f22164a224530879bb5382d0eed36 : SUCCESS in 1h 43m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/eae58f00dd6044399c454b7c9ca55ec1 : SUCCESS in 1h 20m 50s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/93219ad712bf43f79e56bcede6141b1b : SUCCESS in 1h 33m 23s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/8137b65780404c5aa08d4debcb21da91 : SUCCESS in 1h 22m 43s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/964cbcd4be344f6fa330ab97e60e1fc5 : FAILURE in 1h 40m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4e67f6e288894024b50807ba1c17d69b : SUCCESS in 1h 26m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/080483f3211b40818d337b1c2d5991f1 : FAILURE in 1h 45m 59s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/faa7a49f612c4a209a547b24d3a58655 : SUCCESS in 1h 00m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/72000931861147358c57a1555f98a299 : FAILURE in 2h 55m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/11ab49c268ae411e9e08e517dcfd74b8 : SUCCESS in 28m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"964e9a6b6ec57b2cae45fa3313bab9cddb52fec9","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-01 12:34:53.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/835029/6/upstream-check/manila-cDOT-manila-tempest-plugin-ss/ace0cdd/ : FAILURE in 2m 41s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/835029/6/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/4932732/ : FAILURE in 2m 40s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"c2af5cadf5e2a5c7450638b01b1f0184a3465376","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-02 15:46:27.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/29/835029/6/check/DellEMC_Unity_Manila_dhss_true/050f42d/DellEMC_Unity_Manila_dhss_true/353 : SUCCESS in 58m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/29/835029/6/check/DellEMC_Unity_Manila_dhss_false/0a27f99/DellEMC_Unity_Manila_dhss_false/371 : FAILURE in 28m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"ac6b51573021e6d7de2fb8427792d9c9a11f6b43","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-04 23:18:54.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"1de4bf95805422ef2ed7ac17c67b818fa976f26c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-12 18:45:40.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"295572dede63c119bb5cac2d0820a813afcfd1d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 21:34:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/132a37e63a8943aeb7083b3362c3163f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd74924a502e474097d5335764a402d4 : SUCCESS in 3m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/efeb7818f33340ef991f14800fbecdcb : SUCCESS in 1h 28m 28s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/99ec611c24804078bd7209703675b8e8 : FAILURE in 2h 40m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/af0c78d902a044338e237d48b12b987a : SUCCESS in 2h 07m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cafbfe1d5e1b4416b3aa13d90c88a2a5 : FAILURE in 2h 29m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/dd5a16c6d85c49699ebb63935e79b431 : FAILURE in 2h 02m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e1867498a0d54d0a8f3c1baef15515cf : SUCCESS in 1h 08m 44s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/59b900bbf9894b78957fae91e51ded32 : SUCCESS in 34m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cc7688e5da274f3e8ca20b4bd249141e : SUCCESS in 31m 46s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/13685f81898a4dc2982cc8a097d140b2 : SUCCESS in 1h 00m 09s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/fb11910c926746fd8c3553bb025834cc : SUCCESS in 1h 07m 00s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/1657ba6a36f24a4885495c53cdb9b2c2 : SUCCESS in 1h 02m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ad8dca476802414b8f81f6bb2be55b13 : SUCCESS in 49m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b9f5ab9e40094cb785995eee0df07ba3 : SUCCESS in 21m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"76ac12b6194931ca30aace7e87052b6eff0a605a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-12 23:31:16.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/835029/7/upstream-check/manila-tempest-plugin-ontap-dhss/fd80927/ : FAILURE in 1h 18m 49s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/29/835029/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/bab4e25/ : RETRY_LIMIT in 7m 57s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"2b3eb7260aac4c4b4cdff46e28888edc814e4365","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-13 09:03:31.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/29/835029/7/check/DellEMC_Unity_Manila_dhss_true/d013b77/DellEMC_Unity_Manila_dhss_true/625 : FAILURE in 1h 22m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/29/835029/7/check/DellEMC_Unity_Manila_dhss_false/deb8596/DellEMC_Unity_Manila_dhss_false/654 : SUCCESS in 58m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"a2c2d430b51da3451214a3002039913d8716ced1","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-13 11:46:15.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"da17f545ee1b063776ef4b04e3d1715e5b387a28","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-07-13 18:48:48.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9197fffacaea28fb0aabf77a6bb5f450f6f7546d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 19:35:36.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"dc39cf5d4d6b28b7501d82a22d533e1574ce5ba1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 21:31:33.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eff808c813c47d6bd25552814c95741\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcbd886332d64ef0a8922f490f3be076 : SUCCESS in 4m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e970e368d8fb40888c58648f59159f65 : SUCCESS in 1h 54m 47s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3ae6015abe6b49349c3450b4cfb890d7 : SUCCESS in 28m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/acbeb2262fc14805ac68e2558385a866 : SUCCESS in 44m 01s","accounts_in_message":[],"_revision_number":7},{"id":"a0ba38c959264e6bc2d405171f69ec10a63d3b8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 21:31:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6e4d2b90b96ba3f2e8a4f8a851ecbc25616e3c24","revisions":{"36af43f301c3eda8491b7df0b878d433691d1212":{"kind":"REWORK","_number":1,"created":"2022-03-24 10:13:20.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/29/835029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/835029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/1"}}},"commit":{"parents":[{"commit":"7ad8ca68427499ab1e71fcb1a7de8539a37ac03e","subject":"Make _create_loginable_secgroup public","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7ad8ca68427499ab1e71fcb1a7de8539a37ac03e"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:10:45.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:10:45.000000000","tz":60},"subject":"Remove _create_security_group \u0026 _create_empty_security_group","message":"Remove _create_security_group \u0026 _create_empty_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group\n\t* _create_empty_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I9b3245be474c3e23622de8e80e391247f2cf866f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/36af43f301c3eda8491b7df0b878d433691d1212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/36af43f301c3eda8491b7df0b878d433691d1212"}]},"branch":"refs/heads/master"},"119bc84f1d416d7c2bbece50041a5be51d12e86e":{"kind":"NO_CHANGE","_number":2,"created":"2022-04-01 09:14:46.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/29/835029/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/835029/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/2"}}},"commit":{"parents":[{"commit":"19d35e5464477c3ce733d0f711159774160c271a","subject":"Make _create_loginable_secgroup public","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/19d35e5464477c3ce733d0f711159774160c271a"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:10:45.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:14:46.000000000","tz":0},"subject":"Remove _create_security_group \u0026 _create_empty_security_group","message":"Remove _create_security_group \u0026 _create_empty_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group\n\t* _create_empty_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I9b3245be474c3e23622de8e80e391247f2cf866f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/119bc84f1d416d7c2bbece50041a5be51d12e86e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/119bc84f1d416d7c2bbece50041a5be51d12e86e"}]},"branch":"refs/heads/master","description":"Rebase"},"d3d1a30c598cc33ddcada7e6f5a2c5c9ade23754":{"kind":"NO_CHANGE","_number":3,"created":"2022-04-01 09:16:57.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/29/835029/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/835029/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/3"}}},"commit":{"parents":[{"commit":"ce08132f492710b329903159ab0336899d635e42","subject":"Make _create_loginable_secgroup public","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ce08132f492710b329903159ab0336899d635e42"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:10:45.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:16:57.000000000","tz":0},"subject":"Remove _create_security_group \u0026 _create_empty_security_group","message":"Remove _create_security_group \u0026 _create_empty_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group\n\t* _create_empty_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I9b3245be474c3e23622de8e80e391247f2cf866f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d3d1a30c598cc33ddcada7e6f5a2c5c9ade23754"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d3d1a30c598cc33ddcada7e6f5a2c5c9ade23754"}]},"branch":"refs/heads/master","description":"Rebase"},"2c5e82ed409c4aada2ed3111ecf8aa6b953c4563":{"kind":"NO_CHANGE","_number":4,"created":"2022-04-01 09:18:41.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/29/835029/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/835029/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/4"}}},"commit":{"parents":[{"commit":"eeeff4bf22efa3726e689590a4c1525a2b88de70","subject":"Make _create_loginable_secgroup public","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eeeff4bf22efa3726e689590a4c1525a2b88de70"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:10:45.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:18:41.000000000","tz":0},"subject":"Remove _create_security_group \u0026 _create_empty_security_group","message":"Remove _create_security_group \u0026 _create_empty_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group\n\t* _create_empty_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I9b3245be474c3e23622de8e80e391247f2cf866f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2c5e82ed409c4aada2ed3111ecf8aa6b953c4563"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2c5e82ed409c4aada2ed3111ecf8aa6b953c4563"}]},"branch":"refs/heads/master","description":"Rebase"},"4b91ef8bc5c122772f003fce9651679c56dc8622":{"kind":"NO_CHANGE","_number":5,"created":"2022-04-01 09:21:28.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/29/835029/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/835029/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/5"}}},"commit":{"parents":[{"commit":"689110cbca315ac1f300f9e39f5cef831b6c4e87","subject":"Make _create_loginable_secgroup public","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/689110cbca315ac1f300f9e39f5cef831b6c4e87"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:10:45.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:21:28.000000000","tz":0},"subject":"Remove _create_security_group \u0026 _create_empty_security_group","message":"Remove _create_security_group \u0026 _create_empty_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group\n\t* _create_empty_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I9b3245be474c3e23622de8e80e391247f2cf866f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b91ef8bc5c122772f003fce9651679c56dc8622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b91ef8bc5c122772f003fce9651679c56dc8622"}]},"branch":"refs/heads/master","description":"Rebase"},"8f01795fadad696f1aedb9f30ea9980eb1af4f93":{"kind":"NO_CHANGE","_number":6,"created":"2022-04-01 09:25:06.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/29/835029/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/835029/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/6"}}},"commit":{"parents":[{"commit":"e07afa9cfee32f1f2f02f949f3942588218476f5","subject":"Make _create_loginable_secgroup public","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e07afa9cfee32f1f2f02f949f3942588218476f5"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:10:45.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:25:06.000000000","tz":0},"subject":"Remove _create_security_group \u0026 _create_empty_security_group","message":"Remove _create_security_group \u0026 _create_empty_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group\n\t* _create_empty_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I9b3245be474c3e23622de8e80e391247f2cf866f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8f01795fadad696f1aedb9f30ea9980eb1af4f93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8f01795fadad696f1aedb9f30ea9980eb1af4f93"}]},"branch":"refs/heads/master","description":"Rebase"},"6e4d2b90b96ba3f2e8a4f8a851ecbc25616e3c24":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-07-12 18:45:40.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/29/835029/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/29/835029/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/29/835029/7"}}},"commit":{"parents":[{"commit":"22fde6a5882e189db52a5cf3ed63e185ee929955","subject":"Make _create_loginable_secgroup public","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/22fde6a5882e189db52a5cf3ed63e185ee929955"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:10:45.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-12 18:45:40.000000000","tz":0},"subject":"Remove _create_security_group \u0026 _create_empty_security_group","message":"Remove _create_security_group \u0026 _create_empty_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group\n\t* _create_empty_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I9b3245be474c3e23622de8e80e391247f2cf866f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6e4d2b90b96ba3f2e8a4f8a851ecbc25616e3c24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6e4d2b90b96ba3f2e8a4f8a851ecbc25616e3c24"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
