)]}'
{"id":"openstack%2Fmanila-tempest-plugin~835030","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I18dd2a3d532273c8dfff730b4f13922717c0abf7","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-07-13 21:31:38.000000000","reason":"Change was submitted"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-07-13 21:31:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I18dd2a3d532273c8dfff730b4f13922717c0abf7","subject":"Remove _create_security_group_rule \u0026 _default_security_group","status":"MERGED","created":"2022-03-24 10:16:38.000000000","updated":"2022-07-13 21:31:38.000000000","submitted":"2022-07-13 21:31:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835030-tempest-scenario-manager-cleanup","meta_rev_id":"165d7d44b75c9f453d3735dec996e41e027a1828","_number":835030,"virtual_id_number":835030,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-13 21:31:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-07-13 11:40:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-07-12 18:45:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-12 18:45:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-07-13 11:40:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-24 12:02:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-24 12:06:47.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 13:07:53.000000000","updated_by":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-03-24 14:19:07.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 16:06:31.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-05-04 23:19:07.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-05-12 13:21:18.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-07-12 22:11:31.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-13 18:58:19.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"}],"messages":[{"id":"0142d4e836854cb8bf4c65bc5db659a38664860f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-24 10:16:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"079b3e327b77f0e18370a4cce6c0c2c366ca4a23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 12:02:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a9de5c9e29b4d95bfbae50a1008dabf : SUCCESS in 5m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ad4da01f5bbf496b8f47186365505bc4 : SUCCESS in 30m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fe1fb926e51e4318a4c3da6a77387e88 : SUCCESS in 36m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3c6dee630e084712899748b3c46d3b50 : SUCCESS in 1h 11m 33s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/2263f4d0f48e4ef6bf0ab2db52d0dd31 : SUCCESS in 1h 04m 11s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b906c4cb23d944d89adb4f3b1037922c : SUCCESS in 1h 08m 33s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/32515bbc4ebf4f56964eaa23cfa7d5fd : SUCCESS in 1h 14m 24s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/25f2afd4e0804f0197f482670d6160b0 : SUCCESS in 1h 12m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b967800dbb63412ba5bd6a780bbceaa6 : SUCCESS in 1h 38m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1651524405154aa3aab4bf32786257aa : FAILURE in 1h 41m 56s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7f97432d843e47dcb73d4ff34433fd25 : SUCCESS in 55m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cd03ca132c2d49748ca844c0faf816e2 : FAILURE in 1h 09m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8d0eac3a05954c04b1d9c1aadf76234d : SUCCESS in 24m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e730df37c6e50f605850c187c878f6a1dc9d9fa","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-24 12:06:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/30/835030/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/f15a25c/ : FAILURE in 2m 48s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/30/835030/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/ab8969f/ : FAILURE in 1m 07s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"a0eefce7f8209ef685ec666e8a0460da425ed7e0","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-24 12:27:54.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"7cd6cfdb658a9272c8abbace336d06216f512b51","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-24 14:19:07.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/30/835030/1/check/DellEMC_VNX_Manila/c09c166/DellEMC_VNX_Manila/192 : FAILURE in 2h 15m 36s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"555e001dac83733ebbf5f5eda6aa9bcd9bd4ab46","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-24 16:06:31.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/30/835030/1/check/DellEMC_Unity_Manila_dhss_true/99382dc/DellEMC_Unity_Manila_dhss_true/319 : FAILURE in 7m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/30/835030/1/check/DellEMC_Unity_Manila_dhss_false/338b663/DellEMC_Unity_Manila_dhss_false/326 : SUCCESS in 1h 53m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2c4b7365ddc2511c8c5e565e21364b4d85177a09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:11:28.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"90182034e614676fdcc488095a0339a5738a43a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:14:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"042e72d46ae08e882733c7b6b5893a2098630171","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:17:03.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"7f51bd904bb51c6111a5d8802db50e4313a22dee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:18:46.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"ac469078d7de161860c84de00dc01bb97f0ba546","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:21:33.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"ea3905f8d1e6cc045fc35b5f7eabb051f9e35e6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:25:11.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"21a414b9ef9b7c931ff9d078822bb71fd64b0600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 12:02:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdaa6859ecf04c928b5aef9bdde56ce7 : SUCCESS in 3m 54s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8ac6aa18a57c4d57a27abc6d95d7ba21 : SUCCESS in 35m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a43ee38675324cafbe56f0fd9b2e3988 : SUCCESS in 37m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/48ae4a9d2e9f4df08414c984b9929f8c : SUCCESS in 1h 19m 14s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/76c75ee979ee471887a4233453887949 : SUCCESS in 1h 18m 58s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3723851f98814f8aa84ac504e532621d : SUCCESS in 1h 23m 54s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/64245666866e486c8cb30799dfa95b67 : SUCCESS in 1h 56m 17s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/df0211bb0501448e90915fb18106c8f2 : SUCCESS in 1h 28m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/154514d33ec44c63b4366b9f54387ed6 : SUCCESS in 1h 25m 12s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3560e69dd2a04724a176084e08f887e1 : FAILURE in 1h 46m 39s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b3dcaae36edf4968b689939d0f973655 : SUCCESS in 58m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c5dc070790574db8b01007a754d7f800 : SUCCESS in 2h 29m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a279fe13900848a39ebcd8eb8b99d8f9 : SUCCESS in 26m 19s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"941bbf948d87d82e1e3cb1bc86316ddb5430ee80","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-01 12:05:51.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/30/835030/7/upstream-check/manila-cDOT-manila-tempest-plugin-ss/8f00e2e/ : FAILURE in 2m 42s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/30/835030/7/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/a2c6a74/ : FAILURE in 2m 40s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"9b01eabf99e43d08fcfb0f202758334ac64324cd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-02 16:46:31.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/30/835030/7/check/DellEMC_Unity_Manila_dhss_true/13b72a0/DellEMC_Unity_Manila_dhss_true/354 : SUCCESS in 59m 59s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/30/835030/7/check/DellEMC_Unity_Manila_dhss_false/c9d6c08/DellEMC_Unity_Manila_dhss_false/372 : FAILURE in 28m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"0c0a7c117ff569f8e7fc6cdbdd9e472512487cee","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-04 23:19:07.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"30c6e28e6330473a7b7f995f1d4bee30eedc3be7","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-05-12 13:21:18.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"2322b06fba064f55859d3426b92fd4f338165d42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-12 18:45:50.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"bb7a563455a1ace1bd56d7d7449d7bff1d8dd46b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 20:56:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/574e5666192f49e0bafc59a4f1f38b39\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d6ccebd66c415cb493e2d9355342ed : SUCCESS in 6m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/491eb0bf731242be882e8bdac58bc8ee : SUCCESS in 1h 21m 00s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/605fbb590eee4b019f90647cdf8cf1ad : SUCCESS in 2h 07m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/caef78f59f9845839ad2d276d4d41e9a : SUCCESS in 1h 15m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/76bcd570b6b94b5692dfc8d6cd34031c : SUCCESS in 1h 23m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e5fa0250bdee416aa69be3167ef28eb5 : FAILURE in 1h 51m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/286e082d6abb42a2b7b41547f927215f : SUCCESS in 1h 22m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3f5486e65b9f4b0ab20ec47b89668f35 : SUCCESS in 41m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/928a4d286043417ab89ba5a1607ea311 : SUCCESS in 47m 17s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/fbae0f04ae6c4469aa212e718d2816bb : SUCCESS in 1h 25m 24s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5686058b0c20452d9b68473f9cab45d8 : SUCCESS in 1h 15m 44s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d9469644439f4426b46c5a7d429ade26 : SUCCESS in 1h 23m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/53cbe3d5bd1f4714a8aa68534186b203 : SUCCESS in 1h 01m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e1b5f71e5dbe445193fb5e040a309557 : SUCCESS in 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b9c548bfac5dc1b9a9b6af4d895fa5fe9197f563","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-12 22:11:31.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/30/835030/8/upstream-check/manila-tempest-plugin-ontap-dhss/83ba4d8/ : FAILURE in 43m 36s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/30/835030/8/upstream-check/manila-tempest-plugin-ontap-no-dhss/3722cf9/ : RETRY_LIMIT in 7m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"7d1a1462ed59d80743546f91d6337a4e96369f99","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-13 11:22:00.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/30/835030/8/check/DellEMC_Unity_Manila_dhss_true/71c61f0/DellEMC_Unity_Manila_dhss_true/626 : FAILURE in 2h 18m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/30/835030/8/check/DellEMC_Unity_Manila_dhss_false/a64f68d/DellEMC_Unity_Manila_dhss_false/655 : SUCCESS in 1h 04m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"efe16ca347404b3ddda5cb665620d1d618314ff7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-13 11:40:21.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b13a35d9057307f41faaa71acf937681d1691a85","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-07-13 18:58:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6a035fc2c0b3ab299c1a0673765084b00894254c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 19:35:44.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c83cabee169fb2a0380d0ce1dddffe97f04ca4c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 21:09:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59617ce49c3a477e8826132da5cbdef4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b742221ef74da2b02dfe7e6765a782 : SUCCESS in 3m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/782059492fd44041ac5b8ec7a8d5982e : SUCCESS in 1h 20m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d6ffc330b6ca43d9bb583e29a7330705 : SUCCESS in 2h 05m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0a18a66c627f4d9fbf67b52b70211310 : SUCCESS in 1h 22m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/08780abf7ac24dcdbc56f704ebe25dab : SUCCESS in 1h 53m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b55bc01c4eef4c748572e45650fa7250 : FAILURE in 2h 05m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/26fb2e1e74914c56b16b0710dc8d59d5 : POST_FAILURE in 2h 02m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1dac350681544e84aea90998828b3a3a : SUCCESS in 40m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a9ff83c406d44bae803d1c053e5483cd : FAILURE in 34m 24s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/ec66ff88c0e0440d8ee075d68ae441b9 : SUCCESS in 1h 27m 38s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/79841f2cc86845968fc4f63985e278ce : SUCCESS in 54m 37s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/231bd7494d364be9afc20daa135c0b6f : SUCCESS in 1h 17m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9db48554dbfd459fa6f79a12e5b9d596 : SUCCESS in 44m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2555080ea27c4c6c85f22e4440928c16 : SUCCESS in 33m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5a84673b42965c4cae37cc7fd9453933dfc8092e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 21:31:37.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b47cc1cd34cd4a70aef5cb1b2639a91a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3958a2926a741868885cc6c1e3f16f6 : SUCCESS in 6m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ac7f97173f154e3d8efbcb2161ccb2ab : SUCCESS in 53m 16s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4f21f67e86e143d6a7f8d67f871698e5 : SUCCESS in 32m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b7caf5327ff94f539ea92bb3e4018a2b : SUCCESS in 40m 04s","accounts_in_message":[],"_revision_number":8},{"id":"165d7d44b75c9f453d3735dec996e41e027a1828","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 21:31:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"0cf32fb923b70853c158d36dfe1a6866f1ee9d01","revisions":{"40d871cdcb194a8b1fae53f585d3e6f201a7c748":{"kind":"REWORK","_number":1,"created":"2022-03-24 10:16:38.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/30/835030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/835030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/1"}}},"commit":{"parents":[{"commit":"36af43f301c3eda8491b7df0b878d433691d1212","subject":"Remove _create_security_group \u0026 _create_empty_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/36af43f301c3eda8491b7df0b878d433691d1212"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:14:49.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:14:49.000000000","tz":60},"subject":"Remove _create_security_group_rule \u0026 _default_security_group","message":"Remove _create_security_group_rule \u0026 _default_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group_rule\n\t* _default_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I18dd2a3d532273c8dfff730b4f13922717c0abf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40d871cdcb194a8b1fae53f585d3e6f201a7c748"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40d871cdcb194a8b1fae53f585d3e6f201a7c748"}]},"branch":"refs/heads/master"},"756046d01dcb81add2c8f6b1fa64f277dc3639b7":{"kind":"NO_CHANGE","_number":2,"created":"2022-04-01 09:11:28.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/30/835030/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/835030/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/2"}}},"commit":{"parents":[{"commit":"36af43f301c3eda8491b7df0b878d433691d1212","subject":"Remove _create_security_group \u0026 _create_empty_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/36af43f301c3eda8491b7df0b878d433691d1212"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:14:49.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:11:08.000000000","tz":120},"subject":"Remove _create_security_group_rule \u0026 _default_security_group","message":"Remove _create_security_group_rule \u0026 _default_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group_rule\n\t* _default_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I18dd2a3d532273c8dfff730b4f13922717c0abf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/756046d01dcb81add2c8f6b1fa64f277dc3639b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/756046d01dcb81add2c8f6b1fa64f277dc3639b7"}]},"branch":"refs/heads/master"},"4b2ef338005857d50f83446301b53919d79e9be5":{"kind":"NO_CHANGE","_number":3,"created":"2022-04-01 09:14:57.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/30/835030/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/835030/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/3"}}},"commit":{"parents":[{"commit":"119bc84f1d416d7c2bbece50041a5be51d12e86e","subject":"Remove _create_security_group \u0026 _create_empty_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/119bc84f1d416d7c2bbece50041a5be51d12e86e"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:14:49.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:14:57.000000000","tz":0},"subject":"Remove _create_security_group_rule \u0026 _default_security_group","message":"Remove _create_security_group_rule \u0026 _default_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group_rule\n\t* _default_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I18dd2a3d532273c8dfff730b4f13922717c0abf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b2ef338005857d50f83446301b53919d79e9be5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b2ef338005857d50f83446301b53919d79e9be5"}]},"branch":"refs/heads/master","description":"Rebase"},"74bc8888e5a02c962138485749878da3ba2a2624":{"kind":"NO_CHANGE","_number":4,"created":"2022-04-01 09:17:03.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/30/835030/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/835030/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/4"}}},"commit":{"parents":[{"commit":"d3d1a30c598cc33ddcada7e6f5a2c5c9ade23754","subject":"Remove _create_security_group \u0026 _create_empty_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d3d1a30c598cc33ddcada7e6f5a2c5c9ade23754"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:14:49.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:17:03.000000000","tz":0},"subject":"Remove _create_security_group_rule \u0026 _default_security_group","message":"Remove _create_security_group_rule \u0026 _default_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group_rule\n\t* _default_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I18dd2a3d532273c8dfff730b4f13922717c0abf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/74bc8888e5a02c962138485749878da3ba2a2624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/74bc8888e5a02c962138485749878da3ba2a2624"}]},"branch":"refs/heads/master","description":"Rebase"},"19336e9abf4032befb7dca3a26cfe81dddc8843c":{"kind":"NO_CHANGE","_number":5,"created":"2022-04-01 09:18:46.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/30/835030/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/835030/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/5"}}},"commit":{"parents":[{"commit":"2c5e82ed409c4aada2ed3111ecf8aa6b953c4563","subject":"Remove _create_security_group \u0026 _create_empty_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2c5e82ed409c4aada2ed3111ecf8aa6b953c4563"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:14:49.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:18:46.000000000","tz":0},"subject":"Remove _create_security_group_rule \u0026 _default_security_group","message":"Remove _create_security_group_rule \u0026 _default_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group_rule\n\t* _default_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I18dd2a3d532273c8dfff730b4f13922717c0abf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/19336e9abf4032befb7dca3a26cfe81dddc8843c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/19336e9abf4032befb7dca3a26cfe81dddc8843c"}]},"branch":"refs/heads/master","description":"Rebase"},"59bee2f5578463dc02c4913f4e43470f72fd065e":{"kind":"NO_CHANGE","_number":6,"created":"2022-04-01 09:21:33.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/30/835030/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/835030/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/6"}}},"commit":{"parents":[{"commit":"4b91ef8bc5c122772f003fce9651679c56dc8622","subject":"Remove _create_security_group \u0026 _create_empty_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b91ef8bc5c122772f003fce9651679c56dc8622"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:14:49.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:21:33.000000000","tz":0},"subject":"Remove _create_security_group_rule \u0026 _default_security_group","message":"Remove _create_security_group_rule \u0026 _default_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group_rule\n\t* _default_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I18dd2a3d532273c8dfff730b4f13922717c0abf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/59bee2f5578463dc02c4913f4e43470f72fd065e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/59bee2f5578463dc02c4913f4e43470f72fd065e"}]},"branch":"refs/heads/master","description":"Rebase"},"ba9d9e7e8129dce2c9482fd2b4ef734a449f2555":{"kind":"NO_CHANGE","_number":7,"created":"2022-04-01 09:25:11.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/30/835030/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/835030/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/7"}}},"commit":{"parents":[{"commit":"8f01795fadad696f1aedb9f30ea9980eb1af4f93","subject":"Remove _create_security_group \u0026 _create_empty_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8f01795fadad696f1aedb9f30ea9980eb1af4f93"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:14:49.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:25:11.000000000","tz":0},"subject":"Remove _create_security_group_rule \u0026 _default_security_group","message":"Remove _create_security_group_rule \u0026 _default_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group_rule\n\t* _default_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I18dd2a3d532273c8dfff730b4f13922717c0abf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ba9d9e7e8129dce2c9482fd2b4ef734a449f2555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ba9d9e7e8129dce2c9482fd2b4ef734a449f2555"}]},"branch":"refs/heads/master","description":"Rebase"},"0cf32fb923b70853c158d36dfe1a6866f1ee9d01":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-07-12 18:45:50.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/30/835030/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/835030/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/835030/8"}}},"commit":{"parents":[{"commit":"6e4d2b90b96ba3f2e8a4f8a851ecbc25616e3c24","subject":"Remove _create_security_group \u0026 _create_empty_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6e4d2b90b96ba3f2e8a4f8a851ecbc25616e3c24"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:14:49.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-12 18:45:50.000000000","tz":0},"subject":"Remove _create_security_group_rule \u0026 _default_security_group","message":"Remove _create_security_group_rule \u0026 _default_security_group\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* _create_security_group_rule\n\t* _default_security_group\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I18dd2a3d532273c8dfff730b4f13922717c0abf7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0cf32fb923b70853c158d36dfe1a6866f1ee9d01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0cf32fb923b70853c158d36dfe1a6866f1ee9d01"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
