)]}'
{"id":"openstack%2Fmanila-tempest-plugin~835031","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ibb40445426fcf350ca6999558252f288a65a7dbf","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-07-13 21:31:41.000000000","reason":"Change was submitted"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-07-13 21:31:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibb40445426fcf350ca6999558252f288a65a7dbf","subject":"Remove get_networks","status":"MERGED","created":"2022-03-24 10:21:07.000000000","updated":"2022-07-13 21:31:41.000000000","submitted":"2022-07-13 21:31:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835031-tempest-scenario-manager-cleanup","meta_rev_id":"ee323c80a42e0f873ff5bea5e1cc2c5815e94775","_number":835031,"virtual_id_number":835031,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-13 21:31:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-07-13 11:42:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-07-12 18:46:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-12 18:46:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-07-13 11:42:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-24 12:12:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-24 12:16:08.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 13:08:03.000000000","updated_by":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-03-24 14:27:42.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 16:09:24.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-05-04 23:19:15.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-05-23 18:18:33.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-07-12 22:25:29.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-13 18:47:54.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"}],"messages":[{"id":"481d161499f0176fa50d2032fb50113048232c64","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-24 10:21:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb30a38e6f73fcdd70051e311a260b27916e3cbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 12:12:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c944c509d34b4c3083c43990a1590154 : SUCCESS in 5m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/93b63cba422a4465b91f41ea0f86a182 : SUCCESS in 37m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/acc6d877aac54ad7b978392af9a07b36 : SUCCESS in 39m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f1ff3a70adee4e8aaeacfd4a4d9be083 : SUCCESS in 1h 40m 43s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/af329d43875348b88d0c7752b671a18d : SUCCESS in 1h 11m 34s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/fdfaa5653a344a1da2bff94647d15868 : SUCCESS in 1h 18m 43s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/36dd5c20568a430f93429003d22836bf : SUCCESS in 1h 13m 32s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7d6276c87eb24eb9820a405842518acf : SUCCESS in 1h 28m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a57f9acd2ba34840b6578aa61ca464d4 : SUCCESS in 1h 13m 23s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3b422f29a4fd4bd1a2df2c101a99d0a2 : FAILURE in 1h 36m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6cfa73b6099f427f800036b7be5891f2 : SUCCESS in 1h 01m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1837f90d53a64116a1cfb923e03d6848 : FAILURE in 1h 45m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/75d9a71ecb234130a3265ef3e5394797 : SUCCESS in 30m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"761e48f61d72152d65a99715bbb7434732cda452","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-24 12:16:08.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/835031/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/dfe18ed/ : FAILURE in 2m 52s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/835031/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/99608e6/ : FAILURE in 2m 58s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"dcbcc50ad9d880ed2014f6356d10f696ba5ee646","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-03-24 12:27:49.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"fd36c85b77bcd1cf1151b0b155024eec1318fb40","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-03-24 14:27:42.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/31/835031/1/check/DellEMC_VNX_Manila/43be914/DellEMC_VNX_Manila/193 : FAILURE in 8m 30s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"f489ea16e77136ef22cdbb76b8adbde65b71e01e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-24 16:09:24.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/31/835031/1/check/DellEMC_Unity_Manila_dhss_true/0804dd2/DellEMC_Unity_Manila_dhss_true/320 : FAILURE in 2m 52s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/835031/1/check/DellEMC_Unity_Manila_dhss_false/2a558c3/DellEMC_Unity_Manila_dhss_false/327 : FAILURE in 6m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"077dace17c558b8b1fc34336ef2f47b2ea38d365","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:09:43.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fffe61c1505ee7e62ed2c8d10cae7d77625cbafa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:11:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c36b3f88e44a53bc2a92689fee81838eaa358ca3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:15:02.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"26e918252d2a0d1c56b545d249b9e1793c08c57f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:17:08.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"2ceeaee43a79d97400e7d8d4574809483dabf138","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:18:50.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"9fed2b3b58af854020fbf0929f0b939e98ed4655","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:21:37.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"143c05d8562f63a185083db7b91655ec280e1966","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:25:15.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"6fd27da5303489918cbe1a65a9665534d37e98a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 11:52:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69f55b3549004d50b2c109c6d9087824 : SUCCESS in 4m 31s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4077253c7d3d4000977db4a9dd064ae2 : SUCCESS in 34m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/94c4e3f5272c464aa10f0871b0a2898f : SUCCESS in 32m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/96cc7b803e4544a2b115ab0a650866e7 : SUCCESS in 1h 43m 10s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f453d729db8c4165bf14fb6c395bbcd9 : SUCCESS in 57m 16s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c55a4271a64b40fe9a590713671e2bb8 : SUCCESS in 1h 23m 46s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/089d80613eaa4a88bef68303a13c8895 : SUCCESS in 1h 13m 49s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/228237e74b3c43b394c32d018b6283b5 : SUCCESS in 1h 32m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/466b01fe4b6a44b4a2c14135259f2881 : SUCCESS in 1h 32m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f43d06a3de5944efb356246be5c07ca2 : FAILURE in 1h 53m 19s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1d5aea75baa74e30bea2972b0f2e34b2 : SUCCESS in 1h 00m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c70bad445f354350a7c4fd86779bab51 : SUCCESS in 2h 20m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/aa06ec57a53442bf9426b4191a9f14c1 : SUCCESS in 24m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b59f222a52f39909f578d739b31538f2f8359b4d","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-04-01 11:56:10.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/835031/8/upstream-check/manila-cDOT-manila-tempest-plugin-ss/fbb390b/ : FAILURE in 2m 41s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/835031/8/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/80f4056/ : FAILURE in 2m 49s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"60272f77a31039c476ae28691555b8f973fc564f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-02 17:48:04.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/31/835031/8/check/DellEMC_Unity_Manila_dhss_true/7604735/DellEMC_Unity_Manila_dhss_true/355 : SUCCESS in 1h 01m 26s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/835031/8/check/DellEMC_Unity_Manila_dhss_false/0e673e7/DellEMC_Unity_Manila_dhss_false/373 : FAILURE in 28m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"5f8d2c32500464c24d5726c2ed5b61a26dd4af51","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-04 23:19:15.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"6a01879c02323da340c4eb8b1b00cb1e66236ca8","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-05-23 18:18:33.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"36f51be8afde737b74e32c0ab85983107d24112e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-24 21:13:06.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ab2147189eb1c6770f305492ec74dd51e937d2b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-12 18:46:00.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"c9a98bc31c80a17829436fb6b1d369e8f476cf54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 21:07:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90faf099b4b3457a9c0a54579e017361\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b1f2d62800a4ff49ad47583d35ae583 : SUCCESS in 5m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4f341080ae38493483e414307ba0f2fe : SUCCESS in 1h 17m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7ad0e51ad5264ac1a18ccee8b12a32a3 : SUCCESS in 2h 17m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fb372624ff924eddbf65dc9e8bc0fa39 : SUCCESS in 1h 17m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0ccf2b4ddcc945acb8f42b896874de35 : SUCCESS in 1h 28m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d213fb91cafb4d82bd97ebda841ccdb4 : FAILURE in 1h 58m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5d88c07c4f8b43d98ce954152964d02c : SUCCESS in 1h 21m 39s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d535f466c02941fda4974225afc56bf9 : SUCCESS in 39m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/331e6c9d9a3e4f9c93946e1d7f7d0e78 : SUCCESS in 41m 54s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/6154e12b7b1045ca92653c36281dcc99 : SUCCESS in 1h 30m 06s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c34834c2cd6e479b9c4aa24ba7fe5968 : SUCCESS in 1h 19m 41s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3a21973a9b8b4c9690e47c0c2ef431fd : SUCCESS in 1h 17m 09s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/83f45d37432540a9aac2fd2f85028476 : SUCCESS in 1h 05m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8701d33a74a843c399751e262e98c867 : SUCCESS in 30m 04s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"290661fe7ad499e54e9ae766d2d3f820e334801d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-12 22:25:29.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/835031/9/upstream-check/manila-tempest-plugin-ontap-dhss/8734d14/ : FAILURE in 45m 28s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/31/835031/9/upstream-check/manila-tempest-plugin-ontap-no-dhss/5e6f211/ : RETRY_LIMIT in 12m 48s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"5cbd220a981202b43de5461f926595e66e998641","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-13 11:42:39.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e898e7d39ee19f42fe2a36f86dd29757cbd28b09","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-13 14:10:48.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/31/835031/9/check/DellEMC_Unity_Manila_dhss_true/bd6583c/DellEMC_Unity_Manila_dhss_true/627 : FAILURE in 2h 48m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/31/835031/9/check/DellEMC_Unity_Manila_dhss_false/b1147c8/DellEMC_Unity_Manila_dhss_false/656 : SUCCESS in 1h 13m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"062a02935fdd020d3aa37cca5faacbb3830be213","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-07-13 18:47:54.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cd2f549b859afeb7d6ab1bea19dfbac7295424f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 19:35:45.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"7c1f46b2864de4a552bdf0792af03aa14aab824f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 21:31:41.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d4f405c17304a9985f643552956393c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44f095d2dd2e4aa68a55423e7eacdc2a : SUCCESS in 3m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ef74cbbc79bf43c0a329e9122a52491d : SUCCESS in 1h 10m 38s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2b3608367f1d43418eb1d19f5ffa157f : SUCCESS in 34m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2379e094625c47d586c31abcd02ef519 : SUCCESS in 29m 02s","accounts_in_message":[],"_revision_number":9},{"id":"ee323c80a42e0f873ff5bea5e1cc2c5815e94775","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 21:31:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"d4806396e323bd5ad14403a74156841d252604cf","revisions":{"8b1eaad95c77e88c6d6749f6f604b08a62c57e5e":{"kind":"REWORK","_number":1,"created":"2022-03-24 10:21:07.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/31/835031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/835031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/1"}}},"commit":{"parents":[{"commit":"40d871cdcb194a8b1fae53f585d3e6f201a7c748","subject":"Remove _create_security_group_rule \u0026 _default_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40d871cdcb194a8b1fae53f585d3e6f201a7c748"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:20:17.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:20:17.000000000","tz":60},"subject":"Remove get_networks","message":"Remove get_networks\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* get_networks (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ibb40445426fcf350ca6999558252f288a65a7dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8b1eaad95c77e88c6d6749f6f604b08a62c57e5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8b1eaad95c77e88c6d6749f6f604b08a62c57e5e"}]},"branch":"refs/heads/master"},"1578abdbc44805bdacc4adb7c49040a173081360":{"kind":"NO_CHANGE","_number":2,"created":"2022-04-01 09:09:43.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/31/835031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/835031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/2"}}},"commit":{"parents":[{"commit":"40d871cdcb194a8b1fae53f585d3e6f201a7c748","subject":"Remove _create_security_group_rule \u0026 _default_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/40d871cdcb194a8b1fae53f585d3e6f201a7c748"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:20:17.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:09:22.000000000","tz":120},"subject":"Remove get_networks","message":"Remove get_networks\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* get_networks (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ibb40445426fcf350ca6999558252f288a65a7dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1578abdbc44805bdacc4adb7c49040a173081360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1578abdbc44805bdacc4adb7c49040a173081360"}]},"branch":"refs/heads/master"},"a2294ba8c9e309a998304344ec62178265553e49":{"kind":"NO_CHANGE","_number":3,"created":"2022-04-01 09:11:47.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/31/835031/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/835031/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/3"}}},"commit":{"parents":[{"commit":"756046d01dcb81add2c8f6b1fa64f277dc3639b7","subject":"Remove _create_security_group_rule \u0026 _default_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/756046d01dcb81add2c8f6b1fa64f277dc3639b7"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:20:17.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:11:47.000000000","tz":0},"subject":"Remove get_networks","message":"Remove get_networks\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* get_networks (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ibb40445426fcf350ca6999558252f288a65a7dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a2294ba8c9e309a998304344ec62178265553e49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a2294ba8c9e309a998304344ec62178265553e49"}]},"branch":"refs/heads/master","description":"Rebase"},"3d95f7bf2a8b8f06a578f1cb946092a8095ab736":{"kind":"NO_CHANGE","_number":4,"created":"2022-04-01 09:15:02.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/31/835031/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/835031/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/4"}}},"commit":{"parents":[{"commit":"4b2ef338005857d50f83446301b53919d79e9be5","subject":"Remove _create_security_group_rule \u0026 _default_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4b2ef338005857d50f83446301b53919d79e9be5"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:20:17.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:15:02.000000000","tz":0},"subject":"Remove get_networks","message":"Remove get_networks\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* get_networks (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ibb40445426fcf350ca6999558252f288a65a7dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d95f7bf2a8b8f06a578f1cb946092a8095ab736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d95f7bf2a8b8f06a578f1cb946092a8095ab736"}]},"branch":"refs/heads/master","description":"Rebase"},"914efc04d57694431a7506abd603c45243a08ee3":{"kind":"NO_CHANGE","_number":5,"created":"2022-04-01 09:17:08.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/31/835031/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/835031/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/5"}}},"commit":{"parents":[{"commit":"74bc8888e5a02c962138485749878da3ba2a2624","subject":"Remove _create_security_group_rule \u0026 _default_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/74bc8888e5a02c962138485749878da3ba2a2624"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:20:17.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:17:08.000000000","tz":0},"subject":"Remove get_networks","message":"Remove get_networks\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* get_networks (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ibb40445426fcf350ca6999558252f288a65a7dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/914efc04d57694431a7506abd603c45243a08ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/914efc04d57694431a7506abd603c45243a08ee3"}]},"branch":"refs/heads/master","description":"Rebase"},"ef43c1af5272e605e739c065560ba424451f6c04":{"kind":"NO_CHANGE","_number":6,"created":"2022-04-01 09:18:50.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/31/835031/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/835031/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/6"}}},"commit":{"parents":[{"commit":"19336e9abf4032befb7dca3a26cfe81dddc8843c","subject":"Remove _create_security_group_rule \u0026 _default_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/19336e9abf4032befb7dca3a26cfe81dddc8843c"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:20:17.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:18:50.000000000","tz":0},"subject":"Remove get_networks","message":"Remove get_networks\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* get_networks (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ibb40445426fcf350ca6999558252f288a65a7dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef43c1af5272e605e739c065560ba424451f6c04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ef43c1af5272e605e739c065560ba424451f6c04"}]},"branch":"refs/heads/master","description":"Rebase"},"966a57b0360a7070485c4f9fd67ba0cfb64e7bf5":{"kind":"NO_CHANGE","_number":7,"created":"2022-04-01 09:21:37.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/31/835031/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/835031/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/7"}}},"commit":{"parents":[{"commit":"59bee2f5578463dc02c4913f4e43470f72fd065e","subject":"Remove _create_security_group_rule \u0026 _default_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/59bee2f5578463dc02c4913f4e43470f72fd065e"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:20:17.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:21:37.000000000","tz":0},"subject":"Remove get_networks","message":"Remove get_networks\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* get_networks (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ibb40445426fcf350ca6999558252f288a65a7dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/966a57b0360a7070485c4f9fd67ba0cfb64e7bf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/966a57b0360a7070485c4f9fd67ba0cfb64e7bf5"}]},"branch":"refs/heads/master","description":"Rebase"},"4a2a82a87e9a65df1de247527f441977b8b1417d":{"kind":"NO_CHANGE","_number":8,"created":"2022-04-01 09:25:15.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/31/835031/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/835031/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/8"}}},"commit":{"parents":[{"commit":"ba9d9e7e8129dce2c9482fd2b4ef734a449f2555","subject":"Remove _create_security_group_rule \u0026 _default_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ba9d9e7e8129dce2c9482fd2b4ef734a449f2555"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:20:17.000000000","tz":60},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:25:15.000000000","tz":0},"subject":"Remove get_networks","message":"Remove get_networks\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* get_networks (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ibb40445426fcf350ca6999558252f288a65a7dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a2a82a87e9a65df1de247527f441977b8b1417d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a2a82a87e9a65df1de247527f441977b8b1417d"}]},"branch":"refs/heads/master","description":"Rebase"},"d4806396e323bd5ad14403a74156841d252604cf":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-07-12 18:46:00.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/31/835031/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/31/835031/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/31/835031/9"}}},"commit":{"parents":[{"commit":"0cf32fb923b70853c158d36dfe1a6866f1ee9d01","subject":"Remove _create_security_group_rule \u0026 _default_security_group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0cf32fb923b70853c158d36dfe1a6866f1ee9d01"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-03-24 10:20:17.000000000","tz":60},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-12 18:46:00.000000000","tz":0},"subject":"Remove get_networks","message":"Remove get_networks\n\nAs tempest.scenario.manager was announced stable interface in Tempest 27.0.0[1] it can be now reused in plugins.\n\nReplaced methods:\n\t* get_networks (not used)\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Ibb40445426fcf350ca6999558252f288a65a7dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d4806396e323bd5ad14403a74156841d252604cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d4806396e323bd5ad14403a74156841d252604cf"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
