)]}'
{"id":"openstack%2Fmanila-tempest-plugin~835664","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I0e036619049a72e0b0e2c3aa7507bc2e3d17ee91","project":"openstack/manila-tempest-plugin","branch":"master","topic":"ci","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-06-29 14:46:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0e036619049a72e0b0e2c3aa7507bc2e3d17ee91","subject":"[CI] Reduce test concurrency on generic driver job","status":"ABANDONED","created":"2022-03-29 16:27:58.000000000","updated":"2022-06-29 14:46:39.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d1239bdd74d639bc339a336770f4dcac7613d60d","_number":835664,"virtual_id_number":835664,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-30 04:57:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-29 16:31:40.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-03-29 22:36:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-30 05:00:11.000000000","updated_by":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"reviewer":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"f56e5bdf7c08025c423c94ac6ae636c11d547013","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-29 16:27:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7b6870119f92c833853f12b78f1326d314380d7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-29 16:31:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/64/835664/1/check/DellEMC_Unity_Manila_dhss_true/dd29711/DellEMC_Unity_Manila_dhss_true/326 : FAILURE in 2m 56s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/64/835664/1/check/DellEMC_Unity_Manila_dhss_false/9277a5f/DellEMC_Unity_Manila_dhss_false/332 : FAILURE in 3m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"0921982563edc053f8fd7bc4d690d3f4557fa3e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 22:36:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e32d301c18e4a2f890ffc37e39fce53 : SUCCESS in 5m 02s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7ff7f3e516e94db7a7e7022496f1bc56 : SUCCESS in 36m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9027dbc6aecc44b6a0da063a311b32cc : FAILURE in 20m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c99f801f0dbf4df0a386bb7a9fa43660 : SUCCESS in 1h 37m 18s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d88a765b179e48a0b7f914f48a50fbdf : SUCCESS in 1h 15m 58s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9a95b80f73de426fb8ba820026153e52 : SUCCESS in 1h 32m 47s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/037c2b9aea3c47b7ab68403b9d24b842 : SUCCESS in 1h 22m 11s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2d4cbaddc6184ef8b530f78526dc595b : FAILURE in 1h 09m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/fdb96b8f583f4b789fbad0567df58d31 : FAILURE in 21m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7bd9ef397df04c828ff13dc8138628a6 : FAILURE in 17m 56s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8de83a32b2e546cf963cb9817ee6433c : SUCCESS in 1h 00m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f352d4afef5c4f85baaa04deb3edbaf7 : TIMED_OUT in 3h 06m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9a219be615d3475fb70e08b5b59a3c60 : FAILURE in 23m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f06d5bdb36d2c0cc58e28eac7b0f126b3cadcdb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-03-30 01:53:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73805135da23826a477628c35566dce16c0108cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 04:57:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9df090de5ef4dea9b7d39811f7ed531 : SUCCESS in 5m 25s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/38fd40b39bf64fe395bfaa9527cd81b7 : SUCCESS in 30m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8408c54600584a67bb5814eaaa6cf29f : SUCCESS in 31m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/74cd3c2c78334f3e866b9a9ba1465914 : SUCCESS in 1h 15m 24s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c6674dd9d7ee479782870f8ba4292580 : SUCCESS in 1h 18m 51s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e41dc0fdb44244ffa116e18609bd04d4 : SUCCESS in 1h 02m 58s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/bcbca4bc7dfa4780921741df65fdd687 : SUCCESS in 1h 07m 50s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/588153650e374568a5093119266a9ff6 : SUCCESS in 1h 18m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cf792f325efc476f912c43b120af94a4 : SUCCESS in 1h 23m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/da22279cba1e4c8383291016ac3bcd4b : FAILURE in 1h 59m 40s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c2cd22b09d084e319fa17fdd80931d33 : SUCCESS in 56m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5321dc8238944cb0bac57290938489d4 : SUCCESS in 3h 01m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/73efcb6409fd430794a94229b43f528f : SUCCESS in 24m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78e87f028a6b63a7cafbef65efa1990b30cf995e","author":{"_account_id":25243,"name":"NetApp CI","username":"NetApp-ci","inactive":true,"tags":["SERVICE_USER"]},"date":"2022-03-30 05:00:11.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-cDOT-manila-tempest-plugin-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/835664/1/upstream-check/manila-cDOT-manila-tempest-plugin-ss/705e984/ : FAILURE in 2m 49s\n- manila-cDOT-manila-tempest-plugin-no-ss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/835664/1/upstream-check/manila-cDOT-manila-tempest-plugin-no-ss/02028a1/ : FAILURE in 2m 42s\n\nComment \u0027run-NetApp CI\u0027 to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"5e2edf434a326c6ccc78d9ef965544368c4f8f10","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-03-30 07:33:29.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/64/835664/1/check/DellEMC_Unity_Manila_dhss_true/cc2a4d4/DellEMC_Unity_Manila_dhss_true/332 : ABORTED in 9s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/64/835664/1/check/DellEMC_Unity_Manila_dhss_false/59ad8fd/DellEMC_Unity_Manila_dhss_false/336 : SUCCESS in 1h 57m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d1239bdd74d639bc339a336770f4dcac7613d60d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-29 14:46:39.000000000","message":"Abandoned\n\nNo longer necessary; we\u0027ve split the tests into api and scenario jobs","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5dc004612c2c601b5f370b357bc0c50e860c3b7","revisions":{"b5dc004612c2c601b5f370b357bc0c50e860c3b7":{"kind":"REWORK","_number":1,"created":"2022-03-29 16:27:58.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/64/835664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/64/835664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/835664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/835664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/835664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/835664/1"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-03-29 16:25:33.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-03-29 16:25:33.000000000","tz":330},"subject":"[CI] Reduce test concurrency on generic driver job","message":"[CI] Reduce test concurrency on generic driver job\n\nThe job\u0027s currently unstable and when it fails, we\u0027re\nseeing that the issue is with creating nova VMs/generic\ndriver share servers. Let\u0027s run tests single file\nand evaluate if it improves stability\n\nChange-Id: I0e036619049a72e0b0e2c3aa7507bc2e3d17ee91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5dc004612c2c601b5f370b357bc0c50e860c3b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b5dc004612c2c601b5f370b357bc0c50e860c3b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
