)]}'
{"id":"openstack%2Fmanila-tempest-plugin~836130","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I3f63b3a0d4fd60b7824550f5c3deeb8cdb3d9ca9","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-07-13 21:31:44.000000000","reason":"Change was submitted"},"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-07-13 21:31:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f63b3a0d4fd60b7824550f5c3deeb8cdb3d9ca9","subject":"Remove setup_clients","status":"MERGED","created":"2022-04-01 09:39:52.000000000","updated":"2022-07-14 00:08:58.000000000","submitted":"2022-07-13 21:31:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836130-tempest-scenario-manager-cleanup","meta_rev_id":"95befc88c7b457bd73aba3c87cb94563149b995c","_number":836130,"virtual_id_number":836130,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-13 21:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-07-13 11:44:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-07-12 18:46:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-12 18:46:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-07-13 11:44:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-01 11:32:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-02 18:49:21.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-04-04 11:04:13.000000000","updated_by":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-05-04 23:19:35.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-05-12 13:17:25.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-07-12 21:53:37.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-13 18:46:18.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"}],"messages":[{"id":"6c61c8fe2a6aca62b584810ea4621fa5cdd3e7cc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-01 09:39:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66c64105af4d84163d7060c5c383c861fa4e3c15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 11:32:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1301d7f02b7a44db9558525017057cf1 : SUCCESS in 3m 58s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5bb350814a5044d9ad64629f48d77079 : SUCCESS in 30m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/333f1731a80944498432402ed246e5fa : SUCCESS in 30m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/966ea37803864e1a8d45632428b75121 : FAILURE in 49m 31s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5e48d446a9544ca6a798d9746712feb5 : FAILURE in 48m 54s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a78871af8b454de69fec10d97ef9bb04 : FAILURE in 1h 13m 55s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/e79c3271ab7441a29b6b1872f5ff1b62 : FAILURE in 1h 18m 07s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b35e0bf870624bef9a3bddbad7877e98 : FAILURE in 52m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bd6f194846394c22b0eee30715fc5668 : FAILURE in 1h 02m 38s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9c8e7a5edd86433eb798914ffddccd57 : FAILURE in 1h 24m 35s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/928ecff15ab0447d903d77c571812605 : SUCCESS in 51m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/17cdd1ebdfb54e33818f42cd3268fa86 : FAILURE in 1h 49m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ab9bea8b58a24259895cfe7ee7a586b1 : SUCCESS in 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"513b460f23bccab2ed626cdeb5733ac7d5623cf6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-02 18:49:21.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/30/836130/1/check/DellEMC_Unity_Manila_dhss_true/55c729f/DellEMC_Unity_Manila_dhss_true/356 : SUCCESS in 1h 01m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/30/836130/1/check/DellEMC_Unity_Manila_dhss_false/10bebdb/DellEMC_Unity_Manila_dhss_false/374 : FAILURE in 28m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7393b7ae4e1da5e414e3ac34df5f89a4a89328b2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-04 07:14:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"79bca8d27e6d3293787b1ae022eb7cc23273df85","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-04 07:17:41.000000000","message":"Topic manila_cleanup_3 removed","accounts_in_message":[],"_revision_number":2},{"id":"2eea83b4ecfc07dc90fe0aa9c9aad34eda78e7a1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-04 07:17:51.000000000","message":"Topic set to tempest-scenario-manager-cleanup","accounts_in_message":[],"_revision_number":2},{"id":"4dc15a7c4b9dedcb4274f3e48c235653d5a36a77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 09:15:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42cc6e3739344667b36b05a4946e70cc : SUCCESS in 3m 39s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8da9c64f4887488095191321345f8d91 : SUCCESS in 37m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ee3bff8181de4183aef9291fdff40b00 : SUCCESS in 45m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/20fc4842b3784a20a5ed4f1519d606d0 : SUCCESS in 1h 17m 21s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e1dc014957624248bb22d216dd951d06 : SUCCESS in 1h 14m 05s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/822a82dfe0a046f199c910ff0c60de72 : SUCCESS in 1h 07m 30s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/212d8b494f2f49d8b71d6d885e03f192 : SUCCESS in 1h 20m 15s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3c1dfef131004d9591b6b59f5dc86c70 : SUCCESS in 1h 30m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/df286e8d22f74959b0c50b5f3b9df349 : SUCCESS in 1h 50m 29s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c86fc5be7b564d74b103546f73db073c : FAILURE in 1h 52m 26s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9406297083724cf896624649d28e6227 : SUCCESS in 1h 06m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ce2a9a88b11a43ca9ea61d01c3bd6c8a : FAILURE in 1h 14m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/56ec0289419a42599f6785a3d70a2cea : SUCCESS in 29m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"14b23f199cf1e56b2511d2e81d32a5aa94a2b57e","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-04-04 11:04:03.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"05e8c1cbfd123a3fd7bdf2f62106071e4f141bce","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-04 23:19:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"622a124452059d777116a74b21498a3859001f7e","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-05-12 13:17:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"036891a6d4d4368e108e64f6a085dc930c5a2fcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-12 18:46:12.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c11d741e73cdb6830651a964882872bdbef48c4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 20:35:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92e57282675e427bae157bd1c6e12b47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5fccf17911c49b69d7bfc3c80fca19d : SUCCESS in 3m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dea295bf0d9b4af185afe4a634a10d50 : SUCCESS in 1h 19m 38s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/73683cc2c4a94d9bbe687ebe097ab099 : SUCCESS in 1h 42m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/124fd4ec398b439abca2b231f8fae074 : SUCCESS in 1h 20m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c7bcbc689ed24e85b5edb49449e4fef9 : SUCCESS in 1h 48m 13s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/668f8476903244b4b51d4e5056d2c745 : FAILURE in 1h 18m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6900b529d2e54ed0bba37bb3c35a6c2b : SUCCESS in 1h 14m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dd5404646efa4a539d3faba840b9d027 : SUCCESS in 35m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6d57438e4c2e4e3bbbfd6e53b2a13056 : SUCCESS in 40m 48s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/a11a20a71862473db03e4d5943aa9423 : SUCCESS in 1h 34m 50s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/1cad9cb5da2f483daff3098e659739c4 : SUCCESS in 1h 05m 34s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/2c80f39f779e4c92a3095add81d9917b : SUCCESS in 1h 18m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/73944c67458740759ddab0c75c236a51 : SUCCESS in 53m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0ef6444545304e9daf49088aa13bfaa7 : SUCCESS in 25m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"21a65b913bb1108090c7b9e09cbe4e053f62ee5f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-12 21:53:37.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/30/836130/3/upstream-check/manila-tempest-plugin-ontap-dhss/1f2a99e/ : FAILURE in 1h 15m 09s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/30/836130/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/3ec4560/ : RETRY_LIMIT in 8m 09s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"610ba40148ccfc63bb5f8040a222eb47d9120629","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-13 11:44:31.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"285d396a5291794a0dca11c8be6187f2982f8a28","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-13 16:43:59.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/30/836130/3/check/DellEMC_Unity_Manila_dhss_true/4b5efe9/DellEMC_Unity_Manila_dhss_true/628 : FAILURE in 2h 33m 06s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/30/836130/3/check/DellEMC_Unity_Manila_dhss_false/67bb849/DellEMC_Unity_Manila_dhss_false/657 : SUCCESS in 1h 12m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"eff1ccdffc1a2e849dfbed018477a1327ace4ff0","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-07-13 18:46:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7ca29af42c4b179b33b5143cd7b7addf72fe3a32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 19:35:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bcccdd38bd101bb53aa4b6fcc8a541e1528334c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 21:15:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b216fd0e75946d0832f255ef26e4813\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb4d7969ae074ab2a05c71f19d0eb7e9 : SUCCESS in 5m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/35f0a64eac794bf3b2a90e50a3800242 : SUCCESS in 1h 04m 56s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1257413bc7e94bb2ae90588f10fed6d3 : SUCCESS in 2h 23m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4c4b39ac9b55484d85affbcc5994cb42 : SUCCESS in 1h 08m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/05a33a41a48b40e59f75a2623aeb3a44 : SUCCESS in 1h 09m 57s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4be7a644c1444f57830c805c303307fc : FAILURE in 1h 42m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/462a52b51dfe4aad88a8842257af03c0 : SUCCESS in 1h 18m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e3cb85705e9848e0b396e2b23054d1e7 : SUCCESS in 31m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5084cc52e53f46b3b8aa18906dfd79d5 : SUCCESS in 37m 25s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/8e4ac9c04cc349ffb03e991155eea5ae : SUCCESS in 1h 14m 17s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/13730866f3c34e45a9df84681a5be629 : SUCCESS in 1h 05m 22s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/6beeef3d22b44396b1ca5051bfe08186 : SUCCESS in 1h 09m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ffe07636948a4b438cdcba3ec8ebec9d : SUCCESS in 54m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/89847776e37345e0823d76c08f6350e3 : SUCCESS in 26m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ec2a0fb6022137491eab934393e64482954e7a82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 21:31:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d636c9e48d334205b5de10ccf9322b42\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bfb40443fdd496c81824d772e25aec1 : SUCCESS in 3m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7caecb5b835f4b25a9d62e39ff5d3e5e : SUCCESS in 1h 14m 36s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/145d572efd2a4bb1b87d4e030068c892 : SUCCESS in 37m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/80ea2332522d40eb954eb57d401f0d55 : SUCCESS in 39m 35s","accounts_in_message":[],"_revision_number":3},{"id":"3072a29f44a3eef6f0515358533cbcb0b315bdbc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 21:31:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e632098be2d8a4c434ee2f1dabed0a4d30bfa6f8","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-13 22:34:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/30/836130/3/upstream-check/manila-tempest-plugin-ontap-dhss/8b5c311/ : FAILURE in 1h 17m 18s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/30/836130/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/101b5c3/ : RETRY_LIMIT in 13m 17s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"95befc88c7b457bd73aba3c87cb94563149b995c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-14 00:08:58.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/30/836130/3/check/DellEMC_Unity_Manila_dhss_true/a172f08/DellEMC_Unity_Manila_dhss_true/633 : FAILURE in 1h 13m 31s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/30/836130/3/check/DellEMC_Unity_Manila_dhss_false/23fb1e7/DellEMC_Unity_Manila_dhss_false/665 : SUCCESS in 56m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9bd60a982af21851e09c72bd791cdf3830e456a9","revisions":{"26057968e4d87da0c1d72ededa9089947e9a9595":{"kind":"REWORK","_number":1,"created":"2022-04-01 09:39:52.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/30/836130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/836130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/836130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/836130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/836130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/836130/1"}}},"commit":{"parents":[{"commit":"4a2a82a87e9a65df1de247527f441977b8b1417d","subject":"Remove get_networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a2a82a87e9a65df1de247527f441977b8b1417d"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:38:18.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:38:18.000000000","tz":120},"subject":"Remove setup_clients","message":"Remove setup_clients\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\nDeleted methods:\n        * setup_clients\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I3f63b3a0d4fd60b7824550f5c3deeb8cdb3d9ca9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/26057968e4d87da0c1d72ededa9089947e9a9595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/26057968e4d87da0c1d72ededa9089947e9a9595"}]},"branch":"refs/heads/master"},"f33c6a4d61b85653fd810e4e34a13921cde0212e":{"kind":"REWORK","_number":2,"created":"2022-04-04 07:14:41.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/30/836130/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/836130/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/836130/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/836130/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/836130/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/836130/2"}}},"commit":{"parents":[{"commit":"4a2a82a87e9a65df1de247527f441977b8b1417d","subject":"Remove get_networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a2a82a87e9a65df1de247527f441977b8b1417d"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:38:18.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-04 07:14:20.000000000","tz":120},"subject":"Remove setup_clients","message":"Remove setup_clients\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\nDeleted methods:\n        * setup_clients\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I3f63b3a0d4fd60b7824550f5c3deeb8cdb3d9ca9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f33c6a4d61b85653fd810e4e34a13921cde0212e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f33c6a4d61b85653fd810e4e34a13921cde0212e"}]},"branch":"refs/heads/master"},"9bd60a982af21851e09c72bd791cdf3830e456a9":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-07-12 18:46:12.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/30/836130/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/30/836130/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/836130/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/836130/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/836130/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/30/836130/3"}}},"commit":{"parents":[{"commit":"d4806396e323bd5ad14403a74156841d252604cf","subject":"Remove get_networks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d4806396e323bd5ad14403a74156841d252604cf"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-04-01 09:38:18.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-12 18:46:12.000000000","tz":0},"subject":"Remove setup_clients","message":"Remove setup_clients\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\nDeleted methods:\n        * setup_clients\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: I3f63b3a0d4fd60b7824550f5c3deeb8cdb3d9ca9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9bd60a982af21851e09c72bd791cdf3830e456a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9bd60a982af21851e09c72bd791cdf3830e456a9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
