)]}'
{"id":"openstack%2Fmanila-tempest-plugin~836877","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I7b1af957f08328c642a74ff123deb78e193bbe3a","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1968069","attention_set":{},"removed_from_attention_set":{"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2022-05-12 19:53:25.000000000","reason":"Ashley Rodriguez replied on the change"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-06-26 05:58:04.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-06-24 15:58:53.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-06-26 04:43:14.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I7b1af957f08328c642a74ff123deb78e193bbe3a","subject":"Refactor share metadata tests","status":"MERGED","created":"2022-04-06 18:12:53.000000000","updated":"2022-06-26 05:58:04.000000000","submitted":"2022-06-26 05:58:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"836877-bug/1968069","meta_rev_id":"f8ddbf946ceb9303d69df0dec773784a8c4caefb","_number":836877,"virtual_id_number":836877,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-26 05:58:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":2,"date":"2022-06-24 15:58:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-26 04:43:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"all":[{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-26 04:43:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 21:07:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-12 14:28:58.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2022-05-12 14:28:58.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-05-12 14:28:58.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2022-06-22 23:26:03.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-06-23 02:48:14.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"f6fcaed9d1c1c0f4be07f04bae469e501c7990d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-04-06 18:12:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"852f4760f65812cbaa002ede70ae1d2e16ea8bbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-04-06 18:25:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b6507c244dbe6093250085dfb1684251e6da5d2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-04-06 19:25:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c53055c6589e20f1e600610fa46399c28435d3cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 21:07:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ba021314a014eacb9693e6eb24ddbde : SUCCESS in 3m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fa88ea9b03d64e779d554389bfc10870 : FAILURE in 28m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/89c473d92c9e466c9a4470e380d793f2 : FAILURE in 33m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3cc2b3b249c34589ba94b76715231904 : SUCCESS in 59m 54s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/7bc62994a54946b08d5dc237e796880a : SUCCESS in 1h 13m 06s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/240613c756374096ae2b1d0688dcb56d : SUCCESS in 51m 47s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/00583c6785074014ac16493bfd414da5 : SUCCESS in 53m 26s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1a2cb426a2ec45f0b696683a0e2c3034 : SUCCESS in 1h 12m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c8de618735204b9e9f73224f4bb77585 : FAILURE in 1h 19m 36s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b3b197b08e2d48ba99d1c7c829e16865 : FAILURE in 1h 38m 33s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a3eaf2bfb331480da2fa5ca1afe2ebb9 : FAILURE in 45m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/09aa8639ddfe4d629cb11d4a09161de3 : SUCCESS in 1h 40m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/53f489817bcc4413ba5a4a68b89c2187 : FAILURE in 20m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d97aa67a05508f93cce8cee273ecaecf78faac7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-04-12 16:48:02.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0215bc6cfb52415f90c15f4f20a06507b445f563","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 19:35:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1ba28e59f464c9c9180d1e810f1da8d : SUCCESS in 3m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/835136278a8745f4a060f9194be9990a : SUCCESS in 36m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7e8ce234794442a88db487c7a30d2d58 : SUCCESS in 43m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dcf548c71cb648a2ba278a152f4fbe63 : SUCCESS in 1h 02m 21s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e0767488c06d426e92433940f494b18d : SUCCESS in 1h 11m 46s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3fc4c6cbbc394eb2b07e82ed25fd6a8e : SUCCESS in 1h 14m 48s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/dfbbe54283a44157b9ad84b0d0377f48 : SUCCESS in 1h 07m 17s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/937f8d384c864500a8957d1cfc70d253 : SUCCESS in 1h 36m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/478b3dd09ce64283ab87526cf414deb5 : SUCCESS in 1h 41m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7f4220087062477a816030446fc7e724 : FAILURE in 1h 40m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/382e1fd0f10c47b68bce2e7f0677c1a8 : SUCCESS in 1h 10m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3191c675e1294b4cb81776187943cde2 : FAILURE in 2h 33m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4b3dd69df0ff4922b93295e521a38c13 : SUCCESS in 27m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a9b8bc73dd7cda2e33d8892258731af163a1abee","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-05-12 19:53:25.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dffa68bb5bd1aa543c55603f5bd43e91810ea5cb","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-05-12 19:54:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"893d01c933acc312b96ce530c6f260117e3f6204","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 22:00:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c937c877cbd4426b51292332fa7903a : FAILURE in 3m 24s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/14ca2d8728b64518b24560424caf8fd8 : SUCCESS in 22m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/72754554d61a413a8e33ae288055f9d6 : SUCCESS in 40m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/504631a46b354c0ebe14142bb6d5b1d0 : SUCCESS in 1h 22m 46s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e5d07500c58f401a80149578d2dc39bb : SUCCESS in 1h 07m 20s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/133f9f187a1f426bb06785914e57242d : SUCCESS in 1h 02m 46s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/28dab4885fa947fd8097e8b44fa28591 : SUCCESS in 57m 14s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2b8202f00bd94203a6357d253e02b747 : SUCCESS in 1h 26m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/bcfe9d838c0042c7a33c81711ee46199 : SUCCESS in 1h 50m 37s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9f8e0b008d114dfbb06fd63a1756040f : FAILURE in 2h 00m 44s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d08c74dadf744fc9ab209e7d1589e4a2 : SUCCESS in 1h 08m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2f42fcd9718d4f2eb5c645e2a6a3c062 : FAILURE in 1h 22m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/803d4c2c28f145f2835a46880c67cb3d : SUCCESS in 30m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"732275933cf07a55de7dfb699bd95d167e343130","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-25 21:18:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"64c1397344a2173944fbff3866e08b6b04d9984b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 23:16:35.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb4966e017554013b761922710dafe79 : SUCCESS in 3m 03s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bc56d7be17184f5fa10c0cfd0226294d : SUCCESS in 38m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1e45b5f9d26540f0918b72fcb5e2841a : SUCCESS in 28m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0052aa1d77f74e05b069e1d95dd5f991 : SUCCESS in 1h 18m 05s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/11fdabc8fde047578e7e814ddfd261a9 : SUCCESS in 1h 07m 11s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/0b349ce8beab40f6a67b6d31705f779d : FAILURE in 1h 01m 20s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/873271133df54b1b92512ee81ec837e7 : SUCCESS in 1h 22m 51s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2d16f8d406d94aa6b93667c30a7a0e90 : SUCCESS in 1h 20m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b31e1c692d004e5993f6137bd671f80c : FAILURE in 1h 19m 23s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9e0441224ca5429b91ae27f0c73e9ad3 : FAILURE in 1h 48m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ff2c9d6715e2417dabc255ea1812e9fa : SUCCESS in 54m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d991ec1448a046b5955a0aada1d34bb2 : FAILURE in 1h 51m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ac3bac7601794d4c921774efd6e20410 : SUCCESS in 22m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a9b3cc62b4fb36619267fbcdd2d9af20c75cf379","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-22 18:31:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fd0f4c1218764c5fce4da4a38f2131ec926d9c0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 21:01:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e20b7b6775e46dda05b3dc6f7c7889e : SUCCESS in 2m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7ccef02b3fef47fd8e39ad52a9415d7e : SUCCESS in 39m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e4df8adef1ad4b9994e7507f1fe2f632 : SUCCESS in 40m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3e4f2321348a498fb30995763bc76556 : SUCCESS in 1h 10m 19s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/cd2bdc7cb64e4a2ba283af1da0139ff6 : SUCCESS in 1h 26m 19s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d4c24c1c4c164675a29149a259789dd1 : SUCCESS in 1h 13m 35s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/725ff6bd2c1c4aa490c87387470c42fe : SUCCESS in 1h 19m 00s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2d84b95e51b84674a523f76a0585ad4e : FAILURE in 1h 36m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/48ef1d072ddc4fd2b389387f33e10372 : SUCCESS in 1h 27m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b38a960008164ba1a0bbeff4fee08f40 : FAILURE in 2h 05m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ad58bd7ab1f449caaf439ee33c21d2ac : SUCCESS in 56m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9d519e2ffa8f462cb57cd249f246e681 : FAILURE in 1h 49m 48s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2161af3eb4cd4b1ca9108490ae143a83 : SUCCESS in 2h 25m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/85f7a66ee825487caa24a059ccf7b514 : SUCCESS in 23m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d6cb21deb77460acda38b10158ce943ce5e37b08","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-22 23:26:03.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/836877/4/upstream-check/manila-tempest-plugin-ontap-dhss/9aecc73/ : SUCCESS in 1h 32m 56s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/836877/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/984b334/ : RETRY_LIMIT in 11m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"89e3a288c2e26f3825c809b1530c25afa09f41ae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-23 02:48:14.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/77/836877/4/check/DellEMC_Unity_Manila_dhss_true/f245ca1/DellEMC_Unity_Manila_dhss_true/567 : FAILURE in 3h 32m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/836877/4/check/DellEMC_Unity_Manila_dhss_false/55e6755/DellEMC_Unity_Manila_dhss_false/594 : SUCCESS in 1h 00m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"49f4b75aad072092b40318f55fd67bc333fa6b6f","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-06-23 08:05:53.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8d6081a366a4b7b7d942f4fcaf413386d9483f64","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-23 11:27:41.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"74ac02f738f1a2a60ad21f3bd1272fc9a9b57176","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-23 11:31:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"63149008174a0585ab791ac524d63edf5d28d14c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-23 11:40:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3b6f6f28ef0b3c23fc163e3f4808f34149180f18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 13:24:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee1ea8c3c80a44e68092a6cee50cb68e : SUCCESS in 2m 53s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dd15b94198474397b897fc465b5450b4 : SUCCESS in 33m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/679043c03295449c8eebb3226f4f36df : FAILURE in 39m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ff21dab481ca414f86a20bab4864662b : SUCCESS in 1h 06m 33s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3a98c73f63da4c869268da1a5ff24fff : SUCCESS in 1h 23m 56s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c4a3740afd9d41cfafdcb099fc227556 : SUCCESS in 57m 48s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/514808e8662c4dcbb6e5574f5aa1c2da : SUCCESS in 1h 02m 05s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f0af210844f94242b2144264d604bf05 : SUCCESS in 1h 27m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3ebe611d84654b78b939570e11d58994 : SUCCESS in 1h 43m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7f58fab9ee4e4645b90a31dc0a0c021d : FAILURE in 1h 37m 41s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/29bb4cd5d8244004bd1bdd9beb3aa002 : SUCCESS in 1h 10m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/784a4642b6ca4a64b3f0e7de33f1319f : SUCCESS in 1h 17m 40s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b86b13e319a442d8a2e14a98cbb034b3 : SUCCESS in 1h 50m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a7a131f38d3341a4b0af075de60f9ad3 : SUCCESS in 24m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"498e86d4af0fe1a29cb85c95d918df1c7b139c87","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-23 14:21:41.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/77/836877/5/check/DellEMC_Unity_Manila_dhss_true/5cae73e/DellEMC_Unity_Manila_dhss_true/569 : FAILURE in 2h 50m 09s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/836877/5/check/DellEMC_Unity_Manila_dhss_false/a99224b/DellEMC_Unity_Manila_dhss_false/596 : FAILURE in 1h 05m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"8c9f3bb15d6de7e670e26fddc193f688db15ae97","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-23 18:50:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bd3a923c7219301a1e8ae45f0fe78fdc2875ad7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 21:36:24.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82a0d69f3c31482cb61837d8fc791e16 : SUCCESS in 3m 06s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/16a6ed5e9371408f8efa8ebd665dc6f3 : SUCCESS in 31m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f8b0a2ea28934e1b9442bfab970b0faa : SUCCESS in 31m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b164d8b492334c2992e7fe256f543005 : FAILURE in 1h 00m 05s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/fcbc24c7151b46a0b7702432643d3bf6 : SUCCESS in 1h 22m 47s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3b055146243b4133acf3630eb9765fc7 : SUCCESS in 58m 37s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/9146e2dcbda849f6a5ee441d8426a1a7 : SUCCESS in 1h 12m 06s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/aacc0a7f669b458981571caec5c0295c : SUCCESS in 1h 03m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ce105ce789fa47798cb02ea5dc6fc076 : SUCCESS in 1h 12m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f2133c21c532445082dec810e1a1fe45 : FAILURE in 1h 27m 22s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f7cf2fdb9c874479a3e762629e8680ec : SUCCESS in 42m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/04ee7598cada41e4881392dfbfbab028 : SUCCESS in 2h 01m 16s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2965e50771cf4ed3ba686c06feb251f3 : FAILURE in 2h 43m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/23cf436d37cc4978aae682a94016c768 : SUCCESS in 29m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"92b0e1467c99250349d51d0d9aeff7acabf4c8fc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-24 13:04:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0214bc637f91a5e1b530f89f3a892396e7d35926","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 14:55:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/821f70d1f6a74163b1e87dd19ccf80fa : SUCCESS in 3m 41s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d758bcdebcdc4f408c26ee6be1c53c36 : SUCCESS in 30m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2cbf4fe096ae4e2eb21c7a0afa42335d : SUCCESS in 33m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/550d0c42303f4614adf95aafa274a6c0 : SUCCESS in 1h 09m 41s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/d3f3fc7be57c4291b8001150062fb66a : SUCCESS in 1h 12m 26s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d98207bae674484ea58a3916453040af : SUCCESS in 57m 01s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a7c31590d37f492e807b224a4458adde : SUCCESS in 1h 07m 35s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0984169de78941f7b1df012f2e10b6d8 : SUCCESS in 1h 05m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0601e4c210804336a5d9487c041e5ffd : SUCCESS in 1h 25m 23s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6349f130b2224ef1b638f12d808431f5 : FAILURE in 1h 48m 57s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9bedd64356eb4e51ba20230b9a4beb37 : SUCCESS in 50m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/43cf247987524ba59880cfb41b98590d : SUCCESS in 1h 17m 50s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c6c1ab5ee5824a7797d93112e35db844 : SUCCESS in 1h 44m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/53d26c502327401e8dd840a5962e6f0d : SUCCESS in 23m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4c6118009076966e033b9cf514e2ad6588682edd","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-06-24 15:58:53.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"39aec356b397cb36768b61539dd90c477a8a9dbf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-24 17:42:38.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/77/836877/5/check/DellEMC_Unity_Manila_dhss_true/afd94bb/DellEMC_Unity_Manila_dhss_true/579 : FAILURE in 2h 47m 07s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/836877/5/check/DellEMC_Unity_Manila_dhss_false/bade707/DellEMC_Unity_Manila_dhss_false/607 : SUCCESS in 1h 06m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"6203ce527288b5093383dc72054d85b18a67b3dd","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-06-24 18:05:36.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/836877/5/upstream-check/manila-tempest-plugin-ontap-dhss/1cc5c83/ : TIMED_OUT in 3h 08m 55s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/77/836877/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/b85da60/ : RETRY_LIMIT in 11m 11s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"4aa681243106adbbdb8402d18cba29cce35a534e","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-06-26 04:43:14.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"cf80ecd603b79b6d9cd335802f0907d351a373f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-26 04:43:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b021940014dfc10fc6ae9d23d03f0f3270488712","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-26 05:58:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e035a6c498104f49acfb1966c2b47e3e : SUCCESS in 3m 07s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0848f395b3484a0497cd7f9dd8e89c1b : SUCCESS in 26m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c6c97deb56d24201bb465b4e908e5190 : SUCCESS in 34m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/91404c88c39448a78aa93b469c609878 : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":5},{"id":"f8ddbf946ceb9303d69df0dec773784a8c4caefb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-26 05:58:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8bb17fc562bef136ccf2683aaf32e3a1ddb6017c","revisions":{"c4b6f67cdf78682f19fcd55765a15688d86a3c50":{"kind":"REWORK","_number":1,"created":"2022-04-06 18:12:53.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/77/836877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/836877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/1"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-06 18:12:27.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-06 18:12:27.000000000","tz":330},"subject":"Verify key in GET /v2/shares/\u003cshare\u003e/metadata/{key} API","message":"Verify key in GET /v2/shares/\u003cshare\u003e/metadata/{key} API\n\nthe client\u0027s _parse_body helper strips out the top\nlevel key allowing us to miss bugs associated with\nany key changes from the server [1]. Allow for opt-in\nverification of the key and add this verification\nto the affected API.\n\n[1] https://bugs.launchpad.net/manila/+bug/1968069\n\nChange-Id: I7b1af957f08328c642a74ff123deb78e193bbe3a\nRelated-Bug: #1968069\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c4b6f67cdf78682f19fcd55765a15688d86a3c50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c4b6f67cdf78682f19fcd55765a15688d86a3c50"}]},"branch":"refs/heads/master"},"e2fee919dada2d062d1ff898d2c077abcb3a0416":{"kind":"REWORK","_number":2,"created":"2022-04-06 18:25:29.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/77/836877/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/836877/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/2"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-06 18:12:27.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-06 18:24:44.000000000","tz":330},"subject":"Verify key in GET /v2/shares/\u003cshare\u003e/metadata/{key} API","message":"Verify key in GET /v2/shares/\u003cshare\u003e/metadata/{key} API\n\nthe client\u0027s _parse_body helper strips out the top\nlevel key allowing us to miss bugs associated with\nany key changes from the server [1]. Allow for opt-in\nverification of the key and add this verification\nto the affected API.\n\n[1] https://bugs.launchpad.net/manila/+bug/1968069\n\nChange-Id: I7b1af957f08328c642a74ff123deb78e193bbe3a\nRelated-Bug: #1968069\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e2fee919dada2d062d1ff898d2c077abcb3a0416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e2fee919dada2d062d1ff898d2c077abcb3a0416"}]},"branch":"refs/heads/master"},"7c3d713641da98d9c9396e2915b2d7da2ff19dca":{"kind":"REWORK","_number":3,"created":"2022-04-06 19:25:29.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/77/836877/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/836877/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/3"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-06 18:12:27.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-06 19:24:35.000000000","tz":330},"subject":"Refactor share metadata tests","message":"Refactor share metadata tests\n\n1) Use the shares_v2_client to make metadata API requests.\nThis allows us to test v1 proxies as well where appropriate,\nand actually test the supported version of the API instead\nof the deprecated version\n\n2) Verify key in GET /v2/shares/\u003cshare\u003e/metadata/{key} API\n\nthe client\u0027s _parse_body helper strips out the top\nlevel key allowing us to miss bugs associated with\nany key changes from the server [1]. Allow for opt-in\nverification of the key and add this verification\nto the affected API.\n\n[1] https://bugs.launchpad.net/manila/+bug/1968069\n\nChange-Id: I7b1af957f08328c642a74ff123deb78e193bbe3a\nRelated-Bug: #1968069\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7c3d713641da98d9c9396e2915b2d7da2ff19dca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7c3d713641da98d9c9396e2915b2d7da2ff19dca"}]},"branch":"refs/heads/master"},"ff5e780d9b6483ddf8578db46919875bf79f1d70":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-04-12 16:48:02.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/77/836877/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/836877/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/4"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-06 18:12:27.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-12 16:48:02.000000000","tz":0},"subject":"Refactor share metadata tests","message":"Refactor share metadata tests\n\n1) Use the shares_v2_client to make metadata API requests.\nThis allows us to test v1 proxies as well where appropriate,\nand actually test the supported version of the API instead\nof the deprecated version\n\n2) Verify key in GET /v2/shares/\u003cshare\u003e/metadata/{key} API\n\nthe client\u0027s _parse_body helper strips out the top\nlevel key allowing us to miss bugs associated with\nany key changes from the server [1]. Allow for opt-in\nverification of the key and add this verification\nto the affected API.\n\n[1] https://bugs.launchpad.net/manila/+bug/1968069\n\nDepends-On: Ib5a62599b84ba55617071f8bcd8e2a1a11b2537b\nChange-Id: I7b1af957f08328c642a74ff123deb78e193bbe3a\nRelated-Bug: #1968069\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff5e780d9b6483ddf8578db46919875bf79f1d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ff5e780d9b6483ddf8578db46919875bf79f1d70"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8bb17fc562bef136ccf2683aaf32e3a1ddb6017c":{"kind":"REWORK","_number":5,"created":"2022-06-23 11:31:13.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/77/836877/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/77/836877/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/77/836877/5"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-06 18:12:27.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-06-23 11:30:50.000000000","tz":330},"subject":"Refactor share metadata tests","message":"Refactor share metadata tests\n\n1) Use the shares_v2_client to make metadata API requests.\nThis allows us to test v1 proxies as well where appropriate,\nand actually test the supported version of the API instead\nof the deprecated version\n\n2) Verify key in GET /v2/shares/\u003cshare\u003e/metadata/{key} API\n\nthe client\u0027s _parse_body helper strips out the top\nlevel key allowing us to miss bugs associated with\nany key changes from the server [1]. Allow for opt-in\nverification of the key and add this verification\nto the affected API.\n\n[1] https://bugs.launchpad.net/manila/+bug/1968069\n\nDepends-On: Ib5a62599b84ba55617071f8bcd8e2a1a11b2537b\nChange-Id: I7b1af957f08328c642a74ff123deb78e193bbe3a\nRelated-Bug: #1968069\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8bb17fc562bef136ccf2683aaf32e3a1ddb6017c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8bb17fc562bef136ccf2683aaf32e3a1ddb6017c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}]}],"submit_requirements":[]}
