)]}'
{"id":"openstack%2Fmanila-tempest-plugin~839438","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I44f0bf50c3d7042565dca54949965b2c4df1fb41","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1970661","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-09-14 11:46:38.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-09-14 17:45:54.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}},"hashtags":[],"change_id":"I44f0bf50c3d7042565dca54949965b2c4df1fb41","subject":"[CI] Stop suppressing cleanup errors in CephFS jobs","status":"MERGED","created":"2022-04-26 19:51:16.000000000","updated":"2022-09-15 02:49:22.000000000","submitted":"2022-09-14 21:21:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839438-bug/1970661","meta_rev_id":"2c1ba6c566da67ade700775ec3bbbbe6a734dfe9","_number":839438,"virtual_id_number":839438,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-14 21:21:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-09-14 15:25:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-14 15:29:26.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-14 15:29:26.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-26 21:25:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-26 23:48:32.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-09-13 16:16:18.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-09-14 15:29:26.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2022-09-14 15:32:16.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"970fdbff4de3127e421fb6dbbb9407a092ec18f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-04-26 19:51:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8727ea6ab0fa704a774bba66c9ddf1903aea456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 21:25:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/846ca02d2cd64ffeab12ecb7961345d9 : FAILURE in 1h 33m 24s","accounts_in_message":[],"_revision_number":1},{"id":"fa767c73328306d18cca736269c48d87c7048ab8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-26 23:48:32.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/839438/1/check/DellEMC_Unity_Manila_dhss_true/9f17fa9/DellEMC_Unity_Manila_dhss_true/419 : FAILURE in 3h 57m 00s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/839438/1/check/DellEMC_Unity_Manila_dhss_false/e8d7357/DellEMC_Unity_Manila_dhss_false/438 : SUCCESS in 1h 09m 35s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"058847a1a4bd84d046246bd263001ed0958e56d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-13 16:17:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"10340ccf432026612804d8f8543eb069f2facce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 17:35:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ff1c2f7bf8f41a4b587e790b1a3e868\n\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/db2a4d350cb446b7a15f6b67ce60e9c2 : SUCCESS in 1h 16m 49s","accounts_in_message":[],"_revision_number":2},{"id":"fe411fb75cef9707d90439260cf5959ff1649efe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-13 18:56:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"990059fa4bcc40e905e922b9994065441efa7a50","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-13 20:11:31.000000000","message":"Topic dnm/ci removed","accounts_in_message":[],"_revision_number":3},{"id":"8e4175f8cefa00b4d7271edfb6be5837b7f3465d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-13 20:11:33.000000000","message":"Topic set to bug/1970661","accounts_in_message":[],"_revision_number":3},{"id":"9e8ff3340567c88dbbff62c53ba6b3116b7bc2be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 22:07:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8a0b840c2f6493bb97945a178835868\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/517b41e1823a4f1bb04251a3c30d1768 : SUCCESS in 2m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/31ac4b25b9a64ad3bfecf42970a27f7e : SUCCESS in 1h 19m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c2e9a35ba3b34c409a55a92dbbc19905 : FAILURE in 2h 24m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4766f124ca804128a4c8bb8057bad802 : SUCCESS in 2h 18m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7832c5157f364ee6943e05ea619d6ff9 : SUCCESS in 1h 28m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/de0a6d220aa44d2bb8f25cd718ab5d03 : TIMED_OUT in 3h 04m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f21aecfd1afc4a5a9454b25ec6681ff3 : SUCCESS in 1h 11m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b9fc7eebd11344f39a25cdf4da2c7536 : SUCCESS in 34m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/99b3dfe093b04060ac327d4b8313ff17 : FAILURE in 42m 21s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/e3e03922535445ebaf7470bec8287be0 : SUCCESS in 1h 27m 24s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/4c256af7073c40e6bbe7a191d6e8bb78 : SUCCESS in 1h 10m 36s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/b32c9da2953c4f93be9fb5aae8744c59 : SUCCESS in 1h 09m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/79220aa8325940d9a5677f3526281205 : SUCCESS in 56m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/57aaaf963572450b8d9c7078baf239e0 : SUCCESS in 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d8876892804cb1abfdcd26ed8b509da300c8632d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-14 00:32:07.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/839438/3/check/DellEMC_Unity_Manila_dhss_true/a5a0127/DellEMC_Unity_Manila_dhss_true/847 : FAILURE in 1h 18m 11s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/839438/3/check/DellEMC_Unity_Manila_dhss_false/7076f86/DellEMC_Unity_Manila_dhss_false/895 : SUCCESS in 1h 07m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"cb1a61042d90cf646ee329f2a192f2af28a57afe","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-14 05:30:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dd561a98d1836d977b0eb3af2900005582b8b49f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 07:22:55.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/93c8734a943b4a76bdae0e8258fb5b02\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/191d36a520b842538851c44413abcd02 : SUCCESS in 3m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1c1c9d6127324593a013963e5795857d : SUCCESS in 1h 03m 12s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e708881cbeef412aa08eaba80ce348a2 : FAILURE in 1h 50m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/60e953a59cc84cbea6d05e2dd48b9a36 : SUCCESS in 1h 09m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a72940c698114877a67087b993160f5d : SUCCESS in 1h 18m 10s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a9b11af870ba438c9f1e18708867ecac : FAILURE in 1h 15m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ba89f97c854e4a24be439cda2e20871d : SUCCESS in 1h 12m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c47b145ecc2d4459b65b20aa129e5750 : SUCCESS in 27m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/772b1b14dd4a4d85b537a1f0d4a5da3e : SUCCESS in 36m 02s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/919ab358a05e44a5a470f0b7f2fa222c : SUCCESS in 1h 17m 09s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/21c37f581b5044dd87f91ef1cf1410c5 : SUCCESS in 54m 35s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3e93cde441f740bc98dcde2df4206810 : FAILURE in 1h 00m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3af7df308ed44099979254e4d8826d7d : SUCCESS in 54m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/df07bc7dab9d4c10866f9a885540acb0 : SUCCESS in 21m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1d8bd8c1a5912ea23383fe5bd0a7175aeea53b2d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-09-14 11:46:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"937640901da81432a2f1c61d4afd0f28a7d5d8c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 14:03:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a66d7e66ea94b8b9131b570ed92aaf9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa5376b8d8ba4e8f9037c06cca16ab9d : SUCCESS in 3m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b5ea0821ca34440b89e788c25c168776 : SUCCESS in 1h 08m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/52d0d47c0bd04cd48cb1838a7b893845 : SUCCESS in 2h 10m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a9cf0942b7e5415693f8b32c072e1356 : SUCCESS in 1h 15m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/45b6920a092b451cba3b5f1be03d0684 : SUCCESS in 1h 51m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4cf4c3187ce046ea9fece41fdb1ff639 : FAILURE in 1h 58m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/84e89fec6ace4977899e7fdc123d50e2 : SUCCESS in 1h 20m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/73c9a1252af4446f85278d534907fdbd : SUCCESS in 31m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/02548ab157974e88a655625f73eada99 : SUCCESS in 34m 30s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/deac72be3d3848078696aa67e07a15bd : SUCCESS in 1h 15m 11s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/663d110ab4c748ccbc38335c64bc1a92 : SUCCESS in 1h 06m 54s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/4222d81ef751401ba1908da2fa52ddd1 : SUCCESS in 1h 04m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1b40ea6070714bc0b3d271cb98aa61da : SUCCESS in 53m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/767cfc9afe664f518e8497067b2d6a8a : SUCCESS in 19m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"22ac7247a13f171783a91905abd76333b1b43c98","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-09-14 15:25:01.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"719e56aa939c539a6dd012706b207b434a13087b","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-09-14 15:29:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"961ace48ce8e4861fb0ee872022fd5a69029d124","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 15:30:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ebb03e0f1a607d31f7914393e9e04344b57ba393","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-14 15:32:16.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/839438/3/upstream-check/manila-tempest-plugin-ontap-dhss/319eabe/ : FAILURE in 1h 27m 36s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/839438/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/1cae06f/ : SUCCESS in 1h 18m 23s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"0f1c6abe766790405d84633c9b4da84ded9ee22a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-14 16:06:26.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/839438/3/check/DellEMC_Unity_Manila_dhss_true/7c15f2c/DellEMC_Unity_Manila_dhss_true/854 : FAILURE in 1h 16m 31s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/839438/3/check/DellEMC_Unity_Manila_dhss_false/7788e2d/DellEMC_Unity_Manila_dhss_false/903 : SUCCESS in 1h 00m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"c419be277c287cd3b6d3b90ca6703944fac2b9e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 17:23:25.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb300a76a10347b49749bcea5b27ba7d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7234d29d87e42338299e46af1cb0754 : SUCCESS in 3m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/397936d0e4664c72bc52c88021c64216 : FAILURE in 1h 47m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f1b4b99d7b67406ca0535d9842fdf130 : SUCCESS in 43m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fd9287b6ef3c4de6bbd9e8801d58753a : SUCCESS in 43m 31s","accounts_in_message":[],"_revision_number":3},{"id":"d0e26b133a2f26d54a37dc1f25b3b192b0705ed5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-14 17:45:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"100fd172459426a0885828b8a6cbf108c0b4c2df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 20:08:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee74411204b74346a27255c62b5eff9b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6ac4f6047b14e938a2ed43bc501445a : SUCCESS in 2m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/114b04450f7440128b558af4507ed6b8 : SUCCESS in 1h 20m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8ebd628c52ae4f89a7afa85fa699a347 : SUCCESS in 2h 17m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/32ac1b2fd5c74044a58ed6fa0107f632 : SUCCESS in 1h 18m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2560758f960d4791aa4514cf690241a2 : SUCCESS in 1h 15m 45s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/534a4f3908214f10a7a71483f6fe64a8 : FAILURE in 1h 59m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/35e9d24efb5340f9927f67f6872d1ad9 : SUCCESS in 1h 21m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e7af9e4ae5b547d6900836da9bf9fc43 : SUCCESS in 32m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/431294dc3f854aabb258626f458eff73 : SUCCESS in 39m 18s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3348fde5e06942b99f98d8dfc1f5a14d : SUCCESS in 1h 15m 48s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f8f2e151d33c441284587c1ab5fa69ac : SUCCESS in 1h 04m 42s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3cff52f7d5aa4a00b9fb1fbff83b2cb1 : SUCCESS in 1h 06m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4f6175fec4d34ad7ba366af8d3fc3b2a : SUCCESS in 47m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0be8e67cf315494ba095c760c0a3d700 : SUCCESS in 27m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3645ba7f45c57803c3dbfff6407c87a3d10f4908","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 20:08:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"061a608691f60fa2112dbf24d5ad342ec8fea6e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 21:21:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fa3e1d4b6fb455fb6e7c81af556c867\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9e8252e90fa46d0bbdb428d511eb7a4 : SUCCESS in 3m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3e99e3a029f44a3fb19a12e99db051a3 : SUCCESS in 1h 12m 15s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bb36967687b04d42acde29fcaad65d63 : SUCCESS in 33m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f5ff7f928d824641a00b84ed2660c585 : SUCCESS in 36m 53s","accounts_in_message":[],"_revision_number":3},{"id":"6b771bac70b28f8b20020065f9be846526bb10ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 21:21:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"86a0af8732afa5a209f70cd70805da836cc8b979","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-14 22:04:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/839438/3/upstream-check/manila-tempest-plugin-ontap-dhss/db900de/ : SUCCESS in 1h 54m 57s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/38/839438/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/e2e7e3f/ : SUCCESS in 1h 14m 36s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"2c1ba6c566da67ade700775ec3bbbbe6a734dfe9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-15 02:49:22.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/38/839438/3/check/DellEMC_Unity_Manila_dhss_true/d202270/DellEMC_Unity_Manila_dhss_true/857 : FAILURE in 1h 25m 23s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/38/839438/3/check/DellEMC_Unity_Manila_dhss_false/fc91d49/DellEMC_Unity_Manila_dhss_false/907 : SUCCESS in 59m 25s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"79cad927652abfc99415d32b7d92152b3ecd326d","revisions":{"26f310d88635ce2c4b00f6b483a8981823111cc4":{"kind":"REWORK","_number":1,"created":"2022-04-26 19:51:16.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/38/839438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/839438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/839438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/839438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/839438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/839438/1"}}},"commit":{"parents":[{"commit":"c905e77308d48c4582afed88d52bbf3780895383","subject":"Merge \"Remove create_server and rebuild_server methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c905e77308d48c4582afed88d52bbf3780895383"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-26 19:49:58.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-26 19:50:57.000000000","tz":330},"subject":"[DNM] don\u0027t suppress cephfs cleanup errors","message":"[DNM] don\u0027t suppress cephfs cleanup errors\n\nChange-Id: I44f0bf50c3d7042565dca54949965b2c4df1fb41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/26f310d88635ce2c4b00f6b483a8981823111cc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/26f310d88635ce2c4b00f6b483a8981823111cc4"}]},"branch":"refs/heads/master"},"c2a629d96a3696400f62dfdf11ecc85c1dce6811":{"kind":"REWORK","_number":2,"created":"2022-09-13 16:17:00.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/38/839438/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/839438/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/839438/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/839438/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/839438/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/839438/2"}}},"commit":{"parents":[{"commit":"f67893ccccb59151b69719bf44c3a09aae541b11","subject":"Use OVS with IPv6 data path testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f67893ccccb59151b69719bf44c3a09aae541b11"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-04-26 19:49:58.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-09-13 16:16:49.000000000","tz":-420},"subject":"[DNM] don\u0027t suppress cephfs cleanup errors","message":"[DNM] don\u0027t suppress cephfs cleanup errors\n\nChange-Id: I44f0bf50c3d7042565dca54949965b2c4df1fb41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c2a629d96a3696400f62dfdf11ecc85c1dce6811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c2a629d96a3696400f62dfdf11ecc85c1dce6811"}]},"branch":"refs/heads/master"},"79cad927652abfc99415d32b7d92152b3ecd326d":{"kind":"REWORK","_number":3,"created":"2022-09-13 18:56:56.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/38/839438/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/38/839438/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/839438/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/839438/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/839438/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/38/839438/3"}}},"commit":{"parents":[{"commit":"f67893ccccb59151b69719bf44c3a09aae541b11","subject":"Use OVS with IPv6 data path testing jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f67893ccccb59151b69719bf44c3a09aae541b11"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-09-13 18:56:02.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-09-13 18:56:20.000000000","tz":-420},"subject":"[CI] Stop suppressing cleanup errors in CephFS jobs","message":"[CI] Stop suppressing cleanup errors in CephFS jobs\n\nChange-Id: I44f0bf50c3d7042565dca54949965b2c4df1fb41\nCloses-Bug: #1970661\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/79cad927652abfc99415d32b7d92152b3ecd326d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/79cad927652abfc99415d32b7d92152b3ecd326d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}]}],"submit_requirements":[]}
