)]}'
{"id":"openstack%2Fmanila-tempest-plugin~839506","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ia40e014527756efc9e182d97ffc11dffb4924a64","project":"openstack/manila-tempest-plugin","branch":"master","topic":"access_rule","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-05-01 11:39:17.000000000","reason":"Change was marked work in progress"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-05-01 11:39:17.000000000","reason":"Change was marked work in progress"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-05-01 11:39:17.000000000","reason":"Change was marked work in progress"},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2022-05-01 11:39:17.000000000","reason":"Change was marked work in progress"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-05-01 11:39:17.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Ia40e014527756efc9e182d97ffc11dffb4924a64","subject":"Move access rule creation to the test level method","status":"ABANDONED","created":"2022-04-27 09:59:57.000000000","updated":"2022-10-06 07:30:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"858b73aa6556f13c4e80fa23a02119abbcf1b425","_number":839506,"virtual_id_number":839506,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-27 13:14:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 13:14:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 13:16:09.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-27 13:39:14.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-04-27 13:39:14.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2022-04-27 13:39:14.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-04-27 13:39:14.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-04-27 14:56:55.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"0b675b99d9488239512d164c12302f56c01316d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-04-27 09:59:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abc4d82593739cd40e51806d1f38368fe057d453","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-04-27 10:56:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7606111fbab30495b70cac5226619c94f2ab91ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 13:14:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/107c460c879d42e085aa5584bc6b4ed2 : SUCCESS in 3m 08s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c44e658ca36244d0ad09ac5582ccf2f9 : SUCCESS in 37m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d795d861192f4e0796e7f5c3c1a7c721 : SUCCESS in 47m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9993b753492d42b4a7af84d78dfc56bf : SUCCESS in 1h 24m 00s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/b2c7f159570f488ab68bee9f051f6457 : SUCCESS in 1h 11m 34s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/948475f2bc38406b969bf4f54695475c : SUCCESS in 1h 15m 39s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/2f7adec87dc94ce1a3f9dabb71dee7cb : SUCCESS in 1h 20m 57s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c299001bd394434fbad4099c80e09d85 : POST_FAILURE in 1h 38m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/57d629b6786e408b8aedaf3ee8ef331f : FAILURE in 1h 34m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9b00b8be0dad4ead963d60ae08a9a33b : FAILURE in 1h 45m 08s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b22822044f1a47229fa8603f1ba23919 : POST_FAILURE in 59m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1ec31032124047f8bfeccf7767935504 : FAILURE in 2h 12m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f16968c2244a4db8ae18acf04aa4c8e2 : SUCCESS in 32m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a57018e484a88a644493c911f8e042a271ef9ff","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-04-27 13:16:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/06/839506/2/check/DellEMC_VNX_Manila/1895753/DellEMC_VNX_Manila/229 : FAILURE in 1m 25s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"2ae663db90e5eba291f6de23aef2493db4449d9b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-04-27 14:56:55.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/06/839506/2/check/DellEMC_Unity_Manila_dhss_true/d10273e/DellEMC_Unity_Manila_dhss_true/422 : FAILURE in 3h 59m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/839506/2/check/DellEMC_Unity_Manila_dhss_false/6807806/DellEMC_Unity_Manila_dhss_false/441 : SUCCESS in 1h 07m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"00c952fcb146984303299fee5fc9562c1cad9621","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-05-01 11:39:17.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"858b73aa6556f13c4e80fa23a02119abbcf1b425","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-06 07:30:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0689fa835b7a0fb304171c53f3aacdd04f4d2a0e","revisions":{"84ce8d78530935f5598ed732368c02f3de5b6ca5":{"kind":"REWORK","_number":1,"created":"2022-04-27 09:59:57.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/06/839506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/839506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/839506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/839506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/839506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/839506/1"}}},"commit":{"parents":[{"commit":"c905e77308d48c4582afed88d52bbf3780895383","subject":"Merge \"Remove create_server and rebuild_server methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c905e77308d48c4582afed88d52bbf3780895383"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-04-27 09:46:17.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-04-27 09:46:17.000000000","tz":180},"subject":"Move access rule creation to the test level method","message":"Move access rule creation to the test level method\n\nThe access rule is currently being created in the level of the\nclass in \"resource_setup\" method, however, only one single test\nuses it.\nThis causes to failure with the deletion of the share, the share\nstuck in \"error_deleting\" state.\nMoving the creation of the access rule to the test level prevents\nthis issue.\n\nChange-Id: Ia40e014527756efc9e182d97ffc11dffb4924a64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/84ce8d78530935f5598ed732368c02f3de5b6ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/84ce8d78530935f5598ed732368c02f3de5b6ca5"}]},"branch":"refs/heads/master"},"0689fa835b7a0fb304171c53f3aacdd04f4d2a0e":{"kind":"REWORK","_number":2,"created":"2022-04-27 10:56:45.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/06/839506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/06/839506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/839506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/839506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/839506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/06/839506/2"}}},"commit":{"parents":[{"commit":"c905e77308d48c4582afed88d52bbf3780895383","subject":"Merge \"Remove create_server and rebuild_server methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c905e77308d48c4582afed88d52bbf3780895383"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-04-27 09:46:17.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-04-27 10:56:35.000000000","tz":180},"subject":"Move access rule creation to the test level method","message":"Move access rule creation to the test level method\n\nThe access rule is currently being created in the level of the\nclass in \"resource_setup\" method, however, only one single test\nuses it.\nThis causes to failure with the deletion of the share, the share\nstuck in \"error_deleting\" state.\nMoving the creation of the access rule to the test level prevents\nthis issue.\n\nChange-Id: Ia40e014527756efc9e182d97ffc11dffb4924a64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0689fa835b7a0fb304171c53f3aacdd04f4d2a0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0689fa835b7a0fb304171c53f3aacdd04f4d2a0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
