)]}'
{"id":"openstack%2Fmanila-tempest-plugin~840282","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I8fce553d927da6d94471ecdf4040bbbb4377d9cf","project":"openstack/manila-tempest-plugin","branch":"master","topic":"ordered-ddt-data","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-05-03 13:06:01.000000000","reason":"Goutham Pacha Ravi replied on the change"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-05-03 14:53:13.000000000","reason":"Change was submitted"},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2022-05-03 14:53:13.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-05-03 14:53:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8fce553d927da6d94471ecdf4040bbbb4377d9cf","subject":"Remove unneeded randomness when calling ddt.data()","status":"MERGED","created":"2022-05-03 11:17:23.000000000","updated":"2022-05-03 14:57:28.000000000","submitted":"2022-05-03 14:53:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840282-ordered-ddt-data","meta_rev_id":"8f9daff7b5cd7352f3a4928cf8d31f15b5724d92","_number":840282,"virtual_id_number":840282,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-03 14:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-05-03 13:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-05-03 13:06:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-05-03 13:34:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-03 12:06:53.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-05-03 12:06:53.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2022-05-03 12:06:53.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-05-03 13:19:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-03 13:20:35.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-03 13:34:43.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-05-03 14:57:28.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"0a15429a2cb8c10f2dc218606295997b84212e2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-05-03 11:17:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"509424bdcb8226b24f8a31b4e60dda45b17d16ca","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-03 13:06:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8515d7ba8c0f242093baa52883a0de3b56af1184","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 13:19:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d421694dd83496781a906e7ecb83960 : SUCCESS in 3m 09s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ea9453bf925340a595a700ef1732630c : SUCCESS in 27m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8af256be16b2476a88b7e8e02053c124 : SUCCESS in 39m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5dc1d48938504323abdfad43444c688f : SUCCESS in 1h 37m 49s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c4821abac8cd49688a63cb44bb124997 : SUCCESS in 1h 01m 49s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/38097d7304da45bd849a0ff50d76007e : SUCCESS in 1h 12m 23s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/5e7ae37d084d463f9172148770fe6036 : SUCCESS in 55m 48s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/86c8944afd344d52960b91063434b650 : SUCCESS in 1h 05m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f033a7eab576488598f125a0ea850341 : SUCCESS in 1h 13m 12s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ca618a608c4647a993a0380fc59504ce : FAILURE in 1h 32m 31s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e1a3a1bffefa47c2afee154c6c8096e2 : SUCCESS in 54m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/36fbca2b7f4a4d4a8b7b7b2507dd80cd : SUCCESS in 2h 00m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/249e46bca3464ada8c736b3307fe17fe : SUCCESS in 22m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2b57e5858b2eb942b766e30f7c7fdd337147f03","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-03 13:20:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/82/840282/1/check/DellEMC_VNX_Manila/1e296a1/DellEMC_VNX_Manila/239 : FAILURE in 1m 07s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"de8be83f6a56d737eb5c8f913c1d9132dc0f7cc3","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-05-03 13:34:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54c51a71e8f93df51f6ca33c83111c30495814a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 13:35:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c4d791b9cf902bfb9822f3977fe4da1b6640eca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 14:53:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51a02ba05b154841a0ff27b418ab3667 : SUCCESS in 2m 55s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0f8077eaa5f54551a10357ccfa48afb9 : SUCCESS in 28m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7205d7b79f6d48379709981bfa0df075 : SUCCESS in 30m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8af8bca8da364bdc9e9889a9c93461f2 : SUCCESS in 1h 13m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ff553654ed569b74cc5c7a4b58ee7ab494d78450","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 14:53:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8f9daff7b5cd7352f3a4928cf8d31f15b5724d92","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-03 14:57:28.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/82/840282/1/check/DellEMC_Unity_Manila_dhss_true/3c24a2b/DellEMC_Unity_Manila_dhss_true/438 : FAILURE in 3h 39m 51s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/82/840282/1/check/DellEMC_Unity_Manila_dhss_false/80607e4/DellEMC_Unity_Manila_dhss_false/457 : SUCCESS in 1h 15m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6efb2692e6398a7f1bc919d90e9618f716351787","revisions":{"6efb2692e6398a7f1bc919d90e9618f716351787":{"kind":"REWORK","_number":1,"created":"2022-05-03 11:17:23.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/82/840282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/82/840282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/840282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/840282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/840282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/82/840282/1"}}},"commit":{"parents":[{"commit":"c905e77308d48c4582afed88d52bbf3780895383","subject":"Merge \"Remove create_server and rebuild_server methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c905e77308d48c4582afed88d52bbf3780895383"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2022-05-03 11:16:53.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2022-05-03 11:16:53.000000000","tz":120},"subject":"Remove unneeded randomness when calling ddt.data()","message":"Remove unneeded randomness when calling ddt.data()\n\nThe usage of sets introduces unneeded randomness, and the\nutils.deduplicate function can be used instead.\n\nThis is a follow-up to similar past changes, see:\nI9cbd26016238c25487ac8104c1188cd2cf4f467e\nI15bfbf6e1ca604f2bf2fcf12dcd4e67bc728fe5e\n\nChange-Id: I8fce553d927da6d94471ecdf4040bbbb4377d9cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6efb2692e6398a7f1bc919d90e9618f716351787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6efb2692e6398a7f1bc919d90e9618f716351787"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
