)]}'
{"id":"openstack%2Fmanila-tempest-plugin~840727","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Idd49e22c8dc534a1fe1e4814f233b079cf14bb72","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1916102","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-07-15 13:28:58.000000000","reason":"Change was submitted"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-07-15 13:28:58.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-05-07 07:29:01.000000000","reason":"Goutham Pacha Ravi replied on the change"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-07-15 13:28:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd49e22c8dc534a1fe1e4814f233b079cf14bb72","subject":"Modify security service list test","status":"MERGED","created":"2022-05-05 19:34:53.000000000","updated":"2022-07-15 13:28:58.000000000","submitted":"2022-07-15 13:28:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"840727-bug/1916102","meta_rev_id":"5e59c0ec76a42853ba50ca18759b9b60d4473131","_number":840727,"virtual_id_number":840727,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 13:28:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-05-07 07:29:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-15 12:05:25.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-15 12:05:25.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 21:30:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 21:31:37.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-06 00:01:11.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-05-06 13:01:21.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-05-06 13:01:21.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-05-06 13:01:21.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-07-15 12:05:25.000000000","updated_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"}],"messages":[{"id":"b6606e09d1abd82f15078700eb91dd140afe8b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-05-05 19:34:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff4e1351e963a64d7286bef92732bb94d4386def","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 21:30:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7898cec4c26459da17d31d6c447173f : SUCCESS in 2m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fc141cc304224b859f0ac10e50c318fc : SUCCESS in 32m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d8ec81206b5a4de3b71b22b4d6898a31 : SUCCESS in 30m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0d5cf6b930e746f0bc6e26b39d25258f : SUCCESS in 1h 16m 05s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/9d9950babf3048bc9755df046254de12 : SUCCESS in 1h 04m 30s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/52940353ffbb46de80f29114dded05dd : SUCCESS in 1h 05m 48s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/fc57f48c29df4c0c8d50ce882218b87f : SUCCESS in 1h 00m 05s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/74b1f4ef3ba3456488ec91d3d9dbac9b : SUCCESS in 1h 09m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1956245dd833410aae0389ded79b23d5 : SUCCESS in 1h 21m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/38428a8aeff247ad9619a6fd7e185674 : FAILURE in 1h 40m 40s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cf33ab0c0c354d2ebb7014dd79c8dfcd : SUCCESS in 56m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a4d31e52b1084a6a859e9726aa2c6615 : FAILURE in 1h 53m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7ae95617ae7044dea9d137978a92e9a4 : SUCCESS in 21m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26234fd06fb3b2312412118313b69d4ab956e65a","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-05 21:31:37.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/27/840727/1/check/DellEMC_VNX_Manila/8ebd718/DellEMC_VNX_Manila/246 : FAILURE in 1m 18s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"bb522e6b65b9d27c528600b3eac4b144a735bcec","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-06 00:01:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/27/840727/1/check/DellEMC_Unity_Manila_dhss_true/e92e97a/DellEMC_Unity_Manila_dhss_true/448 : FAILURE in 3h 11m 38s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/27/840727/1/check/DellEMC_Unity_Manila_dhss_false/7da6155/DellEMC_Unity_Manila_dhss_false/467 : SUCCESS in 1h 10m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"85e5f1ff1f0311d87d4bb0211f6db4feb26f99a3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-06 13:57:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f7181fd5cb222f8415a3809fe6e43db0847ad87","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-06 13:59:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f01c0d8301d744b4b651caa18ba1bc974526a7ef","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-05-06 14:04:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0787f980a3db322d9961ef4616a195ebf33fd378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-05-06 17:55:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2cf6ebb81b1896a8c4c6898b0b6a80632ea3e408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 19:32:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0541fa77073d409eac1672972c5bc3f6 : SUCCESS in 3m 04s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6d3b12eb6ce24bc494b9ef67e9b40eac : SUCCESS in 26m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cfefb418ac064fcb8920c55d7a1b61ed : SUCCESS in 28m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fc8ad9f55e11470295ba7186632bc3c1 : SUCCESS in 1h 03m 38s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f0b3aceebbb8416faadbce47d662bb2d : SUCCESS in 55m 19s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/13fe7c0eeec847029d43a3781012e656 : SUCCESS in 1h 07m 02s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/2fcf263c5395435f8a70fb1a1b156987 : SUCCESS in 1h 00m 13s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/22a5fe37dff0463db4838fe5270d7b62 : FAILURE in 1h 07m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/51a6454276774793bf5e5ad42ea71de6 : SUCCESS in 1h 18m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a54a5967616144c19956d9d63b537714 : FAILURE in 1h 35m 29s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f220005257c24ec7be401f55011f7839 : SUCCESS in 55m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0db7de89fde64d23a123ef898678c908 : FAILURE in 1h 29m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f1678d54bb264cf0955c786f43cff843 : SUCCESS in 21m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5fbdfba01ae75c48477768d5c1151afbc0a51b15","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-06 19:34:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/27/840727/2/check/DellEMC_VNX_Manila/f898d0a/DellEMC_VNX_Manila/250 : FAILURE in 1m 21s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":2},{"id":"d618073e990869fcfca2404d5bbce58bd97e6eab","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-06 19:51:35.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f11ac75a90973eebd07a1a73775f996644d8ab10","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-06 19:53:49.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a8a804967b39e07e01779be3d7c44a5e7f3a859d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-05-06 20:19:27.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c49f665274a22455bdd4464f549d2a811d4d5ab4","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-05-06 20:20:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"423053bf38e247ed022ddf0f30a27829ed808f9e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-05-06 20:20:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5155d0ce9b273a445fc9fba52f5bdb2a5376b468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 22:02:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6581283628b5453e910f57ccc4f59e70 : SUCCESS in 3m 09s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3cca536509d041d1ace8b43225a96363 : SUCCESS in 34m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/67af05608e0b43b4a3e3c6c3de06abde : SUCCESS in 39m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/06a30a3415fa4ec5ab39b9e78e8b003d : SUCCESS in 1h 37m 04s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/6b41538097ef461db99a4a8e5bfd4850 : SUCCESS in 1h 13m 15s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3c31d4bd45ba47f097a1086f6bef3795 : SUCCESS in 1h 08m 53s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/65d9186298674701a946e5dc3d104c0a : SUCCESS in 1h 08m 15s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/564401cbe22e466bb979a7142c10a309 : FAILURE in 1h 05m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ed89a9ea48904ebbabce8ad7b91fa5c2 : SUCCESS in 1h 02m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c46c319acc4c4945ae68ffee3d286c4b : FAILURE in 1h 42m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c3fd669a06bd41f197cb8734b0654641 : SUCCESS in 1h 07m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/af0a2ea5ba284d7e9fc16be7cecd8325 : SUCCESS in 1h 38m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e5b8095f1b90421e9b30b66a3faa5420 : SUCCESS in 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"50dc64d0ce15713878d101bb79e8e3307ff65626","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-06 22:04:19.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/27/840727/3/check/DellEMC_VNX_Manila/8305f0f/DellEMC_VNX_Manila/251 : FAILURE in 1m 11s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":3},{"id":"1a6337db678c243a07c68ca763a95a7726ce7087","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-07 05:03:44.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/27/840727/3/check/DellEMC_Unity_Manila_dhss_true/79f6ed9/DellEMC_Unity_Manila_dhss_true/455 : FAILURE in 4h 12m 27s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/27/840727/3/check/DellEMC_Unity_Manila_dhss_false/a520e86/DellEMC_Unity_Manila_dhss_false/476 : SUCCESS in 1h 11m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"bca2be0e03bf59c9b826e4eda18764055c59a122","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-07 07:29:01.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a053022d6979d6ba6308488c0504f0709889a880","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-07 07:29:10.000000000","message":"Topic set to bug/1916102","accounts_in_message":[],"_revision_number":3},{"id":"7bb5ad2831c74d8e994b5f087c2f19a623ad132c","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-07-15 12:05:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ca1a88a09f8e5441bae118c8f35d0c88a7d8e138","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 12:05:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4a0cc33680d186b5925a4232976de7ac235d014f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 13:28:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5342cf3052242d286030ba5681b24b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2590257ea81b4d8f97b11d80a25d8a15 : SUCCESS in 3m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6bf24de758ec4a8aa2b5fb7b9f743600 : SUCCESS in 1h 17m 04s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/436d407db04247fbb463bd6bc797eaea : SUCCESS in 38m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/21b10b834016414199687cc99be9b63c : SUCCESS in 39m 14s","accounts_in_message":[],"_revision_number":3},{"id":"5e59c0ec76a42853ba50ca18759b9b60d4473131","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 13:28:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"71caf46c021ed64ddf3a85a3b41c478ec2845795","revisions":{"03a1ffb51f2e277cc4d4b26f5004fcb84a943bdf":{"kind":"REWORK","_number":1,"created":"2022-05-05 19:34:53.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/27/840727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/27/840727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/840727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/840727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/840727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/840727/1"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-05-05 19:33:12.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-05-05 19:33:12.000000000","tz":-180},"subject":"Perform list all sec services using admin client","message":"Perform list all sec services using admin client\n\nWe currently try to list all security services using the\nshares_client. This change switches that search and makes it start\nusing the admins client instead, considering that this is supposed\nto be an administrator action.\n\nRelated-Bug: #1916102\nChange-Id: Idd49e22c8dc534a1fe1e4814f233b079cf14bb72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/03a1ffb51f2e277cc4d4b26f5004fcb84a943bdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/03a1ffb51f2e277cc4d4b26f5004fcb84a943bdf"}]},"branch":"refs/heads/master"},"315b266a13ea5d24ec4f04e3985e6f923297df73":{"kind":"REWORK","_number":2,"created":"2022-05-06 17:55:27.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/27/840727/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/27/840727/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/840727/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/840727/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/840727/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/840727/2"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-05-05 19:33:12.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-05-06 17:55:15.000000000","tz":-180},"subject":"Perform list all sec services using admin client","message":"Perform list all sec services using admin client\n\nWe currently try to list all security services using the\nshares_client. This change switches that search and makes it start\nusing the admins client instead, considering that this is supposed\nto be an administrator action.\n\nRelated-Bug: #1916102\nChange-Id: Idd49e22c8dc534a1fe1e4814f233b079cf14bb72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/315b266a13ea5d24ec4f04e3985e6f923297df73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/315b266a13ea5d24ec4f04e3985e6f923297df73"}]},"branch":"refs/heads/master"},"71caf46c021ed64ddf3a85a3b41c478ec2845795":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-05-06 20:19:27.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/27/840727/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/27/840727/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/840727/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/840727/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/840727/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/27/840727/3"}}},"commit":{"parents":[{"commit":"5203643c2423c89afc90b1c0045de422c643f85a","subject":"Merge \"Fix NoSuchOptError wrt CONF.validation.ssh_key_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5203643c2423c89afc90b1c0045de422c643f85a"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-05-05 19:33:12.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2022-05-06 20:19:27.000000000","tz":0},"subject":"Modify security service list test","message":"Modify security service list test\n\nSince we decoupled ``share:get_all_security_services`` from\n``context_is_admin``, we started preventing non-admins to use\nthe ``all_tenants`` flag for security service listing.\n\nThis change enhances one of our tests to ensure that in a\nquery to list security services using ``all_tenants``,\nless-privileged users won\u0027t be able to see more security\nservices than they should.\n\nRelated-Bug: #1916102\nChange-Id: Idd49e22c8dc534a1fe1e4814f233b079cf14bb72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/71caf46c021ed64ddf3a85a3b41c478ec2845795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/71caf46c021ed64ddf3a85a3b41c478ec2845795"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}]}],"submit_requirements":[]}
