)]}'
{"id":"openstack%2Fmanila-tempest-plugin~841015","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I366221702b3aa2e78bff82e6b2a543b6f6784f77","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/share-backup","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2024-03-06 13:40:28.000000000","reason":"Change was submitted"},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-02-28 21:31:35.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-03-06 13:40:28.000000000","reason":"Change was submitted"},"34181":{"account":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"last_update":"2023-07-06 12:29:15.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-03-06 13:40:28.000000000","reason":"Change was submitted"},"20104":{"account":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"last_update":"2024-03-06 13:40:28.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2024-03-01 11:35:34.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2024-03-06 13:40:28.000000000","reason":"Change was submitted"},"35677":{"account":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"last_update":"2024-03-06 13:40:28.000000000","reason":"Change was submitted"},"33038":{"account":{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},"last_update":"2024-03-06 13:40:28.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2024-03-06 07:23:54.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I366221702b3aa2e78bff82e6b2a543b6f6784f77","subject":"Add share backup tests.","status":"MERGED","created":"2022-05-07 15:20:36.000000000","updated":"2024-03-06 13:40:28.000000000","submitted":"2024-03-06 13:40:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":110,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841015-bp/share-backup","meta_rev_id":"ed9b449a7388cb42f3223c3a93e2f6c016eac4ff","_number":841015,"virtual_id_number":841015,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-06 13:40:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-02-28 21:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2024-03-01 11:35:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},{"value":2,"date":"2024-03-06 07:23:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2024-03-06 12:09:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"}],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-07 15:21:14.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"state":"REVIEWER"},{"updated":"2022-05-07 15:21:14.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},"state":"REVIEWER"},{"updated":"2022-05-07 15:21:14.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2022-05-07 15:21:14.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":20104,"name":"Dmitrii Galkin","display_name":"Dmitry Galkin","email":"galkindmitrii@gmail.com","username":"galkindmitrii"},"state":"REVIEWER"},{"updated":"2022-05-07 17:26:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-08 03:02:40.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-06-15 17:17:19.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2023-06-19 19:03:17.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-07-06 12:29:15.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":34181,"name":"Manish Honap","email":"mmhonap@gmail.com","username":"mhonap"},"state":"REMOVED"},{"updated":"2023-07-06 12:29:15.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-07-17 17:28:41.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-17 17:37:50.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2023-07-17 17:37:50.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-07-17 17:37:50.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2023-08-22 13:42:10.000000000","updated_by":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"reviewer":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"state":"REVIEWER"},{"updated":"2024-02-15 08:38:30.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"}],"messages":[{"id":"d29f288299d89bd2b4b437703408a1a8030463f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-05-07 15:20:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9ac52b5f933ac23d7d00e9a2c37943e67d19803","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 17:26:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08ad55525c974ff992246a1edbf4b71a : SUCCESS in 3m 37s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b2cf3cfe43504513a3519c8f4dfc9f5c : SUCCESS in 37m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cb3e5d72e398437da6243b5b28c0227a : SUCCESS in 41m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/671ae914ae9e4695a9fae6a4b3baf620 : FAILURE in 1h 27m 17s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/151b9fe4efe1461381840dff88c2095e : FAILURE in 1h 14m 34s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/f1c2953e68cc4723a54712c2c859621a : SUCCESS in 1h 22m 26s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/dbbaa4026d3041268495d3d8373fec9f : SUCCESS in 52m 44s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f3b61f2d372d4a7da5a7c69b7036e830 : FAILURE in 1h 18m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/810587c584864799a3048e222873dffd : SUCCESS in 1h 27m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1ea80359fdec4dcdb13e3bfd906cd5ea : FAILURE in 2h 01m 07s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/be7b10f2af3d47a48c4fd93f15b41cbd : SUCCESS in 53m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cd75e533831d468abb652095d02ddc5e : SUCCESS in 1h 36m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/93f611ec69c04b1da86623e3f2d8d1e2 : SUCCESS in 22m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a604218cb60fb124335f4a408b268a5fadad503","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-08 03:02:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/1/check/DellEMC_Unity_Manila_dhss_true/4a74db8/DellEMC_Unity_Manila_dhss_true/460 : FAILURE in 4h 19m 44s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/1/check/DellEMC_Unity_Manila_dhss_false/8f4ea9e/DellEMC_Unity_Manila_dhss_false/481 : SUCCESS in 1h 07m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"49c2e601606c921f62c0bd7f2e6c1dcd02124314","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2022-05-29 14:37:32.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"96ae673ddc8487f63adb1b95e0fdf628400f3621","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-29 14:38:08.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/2/check/DellEMC_Unity_Manila_dhss_true/3aa7ba2/DellEMC_Unity_Manila_dhss_true/515 : FAILURE in 1s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/2/check/DellEMC_Unity_Manila_dhss_false/4151769/DellEMC_Unity_Manila_dhss_false/538 : FAILURE in 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"ad50fe6a3b47590106b6554eae6f7b930e28d1ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-29 16:53:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fcd85a1c78b4b7f933f0b141d51ebb2 : SUCCESS in 4m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0e5d68509e6e4b41a0df4a8513f4302a : SUCCESS in 23m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5eb76b5bbc7445d6bbb66b9ce3a2180c : SUCCESS in 26m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/76ee0e960d594c7aa99f372ffd4edaf6 : FAILURE in 59m 37s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/299ea8fd9ea04045aff4a52767914796 : SUCCESS in 45m 11s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/652b127417394020ac4ad84a9d422494 : SUCCESS in 57m 34s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/8bbac07e804f4c6b8656f880d0e1732c : SUCCESS in 50m 10s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a5e3701b0ce64f06ad698ae36f562ade : FAILURE in 55m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/acb92a1e266d4e9d954792e0d0bb64b8 : SUCCESS in 1h 01m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7bc17b4c09974d108d6df62282dec1a8 : FAILURE in 1h 26m 05s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/527ef13084404e5692042e837557117c : SUCCESS in 48m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/263350e1560c42c39cbf620c89cea56d : SUCCESS in 2h 09m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d3cef7ed0ac54c469044563cc2adee82 : SUCCESS in 19m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"54296b9df63434380836f700412f020b73d9c5de","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-06-15 17:17:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c47d6e2c26061a32f83b09d70cf3426346634b5","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-06-15 17:19:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4f0decc17d96d359ef7c11f9b8af019c2b65b7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-06-16 09:59:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9a899c5b86358eb57a37aa0f334c8441f4e2cbd3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-16 11:19:01.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/3/check/DellEMC_Unity_Manila_dhss_true/00d2cae/DellEMC_Unity_Manila_dhss_true/178 : FAILURE in 46m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/3/check/DellEMC_Unity_Manila_dhss_false/14266d5/DellEMC_Unity_Manila_dhss_false/194 : SUCCESS in 1h 19m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"35bd379e2419f832f522bd2f143a3ea68286effd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 11:28:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/038be19f7ece4107b509c9ad6bd34ed9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43d579de37e34948ad6234bd7e7232c4 : SUCCESS in 3m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fd83fb53a6234ba689984d80a3806d62 : SUCCESS in 1h 27m 40s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/50067bea293c41dfb9abc84147a241f7 : FAILURE in 39m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/06647bfa7a68492eb444a73ea7b46433 : FAILURE in 36m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9e7e5bf8305948d88898b25f43a79139 : SUCCESS in 1h 06m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5ea1016224e24b54bcfbdf84e4c3c048 : SUCCESS in 1h 10m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/80231c2725524984aef84b205182010f : SUCCESS in 1h 06m 44s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/01bf62e9b8094742ab42e012a51f7da4 : SUCCESS in 24m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9dd0c79c428642e481bca29fdc509d38 : SUCCESS in 37m 08s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/19fa1fa5d5fd4af592e54e5362cce01f : SUCCESS in 1h 14m 39s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/fc0ae5339f7e4b9a88c4daf09e16c9b9 : SUCCESS in 57m 21s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/91d34fac70854939ba955058d35d64ec : FAILURE in 12m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/21a1e575df8c41deb75560c9629449cb : SUCCESS in 44m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ef56504752934f8cae97674f065e741c : SUCCESS in 45m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"366753ae467a708b1fd943237067086d2d6d1b55","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-06-19 15:01:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a75daa8e85bb681912c680a41cd695fd7cf69f1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 16:30:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a991008283aa46dbae8f73f02bfa72c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2fad36f464a40e291a7610702f19bfd : SUCCESS in 2m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fa68581885f64b13a087594fc27b701a : SUCCESS in 1h 07m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/78d93cb737144f429e854966cec93fb6 : FAILURE in 42m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a8dbb53567f54c539c4db6aa1d5723a4 : FAILURE in 29m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/09abe54e3bbf4164a6aea921c5adfeba : SUCCESS in 56m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7e77fde621e14927a62ecd90c2596373 : SUCCESS in 1h 06m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ab91a37c116f490290effc9482d1b579 : FAILURE in 58m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fdd1c9dc2182465fb8f09c4c187fe3e6 : SUCCESS in 26m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9b56e8225c694a9baefe7ed79c95f1a3 : SUCCESS in 32m 50s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/44e43da8becf4c6f893f0e5616e98c5e : SUCCESS in 1h 10m 52s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/555457155e8b4e74afd14ecdc1d2b65f : SUCCESS in 1h 24m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d4858bcc76414bd78b5ec7984b92275a : SUCCESS in 22m 07s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4f9fa8a4ed274ff8a40d9259be06c868 : FAILURE in 41m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/76e5111fb24a4d62bbe94ad65e02774b : SUCCESS in 21m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd2eade5bd93fde82f39aeccb8eeed3db7d039d5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-19 19:03:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/841015/3/upstream-check/manila-tempest-plugin-ontap-dhss/6686b25/ : FAILURE in 2h 28m 31s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/841015/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/1ca9a67/ : SUCCESS in 2h 22m 44s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"9eebe8b08b4e985785817f396ef1459bb15f97a5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-19 19:07:40.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/3/check/DellEMC_Unity_Manila_dhss_true/a460811/DellEMC_Unity_Manila_dhss_true/179 : SUCCESS in 2h 36m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/3/check/DellEMC_Unity_Manila_dhss_false/628804c/DellEMC_Unity_Manila_dhss_false/195 : SUCCESS in 1h 06m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"53dfab370911758ce87221da50b1c1dae80c566b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-07-06 12:28:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f4fac54dd897c3833b255c5c3c0adc4bbaf53ef7","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-07-06 12:29:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dee615afa4a1c87ff9e20fad49168ab56d8c56e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 14:22:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9eaefc9a6eb46b6a0d65f4277b54f92\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf550f9db8f43a4b3c14413cbc3e722 : SUCCESS in 3m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/99c855bceec24d98be796980a5bf85a2 : SUCCESS in 1h 30m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/db958b45baf84c20a8a46ec93222153a : FAILURE in 57m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7642a744d5f94124b0936f00bd36b692 : FAILURE in 47m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/dec58161d27c4f079748511968df41fa : SUCCESS in 1h 44m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f717d192deef4e4394221040cbe7544b : SUCCESS in 1h 31m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bc1b575be1d5499fae067c0dee06d48a : FAILURE in 32m 48s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b466664a403e4df0a9d43c21aa4cee49 : SUCCESS in 40m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9b1921aa56e9417fb1d5f13c0cbb90c1 : SUCCESS in 35m 59s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/97b8a2461347482ebb5ce485f6263ca9 : SUCCESS in 1h 27m 10s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f200687db23c4b6b9ae40aab9d9bf78d : SUCCESS in 57m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6b2381f42d824a2c9eaa7deb6262e613 : SUCCESS in 38m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1bbfd51ced6f4000ac15fe2704cc5e17 : FAILURE in 19m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/34dd747d7efe421694e77187187c8ca1 : FAILURE in 30m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"51a8a5cba6a56911348ed62a7657baea1994920a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-06 15:57:57.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/4/check/DellEMC_Unity_Manila_dhss_true/88e07af/DellEMC_Unity_Manila_dhss_true/211 : SUCCESS in 2h 37m 48s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/4/check/DellEMC_Unity_Manila_dhss_false/40482a8/DellEMC_Unity_Manila_dhss_false/224 : SUCCESS in 1h 32m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"77396f1e06216f9fe78c662bb0896c712192f8c4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-06 17:02:43.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/841015/4/upstream-check/manila-tempest-plugin-ontap-dhss/13c973a/ : FAILURE in 2h 37m 54s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/841015/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/4439792/ : SUCCESS in 2h 39m 39s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"1c5aa41102042a93075a5ac6d16e1d1e0c384f00","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-07-07 11:54:01.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"98481756c08526c5bca11151cfe0efd88ae51b24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-07-17 15:53:37.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"cd37b7c46a9eb33429fb5a9b1c6aab10168bfcfa","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-17 17:28:41.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila https://elab-os-logsrv.delllabs.net/15/841015/5/check/DellEMC_PowerFlex_Manila/0ad944b/DellEMC_PowerFlex_Manila/30 : FAILURE in 31m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"1bd9725c3dd4bec43e91d113e080fa20242fa929","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-07-17 17:37:35.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3f3da86f1dc3703aafd8c12bd77f334a1d2878c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 17:49:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5df36a65d20948f2a8452b8a5440f43d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5bae068d9f648139c46fc399e3ca53a : SUCCESS in 3m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a9b27e6a291b4e4695ec34b4ed4bcf15 : SUCCESS in 1h 55m 28s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/089b9daa28564f4d9fa625f4e3eaf1f0 : FAILURE in 31m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a8778916e7834f73a2d12a0cac63c612 : FAILURE in 32m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2195260d95294ecfaba8a5b3b8d7b464 : SUCCESS in 1h 00m 12s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9d9a0eb8dbee4730a397794be0a233d1 : SUCCESS in 56m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a4e8a45886174827bfc55157d83762b4 : FAILURE in 21m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a1af37388c064be7939f5b2c6a7f0eca : SUCCESS in 40m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a306f29f12364ed4b5f4a977e3355dbf : SUCCESS in 28m 25s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/64e41a14c7a54e0c961c1f07d2b4f451 : SUCCESS in 1h 09m 52s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3cae5a501fd942d5baeaae8a3eb92113 : SUCCESS in 1h 26m 29s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/53e99717dbb44d0aa5f4cb078902510e : SUCCESS in 30m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8ad9ae0d28b0420e802c705a80965dcf : FAILURE in 17m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1639433effa84f1a9adfac08e3fb7442 : SUCCESS in 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e57d562af3cccd178362bc5610343751786566e6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-17 18:27:37.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila https://elab-os-logsrv.delllabs.net/15/841015/5/check/DellEMC_PowerFlex_Manila/90d9727/DellEMC_PowerFlex_Manila/31 : FAILURE in 32m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"df5ce13541692e9d3483594f6de1a9b54c6e5139","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-17 20:31:30.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/841015/5/upstream-check/manila-tempest-plugin-ontap-dhss/3882b3c/ : SUCCESS in 2h 40m 58s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/841015/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/9d5212a/ : FAILURE in 1h 32m 12s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"9a1236f2e369723c4268c484a7f93ac54936deb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-07-18 08:39:45.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"19bc901a13511ec2950d19f93378cd3b9382566d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-18 09:05:25.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila https://elab-os-logsrv.delllabs.net/15/841015/6/check/DellEMC_PowerFlex_Manila/ec759c5/DellEMC_PowerFlex_Manila/39 : ABORTED in 5m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"3014de74f778bc4dae47ea5614ec48a5bfa8b877","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 10:09:09.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/908c7d0112bd48adb8a6a0b9e2eb0901\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e68570879a004e5b907b3a37aa51435f : SUCCESS in 3m 23s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7014c26d5ef54f3aa54f9d8b04741690 : SUCCESS in 1h 22m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e68b1d0df6e74e08ae2081ce12a7fd06 : FAILURE in 51m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/32b1cb91d11740d1b3128e2f42ec4023 : FAILURE in 51m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/559327d2e93944ef8c5e70f6519904d7 : SUCCESS in 1h 10m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/20da6516797840efb64c4c59f2c28829 : SUCCESS in 1h 25m 58s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cec8a5d435794470b7409ced2678cd2b : FAILURE in 28m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6dbc8b12376c4c33a7c90c1dfc2b3747 : FAILURE in 43m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b870fa0d5f4d42cc86bc0abe8abf8a0e : SUCCESS in 31m 57s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/df4951894ec04ddab90a5966d96055ba : SUCCESS in 1h 11m 06s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/01d2b1167d2545cb9e41f73a8d32f809 : SUCCESS in 59m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1f9c2872637d499780b2a5bbd6ca1555 : SUCCESS in 25m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6d81cc9c9c4f47a6a8708d543d39397d : FAILURE in 16m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/01d16f3d6c6b4e92bc6cf65d47e78d3b : SUCCESS in 23m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"356515b464598187b6795c717bf516088db769b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-07-18 10:30:53.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"b6e21d514d5d6d99119a44dcd287048d478a021c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-18 10:57:58.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila https://elab-os-logsrv.delllabs.net/15/841015/7/check/DellEMC_PowerFlex_Manila/9c4130a/DellEMC_PowerFlex_Manila/43 : FAILURE in 26m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"5ce944dfce6671edcaa531b02dc63bc7f2043b59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 12:06:21.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e4c72353d054ac0a1af3bcb8ccdf167\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d09b5b4b9174ba78fe81ce52fd62fb1 : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f8a36322fc85459bac28463ebc6d9d0e : SUCCESS in 57m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/676ce7da621b411fa06e07743d8ba8de : FAILURE in 32m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eba6f0a6ebed411a903d65db9f8fa782 : FAILURE in 31m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/fb2d6ecdfa184670afad6d29775c816f : SUCCESS in 1h 09m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e27329d158eb4c249765a3e7f5697eef : SUCCESS in 1h 20m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/426d0a34afc8412093a8c1b7532a7027 : FAILURE in 18m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dd29a93c04f54eb4bac43c73a27c0256 : FAILURE in 23m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1e522a43a0bc44e58506afd83d1dd1c2 : SUCCESS in 37m 14s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/4f3e4fe0736d47ea8d0b0fd5dd0ced21 : SUCCESS in 1h 29m 00s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3e1cb9a2516a4c1c90a4ba5615ded0e5 : SUCCESS in 1h 24m 55s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c639ed59e607421c9d16c86fc80d5651 : SUCCESS in 35m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/be189f3efa8e4e5da83a777bc1327439 : FAILURE in 18m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ca5fe0d48d764b439d34c91c9dd11bbc : SUCCESS in 20m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1e76996bab5f3dc8cb4fe96401e4544b13c0c55a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-07-18 12:23:01.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"0f7e389503da25c8da965368c6873c9410d507dd","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-18 13:50:15.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila https://elab-os-logsrv.delllabs.net/15/841015/8/check/DellEMC_PowerFlex_Manila/78722b7/DellEMC_PowerFlex_Manila/47 : FAILURE in 31m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"330da9b126d2318e422dde7cc0d3dd54ebc4b091","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 14:44:20.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ede28699760a4721939dd574553aa86b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76913ae5547e4a89a36471f3cf1e6e84 : SUCCESS in 3m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/90e8f4e68e604454a4ba65a629455b5d : SUCCESS in 1h 06m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/aa5afdde319e4b6f91eb27d6f19f3621 : FAILURE in 32m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/293911fffe9543bdb6646c15d9dfa9da : FAILURE in 2h 04m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e3969a4ad45046aa81a4d13c6d84a0d9 : SUCCESS in 1h 06m 25s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3269918924ec41cfadc1dce9670d73c2 : SUCCESS in 1h 06m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c9e61acf1c7544a0a9438eaf826a350c : FAILURE in 17m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/99b205d8cb854f3483f69017127499ed : TIMED_OUT in 1h 05m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d08b48e426d441ecaa25444d3d811ebf : SUCCESS in 48m 14s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/f465192760e44cf7ae4b51132620484c : SUCCESS in 1h 40m 48s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/11325639ae56489cb059446874bfbe01 : SUCCESS in 58m 36s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/5ead7008275c4b6e9753ca9c97eaa3fd : FAILURE in 11m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0366f13416104f89bf5120d40a6306d0 : FAILURE in 17m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/232f41085ffb43bea6ae26c04978b7cc : SUCCESS in 37m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"991849744ceba0da88848ac57cc899988495f6be","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-18 21:13:59.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/8/check/DellEMC_Unity_Manila_dhss_true/042abd1/DellEMC_Unity_Manila_dhss_true/246 : SUCCESS in 3h 44m 45s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/8/check/DellEMC_Unity_Manila_dhss_false/2c76e2f/DellEMC_Unity_Manila_dhss_false/264 : SUCCESS in 57m 52s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"0e1371131a0338ee5571ded72474887b8e3f03e2","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-07-19 11:41:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"53f5f8203130f40f1b24aaad6546ef498d948b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 13:09:29.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/72d147cdd90a49a0ac274d533af64450\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f88135517f94bbcb8a88ee602f24514 : SUCCESS in 3m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/de0015c32a834ca38f799f3ce87e68c3 : SUCCESS in 1h 12m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/23ef11737ed54555a4a0aa17c258ec18 : FAILURE in 33m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0327b6bebaa74ff7b6aa2d73bec34032 : FAILURE in 50m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8cddca8a03a34d2b90be67a571aeffdf : SUCCESS in 58m 35s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/13f484c301084bc9ace980fb42bab50b : SUCCESS in 1h 00m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4898d7579ed2484baecbd3c51ae833d4 : FAILURE in 21m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0cf118f7829541eb9c1c44cfca64ef1b : FAILURE in 26m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/63b7703da75d4495b0dbd9397185502f : SUCCESS in 33m 57s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/f0495ac2212740cab8e0284218a89941 : SUCCESS in 1h 06m 38s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/09b3bf6437e14f3ebd09abc92da62951 : SUCCESS in 1h 20m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2e7650f763fe48fcb2769407c57c1c75 : SUCCESS in 23m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e71759c2c4904fa6a282b815017e4a50 : FAILURE in 19m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/356e234ba33f4723a0bb338eccc882c7 : SUCCESS in 25m 02s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1881fdfe1ec578a5630043581b3f12b6ea6ac74a","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-07-19 14:35:23.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"dc3a649fb56cfc883b8f8f06be2d6c4e8747b48f","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-07-19 14:36:09.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"e779c68be40182190f75e027041e94bff3b4baa6","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-01 10:59:07.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"769f4b3669b27eb721a6a2a80a5b4624bf7a03e6","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-01 10:59:33.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"775ba96de3b5e1a54896a34850fda1f462447b02","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-08-01 11:29:16.000000000","message":"Patch Set 8:\n\n(10 comments)","accounts_in_message":[],"_revision_number":8},{"id":"20cc500f0af3c725efa516704ca43c997f84d026","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-08-01 11:29:29.000000000","message":"Patch Set 8: Code-Review-1","accounts_in_message":[],"_revision_number":8},{"id":"c135cb12017d010c06284b13d5fe17b58c974285","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-01 19:03:41.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e7a2bd7c369be180dfdfec5472e810e72ac64c59","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-01 19:04:17.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"473bbd83321c7105a76ba601ac3dc1bd830d5125","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-09 11:57:31.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"86a05651c0e360c9274c084be67504083528e310","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-09 11:57:56.000000000","message":"Patch Set 8:\n\n(10 comments)","accounts_in_message":[],"_revision_number":8},{"id":"02dff5bda1f45d692c69d3d6376750124794c865","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-09 12:28:56.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/9/check/DellEMC_PowerFlex_Manila_dhss_false/d59ce7a/DellEMC_PowerFlex_Manila_dhss_false/179 : FAILURE in 31m 06s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"19cebb442eb31d72bca5a7bde512ab60e755c04f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 13:59:55.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6bb231f61f94b9db296e13a0151f202\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f082cc7bdc24514a209b71d9eeba4f3 : SUCCESS in 2m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3fe1c1a4bff243b1bf770b42ab6e2c76 : SUCCESS in 1h 41m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/83a40c1f1f804962a2a6f46ffcb6854d : FAILURE in 50m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d4742c65933b45cca17306e39dd105c2 : FAILURE in 46m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1bb23cba89b949b5b4d9d168cf6b193e : SUCCESS in 59m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b2d112677ca544639b9f3da159822439 : FAILURE in 1h 21m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/65bcfc30d641476ab3fb150a05e7a6f8 : FAILURE in 1h 05m 48s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/58f828af68fb46279f144695c17bc827 : FAILURE in 39m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c9ba8144842e467f91a200cf5806d8d9 : SUCCESS in 32m 57s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/7dc99fa164c146adafb54c99fa87bc5e : FAILURE in 1h 55m 28s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5388b4a7018c420c82118e0c2f0f6b39 : SUCCESS in 1h 03m 20s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/4de5b7d9814e4899ba56b28205b932e4 : SUCCESS in 36m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2ec299fee9434be0a8fb7e0ac27d4a0b : SUCCESS in 57m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7eb940b6a86649b3adc06746eabd8487 : SUCCESS in 21m 54s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b9abe7849d544fb6dc620997d1a0c98c27855c84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-09 14:32:57.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"65694d15faeb70053c7a90af3f76c21730020098","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-09 15:04:47.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/10/check/DellEMC_PowerFlex_Manila_dhss_false/1af8dd9/DellEMC_PowerFlex_Manila_dhss_false/180 : FAILURE in 31m 32s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"d405c6de49efdb9056bc4f1c09476e25b472f068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 16:15:23.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b21dce981edd46bfbfe914bf7831574b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e058cb779877464190c535a74a703667 : SUCCESS in 2m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5773be18aca44909919c185fc7c3fc10 : SUCCESS in 1h 25m 17s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/de8663f99c034eadb304f2625d5131cb : FAILURE in 51m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/411ee1da921a43a3989a932bac92e9de : FAILURE in 44m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bebacf4c087c4fe3a8e135c44063ffb0 : FAILURE in 1h 20m 17s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4eb4e94173844264ad424a668915a59e : SUCCESS in 1h 19m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/805770590d744acea390dd81622684a2 : SUCCESS in 1h 31m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a640171510c24ad28adbbea93f0fafc3 : FAILURE in 34m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d26fbfe9664946b289dcc63706600243 : SUCCESS in 39m 07s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/0b28f6c8381e4322a2a8044c5b983286 : SUCCESS in 1h 34m 06s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/834d4ff740ed4758bc61f8b75d687449 : SUCCESS in 1h 11m 09s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0521823c39ac4421bf75a8cc69ac5e1f : SUCCESS in 23m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/08dc4f722335468e9af04169794ac0e7 : SUCCESS in 36m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4657dc1748c743038220f66c49aaf844 : SUCCESS in 30m 29s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"dfdfca5d8618e6bd917563b6b59061cb6a09aad4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-09 17:44:48.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/10/check/DellEMC_Unity_Manila_dhss_true/0d33edd/DellEMC_Unity_Manila_dhss_true/341 : SUCCESS in 3h 11m 31s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/10/check/DellEMC_Unity_Manila_dhss_false/24c4cfa/DellEMC_Unity_Manila_dhss_false/365 : SUCCESS in 1h 06m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"074f48d4d6cb6b06441fac78f2f62fbcc6b68ede","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-10 12:43:29.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bc01d5c4b85948952d29958d7b704a830f2f341b","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-08-10 12:55:49.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"88737161ce3532c1a2af9c05d80959952d5c4cd0","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-08-10 13:01:45.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"dee797e54a5303156ba24ce3ca2eb74e32bbeedb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-18 14:01:18.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"0608d2f0c8477d678493d267dfe6e226feab040c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-18 14:03:37.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"3b25b7b3edffa69df4f91dda2e303e0ab7b5425c","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-18 14:03:42.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"20087e7c04bc22ba4a99e73910ecce58031d4dfa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-18 14:28:30.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/12/check/DellEMC_Unity_Manila_dhss_true/1688b7f/DellEMC_Unity_Manila_dhss_true/421 : FAILURE in 24m 34s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/12/check/DellEMC_Unity_Manila_dhss_false/54444a0/DellEMC_Unity_Manila_dhss_false/446 : FAILURE in 11m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"90ee939331c680915292ab1c7df3f3f8c10c45c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 15:49:07.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/72c2c8fef6ae493a8da207121dac2592\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a719bd7a5a9482b8d4586d2b71d9ee5 : SUCCESS in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d01a593783dc4777bd00d626c3e78a24 : SUCCESS in 1h 01m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a2b2aced50fe44c78192bbb55f60c652 : FAILURE in 40m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c0966c67dd484020b08f4fa245b27604 : FAILURE in 38m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/da68a8de7df84f38817256100ee60b8e : SUCCESS in 55m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e011218587ca4b958e12697ff2909de7 : FAILURE in 17m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/18528cbba32440e8a8a0c168e3ecac9e : FAILURE in 57m 03s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1ce37036a7ff4e21a563c7b6621c3b0b : FAILURE in 36m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/644dab46b733440fa0ef8857e9a91020 : SUCCESS in 41m 24s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/df46f9e86a0545bf8d5c26f9bd9a5768 : SUCCESS in 1h 20m 40s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/b2179b29ceff40ecbad00ef150ac5ded : SUCCESS in 1h 44m 30s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/1400815b379b45a19cb37dd944449336 : SUCCESS in 38m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a742e0a1166a4618915a3e88d10b7551 : SUCCESS in 58m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2525e3f5b6374547ad15b6f54422daed : FAILURE in 19m 33s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"a436e7a65eb776e6c4765d3df30760c98dc6d902","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-18 16:07:02.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"55764eb8eb5d0e77752998c1612cc19a731e2782","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-18 22:56:23.000000000","message":"Patch Set 12: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":12},{"id":"7594b77cad2785fadc963b15e5f50a8fed5939ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-21 16:14:16.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"15ce1c5d41a49baa0500f1ffd21228b8fa5bac57","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-21 16:49:21.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/13/check/DellEMC_Unity_Manila_dhss_true/8a16415/DellEMC_Unity_Manila_dhss_true/435 : FAILURE in 34m 47s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/13/check/DellEMC_Unity_Manila_dhss_false/fb3953f/DellEMC_Unity_Manila_dhss_false/461 : FAILURE in 12m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"e35fb8167b4263a9ba3d785bc73d23470eb83163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 17:49:07.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2dd0972f522242f48abb9fd3034aa7f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfa1b8b56fa14bd0a2f471fcf897e87e : SUCCESS in 2m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5c20fee35f2b4f6da5f4b8c809958eda : SUCCESS in 1h 33m 20s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b1dad466ca664b5d921b08f952393500 : FAILURE in 45m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/47162123b4fb453096fa63308989bd6c : FAILURE in 32m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/80bf1942ef1442f4a3b6b93b79525f94 : SUCCESS in 1h 05m 41s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3e0eec59caa8493588d9797972a191f6 : FAILURE in 29m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8b26623c74c64ddaa57a814d5712aba2 : SUCCESS in 1h 14m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/16969de7199e4a238acd3b61e8430fa7 : FAILURE in 25m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/98b57861aafb46d197df81904ef597be : SUCCESS in 42m 45s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/a788c7ecc846498ebce0480b8ff14551 : SUCCESS in 1h 17m 17s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5705e232e3494b9b86fd38a6fdad4a26 : SUCCESS in 1h 17m 36s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/a420050ba4c04b9287d8a7d6500bd263 : SUCCESS in 22m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cca0b583e4b64ab4867456f172ff3446 : FAILURE in 55m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a3193c56fc584d81a106a221ffa55cf0 : FAILURE in 29m 29s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8d65e5aaf29e823c5345b09ff9f8fde0c4b9b999","author":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"date":"2023-08-22 13:42:10.000000000","message":"Patch Set 13: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":13},{"id":"1156d8a63a26bb96b592052da77da50a16eb0309","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-22 16:10:47.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"a948f46e477028d032aba953393191411eb74eec","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-22 16:30:34.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/14/check/DellEMC_Unity_Manila_dhss_true/cb1e4f9/DellEMC_Unity_Manila_dhss_true/440 : FAILURE in 19m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/14/check/DellEMC_Unity_Manila_dhss_false/22a78f1/DellEMC_Unity_Manila_dhss_false/466 : FAILURE in 12m 27s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"459c9635dee6380019fa4b3ca36290b9cad20f4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 17:37:45.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/acea3f19a21e4e69bbe31017c15c4a17\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c178fc700187443e8431dde580ede980 : SUCCESS in 2m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cb7a1c3690ae4f3fb71c1b8101c6dd2c : SUCCESS in 55m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/387e80df2c084880b1ca31dd9d94cd6e : FAILURE in 35m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3eefe7b34a06401fb10be67979dbf71b : FAILURE in 53m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/dcfd6720f2744854a8cd7b70be2e418a : SUCCESS in 48m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8d12901b12e440a0b415b0e2c7d48628 : FAILURE in 24m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3f05d707c6194da2bc9f7de3e88d536e : FAILURE in 1h 21m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/72108cad6f8146a0a408702332678b6a : FAILURE in 38m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ea54ed2a7ecd44929340e3e2d978d8d3 : SUCCESS in 34m 16s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/63b7a2b7efd44a3898e3aca826f21f6f : SUCCESS in 1h 07m 23s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/18f106f38e0c4ff688de4c8c4c0d1ad1 : SUCCESS in 1h 09m 35s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2309c714749a490194485715010cb56f : SUCCESS in 29m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0411c5708d2640c29d2797f70fd9ff21 : FAILURE in 59m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cb2be0c3573343b9ad61cc972ced4eab : FAILURE in 28m 54s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a492d3b4f2e92e11262b60c0038716c9c67d252b","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-25 14:28:00.000000000","message":"Patch Set 14:\n\n(17 comments)","accounts_in_message":[],"_revision_number":14},{"id":"b2d8885aca1ea77a975d424572a5f3fb752d505d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-25 20:36:33.000000000","message":"Patch Set 14: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":14},{"id":"5f12edcb23f23add9e368bd0c6fed8f8bb60afd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-26 14:07:23.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"9e596051270ca93da7dbe49aed6187e90c5886d2","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-26 14:07:36.000000000","message":"Patch Set 14:\n\n(9 comments)","accounts_in_message":[],"_revision_number":14},{"id":"e020e03805a306c46e0a17341e1a68a5f94a3b1b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-26 14:26:45.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/15/check/DellEMC_Unity_Manila_dhss_true/c506222/DellEMC_Unity_Manila_dhss_true/471 : FAILURE in 19m 02s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/15/check/DellEMC_Unity_Manila_dhss_false/63bca04/DellEMC_Unity_Manila_dhss_false/497 : FAILURE in 11m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"9032f5e15cd7868a58a9ca9a7e36217b34222e9e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-26 14:41:07.000000000","message":"Patch Set 15:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/15/check/DellEMC_PowerFlex_Manila_dhss_false/5131b44/DellEMC_PowerFlex_Manila_dhss_false/257 : SUCCESS in 33m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":15},{"id":"c4ce35061b7c8364909cb529d5e52f1654954eea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-26 15:41:14.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fabc1be45c4a4e4ead365c6391592a47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f590f72b5bb640acadd6623044f6d9ce : SUCCESS in 2m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/979d91e8bda94b90afb472d87d9842d1 : SUCCESS in 53m 25s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/35652014a96544978fcad30a3e624ce7 : FAILURE in 52m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/551fb1818d94404b8d72b672058e274a : FAILURE in 42m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8f95bc37fecd44248d29be155044ae77 : SUCCESS in 1h 15m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9de09e812826438486315f0476088ead : SUCCESS in 1h 33m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/616e2ad8f96944e29f67f0fd1a272224 : FAILURE in 1h 17m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/caaa2245f9c84224abcd819279432045 : FAILURE in 34m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6b9240cefe55451e888d2776df17327c : SUCCESS in 42m 13s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/f6d31fb9699049bdbc43a06889563454 : SUCCESS in 1h 21m 25s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a7e3b0d9c61042dc97886f3373acda90 : SUCCESS in 49m 17s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7c07ca548a354d57b40009f67caf60ff : SUCCESS in 20m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b3eb40282e214400b73a48bd22e0cf94 : SUCCESS in 52m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/48b727e35db440e298b8c928905c0006 : FAILURE in 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b9359710a8a9305fc54bd983e5e0ba05fe6a9374","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-28 10:28:32.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"ac8ebf0bd419833c528132094d6086e349b51e28","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-28 10:48:29.000000000","message":"Patch Set 16:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/16/check/DellEMC_Unity_Manila_dhss_true/68fcb54/DellEMC_Unity_Manila_dhss_true/472 : FAILURE in 19m 38s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/16/check/DellEMC_Unity_Manila_dhss_false/dd0105e/DellEMC_Unity_Manila_dhss_false/498 : FAILURE in 13m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"01e71e12f70819a812f744434c78d7f85e9c1368","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-28 11:03:18.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/16/check/DellEMC_PowerFlex_Manila_dhss_false/5030620/DellEMC_PowerFlex_Manila_dhss_false/258 : SUCCESS in 34m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":16},{"id":"4850a9a1699771347eff8227d913c519ef21d11a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 12:02:11.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/60046f2c71a740978cfb171427635036\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf716e4088c447d49b32c3fa110b26fa : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fe55b9f0390f4dceb479686f52aae382 : SUCCESS in 1h 07m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cbd5138828b042edbac7658d19f21bcb : FAILURE in 54m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/271bceb270f345b7b43fd52774efd8ea : FAILURE in 49m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6283ce9c03bb44dea5cd3cf960c30ab2 : SUCCESS in 1h 17m 51s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4aa5243006fc40a28101f0ba99727193 : SUCCESS in 1h 22m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7a3ca961d5814f27bc1e86e173a7a884 : SUCCESS in 1h 17m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1f12e90170f74fa0b8969e3ac369ba4c : FAILURE in 36m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/859825df904f4b8ca22d3332cea6de7c : SUCCESS in 37m 36s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/60e8bf1cd9244822835ef26811021e1e : SUCCESS in 1h 27m 25s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/26b56597a82a4fd19821c30be10dcfb5 : SUCCESS in 1h 15m 51s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ec2c4df3f2c2449abda55fa0f4648298 : SUCCESS in 31m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/05757d099cf4424782113e4a809f0a59 : SUCCESS in 1h 02m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/44aba7e4b66443a0a5c31445e21807b5 : FAILURE in 31m 08s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"af0fac052839f2b6889d8faaedcb095a116c6795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-28 14:04:59.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"8d1ac74f0b8f17c9cad3a46c12e321a29ef41f0c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-28 14:22:54.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/17/check/DellEMC_Unity_Manila_dhss_true/fc3c7ba/DellEMC_Unity_Manila_dhss_true/476 : FAILURE in 17m 34s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/17/check/DellEMC_Unity_Manila_dhss_false/edf7534/DellEMC_Unity_Manila_dhss_false/502 : FAILURE in 11m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":17},{"id":"136df1bc09a3383a43f1c5b9c9588f4386a6bac3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-28 14:41:22.000000000","message":"Patch Set 17:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/17/check/DellEMC_PowerFlex_Manila_dhss_false/df97245/DellEMC_PowerFlex_Manila_dhss_false/262 : SUCCESS in 35m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":17},{"id":"f243531ae0385a55095e4989d778507248f0df6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 15:36:53.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c39beeb3be6b429ab138e38be71b72fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f872d6e16e9d403e9c75f1c5ca0056d5 : SUCCESS in 2m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/465473106fb44701a1579e10ad71a4b5 : SUCCESS in 1h 09m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5890a2d8dac14dad8ae5337bdea7773c : FAILURE in 49m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b5984984c4e04ed9b39efe8f00f4632d : FAILURE in 36m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4170af392292419585535ff8c0465394 : SUCCESS in 1h 15m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e967f6020ef141f1a8ccd42bbe42fdfb : SUCCESS in 1h 12m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1b7d6a2db4154414947f14d5a6a49ef6 : FAILURE in 27m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f1d2d1516ba44366801b7324b62b18a2 : FAILURE in 29m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f8bab7c2f181446e8fb2031a023f4a64 : SUCCESS in 26m 37s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/ccb9de374d5d418389dfcc22d9469d6a : SUCCESS in 1h 25m 43s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/574fac8c73b0440ab3df60cc3f6656ce : FAILURE in 1h 00m 28s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/e4cbbdd72837488f80d57ce4fe38ae98 : SUCCESS in 39m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/33c58735320843c0ae9854c02ad96dd1 : SUCCESS in 42m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3641f85ff13b43f8bcecc3e173cf23f6 : FAILURE in 17m 19s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"f1608c2caf3cc2d10a13fc6f2ce74547a2e92126","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-28 17:19:03.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"ada7e68fa247ce5c928658e6085c56c8dcc2c271","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-28 17:40:41.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/18/check/DellEMC_Unity_Manila_dhss_true/3ae880c/DellEMC_Unity_Manila_dhss_true/480 : FAILURE in 21m 20s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/18/check/DellEMC_Unity_Manila_dhss_false/0bf410b/DellEMC_Unity_Manila_dhss_false/506 : FAILURE in 12m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"2f821975aca41f4b24189d409e1c7ec158f872e5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-28 17:53:52.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/18/check/DellEMC_PowerFlex_Manila_dhss_false/e2e3021/DellEMC_PowerFlex_Manila_dhss_false/266 : SUCCESS in 34m 32s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":18},{"id":"c607aeba2ebd8a012fc882a6fbfb1b0ed18fe9ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:51:10.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bee41d681d1e479db93c140564cdd76f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1971dcd26a3a47688a546f015093253e : SUCCESS in 3m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7f15a7eb315444f5be6cfd72eb889db9 : SUCCESS in 1h 15m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fe9d9910aabd429291ce157c9374375d : FAILURE in 48m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0da7812d35044afc88599d0960f8d6f8 : FAILURE in 45m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1b0564be4e844a6d9bed5528e2c8337d : FAILURE in 1h 11m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b442f00c16984271b88264e703cc4099 : SUCCESS in 1h 09m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/77644378b4184975bcfeacedc2186129 : FAILURE in 25m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cd711732647242f6907d92101be6ac27 : FAILURE in 34m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/106c9f30a47a485eb0aaafd40e2037a2 : SUCCESS in 31m 57s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/6096eefa26994cc69b8bb0454d2b751f : SUCCESS in 1h 26m 39s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/bffcfa3965a440c38f78145bb399a76a : SUCCESS in 1h 17m 32s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/51a8b0d40ab24f7dbfb96cb99c23e7bd : SUCCESS in 36m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ff1698e274a84306b4ca9cfd5d0c0f40 : FAILURE in 43m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b2b3269f28de4922b960723ce8a63086 : TIMED_OUT in 1h 02m 45s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"5c7cd6117b672c9035f404ee3b8460435f630a38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-29 10:08:06.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"db58cfb42c30d5825bb8e53ec18796038a2edee9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-29 10:29:26.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/19/check/DellEMC_Unity_Manila_dhss_true/9d71208/DellEMC_Unity_Manila_dhss_true/481 : FAILURE in 21m 05s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/19/check/DellEMC_Unity_Manila_dhss_false/150c632/DellEMC_Unity_Manila_dhss_false/507 : FAILURE in 12m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"8b87b19b69af8ed938d990a94cf091cab84f9fca","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-29 10:43:59.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/19/check/DellEMC_PowerFlex_Manila_dhss_false/9755887/DellEMC_PowerFlex_Manila_dhss_false/267 : SUCCESS in 35m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":19},{"id":"04be5991384c2d521323d93965a53729642641ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 11:44:07.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f2ea27b36d34fbeb322811e6f31791d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d1ad2c8aed4c7c8894255c50168494 : SUCCESS in 2m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/75c1a8198455491aaa6fd1e258efe1f9 : SUCCESS in 1h 05m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9521feb8d89b4231aa27b0a6eedae734 : FAILURE in 49m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b6e6c9b488ae4dfaa31d45ca5b373a8c : FAILURE in 45m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/dd7a3f0e4afc4ccc9e7aa354997c77aa : SUCCESS in 1h 06m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4c8c4a53e99e4648bdcb0ec7ad95fb3e : FAILURE in 1h 14m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cb5e5e89e3b74cc9a252210e119edd7a : FAILURE in 39m 42s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/71d761aa50264336967bc497a87802f8 : FAILURE in 39m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/322a231eecc94c959dcc720ad506d9a4 : SUCCESS in 41m 18s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3afbb1e8041a484bbc52c2d1e79d5186 : SUCCESS in 1h 30m 37s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d0267ffc1e7342c0a78905cfa552edcf : SUCCESS in 1h 13m 33s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0bb24f9b426f47308c9cfc05d36facbe : SUCCESS in 39m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ac520943f65b4a8e95577c81370f3253 : SUCCESS in 50m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7c898df9e66e4c62bd11b88f117abcaa : FAILURE in 22m 28s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"9597fda4fdeb4079e77b20266222ccc0b8945043","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-29 12:01:44.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"3e96835fd1059658aa4e4ae8776aef9d455f4de3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-29 12:21:53.000000000","message":"Patch Set 20:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/20/check/DellEMC_Unity_Manila_dhss_true/5ed3d1e/DellEMC_Unity_Manila_dhss_true/482 : FAILURE in 19m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/20/check/DellEMC_Unity_Manila_dhss_false/785c570/DellEMC_Unity_Manila_dhss_false/508 : FAILURE in 12m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":20},{"id":"e76ba9d96fe23a709fc58957f9d9b28a33906a42","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-29 12:38:25.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/20/check/DellEMC_PowerFlex_Manila_dhss_false/11fd315/DellEMC_PowerFlex_Manila_dhss_false/268 : SUCCESS in 36m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":20},{"id":"eef39f0b978edae694b2170bdee5142622723fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 13:48:58.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3dec149424e34501ba2da487d874d47a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcefa35ec0c54ef68fd73c94ace72b02 : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a44c7db4ce8b492bb4d486ceaba827ed : SUCCESS in 1h 24m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ce716ba79488478d878f26020610fba4 : FAILURE in 56m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/079365e1c9f249eba3f4c7965b69955c : FAILURE in 37m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/70145d29d7994f39ad8d0c4488e38cc1 : SUCCESS in 1h 00m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0fa6aa33878e41a2be8ac76cfe6b27c2 : SUCCESS in 1h 27m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2fbea6f2c65744389c83a41b8dd4f028 : FAILURE in 28m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/10245a846f5f4dc1b3e2e0189d1351d0 : FAILURE in 27m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dee7cbd7351349adbe7447af325c28d4 : SUCCESS in 33m 13s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/350b4bac40ce404e8aba9685ce2d0ea0 : SUCCESS in 1h 45m 46s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/cdd89e1738b34f5c9e490a6c22f7f95b : SUCCESS in 1h 01m 02s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/411fdc50a4b544a698195628d1758da3 : SUCCESS in 25m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b0cca77086e54493a2d6ac53e0ec8ffd : SUCCESS in 44m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/03f20db6bf854d36bdc4846171540e57 : FAILURE in 28m 24s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"0f837cdbdaecd5678a871dac4f10a93c04e3964c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-08-29 16:27:14.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"84d17d008771825789d42d3565b1374c2bb35c8b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-29 16:46:26.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/21/check/DellEMC_Unity_Manila_dhss_true/093a39c/DellEMC_Unity_Manila_dhss_true/484 : FAILURE in 18m 54s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/21/check/DellEMC_Unity_Manila_dhss_false/894463b/DellEMC_Unity_Manila_dhss_false/510 : FAILURE in 12m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":21},{"id":"420073689ee2b93982be7c5f5926a4f8e62c7999","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-29 17:02:40.000000000","message":"Patch Set 21:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/21/check/DellEMC_PowerFlex_Manila_dhss_false/05abc0b/DellEMC_PowerFlex_Manila_dhss_false/270 : SUCCESS in 35m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":21},{"id":"2b73feca713650530c6954715b06c9d28cff73ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:04:59.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3163d52ec9a493fa8e86715ae677b34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaa688b4957b4dec9818756fc5d5e5d7 : SUCCESS in 2m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/05e5563312664ead818ef8c479190b14 : SUCCESS in 1h 07m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/874ddefed8dd453fa98a52bc1ef1fc0b : FAILURE in 44m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ddcaf018ee1346029559cbd34cfe7163 : FAILURE in 39m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2e1b307154bb42f983d1e43bd3127149 : SUCCESS in 56m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/86c0d77912674cc3b35bd833e9616aca : SUCCESS in 1h 23m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2020c42fa80c43e9aefc1c42d74d0caf : FAILURE in 35m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3ab9010e8e6242a883bf69e740feb6a6 : FAILURE in 45m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cba5bbe3f7634f3da6ea4a71f71f2b8b : SUCCESS in 44m 06s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/8aee56bac05644f19670c33ae9290709 : SUCCESS in 1h 32m 13s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/dd83c992f2d74834bb976592b9eef407 : SUCCESS in 1h 26m 57s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/eb70bec3e2e0426d9d552aefdf079e1a : SUCCESS in 27m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6c0f1e472f294e86be92efca6d27ae83 : FAILURE in 1h 00m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/38c667b4feb44b58aba8af99ddc629e8 : FAILURE in 20m 12s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"f658781c5c8d6b5eb50caf1ce25f0ec9accf02cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-09-07 13:12:57.000000000","message":"Uploaded patch set 22: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"297a33c74907695789feb59317a5392a985d3514","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-07 13:43:36.000000000","message":"Patch Set 22:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/22/check/DellEMC_PowerFlex_Manila_dhss_false/f4a97ef/DellEMC_PowerFlex_Manila_dhss_false/331 : SUCCESS in 30m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":22},{"id":"d36b7321c83fb4cc2c34ab69b2163885abae35a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 14:51:34.000000000","message":"Patch Set 22: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f2ecbc16b0f4232af68020fdd95f982\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc30dcf584ad4020b036bae64d375d82 : SUCCESS in 3m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/56000e10ae69472790875ca1fb9df9c7 : SUCCESS in 28m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1ba55efafb294019ae2349f91bd30680 : FAILURE in 48m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/35f64e7270514f6c9fba534e3ceb588f : FAILURE in 29m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c94602a2c47c46aea0881ed3b14afdd8 : SUCCESS in 58m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7ad5c936377a4120bcac95e32412d929 : SUCCESS in 1h 30m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/21a28d31442c4956b0a576d2e5749f95 : SUCCESS in 49m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aa73c79330044229a57ad1876c9561f7 : FAILURE in 27m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/727b2d1f7b9c4ccab05bd0ee1ab5d9f7 : SUCCESS in 47m 57s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2503017c93634dd685a4f702049d6f70 : SUCCESS in 45m 23s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/190ed0079aa94784987e660fc976f7d6 : SUCCESS in 43m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/bb94bb0f13114480a433bcd7899a8f6c : SUCCESS in 28m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/362153632f9140e19f56166e7e3ef499 : SUCCESS in 56m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/80830dc9db1b421b8fc51adaf2de629b : FAILURE in 20m 56s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/manila-tempest-plugin on branch master.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: manila-tempest-plugin-zfsonlinux\n      description: Test ZFSOnLinux multibackend (DHSS\u003dFalse) with postgresql db\n      parent: manila-tempest-plugin-zfsonlinux-base\n      branches: \u0026ubuntu_jammy_test_image_branches ^(?!stable/(yoga|xena|wallaby|victoria|ussuri)).*$\n\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 161, column 3","accounts_in_message":[],"_revision_number":22},{"id":"ed2ca9ea5fd6c06f41d76e2b1f8fafb97deef7fa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-07 21:22:34.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/22/check/DellEMC_Unity_Manila_dhss_true/cbe1a6c/DellEMC_Unity_Manila_dhss_true/554 : FAILURE in 51m 50s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/22/check/DellEMC_Unity_Manila_dhss_false/b5e0d05/DellEMC_Unity_Manila_dhss_false/578 : FAILURE in 2h 50m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":22},{"id":"afcfb345c55042d1b8ed58a559c43e162ad7935a","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-09-19 12:20:24.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"cbdd10fd3430254cbbb77a08ff44d64c8ddf5512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-10-10 13:43:02.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"85368dde651d12e2fbe74d18382e4a168354e772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 16:10:43.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/799f53630bca49d3a4168fbd6a51a3ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/099152c89e744bcf9cf9e15b3cd81b24 : SUCCESS in 3m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d32d17ba8dba478ea3454f00b45c4195 : SUCCESS in 50m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/70fb5146ec0244669302889fbf0422c1 : SUCCESS in 2h 21m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/77be95860abf49a5b9d10de35ee595a8 : SUCCESS in 1h 10m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e9395533240a4b55b4fce6faa318bbb0 : SUCCESS in 58m 20s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/04df48b6c74844dd8bf128a58ad4d70b : SUCCESS in 1h 31m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/71f5868f8c33462ca7a04ea8b37ed9d7 : SUCCESS in 1h 36m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a75dbeb92a884bca86bbc41e0e624183 : FAILURE in 31m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4d1c3baac27d4819bddb1947eb5ba8ac : SUCCESS in 44m 02s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/86b90fd7bd0d41348e0d6211be29f5a6 : SUCCESS in 43m 24s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/807dfa511f9e4ff79520a146ca9c6126 : SUCCESS in 31m 22s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/92633d3b660a43ffb98aa4e9f0da1bd0 : FAILURE in 28m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f1bf43b58606460a8a460321cb2dfc10 : SUCCESS in 37m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5ef65aa45f484748b71fa1da305ee16c : FAILURE in 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"d14ee0bd3fbb105aa83e6a0209c3f9a65da1319b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-10 18:39:38.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/23/check/DellEMC_Unity_Manila_dhss_true/da18508/DellEMC_Unity_Manila_dhss_true/651 : SUCCESS in 4h 30m 38s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/23/check/DellEMC_Unity_Manila_dhss_false/6b9bab9/DellEMC_Unity_Manila_dhss_false/688 : FAILURE in 1h 23m 36s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"30418e25fdd702ac3c47c8abf6999c6a454d7b04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-02-13 08:51:55.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"da40a83cd7db5e6ec416bd9b5176487dec981e2a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-13 09:22:47.000000000","message":"Patch Set 24:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/24/check/DellEMC_PowerFlex_Manila_dhss_false/603a110/DellEMC_PowerFlex_Manila_dhss_false/806 : SUCCESS in 30m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":24},{"id":"7868264dabf86151009a3ecb388aed2aea1cf6f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 10:52:45.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b545c55ffcc24463bf8805bfb873d37b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b3e8845701f4723ba3f4d8385e6c5ed : SUCCESS in 2m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4d57976da8834fabaebd783f01157e32 : SUCCESS in 1h 10m 20s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/85f2a81b9eb24c4e956423f9f9776fca : SUCCESS in 1h 58m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aece6df2d5a8422dbc543a5bfea947e2 : SUCCESS in 1h 14m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/27c8ebb9cdb6455a88528fda405ed402 : SUCCESS in 1h 10m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6c7c497cdc474644ad6f7c0cb7e4c3c2 : FAILURE in 1h 37m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/52bc957638894245b4480c70ef7ff5b4 : FAILURE in 1h 12m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/76920b858da54e138c1a6b494bb9b163 : FAILURE in 33m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/43f0922b36af4c98a75067c8ff1787b0 : SUCCESS in 28m 44s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/e9e19f73e1984565b12e293971fdfc16 : FAILURE in 21m 05s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/185b0f025311419982fba32f545dec21 : SUCCESS in 57m 51s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/039961897fc24255af1f449d19505ee2 : SUCCESS in 35m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c70c1fb0ca8c4b7988c0a172a13681ae : SUCCESS in 53m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/595b922b367b416194487925ed106a4c : FAILURE in 10m 52s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"4380203ebe840a65a055ff5e55665d1f2c65b4ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-02-13 15:38:28.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"a503bd9bb3295a40e6fe3772e1b227147ddf482b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-13 17:12:43.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/25/check/DellEMC_PowerFlex_Manila_dhss_false/de3e909/DellEMC_PowerFlex_Manila_dhss_false/814 : SUCCESS in 32m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":25},{"id":"bc1b7faf4984ee46bd821dfdd9c05c136f784b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 18:17:52.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1e55f74a1144e85a95b7db25904ef2d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/014522f2a2c34ef5b99dd53875162bae : SUCCESS in 4m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e061eebfa81940e98ce9a1f55ed45efa : SUCCESS in 1h 01m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3e417b07ddb94311b3d66fc23a10a77e : SUCCESS in 2h 32m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/343b5cf3ccb2478da8e53965ae0e8b11 : SUCCESS in 1h 23m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8016ac9d2fde4d8a8ae0b36ea1b1310f : SUCCESS in 1h 21m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d241e0fca3f0449eba821dff9e93462d : FAILURE in 1h 49m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ff950bc82c584a1f87d714bad7e582b6 : SUCCESS in 1h 02m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c18dd3a29b2c48418e146674917ad670 : FAILURE in 34m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f3042d20b1634f9ab01d103d6904891f : SUCCESS in 49m 28s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/064a888ffafa45279dba5cb3a9ec53d7 : FAILURE in 19m 29s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d9d723e5bf2040c5a5fa61660cb44d22 : SUCCESS in 1h 07m 26s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/f449c7091c6445e9962601110a67a7f1 : SUCCESS in 38m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0911786c8338421d89b7e1eb9729d786 : FAILURE in 1h 03m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/af3c66368b5743babac2217c122c37cd : FAILURE in 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"d8feecde31afefbfa6dc9cfc5a8037b6d992b204","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-14 03:37:04.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/25/check/DellEMC_Unity_Manila_dhss_true/c836d22/DellEMC_Unity_Manila_dhss_true/813 : SUCCESS in 3h 41m 13s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/25/check/DellEMC_Unity_Manila_dhss_false/87560dc/DellEMC_Unity_Manila_dhss_false/869 : SUCCESS in 49m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":25},{"id":"6d8d18e011a5240e1942117e71681fc785a3154b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-02-14 05:41:21.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"cbf7d526f2926b57f5eff0f0ebc892ad1af01186","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-14 06:13:50.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/26/check/DellEMC_PowerFlex_Manila_dhss_false/5a34844/DellEMC_PowerFlex_Manila_dhss_false/816 : SUCCESS in 32m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":26},{"id":"8ebc0ed33c3cbd3e9db4fa4be6b5f9100f49eb39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 07:58:13.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/062ae71b7da04d86b84878ac56ce7730\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d18fb8528d9249d6bfed5b06593a9282 : SUCCESS in 2m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c1337874fd6044488d6118e1810cbfb4 : SUCCESS in 1h 17m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9677c1e51a5b406892ef34e7f09dabd9 : SUCCESS in 2h 11m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/590b436cff1046f39fc288e73a0b983c : SUCCESS in 1h 23m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1923beae6c7b41fa94b043bcc074752d : SUCCESS in 1h 37m 02s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1d6811fcf3c848359eb196b221911387 : SUCCESS in 1h 35m 00s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2a479ac9edac48a0bf4d97fe14212dff : SUCCESS in 1h 11m 03s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3387c03fe044492dbabd01f792ee9623 : FAILURE in 40m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a2eb3aa059b8401aa155fc5506e4bcab : SUCCESS in 45m 44s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/80940b0e741e4f05ab2058fc7207a01b : SUCCESS in 1h 13m 32s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/19c20e855cb143f690630f721b8721eb : SUCCESS in 1h 21m 45s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/744d8bd1d461422095dbdc7906fc4d80 : SUCCESS in 1h 13m 40s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/98d9a897892144b3b0f77ae82316f6cf : SUCCESS in 34m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6369b7f303ba49fa83bf0d0d595a05d6 : SUCCESS in 56m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4ce7812041134eb990ac7a1d8a720641 : FAILURE in 20m 52s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"717cfe0ba548b03761bcc74bfb15451dafd5a807","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-14 08:10:43.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/26/check/DellEMC_Unity_Manila_dhss_true/81a713b/DellEMC_Unity_Manila_dhss_true/815 : SUCCESS in 2h 29m 07s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/26/check/DellEMC_Unity_Manila_dhss_false/b2fb7ae/DellEMC_Unity_Manila_dhss_false/871 : SUCCESS in 55m 01s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":26},{"id":"b8f230110af21347100edb167aa4d1bd08acac5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-02-14 08:12:40.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"5e98843a0d41e86ca95886274a94e1a19488a045","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-14 08:42:34.000000000","message":"Patch Set 27:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/27/check/DellEMC_PowerFlex_Manila_dhss_false/18acdd1/DellEMC_PowerFlex_Manila_dhss_false/817 : SUCCESS in 29m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":27},{"id":"7afcffdaf6652630ce66e9f4c65722c7ace9891f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-14 11:27:06.000000000","message":"Patch Set 27:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/27/check/DellEMC_Unity_Manila_dhss_true/88c532a/DellEMC_Unity_Manila_dhss_true/816 : SUCCESS in 3h 14m 12s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/27/check/DellEMC_Unity_Manila_dhss_false/368def3/DellEMC_Unity_Manila_dhss_false/872 : SUCCESS in 54m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":27},{"id":"bc97910fb3991b8607ff66a801100b6769fac069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 11:45:30.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f926c8d737374427a8fa06889fa71818\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d8e2b06f80144188a8b1cf7d4c51f3b : SUCCESS in 3m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b5bef295ea16417dbfb9f2a8d1539569 : SUCCESS in 1h 06m 47s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/60c5a4e111304e788bfad08f3fae845e : TIMED_OUT in 3h 03m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9a7c8e43de2f487db73c695e6fa391e0 : SUCCESS in 1h 16m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7f5840341fba4554ad84a734abd3628e : SUCCESS in 1h 01m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f98d55a5eea84f3091b958a677396f4f : SUCCESS in 1h 33m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2aa4cb614f3a4f9bba2570338edcf178 : SUCCESS in 1h 10m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/93ee51cd52064546abbe9ab72b9ca2da : SUCCESS in 53m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1638635f412f4306b3a778617c78945f : SUCCESS in 41m 46s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/eeec9c1a1a9a4c5e8ee2f62a378e2433 : SUCCESS in 1h 22m 02s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/2c75c1da9bde430c8fdc415fefd93647 : SUCCESS in 1h 01m 59s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/108b3e8569214576a096c2423ae45877 : SUCCESS in 1h 22m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d77f75b831ca454fbe97c620f75cd639 : SUCCESS in 36m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/55d6b5c207cf4878b239b29cea1e65a2 : SUCCESS in 49m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7e628df9c1004ffabd162c50736c998a : FAILURE in 22m 28s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"1730186bbef119a0581adda94b842f0cb64506af","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-14 12:15:17.000000000","message":"Patch Set 27:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/27/check/DellEMC_PowerFlex_Manila_dhss_false/65099bd/DellEMC_PowerFlex_Manila_dhss_false/818 : SUCCESS in 29m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":27},{"id":"99a3ac760a664383aabd474909bb7cebffdb85d0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-14 13:57:45.000000000","message":"Patch Set 27:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/27/check/DellEMC_Unity_Manila_dhss_true/26d6932/DellEMC_Unity_Manila_dhss_true/817 : SUCCESS in 2h 12m 01s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/27/check/DellEMC_Unity_Manila_dhss_false/97b5d3c/DellEMC_Unity_Manila_dhss_false/873 : SUCCESS in 59m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":27},{"id":"9bd01208f7b4a48736f0c91f8573b9b8f7895210","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-14 14:50:36.000000000","message":"Patch Set 27:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/841015/27/upstream-check/manila-tempest-plugin-ontap-dhss/375077e/ : POST_FAILURE in 3h 04m 08s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/841015/27/upstream-check/manila-tempest-plugin-ontap-no-dhss/9741b0d/ : SUCCESS in 1h 48m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":27},{"id":"fc0192c7b9d6eb644efb08630b996cc42d7fb1f8","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-02-16 18:02:13.000000000","message":"Patch Set 27: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":27},{"id":"ab14baa95e54cc3c9b1c2424479c748fb6c2b6b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-02-22 09:02:38.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"f776610435c8e038899deed99e811ac48cd4ed90","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-02-22 09:02:39.000000000","message":"Patch Set 27:\n\n(2 comments)","accounts_in_message":[],"_revision_number":27},{"id":"cddd3ff7bf157af1b8c3ecc43a7392aee0df3c04","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-22 09:58:16.000000000","message":"Patch Set 28:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/28/check/DellEMC_PowerFlex_Manila_dhss_false/5ad8ac9/DellEMC_PowerFlex_Manila_dhss_false/865 : SUCCESS in 30m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":28},{"id":"3dce8a298e5850b0ef64ecfb5b55f822fdd9a688","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 11:24:50.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4100a18539fc47c2b912f015e3699417\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06a0b1df3d8e49c98030ba659a3d6828 : SUCCESS in 3m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/80f1e9fc29a345628ff1fbfd42251ea2 : SUCCESS in 1h 14m 11s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f86ee7b927984004b802aee80aaf1297 : SUCCESS in 2h 20m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aad49df1b20341998a7b277797d269f4 : SUCCESS in 1h 22m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/eaf74c05168445c7983f0561cbdc177d : SUCCESS in 1h 19m 11s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3048087d102b46dc951ebed8735ea275 : SUCCESS in 1h 43m 09s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/baa1380974a74bd0b6e825b8b7085ad0 : FAILURE in 1h 31m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aaf0eba81a1a4f14ab87cb74ed80a698 : SUCCESS in 42m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b38d71e460b64bf0bd4498ca41acd02d : SUCCESS in 49m 02s\n- manila-tempest-plugin-lvm-bobcat https://zuul.opendev.org/t/openstack/build/1a9dafa9e73d4b50870fa3e9c61cd3c5 : SUCCESS in 1h 23m 25s\n- manila-tempest-plugin-lvm-antelope https://zuul.opendev.org/t/openstack/build/7c8320fa7b314b5ba12fd1ac768c64ea : SUCCESS in 1h 27m 23s\n- manila-tempest-plugin-lvm-zed https://zuul.opendev.org/t/openstack/build/acc5749cb19f490d9d00ea6bcae4c3b9 : SUCCESS in 1h 23m 48s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/94eeaba4b6ad4ea8a07b0eab29f332fa : SUCCESS in 28m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/128667cace454f198852cbfc4fcfc8da : SUCCESS in 57m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/baf71f8e10804dd28840702015edba1f : FAILURE in 14m 17s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"ee85919514dac123dc43661fbb1b4f7b7db0c55e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-02-22 11:55:14.000000000","message":"Patch Set 28:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/841015/28/check/DellEMC_Unity_Manila_dhss_true/9c3bc53/DellEMC_Unity_Manila_dhss_true/855 : SUCCESS in 2h 43m 36s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/28/check/DellEMC_Unity_Manila_dhss_false/7bfb471/DellEMC_Unity_Manila_dhss_false/911 : SUCCESS in 58m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":28},{"id":"d64e5b412abd5ad682c568816107d4f809df37ef","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-02-22 11:56:25.000000000","message":"Patch Set 28:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/841015/28/check/DellEMC_PowerFlex_Manila_dhss_false/40453ea/DellEMC_PowerFlex_Manila_dhss_false/866 : SUCCESS in 31m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":28},{"id":"78839feecbe82b495897985517b15e77d7d12caa","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-02-22 16:55:20.000000000","message":"Patch Set 28:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/841015/28/upstream-check/manila-tempest-plugin-ontap-dhss/8d4c6c4/ : TIMED_OUT in 3h 15m 21s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/841015/28/upstream-check/manila-tempest-plugin-ontap-no-dhss/53661f9/ : TIMED_OUT in 3h 10m 17s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":28},{"id":"55a11649aee77b1f858f41c0dd365b4ddfa658ed","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-02-28 21:31:35.000000000","message":"Patch Set 28: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"6076d32d829755e045c4f4d03a0c2dd827b32e51","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-03-01 11:35:34.000000000","message":"Patch Set 28: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"8f2fead349300c35bc22272a074fd3e635d15527","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-03-06 07:23:54.000000000","message":"Patch Set 28: Code-Review+2","accounts_in_message":[],"_revision_number":28},{"id":"7abfb36ce621683f18545610da96c890aceb8e43","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2024-03-06 07:24:10.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"4ee3f63bdf4d2b5121bab2be6ae520f35cd342dc","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-03-06 12:09:08.000000000","message":"Patch Set 28: Workflow+1","accounts_in_message":[],"_revision_number":28},{"id":"cf950e4365ab528e7b95c68dcb2afe6e8dc6b03f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 12:11:21.000000000","message":"Patch Set 28: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":28},{"id":"fa793f35be27b76cd5953e0068a3914c19650197","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 13:40:28.000000000","message":"Patch Set 28: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a851acbb3ce940aba029ad7843f11275\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f3e7ba2f8794b0b9df095a5da89aaac : SUCCESS in 3m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e16beb3b9ffc4f99b9fe7765c44f03c9 : SUCCESS in 1h 17m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4cbf8e9e89eb49a7af3d2f5121c82d83 : SUCCESS in 39m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/501666bb5b0b4f1786d3a8dd96af3eb7 : SUCCESS in 47m 40s","accounts_in_message":[],"_revision_number":28},{"id":"ed9b449a7388cb42f3223c3a93e2f6c016eac4ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 13:40:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":28}],"current_revision_number":28,"current_revision":"edd82c73856bbce856e2d69be840062ae352c23b","revisions":{"3720e6dafa4eb1b7f6ef288336c30d0202cb2ef1":{"kind":"REWORK","_number":1,"created":"2022-05-07 15:20:36.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/1"}}},"commit":{"parents":[{"commit":"c0e182601b83bf7aa708a12336a96c37d75da530","subject":"Merge \"Remove _disassociate_floating_ip \u0026 _create_network\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c0e182601b83bf7aa708a12336a96c37d75da530"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 15:19:45.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: I7d10cf47864cd21932315375d84dc728ff738f23\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3720e6dafa4eb1b7f6ef288336c30d0202cb2ef1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3720e6dafa4eb1b7f6ef288336c30d0202cb2ef1"}]},"branch":"refs/heads/master"},"9e8eabeb70f26c3070fda65dd9af6be428434d58":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-29 14:37:32.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/2"}}},"commit":{"parents":[{"commit":"a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab","subject":"Define queues at project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-29 14:34:55.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: I7d10cf47864cd21932315375d84dc728ff738f23\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9e8eabeb70f26c3070fda65dd9af6be428434d58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9e8eabeb70f26c3070fda65dd9af6be428434d58"}]},"branch":"refs/heads/master"},"0a6c1194543bebe47a685c2eb221815ee3c1e25e":{"kind":"REWORK","_number":3,"created":"2023-06-16 09:59:31.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/3"}}},"commit":{"parents":[{"commit":"9cefa379a5b482f19e4f821986b1d92ec5ac84c5","subject":"Merge \"[CI] Fix cephfs-nfs job by running on centos-stream-9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9cefa379a5b482f19e4f821986b1d92ec5ac84c5"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-06-16 09:59:16.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: I7d10cf47864cd21932315375d84dc728ff738f23\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a6c1194543bebe47a685c2eb221815ee3c1e25e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a6c1194543bebe47a685c2eb221815ee3c1e25e"}]},"branch":"refs/heads/master"},"e2fbc82ccf79c02b5ce61670c7a099014f8db1d1":{"kind":"REWORK","_number":4,"created":"2023-07-06 12:28:35.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/4"}}},"commit":{"parents":[{"commit":"9cefa379a5b482f19e4f821986b1d92ec5ac84c5","subject":"Merge \"[CI] Fix cephfs-nfs job by running on centos-stream-9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9cefa379a5b482f19e4f821986b1d92ec5ac84c5"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-07-06 12:28:10.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: I7d10cf47864cd21932315375d84dc728ff738f23\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e2fbc82ccf79c02b5ce61670c7a099014f8db1d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e2fbc82ccf79c02b5ce61670c7a099014f8db1d1"}]},"branch":"refs/heads/master"},"33f8323c9e9e845ff7dd1babbded8b6fbcd7cc92":{"kind":"REWORK","_number":5,"created":"2023-07-17 15:53:37.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/5"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-07-17 15:53:25.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: I7d10cf47864cd21932315375d84dc728ff738f23\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/33f8323c9e9e845ff7dd1babbded8b6fbcd7cc92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/33f8323c9e9e845ff7dd1babbded8b6fbcd7cc92"}]},"branch":"refs/heads/master"},"200b76a2f1f92de7ac8526b46c44c137a282a050":{"kind":"REWORK","_number":6,"created":"2023-07-18 08:39:45.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/6"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-07-18 08:38:52.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: I7d10cf47864cd21932315375d84dc728ff738f23\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/200b76a2f1f92de7ac8526b46c44c137a282a050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/200b76a2f1f92de7ac8526b46c44c137a282a050"}]},"branch":"refs/heads/master"},"36b28d2ea106b7d9dd81b2976f75f1616ba1a013":{"kind":"REWORK","_number":7,"created":"2023-07-18 10:30:53.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/7"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-07-18 10:30:43.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: I7d10cf47864cd21932315375d84dc728ff738f23\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/36b28d2ea106b7d9dd81b2976f75f1616ba1a013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/36b28d2ea106b7d9dd81b2976f75f1616ba1a013"}]},"branch":"refs/heads/master"},"fc24f2379c32347e41f70db1162db8d88b47af63":{"kind":"REWORK","_number":8,"created":"2023-07-18 12:23:01.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/8"}}},"commit":{"parents":[{"commit":"782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb","subject":"[RBAC] Wait for access rules to be applied prior to deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/782a54dbd3c2f38dcb460f24cfdf0407fc37f3bb"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-07-18 12:16:27.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: I7d10cf47864cd21932315375d84dc728ff738f23\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc24f2379c32347e41f70db1162db8d88b47af63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fc24f2379c32347e41f70db1162db8d88b47af63"}]},"branch":"refs/heads/master"},"00b3fe15ff76ffdc0269c2d4917c33ca00bd6e54":{"kind":"REWORK","_number":9,"created":"2023-08-09 11:57:31.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/9"}}},"commit":{"parents":[{"commit":"68fc053a27ea8449d68c7b0fbb4dd3842a09a199","subject":"Merge \"Fixed multiple subnet negative tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/68fc053a27ea8449d68c7b0fbb4dd3842a09a199"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-09 11:57:22.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: I7d10cf47864cd21932315375d84dc728ff738f23\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/00b3fe15ff76ffdc0269c2d4917c33ca00bd6e54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/00b3fe15ff76ffdc0269c2d4917c33ca00bd6e54"}]},"branch":"refs/heads/master"},"fd8c9d978bb022d83ce80e6ff3f38605e92a7341":{"kind":"REWORK","_number":10,"created":"2023-08-09 14:32:57.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/10"}}},"commit":{"parents":[{"commit":"68fc053a27ea8449d68c7b0fbb4dd3842a09a199","subject":"Merge \"Fixed multiple subnet negative tempest test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/68fc053a27ea8449d68c7b0fbb4dd3842a09a199"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-09 14:32:51.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: I7d10cf47864cd21932315375d84dc728ff738f23\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fd8c9d978bb022d83ce80e6ff3f38605e92a7341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fd8c9d978bb022d83ce80e6ff3f38605e92a7341"}]},"branch":"refs/heads/master"},"cf4b596f5717aeac0adef849a087a251f7ec1c6b":{"kind":"REWORK","_number":11,"created":"2023-08-18 14:01:18.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/11"}}},"commit":{"parents":[{"commit":"8d4a2decf09b91787515515294063d703e5c2a2a","subject":"Merge \"Remove Glance v1 APIs tests code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d4a2decf09b91787515515294063d703e5c2a2a"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-18 13:58:50.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: Ice01ab7892b1eb52b3202f2c79957977f73f3aca\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf4b596f5717aeac0adef849a087a251f7ec1c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf4b596f5717aeac0adef849a087a251f7ec1c6b"}]},"branch":"refs/heads/master"},"2679fdedb04f351cac817ffc85984bb64c195ab1":{"kind":"REWORK","_number":12,"created":"2023-08-18 14:03:37.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/12"}}},"commit":{"parents":[{"commit":"8d4a2decf09b91787515515294063d703e5c2a2a","subject":"Merge \"Remove Glance v1 APIs tests code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d4a2decf09b91787515515294063d703e5c2a2a"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-18 14:03:26.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: Ice01ab7892b1eb52b3202f2c79957977f73f3aca\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2679fdedb04f351cac817ffc85984bb64c195ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2679fdedb04f351cac817ffc85984bb64c195ab1"}]},"branch":"refs/heads/master"},"4f753ab31086f3bd4e2ebb410ec260bd6776755b":{"kind":"REWORK","_number":13,"created":"2023-08-21 16:14:16.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/13"}}},"commit":{"parents":[{"commit":"8d4a2decf09b91787515515294063d703e5c2a2a","subject":"Merge \"Remove Glance v1 APIs tests code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d4a2decf09b91787515515294063d703e5c2a2a"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-21 13:09:47.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: Ice01ab7892b1eb52b3202f2c79957977f73f3aca\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4f753ab31086f3bd4e2ebb410ec260bd6776755b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4f753ab31086f3bd4e2ebb410ec260bd6776755b"}]},"branch":"refs/heads/master"},"a3c04fa7c767c9fc359d348d9eee150cb2e2ecc7":{"kind":"REWORK","_number":14,"created":"2023-08-22 16:10:47.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/14"}}},"commit":{"parents":[{"commit":"8d4a2decf09b91787515515294063d703e5c2a2a","subject":"Merge \"Remove Glance v1 APIs tests code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d4a2decf09b91787515515294063d703e5c2a2a"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-22 14:19:10.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: Ice01ab7892b1eb52b3202f2c79957977f73f3aca\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a3c04fa7c767c9fc359d348d9eee150cb2e2ecc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a3c04fa7c767c9fc359d348d9eee150cb2e2ecc7"}]},"branch":"refs/heads/master"},"9d0517750e3214d72b794d14063c7b29ddbcc2b2":{"kind":"REWORK","_number":15,"created":"2023-08-26 14:07:23.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/15"}}},"commit":{"parents":[{"commit":"8d4a2decf09b91787515515294063d703e5c2a2a","subject":"Merge \"Remove Glance v1 APIs tests code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d4a2decf09b91787515515294063d703e5c2a2a"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-26 14:07:15.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: Ice01ab7892b1eb52b3202f2c79957977f73f3aca\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9d0517750e3214d72b794d14063c7b29ddbcc2b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9d0517750e3214d72b794d14063c7b29ddbcc2b2"}]},"branch":"refs/heads/master"},"ee3f269b013daf2557459e037aca1c3346fe54cd":{"kind":"REWORK","_number":16,"created":"2023-08-28 10:28:32.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/16"}}},"commit":{"parents":[{"commit":"8d4a2decf09b91787515515294063d703e5c2a2a","subject":"Merge \"Remove Glance v1 APIs tests code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d4a2decf09b91787515515294063d703e5c2a2a"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-28 10:28:23.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: Ice01ab7892b1eb52b3202f2c79957977f73f3aca\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee3f269b013daf2557459e037aca1c3346fe54cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee3f269b013daf2557459e037aca1c3346fe54cd"}]},"branch":"refs/heads/master"},"7ca626cef56bb01f0d3958be3b59c011c4075456":{"kind":"REWORK","_number":17,"created":"2023-08-28 14:04:59.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/17"}}},"commit":{"parents":[{"commit":"8d4a2decf09b91787515515294063d703e5c2a2a","subject":"Merge \"Remove Glance v1 APIs tests code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8d4a2decf09b91787515515294063d703e5c2a2a"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-28 14:04:49.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: Ice01ab7892b1eb52b3202f2c79957977f73f3aca\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7ca626cef56bb01f0d3958be3b59c011c4075456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7ca626cef56bb01f0d3958be3b59c011c4075456"}]},"branch":"refs/heads/master"},"765bb829d5ef1698fca12189bb77d39b297f9858":{"kind":"REWORK","_number":18,"created":"2023-08-28 17:19:03.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/18"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-28 17:16:22.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: Ice01ab7892b1eb52b3202f2c79957977f73f3aca\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/765bb829d5ef1698fca12189bb77d39b297f9858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/765bb829d5ef1698fca12189bb77d39b297f9858"}]},"branch":"refs/heads/master"},"da9dde5ae25fe450dabeaaed62bd619807d6342d":{"kind":"REWORK","_number":19,"created":"2023-08-29 10:08:06.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/19"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-29 10:07:43.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: Ice01ab7892b1eb52b3202f2c79957977f73f3aca\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da9dde5ae25fe450dabeaaed62bd619807d6342d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/da9dde5ae25fe450dabeaaed62bd619807d6342d"}]},"branch":"refs/heads/master"},"e6dd4066bdd87163d19dc100f10e4070e37071b7":{"kind":"REWORK","_number":20,"created":"2023-08-29 12:01:44.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/20"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-29 11:59:18.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: Ice01ab7892b1eb52b3202f2c79957977f73f3aca\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e6dd4066bdd87163d19dc100f10e4070e37071b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e6dd4066bdd87163d19dc100f10e4070e37071b7"}]},"branch":"refs/heads/master"},"eac1e7c214f770167b5d13c509dbee3f3fb3cd9b":{"kind":"REWORK","_number":21,"created":"2023-08-29 16:27:14.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/21"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-08-29 16:27:06.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nDepends-On: Ice01ab7892b1eb52b3202f2c79957977f73f3aca\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eac1e7c214f770167b5d13c509dbee3f3fb3cd9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/eac1e7c214f770167b5d13c509dbee3f3fb3cd9b"}]},"branch":"refs/heads/master"},"82c4ef086caf21f07c8f21c9267e1ffa506aafd9":{"kind":"NO_CODE_CHANGE","_number":22,"created":"2023-09-07 13:12:57.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/22"}}},"commit":{"parents":[{"commit":"b0ab2128cb84c161a783587aff0637819bd4c8c7","subject":"Merge \"Test resource locks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b0ab2128cb84c161a783587aff0637819bd4c8c7"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-09-07 13:12:39.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82c4ef086caf21f07c8f21c9267e1ffa506aafd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/82c4ef086caf21f07c8f21c9267e1ffa506aafd9"}]},"branch":"refs/heads/master"},"d28fba26a51b88b235317bb0e76a0708b82077a1":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2023-10-10 13:43:02.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/23"}}},"commit":{"parents":[{"commit":"dae4d135da2c7c9cd223db5379e04ccce61f8037","subject":"Merge \"Add share transfer test.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dae4d135da2c7c9cd223db5379e04ccce61f8037"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-10-10 13:42:48.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d28fba26a51b88b235317bb0e76a0708b82077a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d28fba26a51b88b235317bb0e76a0708b82077a1"}]},"branch":"refs/heads/master"},"4a64366766d696f9750264e69b5b41cfec49ec1e":{"kind":"REWORK","_number":24,"created":"2024-02-13 08:51:55.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/24"}}},"commit":{"parents":[{"commit":"d78f3b9dc8f4c23fadc32933fd131614b975fa75","subject":"Merge \"Remove unused babel.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d78f3b9dc8f4c23fadc32933fd131614b975fa75"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-02-13 08:51:04.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a64366766d696f9750264e69b5b41cfec49ec1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4a64366766d696f9750264e69b5b41cfec49ec1e"}]},"branch":"refs/heads/master"},"d48751461a79e3366b05b12e7e18cffab5f4095e":{"kind":"REWORK","_number":25,"created":"2024-02-13 15:38:28.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/25"}}},"commit":{"parents":[{"commit":"d78f3b9dc8f4c23fadc32933fd131614b975fa75","subject":"Merge \"Remove unused babel.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d78f3b9dc8f4c23fadc32933fd131614b975fa75"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-02-13 15:38:08.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d48751461a79e3366b05b12e7e18cffab5f4095e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d48751461a79e3366b05b12e7e18cffab5f4095e"}]},"branch":"refs/heads/master"},"59c741b7334ec405e2edbb72ea2a7118d8d0f7fb":{"kind":"REWORK","_number":26,"created":"2024-02-14 05:41:21.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/26"}}},"commit":{"parents":[{"commit":"d78f3b9dc8f4c23fadc32933fd131614b975fa75","subject":"Merge \"Remove unused babel.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d78f3b9dc8f4c23fadc32933fd131614b975fa75"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-02-14 05:36:48.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/59c741b7334ec405e2edbb72ea2a7118d8d0f7fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/59c741b7334ec405e2edbb72ea2a7118d8d0f7fb"}]},"branch":"refs/heads/master"},"dc75f2728a3c6c8d959b1c89d65f7a38d6348479":{"kind":"REWORK","_number":27,"created":"2024-02-14 08:12:40.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/27"}}},"commit":{"parents":[{"commit":"d78f3b9dc8f4c23fadc32933fd131614b975fa75","subject":"Merge \"Remove unused babel.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d78f3b9dc8f4c23fadc32933fd131614b975fa75"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-02-14 08:12:33.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc75f2728a3c6c8d959b1c89d65f7a38d6348479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc75f2728a3c6c8d959b1c89d65f7a38d6348479"}]},"branch":"refs/heads/master"},"edd82c73856bbce856e2d69be840062ae352c23b":{"kind":"REWORK","_number":28,"created":"2024-02-22 09:02:38.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/15/841015/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/15/841015/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/15/841015/28"}}},"commit":{"parents":[{"commit":"7604da805417aa12f4574458b05f10ffb31693dd","subject":"Remove unneeded randomness when calling ddt.data()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7604da805417aa12f4574458b05f10ffb31693dd"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2022-05-07 14:48:36.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-02-22 09:01:46.000000000","tz":0},"subject":"Add share backup tests.","message":"Add share backup tests.\n\nShare backup tests i.e. create/delete/get share_backups.\n\nChange-Id: I366221702b3aa2e78bff82e6b2a543b6f6784f77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/edd82c73856bbce856e2d69be840062ae352c23b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/edd82c73856bbce856e2d69be840062ae352c23b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
