)]}'
{"id":"openstack%2Fmanila-tempest-plugin~842216","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I5c52c2ee589083e7baf2987c42568282afb1f883","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1936689","attention_set":{"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2023-03-01 21:58:09.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5c52c2ee589083e7baf2987c42568282afb1f883","subject":"[WIP] share group type create validation check","status":"NEW","created":"2022-05-17 18:38:51.000000000","updated":"2024-07-11 20:14:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9d362e565aa16f51850b8304f3cfcf58b76aecaf","_number":842216,"virtual_id_number":842216,"owner":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-22 20:39:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-17 20:28:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-17 20:53:26.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-09-21 16:33:38.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"66ad1814ade44cbecf867f9f9260be9159f9e3f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-05-17 18:38:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"632ffe4f8f6229a6275fed5f077409305966c2d7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-05-17 18:40:04.000000000","message":"Topic set to yoga-m3-bugsquash","accounts_in_message":[],"_revision_number":1},{"id":"b141ee5945dbaf9723e853ab3c0570af2484beff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 20:28:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffb62cc743604fb1aa1708ce44e5781a : SUCCESS in 3m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/eeffc63602004a49af7425b592c56b5d : FAILURE in 29m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c633bf735c554aa79b055e75cb95566e : FAILURE in 32m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9e7be02fd0aa49038bfd4ccde5b2ddfa : FAILURE in 1h 10m 29s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/dd789cd293604bce841aa40ad716d8e9 : FAILURE in 59m 19s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c7baa8b233934734be39bd8e92ed0851 : FAILURE in 1h 21m 18s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/706aa23eed584f4cb3808fb1228b7d0d : FAILURE in 1h 05m 01s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b69a9b86a793457dad1ee03ecc6a16e6 : FAILURE in 1h 06m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0bfd9deb3d854481880b7e9fae9bb4bc : FAILURE in 1h 08m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/14a8d12f274e4c47a5b9da6e7e066d50 : FAILURE in 1h 47m 46s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a05c6d1a0ae449babcedc6483d7e1a2a : FAILURE in 44m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ebfce463661141c9a39ec32cdeae3e80 : FAILURE in 1h 42m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b1948efe12f044fb9a080363b3877156 : FAILURE in 18m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54d04c5098599ac6f8ebe0bd90f7b284d1a15670","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-17 20:53:26.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/16/842216/1/check/DellEMC_Unity_Manila_dhss_true/4eabbc8/DellEMC_Unity_Manila_dhss_true/481 : FAILURE in 2h 14m 20s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/16/842216/1/check/DellEMC_Unity_Manila_dhss_false/19d46f0/DellEMC_Unity_Manila_dhss_false/502 : FAILURE in 1h 03m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"b3c81ef5501997119e9880f4df102745e2e12e10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-05-18 17:29:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"74ff89f919ddd34be43d95a6cb823b59603dc42b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 19:26:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15b74ffa36fa4892b2db8300b041d10b : SUCCESS in 3m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/77948edff9304354a92a684fdc0a6394 : FAILURE in 27m 31s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7f840b877edb42bb9cade7cb29a1a8c4 : FAILURE in 42m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/91c7bbcb14bc453b844221cb66ff645e : FAILURE in 1h 29m 42s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/80d325cee63b4e5290da69cac8ca6558 : FAILURE in 1h 10m 32s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/29d64407a73840bc82961698d0cb5cc9 : FAILURE in 1h 34m 43s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/b494f810b1874765bad99c2552198ac4 : FAILURE in 1h 18m 33s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c68649a915c44448a6842ed6d8e2168e : FAILURE in 1h 26m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/909e9e1e15554d44a389abf8a24943f4 : SUCCESS in 1h 34m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/864fc772494e405da0fc310b231cc4e4 : FAILURE in 1h 51m 43s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a7f8b92c30de4cebac472461da30bcf3 : SUCCESS in 1h 03m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c3338b44f7c0401a9b1d640dc8f8a377 : FAILURE in 1h 34m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/477da20920b8488faf27ce590ba675e7 : SUCCESS in 23m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b985612e099647a1cfa43cf3eb98f473b18d6a6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-18 20:29:13.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/16/842216/2/check/DellEMC_Unity_Manila_dhss_true/8190020/DellEMC_Unity_Manila_dhss_true/483 : FAILURE in 2h 59m 52s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/16/842216/2/check/DellEMC_Unity_Manila_dhss_false/b6a86d9/DellEMC_Unity_Manila_dhss_false/504 : SUCCESS in 1h 02m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"442f425c725c3a02d06fe5d1794b48e2b62b5739","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-09-13 16:21:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"088c57f1aeb9b7709da24aaa0d8fe23c75303e67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 19:31:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebf5328b9dc3406496c3ef8347508523\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c31128bc4d61446890fc3cd49550e4ea : SUCCESS in 3m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/40983098569641a586e51ccfc2db386d : FAILURE in 1h 03m 09s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5cc8762789c54b67884c75411ae1c1bb : FAILURE in 3h 02m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0cb62b78b7544bb0a9b9c84e96d4c751 : SUCCESS in 56m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f4c7ef477fdf4302a3f50fb6a4c484ed : SUCCESS in 1h 48m 36s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c3124a11f2d2483da992e6212cdf91bf : FAILURE in 2h 21m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f3da0bbd23554569832e8ffd97ea8bba : FAILURE in 1h 28m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/293c329aca524b1f9d70112053674e54 : FAILURE in 34m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8af32819d1a1478182b615e6a9c74dce : FAILURE in 37m 31s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/780b38adb13045fd8bbe6b5fd2fd8ca6 : FAILURE in 1h 25m 23s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/558817ec8189435189dde3ca1470a3bc : FAILURE in 1h 31m 11s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e74e6ad2238e4acf838b8842fa9ebb23 : FAILURE in 1h 12m 07s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cd2b59963a9c4c4c9336e7c57e8dc62b : SUCCESS in 54m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7c66361645414b50a7f94a82a972cf87 : SUCCESS in 29m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a9bd8f3f6bac7d6ef05684408d88f11852f16d56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-09-15 15:49:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6b9fbe0a80c9ebabee55d40ae967f771ae280197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2022-09-15 18:01:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"db1d192b2cf2e737dc910a90afc90af2d4ecb861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 20:23:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/650d6a485f0d4820a21f917cc667c468\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2be6653236ba4443a0440400efa1daa7 : SUCCESS in 2m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6261deb2997b4cb3adfe5ab428773924 : FAILURE in 1h 21m 08s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e04721e1fd8c4e319b9ca6595029741a : SUCCESS in 2h 16m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/cce49e16a03e44af90087cbfe683721f : SUCCESS in 1h 35m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8bae701f6e064f9a9abc68b4f659ce1a : SUCCESS in 1h 38m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/eeeee01e7994477092a5ff4388ddcc0f : FAILURE in 1h 08m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cf930ac96f9b4227b0136287150b720b : FAILURE in 1h 35m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/941f3587e41341619613485c256316c2 : FAILURE in 37m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2d4b4771f78641819aa4b2d25fd153b5 : FAILURE in 48m 11s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/4cf686fb75924d96b75e3a798540220f : FAILURE in 1h 18m 32s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5fd148c257074d728999d12167c7e36b : FAILURE in 1h 38m 53s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/205c7e072f6342169cb13010a6964938 : FAILURE in 1h 22m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2bc1049ca75e4af2a68e5bbd382aa852 : SUCCESS in 57m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a0fbfa57a2e24acfa60fc38298315548 : SUCCESS in 34m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"10aa54a3953a43c0779f2ee7a2b5ffd2f97e51db","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-03-01 18:45:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f54ab04790cae4f73cee67fba4b8cb6ac4c05012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 21:58:09.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4790c7980374c379e04a7ec65bad486\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c32bb191f654aa596b970a75e39f5d6 : SUCCESS in 3m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0fe34326afb94e11896b52abf4021848 : FAILURE in 1h 19m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3118ff5a320d4dce817c613cd05c4123 : SUCCESS in 2h 13m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8512dab9ae444b6eb76c761333965e3a : TIMED_OUT in 3h 02m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/df846edb7b9b4ebea5e46616d7c1733a : FAILURE in 1h 50m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b34e78e925d8452db1d71bd23a4d2a1f : FAILURE in 1h 48m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9f473a753efc4607a847c23352c7df10 : FAILURE in 1h 18m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6e08e231b7224b0d95c6f130ec8a7e7e : FAILURE in 26m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b41417e3097b4988aa3308a36f1d17e9 : FAILURE in 32m 53s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/35d192dcf88142da8043eb8f0f3d2b67 : FAILURE in 1h 40m 35s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3103e870fe684cb4a31f4448da752b5e : FAILURE in 1h 10m 22s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8ad0a00543f845abbe89015db29d3749 : SUCCESS in 25m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c77d24d952e24061bad3864880e62e60 : SUCCESS in 42m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/38c0388ac31c4e9eac01605985fb9aa3 : SUCCESS in 22m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9b9fe666a0d0f239a4b00a560308533febb08bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-09-19 15:04:06.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"3aaef6cc8de454b046aec89e70aa0e376b6bd1dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 16:58:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/01d58a0cd0f448c4bb6ba049bff8c6ba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b4127b7edd643fc9112fbbc4608c164 : SUCCESS in 2m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9c46057c4bad47a3a011a1ac90f64113 : FAILURE in 47m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b53baacc4ae8443ba57b804232ba7b76 : SUCCESS in 1h 52m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/dbc756490be34fe88a2e0fae62dee2c8 : SUCCESS in 1h 02m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f0cdee55f4574f1f82d42d1b79f98417 : FAILURE in 1h 19m 11s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ee920563d94d4466b5af10410bf5a87c : SUCCESS in 1h 15m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d7c7bd4ea5604ef5b75154857a58b970 : FAILURE in 54m 59s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8fbfe65b701a401da6e112aa6fee3d8b : FAILURE in 28m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5c76528cb9f24dba8a871ca85dd25fc9 : FAILURE in 32m 19s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/abd63ac66c0c4f28929bd7e5f11d8995 : FAILURE in 40m 49s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/8b88a150575549b69be542fcb5c2d2ba : FAILURE in 36m 31s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b57d46c73ef54b8cb1a45ca87160b6f7 : SUCCESS in 40m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1021459cf4484d438c16c8ebab831ff8 : SUCCESS in 46m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/206d06f62a0142bdab92cb11b0d22930 : FAILURE in 22m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"eca6f78d10a3b2d6de46bbe20c570ee04ec68c4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-09-21 14:31:02.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"dc5673bf8dd9f5c5ec442eddc99a3f87a5161c7b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-21 16:33:38.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/16/842216/6/check/DellEMC_PowerFlex_Manila_dhss_false/8c16ddd/DellEMC_PowerFlex_Manila_dhss_false/448 : SUCCESS in 30m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"e18ea84c6633da6398ce9296101d8c2686a5227b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 17:44:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6249be147ca44a193f338ccf0e02b80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adf3054954144dfeaa6e16eb91c8869a : SUCCESS in 4m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9a9792b949ec4c7c886df2bc946afd18 : FAILURE in 36m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/45701798a70c42489ff02f5fb2a35b4a : TIMED_OUT in 3h 02m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5308fced96a74e308c6f436d54991af0 : TIMED_OUT in 3h 05m 16s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/86cddd375f1c417fa3568b3c0305b8e3 : SUCCESS in 1h 10m 14s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e30e1a59d5ad4bb7a6f9797a1f0d9c06 : SUCCESS in 1h 27m 54s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6313712926ef4f6a9d8b463ac91c990b : FAILURE in 1h 07m 45s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8df40adc27e54584b9bb9e5b91c2e3d2 : FAILURE in 38m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/db5114e7df8742079ef1eebe6fb4ee17 : FAILURE in 42m 36s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/47e9d2be359b492ca65a5139142f3c56 : FAILURE in 42m 15s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/b43f34bed82b44fd985e68b102bf80f7 : FAILURE in 53m 05s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2e0063afc5844345883d926014d3e638 : SUCCESS in 26m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/79d88b34d08d45bf9bd9e2be326d65f6 : FAILURE in 1h 04m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f59b07f543b64100b07776a8d958f596 : FAILURE in 10m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"224ecc961ba7c2aea99b8a8655530c8bba4d532f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-09-21 17:49:11.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"cebffd76dd9eaa9442178e5bfd5ca308761842be","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-21 20:11:26.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/16/842216/7/check/DellEMC_PowerFlex_Manila_dhss_false/47cdf05/DellEMC_PowerFlex_Manila_dhss_false/455 : SUCCESS in 33m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"1622f4779c368fb7a830ec25136fb1dc142bd2e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 20:18:21.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/870092beee3544cfaa800ecc8d6101e2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b46ca78bf1f64360b1169267b5303364 : SUCCESS in 2m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/719d9ce62928499f8ef1ff61892a8715 : FAILURE in 45m 56s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9e8936d469f24e51a4efd5abf201fa20 : SUCCESS in 2h 06m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bece030a6c794374acb7f8fd291fb40c : FAILURE in 2h 21m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/96fb61b8f61a4b83bfdddca3d08691ab : SUCCESS in 1h 18m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f36b163de47b4159925d550ed93d4e20 : SUCCESS in 1h 24m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/92cc593dc5d74ac988d64c9af85b42b4 : FAILURE in 1h 15m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/24fdad12586042659bb7adf7a9f3eece : FAILURE in 27m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e4358f04e24a4ecd84fdfec07aa5c699 : FAILURE in 44m 18s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/09248d307bb7440b8080219456c44bd8 : FAILURE in 49m 44s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5fa0384b9b3f4439a8dcef94adbf3132 : FAILURE in 38m 04s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/db8934fb139346f5841702feea2bed12 : SUCCESS in 27m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e9c6585e1fdd42f383c544b6591e4886 : FAILURE in 1h 08m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f0ee304c203a47cbb059e8a1519f5a3f : FAILURE in 22m 32s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"92c5811985dde92973244a916c9dca1372b7b59d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-22 13:47:04.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/16/842216/7/check/DellEMC_Unity_Manila_dhss_true/578cd1e/DellEMC_Unity_Manila_dhss_true/636 : SUCCESS in 4h 20m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/16/842216/7/check/DellEMC_Unity_Manila_dhss_false/2180a78/DellEMC_Unity_Manila_dhss_false/671 : SUCCESS in 1h 09m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"d402275caa5a5cf3f4b9474e34d223cfe1b3a479","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-09-22 18:53:59.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"638b6b6d018328e256a0a642b783da97750884e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-09-22 19:02:54.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"08ddf5bf47fd4f347aab453ca6bb31cd52e46770","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-22 19:29:44.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/16/842216/9/check/DellEMC_PowerFlex_Manila_dhss_false/a6e1a6d/DellEMC_PowerFlex_Manila_dhss_false/464 : SUCCESS in 26m 34s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"1065750332a8e54900130d35a12cd34163cebf6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 20:39:43.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/15bbccb506e44e6d9987111552de18c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50bf8c0c5e564636bba1d56e3d37a3d7 : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/85261bc9286344c6bb3e47c44634c468 : FAILURE in 41m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a5ae6ef39f4f480bb96e6e3d6615e726 : SUCCESS in 1h 35m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6807966886e34de0b9a41c15cd526fcd : SUCCESS in 1h 23m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/af978f690f7f4414bb97d814803e6a64 : SUCCESS in 49m 44s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f7318fbe7d474280935df64b9ca3afb2 : SUCCESS in 1h 30m 54s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/747722eff01a42e4a23ea51a072997dc : FAILURE in 1h 20m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0b05dcb48bf041d3972441e90b457158 : FAILURE in 28m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9a42641b9c674c24af6f51e8c64a33d8 : FAILURE in 47m 53s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/989b56fcb42f4667bc0b11c7b1e1dae8 : FAILURE in 37m 33s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/bbea284e8b934bc0a22acbda8f4c5b96 : FAILURE in 35m 22s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8363bc25148d4f368260991151001140 : SUCCESS in 32m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/54b03193f63e480fafbc800c3fc7fe97 : FAILURE in 54m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a6611abeab9a4a1c8c31d90643c75724 : FAILURE in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1daab4931730166e25c63f48f898e3712666ea42","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-23 02:32:44.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/16/842216/9/check/DellEMC_Unity_Manila_dhss_true/b08f44a/DellEMC_Unity_Manila_dhss_true/640 : SUCCESS in 3h 05m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/16/842216/9/check/DellEMC_Unity_Manila_dhss_false/5ede348/DellEMC_Unity_Manila_dhss_false/676 : SUCCESS in 1h 18m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"0047254d4d86ede63ebaab1596e59b2d6b2c7e2d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-07-11 20:14:45.000000000","message":"Topic group-validation removed","accounts_in_message":[],"_revision_number":9},{"id":"9d362e565aa16f51850b8304f3cfcf58b76aecaf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2024-07-11 20:14:48.000000000","message":"Topic set to bug/1936689","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"db2d7a2afd09612b69f3f0108f9b8ccd64e358cb","revisions":{"64f431977a59d449326833d945ccc2d9d7b29086":{"kind":"REWORK","_number":1,"created":"2022-05-17 18:38:51.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/16/842216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/842216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/1"}}},"commit":{"parents":[{"commit":"ba5810ecafd4c873a44bcd949a271063ed1f09cf","subject":"Merge \"Use allow_access helper method\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ba5810ecafd4c873a44bcd949a271063ed1f09cf"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-05-17 18:30:25.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-05-17 18:30:25.000000000","tz":0},"subject":"[WIP] share group type create validation check","message":"[WIP] share group type create validation check\n\nchanges create_share_type() to allow for share group\ntype validations\n\nChange-Id: I5c52c2ee589083e7baf2987c42568282afb1f883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/64f431977a59d449326833d945ccc2d9d7b29086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/64f431977a59d449326833d945ccc2d9d7b29086"}]},"branch":"refs/heads/master"},"33eb6569ab46337e926baecab20b52057e00e77a":{"kind":"REWORK","_number":2,"created":"2022-05-18 17:29:05.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/16/842216/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/842216/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/2"}}},"commit":{"parents":[{"commit":"45cd35173701c626dd35b846ae610a01924756ae","subject":"Merge \"Remove create_floating_ip \u0026 get_server_ip methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/45cd35173701c626dd35b846ae610a01924756ae"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-05-17 18:30:25.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-05-18 17:22:06.000000000","tz":0},"subject":"[WIP] share group type create validation check","message":"[WIP] share group type create validation check\n\nchanges create_share_type() to allow for share group\ntype validations\n\nChange-Id: I5c52c2ee589083e7baf2987c42568282afb1f883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/33eb6569ab46337e926baecab20b52057e00e77a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/33eb6569ab46337e926baecab20b52057e00e77a"}]},"branch":"refs/heads/master"},"4ba1bdc9c3a766ef0eb786a398841b53cc4b8f90":{"kind":"REWORK","_number":3,"created":"2022-09-15 15:49:22.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/16/842216/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/842216/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/3"}}},"commit":{"parents":[{"commit":"2b3fcc571f217c14042f0d6991f59087fbfb0385","subject":"Merge \"Use ddt.named_data decorator to avoid random test names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b3fcc571f217c14042f0d6991f59087fbfb0385"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-05-17 18:30:25.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-09-15 15:49:01.000000000","tz":0},"subject":"[WIP] share group type create validation check","message":"[WIP] share group type create validation check\n\nchanges create_share_type() to allow for share group\ntype validations\n\nChange-Id: I5c52c2ee589083e7baf2987c42568282afb1f883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4ba1bdc9c3a766ef0eb786a398841b53cc4b8f90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4ba1bdc9c3a766ef0eb786a398841b53cc4b8f90"}]},"branch":"refs/heads/master"},"0a9f374e652d6f1d5d7bddc3c1676761c7997763":{"kind":"REWORK","_number":4,"created":"2022-09-15 18:01:49.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/16/842216/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/842216/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/4"}}},"commit":{"parents":[{"commit":"2b3fcc571f217c14042f0d6991f59087fbfb0385","subject":"Merge \"Use ddt.named_data decorator to avoid random test names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b3fcc571f217c14042f0d6991f59087fbfb0385"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-05-17 18:30:25.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-09-15 18:01:01.000000000","tz":0},"subject":"[WIP] share group type create validation check","message":"[WIP] share group type create validation check\n\nchanges create_share_type() to allow for share group\ntype validations\n\nChange-Id: I5c52c2ee589083e7baf2987c42568282afb1f883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a9f374e652d6f1d5d7bddc3c1676761c7997763"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0a9f374e652d6f1d5d7bddc3c1676761c7997763"}]},"branch":"refs/heads/master"},"0f6d8ce4030467adc25572619add9fbe3fa5ecbe":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-09-19 15:04:06.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/16/842216/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/842216/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/5"}}},"commit":{"parents":[{"commit":"2b3fcc571f217c14042f0d6991f59087fbfb0385","subject":"Merge \"Use ddt.named_data decorator to avoid random test names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b3fcc571f217c14042f0d6991f59087fbfb0385"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-05-17 18:30:25.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2023-09-19 15:04:06.000000000","tz":0},"subject":"[WIP] share group type create validation check","message":"[WIP] share group type create validation check\n\nchanges create_share_type() to allow for share group\ntype validations\n\nPartial-Bug: #1936689\nChange-Id: I5c52c2ee589083e7baf2987c42568282afb1f883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f6d8ce4030467adc25572619add9fbe3fa5ecbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0f6d8ce4030467adc25572619add9fbe3fa5ecbe"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3ed100582376d842cbbec4c4ef782c552bcb38b5":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-09-21 14:31:02.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/16/842216/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/842216/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/6"}}},"commit":{"parents":[{"commit":"2b3fcc571f217c14042f0d6991f59087fbfb0385","subject":"Merge \"Use ddt.named_data decorator to avoid random test names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b3fcc571f217c14042f0d6991f59087fbfb0385"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-05-17 18:30:25.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","date":"2023-09-21 14:31:02.000000000","tz":0},"subject":"[WIP] share group type create validation check","message":"[WIP] share group type create validation check\n\nchanges create_share_type() to allow for share group\ntype validations\n\nPartial-Bug: #1936689\nDepends-On: Ifc98b8f5fdfd1047a8680beb36bf1601091d3217\nChange-Id: I5c52c2ee589083e7baf2987c42568282afb1f883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3ed100582376d842cbbec4c4ef782c552bcb38b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3ed100582376d842cbbec4c4ef782c552bcb38b5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2df086d09920679810c229de657e6eae1c58be55":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-09-21 17:49:11.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/16/842216/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/842216/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/7"}}},"commit":{"parents":[{"commit":"2b3fcc571f217c14042f0d6991f59087fbfb0385","subject":"Merge \"Use ddt.named_data decorator to avoid random test names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b3fcc571f217c14042f0d6991f59087fbfb0385"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-05-17 18:30:25.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","date":"2023-09-21 17:49:11.000000000","tz":0},"subject":"[WIP] share group type create validation check","message":"[WIP] share group type create validation check\n\nchanges create_share_type() to allow for share group\ntype validations\n\nPartial-Bug: #1936689\nChange-Id: I5c52c2ee589083e7baf2987c42568282afb1f883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2df086d09920679810c229de657e6eae1c58be55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2df086d09920679810c229de657e6eae1c58be55"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f705ccf74627af891ecbab1d9f01dee5254aa73a":{"kind":"REWORK","_number":8,"created":"2023-09-22 18:53:59.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/16/842216/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/842216/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/8"}}},"commit":{"parents":[{"commit":"2b3fcc571f217c14042f0d6991f59087fbfb0385","subject":"Merge \"Use ddt.named_data decorator to avoid random test names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b3fcc571f217c14042f0d6991f59087fbfb0385"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-05-17 18:30:25.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","date":"2023-09-22 18:53:58.000000000","tz":0},"subject":"[WIP] share group type create validation check","message":"[WIP] share group type create validation check\n\nchanges create_share_type() to allow for share group\ntype validations\n\nPartial-Bug: #1936689\nChange-Id: I5c52c2ee589083e7baf2987c42568282afb1f883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f705ccf74627af891ecbab1d9f01dee5254aa73a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f705ccf74627af891ecbab1d9f01dee5254aa73a"}]},"branch":"refs/heads/master"},"db2d7a2afd09612b69f3f0108f9b8ccd64e358cb":{"kind":"REWORK","_number":9,"created":"2023-09-22 19:02:54.000000000","uploader":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"ref":"refs/changes/16/842216/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/16/842216/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/16/842216/9"}}},"commit":{"parents":[{"commit":"2b3fcc571f217c14042f0d6991f59087fbfb0385","subject":"Merge \"Use ddt.named_data decorator to avoid random test names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b3fcc571f217c14042f0d6991f59087fbfb0385"}]}],"author":{"name":"Ashley Rodriguez","email":"ashrod98@gmail.com","date":"2022-05-17 18:30:25.000000000","tz":0},"committer":{"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","date":"2023-09-22 19:02:54.000000000","tz":0},"subject":"[WIP] share group type create validation check","message":"[WIP] share group type create validation check\n\nchanges create_share_type() to allow for share group\ntype validations\n\nPartial-Bug: #1936689\nChange-Id: I5c52c2ee589083e7baf2987c42568282afb1f883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/db2d7a2afd09612b69f3f0108f9b8ccd64e358cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/db2d7a2afd09612b69f3f0108f9b8ccd64e358cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
