)]}'
{"id":"openstack%2Fmanila-tempest-plugin~842708","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I63d331e575899fa962daf022eaf6cace96a703c7","project":"openstack/manila-tempest-plugin","branch":"master","topic":"fix-queue-config","attention_set":{},"removed_from_attention_set":{"32411":{"account":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"last_update":"2022-05-22 09:49:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I63d331e575899fa962daf022eaf6cace96a703c7","subject":"Define queues at project level","status":"MERGED","created":"2022-05-20 13:10:55.000000000","updated":"2022-05-22 09:49:00.000000000","submitted":"2022-05-22 09:49:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842708-fix-queue-config","meta_rev_id":"2530273961dd86ba89a9abcd27dc8106931e1b68","_number":842708,"virtual_id_number":842708,"owner":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-22 09:49:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2022-05-22 08:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2022-05-22 08:40:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-20 15:25:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-20 15:27:09.000000000","updated_by":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-20 21:01:36.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-05-22 08:40:52.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"fbd9e0724d13008afe5aac9473de7d6aa151d8b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2022-05-20 13:10:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8b4c11f4e30b4770c57182149ca18a0f33d355f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 15:25:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac45faec345d40a2bf2a1ba72aa6a714 : SUCCESS in 2m 40s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/abdfa525b58c4e5fa15949fc5b0428f2 : SUCCESS in 40m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5681db395ed345eb9fcc7f2e1f63886e : SUCCESS in 39m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0fa0283455a64b8b9110f0a0f490d489 : SUCCESS in 1h 31m 22s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/bbb36a3e50c741e3b9b14f7a0b4b1c2f : SUCCESS in 1h 09m 26s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/347d82019ee64e23b2eb84b1467f57b8 : SUCCESS in 1h 23m 27s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/8ac21330deb54e8298575b13c7488ed9 : SUCCESS in 1h 32m 42s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e5a0cc2b12fd4c3989ef1f15adae1bd3 : SUCCESS in 1h 00m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/baa6b9bd55374c4f83f53c0df8dafc98 : SUCCESS in 1h 47m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a426c03071f44d4592c7820a73c70f9b : FAILURE in 1h 43m 23s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7f965c80d1564934a91a11abb7b850b3 : SUCCESS in 1h 09m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/378e7e39736d4fcc95790267e27f2130 : FAILURE in 2h 10m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ad17edf949814f10918f0b5643220568 : SUCCESS in 28m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9a16f8fb5032e0fca4445a59a350dd4e1ead6bb","author":{"_account_id":12017,"name":"DellEMC VNX CI","email":"emc.vnx.ci@emc.com","username":"emc-vnx-ci","tags":["SERVICE_USER"]},"date":"2022-05-20 15:27:09.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_VNX_Manila https://elab-os-logsrv.delllabs.net/08/842708/1/check/DellEMC_VNX_Manila/31a357c/DellEMC_VNX_Manila/270 : FAILURE in 1m 13s\n\nLeave a comment with \u0027run-DellEMC VNX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VNX_CI","accounts_in_message":[],"_revision_number":1},{"id":"9b013050a6c12b834344e19839f0ae330ce210a7","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-20 21:01:36.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/08/842708/1/check/DellEMC_Unity_Manila_dhss_true/a03ccaf/DellEMC_Unity_Manila_dhss_true/487 : FAILURE in 4h 08m 51s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/842708/1/check/DellEMC_Unity_Manila_dhss_false/5db3c4c/DellEMC_Unity_Manila_dhss_false/508 : SUCCESS in 1h 13m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7db0db0b472b83331dc2be9bb3a3f247ec8779b7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-22 08:40:43.000000000","message":"Topic set to fix-queue-config","accounts_in_message":[],"_revision_number":1},{"id":"15fb7f99956658de7b3cb270eea0aa61d99dbcaa","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-05-22 08:40:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"040d6193b2235686ddfd43853cfbd0f80c510f42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 08:41:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f24af5ceef050437c8f0b7a05fbf0c2e9620b49c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 09:49:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af130892daf444fa9a53a9a0bbb8d425 : SUCCESS in 3m 23s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/768e46f62f954e05a6961bee49a2c43c : SUCCESS in 31m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/058a816b69454f0c958071028d6ff132 : SUCCESS in 35m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/43f4178dfcdd461bbc3c382a68d9318f : SUCCESS in 1h 06m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2530273961dd86ba89a9abcd27dc8106931e1b68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-22 09:49:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab","revisions":{"a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab":{"kind":"REWORK","_number":1,"created":"2022-05-20 13:10:55.000000000","uploader":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"ref":"refs/changes/08/842708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/08/842708/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/842708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/842708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/842708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/08/842708/1"}}},"commit":{"parents":[{"commit":"45cd35173701c626dd35b846ae610a01924756ae","subject":"Merge \"Remove create_floating_ip \u0026 get_server_ip methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/45cd35173701c626dd35b846ae610a01924756ae"}]}],"author":{"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","date":"2022-05-20 13:10:04.000000000","tz":-180},"committer":{"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","date":"2022-05-20 13:10:04.000000000","tz":-180},"subject":"Define queues at project level","message":"Define queues at project level\n\nDefining queues at project.pipeline level was deprecated in release\n4.1.0 of Zuul [1] and support for it will (probably) be dropped in\nrelease 7.0.0. [2][3]\n\n[1] https://zuul-ci.org/docs/zuul/latest/releasenotes.html#relnotes-4-1-0-deprecation-notes\n[2] https://lists.openstack.org/pipermail/openstack-discuss/2022-May/028603.html\n[3] https://lists.zuul-ci.org/pipermail/zuul-discuss/2022-May/001801.html\n\nSigned-off-by: Eduardo Santos \u003ceduardo.experimental@gmail.com\u003e\nChange-Id: I63d331e575899fa962daf022eaf6cace96a703c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
