)]}'
{"id":"openstack%2Fmanila-tempest-plugin~843047","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iccd28f79f97fc50fa9e3fb1ad4c13192c3ee8776","project":"openstack/manila-tempest-plugin","branch":"master","topic":"ussuri-pin-tempest","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-05-25 14:48:27.000000000","reason":"Ghanshyam Mann replied on the change"}},"hashtags":[],"change_id":"Iccd28f79f97fc50fa9e3fb1ad4c13192c3ee8776","subject":"DNM: Testing tempest pin on stable/ussuri","status":"ABANDONED","created":"2022-05-24 01:56:02.000000000","updated":"2022-05-31 21:28:18.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"908d3189cbc820c1a2df5988fd7d2f0b52d18538","_number":843047,"virtual_id_number":843047,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-25 04:26:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-24 04:21:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-24 04:44:58.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"63e6cea24d0321f40621273c3067638fcb77dfed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-24 01:56:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94881a747dee923825705d56a0979c6d27b39016","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 04:21:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c160e74a1719406dba8500278bd63d81 : SUCCESS in 3m 58s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/828234e479914dd4a72a5ee77fb157c6 : SUCCESS in 27m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0bf9c49065ae442da30f494255330260 : SUCCESS in 29m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d00a53d07d3a4da7bc6621f74a47e922 : SUCCESS in 1h 16m 37s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a6d39f4368dd417fb0c27b5d141a9f4f : FAILURE in 1h 03m 08s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/58a557a29a9a4f8ca31a86a843b7ceb0 : SUCCESS in 1h 13m 36s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/4c0e1be8c89b414eb689a9d05dfde387 : SUCCESS in 1h 07m 04s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/832ef757c7b7410da4f749fe02fa713b : SUCCESS in 59m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/755eb0ccc7ef4df2a43e1d414561bce1 : SUCCESS in 1h 03m 08s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/69073cb26ac54866900d02482879baae : FAILURE in 1h 24m 21s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/60de8b46feaa4d91aa710779854ca237 : SUCCESS in 46m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8e408c22f15a4215a9f142c36dc5fb2f : FAILURE in 2h 22m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ab9ae89bb0e24676b2ec6a0cd29b2a7e : SUCCESS in 18m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8221c7ffd894c49ac6339764f4db2add084a3858","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-24 04:44:58.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/47/843047/1/check/DellEMC_Unity_Manila_dhss_true/b9e590e/DellEMC_Unity_Manila_dhss_true/490 : FAILURE in 2h 48m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/47/843047/1/check/DellEMC_Unity_Manila_dhss_false/07cb2d8/DellEMC_Unity_Manila_dhss_false/511 : SUCCESS in 1h 13m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"87e18edbcd32ba738565829bcf5d86c87854a25c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-24 18:12:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03ff0bf8ecdd2afbceb0aa9e3cf2a4685b8865a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 20:12:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6017ddb78cf47c0bbf01ec25cded3dd : SUCCESS in 4m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2d5fec9d6e144e5ea82b65ebbae88c70 : SUCCESS in 40m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/aed41c50a5c84be495370ade5baa48df : SUCCESS in 39m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e499a40f1f0a43e7bef03c47cc30f0c3 : POST_FAILURE in 1h 42m 35s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c76610ae5bbb42b98bd1145f4dc0e4fc : SUCCESS in 1h 20m 14s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/7a16a6ffca3a42f687c746bc50c7db55 : SUCCESS in 1h 20m 59s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/74ba6cd8081b461c8fe9418acf734a22 : SUCCESS in 1h 01m 07s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/51380c96bc8a493daefeee7498f4c895 : SUCCESS in 1h 02m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8073291040df48ff9f05020ffb5f8fe5 : SUCCESS in 1h 16m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cda97502525e4a438d4e1180223965b4 : FAILURE in 1h 51m 31s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/693a5b5f17054dbf9d81ace106f78af0 : SUCCESS in 54m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/74e862591f8c4fef9f04562f74e2c73d : SUCCESS in 1h 42m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4b03d22f8c9d4ced908cfb0cde9554b1 : SUCCESS in 22m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"072f7bf6d80b08d0af3e62e4c6ed9419c79ba44e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-25 02:35:08.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ddc5283bc8de181d70c439c8349ffc7616d8a7d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 04:26:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f6308dae05c48b2916868683c3fd283 : SUCCESS in 3m 52s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/74b63203709f4c968339b097baf0de6d : SUCCESS in 34m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f921c7311e3a4e18b4e72c24f70648a1 : SUCCESS in 46m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a26c7a95fc784f60be03b2b44a1832d5 : SUCCESS in 1h 26m 05s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d7dc8cea0cf74634b5b47f39f413b5d4 : FAILURE in 1h 04m 14s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c3b7c987670143c3bca4df3df019a8a5 : SUCCESS in 1h 08m 33s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/21bac71eb2b643a0a55fb9e7c50230e9 : SUCCESS in 1h 03m 27s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/efa9df1f00c74a179ebd55f107fc8a53 : SUCCESS in 1h 11m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/5df1cefe03874d518d6648d8e7a6183e : FAILURE in 1h 16m 55s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fc4855985bc14ccf895a50d175192330 : FAILURE in 1h 36m 49s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4eb8ac28d3d24ba0abccd439a8d7e413 : SUCCESS in 59m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ab084ac9010c4c52a1459b37b1c09c96 : FAILURE in 1h 26m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/363a5050058641378017337e495bc1e0 : SUCCESS in 25m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"452bb1c69b83e53243352b3c18426cb20538abbe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-25 07:08:48.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/47/843047/2/check/DellEMC_Unity_Manila_dhss_true/3acdd87/DellEMC_Unity_Manila_dhss_true/495 : FAILURE in 3h 33m 02s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/47/843047/2/check/DellEMC_Unity_Manila_dhss_false/f5dd588/DellEMC_Unity_Manila_dhss_false/516 : SUCCESS in 1h 45m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"002631b3699466fb7af31a713aac88dfa536c726","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-25 14:48:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"908d3189cbc820c1a2df5988fd7d2f0b52d18538","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-31 21:28:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"feb851acdee1c50c232917ccda0f40bbb3214e61","revisions":{"dbf8f0f4d913910086bc54cc2363ac82e1e8ab96":{"kind":"REWORK","_number":1,"created":"2022-05-24 01:56:02.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/843047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/47/843047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/843047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/843047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/843047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/843047/1"}}},"commit":{"parents":[{"commit":"a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab","subject":"Define queues at project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-24 01:55:13.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-24 01:55:13.000000000","tz":-300},"subject":"DNM: Testing tempest pin on stable/ussuri","message":"DNM: Testing tempest pin on stable/ussuri\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/843045\nChange-Id: Iccd28f79f97fc50fa9e3fb1ad4c13192c3ee8776\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dbf8f0f4d913910086bc54cc2363ac82e1e8ab96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dbf8f0f4d913910086bc54cc2363ac82e1e8ab96"}]},"branch":"refs/heads/master"},"feb851acdee1c50c232917ccda0f40bbb3214e61":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-25 02:35:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/47/843047/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/47/843047/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/843047/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/843047/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/843047/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/47/843047/2"}}},"commit":{"parents":[{"commit":"a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab","subject":"Define queues at project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-24 01:55:13.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-05-25 02:35:07.000000000","tz":0},"subject":"DNM: Testing tempest pin on stable/ussuri","message":"DNM: Testing tempest pin on stable/ussuri\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/838051\nChange-Id: Iccd28f79f97fc50fa9e3fb1ad4c13192c3ee8776\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/feb851acdee1c50c232917ccda0f40bbb3214e61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/feb851acdee1c50c232917ccda0f40bbb3214e61"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
