)]}'
{"id":"openstack%2Fmanila-tempest-plugin~843307","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I949f6fc960c93084efec7ab36cd9f98b636513e5","project":"openstack/manila-tempest-plugin","branch":"master","topic":"victoria-pin-tempest","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-06-08 17:35:46.000000000","reason":"Carlos Eduardo replied on the change"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-06-09 16:29:40.000000000","reason":"Douglas Viroel replied on the change"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-06-09 17:38:43.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-06-09 17:38:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I949f6fc960c93084efec7ab36cd9f98b636513e5","subject":"Adjust stable jobs per maintenance transitions","status":"MERGED","created":"2022-05-25 15:55:27.000000000","updated":"2022-06-09 17:38:43.000000000","submitted":"2022-06-09 17:38:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843307-victoria-pin-tempest","meta_rev_id":"a1486e7007640c2a59e4b528ed6174561dca9813","_number":843307,"virtual_id_number":843307,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-09 17:38:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-06-08 17:35:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-06-09 16:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2022-06-09 15:05:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2022-06-09 16:29:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-25 18:53:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-26 03:40:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-06-08 12:34:35.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-06-08 17:31:52.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-06-08 17:31:52.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-06-09 15:05:43.000000000","updated_by":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"reviewer":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"state":"REVIEWER"}],"messages":[{"id":"8fabc33209ed6a4cc33959e7c736ed01c64431c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-25 15:55:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"832fd609e38c56c4e9e315e986b69af9f3e75fcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 18:53:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f0c12f9f07b4758ab0bfec433fdad32 : SUCCESS in 5m 15s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4a3770970dc04c19b5e4724c9ee0be41 : SUCCESS in 30m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b65f10363e174010ba778ee488c8083f : SUCCESS in 36m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f60177a265cb483885285dd7f0a2755a : SUCCESS in 1h 35m 40s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/6061a13d96944bc285970b8724b85bc3 : SUCCESS in 1h 01m 22s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3ad5872704aa4e459825acab325ab660 : SUCCESS in 1h 11m 57s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/7a8ffed8e7ca401d9b354d92d4664891 : FAILURE in 15m 45s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2af614644bc5493c98c7d4800d95c71a : SUCCESS in 1h 23m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/8e0d3f4e53224a51817e4a103ff1b93b : SUCCESS in 1h 28m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8c6703628af44c799ab5b2ae6baccb8e : FAILURE in 2h 23m 17s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/036a432dbbe74a0bb78e7e2e3a0a16fa : SUCCESS in 54m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/985c4f5939624f2cb57a8af886fb0bb9 : FAILURE in 2h 48m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9cfd277dcf9b432292117f3da1847f2b : SUCCESS in 36m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"95c214ea015ca4f38240440cec6fc14bb9b21377","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-25 21:40:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2d3b954e7851f771484fc7333075ccdeab0edfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 23:28:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5b1f408d5c347d492fb4138c9f71c09 : SUCCESS in 3m 37s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c2bdeedc8ec246688df685dfc76b441d : SUCCESS in 38m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/da7ad4715a284ce898d423b42fb13e32 : SUCCESS in 35m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4fc97c97f9f340149434696e5604114d : SUCCESS in 1h 17m 56s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/800b50200a624193922b1761c4ac9a1a : SUCCESS in 1h 08m 43s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/0ec01e5ce28c4caf8ac9833d263da6e2 : FAILURE in 1h 22m 44s\n- manila-tempest-plugin-lvm-victoria https://zuul.opendev.org/t/openstack/build/6ef182c3b888494b911b52a370cd0666 : SUCCESS in 1h 19m 12s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fec02e345efa45519c3d06b06c49f47d : SUCCESS in 1h 28m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/e321ccf0c1814694ad95c70d2ec10c2c : SUCCESS in 1h 20m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f2cba358a9f24357aa231c64579aec81 : FAILURE in 1h 32m 37s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/65d06cef445b455bb23af831d9c8da86 : SUCCESS in 56m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1895fb4dc69948c2b03b312e92fee555 : FAILURE in 1h 35m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2e28930e50c446a7861271db0d1edfce : SUCCESS in 32m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ad0b0e7fed7e002e16b8fbfacd511d0b5ddc3d4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-26 03:40:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/07/843307/1/check/DellEMC_Unity_Manila_dhss_true/c88b84d/DellEMC_Unity_Manila_dhss_true/502 : FAILURE in 4h 03m 02s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/07/843307/1/check/DellEMC_Unity_Manila_dhss_false/b09b412/DellEMC_Unity_Manila_dhss_false/523 : SUCCESS in 1h 04m 04s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"20b576139ecaddfbb0a6795bf9e9f8136b34da1a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-26 22:07:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8fa5dab5e531a64d9d49594cada492b5e862943","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-31 20:56:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"edbb9fdb1947e07449376ead6388ed5833e8a16f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-31 21:07:46.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"60592dd8298a2be3ab24398f5f3317c376090f70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 23:20:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a5fab893f2845f49b0a292c603d174a : SUCCESS in 2m 47s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a405f67a655645919081a33bef7e6e95 : SUCCESS in 28m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ada53c113cf9448586e147af77956659 : SUCCESS in 32m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e0af38f314134362943bc31f7810bc06 : FAILURE in 1h 14m 46s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3e69b4024ef94ada9505ec87440a1b1c : FAILURE in 27m 16s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/663db5be335d49b2bbb367338b9c1200 : SUCCESS in 1h 06m 53s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f2573554e1b3493893fd7aadcd1315cc : SUCCESS in 1h 12m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/96507b4f397b4d23b654c5415cbe1086 : SUCCESS in 1h 16m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/514b6ee271be43b39696af3d00805574 : FAILURE in 2h 08m 00s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/bb9070b9bdf44d3ea5179a040d513f29 : SUCCESS in 56m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/82c3fc7b31794fe2923cc24705b9fce6 : SUCCESS in 1h 58m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/19d53c48f34e4c1f95c6787703770520 : SUCCESS in 19m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2fb311809f9460d610d3bd4669c87277a7dbf58b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-05-31 23:32:53.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/07/843307/3/check/DellEMC_Unity_Manila_dhss_true/564b40d/DellEMC_Unity_Manila_dhss_true/521 : FAILURE in 2h 24m 54s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/07/843307/3/check/DellEMC_Unity_Manila_dhss_false/7cd9f4b/DellEMC_Unity_Manila_dhss_false/544 : SUCCESS in 59m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"38a5623dddd97d60835d5521daf03074ea403d78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-08 12:34:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"969ace67fdb0d2a31beb84edb4cb5fcec1663e49","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-08 12:36:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"099d99924344d6cc3352e2c267ea5e154c466570","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-08 14:48:04.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/07/843307/4/check/DellEMC_Unity_Manila_dhss_true/73e98ed/DellEMC_Unity_Manila_dhss_true/534 : FAILURE in 2h 13m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/07/843307/4/check/DellEMC_Unity_Manila_dhss_false/23237e6/DellEMC_Unity_Manila_dhss_false/557 : SUCCESS in 1h 00m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"023655adcf9209747a9954332116472ce5544387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 15:02:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74f4879b0a194dcca1e7e4d68a5185b2 : SUCCESS in 4m 15s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/952c09e0c7674b0e8a96f8af93fbb7f5 : SUCCESS in 29m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ec8e866af2bf41c59e53eb934afebd06 : SUCCESS in 33m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e07f54fcdea44c148adf9e04c4a76710 : SUCCESS in 1h 12m 24s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/b7138ce39b6246049ade8adeefc369dc : SUCCESS in 1h 29m 17s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/46a7ff6742384e37a4d61c14dab8dc9f : SUCCESS in 1h 09m 12s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/6d02b9bd4075467e9039402fa525e9cf : SUCCESS in 1h 11m 58s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/42df602c22804f729b5c1e707218dcfa : SUCCESS in 1h 11m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f4454c72e069448281a38f31539b4741 : SUCCESS in 1h 23m 54s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1fa0da1ad9f344d3835a342116293be1 : FAILURE in 27m 34s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7a7c7324781545a69c63d718f9234372 : SUCCESS in 1h 12m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1a97da7a2ea44a1b8bdde107b006c987 : SUCCESS in 1h 55m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3e8f70a989064baeaa75e9353fc16700 : SUCCESS in 33m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0232a91d854690b168643fc3741722f38fc17f57","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-06-08 17:35:46.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6a276de4d065fc7ab7a10082789d79cf86e22ac1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-08 17:40:14.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/07/843307/4/check/DellEMC_Unity_Manila_dhss_true/be659c7/DellEMC_Unity_Manila_dhss_true/535 : FAILURE in 2h 37m 56s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/07/843307/4/check/DellEMC_Unity_Manila_dhss_false/db3ccba/DellEMC_Unity_Manila_dhss_false/558 : SUCCESS in 1h 01m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"943ca173361fcc335f16c5f486e7b925809de53d","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2022-06-09 15:05:43.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"78408013dab56cf572b162ed1992e0538cdf56d6","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-06-09 16:29:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cb2c5e667f50fe39bd7ec1918e15fba7fe478142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 16:30:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ce2bb5914948954188dea715d3464427f5b4580b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 17:38:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdb590f809c148c1b3d03f74284dd2f5 : SUCCESS in 3m 13s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/10718fa120914927a613383bea869c98 : SUCCESS in 28m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4ff928995f034937b4429ca25a16b349 : SUCCESS in 28m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f38cbdc301824b378ef2f0a90d257ad4 : SUCCESS in 1h 07m 37s","accounts_in_message":[],"_revision_number":4},{"id":"a1486e7007640c2a59e4b528ed6174561dca9813","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 17:38:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5019418ec60ca0ce1b0ac5d0382bc32b59dc47cc","revisions":{"4350e9705feeca251f7bdc866d0ce65b06484bb3":{"kind":"REWORK","_number":1,"created":"2022-05-25 15:55:27.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/07/843307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/07/843307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/1"}}},"commit":{"parents":[{"commit":"a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab","subject":"Define queues at project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-25 15:55:19.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-25 15:55:19.000000000","tz":-300},"subject":"DNM: testing Tempest pin in stable/victoria","message":"DNM: testing Tempest pin in stable/victoria\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/843295\nChange-Id: I949f6fc960c93084efec7ab36cd9f98b636513e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4350e9705feeca251f7bdc866d0ce65b06484bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4350e9705feeca251f7bdc866d0ce65b06484bb3"}]},"branch":"refs/heads/master"},"cf652273bc6211743f3d1635d2148d425ac1d1b3":{"kind":"REWORK","_number":2,"created":"2022-05-31 20:56:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/07/843307/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/07/843307/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/2"}}},"commit":{"parents":[{"commit":"a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab","subject":"Define queues at project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-31 20:56:03.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-31 20:56:03.000000000","tz":-300},"subject":"Remove stable/victoria jon from master gate","message":"Remove stable/victoria jon from master gate\n\nstable/victoria is in EM state and Tempest is pinned\nthere so we do not need to test the stable/victoria\njobs on master gate.\n\nChange-Id: I949f6fc960c93084efec7ab36cd9f98b636513e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf652273bc6211743f3d1635d2148d425ac1d1b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cf652273bc6211743f3d1635d2148d425ac1d1b3"}]},"branch":"refs/heads/master"},"91e6b67759267b6683f3ea014ddb7756f50ecfec":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-05-31 21:07:46.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/07/843307/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/07/843307/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/3"}}},"commit":{"parents":[{"commit":"a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab","subject":"Define queues at project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-31 20:56:03.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-05-31 21:07:46.000000000","tz":0},"subject":"Remove stable/victoria job from master gate","message":"Remove stable/victoria job from master gate\n\nstable/victoria is in EM state and Tempest is pinned\nthere so we do not need to test the stable/victoria\njobs on master gate.\n\nChange-Id: I949f6fc960c93084efec7ab36cd9f98b636513e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/91e6b67759267b6683f3ea014ddb7756f50ecfec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/91e6b67759267b6683f3ea014ddb7756f50ecfec"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5019418ec60ca0ce1b0ac5d0382bc32b59dc47cc":{"kind":"REWORK","_number":4,"created":"2022-06-08 12:34:35.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/07/843307/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/07/843307/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/07/843307/4"}}},"commit":{"parents":[{"commit":"a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab","subject":"Define queues at project level","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a6d89a15aa49df0a10966408ddbe2c9f9e5ec0ab"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-31 20:56:03.000000000","tz":-300},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-06-08 12:34:16.000000000","tz":330},"subject":"Adjust stable jobs per maintenance transitions","message":"Adjust stable jobs per maintenance transitions\n\nstable/yoga is the newest stable branch, and needs\na job testing this branch. stable/victoria is in EM\nstate and we don\u0027t support using manila-tempest-plugin\u0027s\nmaster branch against EM branches. Tempest is pinned\nas well for this EM branch.\n\nChange-Id: I949f6fc960c93084efec7ab36cd9f98b636513e5\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5019418ec60ca0ce1b0ac5d0382bc32b59dc47cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/5019418ec60ca0ce1b0ac5d0382bc32b59dc47cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
