)]}'
{"id":"openstack%2Fmanila-tempest-plugin~845286","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Iec0376fbfe12d0ccb20bc0b9215abdd590e2a2bc","project":"openstack/manila-tempest-plugin","branch":"master","topic":"tempest-scenario-manager-cleanup","attention_set":{},"removed_from_attention_set":{"33732":{"account":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"last_update":"2022-07-13 11:39:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iec0376fbfe12d0ccb20bc0b9215abdd590e2a2bc","subject":"Set minimal tempest version to 27.0.0","status":"ABANDONED","created":"2022-06-09 15:18:24.000000000","updated":"2022-07-13 11:39:35.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"c968b6737ecc1f4123d307a4410dad7d16adc135","_number":845286,"virtual_id_number":845286,"owner":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-09 15:32:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-06-09 15:32:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-09 18:05:33.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-07-12 16:45:28.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"79dfbd9bd2deeac78072f953e57b2dcd1d646984","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-06-09 15:18:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c470cd19cfa9edbb478159da5d6d81709eea095","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-06-09 15:22:19.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"30d99e7084f63da44ab21b9a5316b93f1f2f9c93","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"date":"2022-06-09 15:22:57.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a88ace5b2daef0aea003d89bdf43abc7aa182f4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 15:32:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bc327fc6ffb54cfd9a9c9325f21a68dd : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/306073ff7b374a84b30b16aa0df30a13 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":3},{"id":"6363cfe45eab376d452f8511a398e6cccc2f171e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-09 18:05:33.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/86/845286/3/check/DellEMC_Unity_Manila_dhss_true/95d633f/DellEMC_Unity_Manila_dhss_true/540 : FAILURE in 2h 42m 04s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/86/845286/3/check/DellEMC_Unity_Manila_dhss_false/cd91fd2/DellEMC_Unity_Manila_dhss_false/563 : SUCCESS in 50m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"c0e88d8a7ec8a517d9c8379d3b7c04a1fdc521fb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-12 16:45:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c968b6737ecc1f4123d307a4410dad7d16adc135","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-13 11:39:35.000000000","message":"Abandoned\n\nI see that this change was already done as part of Ie17c1d9349e98ce016801bf47c1cb9dba18b51f6 \n\nhttps://github.com/openstack/manila-tempest-plugin/commit/c1f7b9024480ceb2be091f68869d9c003af9577c","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6a27f4ef4599dfe62f56410f8831063751389fc9","revisions":{"e7631d8ade9af8665f47066ab9b3742d77ea0362":{"kind":"REWORK","_number":1,"created":"2022-06-09 15:18:24.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/86/845286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/845286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/845286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/845286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/845286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/845286/1"}}},"commit":{"parents":[{"commit":"f33c6a4d61b85653fd810e4e34a13921cde0212e","subject":"Remove setup_clients","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f33c6a4d61b85653fd810e4e34a13921cde0212e"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-06-09 15:18:09.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-06-09 15:18:09.000000000","tz":120},"subject":"Set minimal tempest version to \u003e\u003d27.0.0","message":"Set minimal tempest version to \u003e\u003d27.0.0\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\nSome methods replaced in scenario-manager-cleanup effor won\u0027t work\nwith tempest version lower than 27.0.0 therefore minimal tempest\nversion must be updated in tempest plugins\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Iec0376fbfe12d0ccb20bc0b9215abdd590e2a2bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e7631d8ade9af8665f47066ab9b3742d77ea0362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e7631d8ade9af8665f47066ab9b3742d77ea0362"}]},"branch":"refs/heads/master"},"d055ee67fcfb7c0884233a7ba34b519c485accea":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-06-09 15:22:19.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/86/845286/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/845286/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/845286/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/845286/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/845286/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/845286/2"}}},"commit":{"parents":[{"commit":"f33c6a4d61b85653fd810e4e34a13921cde0212e","subject":"Remove setup_clients","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f33c6a4d61b85653fd810e4e34a13921cde0212e"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-06-09 15:18:09.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-06-09 15:22:19.000000000","tz":0},"subject":"Set minimal tempest version to 27.0.0","message":"Set minimal tempest version to 27.0.0\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\nSome methods replaced in scenario-manager-cleanup effor won\u0027t work\nwith tempest version lower than 27.0.0 therefore minimal tempest\nversion must be updated in tempest plugins\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Iec0376fbfe12d0ccb20bc0b9215abdd590e2a2bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d055ee67fcfb7c0884233a7ba34b519c485accea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d055ee67fcfb7c0884233a7ba34b519c485accea"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6a27f4ef4599dfe62f56410f8831063751389fc9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-06-09 15:22:57.000000000","uploader":{"_account_id":33732,"name":"Roman Popelka","email":"rpopelka@redhat.com","username":"rpopelka"},"ref":"refs/changes/86/845286/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/86/845286/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/845286/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/845286/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/845286/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/86/845286/3"}}},"commit":{"parents":[{"commit":"f33c6a4d61b85653fd810e4e34a13921cde0212e","subject":"Remove setup_clients","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f33c6a4d61b85653fd810e4e34a13921cde0212e"}]}],"author":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-06-09 15:18:09.000000000","tz":120},"committer":{"name":"Roman Popelka","email":"rpopelka@redhat.com","date":"2022-06-09 15:22:57.000000000","tz":0},"subject":"Set minimal tempest version to 27.0.0","message":"Set minimal tempest version to 27.0.0\n\nAs tempest.scenario.manager was announced stable interface in Tempest\n27.0.0[1] it can be now reused in plugins.\n\nSome methods replaced in scenario-manager-cleanup effort won\u0027t work\nwith tempest version lower than 27.0.0 therefore minimal tempest\nversion must be updated in tempest plugins\n\nEtherpad concerning this effort:\nhttps://etherpad.opendev.org/p/tempest-scenario-manager-cleanup\n\n[1] https://docs.openstack.org/releasenotes/tempest/v27.0.0.html#release-notes-27-0-0\n\nChange-Id: Iec0376fbfe12d0ccb20bc0b9215abdd590e2a2bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6a27f4ef4599dfe62f56410f8831063751389fc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6a27f4ef4599dfe62f56410f8831063751389fc9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
