)]}'
{"id":"openstack%2Fmanila-tempest-plugin~845663","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I7cae098a8240d6702c79b5ef0b4f5823138c4e36","project":"openstack/manila-tempest-plugin","branch":"master","topic":"DNM","attention_set":{},"removed_from_attention_set":{"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-10-18 09:09:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7cae098a8240d6702c79b5ef0b4f5823138c4e36","subject":"DNM","status":"ABANDONED","created":"2022-06-14 04:20:24.000000000","updated":"2022-10-18 09:09:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1c214289f697c953cef9ac946981f1e9fd5d1afe","_number":845663,"virtual_id_number":845663,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-14 06:25:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-14 06:25:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-14 18:27:18.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"911f63c0a0af72a69208bf665622e37273dc5300","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-06-14 04:20:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b53bb47ad1a892e47aa48ae9d0c0a0e6ced7c533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 06:25:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ef842214473403f80c6d41bcc91e292 : SUCCESS in 3m 52s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/98f3f95db80849659907450b5c6c9946 : SUCCESS in 25m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8ea1669ce2964450a60a295a8f502cb9 : SUCCESS in 34m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8899f12cf22a464baf7aff66046219d8 : SUCCESS in 1h 01m 37s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3ff7eeb39e264e5bb9bbef4ecbd8e437 : FAILURE in 40m 26s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/cf4a618d8f2b494bb859e1b4affc4817 : FAILURE in 34m 14s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/186149da076b4e708655fb350e598fce : FAILURE in 48m 43s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4c5a23386cec4856a21383834b439672 : FAILURE in 1h 13m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/aaac4f9e07ec4f65a36a2590fc2beb11 : SUCCESS in 1h 23m 33s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/714d411d47564df88d1d794e5f70b7d9 : FAILURE in 1h 57m 58s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a26e4ddd2c084c8d8d37725f70e57297 : SUCCESS in 50m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8e52dcb7740940d6aa4a93bca48d7497 : SUCCESS in 1h 28m 36s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/81f7644768fb4042ba23b687198dbcaf : SUCCESS in 1h 41m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fb66310ad0b448d4bfbd67b959a6f6bb : SUCCESS in 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7bd5f3eaf01a7ae50506d70d9740e56a5488fb9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-14 18:27:18.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/63/845663/1/check/DellEMC_Unity_Manila_dhss_true/4af44ff/DellEMC_Unity_Manila_dhss_true/549 : FAILURE in 6h 06m 53s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/63/845663/1/check/DellEMC_Unity_Manila_dhss_false/b8fd167/DellEMC_Unity_Manila_dhss_false/572 : SUCCESS in 3h 22m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"1c214289f697c953cef9ac946981f1e9fd5d1afe","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-18 09:09:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30df59baf7280337b5891a260bf822c0beef061c","revisions":{"30df59baf7280337b5891a260bf822c0beef061c":{"kind":"REWORK","_number":1,"created":"2022-06-14 04:20:24.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/63/845663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/63/845663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/845663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/845663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/845663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/63/845663/1"}}},"commit":{"parents":[{"commit":"cc8a651faeac27731a981eb54765da72a162d93a","subject":"Split generic job by tests and reduce concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc8a651faeac27731a981eb54765da72a162d93a"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-06-14 04:20:06.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-06-14 04:20:06.000000000","tz":180},"subject":"DNM","message":"DNM\n\nChange-Id: I7cae098a8240d6702c79b5ef0b4f5823138c4e36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/30df59baf7280337b5891a260bf822c0beef061c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/30df59baf7280337b5891a260bf822c0beef061c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
