)]}'
{"id":"openstack%2Fmanila-tempest-plugin~845723","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ie9f03c0d3c18aa95b1e0fc66441978c86b79e39b","project":"openstack/manila-tempest-plugin","branch":"master","topic":"master2","attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-06-14 11:54:30.000000000","reason":"Goutham Pacha Ravi replied on the change"},"30409":{"account":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"last_update":"2022-06-14 12:54:19.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie9f03c0d3c18aa95b1e0fc66441978c86b79e39b","subject":"add list shares without details filter testcase","status":"NEW","created":"2022-06-14 10:15:25.000000000","updated":"2022-06-15 05:27:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2c20ebe684558a872f853d9bde16f68d7f6cad3e","_number":845723,"virtual_id_number":845723,"owner":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-15 05:27:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-14 11:54:30.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-06-14 12:54:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d100108d1bc8c45da4aa511a4417c127c5fdeac9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2022-06-14 10:15:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97c187c0e59e17e4c290c22daa31e5664c4c2ce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 12:54:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/559241c0fd0043179dfc5abcc668a6e1 : SUCCESS in 4m 05s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/058ec313949e4460b203646043d2d490 : FAILURE in 30m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b6ea08fec3464dcf859214b1a264e964 : FAILURE in 38m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/23f54a3153d745ac993cec2b0a89be4a : SUCCESS in 1h 03m 07s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/252b62b1b5b8443e9eeee7bb48b4bdb1 : FAILURE in 38m 04s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/856250a3063941edb390a748ca845a12 : FAILURE in 36m 25s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c6692434320e4bec9852fc46563ce1f5 : FAILURE in 38m 46s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0175e4e1686a4be0a3afc846457a4e57 : SUCCESS in 1h 10m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/2b0dfa68c7c44aeda800af9562d81801 : FAILURE in 1h 20m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bd0a3ce3644d4dc7aa6610d5218233b0 : FAILURE in 1h 26m 03s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cb22a6de065e48598581261da8dfb8a3 : FAILURE in 51m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3c0e3f6c7770498fb6ea2a16c0b56028 : SUCCESS in 1h 05m 50s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7c0d8e2b6abe44e5b7ad417c873cda98 : FAILURE in 2h 33m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d83223010d2b49b9ae77815600faf793 : FAILURE in 21m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3cede6dd14ac0989f8c53f7a267fe1a1b1721238","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2022-06-15 02:25:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2c20ebe684558a872f853d9bde16f68d7f6cad3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 05:27:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c26940d9f44f4d91a8def8a5d0524943 : SUCCESS in 5m 33s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/577bab6d95cd4c389b3143edfb2f862f : SUCCESS in 42m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7d01e67e76324fd087fc7f820036c959 : SUCCESS in 33m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e04f42215a57444cab9b07a7d430054e : FAILURE in 1h 06m 00s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/1db66a8a9cc5473e861f82db1030d172 : FAILURE in 46m 03s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/222da6fb706f4c6e8f9a729864a8ebb8 : FAILURE in 44m 26s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/fe62d1b884b243a1a892ee96bbcb13e6 : FAILURE in 44m 04s\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/732d7bb4cd9a41d99d3c0d1ecc7c6d9e : FAILURE in 1h 15m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/08e0cae723c7460e966588a1f9a5d49f : SUCCESS in 1h 35m 37s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ea35857cb4ad426c94af10bf9c91cbbc : FAILURE in 1h 12m 19s (non-voting)\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/764eae65bf3642d6bb2e293ccdb9fd33 : SUCCESS in 50m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/27f22c4ab298432cb97eaf337393000f : SUCCESS in 1h 40m 22s (non-voting)\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/315614c118744f3aac8985152c3b4fe5 : FAILURE in 2h 50m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/25bcef6c8dd54972810bbc77aff725b7 : FAILURE in 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6150e206a342d67dbd8e77dd1d3e5536100ee9aa","revisions":{"2dff67e635caab35f7abd516c8d5aa57f908a234":{"kind":"REWORK","_number":1,"created":"2022-06-14 10:15:25.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/23/845723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/23/845723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/845723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/845723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/845723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/845723/1"}}},"commit":{"parents":[{"commit":"cc8a651faeac27731a981eb54765da72a162d93a","subject":"Split generic job by tests and reduce concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc8a651faeac27731a981eb54765da72a162d93a"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2022-06-14 10:12:42.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2022-06-14 10:12:51.000000000","tz":480},"subject":"add list shares without details filter testcase","message":"add list shares without details filter testcase\n\nadd list shares without details filter by existed name and fake name\ntestcase\n\nChange-Id: Ie9f03c0d3c18aa95b1e0fc66441978c86b79e39b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2dff67e635caab35f7abd516c8d5aa57f908a234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2dff67e635caab35f7abd516c8d5aa57f908a234"}]},"branch":"refs/heads/master"},"6150e206a342d67dbd8e77dd1d3e5536100ee9aa":{"kind":"REWORK","_number":2,"created":"2022-06-15 02:25:48.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/23/845723/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/23/845723/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/845723/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/845723/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/845723/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/23/845723/2"}}},"commit":{"parents":[{"commit":"cc8a651faeac27731a981eb54765da72a162d93a","subject":"Split generic job by tests and reduce concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cc8a651faeac27731a981eb54765da72a162d93a"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2022-06-14 10:12:42.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2022-06-15 02:24:26.000000000","tz":480},"subject":"add list shares without details filter testcase","message":"add list shares without details filter testcase\n\nadd list shares without details filter by existed name and fake name\ntestcase\n\nChange-Id: Ie9f03c0d3c18aa95b1e0fc66441978c86b79e39b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6150e206a342d67dbd8e77dd1d3e5536100ee9aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/6150e206a342d67dbd8e77dd1d3e5536100ee9aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
