)]}'
{"id":"openstack%2Fmanila-tempest-plugin~848333","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I4abf0d60e6662c6f409bfa6506ab51ea717dc8cf","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-07-01 19:24:12.000000000","reason":"Change was submitted"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-07-01 18:23:44.000000000","reason":"\u003cGERRIT_ACCOUNT_30002\u003e replied on the change","reason_account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}},"hashtags":[],"change_id":"I4abf0d60e6662c6f409bfa6506ab51ea717dc8cf","subject":"Follow up for CI jobs refactoring","status":"MERGED","created":"2022-06-30 19:59:50.000000000","updated":"2022-07-01 19:24:12.000000000","submitted":"2022-07-01 19:24:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848333","meta_rev_id":"9b21b39684f96df1a43f34f34a0ccf9ad4fd310e","_number":848333,"virtual_id_number":848333,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-01 19:24:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":2,"date":"2022-07-01 18:23:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2022-06-30 20:14:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2022-06-30 20:20:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":1,"date":"2022-07-01 18:23:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-30 20:14:00.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-06-30 20:20:02.000000000","updated_by":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"reviewer":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"state":"REVIEWER"},{"updated":"2022-06-30 22:14:34.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-06-30 22:59:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-01 01:25:34.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-07-01 14:55:19.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"1b396c90f91ff332244c0eaa68b367913e62341f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-06-30 19:59:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddfbf0fee3dbc5595549d0d770e7616448bed2dc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-06-30 20:14:00.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce1cb159991ad668b852504aa9200407a8ae5efa","author":{"_account_id":32411,"name":"Eduardo Santos","email":"eduardo.experimental@gmail.com","username":"ecsantos"},"date":"2022-06-30 20:20:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ffa9dfb46004bedd525e3935faac7563d550231","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-06-30 22:14:34.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/33/848333/1/check/DellEMC_Unity_Manila_dhss_true/ce0bc89/DellEMC_Unity_Manila_dhss_true/592 : FAILURE in 2h 14m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/33/848333/1/check/DellEMC_Unity_Manila_dhss_false/c868206/DellEMC_Unity_Manila_dhss_false/620 : SUCCESS in 1h 02m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"f0f2e41de0ee57e7ee2028bb51739ae363c5dc2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 22:59:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ac28dc7c6a4dd289c2f8f3653417a5 : SUCCESS in 3m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1a60bfa06d4d44f4ada6d97565bbade4 : SUCCESS in 1h 17m 14s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4e0ac18dc13649c4bfc55fe6c07f1293 : FAILURE in 2h 52m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aa4e02b7ebb54f29b88d4101668d0abd : FAILURE in 2h 01m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/48e65934b32b4dd7b534a6eaf75684a3 : SUCCESS in 1h 22m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/05b2ae445f4d47edb63a44c8737da34c : FAILURE in 1h 46m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b8c262ae48a64a1782bf23a6f13d5408 : SUCCESS in 1h 11m 08s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/790063e81b6444e68901d481e6536256 : SUCCESS in 38m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e45ce46cb51f4f0fa66279e5bfbfd4bc : SUCCESS in 42m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8865893914f94b39928549e521df25e1 : SUCCESS in 56m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/bcce17deb9584a29ae945d20de62e5fe : FAILURE in 27m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4baa217a51b60919315c4d2dcbd0c2d568057398","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-07-01 01:25:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/848333/1/upstream-check/manila-tempest-plugin-ontap-dhss/ded32b6/ : FAILURE in 1h 29m 35s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/848333/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/607c489/ : SUCCESS in 1h 20m 59s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"ff609f79a4ec1b6e1274018127f290be525ef280","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-01 04:54:04.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/33/848333/1/check/DellEMC_Unity_Manila_dhss_true/dfaab5d/DellEMC_Unity_Manila_dhss_true/595 : FAILURE in 2h 27m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/33/848333/1/check/DellEMC_Unity_Manila_dhss_false/dce5cde/DellEMC_Unity_Manila_dhss_false/623 : SUCCESS in 1h 09m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"6ec7bfb0bb376253c5ec55207a3287280b7e5042","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-07-01 18:23:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1981672588df60f844b049e2b7ab23cb827bed8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 18:24:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8cc51bb0086c98ba06ca7f9424b7ca48542e7238","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 19:24:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb21459c9da1416299544bba6b5eed20 : SUCCESS in 3m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5d36f5384573455e84113960504c3f7c : SUCCESS in 58m 22s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/57dd33b1579b45fc879033e290ad3a94 : SUCCESS in 28m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e94dc47745404c9ca135b2cba018c540 : SUCCESS in 25m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9b21b39684f96df1a43f34f34a0ccf9ad4fd310e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 19:24:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d12285b25b40031ec90743c45b095d1bbcba4ca4","revisions":{"d12285b25b40031ec90743c45b095d1bbcba4ca4":{"kind":"REWORK","_number":1,"created":"2022-06-30 19:59:50.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/33/848333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/33/848333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/848333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/848333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/848333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/848333/1"}}},"commit":{"parents":[{"commit":"b52f1abc70f7a3cf7aab43c6fdeb3a7473d41153","subject":"Merge \"Change the set of jobs in the stable branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/b52f1abc70f7a3cf7aab43c6fdeb3a7473d41153"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-06-30 19:58:08.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2022-06-30 19:59:33.000000000","tz":-180},"subject":"Follow up for CI jobs refactoring","message":"Follow up for CI jobs refactoring\n\nThis changes addresses few comments left in the change that\nrefactored some of our CI jobs [1], regarding a more suggestive\nanchor name and a comment to give more context for code reviewers\nin the future.\n\n[1] https: //review.opendev.org/c/openstack/manila-tempest-plugin/+/847366\n\nChange-Id: I4abf0d60e6662c6f409bfa6506ab51ea717dc8cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d12285b25b40031ec90743c45b095d1bbcba4ca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d12285b25b40031ec90743c45b095d1bbcba4ca4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[]}
