)]}'
{"id":"openstack%2Fmanila-tempest-plugin~849828","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I720e1ecbe94020500457b48e3be323b1bceae1b5","project":"openstack/manila-tempest-plugin","branch":"master","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-07-18 11:04:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I720e1ecbe94020500457b48e3be323b1bceae1b5","subject":"DNM: Testing all jobs","status":"ABANDONED","created":"2022-07-15 13:28:20.000000000","updated":"2022-07-18 11:04:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f6c083e63fad7e08c4cfdcdb5ce60be2d46066cd","_number":849828,"virtual_id_number":849828,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-15 16:38:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 16:38:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 21:49:50.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"86dba284572410b76b664ee0ee42c356b1bab76b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-15 13:28:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ddb1e82ad4de0259be69dc1079c6e18f9a42d5e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-15 13:29:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"323ab9218e1a4b7b3aeb0406e25934552aeee259","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-15 13:29:28.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"1d8fc9e857976d27710367f31f381a82d3abbc20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 16:38:06.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62f7a8c31351417f80e4cc3f9ff845a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7772164d9bae4e8b9494153878f83c66 : FAILURE in 2m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d24a5ad0913544c992fb35522624208d : SUCCESS in 1h 15m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0f9d827d5b4c48d4af76310f190b343f : FAILURE in 3h 01m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eafceff0fb374e3ab58df2524d1c2695 : SUCCESS in 1h 38m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/57b5938468754dbe90a8722dc9dfcb52 : SUCCESS in 1h 48m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e50a958c57c54458ba43f8e0243c628d : FAILURE in 2h 02m 58s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a1e902038db5489182961263073293f7 : SUCCESS in 1h 12m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5d857974bca0414db169e73cf9166c1b : SUCCESS in 34m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/68093961636d4dec8acae1175d058165 : SUCCESS in 38m 42s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3c4b0e34a3fa43deaa38f668e9985f4f : SUCCESS in 1h 51m 34s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/4b5011a647df4a4784549f1cafe14e63 : SUCCESS in 1h 14m 08s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/8913053b570e4ce184b80089f4ea5f54 : SUCCESS in 1h 19m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/77b8b9bd55304490aadeed412865d6e1 : SUCCESS in 52m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2c930b8a130641da8855ccbe356266db : FAILURE in 30m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"44c518064173610dff064f09dbe4baabf69d969a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-15 21:49:50.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/28/849828/2/check/DellEMC_Unity_Manila_dhss_true/0d676c7/DellEMC_Unity_Manila_dhss_true/638 : FAILURE in 2h 31m 36s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/28/849828/2/check/DellEMC_Unity_Manila_dhss_false/d40bbce/DellEMC_Unity_Manila_dhss_false/670 : SUCCESS in 1h 02m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"f6c083e63fad7e08c4cfdcdb5ce60be2d46066cd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-18 11:04:12.000000000","message":"Abandoned\n\nPatch has now merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aae85435e2607bd03686d4696d268d5ad6f81677","revisions":{"4835226ee123817c7140f6c467caadab42fb482e":{"kind":"REWORK","_number":1,"created":"2022-07-15 13:28:20.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/28/849828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/28/849828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/849828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/849828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/849828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/849828/1"}}},"commit":{"parents":[{"commit":"0e4a3d2a877c4e85d73e3f3a318c5dfaa4624a27","subject":"Update tempest version required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0e4a3d2a877c4e85d73e3f3a318c5dfaa4624a27"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-15 13:28:20.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-15 13:28:20.000000000","tz":0},"subject":"Testing all jobs","message":"Testing all jobs\n\nChange-Id: I720e1ecbe94020500457b48e3be323b1bceae1b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4835226ee123817c7140f6c467caadab42fb482e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/4835226ee123817c7140f6c467caadab42fb482e"}]},"branch":"refs/heads/master"},"aae85435e2607bd03686d4696d268d5ad6f81677":{"kind":"REWORK","_number":2,"created":"2022-07-15 13:29:15.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/28/849828/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/28/849828/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/849828/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/849828/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/849828/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/28/849828/2"}}},"commit":{"parents":[{"commit":"0e4a3d2a877c4e85d73e3f3a318c5dfaa4624a27","subject":"Update tempest version required","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/0e4a3d2a877c4e85d73e3f3a318c5dfaa4624a27"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-15 13:28:20.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-15 13:29:14.000000000","tz":0},"subject":"DNM: Testing all jobs","message":"DNM: Testing all jobs\n\nChange-Id: I720e1ecbe94020500457b48e3be323b1bceae1b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/aae85435e2607bd03686d4696d268d5ad6f81677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/aae85435e2607bd03686d4696d268d5ad6f81677"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
