)]}'
{"id":"openstack%2Fmanila-tempest-plugin~850064","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I6cfc413b3588e5d75a6ac6c9ebe6835541d4c4bc","project":"openstack/manila-tempest-plugin","branch":"master","topic":"stable-jobs","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-07-28 20:11:42.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-08-03 20:04:59.000000000","reason":"\u003cGERRIT_ACCOUNT_30002\u003e replied on the change","reason_account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-08-03 21:20:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6cfc413b3588e5d75a6ac6c9ebe6835541d4c4bc","subject":"[CI] Drop stable/victoria job configuration","status":"MERGED","created":"2022-07-15 19:10:54.000000000","updated":"2022-08-03 21:20:21.000000000","submitted":"2022-08-03 21:20:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850064-stable-jobs","meta_rev_id":"46201b45f8a2a3462742d581204c8e0ac2311e44","_number":850064,"virtual_id_number":850064,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-03 21:20:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-07-28 20:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-08-03 20:04:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2022-08-03 20:04:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 19:11:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-16 00:39:49.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-07-18 11:03:37.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-07-28 19:14:16.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"0ba0a1e6165fe7d8d969a655360a833fb32c96e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-15 19:10:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d421c098a0f436dce716b8eac6c235afac4f9131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 19:11:15.000000000","message":"Patch Set 1: Verified-1\n\n(11 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 67, column 5\ndid not find expected key\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 74, column 7","accounts_in_message":[],"_revision_number":1},{"id":"f3313a8fdebfe2245bb621430b450bc6cc13bcfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-15 19:15:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6ded140b1fb3373ee64fa9ba55321f55c2fc4664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 19:16:00.000000000","message":"Patch Set 2: Verified-1\n\n(10 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/manila-tempest-plugin on branch master.  The error was:\n\n  while parsing a block mapping\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 67, column 5\ndid not find expected key\n  in \"openstack/manila-tempest-plugin/zuul.d/manila-tempest-jobs.yaml@master\", line 74, column 7","accounts_in_message":[],"_revision_number":2},{"id":"50147d58c2d0f46cb3228537383bd29ab9658e8f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-16 00:39:49.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/64/850064/2/check/DellEMC_Unity_Manila_dhss_true/584cd01/DellEMC_Unity_Manila_dhss_true/639 : FAILURE in 2h 49m 59s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/64/850064/2/check/DellEMC_Unity_Manila_dhss_false/6dc8e5d/DellEMC_Unity_Manila_dhss_false/672 : SUCCESS in 58m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"9fda1e1506cb756e8a78e038c0962ae217b6ee13","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-18 11:03:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b9ace8037bd469935128502440b3321e4a36649","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-07-18 16:30:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be84259f84b2471d074cb0c6d06ebf11782d3234","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-19 18:33:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"875e6a05e905b788c4d4f59886711959aa221bb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 18:33:52.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":3},{"id":"9c4590cf82d1bda876fbddb19f84c16f9e050308","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-19 18:35:55.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a08a4ce0f591c1ea8a2449475f99dd4207e9f5b6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-19 18:45:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"36f30a037dfd4312b632938f136fa060343646ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 20:39:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4d55fd340f94eec80b3b0f8b27dad5b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a68d80bebc7470db5bbcc9521032ec7 : SUCCESS in 4m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bff31837453c4a988ed26eb37bee8748 : SUCCESS in 1h 22m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/925292f6c3d94df792740fc74f519316 : SUCCESS in 1h 51m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7b66590cbb0d458c9d069bb3c46bfa50 : SUCCESS in 1h 04m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/efddb9bfc76743918fb57704903afa39 : FAILURE in 1h 16m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9e2ef233b0114fdba71868a88ed4680c : FAILURE in 1h 37m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/37a7e5475a194bb0bf21e44e5ec1d786 : SUCCESS in 1h 07m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/46c1a8c8a01f4e5284cebaaef50a49a3 : SUCCESS in 29m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d0f32779434a4b4d99dc9602a9cd38d7 : SUCCESS in 40m 52s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/268eb1756e5242229effeb9c1488ca8c : SUCCESS in 1h 06m 11s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d1e6a4a7212a44b59102fec1ecbb6681 : SUCCESS in 1h 03m 08s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/c907c2e8d1b245a28e842c2b49cd78b8 : SUCCESS in 56m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f838aef57fdc4e7681569f144d6a8743 : SUCCESS in 53m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7815591ec89a4f42976d0f7bfd12e52f : SUCCESS in 26m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8af6b0f77b8598326f49e51a99ba4b24ff607837","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-07-19 23:15:20.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/64/850064/3/check/DellEMC_Unity_Manila_dhss_true/5719e2a/DellEMC_Unity_Manila_dhss_true/649 : FAILURE in 2h 35m 44s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/64/850064/3/check/DellEMC_Unity_Manila_dhss_false/9350c1b/DellEMC_Unity_Manila_dhss_false/682 : SUCCESS in 1h 01m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"4a3d145654d13bacd0bc2c6d9781590d0093fc68","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-20 14:56:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f8dfca0bd4d3732a365f850266033df8ff1b6ac6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-07-20 14:56:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"008d13eaaa9d1aa2a5bed6e5072b4a36ac7e7a3e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-07-28 20:11:42.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"77c8f7edd2f590300819839860e4b12907ef857a","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-08-03 20:04:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f9f83f76414f0720d4a89ce9d15c8c683b3bf0b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 20:05:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"766ee03a8d1d96693acf152f9db6247388d9f06c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 21:20:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3e4db1a1c8d4f78bdc7419572bd7723\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c7d4b9dcd3c497b95c55990d1e81d85 : SUCCESS in 2m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e0f638920d054fc89bfc7a41879986b1 : SUCCESS in 1h 10m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/19e7af5fbba14d45a0cc2ef5c60686f2 : SUCCESS in 39m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3db4f95bfaa941ce8406eda5b224e3ab : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":3},{"id":"46201b45f8a2a3462742d581204c8e0ac2311e44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 21:20:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9ac47cbab87b03e960f492b4a3a030eb068e9f23","revisions":{"ee7a3d939c2da62341efd88f122e50c9b8dbc34c":{"kind":"REWORK","_number":1,"created":"2022-07-15 19:10:54.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/64/850064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/64/850064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/850064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/850064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/850064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/850064/1"}}},"commit":{"parents":[{"commit":"07f39f392c9abf33a49dc5c963c5907d7dc76579","subject":"Merge \"Remove setup_clients\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/07f39f392c9abf33a49dc5c963c5907d7dc76579"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-15 19:10:28.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-15 19:10:28.000000000","tz":330},"subject":"[CI] Pin manila-tempest-plugin in stable/victoria","message":"[CI] Pin manila-tempest-plugin in stable/victoria\n\nTempest is pinned to a 26.1.0 on stable/victoria\nbranch [1]. manila-tempest-plugin in Zed requires a newer\ntempest release [2] than this one. So, pin\nmanila-tempest-plugin as well to a version that can still\nwork with the pinned version of tempest.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/838051\n[2] https://review.opendev.org/c/openstack/manila-tempest-plugin/+/849995\n\nChange-Id: I6cfc413b3588e5d75a6ac6c9ebe6835541d4c4bc\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee7a3d939c2da62341efd88f122e50c9b8dbc34c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/ee7a3d939c2da62341efd88f122e50c9b8dbc34c"}]},"branch":"refs/heads/master"},"a40b395cc6a18f63c1a05feb941d58f2e456d24d":{"kind":"REWORK","_number":2,"created":"2022-07-15 19:15:33.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/64/850064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/64/850064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/850064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/850064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/850064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/850064/2"}}},"commit":{"parents":[{"commit":"07f39f392c9abf33a49dc5c963c5907d7dc76579","subject":"Merge \"Remove setup_clients\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/07f39f392c9abf33a49dc5c963c5907d7dc76579"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-15 19:10:28.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-15 19:15:12.000000000","tz":330},"subject":"[CI] Pin manila-tempest-plugin in stable/victoria","message":"[CI] Pin manila-tempest-plugin in stable/victoria\n\nTempest is pinned to a 26.1.0 on stable/victoria\nbranch [1]. manila-tempest-plugin in Zed requires a newer\ntempest release [2] than this one. So, pin\nmanila-tempest-plugin as well to a version that can still\nwork with the pinned version of tempest.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/838051\n[2] https://review.opendev.org/c/openstack/manila-tempest-plugin/+/849995\n\nChange-Id: I6cfc413b3588e5d75a6ac6c9ebe6835541d4c4bc\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a40b395cc6a18f63c1a05feb941d58f2e456d24d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a40b395cc6a18f63c1a05feb941d58f2e456d24d"}]},"branch":"refs/heads/master"},"9ac47cbab87b03e960f492b4a3a030eb068e9f23":{"kind":"REWORK","_number":3,"created":"2022-07-19 18:33:20.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/64/850064/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/64/850064/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/850064/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/850064/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/850064/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/64/850064/3"}}},"commit":{"parents":[{"commit":"07f39f392c9abf33a49dc5c963c5907d7dc76579","subject":"Merge \"Remove setup_clients\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/07f39f392c9abf33a49dc5c963c5907d7dc76579"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-15 19:10:28.000000000","tz":330},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-07-19 18:32:32.000000000","tz":330},"subject":"[CI] Drop stable/victoria job configuration","message":"[CI] Drop stable/victoria job configuration\n\nWe\u0027ll need to pin manila-tempest-plugin in stable/victoria\nwhen testing manila [1]; since the configuration is a special\nsnowflake for this branch alone, lets not carry it in this\nbranchless repository any more.\n\nDepends-On: I23827477d3c25eb2b10bea6d62885d50a28ec0a0\n[1] https://review.opendev.org/c/openstack/manila/+/850415\n\nChange-Id: I6cfc413b3588e5d75a6ac6c9ebe6835541d4c4bc\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ac47cbab87b03e960f492b4a3a030eb068e9f23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9ac47cbab87b03e960f492b4a3a030eb068e9f23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[]}
