)]}'
{"id":"openstack%2Fmanila-tempest-plugin~853367","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I5b37f31e302ce75650f817a937d8f17ee58887b2","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1940324","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-08-18 11:15:28.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-08-19 22:31:19.000000000","reason":"\u003cGERRIT_ACCOUNT_30002\u003e replied on the change","reason_account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-08-24 21:05:34.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2022-08-24 08:09:31.000000000","reason":"\u003cGERRIT_ACCOUNT_30407\u003e replied on the change","reason_account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-08-24 20:03:17.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2022-08-24 22:38:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5b37f31e302ce75650f817a937d8f17ee58887b2","subject":"Use OVS with IPv6 data path testing jobs","status":"MERGED","created":"2022-08-16 21:07:23.000000000","updated":"2022-08-24 22:38:52.000000000","submitted":"2022-08-24 22:38:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"853367-bug/1940324","meta_rev_id":"e7320027b56220773c67fab64a55c518f73974ae","_number":853367,"virtual_id_number":853367,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-24 22:38:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-08-19 22:31:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2022-08-24 08:09:31.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":2,"date":"2022-08-24 20:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2022-08-24 21:05:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-16 22:15:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-16 22:24:51.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-08-17 14:30:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-08-19 18:46:25.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2022-08-19 18:46:25.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-08-19 18:49:11.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-08-19 18:49:11.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2022-08-19 21:49:29.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"860280bec8213e7cf677101ad2c5bf0caa022297","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-08-16 21:07:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b04e5c4161feaec535d2d22ebcedfc233c03420","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 22:15:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f07fb3fd6384b28862e91117b42f9f1\n\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c3b6699694124ca78ec3abc0bb387cfc : SUCCESS in 1h 06m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ef7e44dcff364226bf8bf1b3d79f865c : SUCCESS in 58m 49s","accounts_in_message":[],"_revision_number":1},{"id":"29c0c6d0f4afa9b163c20389e5231d2463b6c33a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-16 22:24:51.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/67/853367/1/check/DellEMC_Unity_Manila_dhss_true/e5ccd0c/DellEMC_Unity_Manila_dhss_true/710 : FAILURE in 1h 17m 12s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/67/853367/1/check/DellEMC_Unity_Manila_dhss_false/2f94bc7/DellEMC_Unity_Manila_dhss_false/747 : FAILURE in 16m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"f99131c3a7121364823e7cb292ee4c4720e99e09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-08-17 00:50:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5114568c279fbc8bfa7d8eb5791d39dc90a561e0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-17 02:02:46.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/67/853367/2/check/DellEMC_Unity_Manila_dhss_true/94cb490/DellEMC_Unity_Manila_dhss_true/712 : FAILURE in 1h 12m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/67/853367/2/check/DellEMC_Unity_Manila_dhss_false/40284b6/DellEMC_Unity_Manila_dhss_false/749 : FAILURE in 14m 37s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"eb4e622e17930f2e742e1cda24526a136d816aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 02:59:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6a4b310786f442aa6327971dd34329e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49ce03a48dab4d25858cb2d8f7c49ca0 : SUCCESS in 3m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5f07e05572c1413bbd633fd7e553937e : FAILURE in 1h 28m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ae8893da8e404591ac71f40374d1c339 : FAILURE in 2h 07m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1d69efe1c0e148e88187f503b5da31c8 : SUCCESS in 1h 04m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b83c20c5ba264effb47a33a2b1b3d3f3 : SUCCESS in 1h 25m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/69c18db40b864ca1bc6192d635705337 : FAILURE in 1h 49m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6b69ea5da76e4005b5a1c9270157ad33 : SUCCESS in 1h 00m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ed9399b1ee3c4c1291c9eb9b5d6a34d9 : SUCCESS in 29m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/953aaf23ce68454d80fdd65c2f8e6f0d : SUCCESS in 31m 07s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/1c84f9b1f52147b8b6ba6a69a4aa58eb : SUCCESS in 1h 19m 28s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/02fa76fc0b264b5eb64ea0eca88ac370 : FAILURE in 1h 05m 07s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/f8f1eeb296c742e0af0fb4f5c7c0630f : SUCCESS in 1h 19m 51s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dbd5afd348ad4c8ba542598ab7350bc4 : SUCCESS in 44m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/27cb77c7703744daad14f3fd514534a7 : SUCCESS in 25m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9609c29a43327ce66f57151d37d7205778b13bd4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-08-17 20:50:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df56d430e9de7fa6317b743c048d1bbe8bc353f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 23:39:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5577d2aad122413781a32356b0a188eb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39d1e4efa6c44bcd9daaf056d3b56786 : SUCCESS in 3m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ecf6dc1402a74cbf97695b7b4e125350 : SUCCESS in 1h 20m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/39604a96772a473d9e8d2c536a2cffca : FAILURE in 2h 25m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6d6f20b5b38e4ef5946141412f38a211 : SUCCESS in 1h 00m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f9fb01f23d7e4cec8cc521f8ffe46b0b : SUCCESS in 1h 19m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3cc49c871ac3442da1e00c577071aec4 : FAILURE in 2h 47m 59s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cf11153fcb8945bc9742624280d13aa9 : FAILURE in 1h 14m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0f311ba4ab464c4590a601fccbfa04e2 : SUCCESS in 30m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/56edbb79c9e9497f9909c169d3d55c80 : SUCCESS in 31m 28s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/3682970fdb5445a6bfb725c73100fc73 : SUCCESS in 1h 17m 35s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a79e7713040b42c1a2625aaa9e787ab5 : SUCCESS in 1h 08m 44s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/6f1c26c2e4cf4f74af3b7d51315b4838 : SUCCESS in 1h 13m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9f6cea1d5a6542f8826099f427e780b2 : SUCCESS in 47m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8a0232d606464120bd83143a1ad2d6fd : SUCCESS in 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d8a4fbadac577da971b82f00429f227dc783e2f9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-18 01:05:59.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/67/853367/2/check/DellEMC_Unity_Manila_dhss_true/950c624/DellEMC_Unity_Manila_dhss_true/724 : FAILURE in 1h 16m 40s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/67/853367/2/check/DellEMC_Unity_Manila_dhss_false/60c8bff/DellEMC_Unity_Manila_dhss_false/761 : FAILURE in 17m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"590be452dc26b1cee59f19466980c5da7e928c7f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-18 11:15:28.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2187bc53cdac6bcc961152f4279a84b22e44ea0e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-08-19 19:27:26.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8e08991378454b6cd4cd1e58092780f86ede832a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-08-19 19:33:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6a1f09530c3d1890fd3dc1eaf2db7439a3831b32","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-19 19:34:25.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/67/853367/3/check/DellEMC_Unity_Manila_dhss_true/6bf618c/DellEMC_Unity_Manila_dhss_true/735 : FAILURE in 2s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/67/853367/3/check/DellEMC_Unity_Manila_dhss_false/5c6ea3f/DellEMC_Unity_Manila_dhss_false/772 : FAILURE in 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"a47c1a270815005eacf9ea4d9ecbf4b80832f3d3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-08-19 19:35:56.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2f24fe0b04d4a1f5408a58b75682a49fe7801270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 21:35:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64deacf44a6b43f69f424cd72181234b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da20a8b4509448f0800a5106c38366cb : SUCCESS in 3m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a8e9027ae3de455884ce776ca8ed5815 : SUCCESS in 1h 10m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7af7d99df6d6494a98e5728d6fa76ed1 : SUCCESS in 1h 51m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/597ebfc6739740b6a51e94ecc0c7ae4b : SUCCESS in 1h 14m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c44344de5097485a9bdb7757fc2eb09d : SUCCESS in 1h 17m 12s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6024857cd3df4ed4977cbc2afdc0d291 : FAILURE in 1h 42m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/71a0877ca1a0414c93b6cc5b6966db07 : FAILURE in 1h 09m 55s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/35b66224a13247cd83cb305931cc7795 : SUCCESS in 30m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2b5244258b2c49c095ea7680a041c707 : SUCCESS in 31m 11s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/0047172d87a14a40af46d0b313579a1b : SUCCESS in 1h 27m 06s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/fb614c5ff3d143eb94f0dc4296c9203b : SUCCESS in 1h 00m 14s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/2856ad86593b49b487b683c7e1856099 : SUCCESS in 1h 06m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a9e0887cdd1b455b8e43d820a938aef8 : FAILURE in 55m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1fe19729f34a44e7a9a6b29999c9ad32 : SUCCESS in 22m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"54abc0d5e926c0da3bfeecfa57117151f4214708","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-08-19 21:36:05.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/67/853367/3/check/DellEMC_Unity_Manila_dhss_true/7dcdc02/DellEMC_Unity_Manila_dhss_true/738 : FAILURE in 1s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/67/853367/3/check/DellEMC_Unity_Manila_dhss_false/82bb6da/DellEMC_Unity_Manila_dhss_false/775 : FAILURE in 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"d7b9e819f4a1221a1f082f6422b233da302e12ec","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-08-19 21:49:29.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/853367/3/upstream-check/manila-tempest-plugin-ontap-dhss/fd49851/ : RETRY_LIMIT in 7m 52s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/853367/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/6d2fbd5/ : RETRY_LIMIT in 12m 40s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"2e16ff522d3dd967fb691b1b79af7c553083acfc","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-08-19 22:31:19.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a89e2e7851fcbe00d0c7eba795ff0ad856f86b1f","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-08-24 08:09:31.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cff8caf7a7b9070fce09cfc47d82d8c383e8d0de","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-08-24 20:03:17.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7d3941b63e0ca8eed14e552e08d0defdacb5abb0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-08-24 21:05:34.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dd287132355b28062856b938c66a07b620856a75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 21:06:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c6b7400ef52e98a544e2b56c9306a5bca7e641a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 22:38:51.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f34605dc4e74d5b96b6aa80a2843954\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1beafb6737224db6a7a1a2b52f18b481 : SUCCESS in 3m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/27de1f8ff4814085b20af09085635a96 : SUCCESS in 1h 32m 00s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9f291443f0a6429cba9d59016ad1abe9 : SUCCESS in 29m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c4b305787b40411db585e99c95124477 : SUCCESS in 34m 04s","accounts_in_message":[],"_revision_number":3},{"id":"e7320027b56220773c67fab64a55c518f73974ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 22:38:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f67893ccccb59151b69719bf44c3a09aae541b11","revisions":{"70ff12b1824729f2940b73e83bf9b6bde2078ae9":{"kind":"REWORK","_number":1,"created":"2022-08-16 21:07:23.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/67/853367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/67/853367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/853367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/853367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/853367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/853367/1"}}},"commit":{"parents":[{"commit":"c81501c91051d47cb26be93d0b9b84dcff84f3e6","subject":"remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c81501c91051d47cb26be93d0b9b84dcff84f3e6"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-08-16 21:07:06.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-08-16 21:07:06.000000000","tz":-420},"subject":"[DNM] Enable OVS for IPv6 testing with LVM","message":"[DNM] Enable OVS for IPv6 testing with LVM\n\nChange-Id: I5b37f31e302ce75650f817a937d8f17ee58887b2\nCloses-Bug: #1940324\nRelated-Bug: #1959666\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70ff12b1824729f2940b73e83bf9b6bde2078ae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/70ff12b1824729f2940b73e83bf9b6bde2078ae9"}]},"branch":"refs/heads/master"},"3a3210b00626b5167c9a622472faf26b52a8a5df":{"kind":"REWORK","_number":2,"created":"2022-08-17 00:50:17.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/67/853367/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/67/853367/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/853367/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/853367/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/853367/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/853367/2"}}},"commit":{"parents":[{"commit":"c81501c91051d47cb26be93d0b9b84dcff84f3e6","subject":"remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c81501c91051d47cb26be93d0b9b84dcff84f3e6"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-08-16 21:07:06.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-08-17 00:50:01.000000000","tz":-420},"subject":"Use OVS with IPv6 data path testing jobs","message":"Use OVS with IPv6 data path testing jobs\n\nWe run IPv6 data-path tests with LVM and\nCeph-NFS drivers by setting up bgp to\nadvertise tenant network routes to the host.\nBug #1959666 prevents the use of OVN with\nneutron-dynamic-routing. The underlying neutron\nML2 mechanism driver doesn\u0027t matter to\nmanila; so lets use OVS in the meanwhile with\nthese test jobs.\n\nCloses-Bug: #1940324\nRelated-Bug: #1959666\nChange-Id: I5b37f31e302ce75650f817a937d8f17ee58887b2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a3210b00626b5167c9a622472faf26b52a8a5df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3a3210b00626b5167c9a622472faf26b52a8a5df"}]},"branch":"refs/heads/master"},"f67893ccccb59151b69719bf44c3a09aae541b11":{"kind":"REWORK","_number":3,"created":"2022-08-19 19:33:54.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/67/853367/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/67/853367/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/853367/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/853367/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/853367/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/67/853367/3"}}},"commit":{"parents":[{"commit":"c81501c91051d47cb26be93d0b9b84dcff84f3e6","subject":"remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c81501c91051d47cb26be93d0b9b84dcff84f3e6"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-08-16 21:07:06.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-08-19 19:33:36.000000000","tz":-420},"subject":"Use OVS with IPv6 data path testing jobs","message":"Use OVS with IPv6 data path testing jobs\n\nWe run IPv6 data-path tests with LVM and\nCeph-NFS drivers by setting up bgp to\nadvertise tenant network routes to the host.\nBug #1959666 prevents the use of OVN with\nneutron-dynamic-routing. The underlying neutron\nML2 mechanism driver doesn\u0027t matter to\nmanila; so lets use OVS in the meanwhile with\nthese test jobs.\n\nCloses-Bug: #1940324\nRelated-Bug: #1959666\nChange-Id: I5b37f31e302ce75650f817a937d8f17ee58887b2\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f67893ccccb59151b69719bf44c3a09aae541b11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f67893ccccb59151b69719bf44c3a09aae541b11"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[]}
