)]}'
{"id":"openstack%2Fmanila-tempest-plugin~857933","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I27fdd7dfffeb15965b66dbb3f6b1568c11ff9ad4","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1901210","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-09-16 12:08:21.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-09-16 12:08:21.000000000","reason":"Change was submitted"},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2022-09-16 11:05:16.000000000","reason":"\u003cGERRIT_ACCOUNT_6413\u003e replied on the change","reason_account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}},"hashtags":[],"change_id":"I27fdd7dfffeb15965b66dbb3f6b1568c11ff9ad4","subject":"Expect 404 for unauthorized share access","status":"MERGED","created":"2022-09-15 16:36:54.000000000","updated":"2022-09-16 12:08:21.000000000","submitted":"2022-09-16 12:08:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857933-bug/1901210","meta_rev_id":"0c84f66f69f82c654fb10c59363d005fd163d89d","_number":857933,"virtual_id_number":857933,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-16 12:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-09-15 22:52:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-16 11:05:16.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-16 11:05:16.000000000","_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-15 18:38:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-15 21:01:54.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-09-15 22:52:22.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-09-16 00:32:48.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-09-16 10:21:31.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2022-09-16 10:21:31.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"aa75bbca2af903cdf69cb9d0f341581a903b991b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-15 16:36:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bae5e1e5e0db72e9cd63ac2aea902ce4bc8c8e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 18:38:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc7bbcd2b5ac4f7d81167ecec4290d04\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c47d16871784c939f1d076f5d4c0e06 : SUCCESS in 3m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f2ea4ca5d73c488ab24cbddf6c3263e9 : FAILURE in 1h 09m 40s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8116637df14b45a1bbbcc46184b09f62 : FAILURE in 1h 59m 56s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2d7b2d66e9b24f57b40390b2dd0f67e9 : SUCCESS in 1h 21m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/3d3b2aac5a91434cbd6b855bf5fb53cf : SUCCESS in 1h 16m 26s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f7b1322cb6f0448ca80cae3acac76751 : FAILURE in 1h 56m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/49f707486c3e47148392ddb527909838 : FAILURE in 1h 32m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/966f3f9d86ce4b00bc21736eb09160ae : SUCCESS in 33m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a1247d591b3a4129b66c1398525fdc22 : SUCCESS in 44m 02s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/5ec58254fbc440ee9760813ae4bbe9ab : SUCCESS in 1h 08m 35s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/cd308f69dd534718945f5502add6c538 : SUCCESS in 1h 06m 34s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/f1f7294473804b3a868cd4b3a0a30c34 : SUCCESS in 1h 06m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ed3a3fed7ebc45b2bb99845ab431d8b8 : SUCCESS in 55m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b048e17616154a8d8864484443be69de : SUCCESS in 27m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c1849e846da4944914f2daaaf6005425e002761","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-09-15 18:40:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6fdc93445ed79998a54e16132be60b1db91c833f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 20:42:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6821d9235e7244fd9ef77ab503c56041\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b90c1576f7bd4eb98e355adc442bfd97 : SUCCESS in 3m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e7742130995647ad9be351ec75cc9c57 : SUCCESS in 1h 13m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/70d32c33e7094e10a2ac422dec857a05 : SUCCESS in 2h 00m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b9e7ece1506b4cf3b39561eba883abfb : SUCCESS in 1h 03m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6229ab051c5b43d4b18c1135c5de07b7 : FAILURE in 1h 08m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a85526dbc4bf498dbdacfe996d38761e : FAILURE in 1h 44m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/63b3e1b4343f491eac6865e431ed52a9 : SUCCESS in 1h 17m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/38b57d30ef514c68b0c0054e31193318 : SUCCESS in 33m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b2cff1b875be41babf5d1daf22793bf3 : SUCCESS in 42m 33s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/40e906c125fe4c20b4053a8c506a9799 : SUCCESS in 1h 07m 01s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/1e565686edc145d2ad9289cab2c87ebe : SUCCESS in 57m 43s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/244f11035f8e4099a48c46871504218b : SUCCESS in 1h 03m 17s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1168ab1eede04842a616e6a47945dc73 : SUCCESS in 58m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3753b5d3a93d44f0a97581849e35ac58 : SUCCESS in 24m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dbb01cf42f41e7cfa39532ba98709083782eb48a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-15 21:01:54.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/857933/1/upstream-check/manila-tempest-plugin-ontap-dhss/ee20938/ : FAILURE in 10m 33s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/33/857933/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/338581c/ : FAILURE in 13m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"a48cc449bdb408f06d13e02f94850bc70b7f61d7","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-09-15 22:52:22.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97d18761f54b4079bf79849e330e66c827c2f02d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-16 00:32:48.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/33/857933/1/check/DellEMC_Unity_Manila_dhss_true/a99a0db/DellEMC_Unity_Manila_dhss_true/871 : FAILURE in 1h 38m 26s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/33/857933/1/check/DellEMC_Unity_Manila_dhss_false/087e81e/DellEMC_Unity_Manila_dhss_false/924 : FAILURE in 45m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"71fb6f3b820b11173ea9a43619fdb4c8cfea996d","author":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"date":"2022-09-16 11:05:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ac0b83da1134ac98b904d7c6391028a189eca2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 11:05:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e0d11146af4b87a8d41dfa0709ea6183dc9e25c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 12:08:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/742a547ebb814c95ba0e95a4abb99636\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb3c8cdafd194b47b95a735f4fc5a7e2 : SUCCESS in 3m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b9e26461229840bcb826cb100cffa1b6 : SUCCESS in 1h 00m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b5228f392e774b6fa3576833bf90144d : SUCCESS in 27m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b849cc9740e54b308f9f9ef281f72b77 : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0c84f66f69f82c654fb10c59363d005fd163d89d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 12:08:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a90e3251d655ada1499c6f627a66036b8ee38a6f","revisions":{"a90e3251d655ada1499c6f627a66036b8ee38a6f":{"kind":"REWORK","_number":1,"created":"2022-09-15 16:36:54.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/33/857933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/33/857933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/857933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/857933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/857933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/33/857933/1"}}},"commit":{"parents":[{"commit":"2b3fcc571f217c14042f0d6991f59087fbfb0385","subject":"Merge \"Use ddt.named_data decorator to avoid random test names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b3fcc571f217c14042f0d6991f59087fbfb0385"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-09-15 16:36:32.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-09-15 16:36:32.000000000","tz":-420},"subject":"Expect 404 for unauthorized share access","message":"Expect 404 for unauthorized share access\n\nAs a fix to bug #1901210, manila\u0027s API now responds with\na 404 if a user doesn\u0027t have access to a non-public share to\nprevent an existence detection attack. Since manila tempest tests\ncan be used for older branches without the fix, we must accommodate\nboth error codes - 403 and 404 in our tests for a while.\n\nChange-Id: I27fdd7dfffeb15965b66dbb3f6b1568c11ff9ad4\nPartial-Bug: #1901210\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a90e3251d655ada1499c6f627a66036b8ee38a6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a90e3251d655ada1499c6f627a66036b8ee38a6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}}]}],"submit_requirements":[]}
