)]}'
{"id":"openstack%2Fmanila-tempest-plugin~858351","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I74f0a079edb59e376d045fe9e9fd781acd70249d","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bp/transfer-share-between-project","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-09-22 12:48:34.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-09-21 02:13:22.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2023-09-22 13:54:41.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2023-09-22 13:54:41.000000000","reason":"Change was submitted"},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2023-09-22 13:54:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I74f0a079edb59e376d045fe9e9fd781acd70249d","subject":"Add share transfer test.","status":"MERGED","created":"2022-09-20 09:14:03.000000000","updated":"2023-09-22 13:54:41.000000000","submitted":"2023-09-22 13:54:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":96,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"858351-bp/transfer-share-between-project","meta_rev_id":"11f7971b25eeded4e980d89b490d73200aae6faa","_number":858351,"virtual_id_number":858351,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-22 13:54:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-09-22 12:48:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-09-21 02:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-09-22 12:48:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35049,"name":"Helena","email":"helenamylena@gmail.com","username":"helenadantas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 11:27:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-20 12:28:28.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-09-21 10:22:12.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-02-02 00:29:11.000000000","updated_by":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"reviewer":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"state":"CC"},{"updated":"2023-02-04 02:26:12.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-02-13 18:09:58.000000000","updated_by":{"_account_id":35049,"name":"Helena","email":"helenamylena@gmail.com","username":"helenadantas"},"reviewer":{"_account_id":35049,"name":"Helena","email":"helenamylena@gmail.com","username":"helenadantas"},"state":"CC"},{"updated":"2023-02-17 14:15:09.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2023-02-20 22:40:59.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2023-02-20 22:40:59.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"state":"REVIEWER"},{"updated":"2023-09-15 07:35:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-15 10:13:10.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-09-22 12:48:34.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"b6905a5151fa52ba79e6546e0094305bcab7ec50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-09-20 09:14:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0b23bf75a1c926d99aa267146e282fd422c3cd9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-09-20 09:14:37.000000000","message":"Topic set to bp/transfer-share-between-project","accounts_in_message":[],"_revision_number":1},{"id":"06f04a62a3f2446be6035baac1a5d15cf55bd985","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-09-20 09:16:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e05ac12e0a7e9a0993548eb0f7db08d467e9b76a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 11:27:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0894f7fe31ce44c2b493b4a40a7a1bc3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bf460b7da1d40218d36d4620cf48801 : SUCCESS in 3m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/601101309ffe42dda8bce56953c1992d : SUCCESS in 1h 23m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/39e88d7f6ea142bfa2a17e27b02aaf4f : SUCCESS in 2h 04m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6f89068b187749529c77c8dd752d993b : SUCCESS in 1h 11m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/f258f2175ed24b43b32b48a7cce603bc : FAILURE in 1h 19m 00s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3e8b8904afb5425eaf676811e9b523b2 : FAILURE in 1h 38m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3560225b2d5847929cb56bb75ed09594 : FAILURE in 1h 35m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/292ae0a9812b40188327337384875bc4 : FAILURE in 27m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e76f193e6ca14ee39de75cbe46a030b5 : FAILURE in 33m 39s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/11b39543f7674e8f92cf8069e5dbf6bf : SUCCESS in 1h 13m 55s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/4f0d774b4a644f45a4067d9e943d08cf : SUCCESS in 57m 52s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d7afe75e0dfc4dc9a9310d01459f5753 : SUCCESS in 1h 04m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f193399256284366992368d0beaec66d : FAILURE in 50m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/30bdb0bf5a484abcba3b9fafae761128 : FAILURE in 25m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"82269e90201865ea1519d31d21831665828874ca","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-20 12:28:28.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/2/check/DellEMC_Unity_Manila_dhss_true/f13fba2/DellEMC_Unity_Manila_dhss_true/898 : FAILURE in 3h 11m 27s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/2/check/DellEMC_Unity_Manila_dhss_false/4e4fc5f/DellEMC_Unity_Manila_dhss_false/955 : FAILURE in 1h 03m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"3798f02aeccea08559a9c1c75d0629660820f20a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-09-20 14:34:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"72f2ffb9126662e94c45a4c3bd3cefdc79eb0da4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-20 14:50:50.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"b8613dab6e7ffffd0536156f54761afda3f53a17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 14:50:55.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/843832 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"4b67516355ad80911c211e24454505a45945f14d","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-09-20 14:52:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2c5ce4d9959ddfade6259b228228408e83cbce6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-09-20 16:27:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c2f6ed9616cb93ab38df23fabd15d7f0228f46bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 18:23:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c698aa340b0b460e9ac318fca98cda69\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1264cbee3b7547e5b8d14cab94edf244 : SUCCESS in 3m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5ce87d2275f74f3f8c253b8431d86a15 : SUCCESS in 52m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d99c970201904d91a10ce6e1dd286443 : SUCCESS in 1h 34m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/58c165e8ca354a07a982929dfb00a52d : SUCCESS in 1h 10m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/6744b0fce3744659ab809c158497ad7a : FAILURE in 1h 01m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3d8eabf398484ed8af18f304619e21a6 : FAILURE in 1h 54m 07s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/818da9fb4b264f40bcb7b91b2c7033e0 : FAILURE in 1h 14m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a63d66ac467b4751ad2a7a20cef0df75 : FAILURE in 31m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fc17c5597f894726a8e3cf579da7b3df : FAILURE in 44m 35s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/72e91005598441389488a0e05bdfe433 : SUCCESS in 1h 22m 43s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f1d52777be9e4423bd771bfc40762e4f : FAILURE in 1h 10m 39s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/32b6efa652d446f4a125dd1f1a194876 : SUCCESS in 1h 07m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/40718c29f13c40b3bc17cde84b31e678 : FAILURE in 1h 01m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1fcce893615a4678b6bd8d077f8c7d0b : FAILURE in 23m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"01065656510130f9a3dc1233b0ea7c9ee9ae9d4b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-21 01:22:08.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/4/check/DellEMC_Unity_Manila_dhss_true/ef97586/DellEMC_Unity_Manila_dhss_true/902 : FAILURE in 3h 08m 52s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/4/check/DellEMC_Unity_Manila_dhss_false/e7a8f69/DellEMC_Unity_Manila_dhss_false/959 : FAILURE in 1h 32m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"c59b6d81e53316c09354b09c8ab337c0d57df8a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-09-21 02:53:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3a05b47983410fb0efcbd621561f28e4976d1f84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-09-21 03:51:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"355792e1430d9a606933ab9190016df90ca36f06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-09-21 04:51:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7e032c01552248a10ded9d55502b06d63de2ef6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-09-21 05:57:49.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6dc6c6fbaa7ad5700cfd800352813760de19849a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 08:05:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed9dfd504d64efea286ad99f2dde66f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3488974e377040db8f95ac9eb9dbe933 : SUCCESS in 5m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3fe1ce969c4a40c3872a5db0640ef5a5 : SUCCESS in 1h 25m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d096b49e138f411ea33b74727299d689 : SUCCESS in 2h 00m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/87dc371fb20b4fbfadc027b67e0f5d22 : SUCCESS in 1h 20m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/cf0d73bb212f4699bdea546ee00e1001 : SUCCESS in 1h 30m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/66d2f23bc07f4ea38a47837e7ae0de44 : FAILURE in 1h 56m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b6669fdd27524fc683d5eae23727e30c : FAILURE in 1h 22m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f434701b6d7549f5bd467a738678063c : SUCCESS in 37m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/310315adc7e84c879ee3af49c5206fc8 : SUCCESS in 44m 41s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/a5222e28c9534bd684d54fe565b660b5 : SUCCESS in 1h 34m 52s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/40c48ca28f80420780aaccdae552ca63 : SUCCESS in 1h 11m 37s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a0ac3c9944444d2390d500713c66a543 : SUCCESS in 1h 17m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9006183806954595a2052028e0469ece : SUCCESS in 58m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6032034c0b3842dfa19eb566bd53b499 : SUCCESS in 31m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ca6bf4b77aa18c688a457ba1b3b8eee8918fbfcb","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-09-21 09:28:52.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/8/check/DellEMC_Unity_Manila_dhss_true/9b6c494/DellEMC_Unity_Manila_dhss_true/905 : FAILURE in 3h 30m 46s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/8/check/DellEMC_Unity_Manila_dhss_false/b45fc56/DellEMC_Unity_Manila_dhss_false/964 : FAILURE in 1h 27m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"2fe14c9a3c198c8a597c9a1abeaf9f0ac4dc6ff6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-09-21 10:22:12.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/8/upstream-check/manila-tempest-plugin-ontap-dhss/b39c67b/ : FAILURE in 2h 16m 00s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/8/upstream-check/manila-tempest-plugin-ontap-no-dhss/b318f44/ : SUCCESS in 1h 22m 49s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"fe9bf8ed4307272e13524ec44a307a41af93bd07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-11-16 04:35:00.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"bcecf52312e48effe24a85867299e9632fb52522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 06:19:47.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a1e20d693b14b2d9c65b6dae294578e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc031de293443b4a2dcdbc4fe673ec7 : SUCCESS in 3m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d7bb5b05f5f3479780f071c9e0b7e60d : SUCCESS in 1h 01m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/15bbdd208c2f43b99c5542aeb2e62752 : SUCCESS in 1h 43m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fbc8fdfadc924f7aa321624cf7b27778 : SUCCESS in 1h 07m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4040c8a8b2794395aaa3ede7bd5c6980 : FAILURE in 1h 31m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/876a14573c1e4d3fa8a74414e4f29dbf : FAILURE in 1h 14m 54s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/90479b6503534c1bb4f6f7749a87326e : SUCCESS in 1h 06m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b5273130d646453d885ae2e1ce23a1f8 : SUCCESS in 28m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3049a79672b84cefb20e5ec9b4906bcf : SUCCESS in 30m 16s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/12bf280f3e1543f99260e2cce5c96115 : SUCCESS in 1h 08m 42s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/678dee93cc514e3ba2887ac4b302a5e9 : FAILURE in 24m 49s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/370900bbedc44e3a9bcd54ba97c4f5d1 : SUCCESS in 1h 00m 33s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/8acb567a24e14e839c490d3826223217 : SUCCESS in 22m 30s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a7cf670899124ec68b602a5bdbbce9bb : SUCCESS in 46m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0d1dd601098445469dab5e5a10402ab9 : SUCCESS in 20m 55s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f997a1b342cbdc596d0836e11c18b1b87d53689a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-16 08:10:08.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/9/check/DellEMC_Unity_Manila_dhss_true/f2d13ed/DellEMC_Unity_Manila_dhss_true/1019 : FAILURE in 3h 34m 32s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/9/check/DellEMC_Unity_Manila_dhss_false/3621b68/DellEMC_Unity_Manila_dhss_false/1086 : FAILURE in 1h 19m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"65d9472d4831520f82f1b9a6c8bc791b53dd2e78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2022-12-11 06:49:14.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"0802af1cde669f779ee9b6df2ae9913e9d27eee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 08:49:00.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b46fd86fac44390a134da911834e69a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10930660319e47369f513bd3d618181e : SUCCESS in 3m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/97067c38fbd34ac3abf8e1c4236ca4a4 : SUCCESS in 1h 04m 34s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/80dfcb4e530e4e87920391db81637819 : SUCCESS in 1h 58m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2219665bff0c4a229b24c53e2db2275f : SUCCESS in 1h 14m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/69caf7838731486c9a5ca876e035bd56 : SUCCESS in 1h 08m 06s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/db20221d362c4c31a82ea1b4f8c9f9f2 : FAILURE in 1h 47m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5f30875f8de94ad397b15205e7af5330 : SUCCESS in 1h 01m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d836e47e31b3468c9328a6125d121b3b : SUCCESS in 27m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/103ae144c6ca4fbd9ce9b7cc67a947e0 : SUCCESS in 24m 52s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/a21e2fd11cd84bbba402cc1d34da9481 : SUCCESS in 1h 16m 55s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/bc133c86b4df4a92b1aeb8f81712d774 : SUCCESS in 48m 51s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/bbd67886c12949e69faf519395fd9b81 : SUCCESS in 1h 05m 46s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7d74619a2a8742fe829ef2ccd4b461d5 : SUCCESS in 23m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/897c49e0f2104f06a4b4206612c49881 : SUCCESS in 46m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f1bcf8f964f748c8b2c1db8c211de1c1 : SUCCESS in 16m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d2399ec2a5185ca0e4948a62e13f9bdc26217e3c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-12-11 10:34:37.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/10/upstream-check/manila-tempest-plugin-ontap-dhss/d0c7189/ : SUCCESS in 1h 44m 37s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/10/upstream-check/manila-tempest-plugin-ontap-no-dhss/bf0cbea/ : RETRY_LIMIT in 8m 15s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"c2dfb8cbec05fcf6174ff635ad34ce3ee50589b9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-12-11 10:48:00.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/10/check/DellEMC_Unity_Manila_dhss_true/0b89dd0/DellEMC_Unity_Manila_dhss_true/1075 : FAILURE in 3h 58m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/10/check/DellEMC_Unity_Manila_dhss_false/965f9d7/DellEMC_Unity_Manila_dhss_false/1147 : SUCCESS in 52m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"7e7da08d5b1827d8cb86a5bef733ac54d8e8e40a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-01-29 02:33:35.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"1ec30aba71e575f4aebc2c7867a96754d54587a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-29 04:44:24.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/950292cb8c9a40efa65b97989e2e097d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/589c8efcd50949d394ba13ef88e0daea : SUCCESS in 2m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ce7f4097273f4df4942ba1b8e59e01c2 : SUCCESS in 1h 39m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a9f2fbff5e8d48fe9302e9feb7b851db : FAILURE in 2h 07m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/be46c9db42d541f19b59bf8afc158796 : SUCCESS in 1h 01m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/7aa7415c6e9243a8bc6c44d497dd4680 : SUCCESS in 1h 11m 47s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2cf3baf20f614a5fbef3779ad8135f40 : SUCCESS in 1h 28m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ccf7caa510ca4bbdadac94ce7245d224 : SUCCESS in 1h 09m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/95db202896554a51a744ca26fc0f1c24 : SUCCESS in 29m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a81479335e494880a88b12de65f14b96 : SUCCESS in 34m 29s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/421c217aa9674ff789fe60c0a50b689c : SUCCESS in 1h 17m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/0340e41fd2374a97b7e00845428ac00a : SUCCESS in 1h 11m 30s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/424586e6667f4815b00a2dc9704709a2 : SUCCESS in 1h 15m 39s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0945385bd9eb436f83b6ef844714475c : SUCCESS in 30m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8e62a3d090564e34aa4f85dc4ee92951 : SUCCESS in 49m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e716b9d395994edc936a78fb12ec4d0b : SUCCESS in 23m 33s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"fb411a2481f9856d5e798087368b17f513776749","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-01-29 06:21:03.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/11/upstream-check/manila-tempest-plugin-ontap-dhss/14b7172/ : FAILURE in 1h 35m 35s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/11/upstream-check/manila-tempest-plugin-ontap-no-dhss/0be3a8c/ : SUCCESS in 1h 27m 41s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"bd94a0862f823dfba4442832010af1cd0a03bc6d","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-02-02 00:29:11.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"88668e50c83d72f31470c281687645af36b4bffb","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-02-02 00:29:22.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e57101c4dc61db83c84b53a250bd74b1e0cb97c5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-02-02 02:38:43.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/11/upstream-check/manila-tempest-plugin-ontap-dhss/1bbb605/ : SUCCESS in 2h 08m 13s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/11/upstream-check/manila-tempest-plugin-ontap-no-dhss/80b612f/ : SUCCESS in 1h 22m 53s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"12edca836c60c8bc0063d8defe99cd5a4583fad3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-04 02:26:12.000000000","message":"Patch Set 11: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"44817661fb077256aa556bb00f806462eea1a55f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-06 08:07:42.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"7a5fe7addb77a3e18439863d21489bc492274ad0","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-06 08:08:42.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"a6cb96a9fd316e83507661218cc235d7bf8ec112","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-06 09:05:33.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"3d4fafe74a43a153767cfb281f0074b2693bb8f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 09:05:38.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/843832 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"36fb66fe505a9770d487edea2d95688dae04229c","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-06 09:40:54.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"6a4f9af8feae2e61a14f7982279eabb8e0fa8196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 10:01:42.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/843832 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"f295b7463227f85ced6e1ba45b766f6cd4bfca7f","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-06 10:02:15.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ce7b50daa9784b4ba8c115aacd1b2e92dd585a4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-06 15:25:02.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfc53d026ca04111a1c89686805b84c3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb4223fbf8954518a08fb5263f2d1252 : SUCCESS in 3m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/62dcf4a9a8f743e9a6dbfd60b2739eca : SUCCESS in 59m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4f9e45f506a740609f59309c2d7591ce : FAILURE in 1h 54m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f6ce87cfc8de4315809370b6e4e5018a : SUCCESS in 1h 09m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/08247914704342909c16a84717bce551 : FAILURE in 1h 10m 19s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/76a737f89f1f4eda88fa28e03e4f9b5d : FAILURE in 1h 52m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fcab994f99704f75bd798db19b81ad96 : SUCCESS in 1h 07m 08s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7a7b3d1ea84545a58995399d4d8246d5 : FAILURE in 39m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/51b81e10d53842a29bdbde77dda92b72 : SUCCESS in 45m 12s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/2e4d8dd3510b44f18b20ce30fa618600 : FAILURE in 1h 49m 38s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/db5313510bc341fbbb616475ed653cbb : SUCCESS in 1h 19m 05s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d3f30533343a4b61aa3daba7981c3415 : FAILURE in 58m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/23d5583064f34e0a8d9c1f294243470b : SUCCESS in 29m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c4d4f9f4911e4aaaa914019774368d15 : SUCCESS in 48m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/58ecdf41696946af9940ccc1c0d1f63b : FAILURE in 58m 15s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f18f1c5dd1336b3804f242fcc762fd625f248382","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-07 00:47:29.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"1fa2e5e6036a5637cd1f45baf82811c13559b7d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 02:46:19.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ac5ea326dd44ba4aa86e8a1306e6965\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1363989a3714428adddcbd68f2412b5 : SUCCESS in 3m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5fd1e548e97f48aeade123ddb1f4556a : SUCCESS in 1h 07m 28s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b695f93c7db849d7b222550846e4d707 : SUCCESS in 1h 57m 24s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a3de4b72e2e342c49ff9d29046584fe2 : SUCCESS in 56m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1438b61a1b4449eab966390d8198fa41 : FAILURE in 1h 15m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/31be5f5a5da648148ae50d0d297520bf : FAILURE in 1h 54m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/90b4a82e2cf548f394970b2c173d5346 : SUCCESS in 1h 22m 48s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8c901a54aa67431cbe258fe6a4d3a4ba : FAILURE in 47m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ec6311a2e9444508b0aa1fbbf29de3f7 : SUCCESS in 26m 30s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/4772bd3aabc34865b199576061f3e496 : SUCCESS in 1h 05m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/b3cf1f5117594cd6872a6cd91491b6f2 : SUCCESS in 1h 02m 51s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/39d84c680e8c40c885bd7ea89eb239df : FAILURE in 20m 43s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/6471e1484e64488aa7d502980213bef5 : SUCCESS in 20m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3ef60cdf342b414aa7c6b467deb5b582 : SUCCESS in 1h 02m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4cf740f15620494eb7e2ae71ef865cc6 : FAILURE in 51m 03s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"d0c03b5aa130bd5cc4c5e47d99829a02e6591c07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-07 09:15:39.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"d783a7dfebc573cfe0f42d6d1e4f35f85e278396","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-07 10:23:22.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/13/check/DellEMC_Unity_Manila_dhss_true/24948df/DellEMC_Unity_Manila_dhss_true/1195 : FAILURE in 1h 07m 21s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/13/check/DellEMC_Unity_Manila_dhss_false/d6b5471/DellEMC_Unity_Manila_dhss_false/1294 : SUCCESS in 1h 05m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"5307bea9104026acd60467d1a83fd9401ab9cdbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-07 14:27:38.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/245138eadf4e40d7834347241aa62527\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc74ad3c4833462a9faa502e87f62334 : SUCCESS in 2m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b9cab7a9f3c3456ba476abe426834607 : SUCCESS in 1h 00m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d1c1898c17b243f99b31628a04025161 : SUCCESS in 2h 22m 30s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9c3b399c3f954e839d2e2866b13706c9 : SUCCESS in 1h 16m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/14a80c69db2c45529a21b56666c3d22d : FAILURE in 1h 17m 59s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d8a92f58141a44f8a83d59f971fbdfca : FAILURE in 1h 54m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3565eb73619046eda8aca7f7a6e17dab : FAILURE in 1h 19m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/22d2ed4d610a447ea0714abc2d5094cc : FAILURE in 40m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f317c6ede73842bf89917e9b3f82b8a2 : SUCCESS in 36m 06s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/816658f4e90948b7b6adfe521fa6dab9 : SUCCESS in 1h 06m 21s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/39e425ee459b4c42b011cfa3c097e946 : SUCCESS in 1h 04m 51s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/4d229847d1e344f6bc16888a752d7f48 : FAILURE in 37m 54s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2b312934839e4b97ac12e596414704fd : SUCCESS in 30m 30s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a5c98c60e34944399a808cf27e1d2d05 : SUCCESS in 51m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7c15737e85f94ac8bf99d617ed135b28 : SUCCESS in 17m 25s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"845eb882906072d703ad61c7aadee39d2bb72637","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-08 00:52:07.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"1d7f100e474b7446621a76f6bb5f24e5f5a4a7db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-08 00:57:58.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"13ef9ec376b4f4d11603d58976abc482be85596a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-08 01:57:20.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/14/check/DellEMC_Unity_Manila_dhss_true/a3cf14e/DellEMC_Unity_Manila_dhss_true/1197 : FAILURE in 59m 02s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/14/check/DellEMC_Unity_Manila_dhss_false/29cdc45/DellEMC_Unity_Manila_dhss_false/1296 : SUCCESS in 54m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"e185d4f1cb8747e79d8ffe2e53a67f9f54c4a7c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 02:43:54.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dee4317815314f059aad5f6a3e30eed7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/213f772026c143599fd8d33d4784c7fe : SUCCESS in 2m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/26f857f62e1640bd9894727bd3d9a9d1 : SUCCESS in 1h 10m 47s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/eb156a3645194b978774cc27092ed8a6 : FAILURE in 41m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/85c92a8df87e470eb626e529d14a449a : SUCCESS in 1h 07m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/fb57537f0b7f485c8537f36d964741b8 : FAILURE in 1h 28m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3096bd48c87b4f20b7d7b96ea50cade6 : FAILURE in 1h 44m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6475caa6d5b54474bb690b881cab1f70 : SUCCESS in 1h 10m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/eede9cdc821e463083da0fa555d1be3a : FAILURE in 27m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/45362366e9f74312b4ba77c803a8907a : SUCCESS in 32m 18s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/772cac95654841a088c3791d4611b6a3 : SUCCESS in 1h 23m 15s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/422a840e347e4db39667087744d4f415 : SUCCESS in 1h 02m 17s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/e621bed1873743959e755fa019b23462 : FAILURE in 31m 56s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/dfd3f7408ca746c8ae8823c9dab0693a : SUCCESS in 20m 50s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0b2dc1ae90a44732b7d68c3dc5ae5ebd : SUCCESS in 39m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c00d82d07f054f04af258b539fdb3c18 : SUCCESS in 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e0993ee3dcc1814e18ae38a02478a374070860cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-08 06:59:05.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"1f5fbba701495219d5468b8ee1df81f86740c7d8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-08 08:02:41.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/15/check/DellEMC_Unity_Manila_dhss_true/4e3162d/DellEMC_Unity_Manila_dhss_true/1198 : FAILURE in 1h 03m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/15/check/DellEMC_Unity_Manila_dhss_false/3fd9c7c/DellEMC_Unity_Manila_dhss_false/1297 : SUCCESS in 1h 00m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"b0bd76c3894af430cda84c4283cd5c2348dd6513","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-08 11:38:55.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70c7716fb9a94098b36ede7287b6c283\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78d436ba5a034775a3705941d81d7bc4 : SUCCESS in 2m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f927a564e4ae4ce5b55b711b55061f40 : SUCCESS in 1h 26m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/541ea7de3c02460f996e1c4b1c3d9e79 : SUCCESS in 2h 27m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/326af152aedf480fb699933183232c22 : TIMED_OUT in 3h 05m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/b1de1de0640947c2873167e6906de4c3 : SUCCESS in 1h 51m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/05f05cf6da614352bf1b31f0fcb418c5 : SUCCESS in 1h 31m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4de6a308f8294e8387113e8f42863ff0 : FAILURE in 1h 21m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1722c739f05b4aa7b1a6e97fff384802 : SUCCESS in 40m 01s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/39a2205b804e44768889916d8359803a : SUCCESS in 39m 36s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/88ce0ddda6ff4255a55a5e640bb25fce : SUCCESS in 1h 33m 18s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/539cb80f3af34fd798654a5b5ee0f827 : SUCCESS in 1h 07m 58s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/0191e416004240f5a2be302343962bf8 : SUCCESS in 22m 31s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/38cc212c4f0a4dd0a51d9a6b3d3fffcd : SUCCESS in 46m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9e76e6e13f1a42ecb534aa85d86f117a : SUCCESS in 15m 43s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b971cd85876ffe89018ed1de9608cef169d8acaf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-08 13:25:15.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/15/check/DellEMC_Unity_Manila_dhss_true/6621786/DellEMC_Unity_Manila_dhss_true/1202 : FAILURE in 57m 24s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/15/check/DellEMC_Unity_Manila_dhss_false/823c661/DellEMC_Unity_Manila_dhss_false/1301 : SUCCESS in 1h 10m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":15},{"id":"9f593094f5d5292b62799be58024ab31eda8ed23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-16 13:35:03.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"e7944e5d14af572b919385b6595bdc022cb38902","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-16 14:25:44.000000000","message":"Patch Set 16:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"45f56a13ae0a8fa39ec1cad3ba0d955aebe0be5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:25:49.000000000","message":"Patch Set 16: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/843832 is needed.","accounts_in_message":[],"_revision_number":16},{"id":"64a3304a24e18f2225291357d47475de05de09fc","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-16 14:26:59.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"4590f4848526bb20cf23761d86719c0d38a741e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 16:00:14.000000000","message":"Patch Set 16:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/843832 is needed.","accounts_in_message":[],"_revision_number":16},{"id":"9ab533e73eb2ced2a9f69e498d44a90fa2b3e419","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-17 03:00:31.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"bd4e84bed91648c40f422999464b9d33cc21a7b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 04:11:19.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6edf3505154b4f25857a663ceb5949f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/898f46dd2469492aa7a2e97c82518d2c : SUCCESS in 3m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6af0036195d144f4863e2eddc897000f : FAILURE in 20m 14s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f6c236b8b8b14aaea95ff07006fea650 : FAILURE in 22m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5ee650a8c0284c11840f57372f3e0d7c : FAILURE in 18m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/9370a7972967487ca13a3b430dcbdeb4 : FAILURE in 19m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/aa0f3fa1d44643738332fb7fe21ac619 : FAILURE in 24m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e0cbfa9a7b934807a683c0d2fb8068c8 : FAILURE in 14m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b68c9843b46549dfa89005aa5abbdbbb : FAILURE in 11m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/19c04a041efe42dcb3f8f81919510800 : FAILURE in 11m 50s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/39e26b7352f647278ca21daa5e1fb850 : SUCCESS in 1h 07m 42s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/73edd68165854fefb1a315d8b808ee74 : SUCCESS in 59m 27s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/83d3f1620b1b4eb28e261ddff05fd6da : FAILURE in 14m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/76b0668fcbaf41dea14f5c23d18044f5 : FAILURE in 20m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/39fceccc9fc24f7d9034e009aac50201 : FAILURE in 20m 09s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"1c818a3a58444426ec318af286b057148084a48d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-17 04:20:46.000000000","message":"Patch Set 17:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":17},{"id":"ca48705f49fb908ef2db8eaf868cca8a93c73790","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-17 05:54:37.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"11eb1f4eeaa3eff995d791a99ada83ffb94a9fa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 07:56:47.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bff841d812974454ae6f87c4ab53224f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fce3189213ef4024bb08cc59332c06fa : SUCCESS in 3m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d36a437ba7e740aea6e2f8429dff7f12 : SUCCESS in 1h 16m 08s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dc9437954f414ddb83dd265afc3d2379 : SUCCESS in 2h 00m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/026079c7c99b463482ad713f8d6607d3 : SUCCESS in 1h 01m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/494c1ab02a7841baa3ecad8807ad62d5 : SUCCESS in 1h 07m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/99e2de9e2ab442028806607a55a75c5e : FAILURE in 1h 40m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f18060f51fbe401a83fec80a08c77684 : FAILURE in 58m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b7a76d85c6ee4214b19610ecb509a890 : FAILURE in 31m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/32c33ad99f6c4ee8b93dec5f7335e4ad : FAILURE in 32m 24s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/4b1f3df0479a43388243d76f32f59c58 : SUCCESS in 1h 08m 38s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/e0043fae0a6949ac931fde7735a88fc4 : SUCCESS in 1h 06m 50s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2e45314047d747f4b7206d7c167e44f4 : SUCCESS in 36m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e585587ff82e4970899889fccf0daa01 : SUCCESS in 50m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6af44ef2d7df498385d77d963ff9fcbb : SUCCESS in 20m 26s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"1cb36ec1a705bbfc4db5b19478acd6dea5af0ce3","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-17 08:11:45.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"84d0e2a2c4226a32fe57281718682f9dab7871ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 11:22:10.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b57ffc2eaae6434b9be91b1f0fbc66f4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8065dde4620344c59a7803dd92fc3a37 : SUCCESS in 2m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/73e8918ce0054826a0aa32b2e90804b4 : SUCCESS in 1h 00m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/84410176fcbe42fa87a2c792f155f3f7 : TIMED_OUT in 3h 02m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/97a0128af5c7410eb50b7b84bd12040f : SUCCESS in 1h 32m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/4b11d3c7ff294ffea8c75821648afec2 : SUCCESS in 1h 05m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2e01409fadef4b56b1606ad38f092d4e : FAILURE in 2h 15m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/44bdfcb88e5b432caecb4d2547cbab7d : FAILURE in 1h 12m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4826429bb20944f79bf8d577a4a05892 : FAILURE in 33m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1e9fc55b6470451f877a893b4cdc8a20 : FAILURE in 35m 05s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/02d4d229aaa94a8ea833774bbfefc8f7 : SUCCESS in 1h 08m 36s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c85601a43c4047d293d6e9cfbe260dab : SUCCESS in 58m 41s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/94bf7d37c67c48fbaa8c31d97ed1f545 : SUCCESS in 32m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/97c219544ab74ec4b3ab15d4ad8ddb90 : SUCCESS in 48m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/76f2879605dd4377b983a7e66e185d32 : SUCCESS in 20m 22s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ef7f9895ab01a57f1de56a713f8e3bf39cac05d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-17 11:52:00.000000000","message":"Patch Set 18: Commit message was updated.","accounts_in_message":[],"_revision_number":18},{"id":"de8c227b378a2cf78b7cb40490e9e59e9b42cde5","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-17 14:15:09.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"f4fefc6b94fd54eb6bf198b071f8ed0f8ca7a80f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-02-17 14:15:37.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"f58fb402fc58ade99d832aa886c414851c0e4582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:28:35.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a697ce2dc7846da9f1510c3fc4161d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f64e8c125e497eb2f53f1c2506caa8 : SUCCESS in 3m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/247f86a495ce4af58f8a428c383d6079 : SUCCESS in 1h 11m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7dda498830b54e9f8132aca443666744 : SUCCESS in 2h 34m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/422396a5483740b2b3a1b7c44e76dd9c : SUCCESS in 1h 33m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a69f6930fc364826b75022cfd5a6fda4 : SUCCESS in 1h 35m 49s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ca869b0a09224c77920562fa16d9cf4e : FAILURE in 2h 31m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/91cd22a541bb4e98b551c348e1e68725 : SUCCESS in 1h 11m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4cffb040fa25446b844dcf944db3117e : SUCCESS in 1h 01m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4240878cf55a48abb62e2daad39cfdde : SUCCESS in 37m 55s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/e39c6574823c474d8c14eb4bd316ad18 : SUCCESS in 1h 21m 29s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/22fda0787e7a4eab8692df86405bb7cc : SUCCESS in 1h 02m 40s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/b74bd364746847aea913bd360172e660 : SUCCESS in 33m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4d41581c0a0c4677a9acd26591fcb92b : SUCCESS in 48m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8101bda20084480aa278de8962271dfc : SUCCESS in 28m 48s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"380ffad96f963afca7206afb4b9e4165a4a43b8d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-17 16:02:38.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/18/check/DellEMC_Unity_Manila_dhss_true/4059565/DellEMC_Unity_Manila_dhss_true/1259 : FAILURE in 1h 04m 47s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/18/check/DellEMC_Unity_Manila_dhss_false/ca13880/DellEMC_Unity_Manila_dhss_false/1359 : SUCCESS in 54m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":18},{"id":"329944cdf48f9209ec9e23c365cb5c298e368c8b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-02-21 05:18:14.000000000","message":"Patch Set 18: Code-Review-1\n\n(13 comments)","accounts_in_message":[],"_revision_number":18},{"id":"4ac77b6a052310fbc9870c1509ae5e0511c5ab9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-21 09:05:21.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"c1b219b8a56a355b16d55df555f6bf8632147f2c","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-21 09:06:02.000000000","message":"Patch Set 19:\n\n(13 comments)","accounts_in_message":[],"_revision_number":19},{"id":"c19fa9bc95b3f5cf7404bfbe435ebaedd1ea1995","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-21 09:19:07.000000000","message":"Patch Set 18: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":18},{"id":"eabd7d8863ee3fb8e2d6dcb7233e3f1d4fde535a","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-21 09:19:49.000000000","message":"Patch Set 19: Code-Review-1","accounts_in_message":[],"_revision_number":19},{"id":"ea13dce05730fd38bcb8cfd400ed5962a4b9f683","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-21 10:10:16.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/19/check/DellEMC_Unity_Manila_dhss_true/934dcd4/DellEMC_Unity_Manila_dhss_true/1279 : FAILURE in 51m 17s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/19/check/DellEMC_Unity_Manila_dhss_false/7975300/DellEMC_Unity_Manila_dhss_false/1379 : SUCCESS in 1h 04m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"fc9275f6219e4ba95729612ae644eebd7e1c43a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 11:38:12.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1953b059be5345829ff1c6f10105b8f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9459fd24d354bc78c8c756b2d881ae7 : SUCCESS in 2m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b3db78fa7c2d4a71bdbc62309d8a266a : SUCCESS in 1h 11m 34s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/73794fa69b4b4fa8b3ee267306e994eb : SUCCESS in 2h 04m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fd947ab621834421852cde17e9a16ed2 : SUCCESS in 1h 04m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/666106fab58f4965a0cd02b8dfd8ae0a : SUCCESS in 1h 31m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7c8fca45010e4546808f4975b24d6ed3 : FAILURE in 2h 26m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2c222dbb2d9f4a1396403a460af9ee54 : SUCCESS in 1h 04m 44s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f99adea78feb46fcb596564055e41608 : SUCCESS in 36m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/10956a234d8f421eb6d4a8d18cf44a22 : SUCCESS in 39m 39s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/cfca558e58544d28a895ae38e6149d89 : SUCCESS in 1h 47m 15s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d7e7740648e0455594def3c3699ba71c : SUCCESS in 1h 02m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/756a488e8eac4a23ab4c35c16c463e86 : SUCCESS in 34m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d677dfbf49e94c038721cfeaa812bd05 : SUCCESS in 1h 18m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/abe2a891d0ad4e3baa253e690f42023a : SUCCESS in 29m 07s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"38d4c78c3dd57b626f4e64a2f17f33e94ae53480","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-21 13:56:00.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/19/check/DellEMC_Unity_Manila_dhss_true/f3d0ae0/DellEMC_Unity_Manila_dhss_true/1282 : FAILURE in 50m 56s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/19/check/DellEMC_Unity_Manila_dhss_false/5472ab0/DellEMC_Unity_Manila_dhss_false/1382 : SUCCESS in 57m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":19},{"id":"9ce9eb4c30c04dda4d0dbd483220c6708e98cf82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-22 06:33:26.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"024c01be7b0f056dda832602faa2b8077d54929d","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-22 06:34:17.000000000","message":"Patch Set 20:\n\n(8 comments)","accounts_in_message":[],"_revision_number":20},{"id":"47a655d080d83e9086d97876ad8aabd533d51ed1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-22 06:35:52.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"0a7629dbefa9ff19090e911a76971127fb436bdc","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-22 06:57:12.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"24d7f4ad3adb7f1c78d851e02ecb92ebc4c57c98","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-22 07:09:01.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"01f3cb2abfeaefad86b63a354381b72ddce35c00","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-22 07:42:40.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/21/check/DellEMC_Unity_Manila_dhss_true/97652f0/DellEMC_Unity_Manila_dhss_true/1291 : FAILURE in 57m 02s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/21/check/DellEMC_Unity_Manila_dhss_false/873eca7/DellEMC_Unity_Manila_dhss_false/1391 : SUCCESS in 1h 06m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":21},{"id":"a8093425b1704e25f7415efc86fb7aa95ba0024d","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-22 07:43:46.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"4c9e82708667de8eaf2c8f4bf3bd6df607003f54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 09:01:36.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1cd9c3d5a4042e49b9061d745432c03\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44122aff4b4f4727b525a40f72f8829c : SUCCESS in 3m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ed0207dba2e94ce0becf254191b52322 : SUCCESS in 1h 01m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/055db2e719b643848e10331dbcb286ad : FAILURE in 2h 00m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b3fb46b63007438fa0681d44625a4c3a : SUCCESS in 1h 07m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a365c78f1e3a410aa7321994347d95c0 : SUCCESS in 1h 05m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1ba8604a05b048988444b2e5b2fe5c7f : FAILURE in 2h 07m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/85f58b2ba5f24ff79a6ba31a024c70ab : SUCCESS in 54m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cbc02e4ea84b48a29df8c31783a0b29f : SUCCESS in 33m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d50f343962e449b4a916eba40066fbb7 : SUCCESS in 45m 32s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/5d6b8ca2ee3d416fa3ef107fcd9c598e : SUCCESS in 1h 21m 55s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/bf650407ea16415f8542d439935d8443 : SUCCESS in 1h 31m 35s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/dbe033dfef14429fa844fc3b87dc34ff : SUCCESS in 49m 51s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/97a847756b954f6d8496c5c810b3e215 : SUCCESS in 42m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a1f8745ebf6a4864b760e4dcb5a82bd4 : SUCCESS in 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"c3e1e147af3a58828fb5695123c5222151396479","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-22 09:18:35.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"3587520b045f7c5ca86b2c653c7b02e507740b6b","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-22 09:18:53.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"0cc5bb3ab55c723e08d756aebda0197a8c0de921","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-02-22 09:37:31.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"c374de09617a4c04968722fca5ee900134aed09f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-22 09:52:26.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"34cab6076c61a064dd737c60431b036807bea388","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-02-22 09:52:39.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"7b032b80065cb00794a4936665f23be4628973d4","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-22 10:57:38.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/23/check/DellEMC_Unity_Manila_dhss_true/39068e0/DellEMC_Unity_Manila_dhss_true/1294 : FAILURE in 1h 04m 58s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/23/check/DellEMC_Unity_Manila_dhss_false/056f9dc/DellEMC_Unity_Manila_dhss_false/1394 : SUCCESS in 58m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"e62bb5cb500ed2e573335ce74f3683589dfb8627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 12:22:34.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99efbf289b8c480281868917005cb6a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/186ce11c78874da997cb115c41c20e52 : SUCCESS in 3m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5b549b4037e54c3085bce435ac27b065 : SUCCESS in 1h 16m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8b5af00bbcc14550b9174d0c705de1c6 : FAILURE in 2h 15m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/037bb28394114b5db1f3d0025cbb8639 : SUCCESS in 54m 01s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c182999be9704f268fc71e74640c3867 : SUCCESS in 1h 25m 34s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e60faf535cf6430780e9194d31bd86e3 : SUCCESS in 1h 35m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d9041418188d4123af53957820d7e4c7 : SUCCESS in 1h 00m 48s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6f50710a47b0450480d0fe7bfce91267 : SUCCESS in 37m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6a9854cdac794704ad56693c1b389438 : SUCCESS in 33m 44s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/62cd430a7bdf47058e59fe39c565da91 : SUCCESS in 1h 38m 48s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/8c099f6d4ce9404186e11f8bc9bc06a0 : SUCCESS in 53m 30s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9cca29c7f0db42b5acd9a632a1d051e5 : SUCCESS in 31m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/69357724910742be9c73bcf32fba88ec : SUCCESS in 38m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8581482bda87456394a31efe68666ce9 : SUCCESS in 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"132b24c7079947ed68a81d4085654145f3ebda90","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-02-22 13:31:29.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/23/check/DellEMC_Unity_Manila_dhss_true/5e17bec/DellEMC_Unity_Manila_dhss_true/1295 : FAILURE in 1h 08m 20s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/23/check/DellEMC_Unity_Manila_dhss_false/66659ac/DellEMC_Unity_Manila_dhss_false/1395 : SUCCESS in 1h 08m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"0a5390f29d5a307301f2a204ba5c0f1bc975a0b0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-14 00:42:23.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"0abb0e138b3030c06fd519201268ef4873abd070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-15 07:01:07.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"73b67dd474c57d451387b5d3af84da19a718c7cc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-15 07:35:49.000000000","message":"Patch Set 24:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/24/check/DellEMC_PowerFlex_Manila_dhss_false/8957f24/DellEMC_PowerFlex_Manila_dhss_false/388 : SUCCESS in 34m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":24},{"id":"9fd23aed2c9b38a550a130cdf48fd2aed1843cc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 09:34:26.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/640fe065b0304d749043aa087b89866d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/491407e168d14ca685d05dfe71773fe7 : SUCCESS in 2m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/32fbde2627c6444ebb9635e5d6498270 : SUCCESS in 29m 22s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/370ffecb06e140b9924b8bfce30372ba : SUCCESS in 2h 27m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/59b1529b919b44a7a30a1f1f5679cb3a : SUCCESS in 1h 17m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5592cf095dd3417fb64ef9cf685b4850 : SUCCESS in 1h 07m 32s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1258c04362a749f5adb723f9f18ff9e8 : SUCCESS in 1h 34m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8a6c1b1ad09549aa8ff87548eddebf2f : FAILURE in 1h 15m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/518a9aab85a2478c90b9500083bf7497 : SUCCESS in 36m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7c2ffe6073224d72974b0fb8174e0c21 : SUCCESS in 45m 31s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/75c18b6fd9cd45da8d4d9d46d48650bd : SUCCESS in 50m 22s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/5d9e8492f189482c8e04ccf6ede3556c : SUCCESS in 49m 30s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7d4eb58af7a04ef68dd53fd01f1352a3 : SUCCESS in 36m 11s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1dcaa7382e6441328a3e58b82af04ea2 : SUCCESS in 55m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8da79caaa24a4bbb869067821b32b0f2 : FAILURE in 22m 56s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"0ebe75e763c424eeebf681cb682fa850b0312bb7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-15 10:05:43.000000000","message":"Patch Set 24:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/24/check/DellEMC_PowerFlex_Manila_dhss_false/865d153/DellEMC_PowerFlex_Manila_dhss_false/389 : SUCCESS in 31m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":24},{"id":"42ce4f0c4273f6e52566896903547e773edfdb63","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-15 10:13:10.000000000","message":"Patch Set 24:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/24/check/DellEMC_PowerStore_Manila_dhss_false/ff7ab23/DellEMC_PowerStore_Manila_dhss_false/166 : SUCCESS in 38m 31s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":24},{"id":"641e59b98ba7ceadbb8743d5e44a58a4863552d3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-15 12:52:01.000000000","message":"Patch Set 24:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/24/upstream-check/manila-tempest-plugin-ontap-dhss/d0dca00/ : TIMED_OUT in 3h 15m 56s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/24/upstream-check/manila-tempest-plugin-ontap-no-dhss/d8d4b28/ : RETRY_LIMIT in 9m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":24},{"id":"01051b76b0461dacda62d2fbcd2bcb6beb32ed7c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-15 14:43:34.000000000","message":"Patch Set 24:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/24/check/DellEMC_Unity_Manila_dhss_true/b05aa4b/DellEMC_Unity_Manila_dhss_true/594 : SUCCESS in 3h 21m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/24/check/DellEMC_Unity_Manila_dhss_false/b4128c8/DellEMC_Unity_Manila_dhss_false/624 : SUCCESS in 1h 08m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":24},{"id":"a6e4669583f0491bfb9cf79365c121c99f62da2b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-16 17:24:52.000000000","message":"Patch Set 24: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":24},{"id":"c378a1fd2161a93f9903db5b21c7ef1bac94c163","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-18 01:02:33.000000000","message":"Patch Set 25: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"1e5663327ceeb6fa9a33bb627b6922ad1f4d7296","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-18 01:03:04.000000000","message":"Patch Set 25:\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"1cb3e357b2a7ab27c960bc595288955a2d368074","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-18 01:33:53.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/25/check/DellEMC_PowerFlex_Manila_dhss_false/fb0680c/DellEMC_PowerFlex_Manila_dhss_false/396 : SUCCESS in 30m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":25},{"id":"c3e9b55a16c095d67e4cafb5c2790186036ced1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 03:40:50.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/936d7eaa74ba45f8bec9b656557def52\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2656eea9d39476aa085216fdcacec39 : SUCCESS in 3m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a762553168584d1094829b255c85d3c7 : SUCCESS in 29m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/461d3f4885bb499b8f8a01bd8a42a228 : SUCCESS in 2h 32m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b2cb9c2e392e4c9c94930ea1ba3ccf4e : SUCCESS in 1h 13m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d251b1086c5a4db493a2f0d421387dfb : FAILURE in 1h 10m 58s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/bd170cc2c22a40588d1b4e0709b622b6 : SUCCESS in 1h 18m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ebd1f405fa6849f4b8419b05fd131125 : SUCCESS in 1h 30m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1b2956a3d637473dba59bd3b29ceb037 : SUCCESS in 35m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bfc757df57ca4f74b3f656d6924fa26a : SUCCESS in 47m 53s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/eeabc230b78547de82ad3e7c38efeed6 : SUCCESS in 50m 04s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/0056038949774e55a5ea7891c62d4fb0 : SUCCESS in 45m 36s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/c18684df678b4bb595c292472d10f616 : SUCCESS in 25m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/93d8d8ffb3ee4bd0b0fcdbac905ac4ab : FAILURE in 34m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d6521a764d0f4c60aeee84f9afaeeab9 : FAILURE in 10m 10s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"9c6524082a05aeee6744cf6f2a88b988bf6c0c62","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-18 03:57:40.000000000","message":"Patch Set 25:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/25/upstream-check/manila-tempest-plugin-ontap-dhss/429ff0f/ : RETRY_LIMIT in 8m 40s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/25/upstream-check/manila-tempest-plugin-ontap-no-dhss/50045ee/ : RETRY_LIMIT in 9m 45s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":25},{"id":"e6b7b369a0471001abc0dd7cc09d05fabf0fba36","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-18 04:12:08.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/25/check/DellEMC_PowerFlex_Manila_dhss_false/127bd38/DellEMC_PowerFlex_Manila_dhss_false/397 : SUCCESS in 31m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":25},{"id":"026a4d6d971e6953e8a5fdcd0e09e77d5402fc66","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-18 04:18:59.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/25/check/DellEMC_PowerStore_Manila_dhss_false/ceabb99/DellEMC_PowerStore_Manila_dhss_false/170 : SUCCESS in 37m 54s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":25},{"id":"ba87b5222bf5a5df70fe989834532263b8524226","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-18 04:46:18.000000000","message":"Patch Set 25:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/25/check/DellEMC_Unity_Manila_dhss_true/eca7401/DellEMC_Unity_Manila_dhss_true/599 : SUCCESS in 3h 43m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/25/check/DellEMC_Unity_Manila_dhss_false/d6beaa0/DellEMC_Unity_Manila_dhss_false/629 : SUCCESS in 1h 08m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":25},{"id":"82bb8bb9e59864e78e53ca7b5fef048aa0d59b19","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-09-18 07:27:04.000000000","message":"Patch Set 25:\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"a1407c9611b010084099881cc9c4c114dcd59411","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-18 09:33:27.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"75961739df43ba52aaa20688ba8c3fcdf4231b61","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-18 09:33:46.000000000","message":"Patch Set 26:\n\n(3 comments)","accounts_in_message":[],"_revision_number":26},{"id":"0852a97b820cd5750331e67fdb9608ff4bf3451c","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-18 09:34:43.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"d1a0c3fdbb0bd4d458e288b5131dbedfca90efa8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-18 10:07:07.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/26/check/DellEMC_PowerFlex_Manila_dhss_false/50ba519/DellEMC_PowerFlex_Manila_dhss_false/398 : SUCCESS in 33m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":26},{"id":"5790c33a13b7b2295b1a3fa9284e55d936ed0cc9","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-09-18 11:06:48.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"a286a4c1d05c7265b80c6b1451df7a9bd1b31855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 11:28:05.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd700a4f4fe54649b7df2c4cbc9aaaa9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f145b6d5e634590897cd0425e34e82a : SUCCESS in 2m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/03abf66f655544c9bc64a1fb8fb2bbf6 : SUCCESS in 44m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/de2f0f5b36864f04b690ad842da653a0 : SUCCESS in 1h 52m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/38c9983bf38b4aad8e6adadc3d14598d : SUCCESS in 1h 05m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a488db90041f449cbdaa09662486a1ee : SUCCESS in 1h 09m 55s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/1f04a0799d79410b8f36d591a9c98ca2 : SUCCESS in 1h 24m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e26470a436244114a72866e8fffc9336 : SUCCESS in 1h 20m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1705f4a975a640f89d706decb88416e9 : SUCCESS in 40m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ff31b7ca842f4a5dbb99c8967dce3307 : SUCCESS in 32m 11s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/7a0166445d954cd7a58d499f42436a08 : SUCCESS in 43m 30s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/3a2647ace967484fbe3548c00e60d4d9 : SUCCESS in 33m 43s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d136d545a85f439a848d9f462a70ebe7 : SUCCESS in 37m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a05ca5953e214e3bbc04b73fea7449a3 : SUCCESS in 54m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d074dc9697f14f6e9d3cb21d83598c8d : FAILURE in 21m 52s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"2437adc9885f9d82b65825532b9b8b4a2ed2491f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-18 11:38:58.000000000","message":"Patch Set 26:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/26/upstream-check/manila-tempest-plugin-ontap-dhss/6ed2c3b/ : RETRY_LIMIT in 7m 37s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/26/upstream-check/manila-tempest-plugin-ontap-no-dhss/9b7c6d2/ : RETRY_LIMIT in 9m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":26},{"id":"3d8b767deda9e4aa5aa6b08f054dbd77565e2fa1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-18 11:58:20.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/26/check/DellEMC_PowerFlex_Manila_dhss_false/ebfbcd0/DellEMC_PowerFlex_Manila_dhss_false/399 : SUCCESS in 30m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":26},{"id":"7caa135b68ce306a17d746a9492cb1fb0a0604a5","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-18 12:10:29.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/26/check/DellEMC_PowerStore_Manila_dhss_false/cfc6008/DellEMC_PowerStore_Manila_dhss_false/171 : SUCCESS in 42m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":26},{"id":"b0b34157a3f5fe040783876721fe8293ccedb23b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-18 12:54:27.000000000","message":"Patch Set 26:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/26/check/DellEMC_Unity_Manila_dhss_true/5441ae0/DellEMC_Unity_Manila_dhss_true/600 : SUCCESS in 3h 20m 45s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/26/check/DellEMC_Unity_Manila_dhss_false/eeb6f98/DellEMC_Unity_Manila_dhss_false/630 : SUCCESS in 1h 14m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":26},{"id":"d2e427d53621b3b34cc80cbc59b0a3e1de550226","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-18 21:36:57.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"1454670b46da614db99e095bb681b04a9cc26583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-19 01:13:12.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"9fd78cd16eadf9f16b833be7f43bc169fd2b41d2","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-19 01:13:37.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"1f2d6dc625e38a58f40bcb8340ce569f0dc07513","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-19 01:15:29.000000000","message":"Patch Set 28: Commit message was updated.","accounts_in_message":[],"_revision_number":28},{"id":"0f550affa6efce855c2deb6a054b17bb59de267e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-19 01:43:38.000000000","message":"Patch Set 28:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/28/check/DellEMC_PowerFlex_Manila_dhss_false/74a671e/DellEMC_PowerFlex_Manila_dhss_false/407 : SUCCESS in 27m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":28},{"id":"72c9118cd1b196954d929be216a39d6a8508a9f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 03:39:43.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3229567b60194681b384dc850d9d1d89\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2758189e4a47e3b4e35ff3fa96648a : SUCCESS in 2m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/96e6f319d5a34f70b871e932ff7c00bd : SUCCESS in 41m 06s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/106db5cb395b4371be30eba384277003 : SUCCESS in 2h 18m 53s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4741be7e199d4bce800c7fad9f4a7c8b : SUCCESS in 1h 06m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/dcad0049db92461cb03f149d19ba9291 : FAILURE in 1h 05m 28s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ffb60dc5a7624092a03bab1729f2fedc : FAILURE in 1h 44m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/32f14da7a12e4a3b8c03d594048dd566 : SUCCESS in 1h 13m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bfc405cd954f4eed91ab0d0b1db06228 : FAILURE in 22m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/07f48f675aa341b2ac67b49df0ded8c1 : SUCCESS in 35m 20s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/aef813f76aea4392bb7c630740c78489 : SUCCESS in 27m 34s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f70a028bb6ad43dd9e810755142bfc50 : SUCCESS in 28m 44s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ec8828bde62e49e8a79bff72df0edbab : SUCCESS in 24m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/340010baf3d5405bbc0f0dcb0f8b59e2 : SUCCESS in 38m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e0c71ee30a214975a69b1ea783f10f50 : FAILURE in 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"dde115d12540d612a5195b4b0b0f4d4bc3706650","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-19 05:13:05.000000000","message":"Patch Set 28:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/28/check/DellEMC_Unity_Manila_dhss_true/8d3a0ad/DellEMC_Unity_Manila_dhss_true/605 : SUCCESS in 3h 57m 19s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/28/check/DellEMC_Unity_Manila_dhss_false/c2e74ac/DellEMC_Unity_Manila_dhss_false/635 : SUCCESS in 1h 05m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":28},{"id":"627493a493c478fafdd9816d278a76ec1027ffd5","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-19 05:58:15.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"3240ab263926907ba1070074f97116943ce923f6","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-09-19 10:34:07.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"6714ef7960cabf4a670619a0201a0b23c3aac105","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-20 01:41:17.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"db67fe4135d019a7dc90e9ac16a67d9f48747230","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-20 01:41:41.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"c1f1a6d8cea647f77150e3898187311a495ef286","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-20 02:13:35.000000000","message":"Patch Set 29:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/29/check/DellEMC_PowerFlex_Manila_dhss_false/340bd2e/DellEMC_PowerFlex_Manila_dhss_false/417 : SUCCESS in 32m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":29},{"id":"65001085fa305619e2a4d2eca786708e9f29423a","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-20 03:02:41.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"f8daad9ec7b3409ffa83b7fc946075bdf060fd1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 04:01:08.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e420f497b5534d8d8f084aa58af52f9b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/697eda8cd8a240dca469b335aa126ca2 : SUCCESS in 2m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e7e0934c20ac4db6a33c8c3e791e4590 : SUCCESS in 54m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b70f0ff1fc2d4f329565307ac99fd722 : SUCCESS in 2h 14m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e68dd08599554aadb1c038223b07bf67 : SUCCESS in 1h 16m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d8507780d8d14414933d290edb6cf18b : FAILURE in 1h 05m 34s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/23fc741cffae472bb7893fd7b9f40e3c : FAILURE in 1h 19m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/68ebac6e93c844f1b69d047c5829d4f2 : SUCCESS in 1h 20m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3190b6020171465fb2544a1690433f71 : FAILURE in 41m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/27fa819130b140d99a020f4fbd21dda6 : SUCCESS in 34m 40s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/6a280ecd49fe418dbee6d56fe5dcba38 : SUCCESS in 39m 11s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/2f9d9382e1654972af8970b2ec02363a : SUCCESS in 42m 35s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/07b632d0efc74ac48a4af96e17a2c67c : SUCCESS in 37m 30s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3bde76bced5a4ff1985d21143e648508 : FAILURE in 58m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/04d09fad63ae421ab4ba01cf8c65e7f4 : FAILURE in 14m 36s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"2792e497b2392c9ed6812925a26d880d1aa66ae8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-20 04:41:54.000000000","message":"Patch Set 29:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/29/check/DellEMC_Unity_Manila_dhss_true/8fbeaf6/DellEMC_Unity_Manila_dhss_true/613 : FAILURE in 1h 39m 54s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/29/check/DellEMC_Unity_Manila_dhss_false/6d2d18b/DellEMC_Unity_Manila_dhss_false/643 : FAILURE in 44m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":29},{"id":"1e66224cfe730e8aba59b89fc0f96b64b46bb6d0","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2023-09-20 08:24:49.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"a0ef8b803b28e353af38decfdf7ed4295f1aef26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-20 09:56:37.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"98739716a5c95a4204a68eb538e3fbf9ceace425","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-20 10:29:10.000000000","message":"Patch Set 30:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/30/check/DellEMC_PowerFlex_Manila_dhss_false/632f247/DellEMC_PowerFlex_Manila_dhss_false/418 : SUCCESS in 32m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":30},{"id":"5691db19b3515c5d66658da83bb645b9c7833798","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-20 11:10:36.000000000","message":"Patch Set 30:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/30/check/DellEMC_Unity_Manila_dhss_true/fbcab0d/DellEMC_Unity_Manila_dhss_true/614 : FAILURE in 1h 13m 43s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/30/check/DellEMC_Unity_Manila_dhss_false/59cb2ef/DellEMC_Unity_Manila_dhss_false/644 : FAILURE in 45m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":30},{"id":"3075af141324a42493f73d948e018bf5c1665c2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 12:17:07.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3db6a7c5c8c943e1a2c4979cf2a28155\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5979e14f370e4de7815ba6c76a0026f7 : SUCCESS in 3m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/573d785edda64d3aaa8959a6234284b3 : SUCCESS in 46m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b119f36346dc428f80a134784ffc6fa3 : FAILURE in 2h 19m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b345e936f54c4c75b0bba7d8a20d6d9b : SUCCESS in 1h 22m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/160069d121bb46548b9a8019ce3edc08 : SUCCESS in 1h 14m 28s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/45a05f0781f44ab680fc5c97614e6084 : SUCCESS in 1h 42m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5aff4471b68d4ac4afdb113ab75fbfc0 : FAILURE in 1h 24m 50s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/69583ddca9464720b4d38c768881f713 : SUCCESS in 20m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/657a5ee780d04f2799a348357c8cf9a8 : SUCCESS in 33m 40s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/26db9544786b4b059d4597fd3a54b7e1 : SUCCESS in 49m 26s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/a75fd459a64b45eeadbee3c6a3827bcb : SUCCESS in 48m 56s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/84794b2c70e64268b3102f23051185b8 : SUCCESS in 38m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fbd3c95ebe4846a5bce3a6c63f17b6c8 : SUCCESS in 52m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0eb09ed1c2bd480f8016d38277fc63ed : FAILURE in 23m 58s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"a8178b3bdb5bf7ba6a1dd2bae6d91fec69fe818b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-20 12:49:45.000000000","message":"Patch Set 30:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/30/check/DellEMC_PowerFlex_Manila_dhss_false/3c5764b/DellEMC_PowerFlex_Manila_dhss_false/420 : SUCCESS in 32m 24s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":30},{"id":"aa03ab378e6138dcc70b6fc61f66c4d7ed677ef3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-20 13:00:27.000000000","message":"Patch Set 30:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/30/check/DellEMC_PowerStore_Manila_dhss_false/6e6068e/DellEMC_PowerStore_Manila_dhss_false/182 : SUCCESS in 43m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":30},{"id":"9fbed413776c29c2244aec3a4b504dc30b25caa1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-20 14:06:10.000000000","message":"Patch Set 30:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/51/858351/30/check/DellEMC_Unity_Manila_dhss_true/fa1d1a9/DellEMC_Unity_Manila_dhss_true/616 : FAILURE in 1h 12m 46s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/30/check/DellEMC_Unity_Manila_dhss_false/afba91f/DellEMC_Unity_Manila_dhss_false/646 : FAILURE in 48m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":30},{"id":"e2226336422c15e765b5cf9e2c0043fae8178c99","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-20 14:40:54.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/30/upstream-check/manila-tempest-plugin-ontap-dhss/90a8610/ : SUCCESS in 2h 22m 52s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/30/upstream-check/manila-tempest-plugin-ontap-no-dhss/b097436/ : SUCCESS in 2h 17m 04s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":30},{"id":"e2f9829a54725ac0777c2489dde8b1e79d0a387b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-20 20:26:07.000000000","message":"Patch Set 30: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":30},{"id":"7afcb1126ab91b76d50b8b247191a35f437b469c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-21 00:06:51.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"e3c65d82ce44512c643e20eb628bedbe9329326b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-21 01:06:28.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"f4ce5bdc51a675e41ca4eab467cc5373663bb279","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-21 01:49:05.000000000","message":"Patch Set 31:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/31/check/DellEMC_PowerFlex_Manila_dhss_false/4ceb63e/DellEMC_PowerFlex_Manila_dhss_false/430 : SUCCESS in 32m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":31},{"id":"660abb0ce44dca223d7e1347f2ad0f195bc44e32","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-21 02:13:22.000000000","message":"Patch Set 31: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"0a1b62ff1b5bef126a504e7b5062b9adb2214743","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 02:57:33.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb7fee43ce0048439c2ca2ceebc70e08\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db7e4ca308c24d948762a829a47b220c : SUCCESS in 3m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6d7927c27ba249b397fc0075c527df65 : SUCCESS in 47m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2fbe3bb6a6a44b8c83ea8f8137399fd3 : SUCCESS in 1h 48m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/47f28060886340b9accaa2c5cc1acdb8 : SUCCESS in 1h 44m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/381b4e413fb54be988d421d416509f56 : SUCCESS in 48m 55s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/708ebe67ec224153b57ad983f9bd5da9 : SUCCESS in 1h 42m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e9ecb71c9b6241fcae561dde658d2f68 : SUCCESS in 57m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/494a807a767e4adc98161266e3d86e4d : SUCCESS in 28m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/254a54634e204a9e87b5d41d0c28600e : SUCCESS in 34m 39s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/57eaafda5c524e8db02457e18e07af21 : SUCCESS in 39m 10s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/46a8b4c7f3614b6eb0eb65d92d33e17b : SUCCESS in 35m 25s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/09d87199b5cc4d5e977e0bac0a05c465 : SUCCESS in 21m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ddb891e38086438ba70a18f145e11cfe : FAILURE in 34m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4de477f1adc34c589760657633b8eaa4 : FAILURE in 22m 51s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"292182db08af6896bc49f8e066e4d30029653ba6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-21 03:30:41.000000000","message":"Patch Set 31:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/31/check/DellEMC_PowerFlex_Manila_dhss_false/e6374cf/DellEMC_PowerFlex_Manila_dhss_false/432 : SUCCESS in 32m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":31},{"id":"6acc11a2d52a27bbbdb390649c3380fed7cb3b10","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-21 03:33:54.000000000","message":"Patch Set 31:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/51/858351/31/check/DellEMC_PowerStore_Manila_dhss_false/cd81829/DellEMC_PowerStore_Manila_dhss_false/187 : SUCCESS in 36m 03s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":31},{"id":"1e6314fd8c25c647b27bc2cff890af030d2d7160","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-21 05:57:33.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/31/upstream-check/manila-tempest-plugin-ontap-dhss/7498365/ : RETRY_LIMIT in 8m 21s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/51/858351/31/upstream-check/manila-tempest-plugin-ontap-no-dhss/af80b1a/ : SUCCESS in 2h 58m 09s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":31},{"id":"d21b23e9ccde1b7d425a51928fc2bc6a6097efff","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-22 12:48:34.000000000","message":"Patch Set 31: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"e4ad22abbbe6b6a3c3f2400f990802251bc37b09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 12:49:07.000000000","message":"Patch Set 31: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":31},{"id":"35e11797f5e9bee0df214e0993ddec458ab29ffc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 13:54:40.000000000","message":"Patch Set 31: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f58ab03d2e9d4ffaad0ee40036b1a652\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc1dfcda7d0f437e96a94dc8e96753e5 : SUCCESS in 2m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ae34a53537dd43d8a87d86ff69501ab9 : SUCCESS in 42m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c4d28a33f9c041c187b0354ffb5b5ed9 : SUCCESS in 40m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5ed7a2661d354742ada911bf93003f68 : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":31},{"id":"11f7971b25eeded4e980d89b490d73200aae6faa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 13:54:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":31}],"current_revision_number":31,"current_revision":"1227546e7ca39759fb2240c0e8412e1a7a43cd0b","revisions":{"429d57c3330602b01bd96f00f15dad81b8b3e508":{"kind":"REWORK","_number":1,"created":"2022-09-20 09:14:03.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/1"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:23.000000000","tz":480},"subject":"API 2.74. Add tests for share transfer.","message":"API 2.74. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/429d57c3330602b01bd96f00f15dad81b8b3e508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/429d57c3330602b01bd96f00f15dad81b8b3e508"}]},"branch":"refs/heads/master"},"46369d6633d172d6f8540e13adcdd19e5aed024b":{"kind":"REWORK","_number":2,"created":"2022-09-20 09:16:43.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/2"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:16:34.000000000","tz":480},"subject":"API 2.74. Add tests for share transfer.","message":"API 2.74. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/46369d6633d172d6f8540e13adcdd19e5aed024b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/46369d6633d172d6f8540e13adcdd19e5aed024b"}]},"branch":"refs/heads/master"},"fb96794b23d17a47d214acde6b8843a853671f3d":{"kind":"REWORK","_number":3,"created":"2022-09-20 14:34:44.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/3"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 14:34:18.000000000","tz":480},"subject":"API 2.74. Add tests for share transfer.","message":"API 2.74. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fb96794b23d17a47d214acde6b8843a853671f3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/fb96794b23d17a47d214acde6b8843a853671f3d"}]},"branch":"refs/heads/master"},"c5e98928171d99a9477d2fb9ded625b7dd8de6bf":{"kind":"REWORK","_number":4,"created":"2022-09-20 16:27:45.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/4"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 16:27:18.000000000","tz":480},"subject":"API 2.74. Add tests for share transfer.","message":"API 2.74. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c5e98928171d99a9477d2fb9ded625b7dd8de6bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c5e98928171d99a9477d2fb9ded625b7dd8de6bf"}]},"branch":"refs/heads/master"},"1a811e175f347116b4b8c003330e52c7646d59ee":{"kind":"REWORK","_number":5,"created":"2022-09-21 02:53:14.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/5"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-21 02:53:05.000000000","tz":480},"subject":"API 2.74. Add tests for share transfer.","message":"API 2.74. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1a811e175f347116b4b8c003330e52c7646d59ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1a811e175f347116b4b8c003330e52c7646d59ee"}]},"branch":"refs/heads/master"},"c40d0b02847952ca45d25701e284d0f4db99c624":{"kind":"REWORK","_number":6,"created":"2022-09-21 03:51:37.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/6"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-21 03:51:28.000000000","tz":480},"subject":"API 2.74. Add tests for share transfer.","message":"API 2.74. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c40d0b02847952ca45d25701e284d0f4db99c624"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c40d0b02847952ca45d25701e284d0f4db99c624"}]},"branch":"refs/heads/master"},"a019856e0debc11676793061d94af6fb4fc93fe3":{"kind":"REWORK","_number":7,"created":"2022-09-21 04:51:49.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/7"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-21 04:51:35.000000000","tz":480},"subject":"API 2.74. Add tests for share transfer.","message":"API 2.74. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a019856e0debc11676793061d94af6fb4fc93fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/a019856e0debc11676793061d94af6fb4fc93fe3"}]},"branch":"refs/heads/master"},"143fdeeba557ff907387ef1aa3217ff2e0daa669":{"kind":"REWORK","_number":8,"created":"2022-09-21 05:57:49.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/8"}}},"commit":{"parents":[{"commit":"045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/045dd282fa9ec6de8ee888f9ff2c96fbd22cebe1"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-21 05:57:35.000000000","tz":480},"subject":"API 2.74. Add tests for share transfer.","message":"API 2.74. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/143fdeeba557ff907387ef1aa3217ff2e0daa669"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/143fdeeba557ff907387ef1aa3217ff2e0daa669"}]},"branch":"refs/heads/master"},"cfa998a2ed3e14c536716b183f42351247dc6c48":{"kind":"REWORK","_number":9,"created":"2022-11-16 04:35:00.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/9"}}},"commit":{"parents":[{"commit":"121f536ac32649fdc9f13dca681ab3fd69e725e8","subject":"[RBAC] Fix test tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/121f536ac32649fdc9f13dca681ab3fd69e725e8"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-11-16 04:33:37.000000000","tz":480},"subject":"API 2.74. Add tests for share transfer.","message":"API 2.74. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cfa998a2ed3e14c536716b183f42351247dc6c48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/cfa998a2ed3e14c536716b183f42351247dc6c48"}]},"branch":"refs/heads/master"},"1fb32c8e8d095bb984f9d576fed55b7a901bfa4e":{"kind":"REWORK","_number":10,"created":"2022-12-11 06:49:14.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/10"}}},"commit":{"parents":[{"commit":"8dc165888132bbf43c63f2d501717e5308f54956","subject":"Merge \"Fix incorrect url request for share group type extra specs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/8dc165888132bbf43c63f2d501717e5308f54956"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-12-11 06:49:04.000000000","tz":480},"subject":"API 2.74. Add tests for share transfer.","message":"API 2.74. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1fb32c8e8d095bb984f9d576fed55b7a901bfa4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1fb32c8e8d095bb984f9d576fed55b7a901bfa4e"}]},"branch":"refs/heads/master"},"33efe39724c92afa92865d3df043583a1e5a5139":{"kind":"REWORK","_number":11,"created":"2023-01-29 02:33:35.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/11"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-01-29 02:33:10.000000000","tz":480},"subject":"API 2.75. Add tests for share transfer.","message":"API 2.75. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/33efe39724c92afa92865d3df043583a1e5a5139"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/33efe39724c92afa92865d3df043583a1e5a5139"}]},"branch":"refs/heads/master"},"09080e56124948ed17f341ff2c9369d69fe3107a":{"kind":"REWORK","_number":12,"created":"2023-02-06 08:07:42.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/12"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-02-06 08:07:28.000000000","tz":480},"subject":"API 2.75. Add tests for share transfer.","message":"API 2.75. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/09080e56124948ed17f341ff2c9369d69fe3107a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/09080e56124948ed17f341ff2c9369d69fe3107a"}]},"branch":"refs/heads/master"},"51c3e82ab0ed25e0d8b95299eff811f112252b59":{"kind":"REWORK","_number":13,"created":"2023-02-07 09:15:39.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/13"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-02-07 09:15:27.000000000","tz":480},"subject":"API 2.75. Add tests for share transfer.","message":"API 2.75. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/51c3e82ab0ed25e0d8b95299eff811f112252b59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/51c3e82ab0ed25e0d8b95299eff811f112252b59"}]},"branch":"refs/heads/master"},"f1b1843a20ba7bcca3490e6360e855b1de8ae838":{"kind":"REWORK","_number":14,"created":"2023-02-08 00:57:58.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/14"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-02-08 00:57:49.000000000","tz":480},"subject":"API 2.75. Add tests for share transfer.","message":"API 2.75. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1b1843a20ba7bcca3490e6360e855b1de8ae838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f1b1843a20ba7bcca3490e6360e855b1de8ae838"}]},"branch":"refs/heads/master"},"d6ce98a118717f95378d76289335a507f9497e8d":{"kind":"REWORK","_number":15,"created":"2023-02-08 06:59:05.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/15"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-02-08 06:58:25.000000000","tz":480},"subject":"API 2.75. Add tests for share transfer.","message":"API 2.75. Add tests for share transfer.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6ce98a118717f95378d76289335a507f9497e8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d6ce98a118717f95378d76289335a507f9497e8d"}]},"branch":"refs/heads/master"},"991feb1a4c0eca1166fdfedababada75cbe9ba79":{"kind":"REWORK","_number":16,"created":"2023-02-16 13:35:03.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/16"}}},"commit":{"parents":[{"commit":"54e31e8f5b517d2f1b350c091b29aa518f2c8739","subject":"Skip replication test if microversion is not supported","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/54e31e8f5b517d2f1b350c091b29aa518f2c8739"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-02-16 13:34:12.000000000","tz":480},"subject":"Update micversion to 2.76,support share transfer between project.","message":"Update micversion to 2.76,support share transfer between project.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/991feb1a4c0eca1166fdfedababada75cbe9ba79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/991feb1a4c0eca1166fdfedababada75cbe9ba79"}]},"branch":"refs/heads/master"},"d998c5218ca2d4e7e86f7ed0ff6c5f1cb48872c0":{"kind":"REWORK","_number":17,"created":"2023-02-17 03:00:31.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/17"}}},"commit":{"parents":[{"commit":"07f29e172b177f4f91140d470437e5c7c11d73a5","subject":"Merge \"Revert \"Use ddt.named_data decorator to avoid random test names\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/07f29e172b177f4f91140d470437e5c7c11d73a5"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-02-17 03:00:18.000000000","tz":480},"subject":"Update micversion to 2.77,support share transfer between project.","message":"Update micversion to 2.77,support share transfer between project.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d998c5218ca2d4e7e86f7ed0ff6c5f1cb48872c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/d998c5218ca2d4e7e86f7ed0ff6c5f1cb48872c0"}]},"branch":"refs/heads/master"},"c3f38525519c3645475373191cd14b995c167d13":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2023-02-17 11:52:00.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/51/858351/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/18"}}},"commit":{"parents":[{"commit":"07f29e172b177f4f91140d470437e5c7c11d73a5","subject":"Merge \"Revert \"Use ddt.named_data decorator to avoid random test names\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/07f29e172b177f4f91140d470437e5c7c11d73a5"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2023-02-17 11:52:00.000000000","tz":0},"subject":"Update micversion to 2.77,support share transfer between project.","message":"Update micversion to 2.77,support share transfer between project.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nDepends-On: https://review.opendev.org/c/openstack/manila/+/874229\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c3f38525519c3645475373191cd14b995c167d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/c3f38525519c3645475373191cd14b995c167d13"}]},"branch":"refs/heads/master","description":"Edit commit message"},"021ab310e4104d4d1f457d440b17ad1f14203f66":{"kind":"REWORK","_number":19,"created":"2023-02-21 09:05:21.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/19"}}},"commit":{"parents":[{"commit":"07f29e172b177f4f91140d470437e5c7c11d73a5","subject":"Merge \"Revert \"Use ddt.named_data decorator to avoid random test names\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/07f29e172b177f4f91140d470437e5c7c11d73a5"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-02-21 09:05:02.000000000","tz":480},"subject":"Update micversion to 2.77,support share transfer between project.","message":"Update micversion to 2.77,support share transfer between project.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nDepends-On: https://review.opendev.org/c/openstack/manila/+/874229\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/021ab310e4104d4d1f457d440b17ad1f14203f66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/021ab310e4104d4d1f457d440b17ad1f14203f66"}]},"branch":"refs/heads/master"},"01decc0593a1e08d244585e26e7eb56b9fdcf439":{"kind":"REWORK","_number":20,"created":"2023-02-22 06:33:26.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/20"}}},"commit":{"parents":[{"commit":"07f29e172b177f4f91140d470437e5c7c11d73a5","subject":"Merge \"Revert \"Use ddt.named_data decorator to avoid random test names\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/07f29e172b177f4f91140d470437e5c7c11d73a5"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-02-22 06:33:15.000000000","tz":480},"subject":"Update micversion to 2.77,support share transfer between project.","message":"Update micversion to 2.77,support share transfer between project.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nDepends-On: https://review.opendev.org/c/openstack/manila/+/874229\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/01decc0593a1e08d244585e26e7eb56b9fdcf439"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/01decc0593a1e08d244585e26e7eb56b9fdcf439"}]},"branch":"refs/heads/master"},"35d4985ad6b3337e4e625529603e19221a1b00d5":{"kind":"REWORK","_number":21,"created":"2023-02-22 06:35:52.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/21"}}},"commit":{"parents":[{"commit":"07f29e172b177f4f91140d470437e5c7c11d73a5","subject":"Merge \"Revert \"Use ddt.named_data decorator to avoid random test names\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/07f29e172b177f4f91140d470437e5c7c11d73a5"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-02-22 06:35:41.000000000","tz":480},"subject":"Update micversion to 2.77,support share transfer between project.","message":"Update micversion to 2.77,support share transfer between project.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nDepends-On: https://review.opendev.org/c/openstack/manila/+/874229\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/35d4985ad6b3337e4e625529603e19221a1b00d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/35d4985ad6b3337e4e625529603e19221a1b00d5"}]},"branch":"refs/heads/master"},"9131cee00f64be8ef7cc42006a1bf726bec03504":{"kind":"REWORK","_number":22,"created":"2023-02-22 09:18:35.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/22"}}},"commit":{"parents":[{"commit":"07f29e172b177f4f91140d470437e5c7c11d73a5","subject":"Merge \"Revert \"Use ddt.named_data decorator to avoid random test names\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/07f29e172b177f4f91140d470437e5c7c11d73a5"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-02-22 09:18:21.000000000","tz":480},"subject":"Update micversion to 2.77,support share transfer between project.","message":"Update micversion to 2.77,support share transfer between project.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nDepends-On: https://review.opendev.org/c/openstack/manila/+/874229\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9131cee00f64be8ef7cc42006a1bf726bec03504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/9131cee00f64be8ef7cc42006a1bf726bec03504"}]},"branch":"refs/heads/master"},"dff7c2e2dd6a26d89a48ad27f237298a766c2f74":{"kind":"REWORK","_number":23,"created":"2023-02-22 09:52:26.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/23"}}},"commit":{"parents":[{"commit":"07f29e172b177f4f91140d470437e5c7c11d73a5","subject":"Merge \"Revert \"Use ddt.named_data decorator to avoid random test names\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/07f29e172b177f4f91140d470437e5c7c11d73a5"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-02-22 09:52:16.000000000","tz":480},"subject":"Update micversion to 2.77,support share transfer between project.","message":"Update micversion to 2.77,support share transfer between project.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nDepends-On: https://review.opendev.org/c/openstack/manila/+/874229\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dff7c2e2dd6a26d89a48ad27f237298a766c2f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dff7c2e2dd6a26d89a48ad27f237298a766c2f74"}]},"branch":"refs/heads/master"},"f5b7417570011168981f744beee0904700fc5ef0":{"kind":"REWORK","_number":24,"created":"2023-09-15 07:01:07.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/24"}}},"commit":{"parents":[{"commit":"dc97a5179a044000587ac03b310dc62b011bf93f","subject":"Merge \"[CI] Disable glance RBAC enforcement in generic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc97a5179a044000587ac03b310dc62b011bf93f"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-09-15 06:59:40.000000000","tz":480},"subject":"Update micversion to 2.77,support share transfer between project.","message":"Update micversion to 2.77,support share transfer between project.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nDepends-On: https://review.opendev.org/c/openstack/manila/+/874229\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f5b7417570011168981f744beee0904700fc5ef0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/f5b7417570011168981f744beee0904700fc5ef0"}]},"branch":"refs/heads/master"},"7e1ca61974c1903573d4bbf959c48fc9d29513c1":{"kind":"NO_CODE_CHANGE","_number":25,"created":"2023-09-18 01:02:33.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/25"}}},"commit":{"parents":[{"commit":"dc97a5179a044000587ac03b310dc62b011bf93f","subject":"Merge \"[CI] Disable glance RBAC enforcement in generic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc97a5179a044000587ac03b310dc62b011bf93f"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-09-18 01:02:33.000000000","tz":0},"subject":"Add share transfer tests.","message":"Add share transfer tests.\n\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7e1ca61974c1903573d4bbf959c48fc9d29513c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/7e1ca61974c1903573d4bbf959c48fc9d29513c1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"42cff5c935eed092911c813a1d89c16d7b560324":{"kind":"REWORK","_number":26,"created":"2023-09-18 09:33:27.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/26"}}},"commit":{"parents":[{"commit":"dc97a5179a044000587ac03b310dc62b011bf93f","subject":"Merge \"[CI] Disable glance RBAC enforcement in generic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc97a5179a044000587ac03b310dc62b011bf93f"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-09-18 09:33:03.000000000","tz":480},"subject":"Update micversion to 2.77,support share transfer between project.","message":"Update micversion to 2.77,support share transfer between project.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nDepends-On: https://review.opendev.org/c/openstack/manila/+/874229\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/42cff5c935eed092911c813a1d89c16d7b560324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/42cff5c935eed092911c813a1d89c16d7b560324"}]},"branch":"refs/heads/master"},"2b2639c8ef755bb788e1a405c08ab2f3913d9d73":{"kind":"REWORK","_number":27,"created":"2023-09-19 01:13:12.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/27"}}},"commit":{"parents":[{"commit":"dc97a5179a044000587ac03b310dc62b011bf93f","subject":"Merge \"[CI] Disable glance RBAC enforcement in generic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc97a5179a044000587ac03b310dc62b011bf93f"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-09-19 01:12:31.000000000","tz":480},"subject":"Update micversion to 2.77,support share transfer between project.","message":"Update micversion to 2.77,support share transfer between project.\n\nDepends-On: I8facf9112a6b09e6b7aed9956c0a87fb5f1fc31f\nDepends-On: https://review.opendev.org/c/openstack/manila/+/874229\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b2639c8ef755bb788e1a405c08ab2f3913d9d73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/2b2639c8ef755bb788e1a405c08ab2f3913d9d73"}]},"branch":"refs/heads/master"},"e1db08f7b990f6c2443d2ac7849d0b1f4c8bd734":{"kind":"NO_CODE_CHANGE","_number":28,"created":"2023-09-19 01:15:29.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/28"}}},"commit":{"parents":[{"commit":"dc97a5179a044000587ac03b310dc62b011bf93f","subject":"Merge \"[CI] Disable glance RBAC enforcement in generic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc97a5179a044000587ac03b310dc62b011bf93f"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-09-19 01:15:29.000000000","tz":0},"subject":"Add share transfer test.","message":"Add share transfer test.\n\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e1db08f7b990f6c2443d2ac7849d0b1f4c8bd734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e1db08f7b990f6c2443d2ac7849d0b1f4c8bd734"}]},"branch":"refs/heads/master","description":"Edit commit message"},"109695cd929a6a028be5354448f6735e63c1d2a3":{"kind":"REWORK","_number":29,"created":"2023-09-20 01:41:17.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/29"}}},"commit":{"parents":[{"commit":"dc97a5179a044000587ac03b310dc62b011bf93f","subject":"Merge \"[CI] Disable glance RBAC enforcement in generic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc97a5179a044000587ac03b310dc62b011bf93f"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-09-20 01:40:57.000000000","tz":480},"subject":"Add share transfer test.","message":"Add share transfer test.\n\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/109695cd929a6a028be5354448f6735e63c1d2a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/109695cd929a6a028be5354448f6735e63c1d2a3"}]},"branch":"refs/heads/master"},"dec59c1c1d80877fe6c67e463ee1ac0264e7cf60":{"kind":"REWORK","_number":30,"created":"2023-09-20 09:56:37.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/30"}}},"commit":{"parents":[{"commit":"dc97a5179a044000587ac03b310dc62b011bf93f","subject":"Merge \"[CI] Disable glance RBAC enforcement in generic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc97a5179a044000587ac03b310dc62b011bf93f"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-09-20 09:56:16.000000000","tz":480},"subject":"Add share transfer test.","message":"Add share transfer test.\n\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dec59c1c1d80877fe6c67e463ee1ac0264e7cf60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dec59c1c1d80877fe6c67e463ee1ac0264e7cf60"}]},"branch":"refs/heads/master"},"1227546e7ca39759fb2240c0e8412e1a7a43cd0b":{"kind":"REWORK","_number":31,"created":"2023-09-21 01:06:28.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/51/858351/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/51/858351/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/51/858351/31"}}},"commit":{"parents":[{"commit":"dc97a5179a044000587ac03b310dc62b011bf93f","subject":"Merge \"[CI] Disable glance RBAC enforcement in generic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/dc97a5179a044000587ac03b310dc62b011bf93f"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2022-09-20 09:13:17.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-09-21 01:06:06.000000000","tz":480},"subject":"Add share transfer test.","message":"Add share transfer test.\n\nPartially-Implements: blueprint transfer-share-between-project\n\nChange-Id: I74f0a079edb59e376d045fe9e9fd781acd70249d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1227546e7ca39759fb2240c0e8412e1a7a43cd0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/1227546e7ca39759fb2240c0e8412e1a7a43cd0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
