)]}'
{"id":"openstack%2Fmanila-tempest-plugin~861335","triplet_id":"openstack%2Fmanila-tempest-plugin~master~Ib285d9c740558d1ff3e24acb1ccc3de7e1ad562f","project":"openstack/manila-tempest-plugin","branch":"master","topic":"bug/1993283","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-10-18 15:08:58.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"35075":{"account":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"last_update":"2022-10-18 16:09:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib285d9c740558d1ff3e24acb1ccc3de7e1ad562f","subject":"Add a metadata field shown on API \u003e\u003d 2.73","status":"MERGED","created":"2022-10-18 11:16:39.000000000","updated":"2022-10-18 16:09:02.000000000","submitted":"2022-10-18 16:09:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861335-bug/1993283","meta_rev_id":"de9476ef23be25532ad08e287e5b49347019c657","_number":861335,"virtual_id_number":861335,"owner":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-18 16:09:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-10-18 15:08:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-10-18 15:07:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2022-10-18 15:08:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 12:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-18 13:40:25.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-10-18 14:33:30.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2022-10-18 14:42:05.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2022-10-18 15:07:02.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-10-18 15:07:19.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"2a86bb845f4364e823efebaed413f2082957dedc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"date":"2022-10-18 11:16:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8df66ee2fdfda57ea46e5153551723cf7c3577bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 12:55:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edfc3f4ab0b1418abe35b3b1b32476b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/015011e5a6294cb9a2b22dd9560f2d51 : SUCCESS in 2m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e87e4882036542f3b69624a0cf14c9bb : SUCCESS in 54m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/24d7d0d1f7dc4ff2bdc2a9ce3b57c519 : SUCCESS in 1h 37m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4ede8aaa605b4265acc2bf056603f092 : SUCCESS in 56m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/52d9c41bb0334114a14ac29f019d8e22 : SUCCESS in 1h 12m 21s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/495b2684bbbd47819d598f8b587fb042 : FAILURE in 1h 19m 59s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a58de58bae4947a28c990f23cfc416fc : SUCCESS in 1h 30m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ac2373c88a564204ad6012c998bd9c15 : SUCCESS in 34m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f8ee2b6dd6854df9aa1e337b11506196 : SUCCESS in 44m 39s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c37a045d651440fc9312b442e9ed2ac4 : SUCCESS in 1h 18m 28s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/67251467ac5a46f7965f8f36920f7156 : SUCCESS in 1h 29m 26s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/286c388782384b1893e034ba8589e7a3 : SUCCESS in 1h 14m 43s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/9710958c90284fcba381a920237a6f25 : SUCCESS in 31m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8186cbd33a3b4f4e9b810a1ba7f15862 : SUCCESS in 45m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d1a5150a27c74475b2b188f29169bdf0 : SUCCESS in 35m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"21d565564aeba2928b441197ecd8287e27b977fa","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-18 13:40:25.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/35/861335/1/check/DellEMC_Unity_Manila_dhss_true/ca8cc1f/DellEMC_Unity_Manila_dhss_true/979 : FAILURE in 2h 23m 32s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/35/861335/1/check/DellEMC_Unity_Manila_dhss_false/99768db/DellEMC_Unity_Manila_dhss_false/1041 : SUCCESS in 53m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"8dccae723613142223f6b1308c104e0389005de3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-18 14:33:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/861335/1/upstream-check/manila-tempest-plugin-ontap-dhss/427a2f9/ : SUCCESS in 1h 36m 36s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/35/861335/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/a5abfc9/ : FAILURE in 1h 35m 29s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d2b645e1f9287888d00f4940b1aa95dbb384d78c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2022-10-18 14:42:05.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n* Infinidat-NFS http://openstack-ci-logs.aws.infinidat.com/#logs/refs-changes-35-861335-1/ : SUCCESS in 1h 7m 43s\nFor more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"9446365becca4635872b0e979266f7007a11a894","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2022-10-18 15:07:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a4312e77567123bf87a5e1891af9a984f2bffde4","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-10-18 15:08:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8fb1c508970497608495202712e2caf54d92d9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 15:09:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eabbf92530d07af296d52ae05acf46cb46141134","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 16:09:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60ef0c6b7ac04eb28912779ce1791e31\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/784156090d524095bedf92bb655d8fec : SUCCESS in 2m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/007b8025a133487095e7daa71e3592da : SUCCESS in 58m 43s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/33fbcf7464cd4d559712650ab841bc34 : SUCCESS in 23m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0d1923b04e5f4cd39787c772db5f9dee : SUCCESS in 27m 47s","accounts_in_message":[],"_revision_number":1},{"id":"de9476ef23be25532ad08e287e5b49347019c657","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 16:09:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75dbca626e46b26dda38e98b3f98e9f5cf7f5b4c","revisions":{"75dbca626e46b26dda38e98b3f98e9f5cf7f5b4c":{"kind":"REWORK","_number":1,"created":"2022-10-18 11:16:39.000000000","uploader":{"_account_id":35075,"name":"Alexander Deiter","email":"adeiter@infinidat.com","username":"adeiter"},"ref":"refs/changes/35/861335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/35/861335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/861335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/861335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/861335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/35/861335/1"}}},"commit":{"parents":[{"commit":"e344e37d62caf65da9b8e1ba7a6bc925cfded7fc","subject":"Merge \"Add S-RBAC tests for manila\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/e344e37d62caf65da9b8e1ba7a6bc925cfded7fc"}]}],"author":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2022-10-18 11:07:21.000000000","tz":180},"committer":{"name":"Alexander Deiter","email":"adeiter@infinidat.com","date":"2022-10-18 11:07:21.000000000","tz":180},"subject":"Add a metadata field shown on API \u003e\u003d 2.73","message":"Add a metadata field shown on API \u003e\u003d 2.73\n\nFixed SharesActionsTest cases for API version \u003e\u003d 2.73\nsince a metadata field was added in that version.\n\nCloses-bug: #1993283\nSigned-off-by: Alexander Deiter \u003cadeiter@infinidat.com\u003e\nChange-Id: Ib285d9c740558d1ff3e24acb1ccc3de7e1ad562f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/75dbca626e46b26dda38e98b3f98e9f5cf7f5b4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/75dbca626e46b26dda38e98b3f98e9f5cf7f5b4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[]}
