)]}'
{"id":"openstack%2Fmanila-tempest-plugin~862785","triplet_id":"openstack%2Fmanila-tempest-plugin~master~I756779789becbb0aeec74380d40f9c5fb4cd3fe8","project":"openstack/manila-tempest-plugin","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2022-11-08 21:06:24.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2022-11-22 21:47:20.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2022-11-09 20:45:02.000000000","reason":"\u003cGERRIT_ACCOUNT_30025\u003e replied on the change","reason_account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},"19262":{"account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"last_update":"2022-11-22 18:46:26.000000000","reason":"\u003cGERRIT_ACCOUNT_19262\u003e replied on the change","reason_account":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"}}},"hashtags":[],"change_id":"I756779789becbb0aeec74380d40f9c5fb4cd3fe8","subject":"[RBAC] Add share export locations tests","status":"MERGED","created":"2022-10-27 08:27:50.000000000","updated":"2022-11-23 04:00:17.000000000","submitted":"2022-11-22 21:47:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862785-secure-rbac","meta_rev_id":"0e377a1adc54a27ccf9dc15deb423b4e6e569384","_number":862785,"virtual_id_number":862785,"owner":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-22 21:47:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2022-11-08 21:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2022-11-09 20:45:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2022-11-15 21:28:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"CC":[{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-27 09:02:21.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2022-10-27 09:02:21.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2022-10-27 09:02:21.000000000","updated_by":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2022-10-27 10:32:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-27 11:23:16.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2022-10-27 18:36:05.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"b6a32481d1bd98f90b490a4d363b14d1b7e91288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-27 08:27:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3baf793b45de08618b590ab06b391904aab7106","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-27 08:30:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ab52202c8a5232d4bea6cb1e5cf4081be786af2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 10:32:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d4eeb57f71e4157b66b7ebea694a815\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b07774cef6f84e61b267a9369d0551cb : SUCCESS in 2m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9d4da2ff671f498d83b9ac92ba9ea4e7 : FAILURE in 1h 02m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5e015f5945e14859872904bfca9fded2 : SUCCESS in 2h 00m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/06a617eb245244039c2cdc3fc60c8dbd : SUCCESS in 1h 12m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c5d22c815951408fa6e47c029ec7a284 : SUCCESS in 1h 04m 43s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b0ee1eb5d6b641b6966aaaf2eb2a882b : FAILURE in 1h 38m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cc6f654641a74235b60d8509362a1bed : FAILURE in 1h 04m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2b61beb020144126a0a61f2ca2e7ed9f : SUCCESS in 28m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0328d11ff7074966b6444dc11989e2f1 : SUCCESS in 31m 59s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/ae652c3cb5034bfeae0dceb0593db3d4 : FAILURE in 1h 01m 07s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/772bc3bbddd043148fd13b1d15214940 : SUCCESS in 57m 40s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/434c382571ff46bb83124eecb582b4be : SUCCESS in 1h 05m 10s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/2048cfa98281438aa56c159d8a4ec324 : SUCCESS in 22m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ef20313fb9424fea8080618950666c90 : SUCCESS in 44m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/99b99d4837d8400db36884632e1ba408 : SUCCESS in 18m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f193ae6240d595b55ac96179f8cbd86b3009bf77","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-27 10:35:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43c9f8079aa892f91012d46ccd3fd9ed6c516364","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-27 11:23:16.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/85/862785/2/check/DellEMC_Unity_Manila_dhss_true/c10988f/DellEMC_Unity_Manila_dhss_true/1005 : FAILURE in 2h 52m 06s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/85/862785/2/check/DellEMC_Unity_Manila_dhss_false/c966966/DellEMC_Unity_Manila_dhss_false/1069 : SUCCESS in 1h 00m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"1be7ef4c87613e74eb296663a8b8c42e195ff902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 12:44:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81dc738977434566a159d00bfac745b2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d1edbeed28a465ea2fd1c5101d0bb2f : SUCCESS in 2m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a7d170b3968642c8a7be44a837276db1 : SUCCESS in 1h 03m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/828e14126c6244c9b0e7ed5c6cee7cff : FAILURE in 2h 07m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/22e9fdf1b1e74d93b8a032a4585e8540 : SUCCESS in 1h 16m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/77c895108b4e47339cff6d648a0f4adb : SUCCESS in 1h 03m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/847acd6903d947b080552be9bb962fbb : FAILURE in 2h 02m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d30b04ebdc7142e6bf00fba71570d42c : SUCCESS in 58m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fce8b13c9b7447b4a41652d82c4687fb : SUCCESS in 26m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/158374c24d5d44f39b54c6b013cc1462 : SUCCESS in 36m 48s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/7dada96fea2840ed91586f313622aec3 : SUCCESS in 1h 12m 11s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/9a0018ac3f8d40fa9ca4722941190204 : FAILURE in 1h 03m 13s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/87365b1b15334b7fb663551ba1f1bb29 : SUCCESS in 1h 00m 11s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/7ff3a8a678ab4624ac61ce5e744ca454 : SUCCESS in 27m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c9cb9731e84b41c5bb433c066e65b73e : SUCCESS in 46m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/387db9c0d45e4d898120ce667f7fa5aa : SUCCESS in 20m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65403a70a4e26244a1767fdc08f0fcb30e712f92","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-10-27 13:40:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00c3d7936bff83259b6d1a7a83a894f5595cf27a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 16:10:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b437f9369334dbc9637db92096e7f0e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec8252aa562840818fe43523097a3833 : SUCCESS in 3m 19s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/85a958e4468544a49f80a8bb7e935520 : SUCCESS in 1h 05m 22s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a1ad72692943428cafa6b475995d3122 : FAILURE in 2h 25m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3dfcf4b88d314866bc95c5722266083a : SUCCESS in 1h 10m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/ac0ff3ac485e4a1e9f5b48aa0074b202 : FAILURE in 1h 11m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/db6eac46b4884e2398de79ce39b7618f : FAILURE in 1h 32m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7802629a1147476097bcd791a4072b92 : SUCCESS in 1h 25m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/62e46077f9d541898b01f9f01393ad39 : SUCCESS in 34m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1bac97eeca6a41509b980a766cfeffac : SUCCESS in 38m 30s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/17d4aa1726ad47019b0e663f3660f4be : SUCCESS in 1h 36m 52s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/f8544800f79648048c27b542b8de2723 : SUCCESS in 1h 06m 10s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/684944cfd5884513af2f4e44dc0bf739 : SUCCESS in 1h 35m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/40333da6fa424b5389073cbc2dfb44cc : SUCCESS in 29m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/86f599a7593d47d086842fc27a2bc261 : SUCCESS in 52m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1755a66a724f417c93d2c8546ebd5607 : SUCCESS in 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a48300a9e6fdefdc802a36ca5c281985b6b9163","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-10-27 18:36:05.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/85/862785/2/upstream-check/manila-tempest-plugin-ontap-dhss/7128e5b/ : FAILURE in 2h 23m 44s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/85/862785/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/bd959a1/ : FAILURE in 1h 55m 30s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"00f3b1fce17b1194dfa6637d6e0dbfb3f0489dd3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-10-27 21:56:28.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/85/862785/2/check/DellEMC_Unity_Manila_dhss_true/ff77304/DellEMC_Unity_Manila_dhss_true/1009 : FAILURE in 3h 34m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/85/862785/2/check/DellEMC_Unity_Manila_dhss_false/a700d20/DellEMC_Unity_Manila_dhss_false/1075 : SUCCESS in 1h 16m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"a4598ee9ddbe069509e2bdf630ff8a20b4be6112","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2022-11-08 21:06:24.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"338f47b7f1066ef058f96815500a61223161e00a","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-11-09 20:45:02.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9354dfdb4ed855fd79819e44f067278d55dda66","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2022-11-15 21:28:34.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2747e8698e6ad9cf14f20757fab96ea4ab6c5a63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 21:29:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d867782c71cab8e8d9ed09368fd904f90ae6a577","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 22:24:03.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a3779c8bd5e4632b86328b8382065ca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54ed94369a94467cbb9661833cb30604 : SUCCESS in 2m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/53a277d8876a46a39eafce86cd551018 : FAILURE in 53m 21s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9975b84b0b854cd8858df23c320d91e5 : SUCCESS in 23m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/29da99c9999448d88b357a0a5d84c5fc : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":2},{"id":"c6cdb5cee1036f292fbbc25fcbe01f6a6b410e41","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-16 05:22:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d252a2a911f96a6b21043df8add1c8aa7b1ff9a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 07:02:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64a05307b5504e2ca1382e25b09c22a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc06ec2f33fb4a4babab91ea93e4c023 : SUCCESS in 2m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3dbc20afdd89439bad4c930ed59fa329 : FAILURE in 59m 23s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9041a4acfd3940f78016c8c1468697d5 : SUCCESS in 1h 38m 27s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e13927efc40c465989590c6a4737cf27 : SUCCESS in 1h 02m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/059c3a612f0a4e059569cce03426ec48 : SUCCESS in 1h 02m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a005e8b84fbb401fb78960fdd1874fae : FAILURE in 1h 14m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a49608096eef4096a7674dfcc1f981fd : FAILURE in 1h 21m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b044cb8c4502472ba02f01c71ca4fcdb : SUCCESS in 21m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/80360208c73447c287dfe1a782e529a0 : SUCCESS in 33m 26s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/175ea97072f848969e269a0532154aaa : SUCCESS in 1h 05m 12s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/c062fccb470549dfb3d491f378cea6dd : SUCCESS in 1h 15m 42s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/d82932b7485b4a539ed3f2d79a62d739 : SUCCESS in 1h 20m 47s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/ff96be6616714299b83fabaa1fe0a040 : SUCCESS in 24m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8454416c024e4483995f02beed80f6d1 : SUCCESS in 38m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f589845d440d4fb699563e75ab14a7d9 : SUCCESS in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"014b1904883780a8429f6be7696fcf151636c062","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-16 07:43:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"239c2b5841322693a4eeaa967a2c0cfe5493a8b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 09:35:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13703ff6767345a99e3931f38abe4a30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d70c8c7172244c98af3c7019f015e875 : SUCCESS in 2m 29s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fca02d4769aa4f798f627009205dac81 : FAILURE in 1h 13m 38s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/87599fd9ecda49a29b24888992e77f08 : SUCCESS in 1h 38m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d7663692d2c44897b1bca41c156f55bf : SUCCESS in 1h 12m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/0bffbc27d94140a5b35a2ae1bffc75fd : FAILURE in 56m 52s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/815d04f216f742eab83589c0bfaea327 : FAILURE in 1h 49m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fda4e124a04f4c4b81a776e0da780cdb : FAILURE in 1h 06m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/228dab23965346109bec02d7b67d1d5a : SUCCESS in 23m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a51a832d03d4466e99e977f9d157a576 : SUCCESS in 33m 48s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/c11ce8c8a12c420e993251f48379338f : SUCCESS in 1h 31m 15s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/cf6b8aa0323c4b628683644a01a992cd : SUCCESS in 58m 11s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a091ba185be2497096f241d10274ebfb : SUCCESS in 1h 09m 07s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/d864fa72167148b4bececb7cc4158fbb : SUCCESS in 23m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fdab580cbabd4719af17a134e916cc41 : SUCCESS in 39m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/35d426b46f48428088013813f2270368 : SUCCESS in 20m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a80e3b442251b75cf532e376a8a6f9b0669c357","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-17 19:38:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"515af17e27f942ce87c3ecde67f62356aa403e04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 21:24:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/326034abe46d4ded8828596904ff0ba0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/396bcb09fb7440f6beaea624c406b798 : SUCCESS in 3m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4bb9cfea81504794b282f3bc1180a2d4 : SUCCESS in 1h 02m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a73d37888c1443b0b253b8674a4d51b4 : FAILURE in 1h 43m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d562200d4dd542ba9a0e889665dd4db3 : FAILURE in 22m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/c8968fde75404fda95e10b1cc2e968b4 : SUCCESS in 1h 22m 46s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/c2eb0baed134410da5a2e8fe35cb1b70 : FAILURE in 1h 40m 05s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/562dcb9628e745e186f8792e5d527eaf : SUCCESS in 1h 01m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c6aa921fb7a6402d904d2784def86fc6 : SUCCESS in 26m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c78ec44db58a4b158de37dd698567968 : FAILURE in 37m 51s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/e2639d3721f7408cbce1935aa8054cec : SUCCESS in 1h 15m 28s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/0b07c985c59549aab5d9716d971ab709 : SUCCESS in 55m 08s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a4220077a17b48f99ecd7b871e5cf7c4 : SUCCESS in 1h 00m 15s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/072dd1359d064aadbbda265b362c4562 : SUCCESS in 29m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dabf5049e9be4e34b85bc14c7f83702e : SUCCESS in 48m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/40078192fb1c4edba9a9679897b3712c : SUCCESS in 22m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8091762f8fed504fb03c0aaa9fbb31986676ca74","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-21 10:00:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fac539aaeedf7e27756d11503e33a5f2feca3541","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-21 12:29:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f4a07a5002346f4a858916da6f50e6d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd266e4927f34292879a506b47bd1b99 : SUCCESS in 2m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/51e0eaf8aa814b9288e5873f367529db : FAILURE in 23m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d68f8377d1db41bba007ccbb7a3a5afd : SUCCESS in 2h 11m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/320bc9ca9a0249bab8c15381bc913102 : SUCCESS in 1h 13m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/1cdf35eb02ec423cb324d036b5de0213 : FAILURE in 34m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a5b0c693286b4477aeda404b313c2f48 : FAILURE in 17m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/88ac42c45eb142af88113d276f734d01 : SUCCESS in 54m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d4b6be3a039249c595d184b3b097af92 : SUCCESS in 33m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e35451f96bc9495f990eabcaf85e0dbe : SUCCESS in 31m 08s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/cc9b0f1fa7fa4a7b940822a0b3122bd1 : SUCCESS in 1h 05m 09s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/d114278c4375444fa32fcb521f0a989b : SUCCESS in 1h 26m 12s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/a666e422fc8b4001ad49a20b65422ed5 : SUCCESS in 1h 17m 56s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/3ce63e6d426f4424a0bdc117b15f7093 : SUCCESS in 31m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c3ffc24ed52e4c798944fd9ad2b6ccf8 : SUCCESS in 59m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/865801907ca34dc680b60866b1246fb0 : FAILURE in 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c7981a296fba1e366a14b356d850818436a574d4","author":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"date":"2022-11-22 18:46:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e8e40c28c6310b5c22e3b6f33dd9c5e4ec5fa36a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 20:47:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56fd8bfd36ef44fb8c3eb1e783e1e172\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fb7ce0dc0654efdad0a56eef14ee985 : SUCCESS in 3m 17s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d98c25027d3b42a498245be41f417237 : SUCCESS in 1h 06m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8b9f5199793f49abaa7bd7c77425186d : SUCCESS in 1h 55m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0017e9d7ae2d4444b6e083eb02cdddd4 : SUCCESS in 54m 14s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/90bc03e2108c4ae0b81c2c1ccc1c83fa : SUCCESS in 1h 13m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d350d4604b84442fb66bb55105931390 : FAILURE in 1h 27m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dc83cb7e6bf740769a81117e540193c7 : SUCCESS in 57m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9a81afcaa3f845578daf5474164226d6 : SUCCESS in 32m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c6ebe8abf0d94636be6164b18e13656c : SUCCESS in 43m 07s\n- manila-tempest-plugin-lvm-yoga https://zuul.opendev.org/t/openstack/build/b89568c70e6549eab978ae8d3e839ba8 : SUCCESS in 1h 01m 26s\n- manila-tempest-plugin-lvm-xena https://zuul.opendev.org/t/openstack/build/8f7a37871bcc4b23a3bdf2c2827e1f61 : SUCCESS in 1h 07m 20s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/3a354ea75ec7472fb712b8fe5f3162a4 : SUCCESS in 57m 58s\n- manila-tempest-plugin-dummy-no-dhss-rbac https://zuul.opendev.org/t/openstack/build/bcbeb46bf4c94544bf4c9110af9b8729 : SUCCESS in 29m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b441d6a8c15a4a90973d0b2587d15718 : SUCCESS in 44m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8254f813c01142f0b825a159e444ea4a : SUCCESS in 30m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d0402b5754f0c214e96c60e305f2b42fb94a76e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 20:47:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b92d60239bd70fdc99eefb6072c41177089c0669","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 21:47:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0728df9e925b4774b2e6adf8156c5e2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e614866291d49f2a214ba826be71efa : SUCCESS in 3m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/346e9199cb8e4524857a80a652c01205 : SUCCESS in 58m 51s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7ad11ae1567442a29be11a9ce7b3b517 : SUCCESS in 26m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bd6f62b24d2c47baad6b34c1ef59fcd4 : SUCCESS in 30m 15s","accounts_in_message":[],"_revision_number":2},{"id":"96d45d2c17985e028d9195c1d5bfbfc6d9b0f02a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 21:47:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cfcde25567b62d1e550beca3bf0de25a205a0470","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2022-11-22 22:41:47.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/85/862785/2/upstream-check/manila-tempest-plugin-ontap-dhss/2a0354e/ : SUCCESS in 1h 53m 01s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/85/862785/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/56fa912/ : RETRY_LIMIT in 8m 16s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"0e377a1adc54a27ccf9dc15deb423b4e6e569384","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2022-11-23 04:00:17.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/85/862785/2/check/DellEMC_Unity_Manila_dhss_true/f123665/DellEMC_Unity_Manila_dhss_true/1040 : FAILURE in 3h 44m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/85/862785/2/check/DellEMC_Unity_Manila_dhss_false/ab550a0/DellEMC_Unity_Manila_dhss_false/1108 : SUCCESS in 1h 21m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"120862e4c1218503092c5e2b62bbb251b76872b0","revisions":{"28be225786316ffae24ab3806201c152092d2766":{"kind":"REWORK","_number":1,"created":"2022-10-27 08:27:50.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/85/862785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/862785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/862785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/862785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/862785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/862785/1"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-24 12:42:23.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 08:27:40.000000000","tz":180},"subject":"[RBAC] Add share export locations tests","message":"[RBAC] Add share export locations tests\n\nThe tests validate policy of share export locations actions\nfor admin,member and reader users in a project scope.\n\nChange-Id: I756779789becbb0aeec74380d40f9c5fb4cd3fe8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/28be225786316ffae24ab3806201c152092d2766"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/28be225786316ffae24ab3806201c152092d2766"}]},"branch":"refs/heads/master"},"120862e4c1218503092c5e2b62bbb251b76872b0":{"kind":"REWORK","_number":2,"created":"2022-10-27 08:30:55.000000000","uploader":{"_account_id":19262,"name":"Liron Kuchlani","email":"lkuchlan@redhat.com","username":"lkuchlan"},"ref":"refs/changes/85/862785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila-tempest-plugin","ref":"refs/changes/85/862785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/862785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/862785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/862785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila-tempest-plugin refs/changes/85/862785/2"}}},"commit":{"parents":[{"commit":"3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d","subject":"More Snapshot metadata field verification \u003e\u003d 2.73","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/3d2b7bb036fd8fd3bff51fa14e300d1cb480e56d"}]}],"author":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-24 12:42:23.000000000","tz":180},"committer":{"name":"lkuchlan","email":"lkuchlan@redhat.com","date":"2022-10-27 08:30:46.000000000","tz":180},"subject":"[RBAC] Add share export locations tests","message":"[RBAC] Add share export locations tests\n\nThe tests validate policy of share export locations actions\nfor admin,member and reader users in a project scope.\n\nChange-Id: I756779789becbb0aeec74380d40f9c5fb4cd3fe8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/120862e4c1218503092c5e2b62bbb251b76872b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila-tempest-plugin/commit/120862e4c1218503092c5e2b62bbb251b76872b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}}]}],"submit_requirements":[]}
